)]}'
{"id":"openstack%2Fneutron-lib~991965","triplet_id":"openstack%2Fneutron-lib~master~Ie38eee43a22ed47581375a8fb32751a763a54c08","project":"openstack/neutron-lib","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-06-17 16:37:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie38eee43a22ed47581375a8fb32751a763a54c08","subject":"typing: Add types for ACTION_MAP","status":"MERGED","created":"2026-06-06 10:45:38.000000000","updated":"2026-06-17 16:39:48.000000000","submitted":"2026-06-17 16:37:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991965-typing","meta_rev_id":"41eb22321d3c3e910f82574aef366fb1d03f24fd","_number":991965,"virtual_id_number":991965,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-17 16:37:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-06-17 06:40:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-06-17 08:41:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-06-17 08:41:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-06 12:38:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-08 12:50:12.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-06-17 06:40:01.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-06-17 08:41:37.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"0c4402f397dc8a3cdefecd5e133aca13db976390","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-06 10:45:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"406aef0783e90eefa7548ab572a1fd9eefa1f104","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 12:38:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c454c02215aa40dc8c933f3163a01ff6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14aba4cc1b484820b96b95b9a7a3b58e : FAILURE in 6m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f00113658f64f3d933e38fbf49f3640 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/643d68d99ddf4674924ad51080c1e80f : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e09364f2e1fe4fa695d2c69e92b73df6 : SUCCESS in 3m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3369cb29eb0f421c9faed7825fcfd72d : SUCCESS in 2m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/857c5ed3a19047a08a30de283c764127 : SUCCESS in 7m 42s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c47f905d35fd48d9a8646fc2cabe6f04 : SUCCESS in 1h 02m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11368321e7bb4c4187e9bf280781404d : SUCCESS in 5m 27s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9c71e86bd8dd45e3a547b94dda09b45c : SUCCESS in 1h 30m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3322b30cb5d54837825409a2efb86db1 : SUCCESS in 1h 51m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e9c8a2ab46d04206ae0c7ae9460087c7 : SUCCESS in 33m 45s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/0089266c6530481fb1fe4dc6778e3314 : SUCCESS in 26m 37s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/97686cb976544dae8d1bb2efc468e2e8 : SUCCESS in 5m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6d113bb373f148599dc3bddfa05642b6 : FAILURE in 49m 02s","accounts_in_message":[],"_revision_number":1},{"id":"1d0c38a890d57882b1c8304d3c3ba140784f6e79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-08 12:27:24.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"048f87d79ba77280041168912c077a82991cc26b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 14:40:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ae3ab8a30384474a3d10ca78fb1b842\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0a53076baf74e8285bfd86b0dc9b197 : SUCCESS in 10m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8d7c476da064801b2e5d84c1d67cb17 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/918510867c02499a8ceda4f053b0c471 : SUCCESS in 6m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bdf2a1004d064f76a19ac472a03eb6b9 : SUCCESS in 2m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8d1c8857a455407a9af7eabbf4433c55 : SUCCESS in 3m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/81bb2ef6b3b846b2a5842e128e16fd51 : SUCCESS in 4m 19s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50f8e2476be647f3bf9f6bdb1f0bebba : SUCCESS in 1h 46m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf539a481d8845a888d3aafd8f62ab34 : SUCCESS in 5m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/347957e8ce77494abf6aed8963b00413 : SUCCESS in 1h 53m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4851a53ecc814c38b288af202f2a4911 : SUCCESS in 2h 11m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/44586bdda36642c394c6f8f9c8d3118c : SUCCESS in 49m 26s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/362b60782d6145c6aef47033a56647b2 : SUCCESS in 22m 15s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/df6691d203c9496cb5767e492fdc7cd9 : SUCCESS in 7m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7843cd5b0b9b487e88204e1861e2b9a5 : SUCCESS in 1h 27m 15s","accounts_in_message":[],"_revision_number":2},{"id":"6ae5377ae129750c65870aae16be88a1790167a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-16 15:58:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"529b9afdffa008583bdbcac79d04315f66dfba97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 18:08:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c225758206894c29b7803e148f90fb53\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae865c5f5f434f4685c789018b373348 : SUCCESS in 6m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edbed753788043bf9cb315c9806c4f70 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1fdfade2b9c49c791142572e35c6bbe : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/54684449903e4ee9b9c6f3bd161132c6 : SUCCESS in 2m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/971e47e4f3234a3cb251f0c80500989c : SUCCESS in 4m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dcee320a56a44fa286ad83c90c7cc851 : SUCCESS in 3m 56s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ad7e3954aa74c038b2399ec8808feca : SUCCESS in 1h 51m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55ed6e5dfbcd4a7ca8879e03eb142905 : SUCCESS in 3m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0cbb0ddac00f485392fae2bb7d37ca63 : SUCCESS in 1h 21m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/50f81be85bd2486c9452e156fd6e6b96 : SUCCESS in 2h 03m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6fea43b9fcf84899a06d33a642e09365 : SUCCESS in 34m 04s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/372f541cdf894e83994dfbb059bc42af : SUCCESS in 25m 49s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9091843236724c22b1363a2ba1af2d2f : SUCCESS in 6m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c1ea07eed74742b9bfc4f80a6750e9c6 : SUCCESS in 1h 29m 15s","accounts_in_message":[],"_revision_number":3},{"id":"985c90f9945b1815d4f4c38525c095d965b18c29","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-17 06:40:01.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"31a33c0842f1659e50d102d00eff80ba98bf5c2e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-06-17 08:41:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fea81946c3b2b8455f813b3f8cd20b458d9c6926","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 14:25:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cf068ddb3c58ad4e907f316520624880429ec11b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 16:37:54.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35ddb1a71ba544d5896062e9b998a127\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c71bc0370ab41f68628048fe4da71bc : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f93fb9c0ced48f5a3f19b3682779f41 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ed1d062fe7a44e388614cc5f05743b5 : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e15be586e664c058e1f1087ec8b0ac7 : SUCCESS in 3m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96d1c0f4f3c448cf9310382a5bac8818 : SUCCESS in 1h 31m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b689031b99b740518e61e49c62a7ac6a : SUCCESS in 5m 23s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/33a4dc0964444ffcb0bce13114f511e8 : SUCCESS in 1h 29m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/45f393343a374959b54705b9ab66b0eb : SUCCESS in 2h 11m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d0bffd0923b644f184c1f193c6a4eaa6 : SUCCESS in 1h 26m 37s","accounts_in_message":[],"_revision_number":3},{"id":"31e7178cf3a88a5a6345c07e42059fa6381eef92","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 16:37:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"41eb22321d3c3e910f82574aef366fb1d03f24fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 16:39:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2bb6260f2374bffba35719c10893177\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a97b045ddd9a482e87a9f81604e74c60 : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"230d219ddd959e9ceb6e44cf1679bd32d5f3d4fb","revisions":{"3e88ae3e089523e629f67665040e8850a62e71c1":{"kind":"REWORK","_number":1,"created":"2026-06-06 10:45:38.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/991965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/65/991965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/991965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/991965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/991965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/65/991965/1"}}},"commit":{"parents":[{"commit":"89811ccb9b94dfff914b636db2c7950b86aacf39","subject":"typing: Add types for SUB_RESOURCE_ATTRIBUTE_MAP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/89811ccb9b94dfff914b636db2c7950b86aacf39"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-03 16:45:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-06 10:45:20.000000000","tz":60},"subject":"typing: Add types for ACTION_MAP","message":"typing: Add types for ACTION_MAP\n\nThis one is comparatively simple.\n\nChange-Id: Ie38eee43a22ed47581375a8fb32751a763a54c08\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3e88ae3e089523e629f67665040e8850a62e71c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3e88ae3e089523e629f67665040e8850a62e71c1"}]},"branch":"refs/heads/master"},"b20718359533037fb043644c857dedf016469323":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-08 12:27:24.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/991965/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/65/991965/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/991965/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/991965/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/991965/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/65/991965/2"}}},"commit":{"parents":[{"commit":"dd12839743092f3a1026d763346180473e2788d7","subject":"typing: Add types for SUB_RESOURCE_ATTRIBUTE_MAP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/dd12839743092f3a1026d763346180473e2788d7"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-03 16:45:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-08 12:27:10.000000000","tz":60},"subject":"typing: Add types for ACTION_MAP","message":"typing: Add types for ACTION_MAP\n\nThis one is comparatively simple.\n\nChange-Id: Ie38eee43a22ed47581375a8fb32751a763a54c08\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b20718359533037fb043644c857dedf016469323"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b20718359533037fb043644c857dedf016469323"}]},"branch":"refs/heads/master"},"230d219ddd959e9ceb6e44cf1679bd32d5f3d4fb":{"kind":"REWORK","_number":3,"created":"2026-06-16 15:58:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/991965/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/65/991965/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/991965/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/991965/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/991965/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/65/991965/3"}}},"commit":{"parents":[{"commit":"b8010afeb2e2954b84f0cdf8ebfdb5ecb3da7221","subject":"typing: Add types for SUB_RESOURCE_ATTRIBUTE_MAP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b8010afeb2e2954b84f0cdf8ebfdb5ecb3da7221"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-03 16:45:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-16 15:50:19.000000000","tz":60},"subject":"typing: Add types for ACTION_MAP","message":"typing: Add types for ACTION_MAP\n\nThis one is comparatively simple.\n\nChange-Id: Ie38eee43a22ed47581375a8fb32751a763a54c08\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/230d219ddd959e9ceb6e44cf1679bd32d5f3d4fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/230d219ddd959e9ceb6e44cf1679bd32d5f3d4fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
