)]}'
{"id":"openstack%2Fneutron-lib~992461","triplet_id":"openstack%2Fneutron-lib~master~I93f2b1c09d5d8da4adb6a918181ecfb51732186e","project":"openstack/neutron-lib","branch":"master","topic":"bug/2131666","attention_set":{},"removed_from_attention_set":{"15554":{"account":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"last_update":"2026-06-11 22:08:31.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-06-11 08:42:05.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}},"hashtags":[],"change_id":"I93f2b1c09d5d8da4adb6a918181ecfb51732186e","subject":"Extend L2AgentExtension API: handle_switch_restart","status":"MERGED","created":"2026-06-09 14:04:39.000000000","updated":"2026-06-11 22:09:38.000000000","submitted":"2026-06-11 22:08:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"992461-bug/2131666","meta_rev_id":"6a22fd1c2250dea054e89a13d5cb663bda8ebb3f","_number":992461,"virtual_id_number":992461,"owner":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-11 22:08:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2026-06-11 08:42:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-06-11 17:26:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2026-06-11 19:54:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-09 16:30:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-09 21:44:21.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-06-10 08:12:37.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-06-10 09:49:57.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-06-11 08:42:05.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-06-11 17:26:09.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"f3f3411d7ee6582014c0f04f0fb595317038034d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2026-06-09 14:04:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe74b84cd9d9593630f3257ae711037dd9dc471d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 16:30:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8fc65aff69549c9bc874e21438fd3a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76dc7e4ee40548008b7def1219031bed : SUCCESS in 5m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/499e36da996a4bd2a858e0efda6a2ff0 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cada4e7eb9846bea0495a6f6ba7e178 : SUCCESS in 7m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/037ba142928b4977b14cdcd6a9cdcb66 : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a0baefe1137a4638bbe082d258746696 : SUCCESS in 3m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/578533c5d5bb4e77883905b26cb3e7e9 : SUCCESS in 3m 57s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/79f68be1dfe642a18ebcbf420eba90ee : SUCCESS in 1h 00m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fb0b494bc7b4bec9d73b4edb6e5098e : SUCCESS in 4m 54s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/379a45c228d847c9990c8b5587bed60d : FAILURE in 1h 35m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ae3c7b3e48494d7e8249b20faae47039 : SUCCESS in 2h 22m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e444bd4682714158a410742570f8b13b : SUCCESS in 35m 43s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/9615bcc9ab78483ca730ece4868f6ca8 : FAILURE in 25m 50s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3260a96de8064c22a14db6cf2c89f667 : SUCCESS in 5m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0791856c89184ad0be4a9ecdcdf4e67a : SUCCESS in 1h 06m 05s","accounts_in_message":[],"_revision_number":1},{"id":"fa09c6ce4d0d873ed2c36390df614ba6ca72943d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-10 09:49:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"58c85cdf90d99ac7dc6b259024579ff36763065f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2026-06-10 13:28:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"107cc72b7e4eaa6fdedf97c75c13e14ad2bdf126","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2026-06-10 13:31:01.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"86d202f65e345a747ad2f534249368ec9d2bbaaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2026-06-10 15:22:17.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"9a4a9588356ab5de2e27d77cef154fe0a1756086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 17:19:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/042993201ee944449e18b15145da2424\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d9c8a11341c47d89ed6f1797e2ec95b : SUCCESS in 9m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/590881d146cc472581a7e4710cc56fe4 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b80cf50f6cd5444995e5c5aa6410f3d2 : FAILURE in 8m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cbacb12c3bd6403b8010ace194b44bad : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5183ba8a148448649e68c88d34f92091 : SUCCESS in 4m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7fca1190f9184231bd3a699c3581a506 : SUCCESS in 7m 22s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f65e3866246247dc9421139495f214f4 : SUCCESS in 1h 44m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/737ee1b21bff4522a1ee15eae9568abd : SUCCESS in 4m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fa96f9de6d4043dba7a0b6e1482638af : SUCCESS in 1h 33m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/61509fd87d324edc85622ce614495c31 : SUCCESS in 1h 43m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cff49ab1f2a94f53bc4cfd726a769d34 : SUCCESS in 35m 28s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/9c263902dd5d4f0586d778ec596153ba : SUCCESS in 21m 22s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/4fc0c3192a2c444e97c6fdfc0ab6574e : SUCCESS in 5m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/eb4f620771004035813900c151c5bd98 : SUCCESS in 1h 30m 43s","accounts_in_message":[],"_revision_number":3},{"id":"c16ca697965e7c22e06954336af3ba2948b6500b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-11 08:42:05.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"db8d25951a6a3a68de3ca96597f6f77cfebe1259","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-11 17:26:09.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"91adb4dd1f5fa27daffd3d6b071c4b6efbdac3b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 19:43:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5510bf5050043e69086942317d5d566\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/768361c21d9d4e18b9ad0b2b374c753c : SUCCESS in 8m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b379a5c1ee764b33a8276bd5e0e858fb : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/305f9a479f504644a669f01b12a011b0 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/46cd1aecd204446da4cd9346f5aed9f7 : SUCCESS in 2m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/98be7a66671845638da8764491234f0b : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ce813a8550044735bc01a01a8675cc70 : SUCCESS in 6m 33s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0c2fe437c5fa43dda6983967289c76b4 : SUCCESS in 1h 58m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5dda03649a754a949ac11a6519f5485f : SUCCESS in 4m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6ee47d9b4de74d2da80dd5c65cda7da4 : SUCCESS in 1h 33m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/99df139d334a4877912ae95b87975d5c : SUCCESS in 2h 16m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2ef43f1038b2438697af073048cb8443 : SUCCESS in 36m 26s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/7dddfe42a17d47b19d42dab4d385b110 : SUCCESS in 17m 09s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/72d4cffd4bb541b096db5f4d777b1277 : SUCCESS in 4m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7c39717bc4054f6e98cba23fc27662ce : SUCCESS in 46m 49s","accounts_in_message":[],"_revision_number":3},{"id":"bc327f5dc8cd1831bf167f168a37c8a423c98cbd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-11 19:54:10.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"73b9f6a2562b76b01b979a747ec95f1189cb11ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 19:54:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f3909950b8527ee7a45c52be6eaaa4dada99bff4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 22:08:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a093a1839e548e8b23e43b0c2e1070b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c67f1e4e0f184eac8071c957fe275d02 : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd85dcdd3a364c3c81a4be6a207e599b : SUCCESS in 6m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c2b2fda79d948d595d21930a0693000 : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ef2b4d4fb864405bb34b326ce5feb089 : SUCCESS in 3m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dfb47ebe7a5d41a6833d072c6b57eac9 : SUCCESS in 1h 53m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59c75ab0b49e419390bd95f1a0e8d3b7 : SUCCESS in 4m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a4760ff0f7ec445e82b8f7c3187d1b5e : SUCCESS in 1h 53m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a713f27c4116468fbaab0e1ed988e44c : SUCCESS in 2h 13m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/569127525ea6445590e94a380de6cb12 : SUCCESS in 1h 28m 37s","accounts_in_message":[],"_revision_number":3},{"id":"8cbdc6ea925e4b4de8eec82fa82c5a35af291342","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 22:08:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6a22fd1c2250dea054e89a13d5cb663bda8ebb3f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 22:09:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e6cbc55ee534db9908b7958e04d392d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93f36bb4c12b4ce48f908ce5542703a9 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40aa82edd5af4a32b56b0ab551726c6c : SUCCESS in 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fc0559a26b7f65e9830e9ec0c42a255e8ee2a2b8","revisions":{"049fb4ff1251658475d85e6bf9fe53bf943c41d7":{"kind":"REWORK","_number":1,"created":"2026-06-09 14:04:39.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/61/992461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/61/992461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/61/992461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/61/992461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/61/992461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/61/992461/1"}}},"commit":{"parents":[{"commit":"d794fe186a9469c5055684f0a9fab7cfa02a7ae0","subject":"Merge \"api-ref: Add missing zero quota information in subnetpools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d794fe186a9469c5055684f0a9fab7cfa02a7ae0"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2026-04-08 13:47:00.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2026-06-09 14:01:02.000000000","tz":120},"subject":"Extend L2AgentExtension API: handle_switch_restart","message":"Extend L2AgentExtension API: handle_switch_restart\n\nThe method handle_switch_restart may be needed if an agent extension\nneeds (partial) re-initialization when the switch managed by the agent\nis restarted. If your agent extension is not affected, just provide a\nnoop implementation.\n\nWe introduce this method because we realized in bug #2131666, that ovs\nflows installed by agent extension are lost then the managed switch is\nrestarted and we need a way to re-install them.\n\nChange-Id: I93f2b1c09d5d8da4adb6a918181ecfb51732186e\nNeeded-By: https://review.opendev.org/c/openstack/neutron/+/967378\nPartial-Bug: #2131666\nSigned-off-by: Bence Romsics \u003cbence.romsics@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/049fb4ff1251658475d85e6bf9fe53bf943c41d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/049fb4ff1251658475d85e6bf9fe53bf943c41d7"}]},"branch":"refs/heads/master"},"57f39b8a5e37cba7122ef5a7da355e984db95418":{"kind":"REWORK","_number":2,"created":"2026-06-10 13:28:45.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/61/992461/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/61/992461/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/61/992461/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/61/992461/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/61/992461/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/61/992461/2"}}},"commit":{"parents":[{"commit":"e3ef0d1618ae583ce6d4176d3fb737367dfb3b68","subject":"Merge \"Run mypy from tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e3ef0d1618ae583ce6d4176d3fb737367dfb3b68"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2026-04-08 13:47:00.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2026-06-10 11:57:44.000000000","tz":120},"subject":"Extend L2AgentExtension API: handle_switch_restart","message":"Extend L2AgentExtension API: handle_switch_restart\n\nThe method handle_switch_restart is needed if an agent extension needs\n(partial) re-initialization when the switch managed by the agent is\nrestarted. If your agent extension is not affected, there is nothing\nto do.\n\nWe introduce this method because we realized in bug #2131666, that\novs flows installed by agent extension are lost then the managed\nswitch is restarted and we need a way to re-install them.\n\nChange-Id: I93f2b1c09d5d8da4adb6a918181ecfb51732186e\nNeeded-By: https://review.opendev.org/c/openstack/neutron/+/967378\nPartial-Bug: #2131666\nSigned-off-by: Bence Romsics \u003cbence.romsics@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/57f39b8a5e37cba7122ef5a7da355e984db95418"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/57f39b8a5e37cba7122ef5a7da355e984db95418"}]},"branch":"refs/heads/master"},"fc0559a26b7f65e9830e9ec0c42a255e8ee2a2b8":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-06-10 15:22:17.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/61/992461/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/61/992461/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/61/992461/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/61/992461/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/61/992461/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/61/992461/3"}}},"commit":{"parents":[{"commit":"e3ef0d1618ae583ce6d4176d3fb737367dfb3b68","subject":"Merge \"Run mypy from tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e3ef0d1618ae583ce6d4176d3fb737367dfb3b68"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2026-04-08 13:47:00.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2026-06-10 15:21:52.000000000","tz":120},"subject":"Extend L2AgentExtension API: handle_switch_restart","message":"Extend L2AgentExtension API: handle_switch_restart\n\nThe method handle_switch_restart is needed if an agent extension needs\n(partial) re-initialization when the switch managed by the agent is\nrestarted. If your agent extension is not affected, there is nothing\nto do.\n\nWe introduce this method because we realized in bug #2131666, that\novs flows installed by agent extension are lost when the managed\nswitch is restarted and we need a way to re-install them.\n\nChange-Id: I93f2b1c09d5d8da4adb6a918181ecfb51732186e\nNeeded-By: https://review.opendev.org/c/openstack/neutron/+/967378\nPartial-Bug: #2131666\nSigned-off-by: Bence Romsics \u003cbence.romsics@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fc0559a26b7f65e9830e9ec0c42a255e8ee2a2b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fc0559a26b7f65e9830e9ec0c42a255e8ee2a2b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
