)]}'
{"id":"openstack%2Fneutron-lib~994589","triplet_id":"openstack%2Fneutron-lib~master~Ieafd1ae34768199a6554fac1f308fb306b975d27","project":"openstack/neutron-lib","branch":"master","topic":"no-else-return","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-06-25 10:54:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ieafd1ae34768199a6554fac1f308fb306b975d27","subject":"Fix unnecessary-pass pylint warnings","status":"MERGED","created":"2026-06-24 05:47:18.000000000","updated":"2026-06-25 10:56:05.000000000","submitted":"2026-06-25 10:54:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994589-no-else-return","meta_rev_id":"ce3d5451fd54961029c3605a8d95739a251cdd9b","_number":994589,"virtual_id_number":994589,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-25 10:54:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-06-24 06:09:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-06-25 08:35:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-06-25 08:35:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-06-24 06:09:59.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-24 06:09:59.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-06-24 06:09:59.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-06-24 06:09:59.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-06-24 08:14:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-25 08:35:33.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"4d7d7a7669ebca5dc8e03ab5dcc22cd55a5d5704","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-24 05:47:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08c34823aa328eed2d88233a85164737459bcaf0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-24 06:09:59.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"b7cac999bd0763fdc30f836027c3c4358f51fa29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 08:14:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c168301a11b747f6a05d6752f19964b5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f01f3f0351c4795935e9b8388db92fb : SUCCESS in 6m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37ed3bf3dcb042ecb7d4de4e752bba20 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a71ff37d97c412aad41c2d199fe1846 : SUCCESS in 6m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c6144646afdd4bff96ec8c880807b5a0 : SUCCESS in 5m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2313f793534f487fbc42a9b67971bd8f : SUCCESS in 3m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4be680a3f6f04797ae87c9b86bda1c9d : SUCCESS in 5m 53s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5087f3b786284ba79f9646d3de0a65f7 : SUCCESS in 1h 55m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64f057c6d21c4a3b973fe4a50c5a64dd : SUCCESS in 3m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0ba66f3763a24fd3a31ee73ca802038a : SUCCESS in 1h 18m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2be57b3787c143e3b3937e20f6c3edf2 : SUCCESS in 2h 21m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/402e28ce81b14538a739c0a0db67880b : SUCCESS in 31m 23s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/e8ceb1c1ec594becb8e71b0e9480f8e4 : SUCCESS in 26m 31s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/df8b22d0a9814678beca8e0350d1ed3c : SUCCESS in 5m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/96cbd7c446934874848aec3b6aea4a83 : SUCCESS in 1h 32m 56s","accounts_in_message":[],"_revision_number":1},{"id":"70cee5ccc6dd15b5b1c2402484a856adfb1a2e8c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-06-25 08:35:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7d5884b2b5ee1ac22fe6503a3199bebb05a7d39d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 08:36:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"39d237f376dd68e0074771fdd88fe0029a62b8a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 10:54:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf5b556e39b54312b113ed76d2b10315\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfd79d17dff04548928272bfb83ad8ca : SUCCESS in 9m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc0a40fc737841f89fb76a6778fd597f : SUCCESS in 6m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ac996427841423098ab16a416c1265f : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e6c132d0dc784becace2dda465fcbc74 : SUCCESS in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5bca4b93055f4783949963d1cbe203c7 : SUCCESS in 6m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9b01772e5eba4c0ca4f25724ad7927b2 : SUCCESS in 1h 46m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3d1eb64791f4b03bd18fc72eae964b2 : SUCCESS in 4m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b7976017033546e590077c65122062c6 : SUCCESS in 1h 37m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8bd70c07910c417dab26295eaffe39ae : SUCCESS in 1h 57m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4e98d4c4ce30490090844b073bc91a01 : SUCCESS in 1h 35m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d931d5315a4a96155f603bf781f40234618709ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 10:54:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ce3d5451fd54961029c3605a8d95739a251cdd9b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 10:56:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa62376ade764dcc909abf0745275d73\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95d9843e2d2848a79e4d8b24690c7757 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"743bfb4b53bafca9bbe056d67786aa951af3d3ce","revisions":{"743bfb4b53bafca9bbe056d67786aa951af3d3ce":{"kind":"REWORK","_number":1,"created":"2026-06-24 05:47:18.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/89/994589/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/89/994589/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/89/994589/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/89/994589/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/89/994589/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/89/994589/1"}}},"commit":{"parents":[{"commit":"056feb26f799a42253e32c09730d68a3dfaa5f65","subject":"Fix superfluous-parens pylint warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/056feb26f799a42253e32c09730d68a3dfaa5f65"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-24 05:24:10.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-24 05:24:10.000000000","tz":-240},"subject":"Fix unnecessary-pass pylint warnings","message":"Fix unnecessary-pass pylint warnings\n\nLet\u0027s fix them and enable the pylint check.\n\nTrivialFix\n\nChange-Id: Ieafd1ae34768199a6554fac1f308fb306b975d27\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/743bfb4b53bafca9bbe056d67786aa951af3d3ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/743bfb4b53bafca9bbe056d67786aa951af3d3ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
