)]}'
{"id":"openstack%2Fneutron-lib~994590","triplet_id":"openstack%2Fneutron-lib~master~I802fd313f40bcb64e11816ecd94d73afccc7cb05","project":"openstack/neutron-lib","branch":"master","topic":"no-else-return","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-06-25 11:01:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I802fd313f40bcb64e11816ecd94d73afccc7cb05","subject":"Fix no-else-return pylint warnings","status":"MERGED","created":"2026-06-24 05:47:18.000000000","updated":"2026-06-25 11:03:05.000000000","submitted":"2026-06-25 11:01:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994590-no-else-return","meta_rev_id":"2bb03747e0844b215798cb74fc51b75ef5968a27","_number":994590,"virtual_id_number":994590,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-25 11:01:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-06-24 06:10:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-06-25 08:35:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-06-25 08:35:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-06-24 06:10:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-24 06:10:37.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-06-24 06:10:37.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-06-24 06:10:37.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-06-24 08:14:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-25 08:35:45.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"2cd7d520167c96108eac5196cf4bff6f276d6bef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-24 05:47:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce9d6906e1e9ab811011ef69ea931247cd7c9cbb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-24 06:10:37.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"49a402830744bbf41a971325f97dfe170d02706c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 08:14:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/159cd2e44066476b8bccf61dabf5dfa4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/353b1f627d654d258e6cf6261dffc28d : SUCCESS in 8m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/225d7668b6d347a4ad3716c243e3c2c6 : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d6ae79b922c44ddaaa2af5d7dc4dce0 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0dc4bd62666c4e10bae8f0b3534595bd : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c8423c00fcd48bea3174685b4eadd5c : SUCCESS in 3m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cbfe1b7340c44fcd8241ff0acf62a70a : SUCCESS in 5m 53s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78fc97d8e2004581abc3ce9b872ca679 : SUCCESS in 1h 35m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b15a099a565b4f40bfd443fc4129f28b : SUCCESS in 5m 18s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a20a3ae127e6447895d49a6333e35559 : SUCCESS in 1h 28m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b8399020d1a741b2a046822dac8e9f9c : SUCCESS in 2h 26m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/acbe116fdb9749deb1ae386766a8869b : SUCCESS in 30m 21s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/7b809238a8c8478eab4cbfa8aaaa0c9d : SUCCESS in 26m 04s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/99157c671b444438b66e665323613bc9 : SUCCESS in 5m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/352e6efee99f4f609a8d30d547f87dd8 : SUCCESS in 50m 59s","accounts_in_message":[],"_revision_number":1},{"id":"34e975fd620d7b871d54aeffd7a93e980c16e58d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-06-25 08:35:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eaf1bc97b8029bf297257023ed9e583fcec6eafc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 08:36:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ff17f116099e78168589afa9d3b96388e401413","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 11:01:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6633465840f4f85ba8bec3f9344d248\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e08b70a43e4e4976acf7d00d540d8f15 : SUCCESS in 8m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/085c0b0bd1f7429c98fb1f56c41a6a9f : SUCCESS in 7m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/86201ffbf10d4c94ba195319a04df145 : SUCCESS in 5m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/140d76a1f3034161b018140e064aacc0 : SUCCESS in 3m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b91b49a579e34716b90fb3fcc5934f54 : SUCCESS in 6m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eae7af7e3ffc4e0fb6c9976814094e65 : SUCCESS in 1h 41m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/442c2eb04ebf46e194f0e3858d91aa66 : SUCCESS in 5m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b70f2ed8f09844a2a0d9f76253ab56d2 : SUCCESS in 1h 18m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5769f2566b7f43da88fac62f72c6631b : SUCCESS in 2h 24m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b85905eadee64682953082e931997d34 : SUCCESS in 1h 35m 53s","accounts_in_message":[],"_revision_number":1},{"id":"2896be0938c8ec45b2e110b556f723da43e19333","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 11:01:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2bb03747e0844b215798cb74fc51b75ef5968a27","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 11:03:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca19144bb1a44dc2ade0bb41bb54e760\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30b6df7996c34e5cba23839e14c7a4b0 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1af691d06cfc180ab6521c4508eb946bb6e35f1f","revisions":{"1af691d06cfc180ab6521c4508eb946bb6e35f1f":{"kind":"REWORK","_number":1,"created":"2026-06-24 05:47:18.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/90/994590/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/90/994590/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/994590/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/994590/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/90/994590/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/90/994590/1"}}},"commit":{"parents":[{"commit":"743bfb4b53bafca9bbe056d67786aa951af3d3ce","subject":"Fix unnecessary-pass pylint warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/743bfb4b53bafca9bbe056d67786aa951af3d3ce"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-24 05:33:13.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-24 05:46:32.000000000","tz":-240},"subject":"Fix no-else-return pylint warnings","message":"Fix no-else-return pylint warnings\n\nLet\u0027s fix them and enable the pylint check.\n\nAlso start enforcing signature-differs as there\nare no warnings for it.\n\nTrivialFix\n\nChange-Id: I802fd313f40bcb64e11816ecd94d73afccc7cb05\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1af691d06cfc180ab6521c4508eb946bb6e35f1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1af691d06cfc180ab6521c4508eb946bb6e35f1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
