)]}'
{"id":"openstack%2Fneutron-lib~994683","triplet_id":"openstack%2Fneutron-lib~master~I97e7647cefa8404a9ea6f5ba0690f99dc8ad80b0","project":"openstack/neutron-lib","branch":"master","topic":"python-next-jobs","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-06-24 17:15:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I97e7647cefa8404a9ea6f5ba0690f99dc8ad80b0","subject":"pre-commit: Bump versions","status":"MERGED","created":"2026-06-24 11:25:11.000000000","updated":"2026-06-24 17:16:21.000000000","submitted":"2026-06-24 17:15:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994683-python-next-jobs","meta_rev_id":"decb22a1a205817f98af23993c00547289268d25","_number":994683,"virtual_id_number":994683,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-24 17:15:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-06-24 12:58:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-06-24 13:30:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-06-24 13:30:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-24 11:38:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-24 12:58:55.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-06-24 13:30:50.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"cca236ded1250d70d370fd003046c56a543f0daa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-24 11:25:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2268b4e180d6dd6bc3b6a0087e2ef7b19d974558","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 11:38:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c3a1e6c21de4ca0b89daff9a4e76577\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9af7ee5ff1341d68e8eddf228becc95 : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edb79c3e58a2480397de9c1e17ea5a04 : FAILURE in 2m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa088bf0cfc941ddbe0439f87186f8b9 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":1},{"id":"40b7ba1d45e0ffe400fab4fcb2701151a4158bb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-24 12:55:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"d6b4cfe26c2545411ab53d2bf3d75dba16cf06a1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-24 12:58:55.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6946dee039b446f6e1a0a10c3d704d4573bbf405","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-24 13:01:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"34304b02bd7e730a8870c1239d7b7a386e22c781","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-24 13:30:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6acc34c2d1e423e500138d11da33dd9b00ca80a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 15:12:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16eef65618b448288b004fece5ab0962\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b5919f957de405eb5d7ecb3d5fe451f : SUCCESS in 5m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8635fa2eeb8a4ea2a4a9b802df57972d : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04d962a84ac9420e9931b54806bd1166 : SUCCESS in 6m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b20fe7c2abb349eb81bb8c08237cc269 : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b1fe4a3946094ffe8de834d8e19a3308 : SUCCESS in 2m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ddb3c27417b44087bba97f2335d86351 : SUCCESS in 4m 19s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/259910bd1b5543e0991d13b0ef5a0e66 : SUCCESS in 1h 42m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b0aa7d116fe493998d127eb7212e191 : SUCCESS in 4m 48s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6ce30603af3a4c21a3ddd630c44deea9 : SUCCESS in 1h 29m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/797a764dd4654180ab08a17fa32f22ea : SUCCESS in 2h 16m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b97162ad8e37422885d8dd407a8eae27 : SUCCESS in 31m 53s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/b3a698466836434cb593875122cede8e : SUCCESS in 27m 16s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/10981f6b19c04dfbab8a68ccf11e0966 : SUCCESS in 6m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d2ca0eab88d44a18a74210139aae8879 : SUCCESS in 1h 11m 10s","accounts_in_message":[],"_revision_number":2},{"id":"626981d1f49dcbfdc53a3f0e33de07bbb42a340d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 15:13:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"08e337c232fd7c21dfd039b19632b90b8cb371ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 17:15:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/865cdc4549e7453e961a6762e23111e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56db69e0655444cf99296ac2d05b1346 : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fdebeacf3494a34bfc46b54b97771c5 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0fd5ec243cc042b6aebd66ca2f962141 : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/53f8148438974e1198a20331f5ada4ce : SUCCESS in 2m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e60360bf05e4a4d89bddd7644474db4 : SUCCESS in 1h 39m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f58cc82ba15472faabaf70d1eaa0393 : SUCCESS in 6m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fbe036d35c0f4edea3998c00534e505d : SUCCESS in 1h 54m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3c398428a1d2410c8e6af56b7079adb0 : SUCCESS in 2h 01m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9067b223a55841e0b7b1ac3412b9a51d : SUCCESS in 52m 30s","accounts_in_message":[],"_revision_number":2},{"id":"b3ac13b9a8c032fb6defdd884c17acde59194ca4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 17:15:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"decb22a1a205817f98af23993c00547289268d25","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 17:16:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0196db844fe74f5abf6daa8098255895\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97775c93b47a4909b42303cc6ba39fcf : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"57616f6da7b986865552d78847eb84341121bd6a","revisions":{"c88dec08295a110cd3dd778fe2103deaf157e170":{"kind":"REWORK","_number":1,"created":"2026-06-24 11:25:11.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/83/994683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/83/994683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/994683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/994683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/994683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/83/994683/1"}}},"commit":{"parents":[{"commit":"230d219ddd959e9ceb6e44cf1679bd32d5f3d4fb","subject":"typing: Add types for ACTION_MAP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/230d219ddd959e9ceb6e44cf1679bd32d5f3d4fb"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-24 11:22:48.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-24 11:22:48.000000000","tz":60},"subject":"pre-commit: Bump versions","message":"pre-commit: Bump versions\n\nChange-Id: I97e7647cefa8404a9ea6f5ba0690f99dc8ad80b0\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c88dec08295a110cd3dd778fe2103deaf157e170"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c88dec08295a110cd3dd778fe2103deaf157e170"}]},"branch":"refs/heads/master"},"57616f6da7b986865552d78847eb84341121bd6a":{"kind":"REWORK","_number":2,"created":"2026-06-24 12:55:01.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/83/994683/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/83/994683/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/994683/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/994683/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/83/994683/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/83/994683/2"}}},"commit":{"parents":[{"commit":"230d219ddd959e9ceb6e44cf1679bd32d5f3d4fb","subject":"typing: Add types for ACTION_MAP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/230d219ddd959e9ceb6e44cf1679bd32d5f3d4fb"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-24 11:22:48.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-24 12:54:37.000000000","tz":60},"subject":"pre-commit: Bump versions","message":"pre-commit: Bump versions\n\nNote that this highlights a new issue that might need addressing, but\nit\u0027s ignored for now.\n\nChange-Id: I97e7647cefa8404a9ea6f5ba0690f99dc8ad80b0\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/57616f6da7b986865552d78847eb84341121bd6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/57616f6da7b986865552d78847eb84341121bd6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
