)]}'
{"id":"openstack%2Fneutron-lib~995265","triplet_id":"openstack%2Fneutron-lib~master~I583e2e4937bd18f3f044880f95a2b98415440568","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-07-09 14:23:11.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-07-09 23:26:48.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-07-07 09:49:49.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"I583e2e4937bd18f3f044880f95a2b98415440568","subject":"[FWaaS] Set \u0027is_filter\u0027 for the attributes which allows filtering","status":"MERGED","created":"2026-06-29 08:30:53.000000000","updated":"2026-07-09 23:27:44.000000000","submitted":"2026-07-09 23:26:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995265","meta_rev_id":"d25265d7fd2b164dfbc426018e87f4bf2e12674d","_number":995265,"virtual_id_number":995265,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-09 23:26:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-07-09 14:23:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-07-07 09:49:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-07-09 14:23:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-29 08:47:25.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-06-29 10:46:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-29 11:37:03.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"fbbbaf5a759f356b89b813f599c3296a55f9e16a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-29 08:30:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea1ea1d92721ea5e9a132258e5bd5bb56228cab9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-29 08:47:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"87533b574697787a792b5aa0010fb5b9fdf4f111","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 10:46:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/439286260670486982445ed6e2af5c71\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9da5c46f5fa34587b51fd41dab35755a : SUCCESS in 8m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00113e64be8944dbb5367550b63e3b75 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21b1816f620c477d8afc980b8a532fd5 : SUCCESS in 6m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/603feae1f82242118072149da3756006 : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cffa0c155fab4d06ac1dd2522e7f8d6c : SUCCESS in 3m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/328629c5b9d94051a2329ee825302db7 : SUCCESS in 6m 42s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/dcc4b4ce7c1b452fbb060b91d7f23580 : FAILURE in 5m 19s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0361896653254c4ab372889d6f1c50ca : SUCCESS in 1h 55m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b153380bef04e7a87ca41fe5ce7f8f3 : SUCCESS in 3m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f3776bf47c0642b0984ae8672c1c56c7 : SUCCESS in 1h 26m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f4df67d6876343ceb37a680359bcc2be : SUCCESS in 2h 14m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cebfa7294c704bf5a87aa3bbd77ce366 : SUCCESS in 31m 13s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/dd48b549b55d4bc1b2043ad2c93b0044 : SUCCESS in 27m 41s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/561d7206c80f472696703acd20303d07 : SUCCESS in 3m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/aa6308602527461caab3e9a76105bed2 : SUCCESS in 1h 20m 26s","accounts_in_message":[],"_revision_number":1},{"id":"f0e7500899a95e13e594af07c62fc48c5a47d974","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-07-01 08:44:48.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"8a6642b786fdccf1fec54f15955ab4aac4624a11","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-07-01 08:44:57.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d530be83d5f853b25bcedc11050ba00141cb7aa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 10:54:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3dafb4e22244a57a9ea1e47948dc2e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a523fc7ea6be4ca7870d94f271971588 : SUCCESS in 7m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8456b3633bf04b23b08b2f3759eba8b9 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0099a579de444a479a67c4eace3da577 : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b13f5563fc5a41d39cb7eb884c4e1d27 : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8a924a79b55c43328b591125ca90a9a9 : SUCCESS in 3m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8d06567ef4e5479fb95a26b5bd0d3dd9 : SUCCESS in 5m 21s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/cf3a7ff95bff4defa936a531cde73360 : FAILURE in 5m 13s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f0ee56848edf46bc9a7091f55efda630 : SUCCESS in 1h 56m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f89bde6fb24439fbcbbc7378e6926db : SUCCESS in 5m 38s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/850afd687dbf41c190eb9cae7905d0d1 : SUCCESS in 1h 23m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4ed584cf9bab454ba9d91fac444a8bce : SUCCESS in 2h 09m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8fce74a0fb954e1fa4793d316620452c : SUCCESS in 34m 06s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/f9299f48dfc846ee955b1b5116189ae9 : SUCCESS in 23m 12s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/8da0a17a410141c18e9b1f344566da20 : SUCCESS in 8m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/005765e37c84425bb620cf96812642ad : SUCCESS in 1h 35m 19s","accounts_in_message":[],"_revision_number":2},{"id":"c43901e53bc69ae7f4ac56a3dea80a6e64413f64","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-07-07 09:49:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5a0c5e96569f0c7c8cc9db8ce202ee7de1adb292","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-09 14:23:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1dd8833606a203ffb76b40e5e89223a72dc9bfad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 14:23:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b882114b5fb73a4d340ceca3f16260a2a5a3296e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 16:40:33.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e88322f584f24a65909804377be12b51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ee3da80497d4a0b8b9cce45f84f04ab : SUCCESS in 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ce99f933942449894c3ca15b96eb404 : SUCCESS in 6m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be5c006377194dc6837bd58be48e5366 : SUCCESS in 3m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1492ad52994a43c88bc509f04d825230 : SUCCESS in 4m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f869226174cf4a8490e70ad74687006b : SUCCESS in 7m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/74fbcdffd24d47499dc3045508287e2f : SUCCESS in 1h 28m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0294d844a8440d7b4131580fa1bab54 : SUCCESS in 4m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/00647cc5b13743f198c18f8f0414374a : SUCCESS in 1h 23m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/caff89e9fd2f4591aebf3f23ff69fa01 : SUCCESS in 2h 16m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2ae1d565f3464cea9576d5299206d214 : FAILURE in 39m 49s","accounts_in_message":[],"_revision_number":2},{"id":"5d66610ad128afcf2207455560e6ba44777830cf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-09 17:53:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0d70880552251726040765d9f639ad63a2e0da59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 20:49:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39bd904b75134595a3c0c76e1f146080\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c4021f506864ec2bd91f5c347704b4a : SUCCESS in 9m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de4e962defb84b4ea9a67516ff9369d3 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7913388ea93d4f6b9afecbfe584d2532 : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17c17ba2ddb643fa8900607238b34944 : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9ac3c2d0ca704e2eb67c501dfb8cb6a0 : SUCCESS in 3m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ea2c365bae8440bd9688391ed41ccdf4 : SUCCESS in 5m 39s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/c843b8b1c4fa43e9a93e7598d759f29f : FAILURE in 5m 04s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/047a4fd704f744e8a8d890259a86cdcb : SUCCESS in 1h 45m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a328a4cd9a884b539d5cceddc6801535 : SUCCESS in 6m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/07126f3ba8954c7b94836d46894a86e9 : SUCCESS in 1h 19m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0c92fed944ee43fe92eb7b3b8d55d371 : SUCCESS in 1h 58m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d4512009445a4ebda225ab91b15b7cfd : SUCCESS in 31m 40s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/0ab0aba242234cb482acdd52938ae393 : SUCCESS in 24m 00s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/8120bf128e544a458d103a81eaeb77f4 : SUCCESS in 5m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4f468c1578ac4cbd8d8330cdeccd7e1d : SUCCESS in 1h 33m 06s","accounts_in_message":[],"_revision_number":2},{"id":"e3776b6835b99fdca2f96fbcea88590fd7ccb554","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 20:51:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f0087d60aa8a39f522481a0df32045665491b852","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 23:26:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20ed26ca82e4451cb070d0b0a570884d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8de7dcad96414aa8b253e10963b438a9 : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7603acb6f504a928f4a6ee8f6efca78 : SUCCESS in 6m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/88b5088c43b842df84fcc129d9db0e44 : SUCCESS in 3m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68855f66e9ad41b686a27c3417427068 : SUCCESS in 3m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ece2580e29c3404389e0c27fc2362851 : SUCCESS in 4m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e62212bbbd394ebdb42734d3179f0553 : SUCCESS in 1h 40m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d602040d244b494bb247bc8cddd9e949 : SUCCESS in 5m 33s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7a3b3f82d7f246d3a012ddabf9b8d605 : SUCCESS in 1h 55m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9eafb2f0bddd4896b60ab6b242789d41 : SUCCESS in 2h 33m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3401e98b3956449583e33a2af4b0a5ff : SUCCESS in 1h 28m 19s","accounts_in_message":[],"_revision_number":2},{"id":"97a68ae820193d839d6b9b9caadc514e6896f901","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 23:26:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d25265d7fd2b164dfbc426018e87f4bf2e12674d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 23:27:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1bf9d5ae40e453fbb7ffd6c45188d4a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03e4ed2389c24ec8910132bae69eb2b6 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d81239ecb89d62ffacb18ad0044b6188e8d190ed","revisions":{"747b7dc82e370c4a30ad89e605d383e0777e016b":{"kind":"REWORK","_number":1,"created":"2026-06-29 08:30:53.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/65/995265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/65/995265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/995265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/995265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/995265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/65/995265/1"}}},"commit":{"parents":[{"commit":"9e1cb4a06d57a443afc7b565f6816fc6c0152f70","subject":"Merge \"Fix no-else-return pylint warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9e1cb4a06d57a443afc7b565f6816fc6c0152f70"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-06-29 08:29:42.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-06-29 08:30:49.000000000","tz":120},"subject":"[FWaaS] Set \u0027is_filter\u0027 for the attributes which allows filtering","message":"[FWaaS] Set \u0027is_filter\u0027 for the attributes which allows filtering\n\nCloses-bug: #2157781\n\nChange-Id: I583e2e4937bd18f3f044880f95a2b98415440568\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/747b7dc82e370c4a30ad89e605d383e0777e016b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/747b7dc82e370c4a30ad89e605d383e0777e016b"}]},"branch":"refs/heads/master"},"d81239ecb89d62ffacb18ad0044b6188e8d190ed":{"kind":"REWORK","_number":2,"created":"2026-07-01 08:44:48.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/65/995265/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/65/995265/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/995265/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/995265/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/65/995265/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/65/995265/2"}}},"commit":{"parents":[{"commit":"9e1cb4a06d57a443afc7b565f6816fc6c0152f70","subject":"Merge \"Fix no-else-return pylint warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9e1cb4a06d57a443afc7b565f6816fc6c0152f70"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-06-29 08:29:42.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-01 08:44:00.000000000","tz":120},"subject":"[FWaaS] Set \u0027is_filter\u0027 for the attributes which allows filtering","message":"[FWaaS] Set \u0027is_filter\u0027 for the attributes which allows filtering\n\nCloses-bug: #2157781\n\nChange-Id: I583e2e4937bd18f3f044880f95a2b98415440568\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d81239ecb89d62ffacb18ad0044b6188e8d190ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d81239ecb89d62ffacb18ad0044b6188e8d190ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
