)]}'
{"id":"openstack%2Fneutron-lib~996360","triplet_id":"openstack%2Fneutron-lib~master~Ie5e1770740b21ef6efa9ca2e8d92c89a3f83f37c","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2026-07-09 14:47:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie5e1770740b21ef6efa9ca2e8d92c89a3f83f37c","subject":"adjust required Python to 3.11 and newer","status":"MERGED","created":"2026-07-07 20:32:07.000000000","updated":"2026-07-09 14:49:28.000000000","submitted":"2026-07-09 14:47:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996360","meta_rev_id":"420afa37e994b3dff41426158e3a9dfa552fb560","_number":996360,"virtual_id_number":996360,"owner":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-09 14:47:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2026-07-09 12:44:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-07-07 20:35:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2026-07-09 12:44:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-07 20:35:37.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-07-07 22:22:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-09 12:44:01.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"eb99fb111ca574ed8995a6d29d0467166661b543","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-07-07 20:32:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45e56ef8f920b2a1c8f45d0e75d5af87fff63aef","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-07-07 20:35:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f0761ddaa4a76411b3fe3bb037f4902a3ecf9e91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 22:22:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f437404233a48f79e91385fa82e4781\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/92ee3be25bf8420e8a7adf69177ff558 : SUCCESS in 4m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fa2a27fd3a404de083dfeed4b1c744dd : SUCCESS in 1m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/129de49a014f40a5a6a278aab8ba4095 : SUCCESS in 10m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0dd1100477894d99b923010252c97ab6 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b370badadf094775be47261cc9538f01 : SUCCESS in 7m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/babe2ebc7baf412d9ddb0a905348cbff : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b25d815852b444748997a543f4795c22 : SUCCESS in 8m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/55b45f6725334712ac53a06b8b88e848 : SUCCESS in 6m 37s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/328b5fdf2d454425b44acddcc6088bed : FAILURE in 3m 22s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4aaee3fa99b46d4a4ed8ea61e4876b8 : SUCCESS in 1h 19m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a34157f84f204079ba702307decf9dd1 : SUCCESS in 6m 30s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ec001e7ec8d34155920901e2abac10bd : SUCCESS in 1h 39m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3dbc5c31dda148728da70fa04a42b241 : SUCCESS in 1h 48m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c262fc024c9c4eaf860e5c13d85bf6e6 : SUCCESS in 34m 11s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/ec3e98c319c34a26b57bc08e54faaad1 : SUCCESS in 24m 44s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/daa6ae14cb964d3f8a81e6f4d140c1f9 : SUCCESS in 5m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0c2c97b6d126429883c936f8ad357101 : SUCCESS in 1h 25m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e0398e1fc69459480b51c9e0a3fd786607179206","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-07-09 12:44:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5c6c2c7112b3551bb3f28a21ded3d9b86a29832e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 12:44:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"217524301398cff9eb0a507385d2b88d484ecdf3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 14:47:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b51f3198d86d4ca28d94ae507ac856a8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c0d2d145c516494081077fb005c7829c : SUCCESS in 3m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ff2cacb9b6ce4813a54a9644ba02e9aa : SUCCESS in 1m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f7a0921aabe473ea010dc7332fd8817 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2bdfddcbaaf49ffb83a186b2a187a7d : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b84c43e875e54c118bb0e49545e0f224 : SUCCESS in 2m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ac3cc63b29d476e8be52afbcbed5cc9 : SUCCESS in 2m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ba42642bc9874a0dbe0a650b8c90ad7d : SUCCESS in 3m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2bb0ba916c684686bb1143a8f33c7839 : SUCCESS in 1h 54m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad8fb834f09c44b990e936127a3810bf : SUCCESS in 5m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cf898d485fae48b6a781970674998e6d : SUCCESS in 1h 51m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/35a9c39a14ca48fa932597f0746cec31 : SUCCESS in 2h 02m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0a6921a415be495eaef10d7cc821e811 : SUCCESS in 1h 21m 18s","accounts_in_message":[],"_revision_number":1},{"id":"30ca72b446b667fe1a7227e42405adea7ef35640","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 14:47:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"420afa37e994b3dff41426158e3a9dfa552fb560","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 14:49:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b0d432fc0d74fbea00a8da0dbeac2fe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ab7b6ddeb534ac2ab0820912e711d80 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0206cb6dbddc245005a1606a255f20b56bd0b281","revisions":{"0206cb6dbddc245005a1606a255f20b56bd0b281":{"kind":"REWORK","_number":1,"created":"2026-07-07 20:32:07.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/60/996360/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/60/996360/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/996360/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/996360/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/60/996360/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/60/996360/1"}}},"commit":{"parents":[{"commit":"2df8ca85c993ca20cc66c50882c3d69f5ee55ed2","subject":"Merge \"Add is_filter to PVLAN attributes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2df8ca85c993ca20cc66c50882c3d69f5ee55ed2"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-07-07 20:31:12.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2026-07-07 20:32:00.000000000","tz":-300},"subject":"adjust required Python to 3.11 and newer","message":"adjust required Python to 3.11 and newer\n\nIn change I0dafc59cde7d0cf02e4c126c35cee9c1bcb3cda0 Python 3.10 support\nwas removed so we should increase the requires-python to 3.11 and newer.\n\nChange-Id: Ie5e1770740b21ef6efa9ca2e8d92c89a3f83f37c\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0206cb6dbddc245005a1606a255f20b56bd0b281"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0206cb6dbddc245005a1606a255f20b56bd0b281"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
