)]}'
{"id":"openstack%2Fneutron-lib~996435","triplet_id":"openstack%2Fneutron-lib~master~Ic8ebe7a0de21acefc2d38b6cba5b260f619f10f8","project":"openstack/neutron-lib","branch":"master","topic":"fix/sa-orphans","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-07-31 07:53:02.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2026-07-31 11:11:26.000000000","reason":"Change was submitted"},"39125":{"account":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"},"last_update":"2026-07-31 11:11:26.000000000","reason":"Change was submitted"},"11816":{"account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"last_update":"2026-07-31 11:11:26.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-07-31 11:11:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic8ebe7a0de21acefc2d38b6cba5b260f619f10f8","subject":"Revert \"Set confirm_deleted_rows\u003dFalse to StandardAttribute obj\"","status":"MERGED","created":"2026-07-08 10:08:48.000000000","updated":"2026-07-31 11:12:41.000000000","submitted":"2026-07-31 11:11:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"996435-fix/sa-orphans","meta_rev_id":"f2d3394eccd25be477898d1bc3d715c03e5e515a","_number":996435,"virtual_id_number":996435,"owner":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-31 11:11:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-07-23 12:50:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2026-07-30 14:42:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":-1,"date":"2026-07-23 14:20:20.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2026-07-30 14:42:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-07-23 12:50:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-08 10:17:21.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-07-08 12:23:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-20 15:05:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"CC"},{"updated":"2026-07-20 15:05:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"CC"},{"updated":"2026-07-20 15:05:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-07-20 15:05:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2026-07-20 15:05:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-07-20 15:05:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-07-20 15:13:54.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"CC"},{"updated":"2026-07-21 14:06:53.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2026-07-21 14:24:08.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-07-30 14:42:14.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"78287ca84d348a21a5c616580ab74df3ab1e6876","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"},"date":"2026-07-08 10:08:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d1d7536431f77af57c5f445d627ad588374a8d9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-08 10:17:21.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a43c84542b8550686ed1145224ec92e1e017e65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 12:23:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ebc2fcab0250415d8b7fa2825ee96842\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/522934dab71443f696eb60de46b11cac : SUCCESS in 10m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2d8217d8c8f4ed4b8b1d8bfba03ae0d : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0caf223c1a2f4ee9aed4972de683e15f : SUCCESS in 11m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0f544aec7eab4774b93b8ec4e447f71d : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b613bb38a86d4bd4a1fa410735723a5c : SUCCESS in 3m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e17e87c1ef3c40148c90145ae1fc4fc6 : SUCCESS in 6m 05s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/a5cef33c498d4265b11528e8de7cd4bb : FAILURE in 5m 36s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f50545d198454930bc3a266d7a8fb905 : SUCCESS in 1h 34m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c9169e4712743b0862ea0e81224ebbb : SUCCESS in 5m 42s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d6df52ff754a4be08a810b20c5b07411 : FAILURE in 50m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7448c9a20b2a45729568e4f25b9e71c1 : SUCCESS in 2h 11m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/868392b96f714bfa84cb5f5c5e5e6b90 : FAILURE in 51m 34s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/5d0834e7a92d48a5b348b278b180936c : SUCCESS in 26m 39s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/978c455611444c3da926248584b9d29d : SUCCESS in 3m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d6107522bbc94e60b05b3ac68890b902 : SUCCESS in 48m 16s","accounts_in_message":[],"_revision_number":1},{"id":"e4a992279b3b0645b368ce72b3ad8ce68333b00d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"},"date":"2026-07-10 10:51:52.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"fbfeeeb7965151b40acf50338e9f0f8b23bac6f4","author":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"},"date":"2026-07-10 11:12:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d871a5b0c255719c6a506f08d828d02b1a1185ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 13:46:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/67591cfc068f49298ee6f9462e4b3424\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac5a3e1cefde44449baf95f00a10117c : SUCCESS in 5m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29f97ec724014f509a2117649a899e5e : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b76f5cfa59c646328abef66a8ace0c21 : SUCCESS in 7m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25461c2debe348a4982a436f64040271 : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2d071e34e45b432a8f7c1077bdf8ccb2 : SUCCESS in 3m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a5edd28cf37445b3a894d16ca449ef38 : SUCCESS in 4m 02s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/97cc66c13fa540599774859ae431e429 : FAILURE in 6m 11s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25ab2c2569cc439f8ed79fa5efaffdf8 : SUCCESS in 1h 46m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f97daba1c7b4807af5b420cd8eb2945 : SUCCESS in 5m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2286dd1f631c4ab9a2eeaad4fb15c1a5 : SUCCESS in 1h 34m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/76d1010eeccc4d2a9dad25649e113b6c : SUCCESS in 2h 53m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e83b5aad5c2e42178b12487f5669c981 : SUCCESS in 45m 00s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/7b77f15310cb40f58dd79a23c9be7cc9 : SUCCESS in 19m 05s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/72a9c8cb45414d6da532af7a4c51615b : SUCCESS in 6m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/dec7f50757294f93a4982ece4fa58131 : FAILURE in 53m 20s","accounts_in_message":[],"_revision_number":2},{"id":"962bdd84a3367d6675974f53d5dbad42411d9c5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"},"real_author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-20 15:01:10.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_39125\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"}],"_revision_number":3},{"id":"22e36706654bae1fddf2372f15d6a4ea7f822ba0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-20 15:04:43.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3f25eb052b2101260861830066f9e10a5c9f7cf9","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2026-07-20 15:13:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6ec4957599459bcc6fdf571ec5464b86b07ca22a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 16:48:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/449efee9ff6349ea95ac7c2f167136f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/006c7cd232be4ca09c995d47afafde26 : SUCCESS in 8m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9b63d31ce484be28192f8b1f3af4bdc : FAILURE in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b4e38e778c840469c03e19ab4303102 : SUCCESS in 6m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3669577bb0a46f7888092f4f8e9e663 : FAILURE in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c06683cc2984d3eafca8428ccd39f93 : FAILURE in 3m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/41da8ebb363042debb47f53491bbaf69 : FAILURE in 7m 10s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/8de2921e75864850b977d5a25e9359df : FAILURE in 6m 36s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb124c0ba3794e5b9fc687c764078a44 : SUCCESS in 1h 30m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da73c6e06cbf47b5a5c9fd63d966c0da : SUCCESS in 4m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/63f71ce4a884459c9be284f9882978f4 : SUCCESS in 1h 10m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1b208e7c23bc4d1497ce3784e165a90f : SUCCESS in 1h 45m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7478d11ef6314303ba39223fb3b29b6c : SUCCESS in 30m 21s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/5d8d7b49262f452bb171d8356c96c2a8 : SUCCESS in 26m 05s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/43c98288d3cf42549915aee2b03c9208 : FAILURE in 7m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bbcb4f1ab4a54170bb75cd93532a07e2 : FAILURE in 48m 30s","accounts_in_message":[],"_revision_number":3},{"id":"1545ccca5df5e5c62480a45e1e00aa478ae730eb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-07-20 19:50:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"79ab648fd81c7de2d2b6cf31de4ae2584c817de2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"},"real_author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-21 08:22:02.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_39125\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"}],"_revision_number":4},{"id":"cf68085221cc2652b0f6e7c24a3238a0f223a374","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 10:36:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0e7bbc9b9e944b9a83ce4443cd2af60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/986fc8822b814a2b897d91da0c8e1b09 : SUCCESS in 9m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1ae1220785b4792a92a4d8acef9cc20 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf6b5fb271354db3a61bf0d15b71e66f : SUCCESS in 6m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e27e092f675d4b9097f9cae242ac771f : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0f0d032ef4ba433c9f4426e31d178836 : SUCCESS in 6m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9fc0351a361747ea82e66e59abcb8875 : SUCCESS in 6m 10s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/005c1f8bf0b84a24b2b26399d3f0b6e0 : FAILURE in 5m 44s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba67e37437e1409281059ab098adc19e : SUCCESS in 1h 17m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfa212a54cd44af4a1086f8030b37a1a : SUCCESS in 4m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5ffcc28a676a49d99bc41c515b462561 : SUCCESS in 1h 31m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/45e64a80e99b4cb281d3f0cc99f40424 : SUCCESS in 2h 07m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ae133fbaa31446a1a24e88a93cd49e90 : SUCCESS in 44m 57s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/bb08c73865324e41b96bdba33fb705f3 : SUCCESS in 26m 00s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/24201443796543b29f3000020f222a2e : SUCCESS in 7m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a3d47746a1eb408c927fc4f53590ea60 : SUCCESS in 1h 37m 47s","accounts_in_message":[],"_revision_number":4},{"id":"5095ae87368a558ee3e7c12c22efae3ac592218d","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-07-21 14:06:53.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"81f85067ad88e5158812e2bfb203fd305921a922","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-21 14:08:02.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e2f8b7d0b21a28c3a4b1f63734c33f968e2b4892","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-07-21 14:24:08.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c7416b9a05b01dcd2c64eef617711858b8faed4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 14:25:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6953de9b8cdf3f72c1e595bac3407fef8c97feb6","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2026-07-21 14:29:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"01e944e4598fcecfc153f530f9001bb7cb8a74b9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-21 15:07:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4315be30e481b1a3e410e0b8b271ce96b5e7ac4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 16:15:14.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f20c2e30ecd24cbe981d96108606cff6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbd6749adcec4f4393b3e66c80f82a17 : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d87a34d2e8f4dd19f19098903289511 : SUCCESS in 7m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c6da2232b7b468899a13c792dc1b9ac : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3d147fd13b75443b8d0d65f4673cd52b : SUCCESS in 5m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f801e0fa794a47fdb651d544e7d3148e : SUCCESS in 7m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9cf226f488342bfb5b68486946f78e5 : SUCCESS in 44m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fe94fdd0d30423d82dbed48b46e83f4 : SUCCESS in 5m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/655a2efff64646969bb68245449d5fdb : SUCCESS in 1h 04m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6eece486e18e45de8bfd5ec2c5676fb9 : SUCCESS in 1h 49m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/966167435676453f8b1223827302b43d : FAILURE in 58m 57s","accounts_in_message":[],"_revision_number":4},{"id":"c877bbc8505e9d193b52605ab0648deb75d9d84b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-07-21 17:20:28.000000000","message":"Patch Set 4: -Code-Review -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"212ea10d6658a2365d36420675102a8735da4823","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2026-07-21 18:05:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2bad44acced05d5aefd032c39b9603f0b12dbe5a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-22 15:08:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"88edd94c521f76154cc935b5d7de4732301004b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 17:54:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e15455aea4d4dc6ae03287197d9e262\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8936506cb241452db8f58abf17dcb0ba : SUCCESS in 8m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09fcf80ec6634246b39ae982f17e7c8e : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc7007ba17814ed0825ae30983ea6b31 : SUCCESS in 6m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/165fded96995403198f64135641e1a89 : SUCCESS in 6m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b1a146c0cfd34dceb6558c81257b3c57 : SUCCESS in 4m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/657e130b1fab41bcb8d3f1a0f99363de : SUCCESS in 7m 51s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/95475f949a9246cc990ffc4c8d6d68ff : FAILURE in 5m 38s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/025635c15c3743deaa5af70dc71534a9 : SUCCESS in 1h 12m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fc1451c1f8b44c1bc909e19fb267eff : SUCCESS in 5m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/34c346676aa344caab2490a63902571a : SUCCESS in 1h 02m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/45edd414c9a84693a4b90cf24563f474 : SUCCESS in 2h 39m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/62e9c01cf5e746ebbd062321192778d8 : SUCCESS in 34m 33s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/8503f28a396f48a885f599794777317e : SUCCESS in 25m 31s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/5450cd2cd6574bf59b5c439a51189771 : SUCCESS in 6m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/088bcd58959144e48027e99ab423ccc0 : FAILURE in 1h 02m 39s","accounts_in_message":[],"_revision_number":4},{"id":"1838f7a485916b47636c3e613079d03667e6e644","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-23 08:45:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"45206a7aee03863299649f0b4edb86f433052e8e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-07-23 09:04:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0b02cc9efb4508d950b9c9e174cc12406b190485","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-23 12:50:31.000000000","message":"Patch Set 5: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"2b6fdef279bbd194b836e3caa3844fef53191af4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-23 12:50:37.000000000","message":"Patch Set 5: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":5},{"id":"9637df8220c56637bbd4a1ca83a28085311d0480","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-23 12:51:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2afe29ff741b0d9c84a3079e3207b1bf3839751e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-07-23 14:20:20.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fdf198b4bca12962a81801b9bbfa63ef4a666d3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 14:36:51.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56cca53cf48e4cf7862d162757be40c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0db82ca7b328456f8c4d16093977209a : SUCCESS in 10m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/342e8dec87f749bdb44a69685ef02029 : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f091cced50e641e0b968eb0589b108f9 : SUCCESS in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/98d444c12eb84b969176946a0a9ba6f1 : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/418664f7dbbe43bb951f91f20d707a5d : SUCCESS in 2m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3f6ea404bc334cdb89e4e074cadb95ba : SUCCESS in 3m 52s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/44c74da691674bde87d5f07b6dd05dde : FAILURE in 5m 53s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe8a3b9f3b664ca7b7466302f716e083 : SUCCESS in 50m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c56fe31d9e240d0808e9f594f1c12e4 : SUCCESS in 3m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a764cd09e53547bd86e376c75f6f8ba0 : SUCCESS in 1h 03m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1aac0aee97374647b1e789a3fbd6f8bf : SUCCESS in 1h 45m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/086a80acc9b0430cbf941d93bc39d147 : SUCCESS in 30m 35s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/50d49087cf1740408e95f708b70c823d : SUCCESS in 23m 10s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/997753ef4c8849698fc2a29b9e034d30 : SUCCESS in 5m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8b8dd4a89e1342b8ba77f0046b652f45 : FAILURE in 1h 00m 09s","accounts_in_message":[],"_revision_number":5},{"id":"93123160f502385cc41b49cab70418de2bf5f14e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-24 07:59:23.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0c9828ca067865dbd84e90758874b0bef19b7718","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 09:41:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55065587f1434be4ab2ed499b86dd6b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da184cc49d0443ecadc8bac3f0d393cc : SUCCESS in 4m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4a846265d5343d19d27b2e18a5970a4 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4764fcec28af4c179b73515624f87a3e : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6779e9d6c564416cb35da5f19eef58ce : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/31380eb402d242bc81fc0c9cf97002ca : SUCCESS in 3m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8598604274824efcab98cb7d1a4f1b3d : SUCCESS in 8m 02s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/b2d7b74e9b2342deb2089159a9a6ddea : FAILURE in 5m 59s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b3e95a7aa554fa187c1e9a62f0beaac : SUCCESS in 40m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65e66cfe40c24646b5e9a847e75df963 : SUCCESS in 6m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/57cc568c5b4046799d2d8e7687f6daae : SUCCESS in 1h 15m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5c18351a134448e3b5857d0795fc5323 : SUCCESS in 1h 39m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d75ca89c6405459583f1a94f0a855f50 : SUCCESS in 33m 54s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/d2af056dfd5a4b67af710ecf58aba822 : SUCCESS in 16m 29s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/2e491faaa6c448338606593bd8a784d8 : SUCCESS in 3m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3c05469fbb6c42d2825d97a563afa739 : SUCCESS in 1h 33m 02s","accounts_in_message":[],"_revision_number":5},{"id":"b0dce4800e2394afbf4f7c181961b2a5defa6a90","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-27 08:43:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5de53bdeabc5a669ce2f999e9035db3e29c21cb0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-07-30 14:42:14.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5b58abf552d37376e0244b74411ef8bf5e4398b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 14:42:17.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/775e4a3ea88a4b1ab6d4bace73555afd\n\n\nWarning:\n  Change 998421,4 in project openstack/neutron does not share a change queue with 996435 in project openstack/neutron-lib","accounts_in_message":[],"_revision_number":5},{"id":"19958e6d235327f7883c252b099daf66858dfc64","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-31 07:53:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6d3b2f371b978b6ec2b7395bc49694c61e372c98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 09:23:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f76c578959104141afefffe665418025\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e84d5f55192c4355adca3ed2ac5a84ce : SUCCESS in 8m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9ecd1057cb441618627dcf1247c52ab : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/266ac2108a2d45b297d14f7350aa2ab5 : SUCCESS in 6m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/40b9a97cb1044b4f9ce3147ed062bade : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6599d963cb654204b9a6bb034f72f7bc : SUCCESS in 3m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0a82de72967749f0afd27c561942bcdb : SUCCESS in 5m 55s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/1a37ef9bf7864208a35c7ec157f244a2 : FAILURE in 5m 30s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/33e0055c8e2d41ccb4488cb137313950 : SUCCESS in 50m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdea91de880243538811c611a3e2c507 : SUCCESS in 4m 54s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9232bd62c48f445991bde4bf7d8b6bff : SUCCESS in 58m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f0d5b455ddf041038808687a82d4d597 : SUCCESS in 1h 29m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a3bd01d33398498faed2f2ff1b835623 : SUCCESS in 29m 53s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/cf20815b7563488180ea40c4d2a1078a : SUCCESS in 20m 06s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/aecbfff98950409aba9daa0257dab3af : SUCCESS in 5m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7feac56a2d544911b51ef0cbf8b9a601 : SUCCESS in 54m 36s","accounts_in_message":[],"_revision_number":5},{"id":"a3258dce8a79ee357a9af73ea4d63db3519bf5b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 09:24:10.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"69115676f7809e2664edc03a25c9c73aee7cdd02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 11:11:26.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72370a8f10c84bfa9bbbd45ce998b6f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2547edd5155747728e6921f9673d7f5d : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2cce96509e54df4ac9df58ef08de899 : SUCCESS in 7m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/adf7faa6655f4c9eb4ca6387f24a92d4 : SUCCESS in 3m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e6be02fdb6c4fb3aa2e5c74f9214e10 : SUCCESS in 3m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/875b7657630f4e61b27037299c0b4206 : SUCCESS in 6m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a0a5c99d20de4027b74d9dba8383c293 : SUCCESS in 44m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/242c312f4fa04e3f9dd0de584b817215 : SUCCESS in 5m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5a5c98a1978d45ff829ee2b1d806399a : SUCCESS in 1h 10m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dd6db41743794675ae3b09854996eebe : SUCCESS in 1h 46m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ba723e608eb84567ab0a0878a747afa5 : SUCCESS in 1h 12m 52s","accounts_in_message":[],"_revision_number":5},{"id":"c288e947ffa808b6bb7cd70756b186442c10cef8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 11:11:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"f2d3394eccd25be477898d1bc3d715c03e5e515a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 11:12:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02c206937dad4e839ddf58bd60fcb729\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b11060c8e0cb4b9082a75d5f6b3e81ea : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b45adc9c943b90485f0326549034c9ab5bf9f9cd","revisions":{"bce41042643dae9d9e7dc7d8ef63717e75ec4259":{"kind":"REWORK","_number":1,"created":"2026-07-08 10:08:48.000000000","uploader":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"},"ref":"refs/changes/35/996435/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/35/996435/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/996435/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/996435/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/996435/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/35/996435/1"}}},"commit":{"parents":[{"commit":"2df8ca85c993ca20cc66c50882c3d69f5ee55ed2","subject":"Merge \"Add is_filter to PVLAN attributes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2df8ca85c993ca20cc66c50882c3d69f5ee55ed2"}]}],"author":{"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","date":"2026-07-08 08:34:07.000000000","tz":120},"committer":{"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","date":"2026-07-08 09:20:23.000000000","tz":120},"subject":"Revert \"Set confirm_deleted_rows\u003dFalse to StandardAttribute obj\"","message":"Revert \"Set confirm_deleted_rows\u003dFalse to StandardAttribute obj\"\n\nThis reverts commit 0e2f7cb4d516fb5660ab017be4ff6c7fe254288e.\n\nSetting confirm_deleted_rows to False suppresses the StaleDataError that\nwould be raised if we fail to delete a standardattributes row. In that\ncase the actual resource i.e. the corresponding port will be deleted but\nthe standardattributes row is left orphaned in the database and needs to\nbe cleaned up manually. This also affects the OVN maintenance task which\nwill not identify the resource as requiring a delete action via a\nremaining ovn_revision_numbers row as long as the standardattributes row\nexists.\n\nDeleting the standardattributes row fails when there is a concurrent\nupdate and delete such that the update bumps the revision number after\nthe delete has read the standardattributes row. The delete of the\nstandardattribute row will explicitly match against the expected\nrevision number in the delete query, which no longer matches after the\nupdate has bumped it.\n\nCo-Authored-By: Felix Huettner \u003cfelix.huettner@digits.schwarz\u003e\nSigned-off-by: Felix Moebius \u003cfelix.moebius@digits.schwarz\u003e\nChange-Id: Ic8ebe7a0de21acefc2d38b6cba5b260f619f10f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bce41042643dae9d9e7dc7d8ef63717e75ec4259"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bce41042643dae9d9e7dc7d8ef63717e75ec4259"}]},"branch":"refs/heads/master"},"c16e36980647afc2a4bdaaf15a6f889a6abd825f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-07-10 10:51:52.000000000","uploader":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"},"ref":"refs/changes/35/996435/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/35/996435/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/996435/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/996435/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/996435/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/35/996435/2"}}},"commit":{"parents":[{"commit":"2df8ca85c993ca20cc66c50882c3d69f5ee55ed2","subject":"Merge \"Add is_filter to PVLAN attributes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2df8ca85c993ca20cc66c50882c3d69f5ee55ed2"}]}],"author":{"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","date":"2026-07-08 08:34:07.000000000","tz":120},"committer":{"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","date":"2026-07-10 10:48:50.000000000","tz":120},"subject":"Revert \"Set confirm_deleted_rows\u003dFalse to StandardAttribute obj\"","message":"Revert \"Set confirm_deleted_rows\u003dFalse to StandardAttribute obj\"\n\nThis reverts commit 0e2f7cb4d516fb5660ab017be4ff6c7fe254288e.\n\nSetting confirm_deleted_rows to False suppresses the StaleDataError that\nwould be raised if we fail to delete a standardattributes row. In that\ncase the actual resource i.e. the corresponding port will be deleted but\nthe standardattributes row is left orphaned in the database and needs to\nbe cleaned up manually. This also affects the OVN maintenance task which\nwill not identify the resource as requiring a delete action via a\nremaining ovn_revision_numbers row as long as the standardattributes row\nexists.\n\nDeleting the standardattributes row fails when there is a concurrent\nupdate and delete such that the update bumps the revision number after\nthe delete has read the standardattributes row. The delete of the\nstandardattribute row will explicitly match against the expected\nrevision number in the delete query, which no longer matches after the\nupdate has bumped it.\n\nCloses-bug: #2160292\nCo-Authored-By: Felix Huettner \u003cfelix.huettner@digits.schwarz\u003e\nSigned-off-by: Felix Moebius \u003cfelix.moebius@digits.schwarz\u003e\nChange-Id: Ic8ebe7a0de21acefc2d38b6cba5b260f619f10f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c16e36980647afc2a4bdaaf15a6f889a6abd825f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c16e36980647afc2a4bdaaf15a6f889a6abd825f"}]},"branch":"refs/heads/master"},"bb8fae905e47239a265f83217029fb9ec187cb9f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-07-20 15:01:10.000000000","uploader":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"},"real_uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/35/996435/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/35/996435/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/996435/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/996435/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/996435/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/35/996435/3"}}},"commit":{"parents":[{"commit":"7a0b2fe4fb7076eaa61f266eff73669937f6f01d","subject":"Merge \"Remove unused VPN exception classes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7a0b2fe4fb7076eaa61f266eff73669937f6f01d"}]}],"author":{"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","date":"2026-07-08 08:34:07.000000000","tz":120},"committer":{"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","date":"2026-07-20 15:01:10.000000000","tz":0},"subject":"Revert \"Set confirm_deleted_rows\u003dFalse to StandardAttribute obj\"","message":"Revert \"Set confirm_deleted_rows\u003dFalse to StandardAttribute obj\"\n\nThis reverts commit 0e2f7cb4d516fb5660ab017be4ff6c7fe254288e.\n\nSetting confirm_deleted_rows to False suppresses the StaleDataError that\nwould be raised if we fail to delete a standardattributes row. In that\ncase the actual resource i.e. the corresponding port will be deleted but\nthe standardattributes row is left orphaned in the database and needs to\nbe cleaned up manually. This also affects the OVN maintenance task which\nwill not identify the resource as requiring a delete action via a\nremaining ovn_revision_numbers row as long as the standardattributes row\nexists.\n\nDeleting the standardattributes row fails when there is a concurrent\nupdate and delete such that the update bumps the revision number after\nthe delete has read the standardattributes row. The delete of the\nstandardattribute row will explicitly match against the expected\nrevision number in the delete query, which no longer matches after the\nupdate has bumped it.\n\nCloses-bug: #2160292\nCo-Authored-By: Felix Huettner \u003cfelix.huettner@digits.schwarz\u003e\nSigned-off-by: Felix Moebius \u003cfelix.moebius@digits.schwarz\u003e\nChange-Id: Ic8ebe7a0de21acefc2d38b6cba5b260f619f10f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bb8fae905e47239a265f83217029fb9ec187cb9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bb8fae905e47239a265f83217029fb9ec187cb9f"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"2df8ca85c993ca20cc66c50882c3d69f5ee55ed2","ours":"c16e36980647afc2a4bdaaf15a6f889a6abd825f","theirs":"7a0b2fe4fb7076eaa61f266eff73669937f6f01d","merge_strategy":"recursive","contains_conflicts":false}},"01c0a7c81e0351c3308c5e32ebfe0da3743478b2":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-07-21 08:22:02.000000000","uploader":{"_account_id":39125,"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","username":"felixmoebius"},"real_uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/35/996435/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/35/996435/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/996435/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/996435/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/996435/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/35/996435/4"}}},"commit":{"parents":[{"commit":"1394940afec5850ee1a60d2e5557842ff12b84b4","subject":"Remove disabling the scope enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1394940afec5850ee1a60d2e5557842ff12b84b4"}]}],"author":{"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","date":"2026-07-08 08:34:07.000000000","tz":120},"committer":{"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","date":"2026-07-21 08:22:02.000000000","tz":0},"subject":"Revert \"Set confirm_deleted_rows\u003dFalse to StandardAttribute obj\"","message":"Revert \"Set confirm_deleted_rows\u003dFalse to StandardAttribute obj\"\n\nThis reverts commit 0e2f7cb4d516fb5660ab017be4ff6c7fe254288e.\n\nSetting confirm_deleted_rows to False suppresses the StaleDataError that\nwould be raised if we fail to delete a standardattributes row. In that\ncase the actual resource i.e. the corresponding port will be deleted but\nthe standardattributes row is left orphaned in the database and needs to\nbe cleaned up manually. This also affects the OVN maintenance task which\nwill not identify the resource as requiring a delete action via a\nremaining ovn_revision_numbers row as long as the standardattributes row\nexists.\n\nDeleting the standardattributes row fails when there is a concurrent\nupdate and delete such that the update bumps the revision number after\nthe delete has read the standardattributes row. The delete of the\nstandardattribute row will explicitly match against the expected\nrevision number in the delete query, which no longer matches after the\nupdate has bumped it.\n\nCloses-bug: #2160292\nCo-Authored-By: Felix Huettner \u003cfelix.huettner@digits.schwarz\u003e\nSigned-off-by: Felix Moebius \u003cfelix.moebius@digits.schwarz\u003e\nChange-Id: Ic8ebe7a0de21acefc2d38b6cba5b260f619f10f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/01c0a7c81e0351c3308c5e32ebfe0da3743478b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/01c0a7c81e0351c3308c5e32ebfe0da3743478b2"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"7a0b2fe4fb7076eaa61f266eff73669937f6f01d","ours":"bb8fae905e47239a265f83217029fb9ec187cb9f","theirs":"1394940afec5850ee1a60d2e5557842ff12b84b4","merge_strategy":"recursive","contains_conflicts":false}},"b45adc9c943b90485f0326549034c9ab5bf9f9cd":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-07-23 12:50:31.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/35/996435/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/35/996435/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/996435/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/996435/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/35/996435/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/35/996435/5"}}},"commit":{"parents":[{"commit":"1394940afec5850ee1a60d2e5557842ff12b84b4","subject":"Remove disabling the scope enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1394940afec5850ee1a60d2e5557842ff12b84b4"}]}],"author":{"name":"Felix Moebius","email":"felix.moebius@digits.schwarz","date":"2026-07-08 08:34:07.000000000","tz":120},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-07-23 12:50:31.000000000","tz":0},"subject":"Revert \"Set confirm_deleted_rows\u003dFalse to StandardAttribute obj\"","message":"Revert \"Set confirm_deleted_rows\u003dFalse to StandardAttribute obj\"\n\nThis reverts commit 0e2f7cb4d516fb5660ab017be4ff6c7fe254288e.\n\nSetting confirm_deleted_rows to False suppresses the StaleDataError that\nwould be raised if we fail to delete a standardattributes row. In that\ncase the actual resource i.e. the corresponding port will be deleted but\nthe standardattributes row is left orphaned in the database and needs to\nbe cleaned up manually. This also affects the OVN maintenance task which\nwill not identify the resource as requiring a delete action via a\nremaining ovn_revision_numbers row as long as the standardattributes row\nexists.\n\nDeleting the standardattributes row fails when there is a concurrent\nupdate and delete such that the update bumps the revision number after\nthe delete has read the standardattributes row. The delete of the\nstandardattribute row will explicitly match against the expected\nrevision number in the delete query, which no longer matches after the\nupdate has bumped it.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/998421\n\nCloses-bug: #2160292\nCo-Authored-By: Felix Huettner \u003cfelix.huettner@digits.schwarz\u003e\nSigned-off-by: Felix Moebius \u003cfelix.moebius@digits.schwarz\u003e\nChange-Id: Ic8ebe7a0de21acefc2d38b6cba5b260f619f10f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b45adc9c943b90485f0326549034c9ab5bf9f9cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b45adc9c943b90485f0326549034c9ab5bf9f9cd"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
