)]}'
{"id":"openstack%2Fneutron-lib~998059","triplet_id":"openstack%2Fneutron-lib~master~Ia245af80e6ce40d8a1972ea6efe6585734e81a42","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-07-21 09:55:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia245af80e6ce40d8a1972ea6efe6585734e81a42","subject":"Remove disabling the scope enforcement","status":"MERGED","created":"2026-07-20 19:49:29.000000000","updated":"2026-07-21 09:56:10.000000000","submitted":"2026-07-21 09:55:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998059","meta_rev_id":"d5111c927c6e2fa2c5c18eca9a3b38d26394cce7","_number":998059,"virtual_id_number":998059,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-21 09:55:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-07-21 08:21:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-07-21 08:21:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-07-20 20:57:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-20 22:30:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-21 08:21:16.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"bc04334f475dc484ab41af9295c8bc8d133d01fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-07-20 19:49:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90b2341c0ffe9181121b9ce060dfec0449e583e4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-07-20 19:50:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"762149da62f72393667f71e9aed5169dc7afffcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-07-20 20:19:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"315aa517693932d7b813c88e1a3045c3eaefac1f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-07-20 20:57:16.000000000","message":"Patch Set 2: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1366b22661caed4cc7be9c72494ff15c22f2cba6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-07-20 20:57:32.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"fd5a356a543bae5adc5c76df1e9be52b7ca7ded4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 22:30:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7eb4b0574144fe7a39bedfce798dc17\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/871153daba2a4d73ad950b542400fc35 : SUCCESS in 2m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3efd678b46d491eb76342a045e62f3f : SUCCESS in 8m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2d1db01b53a49889819f2b87cec639a : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6265b72eca984236802951402e2d95ec : SUCCESS in 7m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a9a55ca9a39147b99bd0ae1fc3ba48fe : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/acecbd9e31f747fcae007b0c7e40f927 : SUCCESS in 5m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f0a8f20e88bc480b90f23117aab06a7d : SUCCESS in 7m 28s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/8e2b1a24e85b45dba7e9aa9e0a469a45 : FAILURE in 5m 32s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f26df9536eea468685464a0dae746b38 : SUCCESS in 1h 27m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6b9a4f93a224f2898428cda2cf9a8f2 : SUCCESS in 4m 48s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/f697f307e6b24ef6a01a7a2665d9cebf : SUCCESS in 22m 26s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/1c03a348807744f7a87d5f34189f4796 : SUCCESS in 6m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e4f284a13d514b5c9c779d31ab3102ba : SUCCESS in 2h 04m 30s","accounts_in_message":[],"_revision_number":2},{"id":"cfd25868b3f79d6aa6952f2934e963bbfcdce4f4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-21 08:21:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4fd0666f6ef95e0a8d3f8a4fdac3324b28d77c67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 08:22:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"99991edc158423326c29e4c222178c9201f1b1f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 09:55:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1614bcc3d2fc4cddb1c02738d0cb5137\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cc2ea390925d45aa9e21ab0db5d328ab : SUCCESS in 2m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/761cfef5bed14635a80dedcec6dddbf2 : SUCCESS in 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d1899f5c9944838851660ae0868637a : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0518b6718cf745809831d178c875e0d7 : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/87be8818d4fe402ab6e5c66f76950f56 : SUCCESS in 3m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f758c6dc247042c49b28dc0c7d50d8ea : SUCCESS in 5m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc1c3cbc94f74adea699df984cf6fa30 : SUCCESS in 1h 31m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55b948df70ca44e8b9f168bf3c2bfeae : SUCCESS in 4m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/676a9b2cfbef43c29ab2d7a8e350d51b : SUCCESS in 53m 02s","accounts_in_message":[],"_revision_number":2},{"id":"2542a9a6076f43271b01bcb90d5f90dcd0b4f48c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 09:55:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d5111c927c6e2fa2c5c18eca9a3b38d26394cce7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 09:56:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fd97bd90ed8451fae5c06e318bc7c74\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9537f3243c7e464fbbc5a844ad943de9 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b48f4038a7b4749a06dcdbec3a0eb29 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1394940afec5850ee1a60d2e5557842ff12b84b4","revisions":{"4d9a87e2ed115295736e87a52d91c7b6fa06e6c7":{"kind":"REWORK","_number":1,"created":"2026-07-20 19:49:29.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/59/998059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/59/998059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/998059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/998059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/998059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/59/998059/1"}}},"commit":{"parents":[{"commit":"7a0b2fe4fb7076eaa61f266eff73669937f6f01d","subject":"Merge \"Remove unused VPN exception classes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7a0b2fe4fb7076eaa61f266eff73669937f6f01d"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-07-20 19:44:16.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-07-20 19:49:05.000000000","tz":-240},"subject":"Remove disabling the scope enforcement","message":"Remove disabling the scope enforcement\n\nRBAC scope enforcement was controlled by the oslo.policy\ndeprecated config option enforce_scope which has been removed\nin oslo.policy 6.0.0. That removes the ability to disable\nthe scope enforcement which is the end goal of SRBAC.\n\n- https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html#id5\n\nAdding upgrade impact in releasenotes with olso.policy version\nbump in requirements.txt.\n\nRemove it from unit test that was setting it, otherwise\nit is unused in Neutron as we are already doing policy\nscope enforcement.\n\nChange-Id: Ia245af80e6ce40d8a1972ea6efe6585734e81a42\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4d9a87e2ed115295736e87a52d91c7b6fa06e6c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4d9a87e2ed115295736e87a52d91c7b6fa06e6c7"}]},"branch":"refs/heads/master"},"1394940afec5850ee1a60d2e5557842ff12b84b4":{"kind":"REWORK","_number":2,"created":"2026-07-20 20:19:14.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/59/998059/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/59/998059/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/998059/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/998059/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/998059/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/59/998059/2"}}},"commit":{"parents":[{"commit":"7a0b2fe4fb7076eaa61f266eff73669937f6f01d","subject":"Merge \"Remove unused VPN exception classes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7a0b2fe4fb7076eaa61f266eff73669937f6f01d"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-07-20 19:44:16.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-07-20 20:19:01.000000000","tz":-240},"subject":"Remove disabling the scope enforcement","message":"Remove disabling the scope enforcement\n\nRBAC scope enforcement was controlled by the oslo.policy\ndeprecated config option enforce_scope which has been removed\nin oslo.policy 6.0.0. That removes the ability to disable\nthe scope enforcement which is the end goal of SRBAC.\n\n- https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html#id5\n\nAdding upgrade impact in releasenotes with olso.policy version\nbump in requirements.txt.\n\nRemove it from unit test that was setting it, otherwise\nit is unused in Neutron as we are already doing policy\nscope enforcement.\n\nChange-Id: Ia245af80e6ce40d8a1972ea6efe6585734e81a42\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1394940afec5850ee1a60d2e5557842ff12b84b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1394940afec5850ee1a60d2e5557842ff12b84b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
