)]}'
{"id":"openstack%2Fneutron-lib~999401","triplet_id":"openstack%2Fneutron-lib~stable%2F2026.1~I583e2e4937bd18f3f044880f95a2b98415440568","project":"openstack/neutron-lib","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-08-26 17:10:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I583e2e4937bd18f3f044880f95a2b98415440568","subject":"[FWaaS] Set \u0027is_filter\u0027 for the attributes which allows filtering","status":"MERGED","created":"2026-07-31 10:41:22.000000000","updated":"2026-08-26 17:11:45.000000000","submitted":"2026-08-26 17:10:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999401","meta_rev_id":"fdafe3733c65a3ccef14e3ae8128cf977e709eb7","_number":999401,"virtual_id_number":999401,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-26 17:10:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-07-31 14:37:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-08-03 07:51:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-08-03 07:51:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-31 10:50:40.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-07-31 12:41:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-03 07:51:27.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"254256502203c9ff0c7cce2a7ea58e3166020fbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-07-31 10:41:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e044efd23ddc2cdd8f9010192204bf21d332669e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-31 10:50:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1d2cc73f3312a24b90bb4b01b926f14270718211","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 12:41:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/49910b7711fe4fbbab3e7dddfc77a172\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21434c95b03d46dc82146852b63c29b1 : SUCCESS in 8m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d7c5f0c7ab14cf1af53239f77d395cb : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ecab3c9f6314b67a1f04dedd430ad8a : FAILURE in 7m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f5f4c02ad7d640dba2cca53b093949c3 : SUCCESS in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6b1f695c49f84582bf3ba07059f84582 : SUCCESS in 4m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/22b7522c895749f8ac9d3e14d65c66e4 : SUCCESS in 5m 54s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/73c1e4bf66704aae949f9dca40651699 : SUCCESS in 1h 45m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b38f5b013c66459597f9809ea2f6e462 : SUCCESS in 4m 37s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/ea6b5bcfaaf943f29ab2897006c71687 : SUCCESS in 1h 25m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/3415f892791c496db08caca817c03ac0 : SUCCESS in 1h 35m 24s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/efac23da7c3042278125485249f74a2d : SUCCESS in 1h 58m 25s\n- neutron-tempest-plugin-designate-scenario-2026-1 https://zuul.opendev.org/t/openstack/build/1ae7ff7de9c5404b8a30131a45ba49b8 : SUCCESS in 33m 36s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/7360516c94b143bb8b8321add1dc8be1 : SUCCESS in 23m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c58d45e79c5247899f4971df07422824 : SUCCESS in 1h 10m 45s","accounts_in_message":[],"_revision_number":1},{"id":"91ea4a865ce3defcace059caa168878ba7506e44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-07-31 14:37:17.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"10f0162d3bf7a2179a60eb2011339179f035f0f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 16:33:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/add669f97ac242848290f3543e7b16c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ab66be7a6c64ef7ae993f715ca381b8 : SUCCESS in 8m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c36a04e87584c36999d6aefcebdedc8 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdc28e5ae7a34218b00036973c7821f2 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2177d50ef9f4db4a788c7cea81fc5c5 : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eec692e0ee304451990977616d842528 : SUCCESS in 4m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e3321c2fe48c422da00dd0c1d9b879b3 : SUCCESS in 6m 02s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4097afe18dc42af96231c7e5fdc3732 : SUCCESS in 1h 50m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/294c7f5100cd4675a46f90b32429312d : SUCCESS in 4m 15s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/9134fe7b9b3d47b294953ca98335da8a : SUCCESS in 1h 40m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/6284de8764c24b2bbf308a5907818368 : SUCCESS in 1h 34m 19s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/a4222f363bae4ae58851cbc001be487b : SUCCESS in 1h 39m 22s\n- neutron-tempest-plugin-designate-scenario-2026-1 https://zuul.opendev.org/t/openstack/build/b2ce15283747451fbc21f9c23ff97f93 : SUCCESS in 33m 45s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/d3562cd5de6548dc818efc3bc13caa03 : SUCCESS in 15m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/27a1f8079e7c442eb3acea6d95919f3e : SUCCESS in 43m 14s","accounts_in_message":[],"_revision_number":2},{"id":"22b3b5d2ef38cfdee780523697340ee6572dca9c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-08-03 07:51:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bc6e537c0df3b96ea062f2eb2aa321743541a52e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 14:42:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"669e1614e50d7356668d87bad2829f2df7790e73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 17:10:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22426e53c9244a72bab984173e5137bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63c61532634343c4bedf9aa413c7b1a2 : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a106aaa72d0a47f3987e1acf2cbf4e1c : SUCCESS in 7m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/246e32ae2bea488bac6f1b487fb00e3a : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/12e432949f7248b198960ef4abdc6e22 : SUCCESS in 3m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e816a132ac094313b83d54507c8e7e87 : SUCCESS in 1h 48m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38d998a2fe2b4aa39785322989bb357f : SUCCESS in 5m 46s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/1bb1683510fb46d0bda7c551d2bc309e : SUCCESS in 2h 26m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/187835a1056a450098033e66db216819 : SUCCESS in 1h 22m 06s","accounts_in_message":[],"_revision_number":2},{"id":"de522f034fe774cd45d712952722a4e728ebc127","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 17:10:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"fdafe3733c65a3ccef14e3ae8128cf977e709eb7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 17:11:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/722d14b46e2a4ade84aefb51000c06ab\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72217ae5929f45e78c4acbd2f634312e : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"67fcdde5cda6e8d93abd5333d202fd90a5177c71","revisions":{"199de3686be7bc55d7d96b2ec426a0fb098339d5":{"kind":"REWORK","_number":1,"created":"2026-07-31 10:41:22.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/01/999401/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/01/999401/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/01/999401/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/01/999401/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/01/999401/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/01/999401/1"}}},"commit":{"parents":[{"commit":"7e11b17b3a1b8d9ccfca28ea4f39ed51ab39cfc5","subject":"Fix address‑groups API path and request body format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7e11b17b3a1b8d9ccfca28ea4f39ed51ab39cfc5"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-06-29 08:29:42.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-31 10:39:28.000000000","tz":120},"subject":"[FWaaS] Set \u0027is_filter\u0027 for the attributes which allows filtering","message":"[FWaaS] Set \u0027is_filter\u0027 for the attributes which allows filtering\n\nCloses-bug: #2157781\n\nConflicts:\n    neutron_lib/api/definitions/firewall_v2.py\n\nChange-Id: I583e2e4937bd18f3f044880f95a2b98415440568\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n(cherry picked from commit d81239ecb89d62ffacb18ad0044b6188e8d190ed)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/199de3686be7bc55d7d96b2ec426a0fb098339d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/199de3686be7bc55d7d96b2ec426a0fb098339d5"}]},"branch":"refs/heads/stable/2026.1"},"67fcdde5cda6e8d93abd5333d202fd90a5177c71":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-07-31 14:37:17.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/01/999401/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/01/999401/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/01/999401/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/01/999401/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/01/999401/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/01/999401/2"}}},"commit":{"parents":[{"commit":"521839c7e5a01da704d2c1dc8a4eebc4d9dda4bb","subject":"Set neutron-lib as dependency for hacking pre-commit hook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/521839c7e5a01da704d2c1dc8a4eebc4d9dda4bb"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-06-29 08:29:42.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-07-31 14:37:17.000000000","tz":0},"subject":"[FWaaS] Set \u0027is_filter\u0027 for the attributes which allows filtering","message":"[FWaaS] Set \u0027is_filter\u0027 for the attributes which allows filtering\n\nCloses-bug: #2157781\n\nConflicts:\n    neutron_lib/api/definitions/firewall_v2.py\n\nChange-Id: I583e2e4937bd18f3f044880f95a2b98415440568\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n(cherry picked from commit d81239ecb89d62ffacb18ad0044b6188e8d190ed)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/67fcdde5cda6e8d93abd5333d202fd90a5177c71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/67fcdde5cda6e8d93abd5333d202fd90a5177c71"}]},"branch":"refs/heads/stable/2026.1","description":"Rebase","conflicts":{"base":"7e11b17b3a1b8d9ccfca28ea4f39ed51ab39cfc5","ours":"199de3686be7bc55d7d96b2ec426a0fb098339d5","theirs":"521839c7e5a01da704d2c1dc8a4eebc4d9dda4bb","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
