)]}'
{"id":"openstack%2Fneutron-lib~999637","triplet_id":"openstack%2Fneutron-lib~master~Ic84d6b6725c79a972a0c66a2ac8becbf04b47447","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-10 00:52:18.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic84d6b6725c79a972a0c66a2ac8becbf04b47447","subject":"DNM: Demonstrate local hacking checks not running","status":"ABANDONED","created":"2026-08-03 17:21:18.000000000","updated":"2026-08-10 00:52:18.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"985c7ebb190367fd1b5c637633fcb359e2837910","_number":999637,"virtual_id_number":999637,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-03 19:06:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-03 19:06:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9b303191738fc9a9b7498470862995aa5b036895","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-03 17:21:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"086b6d9e609909f7404161e8f12e8033e3d238c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 19:06:02.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/adcb25f288a8432bb2ca77fc7f902968\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4413369782b742eea0086edee72c41f0 : SUCCESS in 7m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48be516133304ebcab10a24bacd0137e : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08c5178a09c64d5bba4c139a97ae68aa : FAILURE in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b25aed20e074c0ab1a802ddf2db3ed3 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb0f341b44204168a902594035dd21d0 : SUCCESS in 3m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0e92f3931db2407582e9820cafd71475 : SUCCESS in 6m 30s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/41bca09d75f140eaaa7858fabba6bd28 : FAILURE in 5m 18s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7f54672c605416cb3fb68610262ac0c : SUCCESS in 52m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7c80c2192d54a02a0007383e8fca954 : SUCCESS in 4m 35s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/eb9aa93a3220463f8797ffd9cab8bfdc : SUCCESS in 56m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bcfd494c55984a588cbe4ae5491bb392 : SUCCESS in 1h 43m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/160eae2b781c4e21b0bc660c26765bab : SUCCESS in 30m 36s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/8bf287690e09426ab1256a95f3a2c6a4 : SUCCESS in 16m 48s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/7b1a4661692b4dc4b5c8194acdf3f073 : SUCCESS in 2m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/35155ec2bc9c4675a489aa83a4cf8dc1 : SUCCESS in 56m 09s","accounts_in_message":[],"_revision_number":1},{"id":"985c7ebb190367fd1b5c637633fcb359e2837910","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-10 00:52:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4cbb431091dab5b0a956d0ec2c434506d0d4ccc1","revisions":{"4cbb431091dab5b0a956d0ec2c434506d0d4ccc1":{"kind":"REWORK","_number":1,"created":"2026-08-03 17:21:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/37/999637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/37/999637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/37/999637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/37/999637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/37/999637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/37/999637/1"}}},"commit":{"parents":[{"commit":"92b5fb83df3d481d27bd2244d34cd3839c6ba625","subject":"Set neutron-lib as dependency for hacking pre-commit hook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/92b5fb83df3d481d27bd2244d34cd3839c6ba625"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-03 17:20:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-03 17:20:28.000000000","tz":540},"subject":"DNM: Demonstrate local hacking checks not running","message":"DNM: Demonstrate local hacking checks not running\n\nChange-Id: Ic84d6b6725c79a972a0c66a2ac8becbf04b47447\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4cbb431091dab5b0a956d0ec2c434506d0d4ccc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4cbb431091dab5b0a956d0ec2c434506d0d4ccc1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
