)]}'
{"id":"openstack%2Fneutron-lib~999731","triplet_id":"openstack%2Fneutron-lib~master~I803527a1edcbc976d7e95e6a5be248d91bc9a40b","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-08-10 15:23:20.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-08-07 14:52:45.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I803527a1edcbc976d7e95e6a5be248d91bc9a40b","subject":"Add optional plugin_conf to the BaseOvnDbSynchronizer class","status":"MERGED","created":"2026-08-04 14:34:18.000000000","updated":"2026-08-10 15:24:20.000000000","submitted":"2026-08-10 15:23:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"999731","meta_rev_id":"cb21b432fa075ad2e1bd378895fd411270e95425","_number":999731,"virtual_id_number":999731,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-10 15:23:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-08-05 09:54:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-08-07 14:52:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-08-10 13:35:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-08-05 09:54:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-04 16:22:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-05 09:54:07.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-08-05 13:18:16.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"f67919b25314d119492c229763b03f6a44322877","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-08-04 14:34:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0bd9fffa7dc6f5623e088e01876087543e3d1924","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 16:22:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22a5f7c6dd0c441788f4ca5bd918df12\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb529c4361904dec84fa0a1994980397 : SUCCESS in 7m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf1b95f4753a404aa5c4837858523c69 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b95bed4c34d408fb1953465d8d32348 : SUCCESS in 8m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a235e946dff1428aa0637be7ab1a64c8 : SUCCESS in 2m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/617d870eebbc4d00b995a0b2ad545ed5 : SUCCESS in 3m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/13188ffa354f412ab5390a4268d9ac08 : SUCCESS in 8m 02s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/02342d371a1948b4b27f591985d3b124 : FAILURE in 7m 19s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c0346137c8ea40869bc50546a32ec7fc : SUCCESS in 1h 37m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba8d8800d9bf4c0c815c6d5ed41d908d : SUCCESS in 4m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/232b7314d036460698e35f44bb63ba31 : SUCCESS in 1h 04m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b77b1f1e7621453c886965aa590c5f8a : SUCCESS in 1h 46m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8949b55716df45828446d55345be94ce : SUCCESS in 31m 35s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/496c87e212804088957e04110bcd86cb : SUCCESS in 24m 40s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/232102ee7fae4069930cc319bcc2df67 : SUCCESS in 5m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d25ab7f3acf24fdea2291eec8d7e3c6f : SUCCESS in 52m 28s","accounts_in_message":[],"_revision_number":1},{"id":"951517fdbfaae2146fa3398938ffcaf25f26a17b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-08-05 09:54:07.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29344a734f7d40e1650ae3a4b944b02a2db9235b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-08-05 13:18:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5ec50b9b0fc5939738c04cb5789502192b957269","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-08-05 13:37:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a716955b00551afbae850467d1cf62378c72ee56","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-08-07 14:52:45.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d8949d20022ca29b8da4cfe9d5c35be66ba2990a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-08-10 13:35:21.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9481f707c1a91b06412102d0b9558927eb92fdc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 13:35:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a0ee4fe7be0cbc3205f5d1e23899ad6afd6ecdea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 15:23:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b94a2f632b094988a4ccb849cbf79a98\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c2f267c01af4b769d1da08455915b44 : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5c35304bc494d3195bcdfcd8e3f8960 : SUCCESS in 6m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1dd7bb96097449e9aabc62b930b62bc0 : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/71e74764efb04424aac48f7fb4e195f6 : SUCCESS in 3m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ccea73bb9625486d9fbef1b9f202da59 : SUCCESS in 6m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c038303d2ad34998ac308a8ffdb31780 : SUCCESS in 1h 42m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/caf30df468754f5296a34dd81aa73a14 : SUCCESS in 3m 31s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4810417c469344c78fdffdcfa400acf8 : SUCCESS in 1h 06m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/281ac6e15446448d86a945a75a7483bc : SUCCESS in 1h 29m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ac49a5af0f4a4cf8819119110cf18729 : SUCCESS in 1h 06m 24s","accounts_in_message":[],"_revision_number":1},{"id":"09083b180cbf5c0afc7cc95a34b7f7c74f32ae22","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 15:23:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cb21b432fa075ad2e1bd378895fd411270e95425","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 15:24:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4140e50d6867444cb9ba6a2d6b413e82\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81f75b8791284c62abf3a076384232b5 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b58c72969b4033a54f98ebf449db52903e388196","revisions":{"b58c72969b4033a54f98ebf449db52903e388196":{"kind":"REWORK","_number":1,"created":"2026-08-04 14:34:18.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/31/999731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/31/999731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/31/999731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/31/999731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/31/999731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/31/999731/1"}}},"commit":{"parents":[{"commit":"92b5fb83df3d481d27bd2244d34cd3839c6ba625","subject":"Set neutron-lib as dependency for hacking pre-commit hook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/92b5fb83df3d481d27bd2244d34cd3839c6ba625"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-08-04 14:29:04.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-08-04 14:34:12.000000000","tz":120},"subject":"Add optional plugin_conf to the BaseOvnDbSynchronizer class","message":"Add optional plugin_conf to the BaseOvnDbSynchronizer class\n\nThis allows ovn-db-sync plugins provided by external projects to specify\ntheir own config files which will be loaded aside of the Neutron config\nfiles.\nThis new option can be used for example by the ovn-octavia-provider\u0027s\novn-db-sync plugin to provide Octavia configuration file. That file\ncan\u0027t be loaded simply with `--config-option` to the\nneutron-ovn-db-sync-util tool as that would overwrite some options\nlike e.g. database connection string which are common in both Octavia\nand Neutron.\n\nAssisted-by: Cursor Composer 2.5\n\nRelated-bug: #2156035\n\nChange-Id: I803527a1edcbc976d7e95e6a5be248d91bc9a40b\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b58c72969b4033a54f98ebf449db52903e388196"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b58c72969b4033a54f98ebf449db52903e388196"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
