)]}'
{"id":"openstack%2Fneutron-lib~999816","triplet_id":"openstack%2Fneutron-lib~master~Idc2df24e04374ebb3926afb8f152d8902dc27201","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-08-07 07:21:36.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2026-08-07 18:51:51.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-08-07 18:51:51.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-08-07 13:11:23.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Idc2df24e04374ebb3926afb8f152d8902dc27201","subject":"Add \"writeable FIP router_id\" API extension definition","status":"MERGED","created":"2026-08-05 12:42:30.000000000","updated":"2026-08-07 18:52:59.000000000","submitted":"2026-08-07 18:51:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999816","meta_rev_id":"c020d77c4db60f7354d45a51c6507ac4b04815a6","_number":999816,"virtual_id_number":999816,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-07 18:51:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-08-07 07:21:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2026-08-07 13:11:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2026-08-07 13:11:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-08-07 07:21:36.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-05 12:56:12.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-08-05 13:00:03.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-08-05 13:00:03.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-08-05 15:02:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-06 22:10:47.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-08-06 22:19:42.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"e338bb501f36b5e67fca28b428f6c9889adf0304","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-08-05 12:42:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"718ab3dff58b8eb864ac00d8de34347c1669aff2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-08-05 12:56:12.000000000","message":"Patch Set 1: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"bbd7350dcd013bee92529f7bf78f7a6f7e78d7de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-08-05 12:59:13.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"971e77a362e8c87b3eea2909b7658ec0a05129f5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-08-05 12:59:20.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1130b180f3b961b9856916d8a33fe24b38abb408","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-08-05 13:00:03.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"351bcdb9a103a4eac5f87f9fd9e2f20e7840c0de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 15:02:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f243ee7bb5142a78378385b3f489d35\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dab27b996b0d48f499e7599dff114545 : FAILURE in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58f890bf4d4247618c97908106bdeb92 : SUCCESS in 4m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/633dba4225bd441c99630b1763977f6d : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06eb234247cd46a59fe2e2dec87dcf81 : SUCCESS in 6m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ef9b9357f2d54acd84ed789203df7fda : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5973b4e172645588873f98779148788 : SUCCESS in 2m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9fde5cdeceaa4a74b9c046f443b42fd2 : SUCCESS in 8m 33s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/087feb7fbb374c84833c25d7d7dede8d : FAILURE in 5m 43s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91d01be85b834f7d91de2cb333fd47f3 : SUCCESS in 1h 37m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22360b69622e47c8b0055be92dc2216c : SUCCESS in 5m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/63000952df414814b426c5614a9c8db7 : SUCCESS in 1h 07m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d64e72c710064926a9c85b05e1e100cc : SUCCESS in 1h 56m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/235110eabcc946dfac49d69c87ce4797 : SUCCESS in 31m 21s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/18410bea881f402d9deb4c20f47243f8 : SUCCESS in 31m 00s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/5bb2d7648cab495a953ffca3b65f75dd : SUCCESS in 3m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/385d9ef2d42349f0a5c01ea1fe2a6c3f : SUCCESS in 1h 56m 24s","accounts_in_message":[],"_revision_number":2},{"id":"38d59385dcca4b3b4c4da103747f837f967eae43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-08-06 08:38:25.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"a756124ba26ce9e0bd55efb3af9c29bbd6510233","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 10:39:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4578fd1081f94146862602c1bdcb7cf9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d507905ef22541ae8254deb111278c43 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0834645f914542a5a1629f28dbf32238 : SUCCESS in 8m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fade75aafe84caca9c80fb8d8e71144 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fe5bbcea1214116a84ac946919aea62 : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ed248b56c86c43ce926864ea783b8c25 : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e67afe95c81e472d87ac27ea067888a8 : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d0a615daeec2475289ceba0b7ff1b296 : SUCCESS in 3m 56s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/860c78c53b0849faa377a338a08b03ce : FAILURE in 7m 10s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/68042b8155f646da9e691d57beb055ea : FAILURE in 35m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfb65695a9d34f4c8d3943b77a2b94bb : SUCCESS in 3m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c5569055ff074d37834663425c7ddfab : FAILURE in 26m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8b5a2d7dea984c8ea05576ce79cab877 : SUCCESS in 2h 00m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/912614a3807446fb8a7d6291c459d29f : SUCCESS in 30m 21s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/2765dad9a3fb49c6b678ba831a6e43f5 : SUCCESS in 17m 30s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/8ff8b21b193f456da6a2a3b023b1a90c : SUCCESS in 5m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/29e70b4db5da4244b5b9d431aed8386b : SUCCESS in 1h 28m 03s","accounts_in_message":[],"_revision_number":3},{"id":"2ca767d14d0a2360c85d998ff9d687da7f4c4749","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-08-06 13:27:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"16a6bb9733e6630e1adc2d6c182570b64550ea31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 15:24:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a000bd0b9d54b799b22a730cea70564\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c4c62a9be86946e386f67d1e37817731 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7be9278a6d4477885ce8943a876a0f8 : SUCCESS in 8m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8ddec899b9b437c911a2fcb3d384892 : SUCCESS in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73db2c9824614794ab1c9dd13a0c62e3 : SUCCESS in 7m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5644704968884d56b36988671e4d99dd : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/34b4cd09c2bc4693968ddd869480db6d : SUCCESS in 2m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6ec25698894c4a39809ce23d6ba4aa1f : SUCCESS in 7m 51s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/a07845d75193429b97650532279c9df2 : FAILURE in 7m 45s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/622ae9a8961248f8a7e91217c23d4a1f : SUCCESS in 47m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48b9d8317e724c16a8fe27685c99a616 : SUCCESS in 5m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a5e0a957db194863a522461360ef821b : SUCCESS in 57m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fff3f540656f4716bccb76c4ee44f95f : SUCCESS in 1h 54m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/93380916efa1496dbc95a042bce0fd3c : SUCCESS in 31m 15s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/f2db532576e743bdb76d954cf23290f9 : SUCCESS in 17m 44s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/a9d2c3f9a96d4c74a6ea2e89271db56d : SUCCESS in 3m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9cf05d67c385452290e436ff2f6f37ff : SUCCESS in 53m 06s","accounts_in_message":[],"_revision_number":3},{"id":"c16abeca54feb97ac3b616fa691285d19155ffc1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-08-06 22:10:47.000000000","message":"Patch Set 3: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8f8db1327452da60bff10d8a56463ae3a6eb26a2","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2026-08-06 22:19:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a02861f792844bd67a5e7e0f8fc9b0a8a7c76a29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-08-07 07:14:39.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"3f5fbbc267dea5575cb5d41dc85df64a70a35d52","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-08-07 07:15:03.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f7d2164d4a18d9dd3fae15d1576157c016d0959d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-08-07 07:21:36.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"e077eb7a27ef4c4fa725a2908b31b8f6a6ae7bc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 08:59:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e077099522a748e287080ca9bd735c31\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2b3c0cf6b0a5461097f985f3d6443450 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ef51139d81244a691ebbb7777af0128 : SUCCESS in 8m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f12a4239d79449eb4f48672abbdaadf : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd1d04ec8bc24490b8a9a88cd6426844 : SUCCESS in 8m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f038f6f0fd83442089486b0d35b9d390 : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b7d2b7ec56284f298bb5bc8b9f53f411 : SUCCESS in 3m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ff2a99dac2e844aab147c8ba4147fc1e : SUCCESS in 4m 52s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/ed403b8bafa74dba9b17bcae41197a68 : FAILURE in 4m 44s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/15b591bc6e3e4f1194a534166bec429b : SUCCESS in 1h 30m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/670b49e95963486383e36d65b3ba0414 : SUCCESS in 3m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c26d068901c44b76b340f70982ee4a46 : SUCCESS in 1h 06m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6ccde7375d554815a42096d2481c7271 : SUCCESS in 1h 44m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6e2f9ffa6804444cbe1c4b11dd5d50c4 : SUCCESS in 29m 28s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/8f471456fbb14d2686a0c58520f69fb4 : SUCCESS in 27m 42s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/91341251538545e680926806c00fd66f : SUCCESS in 4m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3824cc9a27fd4b2e947ebf7b48da2c1a : SUCCESS in 1h 15m 08s","accounts_in_message":[],"_revision_number":4},{"id":"766cd7eecbc90fc4e2933cd1d2cb5aa200501c28","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-08-07 13:11:23.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"cc1da353f1c1de1dc73eee9c0a47531932b8d77f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 13:12:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d0abc2d425e6ab174ad8a56e855b0f3a11139d5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 15:06:54.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14e00acca68c4dff85d0be73fd44ded8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f296e2dc2a8b4615865d8940a5896c21 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25f0a84923b7442a984eb003ea1d18b9 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddcd95b65bb94e6b821dcfd6cdd63019 : SUCCESS in 5m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7da508f6152643cf9442cd814c7e198a : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2e066674b07642a0bb3015d80a46ec94 : SUCCESS in 3m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/18655558ba684c62a0a742adf49ead53 : SUCCESS in 6m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b981873b8f97483ea2fb5623ff6aac11 : SUCCESS in 1h 26m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1050e3647045400bb9b3a291469c75cb : SUCCESS in 3m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e1783f8a41224387b13aa422b512dc76 : SUCCESS in 1h 01m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5de0961632a24a5eb7e671c73f3a8e4e : FAILURE in 1h 54m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/69a8e8f1171744cbbc9d308679a2f1b2 : SUCCESS in 1h 29m 43s","accounts_in_message":[],"_revision_number":4},{"id":"2c8a97b35071285aec6d7bf7ccfac3fbc313e9c9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-08-07 15:10:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"faeb85806a93d9ce7750753fd10b2661bb34f1b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 17:02:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4226e6ecfcbf436f97831fd6ba533ae9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d1096671d1894f8fbe57b6d93d09f7e1 : SUCCESS in 2m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/009d3737a408473e96afb6d7fb551fe1 : SUCCESS in 5m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f1ce714d51c467da70816bdff632a4a : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f274b79602046be887efdf0dc77511a : SUCCESS in 5m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/11c3ec8e247e4db89e4bdf798e56f593 : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f6428e9abd4644b28087673d910b3af4 : SUCCESS in 2m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b9e1effb70e54f72bf33bad1c442a27d : SUCCESS in 4m 57s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/557deccfc8124d47820cf6bf85c9a405 : FAILURE in 3m 26s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e67738d82fc4d1a9c860dd200eb2032 : SUCCESS in 45m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72d466c0303342dcb13ff2d9c924c2d5 : SUCCESS in 3m 24s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7908906e789e436fb99ab7a82e543144 : SUCCESS in 1h 03m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1cdb4a5f788449d0bd2c1b88ed9a862d : SUCCESS in 1h 51m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e3ac71915c02441b93cff72eb6ad31fd : SUCCESS in 33m 16s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/f7af2a2b22a34ff08c6acdd9189e09ae : SUCCESS in 24m 27s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/c0389afee47e462e85bad2d7454f44b8 : SUCCESS in 3m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9d6ab15908d941dbb4c78fd3b4a5cf15 : SUCCESS in 59m 14s","accounts_in_message":[],"_revision_number":4},{"id":"c6f37397780184aca1b099381d331b4435a45496","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 17:02:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ae355254574820c3ac6f6c9ed49859bac73790d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 18:51:51.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c51665eeb8404ab6be81d2249fe13f1c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2daecaac48454d6a99a0acdfebbde0d6 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5964f156e0f948b0a8d1d8a85785d4c0 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/126585d3b7a742dd8457d8fe2e79f92c : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/31c55b75d5e440a48cae238f10f44aad : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2772e6548c8e46e899b63aaac217f4b8 : SUCCESS in 3m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b668dc8fd6004ad29faba6387ea0efdf : SUCCESS in 3m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8229dd5357954b4bb2f941a3f4b147d2 : SUCCESS in 1h 19m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b30bec0f8d24f718869861d4dcff46e : SUCCESS in 3m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f4a37a95b1e14987a2ea104bafe97901 : SUCCESS in 58m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9aad2e7cc32d49e09110f6145cd835fb : SUCCESS in 1h 48m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e0521110bd2c4627999062229a3be689 : SUCCESS in 1h 15m 13s","accounts_in_message":[],"_revision_number":4},{"id":"18431636306e691f849e86e101e34e41454b7044","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 18:51:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"c020d77c4db60f7354d45a51c6507ac4b04815a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 18:52:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29c2118474aa4f9db529b780f9f0ac6a\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/73a304dbf8f642d8b6e5711d0ea594c0 : SUCCESS in 41s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1714f9d4a0584296809af2b3e45f7d65 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9908d055eac4f7c8843e8aeb7d75ed1 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"976e82928d010d354486a5702d028af05adff0b1","revisions":{"bb4333c18f9cdefd97520f8cd5a1f3dbd490d128":{"kind":"REWORK","_number":1,"created":"2026-08-05 12:42:30.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/16/999816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/16/999816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/16/999816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/16/999816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/16/999816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/16/999816/1"}}},"commit":{"parents":[{"commit":"4fcaf13754b4876ec749ac6380bc9792bd2c014c","subject":"Merge \"bgp: Add leak_routes subnet attribute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4fcaf13754b4876ec749ac6380bc9792bd2c014c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-08-05 12:39:03.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-08-05 12:39:03.000000000","tz":120},"subject":"Add \"writeable FIP router_id\" API extension definition","message":"Add \"writeable FIP router_id\" API extension definition\n\nThis patch adds definition of the new API extension. That extension\nmakes `router_id` attribute of Floating IP to be writable with POST and\nPUT requests and doing that will skip existing Neutron validation if\nFlaoting IP\u0027s network and tenant network have direct connection in any\nrouter_id. Instead user will be able to tell explicitly which router\nwill be used to do NAT for that Floating IP.\nSpec for that new feature was proposed in [1].\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/996269\n\nRelated-bug: #2072505\n\nChange-Id: Idc2df24e04374ebb3926afb8f152d8902dc27201\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bb4333c18f9cdefd97520f8cd5a1f3dbd490d128"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bb4333c18f9cdefd97520f8cd5a1f3dbd490d128"}]},"branch":"refs/heads/master"},"54c3f07bd4920a055e22f1521364d79e4c1992ef":{"kind":"REWORK","_number":2,"created":"2026-08-05 12:59:13.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/16/999816/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/16/999816/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/16/999816/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/16/999816/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/16/999816/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/16/999816/2"}}},"commit":{"parents":[{"commit":"4fcaf13754b4876ec749ac6380bc9792bd2c014c","subject":"Merge \"bgp: Add leak_routes subnet attribute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4fcaf13754b4876ec749ac6380bc9792bd2c014c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-08-05 12:39:03.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-08-05 12:58:58.000000000","tz":120},"subject":"Add \"writeable FIP router_id\" API extension definition","message":"Add \"writeable FIP router_id\" API extension definition\n\nThis patch adds definition of the new API extension. That extension\nmakes `router_id` attribute of Floating IP to be writable with POST and\nPUT requests and doing that will skip existing Neutron validation if\nFloating IP\u0027s network and tenant network have direct connection in any\nrouter_id. Instead user will be able to tell explicitly which router\nwill be used to do NAT for that Floating IP, bypassing the existing\nvalidation that requires a direct subnet connection between the floating\nIP\u0027s external network and the tenant network through a router.\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/996269\n\nRelated-bug: #2072505\n\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Idc2df24e04374ebb3926afb8f152d8902dc27201\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/54c3f07bd4920a055e22f1521364d79e4c1992ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/54c3f07bd4920a055e22f1521364d79e4c1992ef"}]},"branch":"refs/heads/master"},"653046f7ae6291102dc89129f0aba8f7a58690f3":{"kind":"REWORK","_number":3,"created":"2026-08-06 08:38:25.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/16/999816/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/16/999816/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/16/999816/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/16/999816/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/16/999816/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/16/999816/3"}}},"commit":{"parents":[{"commit":"4fcaf13754b4876ec749ac6380bc9792bd2c014c","subject":"Merge \"bgp: Add leak_routes subnet attribute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4fcaf13754b4876ec749ac6380bc9792bd2c014c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-08-05 12:39:03.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-08-06 08:38:09.000000000","tz":120},"subject":"Add \"writeable FIP router_id\" API extension definition","message":"Add \"writeable FIP router_id\" API extension definition\n\nThis patch adds definition of the new API extension. That extension\nmakes `router_id` attribute of Floating IP to be writable with POST and\nPUT requests and doing that will skip existing Neutron validation if\nFloating IP\u0027s network and tenant network have direct connection in any\nrouter_id. Instead user will be able to tell explicitly which router\nwill be used to do NAT for that Floating IP, bypassing the existing\nvalidation that requires a direct subnet connection between the floating\nIP\u0027s external network and the tenant network through a router.\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/996269\n\nRelated-bug: #2072505\n\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Idc2df24e04374ebb3926afb8f152d8902dc27201\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/653046f7ae6291102dc89129f0aba8f7a58690f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/653046f7ae6291102dc89129f0aba8f7a58690f3"}]},"branch":"refs/heads/master"},"976e82928d010d354486a5702d028af05adff0b1":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-08-07 07:14:39.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/16/999816/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/16/999816/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/16/999816/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/16/999816/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/16/999816/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/16/999816/4"}}},"commit":{"parents":[{"commit":"4fcaf13754b4876ec749ac6380bc9792bd2c014c","subject":"Merge \"bgp: Add leak_routes subnet attribute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4fcaf13754b4876ec749ac6380bc9792bd2c014c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-08-05 12:39:03.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-08-07 07:14:39.000000000","tz":0},"subject":"Add \"writeable FIP router_id\" API extension definition","message":"Add \"writeable FIP router_id\" API extension definition\n\nThis patch adds definition of the new API extension. That extension\nmakes `router_id` attribute of Floating IP to be writable with POST and\nPUT requests and doing that will skip existing Neutron validation if\nFloating IP\u0027s network and project network have direct connection in any\nrouter_id. Instead user will be able to tell explicitly which router\nwill be used to do NAT for that Floating IP, bypassing the existing\nvalidation that requires a direct subnet connection between the floating\nIP\u0027s external network and the project network through a router.\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/996269\n\nRelated-bug: #2072505\n\nAssisted-by: Cursor Composer 2.5\n\nChange-Id: Idc2df24e04374ebb3926afb8f152d8902dc27201\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/976e82928d010d354486a5702d028af05adff0b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/976e82928d010d354486a5702d028af05adff0b1"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
