)]}'
{"id":"openstack%2Fneutron-specs~788821","triplet_id":"openstack%2Fneutron-specs~master~Ic8db22d1b6570f68bd6400ecc653dc893a4b6184","project":"openstack/neutron-specs","branch":"master","topic":"off-path-smartnic-dpu-port-binding-with-ovn","attention_set":{},"removed_from_attention_set":{"13686":{"account":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"last_update":"2021-11-09 21:57:53.000000000","reason":"Change was submitted"},"24824":{"account":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"last_update":"2021-11-09 21:57:53.000000000","reason":"Change was submitted"},"9531":{"account":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"last_update":"2021-11-05 01:47:50.000000000","reason":"LIU Yulong replied on the change"}},"hashtags":[],"change_id":"Ic8db22d1b6570f68bd6400ecc653dc893a4b6184","subject":"Off-path SmartNIC Port Binding with OVN","status":"MERGED","created":"2021-04-29 18:54:18.000000000","updated":"2021-11-09 21:59:28.000000000","submitted":"2021-11-09 21:57:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":12,"has_review_started":true,"submission_id":"788821-off-path-smartnic-dpu-port-binding-with-ovn","meta_rev_id":"9a88cfdb5c2acd5d2bb8d8328969e0771502c8be","_number":788821,"virtual_id_number":788821,"owner":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-09 21:57:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2021-11-09 21:51:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2021-11-03 12:14:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2021-11-05 01:47:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":1,"date":"2021-11-09 21:51:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"reviewers":{"REVIEWER":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-29 18:56:58.000000000","updated_by":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"REVIEWER"},{"updated":"2021-04-29 18:56:58.000000000","updated_by":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"reviewer":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"state":"REVIEWER"},{"updated":"2021-04-29 18:56:58.000000000","updated_by":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"reviewer":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"state":"REVIEWER"},{"updated":"2021-04-29 18:56:58.000000000","updated_by":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"reviewer":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"state":"REVIEWER"},{"updated":"2021-04-29 19:05:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-29 19:13:59.000000000","updated_by":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"reviewer":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"state":"REVIEWER"},{"updated":"2021-05-04 07:19:14.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2021-05-10 10:07:33.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"CC"},{"updated":"2021-05-11 21:01:18.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2021-05-25 12:06:40.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-05-27 16:50:57.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-07-09 14:36:33.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-07-29 22:16:12.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REMOVED"},{"updated":"2021-11-03 11:00:29.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-11-05 01:47:50.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2021-11-09 21:51:52.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"67e34613c48d38f5bea254d624398b3b638520d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2021-04-29 18:54:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b6696b4174a23477ea3c2c1ce1026931b478d2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 19:05:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2357e86a927466da951c165c26aaf98 : SUCCESS in 8m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/473f6fe8581c44e192d67b1d9ecc3a12 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1},{"id":"791f21e95bc0f5e0054aedcb0beb8bec1d10ae53","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-04-29 19:13:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d55b63acc028b61a260c1469c3070e132ec5498f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-04 07:19:14.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nJust started the review","accounts_in_message":[],"_revision_number":1},{"id":"7ec4d97515d00212e55cbca97b9962d3f3796a1f","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2021-05-04 07:42:33.000000000","message":"Patch Set 1:\n\n(1 comment)\n\n\u003e Patch Set 1:\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e Just started the review\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"f70576ddeb71cdf23f377895d149313517458bb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2021-05-04 07:43:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"001e3ed2989c766d86f4c118deb9d74d80ff8c73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 07:49:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97f220cd25af41dcb3f669f677129c33 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a62efd68ce2144709879ca648b2e06a8 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":2},{"id":"95fd46fd0a71ecd7f28715ce738a3b519fe20d77","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-05-06 14:43:08.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6e2c62609cbed15349bd191083acb87b181f7f78","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-05-10 10:07:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c24181edd1cf5ab44fd99a7cb03773800d980384","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2021-05-10 10:33:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c83406ef4abef5c765363771b201f0927988d1ae","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-11 21:01:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5bd5e9d450387d04021a1aec388b55a734f9c621","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2021-05-12 08:00:05.000000000","message":"Patch Set 2:\n\nTo demonstrate what we are proposing to do I dusted off the PoC code we did prior to writing this spec:\n\nnova-compute PoC: https://pastebin.ubuntu.com/p/kbZ3zz7rPJ/\nneutron-server PoC: https://pastebin.ubuntu.com/p/gKKvTtYTFH/\nfixture ran on third host for purpose of PoC: https://pastebin.ubuntu.com/p/JQV6hywv7V/\n\nAfter Nova plugs or not plugs a VIF it will already today wait for a notification from Neutron about success of port binding. Neutron does not emit this notification until it has feedback from its agents or OVN that the port is successfully bound, so even if Nova does not do any plugging it will not resume the instance until networking is fully plumbed.","accounts_in_message":[],"_revision_number":2},{"id":"52a362787214359f1a166851786c4709dcece0bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2021-05-27 16:50:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6e3e37831a769dbade8667e340b1614c32bdb9bf","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2021-05-27 16:50:41.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"cb10f5ffe4e838d42df4ceb7b17fa188e282e142","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2021-05-27 16:52:47.000000000","message":"Patch Set 3:\n\nSorry about the lag. Been addressing the Nova spec comments and now incorporated the latest changes based on discussions there.","accounts_in_message":[],"_revision_number":3},{"id":"bbb006cd8c84b99ab63dd74d279e4cde6231b0a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 16:58:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b38f4dd19314c90b241a1814c72f873 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/030c320dd3704bdc8e6876ac1c799209 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":3},{"id":"272d8b5408bf4164af6e406b441f6ffd13cdf621","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2021-05-28 08:37:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9b5f0b5797bc1add38782b5a4cda62880c6b9d2d","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2021-05-28 08:38:22.000000000","message":"Patch Set 4:\n\nAdded a link to the representor port plugging proposal in OVN (by Frode Nordahl).\n\nhttps://patchwork.ozlabs.org/project/ovn/patch/20210509140305.1910796-1-frode.nordahl@canonical.com/","accounts_in_message":[],"_revision_number":4},{"id":"809806748976448b4a71a52847a6cb277d1765c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 08:42:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c29faa3ccbd14e7ca731f352e7ef136e : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96411cf1a2a74b60886b1cccf8cc5063 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":4},{"id":"fcab74a4bde1a7f370c876fcb243197c170c8e9c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-01 11:32:41.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"feaa2d82e6f8ad6d096da890d41435e031c7baa6","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2021-06-03 14:43:41.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"90ae12157c83f8423248b5f4eedd4e5b88670934","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2021-06-04 05:53:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"daf3f5121e6c53fa9274e3335c4b07a9a38edf56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 05:59:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a31aa9826aa043c8921311a24be2a137 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/830579cd9b0c4799a6f4eb88d1b0405f : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":5},{"id":"6bad3458b7be4d6898b13171fe180d3e42ae8c64","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-08 09:53:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"42e40eb5496e35518d8089373ec2c21ed6f7e8b7","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2021-07-09 14:08:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9c568317e838854257479fc002ddda50fa37fd19","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-09 14:36:33.000000000","message":"Patch Set 5: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"47178d61abd5b3f99953dd600429ba8126c491e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2021-07-13 08:45:45.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"033340fd7b1e01f54741bb1d973a69162f37b2b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2021-07-13 08:47:16.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7f32e872ce12a06d3a9ff6f2944031cf1c2d7438","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2021-07-13 08:47:42.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"0af74e0cf620909881f73c0c269752f0eb8cb2a2","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2021-07-13 08:49:27.000000000","message":"Patch Set 7:\n\n* Updated the IRC details to account for freenode -\u003e {oftc, libera} change;\n* Made notes about ovn-vif discussions upstream OVN.","accounts_in_message":[],"_revision_number":7},{"id":"aa7a80cb37f6bf982796a0c8dede55d3ffd66181","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 09:03:52.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd6cde2ab18a49afb798c7f38a7062bf : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb598b7d9bce418391350cb4e8f38e01 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":7},{"id":"29e08b8e38edd9bfcaa993cc6b0957d9ab0ecbae","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-29 22:16:12.000000000","message":"Removed reviewer Brian Haley.","accounts_in_message":[],"_revision_number":7},{"id":"1714ef5390a8bb0bdf0448125d78a168c45c4903","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2021-10-20 10:08:12.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"730b9f22363975b618ac9500d76c7499f8c4ba9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 10:16:15.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81f5eca3f54a4cc5a7d00a84445b4d53 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77021e2f3a0e470f87946e42579712a9 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":8},{"id":"bceadff94935621d86d708f4f0414ffed2c1e13d","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-10-28 08:45:03.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e74d83f24e14cb8570056face902a3d2dd558686","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2021-10-28 10:10:51.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fdff2fba10e60c6f0254c966b5312c18c8489b05","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2021-10-29 13:57:26.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"35324874926d202b84ec19d9d857269a257a79dc","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2021-10-29 14:01:26.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2556a32b4e67e2da98305c4c8a8a3c6a4b61cb06","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-03 11:00:29.000000000","message":"Patch Set 8: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"0cdaf08264276d088a93a59a420255abb9db8533","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2021-11-03 12:06:42.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"1ee516004aaa72da1b494bc41c19dfad3869c393","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-03 12:14:15.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"73462d158b588275a757705e3cc511f5c3644667","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 12:18:54.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e0c0db414c84d23ac1bcb1d85e319c1 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4504fb0d412e49d697ce2f53d8cee442 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":9},{"id":"513af149269501af32ae5522623f9c00798a3037","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-11-05 01:47:50.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"e6a0243544e3572d35a3f570f47cc8ed2681a478","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2021-11-08 10:01:05.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"8b6adae99ec816ea748de37dd2481e5d0d4abc2d","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-11-09 21:51:52.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"b9f9f35a8c335cbb2a9ec34db85740ec012af3af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 21:52:23.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"e91f7ff057437735b4c46a017ce3f99fbbc15796","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 21:57:52.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c92fa4abcba9488f92554110dbd4f6f2 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97a5de596220457d86c607fe1c2e3d90 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":9},{"id":"1732ec429faad39dc75c453097a130352641107d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 21:57:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"9a88cfdb5c2acd5d2bb8d8328969e0771502c8be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 21:59:28.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/59537e332cbb42ba8861f46502dfc8ce : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"8ff7a77df9ec4fb3b5cb21ca42fc302b855a63fb","revisions":{"ad79d6f7e2fb3c528a74aa9395959d287a656129":{"kind":"REWORK","_number":1,"created":"2021-04-29 18:54:18.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/21/788821/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-specs","ref":"refs/changes/21/788821/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/1"}}},"commit":{"parents":[{"commit":"e3f87af3bc32fb19d81acdadd4d69709e10e6c69","subject":"Spec folder for Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/e3f87af3bc32fb19d81acdadd4d69709e10e6c69"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2021-04-29 18:35:04.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2021-04-29 18:54:16.000000000","tz":180},"subject":"Off-path SmartNIC Port Binding and Offload","message":"Off-path SmartNIC Port Binding and Offload\n\nhttps://blueprints.launchpad.net/neutron/+spec/off-path-smartnic-port-binding-and-offload\n\nOff-path SmartNICs (or DPUs) introduce an architecture change where\nnetwork agents responsible for eSwitch configuration and representor\ninterface plugging run on a separate SoC with its own CPU, memory and\nthat runs a separate OS kernel. The side-effect of that is that\nhypervisor hostnames no longer match SmartNIC (DPU) hostnames which are\nseen by OVS and OVN agents while the existing port binding code relies\non that. The goal of this specification is to introduce changes into the\nOVN mechanism driver code necessary to extend the existing hardware\noffload-related code to cope with the hostname mismatch and related\ndesign challenges while reusing the rest of the code. To do that, the\nOVN mechanism driver code is extended to accept additional information\nfrom port updates coming from Nova in a port\u0027s \"binding:profile\"\nattribute in order to identify a correct transport node (chassis) and\nrepresentor port at the transport node host side.\n\nChange-Id: Ic8db22d1b6570f68bd6400ecc653dc893a4b6184\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/ad79d6f7e2fb3c528a74aa9395959d287a656129"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/ad79d6f7e2fb3c528a74aa9395959d287a656129"}]},"branch":"refs/heads/master"},"2e469faff5ae3a7541506451c17a923799468b8f":{"kind":"REWORK","_number":2,"created":"2021-05-04 07:43:32.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/21/788821/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-specs","ref":"refs/changes/21/788821/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/2"}}},"commit":{"parents":[{"commit":"e3f87af3bc32fb19d81acdadd4d69709e10e6c69","subject":"Spec folder for Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/e3f87af3bc32fb19d81acdadd4d69709e10e6c69"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2021-04-29 18:35:04.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2021-05-04 07:38:07.000000000","tz":180},"subject":"Off-path SmartNIC Port Binding and Offload","message":"Off-path SmartNIC Port Binding and Offload\n\nhttps://blueprints.launchpad.net/neutron/+spec/off-path-smartnic-port-binding-and-offload\n\nOff-path SmartNICs (or DPUs) introduce an architecture change where\nnetwork agents responsible for eSwitch configuration and representor\ninterface plugging run on a separate SoC with its own CPU, memory and\nthat runs a separate OS kernel. The side-effect of that is that\nhypervisor hostnames no longer match SmartNIC (DPU) hostnames which are\nseen by OVS and OVN agents while the existing port binding code relies\non that. The goal of this specification is to introduce changes into the\nOVN mechanism driver code necessary to extend the existing hardware\noffload-related code to cope with the hostname mismatch and related\ndesign challenges while reusing the rest of the code. To do that, the\nOVN mechanism driver code is extended to accept additional information\nfrom port updates coming from Nova in a port\u0027s \"binding:profile\"\nattribute in order to identify a correct transport node (chassis) and\nrepresentor port at the transport node host side.\n\nNova spec: https://review.opendev.org/c/openstack/nova-specs/+/787458\n\nChange-Id: Ic8db22d1b6570f68bd6400ecc653dc893a4b6184\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/2e469faff5ae3a7541506451c17a923799468b8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/2e469faff5ae3a7541506451c17a923799468b8f"}]},"branch":"refs/heads/master"},"94980b17b321b4d0de04298a3ba8b9c2e640f834":{"kind":"REWORK","_number":3,"created":"2021-05-27 16:50:32.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/21/788821/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-specs","ref":"refs/changes/21/788821/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/3"}}},"commit":{"parents":[{"commit":"e3f87af3bc32fb19d81acdadd4d69709e10e6c69","subject":"Spec folder for Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/e3f87af3bc32fb19d81acdadd4d69709e10e6c69"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2021-04-29 18:35:04.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2021-05-27 16:49:38.000000000","tz":180},"subject":"Off-path SmartNIC Port Binding and Offload","message":"Off-path SmartNIC Port Binding and Offload\n\nhttps://blueprints.launchpad.net/neutron/+spec/off-path-smartnic-dpu-port-binding\n\nOff-path SmartNIC DPUs introduce an architecture change where network\nagents responsible for NIC switch configuration and representor\ninterface plugging run on a separate SoC with its own CPU, memory and\nthat runs a separate OS kernel. The side-effect of that is that\nhypervisor hostnames no longer match SmartNIC DPU hostnames which are\nseen by ovs-vswitchd and OVN [3]_ agents while the existing port binding\ncode relies on that. The goal of this specification is to introduce\nchanges necessary to extend the existing hardware offload code to cope\nwith the hostname mismatch and related design challenges while reusing\nthe rest of the code. To do that, PCI(e) add-in card tracking is\nintroduced for boards with unique serial numbers so that it can be used\nto determine the correct hostname of a SmartNIC DPU which is responsible\nfor a particular VF. Additionally, more information is suggested to be\npassed in the \"binding:profile\" during a port update to facilitate\nrepresentor port plugging.\n\nNova spec: https://review.opendev.org/c/openstack/nova-specs/+/787458\nNova BP: https://blueprints.launchpad.net/nova/+spec/integration-with-off-path-network-backends\n\nChange-Id: Ic8db22d1b6570f68bd6400ecc653dc893a4b6184\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/94980b17b321b4d0de04298a3ba8b9c2e640f834"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/94980b17b321b4d0de04298a3ba8b9c2e640f834"}]},"branch":"refs/heads/master"},"348058fa8eac72f1df9d897e45fe60f5c9bb356b":{"kind":"REWORK","_number":4,"created":"2021-05-28 08:37:19.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/21/788821/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-specs","ref":"refs/changes/21/788821/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/4"}}},"commit":{"parents":[{"commit":"e3f87af3bc32fb19d81acdadd4d69709e10e6c69","subject":"Spec folder for Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/e3f87af3bc32fb19d81acdadd4d69709e10e6c69"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2021-04-29 18:35:04.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2021-05-28 08:37:12.000000000","tz":180},"subject":"Off-path SmartNIC Port Binding","message":"Off-path SmartNIC Port Binding\n\nhttps://blueprints.launchpad.net/neutron/+spec/off-path-smartnic-dpu-port-binding\n\nOff-path SmartNIC DPUs introduce an architecture change where network\nagents responsible for NIC switch configuration and representor\ninterface plugging run on a separate SoC with its own CPU, memory and\nthat runs a separate OS kernel. The side-effect of that is that\nhypervisor hostnames no longer match SmartNIC DPU hostnames which are\nseen by ovs-vswitchd and OVN [3]_ agents while the existing port binding\ncode relies on that. The goal of this specification is to introduce\nchanges necessary to extend the existing hardware offload code to cope\nwith the hostname mismatch and related design challenges while reusing\nthe rest of the code. To do that, PCI(e) add-in card tracking is\nintroduced for boards with unique serial numbers so that it can be used\nto determine the correct hostname of a SmartNIC DPU which is responsible\nfor a particular VF. Additionally, more information is suggested to be\npassed in the \"binding:profile\" during a port update to facilitate\nrepresentor port plugging.\n\nNova spec: https://review.opendev.org/c/openstack/nova-specs/+/787458\nNova BP: https://blueprints.launchpad.net/nova/+spec/integration-with-off-path-network-backends\n\nChange-Id: Ic8db22d1b6570f68bd6400ecc653dc893a4b6184\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/348058fa8eac72f1df9d897e45fe60f5c9bb356b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/348058fa8eac72f1df9d897e45fe60f5c9bb356b"}]},"branch":"refs/heads/master"},"092462342f3a5b25f19218f77732bf9397ee7740":{"kind":"REWORK","_number":5,"created":"2021-06-04 05:53:54.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/21/788821/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-specs","ref":"refs/changes/21/788821/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/5"}}},"commit":{"parents":[{"commit":"e3f87af3bc32fb19d81acdadd4d69709e10e6c69","subject":"Spec folder for Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/e3f87af3bc32fb19d81acdadd4d69709e10e6c69"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2021-04-29 18:35:04.000000000","tz":180},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2021-06-04 05:53:52.000000000","tz":120},"subject":"Off-path SmartNIC Port Binding","message":"Off-path SmartNIC Port Binding\n\nhttps://blueprints.launchpad.net/neutron/+spec/off-path-smartnic-dpu-port-binding\n\nOff-path SmartNIC DPUs introduce an architecture change where network\nagents responsible for NIC switch configuration and representor\ninterface plugging run on a separate SoC with its own CPU, memory and\nthat runs a separate OS kernel. The side-effect of that is that\nhypervisor hostnames no longer match SmartNIC DPU hostnames which are\nseen by ovs-vswitchd and OVN [3]_ agents while the existing port binding\ncode relies on that. The goal of this specification is to introduce\nchanges necessary to extend the existing hardware offload code to cope\nwith the hostname mismatch and related design challenges while reusing\nthe rest of the code. To do that, PCI(e) add-in card tracking is\nintroduced for boards with unique serial numbers so that it can be used\nto determine the correct hostname of a SmartNIC DPU which is responsible\nfor a particular VF. Additionally, more information is suggested to be\npassed in the \"binding:profile\" during a port update to facilitate\nrepresentor port plugging.\n\nNova spec: https://review.opendev.org/c/openstack/nova-specs/+/787458\nNova BP: https://blueprints.launchpad.net/nova/+spec/integration-with-off-path-network-backends\n\nChange-Id: Ic8db22d1b6570f68bd6400ecc653dc893a4b6184\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/092462342f3a5b25f19218f77732bf9397ee7740"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/092462342f3a5b25f19218f77732bf9397ee7740"}]},"branch":"refs/heads/master"},"79de2fa60ce08facf98b2b57316529ab6872f26c":{"kind":"REWORK","_number":6,"created":"2021-07-13 08:45:45.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/21/788821/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-specs","ref":"refs/changes/21/788821/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/6"}}},"commit":{"parents":[{"commit":"e3f87af3bc32fb19d81acdadd4d69709e10e6c69","subject":"Spec folder for Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/e3f87af3bc32fb19d81acdadd4d69709e10e6c69"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2021-04-29 18:35:04.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2021-07-13 08:43:45.000000000","tz":180},"subject":"Off-path SmartNIC Port Binding with OVN","message":"Off-path SmartNIC Port Binding with OVN\n\nhttps://blueprints.launchpad.net/neutron/+spec/off-path-smartnic-dpu-port-binding-with-ovn\n\nOff-path SmartNIC DPUs introduce an architecture change where network\nagents responsible for NIC switch configuration and representor\ninterface plugging run on a separate SoC with its own CPU, memory and\nthat runs a separate OS kernel. The side-effect of that is that\nhypervisor hostnames no longer match SmartNIC DPU hostnames which are\nseen by ovs-vswitchd and OVN [3]_ agents while the existing port binding\ncode relies on that. The goal of this specification is to introduce\nchanges necessary to extend the existing hardware offload code to cope\nwith the hostname mismatch and related design challenges while reusing\nthe rest of the code. To do that, PCI(e) add-in card tracking is\nintroduced for boards with unique serial numbers so that it can be used\nto determine the correct hostname of a SmartNIC DPU which is responsible\nfor a particular VF. Additionally, more information is suggested to be\npassed in the \"binding:profile\" during a port update to facilitate\nrepresentor port plugging.\n\nNova spec: https://review.opendev.org/c/openstack/nova-specs/+/787458\nNova BP: https://blueprints.launchpad.net/nova/+spec/integration-with-off-path-network-backends\n\nChange-Id: Ic8db22d1b6570f68bd6400ecc653dc893a4b6184\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/79de2fa60ce08facf98b2b57316529ab6872f26c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/79de2fa60ce08facf98b2b57316529ab6872f26c"}]},"branch":"refs/heads/master"},"e7a0d95ea4c91c9b0cc822d55ab0158116a86acb":{"kind":"REWORK","_number":7,"created":"2021-07-13 08:47:16.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/21/788821/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-specs","ref":"refs/changes/21/788821/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/7"}}},"commit":{"parents":[{"commit":"e3f87af3bc32fb19d81acdadd4d69709e10e6c69","subject":"Spec folder for Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/e3f87af3bc32fb19d81acdadd4d69709e10e6c69"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2021-04-29 18:35:04.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2021-07-13 08:47:06.000000000","tz":180},"subject":"Off-path SmartNIC Port Binding with OVN","message":"Off-path SmartNIC Port Binding with OVN\n\nhttps://blueprints.launchpad.net/neutron/+spec/off-path-smartnic-dpu-port-binding-with-ovn\n\nOff-path SmartNIC DPUs introduce an architecture change where network\nagents responsible for NIC switch configuration and representor\ninterface plugging run on a separate SoC with its own CPU, memory and\nthat runs a separate OS kernel. The side-effect of that is that\nhypervisor hostnames no longer match SmartNIC DPU hostnames which are\nseen by ovs-vswitchd and OVN [3]_ agents while the existing port binding\ncode relies on that. The goal of this specification is to introduce\nchanges necessary to extend the existing hardware offload code to cope\nwith the hostname mismatch and related design challenges while reusing\nthe rest of the code. To do that, PCI(e) add-in card tracking is\nintroduced for boards with unique serial numbers so that it can be used\nto determine the correct hostname of a SmartNIC DPU which is responsible\nfor a particular VF. Additionally, more information is suggested to be\npassed in the \"binding:profile\" during a port update to facilitate\nrepresentor port plugging.\n\nNova spec: https://review.opendev.org/c/openstack/nova-specs/+/787458\nNova BP: https://blueprints.launchpad.net/nova/+spec/integration-with-off-path-network-backends\n\nChange-Id: Ic8db22d1b6570f68bd6400ecc653dc893a4b6184\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/e7a0d95ea4c91c9b0cc822d55ab0158116a86acb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/e7a0d95ea4c91c9b0cc822d55ab0158116a86acb"}]},"branch":"refs/heads/master"},"cd755464825baaf5dfa1d77c3e696ed6fe987b5a":{"kind":"REWORK","_number":8,"created":"2021-10-20 10:08:12.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/21/788821/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-specs","ref":"refs/changes/21/788821/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/8"}}},"commit":{"parents":[{"commit":"21b6188e86cf3cd1b095dc061da10bb447fec57a","subject":"Spec folder for Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/21b6188e86cf3cd1b095dc061da10bb447fec57a"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2021-04-29 18:35:04.000000000","tz":180},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2021-10-20 10:07:24.000000000","tz":120},"subject":"Off-path SmartNIC Port Binding with OVN","message":"Off-path SmartNIC Port Binding with OVN\n\nhttps://blueprints.launchpad.net/neutron/+spec/off-path-smartnic-dpu-port-binding-with-ovn\n\nOff-path SmartNIC DPUs introduce an architecture change where network\nagents responsible for NIC switch configuration and representor\ninterface plugging run on a separate SoC with its own CPU, memory and\nthat runs a separate OS kernel. The side-effect of that is that\nhypervisor hostnames no longer match SmartNIC DPU hostnames which are\nseen by ovs-vswitchd and OVN [3]_ agents while the existing port binding\ncode relies on that. The goal of this specification is to introduce\nchanges necessary to extend the existing hardware offload code to cope\nwith the hostname mismatch and related design challenges while reusing\nthe rest of the code. To do that, PCI(e) add-in card tracking is\nintroduced for boards with unique serial numbers so that it can be used\nto determine the correct hostname of a SmartNIC DPU which is responsible\nfor a particular VF. Additionally, more information is suggested to be\npassed in the \"binding:profile\" during a port update to facilitate\nrepresentor port plugging.\n\nNova spec: https://review.opendev.org/c/openstack/nova-specs/+/787458\nNova BP: https://blueprints.launchpad.net/nova/+spec/integration-with-off-path-network-backends\n\nChange-Id: Ic8db22d1b6570f68bd6400ecc653dc893a4b6184\nCloses-Bug: #1932154\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/cd755464825baaf5dfa1d77c3e696ed6fe987b5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/cd755464825baaf5dfa1d77c3e696ed6fe987b5a"}]},"branch":"refs/heads/master"},"8ff7a77df9ec4fb3b5cb21ca42fc302b855a63fb":{"kind":"REWORK","_number":9,"created":"2021-11-03 12:06:42.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/21/788821/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-specs","ref":"refs/changes/21/788821/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-specs refs/changes/21/788821/9"}}},"commit":{"parents":[{"commit":"40f6c1663d045fbb23dcb64f4fc0227f6f063dd4","subject":"Neutron VLAN networks with QinQ enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/40f6c1663d045fbb23dcb64f4fc0227f6f063dd4"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2021-04-29 18:35:04.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2021-11-03 12:06:23.000000000","tz":180},"subject":"Off-path SmartNIC Port Binding with OVN","message":"Off-path SmartNIC Port Binding with OVN\n\nhttps://blueprints.launchpad.net/neutron/+spec/off-path-smartnic-dpu-port-binding-with-ovn\n\nOff-path SmartNIC DPUs introduce an architecture change where network\nagents responsible for NIC switch configuration and representor\ninterface plugging run on a separate SoC with its own CPU, memory and\nthat runs a separate OS kernel. The side-effect of that is that\nhypervisor hostnames no longer match SmartNIC DPU hostnames which are\nseen by ovs-vswitchd and OVN agents while the existing port binding\ncode relies on that. The goal of this specification is to introduce\nchanges necessary to extend the existing hardware offload code to cope\nwith the hostname mismatch and related design challenges while reusing\nthe rest of the code. To do that, PCI(e) add-in card tracking is\nintroduced for boards with unique serial numbers so that it can be used\nto determine the correct hostname of a SmartNIC DPU which is responsible\nfor a particular VF. Additionally, more information is suggested to be\npassed in the \"binding:profile\" during a port update to facilitate\nrepresentor port plugging.\n\nWIP code: https://review.opendev.org/c/openstack/neutron/+/808961\nNova spec: https://review.opendev.org/c/openstack/nova-specs/+/787458\nNova BP: https://blueprints.launchpad.net/nova/+spec/integration-with-off-path-network-backends\n\nNeeded-By: I07ef52769da72cde8867f996111b7df4a80e4d79\nChange-Id: Ic8db22d1b6570f68bd6400ecc653dc893a4b6184\nCloses-Bug: #1932154\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/8ff7a77df9ec4fb3b5cb21ca42fc302b855a63fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-specs/commit/8ff7a77df9ec4fb3b5cb21ca42fc302b855a63fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
