)]}'
{"id":"openstack%2Fneutron-tempest-plugin~1000968","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I6bd79e7bfbc5e496009fed53dfe8d20f5f3af748","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-08-25 12:47:33.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-08-25 15:35:21.000000000","reason":"removed on reply"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-08-26 00:04:47.000000000","reason":"Change was submitted"},"31291":{"account":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"last_update":"2026-08-26 00:04:47.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-08-25 14:49:33.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I6bd79e7bfbc5e496009fed53dfe8d20f5f3af748","subject":"Fix admin router cleanup and remove BaseRouterTest class","status":"MERGED","created":"2026-08-14 14:11:45.000000000","updated":"2026-08-26 00:04:47.000000000","submitted":"2026-08-26 00:04:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1000968","meta_rev_id":"2a1509ea656e128642c485eaa90a7be605843a37","_number":1000968,"virtual_id_number":1000968,"owner":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-26 00:04:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-08-25 12:47:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":-1,"date":"2026-08-25 15:35:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-08-25 14:49:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-25 19:42:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2026-08-25 14:49:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-08-25 12:47:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"}],"CC":[{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-14 16:56:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-14 19:35:07.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-08-15 00:51:28.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-08-23 17:26:47.000000000","updated_by":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"reviewer":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"state":"CC"},{"updated":"2026-08-24 10:54:18.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-08-24 11:13:00.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-08-24 15:05:58.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-08-25 19:42:31.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"CC"},{"updated":"2026-08-25 19:42:36.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"}],"messages":[{"id":"f4ecfdd9d65c6b937ec6fa02cc864dc15c4bf8d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-08-14 14:11:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8687feef980caf3d5cd92184b1780a4be7865072","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-08-14 14:15:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a8a164063fe2cdc1fb607ae55435b81e1f953b33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-08-14 14:26:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"73732568f3988c2350902556adc3f4cf019e9ddd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 16:56:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b5f1b9578ef45c7ba11c036a71b5c55\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f70047a3124345ff9fde665ce1984d73 : SUCCESS in 2m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e32f30fd0bd942f482e63fdbf35eca00 : SUCCESS in 1h 11m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c7ee149e3d1b4fa185ee5f41ead545a5 : SUCCESS in 1h 42m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/101a5517c42a4f53810ef2da1589b3d0 : SUCCESS in 30m 13s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/420f6ca78d8d4da29497ed27d0cd396b : FAILURE in 1h 33m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/bcafe49f468448c8ad974e2bd3a10929 : SUCCESS in 1h 31m 43s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/bc881857ee9a4b759a5a5cf2ee7093b6 : SUCCESS in 2h 29m 22s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/d6f10df027b74384807a80dc5859e565 : SUCCESS in 33m 28s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/7ddfe1914c9e4c9ca5ee3e11f823bd80 : SUCCESS in 1h 11m 37s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/52d7eddccef24f1eb6045df64b44f5f3 : SUCCESS in 1h 20m 24s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/cc9a8a2c7bc8424eb1daf481a0223e13 : SUCCESS in 1h 23m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/59a67984bd8e4f26adc4d7ffe6cfae77 : SUCCESS in 1h 57m 01s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/9d60def31aa34949a3891dbf75661834 : SUCCESS in 2h 04m 45s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/0aebb7d548bc489186a22721a3ccc677 : SUCCESS in 31m 50s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/0414cf1f814d4b5195239a15d9c3d8d5 : SUCCESS in 1h 28m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/0e5a3cc235f04bfb8b9ed26a2bcd83d5 : SUCCESS in 1h 32m 51s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/54e3113eaf8c458da8a2e23bf32d109b : SUCCESS in 1h 55m 05s\n- neutron-tempest-plugin-designate-scenario-2026-1 https://zuul.opendev.org/t/openstack/build/7760ba9685164a3fbe8813a29adc23d6 : SUCCESS in 32m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc977d502645420a9e4a35d825b9d399 : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6676f427051443408446cd19bddfe2bf : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":3},{"id":"22e92e655cc5cbefa1a576989f3d3586f1019fed","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-15 00:51:28.000000000","message":"Patch Set 3: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/e504d758d8a1424295aea60222cbb7a5\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/aaf192c1264e410c84784cedb3bc02ad : SUCCESS in 9h 47m 06s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/2b6e423c7260475783ff16ba7cafe90a : FAILURE in 3h 53m 50s","accounts_in_message":[],"_revision_number":3},{"id":"ec98fda382e6cf5b27cf4be98e4caf4316e1129b","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-08-17 10:59:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6c6f37922ff6556b69d5cede8e9a44f5c9e436c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 13:02:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f173a505884a44e3aaeed538710047d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a68d59641c2841e58be0a067b9205d94 : SUCCESS in 2m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/25095675aeb146b2b87aa26cc4d6167c : SUCCESS in 1h 11m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bc447e4add0f4091812b9d2fecb564a7 : SUCCESS in 1h 45m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4ef163bc64e84d969e7394962cf2b190 : SUCCESS in 34m 51s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/13bb662e41824e88bfb216892c02a470 : SUCCESS in 1h 39m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/c73e17436d5841f38c2c913a5e071f9b : SUCCESS in 1h 46m 02s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/30222de3772c4e77bf8c034c6626794c : SUCCESS in 1h 57m 44s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/7dbca393ce2a479cb0ee94bcf48fff14 : SUCCESS in 32m 48s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/e3685804849d4079814631d6e0f5576c : SUCCESS in 1h 34m 36s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/0af23dd49270485aaa522e410f5e0cd7 : SUCCESS in 1h 23m 05s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/7ecb2e51f25d4e9fbd7cba04670141c3 : SUCCESS in 1h 21m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/f78ebb0874ad4300aa3b1cd4ab23ab6d : SUCCESS in 1h 50m 35s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/32fcb176c7214b5b9862849748b14af7 : SUCCESS in 1h 56m 37s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/490fe9bd66c8418ba2a8493183a67514 : SUCCESS in 32m 38s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/134eeb31e37d4ad9bccca5349a5b8a46 : SUCCESS in 1h 36m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/7d7ebcc0b70849bbafd5c259f6dade4a : SUCCESS in 1h 55m 48s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/616c252ce585496896a4f9ad95a18b4a : SUCCESS in 2h 01m 16s\n- neutron-tempest-plugin-designate-scenario-2026-1 https://zuul.opendev.org/t/openstack/build/2cb3ea88200c492d840bb85db1b992e0 : SUCCESS in 37m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31ed54073ff84d01acd74807b9cf4b11 : SUCCESS in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c17619420294e55998218cfae3024e3 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":3},{"id":"0e732d1b2d3bdb23e0880fb605e18c450e36218e","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-17 17:18:27.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/75de17cf01ac40a59eaa12a078672115\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/7538da9b227a458598f354a2e4ca610e : SUCCESS in 4h 12m 10s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/9d8e54ed5a994988ba71857096ba17c0 : FAILURE in 3h 47m 03s","accounts_in_message":[],"_revision_number":3},{"id":"4a29411ad816e07c40020756aa5af9f13f29e5d8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-08-24 10:54:18.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"8b970eb90723e2d5f8201859856d5ccb7726936c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-08-24 10:55:21.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"ec1915a94b92e10b8cff1dcbc5313c5fd3f56877","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-08-24 10:57:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4b81caf2cffae1ddfbe4e75b627c691d64292631","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-08-24 10:59:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"90b75d9438052c337da89942683d25ee06554d06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-08-24 11:03:44.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"bf95bd1779d7c85fd9e6f734dde4fd2a2e5b1136","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-08-24 11:04:11.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"6afcb67df3948e9bbe0807c0fd46ce09b74a9d6a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-08-24 11:13:00.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fb944ccad39ed94055b301f31a59fc8ef9465208","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 13:17:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/07596f26d3524c12b84d4c6aa4bb848d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eab8adaca634ae7bf59f43dda460e42 : SUCCESS in 2m 23s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0f6e0d83c13e4aa987e158c1ba58abd3 : SUCCESS in 1h 02m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a824aa98cd794fdc884ef198ffa5459c : SUCCESS in 1h 59m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4872648e33c6465a8fb6127166e34948 : SUCCESS in 29m 57s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/4a5fbecccc7f49f4bee4c01133d29321 : SUCCESS in 1h 30m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/361bc04e8a3b4e3199a65f3e96390c74 : SUCCESS in 2h 01m 11s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/a8200370c99449f09aec4898993c0857 : SUCCESS in 1h 58m 12s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/5ce103a6b554489bbbb786babf44e5b4 : SUCCESS in 31m 19s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/adc142f7a0884c75af413b0c7b174592 : SUCCESS in 1h 09m 56s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/b425953039a84935a03cbc5d5e24addf : POST_FAILURE in 1h 15m 54s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/b7a9fdac2a3a4611aba4a26527492fdf : SUCCESS in 1h 19m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/f9b53cf7dab14f109a148a45204f2b54 : SUCCESS in 1h 53m 25s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/d85d4790ad154b0a9a6a82d3f0ed6a17 : SUCCESS in 1h 54m 12s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/9596fed558f94a82874e7eb82502ad35 : SUCCESS in 38m 52s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/ea8c71fad02c4f589136b82d556ee7f0 : SUCCESS in 1h 39m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/c80e23b0fe3a4effa49b65723105ad3f : SUCCESS in 2h 08m 10s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/09ec28600aa043349878d6fe0768b358 : SUCCESS in 2h 12m 22s\n- neutron-tempest-plugin-designate-scenario-2026-1 https://zuul.opendev.org/t/openstack/build/0e54b689b2c448018dfa13e4c33f1f5a : SUCCESS in 31m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a73595174e744567b4403da074609069 : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca74c8c3b55041e69578620a61350df3 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":4},{"id":"5a13a27fa41e32cd4fdac53326f576fe271a3077","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-08-24 13:24:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1584348e69e3887539559f8180bcbab7fe438842","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-08-24 14:27:13.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4a2185e3d45df90234e62c3bd676f115d789dcba","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-24 14:51:47.000000000","message":"Patch Set 4: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/c31b7473becd446f883afba96dafd43c\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/d0fcd6c65a024479b94be28afd763dba : SUCCESS in 2h 17m 41s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/742f39ebd2bc464ba516228ca2d4bd7f : FAILURE in 1h 59m 51s","accounts_in_message":[],"_revision_number":4},{"id":"61dcf7c0449ced623b48507f63cfb72b117b62b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-08-24 14:52:45.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1, Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"d461940e3c8491f5463d081478ae982d4b511a55","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-08-24 14:55:12.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5f2daa0e2f0f396057d0f68cb87274f56d1fdc33","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-08-24 15:05:58.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f35f44e7dcb69c1f3b5fb9c1b4fdaea32801f1af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 17:20:03.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6bad730b1be542f89fbf70a41dd4ecdf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/674506d29cb549359438c2cbe9933b05 : SUCCESS in 4m 23s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d482161c23bf4d4bb22cd6cd6c4dbe27 : FAILURE in 1h 15m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/96f6b57dd0ed476091d9dab57df8d5f0 : FAILURE in 1h 34m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0bb988d3e80448afbf4f41899c02c86c : SUCCESS in 34m 13s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/d66d4f36e70440dfaee501e4262657da : FAILURE in 1h 36m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/37e6369f11054602ae688f242c29f705 : FAILURE in 1h 47m 53s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/c82885c6122841f9bb287e0739f91bed : FAILURE in 2h 02m 12s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/e3bf8909766b458eb43c5a8f0e778e4e : SUCCESS in 41m 51s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/b44e53113248486bb53b0d51f0033a5a : SUCCESS in 1h 30m 28s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/3482a27d7d964e7686bb7698e9c6c34f : SUCCESS in 1h 40m 36s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/51f9823ac2324333a30031ed71a4db45 : FAILURE in 1h 29m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/78904a3e9cc24d1082b9aa5bd08344ed : FAILURE in 1h 49m 19s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/f3710f62b0414324a3b338d2930147e8 : FAILURE in 1h 44m 28s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/bf44059f36634ae2b04a844c26586c64 : SUCCESS in 34m 10s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/1fab603f3ac64dfeaeb097de86364089 : FAILURE in 1h 33m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/363175a627e94bc9b4de0045ddc8aacb : FAILURE in 1h 38m 08s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/09bd30f52bff40c6839116c949a956f6 : FAILURE in 2h 21m 21s\n- neutron-tempest-plugin-designate-scenario-2026-1 https://zuul.opendev.org/t/openstack/build/218a170a7bef4bb48b6ad3b723a11cb4 : SUCCESS in 45m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1383d0809263486c8bd9641c700a8bf5 : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36da79f8bbea41cbb5b92fb93236def2 : SUCCESS in 4m 11s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/13955ae7f05e49aca05f9245b71f0c12 : SUCCESS in 1h 13m 59s\n- neutron-tempest-plugin-sfc-2025-1 https://zuul.opendev.org/t/openstack/build/67d278d52f80464b9a8ce2229e9ed858 : SUCCESS in 1h 53m 48s\n- neutron-tempest-plugin-sfc-2025-2 https://zuul.opendev.org/t/openstack/build/ad47a0fde78b4014b6fa0e2fbe299141 : SUCCESS in 1h 26m 40s\n- neutron-tempest-plugin-sfc-2026-1 https://zuul.opendev.org/t/openstack/build/0253b5df502742bf8f418ba68a3a88ea : SUCCESS in 1h 28m 58s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d702a04b65c3436ca83ed86a46b153b2 : SUCCESS in 1h 09m 15s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-1 https://zuul.opendev.org/t/openstack/build/a020ca70747945419ae02ed1adcce58a : SUCCESS in 1h 44m 47s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-2 https://zuul.opendev.org/t/openstack/build/214d875292d847b99710968acae59ffb : SUCCESS in 1h 38m 16s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2026-1 https://zuul.opendev.org/t/openstack/build/89ab5046d1bf4cbb808474a7e27d9f57 : SUCCESS in 1h 44m 40s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/e072227418d640dfb49c373fe0fabca3 : SUCCESS in 55m 03s\n- neutron-tempest-plugin-dynamic-routing-2025-1 https://zuul.opendev.org/t/openstack/build/79c5c8d23cc44dde8df5bbaf1d389961 : SUCCESS in 57m 23s\n- neutron-tempest-plugin-dynamic-routing-2025-2 https://zuul.opendev.org/t/openstack/build/170a9c0670894e79acc0426a9702ea57 : SUCCESS in 44m 58s\n- neutron-tempest-plugin-dynamic-routing-2026-1 https://zuul.opendev.org/t/openstack/build/0de27342151d4ea2af1b7727906996ed : SUCCESS in 43m 43s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/b9f4bbd05f844970975bbc38aa4cc6d1 : SUCCESS in 34m 26s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/43430dd2bd90404dba7ab6ce4bafe57a : SUCCESS in 33m 46s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/78a2f6fbcc164c74bb8427d3e79a1931 : SUCCESS in 42m 04s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/b022c130c7564c08bac9aa5c96f832f7 : SUCCESS in 26m 38s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/ef6f436278ee4ecbb36d9418dffe9ae4 : SUCCESS in 59m 12s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/54ac1195eda14fa9b7db2cb598fce675 : SUCCESS in 48m 08s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/d5a2fafcf767425bb4785d3709de68b7 : SUCCESS in 56m 59s\n- neutron-tempest-plugin-vpnaas-2025-1 https://zuul.opendev.org/t/openstack/build/0eb0d820072e418588830e97646e3aa2 : SUCCESS in 44m 09s\n- neutron-tempest-plugin-vpnaas-2025-2 https://zuul.opendev.org/t/openstack/build/0a53f59c27054d65aad5b92440bc78fc : SUCCESS in 34m 40s\n- neutron-tempest-plugin-vpnaas-2026-1 https://zuul.opendev.org/t/openstack/build/f79ec7eb04c0497389c53e1b8a98dd62 : SUCCESS in 23m 25s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/4aad041ed5074ee68980ba67c351cb19 : SUCCESS in 1h 04m 23s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/e9e263dfac2c4cf1a01a4d236a373ff0 : SUCCESS in 1h 17m 12s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/b2e8b4c8095847b1bc5d90d96e1d870b : SUCCESS in 1h 39m 02s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/3b6c06cec42149ecb31e9a2d830eb63d : SUCCESS in 1h 34m 13s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/1e16fc4c044441fb9231886feb7e009e : SUCCESS in 1h 13m 07s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/008d6e04542d4772a2971b38ef1d2a9a : SUCCESS in 1h 40m 25s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/909e24b74bb444c6a75b19e2d21afd7f : SUCCESS in 1h 13m 33s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/0f8141a5a21f4881bad618488c538ef7 : SUCCESS in 1h 20m 42s","accounts_in_message":[],"_revision_number":5},{"id":"5395eec5cf1d46d26d3b3e06eb63f6fad6accbbd","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-24 19:14:03.000000000","message":"Patch Set 5: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/af6543f40b79447682627c9d3bfe3da0\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/016702fbf00948e68b7d0ab5bd85031b : SUCCESS in 4h 19m 44s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/bec7aa67da2e41f49f1852d81c95dbf6 : FAILURE in 4h 04m 09s","accounts_in_message":[],"_revision_number":5},{"id":"bfe640843c8c2d02188a0d2c830b43fdb2802441","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-08-25 06:07:44.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"37a197e2c41e1ad9e40f15125a253f7b5a71a95a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-08-25 07:00:52.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"00b38ae5d0910e7054531bca6dd43fb0078283ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 09:03:32.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/258599b4ef114cb69d7f72541f43a505\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58ab6cad4c924a3a8b0798e4fbeb86f1 : SUCCESS in 3m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6f87ed74955e44d591d6effc483081fb : SUCCESS in 1h 02m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f7f310e8aeaf4157a5bc080fe9840435 : FAILURE in 1h 33m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/23b344fbfb854fb0be8a5de66ec7a322 : SUCCESS in 30m 09s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/624eae8b5fb140d9a07edd4a9467b2d9 : SUCCESS in 1h 18m 59s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/7853c0eb38f24fd38ec2548eede96c89 : SUCCESS in 1h 42m 22s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/2f77381d08464aeeaadf03151e23613a : FAILURE in 2h 00m 18s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/05a8b93524dc40849a5eab17949cd8ff : SUCCESS in 30m 38s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/75c1fcec63ef4927a616071e20c56056 : SUCCESS in 1h 24m 21s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/fbb5ea4143b74f4f9f17bbf2bc64bb81 : SUCCESS in 1h 20m 32s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/2c7f69a736b24bc58aeaa45c3c8fe514 : SUCCESS in 1h 28m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/7c4720241dcc40fa85e1a187d0b4148a : SUCCESS in 1h 48m 03s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/acee3dd7f2304e78a98c8d0c093cb350 : FAILURE in 1h 45m 59s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/5219378f072d4354aea570c793f7a837 : SUCCESS in 30m 06s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/ff79ecbd4a5646edabd7f804125629d4 : SUCCESS in 1h 34m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/0f8c01eebf5f4e3082cea4ff7150726e : SUCCESS in 1h 47m 00s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/77b594a199e846bf9c496ad93b5bd9bf : FAILURE in 1h 56m 42s\n- neutron-tempest-plugin-designate-scenario-2026-1 https://zuul.opendev.org/t/openstack/build/6cff9ec55f254aeb836afd8a9cf5ab32 : SUCCESS in 40m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/970598278ed54a479e32445c934f6e42 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9a83125524a42e285787e5e31a0d99b : SUCCESS in 3m 23s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/d60113c7901e4c39817af20335601b36 : SUCCESS in 1h 36m 05s\n- neutron-tempest-plugin-sfc-2025-1 https://zuul.opendev.org/t/openstack/build/83dc556356a84e2197159ece2cada033 : SUCCESS in 1h 47m 12s\n- neutron-tempest-plugin-sfc-2025-2 https://zuul.opendev.org/t/openstack/build/5d937153978d49a5b2e7493a60261051 : SUCCESS in 1h 00m 58s\n- neutron-tempest-plugin-sfc-2026-1 https://zuul.opendev.org/t/openstack/build/1d8d4ff0d2b54f97a015a2cade7fd50b : SUCCESS in 1h 55m 51s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/0ec834531e8f4b80a1af80a7ca5f5e98 : SUCCESS in 39m 10s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-1 https://zuul.opendev.org/t/openstack/build/2654f645dffa4fee9ad9d4cda4a7fce2 : SUCCESS in 1h 55m 50s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-2 https://zuul.opendev.org/t/openstack/build/872660813c0f4382bb7a827f47bf732d : SUCCESS in 57m 45s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2026-1 https://zuul.opendev.org/t/openstack/build/6075dd652fa048e498fe5429651c776c : SUCCESS in 1h 01m 07s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/8a3cfd781da14d04a4e2b7bed3c32819 : SUCCESS in 54m 27s\n- neutron-tempest-plugin-dynamic-routing-2025-1 https://zuul.opendev.org/t/openstack/build/4b04abdbc8ba4996afec1c3d96e025b3 : SUCCESS in 52m 28s\n- neutron-tempest-plugin-dynamic-routing-2025-2 https://zuul.opendev.org/t/openstack/build/344444476b7045db9ae6fb4f1a3840bb : SUCCESS in 28m 17s\n- neutron-tempest-plugin-dynamic-routing-2026-1 https://zuul.opendev.org/t/openstack/build/58969661ef08416091efc10156a4d36e : SUCCESS in 26m 39s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/ca16cb5b053d4b3784fd098b9aa7144f : SUCCESS in 29m 36s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/c98e9606c6164bfe844071ec3dce0e32 : SUCCESS in 29m 08s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/cc16db83c103480c80be1b9ad4c30aa0 : SUCCESS in 23m 11s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/8c188ad5095447fd820e0830aa649a9e : SUCCESS in 52m 27s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/8f11aa5ff041440c98f35fa2406c9c55 : SUCCESS in 51m 27s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/e99565a06bbc40508e82d11bd81f86c9 : SUCCESS in 33m 04s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/a501a41a7be94d5ea5a84706eb4dfca6 : SUCCESS in 53m 18s\n- neutron-tempest-plugin-vpnaas-2025-1 https://zuul.opendev.org/t/openstack/build/9fa2a34ecbe54f4d83a6052bac294049 : SUCCESS in 23m 11s\n- neutron-tempest-plugin-vpnaas-2025-2 https://zuul.opendev.org/t/openstack/build/552db03335f6435282dd6aec65af5d08 : SUCCESS in 52m 21s\n- neutron-tempest-plugin-vpnaas-2026-1 https://zuul.opendev.org/t/openstack/build/50734a631fe6424ab80d53741d0faf37 : SUCCESS in 50m 28s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/689d1166e05748b2990b25a20e37f97b : SUCCESS in 1h 27m 10s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/2d3fa26b04ba4625b7ca0ca3f0eac5d2 : SUCCESS in 1h 18m 10s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/0bd23380ee5044f5b65d623ef350a604 : SUCCESS in 1h 31m 42s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/124b17a181c142d6b7a4a4d2826c1f8b : SUCCESS in 1h 31m 29s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/c7fb3f6a690a4273bf6743fc9d953626 : SUCCESS in 1h 02m 15s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/f609adc4177b43daa6d6f5f57bbdfe9b : SUCCESS in 1h 28m 45s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/3d80a3183477414783edc8d73bee57dc : SUCCESS in 1h 07m 12s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/2234492609f64d25a30aa4c6ae705dd6 : SUCCESS in 1h 23m 58s","accounts_in_message":[],"_revision_number":6},{"id":"7db60f69999e6fee497ab4de70afaf0cf9f36e24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-08-25 09:32:19.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"e61981d5824d3f9249abc0c0ea667af46bae0308","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-08-25 11:06:07.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a20766eedf7c559e26dfc537236f6cb135710364","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-08-25 12:47:33.000000000","message":"Patch Set 8: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0160e16a3446dc795de9bac3469c1d1a6cbfa1f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 13:29:15.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46e3d4e64a174d8ebd28e84b84366a99\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7e844727ff74fbb9caee1a9a80b2e93 : SUCCESS in 4m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0ce085df9c4e4502b1c4d5ec730c6d4e : SUCCESS in 59m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e83ff49c728946edb637a0db3d913201 : SUCCESS in 1h 56m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3121f7a17bbd45c4aa18f6d40a691783 : SUCCESS in 30m 12s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/5602e7339f1144dba64668084729aee0 : SUCCESS in 1h 41m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/7000fe4915b848a4ba725a419877c260 : SUCCESS in 2h 14m 38s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/7a753ccb47564ce3a9bea0f5d93d3f4b : SUCCESS in 2h 06m 58s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/2559a00909a64e67b3e7f028064f554e : SUCCESS in 43m 26s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/cb94f4d21c1b4c5e95c85d3c7ade7c9c : SUCCESS in 1h 29m 40s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/c7ea60eb48454f32b156ed16b0e8aa66 : SUCCESS in 1h 20m 05s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/bded4f9abe364046a6fa3931949e5ad1 : SUCCESS in 1h 22m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/d1f558b97d9e473bba62494d2a9a628e : SUCCESS in 2h 00m 28s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/8c76cf77b07c48e6844d8faa7e3d47dc : SUCCESS in 2h 17m 41s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/c32bf8f28ddd46c499b4b63e6effa97e : SUCCESS in 33m 14s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/6c1145df27ff450fab442a5de406f8de : SUCCESS in 1h 24m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/ef5c6ad708884362998c9cdf1ac3f4d9 : SUCCESS in 1h 46m 09s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/654c3e2311fc454a9a4e08a65704b0cf : SUCCESS in 2h 02m 00s\n- neutron-tempest-plugin-designate-scenario-2026-1 https://zuul.opendev.org/t/openstack/build/08863a924cb9431ab6816c566bce5a66 : SUCCESS in 33m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/308b412dc9f04f87b6143fbc3da94bc8 : SUCCESS in 3m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd5252e9136a40259adee4ab2554e462 : SUCCESS in 4m 16s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/f4356bac4fa645b7b21a3fc410312ab6 : SUCCESS in 1h 58m 07s\n- neutron-tempest-plugin-sfc-2025-1 https://zuul.opendev.org/t/openstack/build/13f9d1c383ee4981ba9a3c155bfcf512 : SUCCESS in 1h 41m 32s\n- neutron-tempest-plugin-sfc-2025-2 https://zuul.opendev.org/t/openstack/build/4b841c261c88479188efe4f08f49c799 : SUCCESS in 1h 49m 26s\n- neutron-tempest-plugin-sfc-2026-1 https://zuul.opendev.org/t/openstack/build/0caf6429a5564ba3a9773e28fb464597 : SUCCESS in 1h 09m 52s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7997e0aa26804a129e559cedf7248ef8 : SUCCESS in 1h 28m 50s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-1 https://zuul.opendev.org/t/openstack/build/15b23ab0eda248c1a7755a0e35d17925 : SUCCESS in 1h 30m 51s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-2 https://zuul.opendev.org/t/openstack/build/a91475dc2d0a4314972c804451226fba : SUCCESS in 1h 33m 31s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2026-1 https://zuul.opendev.org/t/openstack/build/40f671a8040b4575b45dbdccbfebd831 : SUCCESS in 1h 34m 30s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/4af457186496468796b82ef87b2db134 : SUCCESS in 44m 23s\n- neutron-tempest-plugin-dynamic-routing-2025-1 https://zuul.opendev.org/t/openstack/build/996bcb8db2c94cd3a7064fb6f54f835f : SUCCESS in 55m 39s\n- neutron-tempest-plugin-dynamic-routing-2025-2 https://zuul.opendev.org/t/openstack/build/677030d2094f4cf085678c77bdb615c8 : SUCCESS in 56m 59s\n- neutron-tempest-plugin-dynamic-routing-2026-1 https://zuul.opendev.org/t/openstack/build/fdc2000f322345e3b3c5ea20c15c488f : SUCCESS in 58m 55s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/a6733cf8cbda4b10a7293db8598d930c : SUCCESS in 29m 56s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/cca1bb5c018b4e4fab1a2ce4a88c09bf : SUCCESS in 29m 15s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/c0a9a4511ccd4f2093f241673a31f389 : SUCCESS in 37m 23s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/c6845eedf9ff4a4ca88013fee35fc04b : SUCCESS in 35m 32s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/aacd99dbaea64887ac099117c48215d6 : SUCCESS in 1h 05m 14s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/fc76d3a379da48e99eb93b393f95a96b : SUCCESS in 52m 38s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/e4e4a76bc7534506b825b388742a1669 : SUCCESS in 1h 00m 26s\n- neutron-tempest-plugin-vpnaas-2025-1 https://zuul.opendev.org/t/openstack/build/aee69cc0c7f5430ebe222c2299e04008 : SUCCESS in 55m 47s\n- neutron-tempest-plugin-vpnaas-2025-2 https://zuul.opendev.org/t/openstack/build/88ddd0719cb64dcda06abc14b89e5a47 : SUCCESS in 49m 59s\n- neutron-tempest-plugin-vpnaas-2026-1 https://zuul.opendev.org/t/openstack/build/a2d5bf175eea4ddb89dfdd4128b7b497 : SUCCESS in 51m 38s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/4a19fe225f0d4f0d86db3f50189852e2 : SUCCESS in 1h 40m 01s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/83002728a9764d108237ea24a05d62cf : SUCCESS in 1h 16m 49s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/dcac7004eff4416e8a27f64a05404ea2 : SUCCESS in 1h 42m 59s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/b9075571d5fa446da9752aedf358a2ff : SUCCESS in 1h 46m 08s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/905722367ac345419f2f3c81b047fbdf : SUCCESS in 1h 42m 27s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/51d9837a5f5347f8bdd73a85789c94d2 : SUCCESS in 1h 32m 21s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/b1b2780ac5d34eedaba4115e0d180d23 : SUCCESS in 1h 21m 46s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/9ce5331c281f4a3386aa4efa0ca772a6 : SUCCESS in 1h 02m 44s","accounts_in_message":[],"_revision_number":8},{"id":"d95f708520b4ecfe16bfa6338fa61ad5d9b70124","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-08-25 14:49:33.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"0e3a6abf4c492e9f3fdcaaf1a9693cb35086532a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 14:50:15.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"f54977a1749c3fdabcb0d1c2bf7a892d0d0361df","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-25 15:35:21.000000000","message":"Patch Set 8: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/32068c0620454062a9354d677532276c\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/e761aa8b7efd4b8da55ae664457e3f4c : SUCCESS in 4h 09m 48s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/af5181df08704ce9a3c5ce129a96bf3f : FAILURE in 3h 54m 49s","accounts_in_message":[],"_revision_number":8},{"id":"a866614a01e9675753dc88afbb91dcdd3771f892","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 17:37:04.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7c50199ab814165b1ce65f70c719d85\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a000dcd221ce455cb93eb279cb8c0255 : SUCCESS in 4m 50s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ba4d9dbcc3b940d58623cdf06ca69121 : SUCCESS in 1h 00m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/590495eaedc7434e8c01c378a60bfe2b : SUCCESS in 2h 03m 21s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/4ed547108473416a861dce4cb3af35cc : SUCCESS in 1h 48m 19s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/0fd5a90249c247dd90f8ccfca9ed0439 : SUCCESS in 1h 48m 57s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/28cd3c80eada427aa695999ed86cc661 : FAILURE in 2h 26m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49cf4bde51c244ed9a9ec82cd319bee6 : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5fe924ade4e4d3ebd8003c756489abe : SUCCESS in 4m 31s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/41becd2a232b4335a14a1e8e8f3c79e0 : SUCCESS in 1h 45m 21s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/ba0cf7845af643e1a6098774b16a035a : SUCCESS in 1h 26m 17s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/359d0d67cd7445e3bc11e2a7fde43e6b : SUCCESS in 58m 34s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/476c1e345d7f4c8ca0a3b738160f8c07 : SUCCESS in 32m 24s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/9235bde84c064a6d91c25cbdc03d0fba : SUCCESS in 52m 56s","accounts_in_message":[],"_revision_number":8},{"id":"d5700cf9147e84189585e88a93d30df868748999","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-08-25 19:42:31.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4ae6f9dbb004d356fdd9afda6933fd07f613d0f2","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-08-25 19:42:36.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"7d85a8fc8d9134620a940db7322a96efe6db9d27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 21:50:47.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac7f4689a309414ba5a307c14cf100e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e23ce6b6a8334eceb919058918e8ffc8 : SUCCESS in 3m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/15e7c0ee38c74cd099c3a3aaf968073e : SUCCESS in 1h 06m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3f1293a4455b4f81bacf1f76c4457920 : SUCCESS in 2h 00m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/02600a6b0fd941d9b46f95b4060c0847 : SUCCESS in 31m 38s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/41ca5debe8ff4c2f8393266ad8a95428 : SUCCESS in 1h 35m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/c8015e1b23934e99971e4c0732613646 : SUCCESS in 1h 32m 25s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/0b18c34ba73944859b66e874b90a7265 : SUCCESS in 2h 05m 44s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/4c6ef4c3e26d480f8c9ff1d18b6bfe46 : SUCCESS in 33m 56s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/77b13e7b5f6d455ca29f8beb73897b13 : SUCCESS in 1h 24m 46s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/0a2b0687ffc84fae9af9b9f159ad4cf3 : SUCCESS in 1h 11m 13s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/68c334bbaba44857966fdc65e1d1f70d : SUCCESS in 1h 48m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/43a7379d30d34148b8ef97271570541e : SUCCESS in 1h 37m 19s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/aeab27aa88f04f019e3296568b016987 : SUCCESS in 1h 56m 06s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/31f363c91d4043bb8e15b01eb18914ca : SUCCESS in 39m 42s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/cf2666e41bab402188732100c851a300 : SUCCESS in 1h 48m 15s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/0ccc80dc82a946aa997f2a24ad215a70 : SUCCESS in 1h 48m 43s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/c00860a5cf074f28868795de38a36da4 : SUCCESS in 1h 39m 31s\n- neutron-tempest-plugin-designate-scenario-2026-1 https://zuul.opendev.org/t/openstack/build/b54d7091a05f469a9ff47d4edf6dfacc : SUCCESS in 33m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8b6e69743694c869261ca7e6b7bf37a : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2f9f3a68fe24f3894826a0abf5ed397 : SUCCESS in 5m 06s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/4f8ee60aeb274d75bf5f294ea1e3e25f : SUCCESS in 1h 37m 56s\n- neutron-tempest-plugin-sfc-2025-1 https://zuul.opendev.org/t/openstack/build/16d1d503452440459ac1ae516cd2f3fe : SUCCESS in 1h 53m 08s\n- neutron-tempest-plugin-sfc-2025-2 https://zuul.opendev.org/t/openstack/build/a3c0458f033f4a17a7aa3263bf548280 : SUCCESS in 1h 14m 22s\n- neutron-tempest-plugin-sfc-2026-1 https://zuul.opendev.org/t/openstack/build/05cd44bc055a40cd990e2ae3da893d8e : SUCCESS in 1h 42m 36s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/6bb32b77423e4ab6a2bc4c110f6c02da : SUCCESS in 53m 44s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-1 https://zuul.opendev.org/t/openstack/build/3cceba39f3764b819f1e37baaaccb699 : SUCCESS in 1h 52m 06s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2025-2 https://zuul.opendev.org/t/openstack/build/c9b953707e0d4179957b4004eac838c2 : SUCCESS in 1h 50m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2026-1 https://zuul.opendev.org/t/openstack/build/84b9d44a3ca24addb2b52566ca328496 : SUCCESS in 1h 14m 20s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/bb1315a2bf08492d9ad6c0502d29c0e1 : SUCCESS in 45m 00s\n- neutron-tempest-plugin-dynamic-routing-2025-1 https://zuul.opendev.org/t/openstack/build/371c7ed7cc0b4fce90e1f9bb5f667804 : SUCCESS in 38m 09s\n- neutron-tempest-plugin-dynamic-routing-2025-2 https://zuul.opendev.org/t/openstack/build/4a4b2dba754c4edabd43facaa5b77be3 : SUCCESS in 1h 05m 34s\n- neutron-tempest-plugin-dynamic-routing-2026-1 https://zuul.opendev.org/t/openstack/build/baae0cce15ef45bd8e1b93ef1edf2d1f : SUCCESS in 1h 00m 38s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/f856ffbac10c49479411d05ccaa071e2 : SUCCESS in 40m 12s\n- neutron-tempest-plugin-fwaas-openvswitch https://zuul.opendev.org/t/openstack/build/7223be40fa4f447cb499b3b3849e67ec : SUCCESS in 30m 20s\n- neutron-tempest-plugin-fwaas-2025-1 https://zuul.opendev.org/t/openstack/build/f90171949cb6494f91569061e6c2d556 : SUCCESS in 49m 48s\n- neutron-tempest-plugin-fwaas-2025-2 https://zuul.opendev.org/t/openstack/build/f9529b4a7cc14a85a7081424aa0e4fdc : SUCCESS in 58m 05s\n- neutron-tempest-plugin-fwaas-2026-1 https://zuul.opendev.org/t/openstack/build/26f9d92d627045a7a291e59c4beca2a8 : SUCCESS in 49m 05s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/76dcb698889540698958af5a2aa7fb22 : SUCCESS in 28m 11s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/82e4174f84b44799947a2da44e80ae25 : SUCCESS in 54m 08s\n- neutron-tempest-plugin-vpnaas-2025-1 https://zuul.opendev.org/t/openstack/build/fa06ee5a3cda488a940296efdb18eaee : SUCCESS in 27m 59s\n- neutron-tempest-plugin-vpnaas-2025-2 https://zuul.opendev.org/t/openstack/build/75a3a74cd071451093ce70396d8d066f : SUCCESS in 55m 55s\n- neutron-tempest-plugin-vpnaas-2026-1 https://zuul.opendev.org/t/openstack/build/5867ccc59d2b4e5c95698641b5f03292 : SUCCESS in 28m 22s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/6005de5f768c4709a1f8d1845cac889d : SUCCESS in 1h 07m 37s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/8b1e9504696e4810aef366be5a4c29f2 : SUCCESS in 40m 08s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/55138865dd064e4f88b6847518a93a17 : SUCCESS in 1h 33m 25s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/4919e9c563a4476f9a934c973279ccb3 : SUCCESS in 1h 15m 21s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/e3d661e966654f8b8a91b5e68e54cf88 : SUCCESS in 1h 27m 47s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/33000c81c2794b4c9842ead671b2f992 : SUCCESS in 45m 55s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/88d692c6d42b409e9ef44c57febdf7de : SUCCESS in 1h 07m 17s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/e91501ae0db1428084bea31adbed15a8 : SUCCESS in 46m 39s","accounts_in_message":[],"_revision_number":8},{"id":"273559419643b5c61ac7248d3773eece8f1270fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 21:51:24.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"840efa5162dae4bded50134609dc0eedef04174a","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-25 23:45:11.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/ed6653b6aaa6454a9abbf823d836494c\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/30bd8b8c956d4a8582a75847f09e630b : SUCCESS in 4h 01m 21s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/2f8bf5a00e1b45ef83c4acbe3cb6b8d6 : FAILURE in 3h 46m 59s","accounts_in_message":[],"_revision_number":8},{"id":"15f22f355aa625abae832fdd0bdc5191c0f05e7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 00:04:47.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/509ba27ffb4147408483bcf4bea1dfcb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79c3ab43b85a4227bda9de8516711f3e : SUCCESS in 5m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1cb4fd97960f4df0b6e8729b15553ca4 : SUCCESS in 59m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b12cace6ecb74885bd351489cdc7657a : SUCCESS in 1h 43m 05s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/f59966847fbe467984f5349b8e9311ff : SUCCESS in 1h 52m 23s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/b8766fb13f85404dab0e43e9afc058d1 : SUCCESS in 1h 50m 38s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/8eafbab3cf724793a555d22c308806b7 : SUCCESS in 2h 12m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4ce165e6d2342e28be0ee861ef96d46 : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/334e96c36c1941cb9d4c3688a7a09417 : SUCCESS in 3m 55s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/38a92963c3ea4eeeaf721cec80448315 : SUCCESS in 46m 13s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/0dee588f38c1402dbeb52db64944dcaf : SUCCESS in 1h 12m 46s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/b8f3b256f4a847aba23ea86d9d054e7a : SUCCESS in 1h 02m 11s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/69ad908003374a2ba8e88cbdab88b410 : SUCCESS in 33m 32s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/aa346ad6114b40608d89f6a9f081b7c3 : SUCCESS in 25m 15s","accounts_in_message":[],"_revision_number":8},{"id":"2a1509ea656e128642c485eaa90a7be605843a37","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 00:04:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"5431c1424afed6f4b55eb69b8d9f2e5e5be76216","revisions":{"01d25dcee5c76f7fd864cfa0fcf421b369dd9ca5":{"kind":"REWORK","_number":1,"created":"2026-08-14 14:11:45.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/68/1000968/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/68/1000968/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/1"}}},"commit":{"parents":[{"commit":"69c70a6c08d165d3a7e1a4f4d51d2efdf1e84107","subject":"Merge \"Add PVLAN cold migration scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/69c70a6c08d165d3a7e1a4f4d51d2efdf1e84107"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-08-14 14:09:59.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-08-14 14:11:38.000000000","tz":120},"subject":"[evpn] Fix router cleanup","message":"[evpn] Fix router cleanup\n\nThe created router was not properly cleaned up during teardown because\nit was created by the admin user. Using the _create_admin_router from\nthe BaseRouterTest class fixes this issue.\n\nChange-Id: I6bd79e7bfbc5e496009fed53dfe8d20f5f3af748\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/01d25dcee5c76f7fd864cfa0fcf421b369dd9ca5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/01d25dcee5c76f7fd864cfa0fcf421b369dd9ca5"}]},"branch":"refs/heads/master"},"a6b4631d6bd384d39209cad4da755e2542177611":{"kind":"REWORK","_number":2,"created":"2026-08-14 14:15:01.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/68/1000968/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/68/1000968/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/2"}}},"commit":{"parents":[{"commit":"69c70a6c08d165d3a7e1a4f4d51d2efdf1e84107","subject":"Merge \"Add PVLAN cold migration scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/69c70a6c08d165d3a7e1a4f4d51d2efdf1e84107"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-08-14 14:09:59.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-08-14 14:14:25.000000000","tz":120},"subject":"[evpn] Fix router cleanup","message":"[evpn] Fix router cleanup\n\nThe created router was not properly cleaned up during teardown because\nit was created by the admin user. Using the _create_admin_router from\nthe BaseRouterTest class fixes this issue.\n\nChange-Id: I6bd79e7bfbc5e496009fed53dfe8d20f5f3af748\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a6b4631d6bd384d39209cad4da755e2542177611"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a6b4631d6bd384d39209cad4da755e2542177611"}]},"branch":"refs/heads/master"},"7524d173171facd537bcb4f73d02180c8700ecac":{"kind":"REWORK","_number":3,"created":"2026-08-14 14:26:04.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/68/1000968/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/68/1000968/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/3"}}},"commit":{"parents":[{"commit":"69c70a6c08d165d3a7e1a4f4d51d2efdf1e84107","subject":"Merge \"Add PVLAN cold migration scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/69c70a6c08d165d3a7e1a4f4d51d2efdf1e84107"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-08-14 14:25:22.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-08-14 14:25:22.000000000","tz":120},"subject":"[evpn] Fix router cleanup","message":"[evpn] Fix router cleanup\n\nThe created router was not properly cleaned up during teardown because\nit was created by the admin user.\n\nChange-Id: I6bd79e7bfbc5e496009fed53dfe8d20f5f3af748\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7524d173171facd537bcb4f73d02180c8700ecac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7524d173171facd537bcb4f73d02180c8700ecac"}]},"branch":"refs/heads/master"},"4dc1d2e9f1a7ec50d8c08fc9d8e3e368e133dfce":{"kind":"REWORK","_number":4,"created":"2026-08-24 11:03:44.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/68/1000968/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/68/1000968/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/4"}}},"commit":{"parents":[{"commit":"460ac52935e08dc9391c91cf231da2fb94100e89","subject":"[CI][2025.1] Skip broken network segment range test in 2025.1 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/460ac52935e08dc9391c91cf231da2fb94100e89"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-08-14 14:25:22.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-08-24 11:03:32.000000000","tz":120},"subject":"[evpn] Fix router cleanup","message":"[evpn] Fix router cleanup\n\nThe created router was not properly cleaned up during teardown because\nit was created by the admin user.\n\nChange-Id: I6bd79e7bfbc5e496009fed53dfe8d20f5f3af748\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4dc1d2e9f1a7ec50d8c08fc9d8e3e368e133dfce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4dc1d2e9f1a7ec50d8c08fc9d8e3e368e133dfce"}]},"branch":"refs/heads/master"},"814862a8fd82d0a0aef9018a262bde5939e027c1":{"kind":"REWORK","_number":5,"created":"2026-08-24 14:52:45.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/68/1000968/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/68/1000968/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/5"}}},"commit":{"parents":[{"commit":"460ac52935e08dc9391c91cf231da2fb94100e89","subject":"[CI][2025.1] Skip broken network segment range test in 2025.1 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/460ac52935e08dc9391c91cf231da2fb94100e89"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-08-14 14:25:22.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-08-24 14:52:32.000000000","tz":120},"subject":"Fix admin router cleanup","message":"Fix admin router cleanup\n\nRouters created with create_admin_router were cleaned up using non-admin\ncredentials, which doesn\u0027t work.\nWith this patch non-admin routers are deleted using non-admin client\nand admin routers are deleted with admin client.\n\nChange-Id: I6bd79e7bfbc5e496009fed53dfe8d20f5f3af748\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/814862a8fd82d0a0aef9018a262bde5939e027c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/814862a8fd82d0a0aef9018a262bde5939e027c1"}]},"branch":"refs/heads/master"},"3bbd558cd69ef2a6d0e211c11e0523c61db7690c":{"kind":"REWORK","_number":6,"created":"2026-08-25 07:00:52.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/68/1000968/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/68/1000968/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/6"}}},"commit":{"parents":[{"commit":"460ac52935e08dc9391c91cf231da2fb94100e89","subject":"[CI][2025.1] Skip broken network segment range test in 2025.1 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/460ac52935e08dc9391c91cf231da2fb94100e89"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-08-14 14:25:22.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-08-25 06:55:20.000000000","tz":120},"subject":"Fix admin router cleanup and remove BaseRouterTest class","message":"Fix admin router cleanup and remove BaseRouterTest class\n\nRouters created with create_admin_router were cleaned up using non-admin\ncredentials, which doesn\u0027t work.\nWith this patch non-admin routers are deleted using non-admin client\nand admin routers are deleted with admin client.\n\nOnce that issue has been fixed on BaseAdminNetworkTest class,\nBaseRouterTest class is not needed at all because routers cleanup\nhappens when resouce_cleanup is called. Only its method _delete_router,\nwhich is slightly different from BaseNetworkTest.delete_router, has been\nadded to RoutersTestHA class.\n\nChange-Id: I6bd79e7bfbc5e496009fed53dfe8d20f5f3af748\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3bbd558cd69ef2a6d0e211c11e0523c61db7690c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3bbd558cd69ef2a6d0e211c11e0523c61db7690c"}]},"branch":"refs/heads/master"},"5f20f1971417432c1d1f8b9b072bdae9ef056631":{"kind":"REWORK","_number":7,"created":"2026-08-25 09:32:19.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/68/1000968/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/68/1000968/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/7"}}},"commit":{"parents":[{"commit":"460ac52935e08dc9391c91cf231da2fb94100e89","subject":"[CI][2025.1] Skip broken network segment range test in 2025.1 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/460ac52935e08dc9391c91cf231da2fb94100e89"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-08-14 14:25:22.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-08-25 09:32:10.000000000","tz":120},"subject":"Fix admin router cleanup and remove BaseRouterTest class","message":"Fix admin router cleanup and remove BaseRouterTest class\n\nRouters created with create_admin_router were cleaned up using non-admin\ncredentials, which doesn\u0027t work.\nWith this patch non-admin routers are deleted using non-admin client\nand admin routers are deleted with admin client.\n\nOnce that issue has been fixed on BaseAdminNetworkTest class,\nBaseRouterTest class is not needed at all because routers cleanup\nhappens when resouce_cleanup is called. Only its method _delete_router,\nwhich is slightly different from BaseNetworkTest.delete_router, has been\nadded to RoutersTestHA class.\n\nChange-Id: I6bd79e7bfbc5e496009fed53dfe8d20f5f3af748\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5f20f1971417432c1d1f8b9b072bdae9ef056631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5f20f1971417432c1d1f8b9b072bdae9ef056631"}]},"branch":"refs/heads/master"},"5431c1424afed6f4b55eb69b8d9f2e5e5be76216":{"kind":"REWORK","_number":8,"created":"2026-08-25 11:06:07.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/68/1000968/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/68/1000968/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/1000968/8"}}},"commit":{"parents":[{"commit":"460ac52935e08dc9391c91cf231da2fb94100e89","subject":"[CI][2025.1] Skip broken network segment range test in 2025.1 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/460ac52935e08dc9391c91cf231da2fb94100e89"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-08-14 14:25:22.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-08-25 11:06:02.000000000","tz":120},"subject":"Fix admin router cleanup and remove BaseRouterTest class","message":"Fix admin router cleanup and remove BaseRouterTest class\n\nRouters created with create_admin_router were cleaned up using non-admin\ncredentials, which doesn\u0027t work.\nWith this patch non-admin routers are deleted using non-admin client\nand admin routers are deleted with admin client.\n\nOnce that issue has been fixed on BaseAdminNetworkTest class,\nBaseRouterTest class is not needed at all because routers cleanup\nhappens when resouce_cleanup is called. Only its method _delete_router,\nwhich is slightly different from BaseNetworkTest.delete_router, has been\nadded to RoutersTestHA class.\n\nChange-Id: I6bd79e7bfbc5e496009fed53dfe8d20f5f3af748\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5431c1424afed6f4b55eb69b8d9f2e5e5be76216"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5431c1424afed6f4b55eb69b8d9f2e5e5be76216"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
