)]}'
{"id":"openstack%2Fneutron-tempest-plugin~1003265","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I2d7f269a08a2146f68b38ca338458ed83d9792f0","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-09-09 12:22:42.000000000","reason":"Change was submitted"},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-09-02 13:35:48.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I2d7f269a08a2146f68b38ca338458ed83d9792f0","subject":"Add scenario test for the indirect Floating IPs API extension","status":"MERGED","created":"2026-09-01 12:21:16.000000000","updated":"2026-09-09 12:22:42.000000000","submitted":"2026-09-09 12:22:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003265","meta_rev_id":"c3af48980623d4e79a0a5b7849d995f54f120663","_number":1003265,"virtual_id_number":1003265,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-09 12:22:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2026-09-09 10:07:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2026-09-04 14:13:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2026-09-08 19:41:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2026-09-09 10:07:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-01 12:26:41.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-09-01 12:26:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-08 19:41:50.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-09-09 10:07:51.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"55bdf166d9fa671c37aa06c5ab6e3b4ba829ff2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-01 12:21:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0e9861812b90d75a453def4df3364475416394a","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-01 12:26:41.000000000","message":"Patch Set 1: Code-Review-1\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/1003264, https://review.opendev.org/c/openstack/neutron/+/1003262 are needed.","accounts_in_message":[],"_revision_number":1},{"id":"f5713fa6ba1e332f5dd14014f0939fc1c888ba72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 12:26:43.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/1003264, https://review.opendev.org/c/openstack/neutron/+/1003262 are needed.","accounts_in_message":[],"_revision_number":1},{"id":"de868506a59ed6be280566cd7b3106401a51eda0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-02 07:38:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b3fe4781f7515ef486650720f2a79ddea75d3ced","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-02 09:37:52.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/1003264, https://review.opendev.org/c/openstack/neutron/+/1003262 are needed.","accounts_in_message":[],"_revision_number":1},{"id":"2a5a5f60857a1f540c60bde3ecd7c855cd377fd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 09:37:53.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/1003264, https://review.opendev.org/c/openstack/neutron/+/1003262 are needed.","accounts_in_message":[],"_revision_number":1},{"id":"3cbca1546647b03a864cffb3701a796f78b34c59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-02 09:38:05.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"feb7a11e5f84ba746d9c997f927a173dc9b61d7e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-02 09:38:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2f51e9fda14677774cf4ae0c7acd22a018a910e3","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 10:26:46.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/1003264, https://review.opendev.org/c/openstack/neutron/+/1003262 are needed.","accounts_in_message":[],"_revision_number":2},{"id":"94e443cf637bc6a87b26b96dffb455d445e760a5","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-02 10:26:47.000000000","message":"Patch Set 2: Code-Review-1\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/1003264, https://review.opendev.org/c/openstack/neutron/+/1003262 are needed.","accounts_in_message":[],"_revision_number":2},{"id":"ed0fde07f1841f9023369349a28c0a7fd04adb57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 10:26:53.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/1003264, https://review.opendev.org/c/openstack/neutron/+/1003262 are needed.","accounts_in_message":[],"_revision_number":2},{"id":"da2fdb2b9d6ddeb2fc1fbaa6a7aac40c9fd6063e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-02 10:27:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"787481efb8a80d88853475a14300991bec3a7e27","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-02 10:27:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f7635919e8191b57a3cd28be3e61fbc6a09fb813","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 12:48:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7a2d9a853e84f998e7204345bcde5e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69efd26ef7a440afb7bf48a061c72afd : SUCCESS in 3m 33s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ae7da10bc1db4e8f80607c4a5a1072d5 : FAILURE in 1h 05m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b036949fb8fa4573bf55ea9a326ab814 : FAILURE in 1h 42m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e57f72992bba4fe78384f9d75e5539e8 : SUCCESS in 33m 36s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/94efec540375484a88b7a84f8f867910 : SUCCESS in 1h 34m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/60b5325a714e47188edf45ad37a0a6d2 : SUCCESS in 1h 56m 50s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/ce6527f367a24b9c9ef24231f5cf0811 : SUCCESS in 1h 53m 40s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/c62e4a01a4964f849aff0041cb249c59 : SUCCESS in 33m 17s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/3933010137174ef8aef5de92dc308843 : SUCCESS in 1h 29m 52s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/2f6621ccb5f54267a85b3ebc706005b6 : SUCCESS in 1h 07m 28s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/2ed5e163c4224994b1b5b4a049b5bcd0 : SUCCESS in 1h 40m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/ac7d5825d8f143b89c89d4de14af1303 : SUCCESS in 2h 13m 08s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/e2f1e97018a643ec8ff6b4b47db1912d : SUCCESS in 1h 46m 41s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/11b6943be3e64bf2ba53152bb447d61a : SUCCESS in 32m 55s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/17f2fcd54f5446d1a2134a91b271bdc5 : SUCCESS in 1h 22m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/f4b209c7de6f4ee9bf362ee7efb70c84 : SUCCESS in 1h 42m 59s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/70f952e8770640b8a26b267ff3e2ec1e : SUCCESS in 2h 19m 28s\n- neutron-tempest-plugin-designate-scenario-2026-1 https://zuul.opendev.org/t/openstack/build/462581f08037424a932479abdb4c5b50 : SUCCESS in 33m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/684d928cae99459c93baeef4c063f1bf : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb934010be7944bca2f68c47ad656632 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":2},{"id":"a49375e7fa89e141de1a5f6b3dcb05b71d17ebdb","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 13:33:18.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/170eb115cfaa4f099d165a04ea861f44\n\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/77c0d28a67794babbe4491e90006ed6b : FAILURE in 1h 26m 26s (non-voting)\n- neutron-tempest-plugin-openvswitch-distributed-dhcp https://zuul.opendev.org/t/openstack/build/81b475d33e2f4be38597c27078d17e1e : FAILURE in 1h 02m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-distributed-dhcp https://zuul.opendev.org/t/openstack/build/6f3c2d6a0bb04ac79cf08f2eed435442 : FAILURE in 1h 09m 36s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/3efec22870da4b42be3c483935f1b18d : FAILURE in 1h 45m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/37d6a1f258f64b149765dc86cafce6b3 : FAILURE in 1h 29m 55s\n- neutron-tempest-plugin-dvr-multinode-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/4fe97f2ef759472186e01b361499f246 : SUCCESS in 2h 59m 18s (non-voting)\n- neutron-tempest-plugin-dvr-multinode-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/6d17b6354e7a42be9c557aaecfce894d : SUCCESS in 2h 44m 01s (non-voting)\n- neutron-tempest-plugin-dvr-multinode-scenario-2026-1 https://zuul.opendev.org/t/openstack/build/22a51977b33b4a0e9b2cb579d10b024b : SUCCESS in 2h 56m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2dd528e66b9d0d133ebde2f5c4dda3115daa9937","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-02 13:35:17.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"f9be8871e5e3122eb6e0649c479f67941a673075","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-02 13:35:48.000000000","message":"Patch Set 3: Code-Review-1\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/1003264, https://review.opendev.org/c/openstack/neutron/+/1003262 are needed.","accounts_in_message":[],"_revision_number":3},{"id":"db63c68498310e97f4798fd41d4eecd182d67bd6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-02 13:36:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"91953bc21cd08e7730d889a8a4e55fdc25c96b38","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-02 13:37:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"49ed16e7aba7235ba859a87d5775cf2a4092947d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 13:38:43.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/1003264, https://review.opendev.org/c/openstack/neutron/+/1003262 are needed.","accounts_in_message":[],"_revision_number":3},{"id":"826ed5a586fd163c197de1d6370cf8ab4b00ef36","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-02 13:42:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3f44e5e448361b6c8d7a99329d620b325382ac27","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-02 13:43:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4aabe8df576467e91df5a6c8e182ab2487cf8fc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 15:51:36.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d40accee70154746bd08b3181cb73904\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b88280650b2c450d80feb823838a2db4 : SUCCESS in 2m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/df2bfabdc50749549f5c6b1ca1380ec6 : SUCCESS in 1h 01m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0ed8d03496804f7caac929df250ac199 : FAILURE in 1h 43m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3f81a4e90dea4a5da4ee3c453d3d4cf2 : SUCCESS in 32m 02s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/f73d7e71c9664d8d89510a73f3d1035d : SUCCESS in 1h 33m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/00f11723a2564e1eace1717d35238a46 : SUCCESS in 1h 54m 27s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/ecef7ff09d2849c2aec360a92736f40f : SUCCESS in 1h 37m 46s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/91f56787878f4b3b8301a57aaa24719c : SUCCESS in 31m 51s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/a004ba834c554d378837deef264f3e30 : SUCCESS in 1h 20m 25s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/3479ab1d16094816888ad0d1aed65ec0 : SUCCESS in 1h 07m 02s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/43e33ecb17af4aea945196ccc2f3aa1b : SUCCESS in 1h 24m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/d7df14cf44844c6b8b0a4a0dc1c970ab : SUCCESS in 1h 50m 07s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/9bb6adb923504c089de063348b3bdce9 : SUCCESS in 1h 55m 35s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/97f5e9947d4b48f084232c9e0395e3fe : SUCCESS in 32m 33s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/a3ac7d40f2dd49a784a01eeccb4cd9b4 : SUCCESS in 1h 22m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/80a660fffe074c71acd39ca2dfe38ad1 : SUCCESS in 1h 52m 02s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/5b0bbcdcdef24a69932e7f5a389e39c7 : SUCCESS in 1h 58m 40s\n- neutron-tempest-plugin-designate-scenario-2026-1 https://zuul.opendev.org/t/openstack/build/d0bfa153f96346948e427e98d8aad6f7 : SUCCESS in 33m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cc40562280c47f0bea52f344bcea707 : SUCCESS in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22d5aa02639d45628bced1591933b9aa : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":3},{"id":"e8cf4296a81310bc6a0e49bb8980eb20c51caa1e","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-02 16:46:10.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/b6df6d68cf134df28c37631642c1d6ad\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/b466775b2568457bade0b02c2d4204d8 : SUCCESS in 3h 07m 46s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/739c2340194848b0802844a19c133417 : FAILURE in 2h 22m 37s","accounts_in_message":[],"_revision_number":3},{"id":"20ee51d56797b9e725f92f00853959377edd830a","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 20:12:02.000000000","message":"Patch Set 3:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9f4d5b4dada4da49289289a05b7409a\n\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/cc12bbfa2a2e4f23ae3fc2fd9a9cd981 : SUCCESS in 2h 32m 48s (non-voting)\n- neutron-tempest-plugin-openvswitch-distributed-dhcp https://zuul.opendev.org/t/openstack/build/f5fdfcc7ee2e42e8a44b774d6af16454 : SUCCESS in 1h 06m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-distributed-dhcp https://zuul.opendev.org/t/openstack/build/cf05d5b5d8bd468299ace8d1dc1610fa : FAILURE in 1h 20m 05s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/2a3cab304e5b4b8fb257c6b11f202faa : FAILURE in 1h 49m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/78abc271ee2743f1a86e49c57ae2f52d : SUCCESS in 1h 16m 27s\n- neutron-tempest-plugin-dvr-multinode-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/1867b2bfce1a4aa2b2d2a96dde620fac : ERROR Failed to update project openstack/glance in 13s (non-voting)\n- neutron-tempest-plugin-dvr-multinode-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/9dbe4435b0f0422bbe52625ffe31607c : FAILURE in 3h 05m 04s (non-voting)\n- neutron-tempest-plugin-dvr-multinode-scenario-2026-1 https://zuul.opendev.org/t/openstack/build/fcb22eed9ae1413a972b2a482bd8515e : FAILURE in 2h 49m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bf42250d10159ca7553ed66e14bfa852b4dc5d09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-03 08:01:20.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"a2a41a06425bf327a20aa61343c280f9026e9241","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 10:24:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/02319c6f0e1b4b1eaf316bb0a04d3c22\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2524ef6003a485085b31a93de486ff9 : SUCCESS in 5m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fb4bffee0d75422fb7ad16c733bbc26c : SUCCESS in 1h 05m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5ae654e33dee49908b7496a482599ebb : FAILURE in 1h 54m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8bd91f7d4ddf42c3b331b4a7e6cbb762 : SUCCESS in 31m 03s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/36edaf31aa6a4a93937139a10fd07b5f : SUCCESS in 1h 21m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/2bb8224828b64c84a2b2eff83d9cb15c : SUCCESS in 1h 56m 37s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/c47395c730af429b94c84f8cbe34dccd : SUCCESS in 2h 06m 29s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/8450f1d2f67045c7bea47d42fd9425a7 : SUCCESS in 33m 26s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/99134def8a7e459c82418d684e1465dd : SUCCESS in 1h 35m 29s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/ca4b279dc8674a21aa5cd04bd199c263 : SUCCESS in 1h 15m 01s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/3f5304340960426199faad2e3af5961c : SUCCESS in 1h 27m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/11cf037ff7d44bc88f7463cd517557cc : SUCCESS in 1h 34m 41s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/d27177fd4440454b8fda59174cbbb1cd : SUCCESS in 2h 19m 37s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/5505326bf2cf45f3ba0614e88b87a95a : SUCCESS in 32m 37s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/7f463e5446c445b58cfe01ad2de790c1 : SUCCESS in 1h 23m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/f791147b9604400bb783aef4d83b020f : SUCCESS in 1h 51m 51s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/466dded5386549a79544e5aa49d56408 : POST_FAILURE in 1h 42m 43s\n- neutron-tempest-plugin-designate-scenario-2026-1 https://zuul.opendev.org/t/openstack/build/2c5ef3ae0e2b4125ae0938a08e53f69b : SUCCESS in 32m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af5ef4b041bb4721bacf92ff0996a85e : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee191e7fc7cc40acbdc5e89fdacfe27e : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":4},{"id":"d9eb1aeead9959310097f6f505dfe32bafb88a1d","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-03 12:24:12.000000000","message":"Patch Set 4: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/6ef0dafeb9a44826a006ea18386200f6\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/97f21c2e53e4406484415d5a226ef509 : SUCCESS in 4h 21m 24s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/3e4f98e369704e378616e46d5da72eba : SUCCESS in 4h 02m 10s","accounts_in_message":[],"_revision_number":4},{"id":"9cb55ae83d25a4be1bd76438d21a966afd8af25f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-04 08:25:09.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"a3bb34b8df38d75be00716148eccbaca42557b94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 10:57:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79f1eed7600f41e9a2ba87bd93da08e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb745aa23c4a4df0ba2c82f85006a0bb : SUCCESS in 3m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c24490af684746879609bcee87c833f3 : SUCCESS in 1h 05m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/55e8d1787c06436f9d994ef4aab70551 : SUCCESS in 2h 14m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8dc1296096b3416191c669e02dae6cf2 : SUCCESS in 32m 37s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/863e497dc0fa40d894f83821e58e8ce8 : SUCCESS in 1h 53m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/f68740776cb341c08a95d43ce52a813b : SUCCESS in 2h 12m 21s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/9e32432c66fe4abcad4fe2ee4aea4f70 : SUCCESS in 2h 23m 08s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/41a3767e68374c7ba54265d40d32d1a9 : SUCCESS in 32m 59s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/831ab361b2734c93ac0b8c825dbd9304 : SUCCESS in 1h 49m 46s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/9d8a622069cc432eb3045dd82636c5d9 : SUCCESS in 1h 19m 24s\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/d185096066734d5ebd3cc747abe56f4a : SUCCESS in 2h 08m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/cb48ebbf4cdf4b9f8ae48695ece72cdb : SUCCESS in 2h 29m 12s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/53945869594a4b71ab3eb32c37842153 : SUCCESS in 2h 17m 42s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/a3951b4a8ec4447b905ba6af408394c5 : SUCCESS in 33m 47s\n- neutron-tempest-plugin-openvswitch-2026-1 https://zuul.opendev.org/t/openstack/build/6aeb3eb3f90b4d3db11650bd05a4a804 : SUCCESS in 1h 52m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2026-1 https://zuul.opendev.org/t/openstack/build/5d4016b5164545fe8bcc58a1ccf071dd : SUCCESS in 2h 23m 54s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/6c790dd6037845fb990bd101ee9f6556 : SUCCESS in 2h 16m 38s\n- neutron-tempest-plugin-designate-scenario-2026-1 https://zuul.opendev.org/t/openstack/build/2da095e275454501b1ead6b47c79d539 : SUCCESS in 33m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95a3a2f799bf4960a8e4df4a571c1ba9 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13d3bbbe8f49414db82ba85bcc06d985 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":5},{"id":"42cce711ee854dd2c4f6ad45285bce14ef03687b","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-04 14:13:57.000000000","message":"Patch Set 5: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/8b449360b9bf472d9c32862b34efe661\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/8398e4bb9072415eb052e82e85dc114c : SUCCESS in 5h 42m 33s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/56dfbbdc7ed841ad8a73ac0d66233070 : SUCCESS in 3h 50m 53s","accounts_in_message":[],"_revision_number":5},{"id":"9f1c56a4b5633a317187152b11f72c151440c45f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-09-08 19:41:50.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"d8fe7946a43b7306cd174a00ba7a9abeac6c25c2","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-09-09 10:07:51.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"e4ac488b22e6119816acbc76c6464d30237e12f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 10:08:28.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"076aa539f64b81b8f267f858a12a2abbb1054585","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 12:22:42.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d952e20542c41cb934dfb4831627261\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66a202f0e71c4297a928f960f7f4ce64 : SUCCESS in 3m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4943c09e87e140c3a3b4e9e571ba60a5 : SUCCESS in 1h 07m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/118198940ad946328ddb4e7689c3004a : SUCCESS in 1h 40m 05s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/aa2bedf2cd0b4b319d696f4869a9171a : SUCCESS in 1h 48m 44s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/91b863d22962415f951771ce5cd86d17 : SUCCESS in 2h 11m 43s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/f32626dbaab34d888c2062d57dcdee67 : SUCCESS in 2h 13m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a860e1e1531487ca98d625518fd6297 : SUCCESS in 3m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bbe4c65ebc54e68bb7206245aba1ce4 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":5},{"id":"c3af48980623d4e79a0a5b7849d995f54f120663","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 12:22:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5221d4432fe334a8941a584d151ae9b33373b0cb","revisions":{"435397131bab94e3c29f4abae9bef2e6c2ded508":{"kind":"REWORK","_number":1,"created":"2026-09-01 12:21:16.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/65/1003265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/65/1003265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/1003265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/1003265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/1003265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/1003265/1"}}},"commit":{"parents":[{"commit":"1e9511ad4bc0971033b7fc0a8a7544835bb3c52c","subject":"Add negative API tests for the indirect Floating IPs API extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1e9511ad4bc0971033b7fc0a8a7544835bb3c52c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-01 12:18:24.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-01 12:21:03.000000000","tz":120},"subject":"Add scenario test for the indirect Floating IPs API extension","message":"Add scenario test for the indirect Floating IPs API extension\n\nThis patch adds scenario test case for indirect Floating IP\nspecified in [1] and implemented in [2] in Neutron.\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/996269\n[2] https://review.opendev.org/c/openstack/neutron/+/1003262\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/1003262\n\nRelated-bug: #2072505\n\nAssisted By: OpenAI GPT-5.6 Terra\n\nChange-Id: I2d7f269a08a2146f68b38ca338458ed83d9792f0\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/435397131bab94e3c29f4abae9bef2e6c2ded508"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/435397131bab94e3c29f4abae9bef2e6c2ded508"}]},"branch":"refs/heads/master"},"6c0f0858497da7fcf25c824626e71de3b72b1f89":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-02 09:38:05.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/65/1003265/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/65/1003265/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/1003265/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/1003265/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/1003265/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/1003265/2"}}},"commit":{"parents":[{"commit":"0ba778221c0807aa8456bf2b30d8d4d1fbf319ed","subject":"Add negative API tests for the indirect Floating IPs API extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0ba778221c0807aa8456bf2b30d8d4d1fbf319ed"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-01 12:18:24.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-02 09:31:25.000000000","tz":120},"subject":"Add scenario test for the indirect Floating IPs API extension","message":"Add scenario test for the indirect Floating IPs API extension\n\nThis patch adds scenario test case for indirect Floating IP\nspecified in [1] and implemented in [2] in Neutron.\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/996269\n[2] https://review.opendev.org/c/openstack/neutron/+/1003262\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/1003262\n\nRelated-bug: #2072505\n\nAssisted By: OpenAI GPT-5.6 Terra\n\nChange-Id: I2d7f269a08a2146f68b38ca338458ed83d9792f0\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6c0f0858497da7fcf25c824626e71de3b72b1f89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6c0f0858497da7fcf25c824626e71de3b72b1f89"}]},"branch":"refs/heads/master"},"cc93d3d9f2b84e63959d33944db9e742d8c07cba":{"kind":"REWORK","_number":3,"created":"2026-09-02 13:35:17.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/65/1003265/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/65/1003265/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/1003265/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/1003265/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/1003265/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/1003265/3"}}},"commit":{"parents":[{"commit":"0ba778221c0807aa8456bf2b30d8d4d1fbf319ed","subject":"Add negative API tests for the indirect Floating IPs API extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0ba778221c0807aa8456bf2b30d8d4d1fbf319ed"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-01 12:18:24.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-02 13:34:56.000000000","tz":120},"subject":"Add scenario test for the indirect Floating IPs API extension","message":"Add scenario test for the indirect Floating IPs API extension\n\nThis patch adds scenario test case for indirect Floating IP\nspecified in [1] and implemented in [2] in Neutron.\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/996269\n[2] https://review.opendev.org/c/openstack/neutron/+/1003262\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/1003262\n\nRelated-bug: #2072505\n\nAssisted By: OpenAI GPT-5.6 Terra\n\nChange-Id: I2d7f269a08a2146f68b38ca338458ed83d9792f0\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/cc93d3d9f2b84e63959d33944db9e742d8c07cba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/cc93d3d9f2b84e63959d33944db9e742d8c07cba"}]},"branch":"refs/heads/master"},"803870c1b28f86beaf8f846a7ac70ef969382314":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-09-03 08:01:20.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/65/1003265/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/65/1003265/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/1003265/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/1003265/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/1003265/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/1003265/4"}}},"commit":{"parents":[{"commit":"f55b1739dfd5f4a734aa777fb5a07340892e081a","subject":"Add negative API tests for the indirect Floating IPs API extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f55b1739dfd5f4a734aa777fb5a07340892e081a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-01 12:18:24.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-03 08:00:58.000000000","tz":120},"subject":"Add scenario test for the indirect Floating IPs API extension","message":"Add scenario test for the indirect Floating IPs API extension\n\nThis patch adds scenario test case for indirect Floating IP\nspecified in [1] and implemented in [2] in Neutron.\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/996269\n[2] https://review.opendev.org/c/openstack/neutron/+/1003262\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/1003262\n\nRelated-bug: #2072505\n\nAssisted By: OpenAI GPT-5.6 Terra\n\nChange-Id: I2d7f269a08a2146f68b38ca338458ed83d9792f0\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/803870c1b28f86beaf8f846a7ac70ef969382314"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/803870c1b28f86beaf8f846a7ac70ef969382314"}]},"branch":"refs/heads/master"},"5221d4432fe334a8941a584d151ae9b33373b0cb":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-09-04 08:25:09.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/65/1003265/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/65/1003265/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/1003265/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/1003265/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/1003265/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/1003265/5"}}},"commit":{"parents":[{"commit":"e0d36c54b08d42b79de40aa286d6ad66636c3b54","subject":"Add negative API tests for the indirect Floating IPs API extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e0d36c54b08d42b79de40aa286d6ad66636c3b54"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-01 12:18:24.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-04 08:25:09.000000000","tz":0},"subject":"Add scenario test for the indirect Floating IPs API extension","message":"Add scenario test for the indirect Floating IPs API extension\n\nThis patch adds scenario test case for indirect Floating IP\nspecified in [1] and implemented in [2] in Neutron.\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/996269\n[2] https://review.opendev.org/c/openstack/neutron/+/1003262\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/1003262\n\nRelated-bug: #2072505\n\nAssisted By: OpenAI GPT-5.6 Terra\n\nChange-Id: I2d7f269a08a2146f68b38ca338458ed83d9792f0\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5221d4432fe334a8941a584d151ae9b33373b0cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5221d4432fe334a8941a584d151ae9b33373b0cb"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"f55b1739dfd5f4a734aa777fb5a07340892e081a","ours":"803870c1b28f86beaf8f846a7ac70ef969382314","theirs":"e0d36c54b08d42b79de40aa286d6ad66636c3b54","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
