)]}'
{"id":"openstack%2Fneutron-tempest-plugin~1003633","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I95394c4ab28bb78fea082844c7a21d675860a97b","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-09-08 15:03:25.000000000","reason":"Change was submitted"},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-09-04 13:51:08.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I95394c4ab28bb78fea082844c7a21d675860a97b","subject":"Turn back on `ovn_router_indirect_snat` in the ovn job","status":"MERGED","created":"2026-09-03 08:01:20.000000000","updated":"2026-09-08 15:03:25.000000000","submitted":"2026-09-08 15:03:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003633","meta_rev_id":"11e2ec02ed861f774f9da3dc3811bc69029f724f","_number":1003633,"virtual_id_number":1003633,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-08 15:03:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-09-07 12:29:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-09-04 13:51:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2026-09-08 12:56:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2026-09-08 12:56:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-09-07 12:29:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 10:05:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-03 12:07:28.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-09-03 21:22:42.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-09-07 12:29:52.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-09-08 12:56:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"27ff728f11cbf0477b229cc9569e1ddc3d4b42e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-03 08:01:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5267e98bde5b6524c07ed6404c8e13e6ae65c57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 10:05:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f9a55b7e9c44117b6d47f1d009bc25b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5abdf24f081b404c859aa99d9dd3a4f7 : SUCCESS in 2m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3c13caa169cc46a88b56930d06e3b17d : FAILURE in 1h 37m 44s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/bc6090522ae945b0a017047ad4915a2f : SUCCESS in 1h 53m 23s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/a0bbf58036944d3e901077e10c2339f6 : SUCCESS in 1h 10m 58s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/98eeed1264a44ac39b34d78c2176b597 : SUCCESS in 2h 01m 32s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/e24d9d1ce71e4d0e80394786ac1e8086 : SUCCESS in 1h 59m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4d15699d721469b862b6c25548402c2 : SUCCESS in 2m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac249024f419476f99622f6ca8d20c8b : SUCCESS in 2m 42s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/78872fd8ac854bb8bd144486241fac80 : SUCCESS in 31m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9a4a32343d2b3e1761131f6a098a04099e40a34f","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-03 12:07:28.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/b78cf832d8524543871e3ae8e3359933\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/c4bd5e9c519a4c4bbffc97acfe8d292d : SUCCESS in 4h 03m 34s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/ebbdb50155ce41728bef8b89ef414815 : SUCCESS in 3h 46m 37s","accounts_in_message":[],"_revision_number":1},{"id":"8c3d864f307d05a0c450e2447cc261d54d6915c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-09-04 08:20:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"59c3941fe3cb291ecb0c1cd7dd484e2a717d8073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 11:10:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d175ff4232944f2afb7b80d0cb86f51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90bfc4ea31b444df9186a936283c4757 : SUCCESS in 4m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/00a97b9339c84b6abc0f65b710785b74 : SUCCESS in 2h 48m 41s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/653ec3a9cd62402dad2d65c66b388d05 : SUCCESS in 2h 27m 13s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/fb9d01a667f6417ebb405f551ba4bbe1 : SUCCESS in 1h 29m 47s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/7997626f27194a998c3d880af9f157de : SUCCESS in 2h 23m 02s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/33ef5c84b3264b76abafc329f4105548 : SUCCESS in 2h 10m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a179d536cf374f90bd66c19a7ba18552 : SUCCESS in 4m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b74c76cf65764a67996ef77f3c354aa4 : SUCCESS in 3m 36s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/c0598047779b440ea92f4c95c9673057 : SUCCESS in 32m 39s","accounts_in_message":[],"_revision_number":2},{"id":"7f4eda3c56e1e1466158daa3d768e600a3705b33","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-04 13:51:08.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/4fe561c0d8dd4e31a700e322540354a1\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/1038c1f207f54b09806e7b5365f487d7 : SUCCESS in 5h 19m 42s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/88bde7eedd864fbca3a2429b34e9cb0a : SUCCESS in 3h 31m 24s","accounts_in_message":[],"_revision_number":2},{"id":"8baa80d352e6780c9d09391c8a74c1ed0c6f3b00","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-07 12:29:52.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"9784a8ffefe14820c70b1f7bb34b021bc7ec9332","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-09-08 12:56:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8b1855bcafdc2842bb228ea81e31c935e28c888c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 12:57:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"16a750cb7c8043cd57a7db905726f026a699fd25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 15:03:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fbc5c83b70d43cb86c5574c3c21a30b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7c53ddab240491da4bdafe9f70ea091 : SUCCESS in 3m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/560ea92368a645859d7ae87bb529541d : SUCCESS in 1h 34m 21s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/8a5fd32c506f41bf8cd11f0c6804f33a : SUCCESS in 1h 57m 06s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/5a62c2fbfe1d42b7b70961ac8a9eaf7a : SUCCESS in 2h 04m 53s\n- neutron-tempest-plugin-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/6bcc68817fb14380a838f589728be65b : SUCCESS in 1h 58m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb7ab49f1f424deab19021325d62864f : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba3996b11694426b8867be1cae450650 : SUCCESS in 2m 35s\n- neutron-tempest-plugin-fwaas-ovn https://zuul.opendev.org/t/openstack/build/2303f78d124948a3a22b0bb8258a8538 : SUCCESS in 31m 04s","accounts_in_message":[],"_revision_number":2},{"id":"11e2ec02ed861f774f9da3dc3811bc69029f724f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 15:03:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2c7bdae8ac095461f5903048769fadc2b3e42427","revisions":{"5967ec0584b0f115822e1c7bec8d466033421702":{"kind":"REWORK","_number":1,"created":"2026-09-03 08:01:20.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/33/1003633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/33/1003633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/33/1003633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/33/1003633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/33/1003633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/33/1003633/1"}}},"commit":{"parents":[{"commit":"69c70a6c08d165d3a7e1a4f4d51d2efdf1e84107","subject":"Merge \"Add PVLAN cold migration scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/69c70a6c08d165d3a7e1a4f4d51d2efdf1e84107"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-03 07:58:56.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-03 07:58:56.000000000","tz":120},"subject":"Turn back on `ovn_router_indirect_snat` in the ovn job","message":"Turn back on `ovn_router_indirect_snat` in the ovn job\n\nIt was disabled due to too old OVN used in Ubuntu (24.03.2) but now it\nseems that there is already OVN 24.03.6 and we can enable it back.\n\nChange-Id: I95394c4ab28bb78fea082844c7a21d675860a97b\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5967ec0584b0f115822e1c7bec8d466033421702"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5967ec0584b0f115822e1c7bec8d466033421702"}]},"branch":"refs/heads/master"},"2c7bdae8ac095461f5903048769fadc2b3e42427":{"kind":"REWORK","_number":2,"created":"2026-09-04 08:20:26.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/33/1003633/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/33/1003633/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/33/1003633/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/33/1003633/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/33/1003633/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/33/1003633/2"}}},"commit":{"parents":[{"commit":"69c70a6c08d165d3a7e1a4f4d51d2efdf1e84107","subject":"Merge \"Add PVLAN cold migration scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/69c70a6c08d165d3a7e1a4f4d51d2efdf1e84107"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-03 07:58:56.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2026-09-04 08:20:22.000000000","tz":120},"subject":"Turn back on `ovn_router_indirect_snat` in the ovn job","message":"Turn back on `ovn_router_indirect_snat` in the ovn job\n\nIt was disabled to skip test `test_nested_snat_external_ip` which is\nfailing with the OVN version provided by the Ubuntu 24.04.\nNow it is better to keep it enabled to be able to test \"indirect\nFloating IP\" functionality which requires nested SNAT to be enabled in\nNeutron but should works fine even with OVN version provided by Ubuntu\n24.04.\n\nSo this patch enables that config option in Neutron and in the same time\nadds the `test_nested_snat_external_ip` to the excluded list in the\njob\u0027s definition.\n\nChange-Id: I95394c4ab28bb78fea082844c7a21d675860a97b\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2c7bdae8ac095461f5903048769fadc2b3e42427"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2c7bdae8ac095461f5903048769fadc2b3e42427"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
