)]}'
{"id":"openstack%2Fneutron-tempest-plugin~1004439","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I695292f10aa5b15499122c3fc92ec945afee7e9b","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/1882220","attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-09-08 19:13:24.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-09-08 12:30:12.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I695292f10aa5b15499122c3fc92ec945afee7e9b","subject":"DNM \u003d\u003d Test ``neutron-tempest-plugin-vpnaas`` scenario classes","status":"NEW","created":"2026-09-07 15:04:14.000000000","updated":"2026-09-08 19:13:24.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"16ee99c4d80ffa4e98921d6ea2b0efa7d9e8af37","_number":1004439,"virtual_id_number":1004439,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-08 14:35:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":-1,"date":"2026-09-08 12:30:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-07 16:34:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-07 18:56:36.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"}],"messages":[{"id":"5dbd53cc23eb678e28897cbd1ee763b474248007","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-07 15:04:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f92fdd7242c60315a77893e860bee1890a802059","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 16:34:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2b0a76d164e4468b4b31de989536a2d\n\n- neutron-tempest-plugin-vpnaas-1 https://zuul.opendev.org/t/openstack/build/34786572c1cf48abbd2bc1503327e903 : SUCCESS in 1h 24m 00s\n- neutron-tempest-plugin-vpnaas-2 https://zuul.opendev.org/t/openstack/build/43ce9d24af8440cf900eedbd6b040160 : SUCCESS in 1h 12m 57s\n- neutron-tempest-plugin-vpnaas-3 https://zuul.opendev.org/t/openstack/build/1b6194c3b30c407fa88f3a31bb7d2aed : SUCCESS in 1h 11m 54s\n- neutron-tempest-plugin-vpnaas-4 https://zuul.opendev.org/t/openstack/build/6f04406e73324aeb9efac21ab6d403b7 : SUCCESS in 1h 10m 34s\n- neutron-tempest-plugin-vpnaas-5 https://zuul.opendev.org/t/openstack/build/90a8d6fb9fd64b218faa6974b15b9af1 : SUCCESS in 1h 02m 33s","accounts_in_message":[],"_revision_number":1},{"id":"92f765f018935c6f777f4d930ee6c3292b544e8c","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-07 18:56:36.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/0df0b6fc9a094907b1de521f0f4315d5\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/adec4de8a76046b0b00b3cf4786c20b3 : SUCCESS in 3h 51m 28s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/4dcb5385692b4973a699026aebb8a269 : SUCCESS in 3h 33m 59s","accounts_in_message":[],"_revision_number":1},{"id":"220c91206c42383e03b27a35701e25274566feb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-08 06:32:29.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"ec8d49a3901b51e24744326a89c899ae90c1c479","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-08 06:37:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0fbc9e48856afc8f8beb11c9bc1132f8191c5cef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-08 07:28:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"efc8dab6e81c75d5fe62e9f3515ef63ce1787310","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-08 08:36:04.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"3c129fa0129eecbe51244b3f22e97ade360e234a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 09:46:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0463c86d2e4406f99eb21c84ea0c9d8\n\n- neutron-tempest-plugin-vpnaas-1 https://zuul.opendev.org/t/openstack/build/4e89cd62e83140e287dd27f520cf381b : SUCCESS in 50m 57s\n- neutron-tempest-plugin-vpnaas-2 https://zuul.opendev.org/t/openstack/build/554ecc7d49fb41d28499b33dcd421654 : SUCCESS in 52m 23s\n- neutron-tempest-plugin-vpnaas-3 https://zuul.opendev.org/t/openstack/build/fd88e5d431f14747a235e393b1adadfd : SUCCESS in 53m 51s\n- neutron-tempest-plugin-vpnaas-4 https://zuul.opendev.org/t/openstack/build/da65e50ef7924abcafc9e5d139761ea3 : SUCCESS in 52m 47s\n- neutron-tempest-plugin-vpnaas-5 https://zuul.opendev.org/t/openstack/build/58ea14bb6ea64982a53d71c677e94e97 : SUCCESS in 1h 04m 04s\n- neutron-tempest-plugin-vpnaas-ovn-1 https://zuul.opendev.org/t/openstack/build/ac56317357bd41c1a72b46a67c37911d : SUCCESS in 52m 13s\n- neutron-tempest-plugin-vpnaas-ovn-2 https://zuul.opendev.org/t/openstack/build/7d0a9422b9ea47c6822e583c2cc3c745 : SUCCESS in 50m 58s\n- neutron-tempest-plugin-vpnaas-ovn-3 https://zuul.opendev.org/t/openstack/build/2f255cc1100a4e6293b15b5aaef7de35 : SUCCESS in 52m 43s\n- neutron-tempest-plugin-vpnaas-ovn-4 https://zuul.opendev.org/t/openstack/build/5d703c97b4324ae8b6e5ea1e494ee0bf : SUCCESS in 52m 03s\n- neutron-tempest-plugin-vpnaas-ovn-5 https://zuul.opendev.org/t/openstack/build/3a8804958bc14fec826f4eb0ea9029c4 : SUCCESS in 53m 04s","accounts_in_message":[],"_revision_number":5},{"id":"d86c03144d4b83224d03ee0095c17bf26272996d","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-08 12:30:12.000000000","message":"Patch Set 5: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/e50a0cf671fe41a59eb659550f6282c1\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/c3a78599844a4ead9f93f27a761ff0fb : SUCCESS in 3h 52m 33s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/f66ff830e6c046f591242c594bd02001 : FAILURE in 3h 37m 48s","accounts_in_message":[],"_revision_number":5},{"id":"76bf0e3a59e726a28be616b277daac5e247a505c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-08 13:33:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c6435c0b910d0fa4ac8509db268f0e806341fb0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 14:35:00.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee2ee9ebfc1d43ccb2195e10d53980eb\n\n- neutron-tempest-plugin-vpnaas-1 https://zuul.opendev.org/t/openstack/build/0dba991ab63f4682867e7cdaff75f6ab : SUCCESS in 53m 10s\n- neutron-tempest-plugin-vpnaas-2 https://zuul.opendev.org/t/openstack/build/1e9f1543b0fd45e9a64537981f32120e : FAILURE in 47m 22s\n- neutron-tempest-plugin-vpnaas-3 https://zuul.opendev.org/t/openstack/build/18c380830976410fb417447fb7a86c31 : SUCCESS in 26m 43s\n- neutron-tempest-plugin-vpnaas-4 https://zuul.opendev.org/t/openstack/build/0a72f1eb9c4e4d7e9fc1531f23a0f8e4 : SUCCESS in 27m 52s\n- neutron-tempest-plugin-vpnaas-5 https://zuul.opendev.org/t/openstack/build/69c9b0fda27f48b2a0a84a40c5f8083e : FAILURE in 53m 49s\n- neutron-tempest-plugin-vpnaas-ovn-1 https://zuul.opendev.org/t/openstack/build/c06e356bc9aa41dba584a8de781d48ea : SUCCESS in 51m 36s\n- neutron-tempest-plugin-vpnaas-ovn-2 https://zuul.opendev.org/t/openstack/build/b21d1b9846a94149b78e207831f2a0ce : FAILURE in 46m 09s\n- neutron-tempest-plugin-vpnaas-ovn-3 https://zuul.opendev.org/t/openstack/build/f2fed2aca08a41f5b2c41be29083162e : SUCCESS in 25m 53s\n- neutron-tempest-plugin-vpnaas-ovn-4 https://zuul.opendev.org/t/openstack/build/4a19a4012a124d5f815c81a052efc2e5 : FAILURE in 42m 05s\n- neutron-tempest-plugin-vpnaas-ovn-5 https://zuul.opendev.org/t/openstack/build/36c1ad5142d4466b8d77864808a5bf79 : SUCCESS in 50m 50s","accounts_in_message":[],"_revision_number":5},{"id":"16ee99c4d80ffa4e98921d6ea2b0efa7d9e8af37","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-08 19:13:24.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/07d909933043461391fae4583645bee3\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/da4a4137532b4d94851e2766f1f8c19e : SUCCESS in 4h 21m 17s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/09e6abd41bca40c4813f261d06e52e49 : FAILURE in 3h 31m 54s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"77eb622a3e816fe2c54635e8f2d5ee77ca6ee9ca","revisions":{"eae77c9efe5d39c493295f01de51ea9b62ed07e5":{"kind":"REWORK","_number":1,"created":"2026-09-07 15:04:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/1004439/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/39/1004439/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/39/1004439/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/39/1004439/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/39/1004439/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/39/1004439/1"}}},"commit":{"parents":[{"commit":"6577c3b05b3f62b17388e18635e7d732f64f90d0","subject":"vpnaas: Use unique CIDRs per scenario test class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6577c3b05b3f62b17388e18635e7d732f64f90d0"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-09-07 15:03:46.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-09-07 15:03:46.000000000","tz":120},"subject":"DNM \u003d\u003d Test ``neutron-tempest-plugin-vpnaas`` scenario classes","message":"DNM \u003d\u003d Test ``neutron-tempest-plugin-vpnaas`` scenario classes\n\nStress-test the unique-CIDR fix by running the VPNaaS CI job five\ntimes in parallel and duplicating each ``Vpnaas*in*`` scenario class\nfive times per job.\n\nRelated-Bug: #1882220\nAssisted-By: Composer 2.5\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I695292f10aa5b15499122c3fc92ec945afee7e9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/eae77c9efe5d39c493295f01de51ea9b62ed07e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/eae77c9efe5d39c493295f01de51ea9b62ed07e5"}]},"branch":"refs/heads/master"},"01b50ae5d912f4796ae33f47e82f3607aa4e791a":{"kind":"REWORK","_number":2,"created":"2026-09-08 06:32:29.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/1004439/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/39/1004439/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/39/1004439/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/39/1004439/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/39/1004439/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/39/1004439/2"}}},"commit":{"parents":[{"commit":"2351b01c1bb077e6d715311468adb6ac28c0d03d","subject":"vpnaas: Use unique CIDRs per scenario test class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2351b01c1bb077e6d715311468adb6ac28c0d03d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-09-07 15:03:46.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-09-08 06:32:15.000000000","tz":120},"subject":"DNM \u003d\u003d Test ``neutron-tempest-plugin-vpnaas`` scenario classes","message":"DNM \u003d\u003d Test ``neutron-tempest-plugin-vpnaas`` scenario classes\n\nStress-test the unique-CIDR fix by running the VPNaaS CI job five\ntimes in parallel and duplicating each ``Vpnaas*in*`` scenario class\nfive times per job.\n\nRelated-Bug: #1882220\nAssisted-By: Composer 2.5\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I695292f10aa5b15499122c3fc92ec945afee7e9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/01b50ae5d912f4796ae33f47e82f3607aa4e791a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/01b50ae5d912f4796ae33f47e82f3607aa4e791a"}]},"branch":"refs/heads/master"},"12c2dedf6c4f91acd05bfd6b2efe0c1130aceaf5":{"kind":"REWORK","_number":3,"created":"2026-09-08 06:37:24.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/1004439/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/39/1004439/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/39/1004439/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/39/1004439/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/39/1004439/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/39/1004439/3"}}},"commit":{"parents":[{"commit":"2351b01c1bb077e6d715311468adb6ac28c0d03d","subject":"vpnaas: Use unique CIDRs per scenario test class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2351b01c1bb077e6d715311468adb6ac28c0d03d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-09-07 15:03:46.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-09-08 06:37:16.000000000","tz":120},"subject":"DNM \u003d\u003d Test ``neutron-tempest-plugin-vpnaas`` scenario classes","message":"DNM \u003d\u003d Test ``neutron-tempest-plugin-vpnaas`` scenario classes\n\nStress-test the unique-CIDR fix by running the VPNaaS CI job five\ntimes in parallel and duplicating each ``Vpnaas*in*`` scenario class\nfive times per job.\n\nRelated-Bug: #1882220\nAssisted-By: Composer 2.5\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I695292f10aa5b15499122c3fc92ec945afee7e9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/12c2dedf6c4f91acd05bfd6b2efe0c1130aceaf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/12c2dedf6c4f91acd05bfd6b2efe0c1130aceaf5"}]},"branch":"refs/heads/master"},"2e64d217cc82fce97899a509c039350d03cca1f7":{"kind":"REWORK","_number":4,"created":"2026-09-08 07:28:09.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/1004439/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/39/1004439/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/39/1004439/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/39/1004439/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/39/1004439/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/39/1004439/4"}}},"commit":{"parents":[{"commit":"0e7f614e39497ba007c70d270f01a945fa3a805d","subject":"vpnaas: Use unique CIDRs per scenario test class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0e7f614e39497ba007c70d270f01a945fa3a805d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-09-07 15:03:46.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-09-08 07:27:40.000000000","tz":120},"subject":"DNM \u003d\u003d Test ``neutron-tempest-plugin-vpnaas`` scenario classes","message":"DNM \u003d\u003d Test ``neutron-tempest-plugin-vpnaas`` scenario classes\n\nStress-test the unique-CIDR fix by running the VPNaaS CI job five\ntimes in parallel and duplicating each ``Vpnaas*in*`` scenario class\nfive times per job.\n\nRelated-Bug: #1882220\nAssisted-By: Composer 2.5\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I695292f10aa5b15499122c3fc92ec945afee7e9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2e64d217cc82fce97899a509c039350d03cca1f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2e64d217cc82fce97899a509c039350d03cca1f7"}]},"branch":"refs/heads/master"},"77eb622a3e816fe2c54635e8f2d5ee77ca6ee9ca":{"kind":"REWORK","_number":5,"created":"2026-09-08 08:36:04.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/1004439/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/39/1004439/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/39/1004439/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/39/1004439/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/39/1004439/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/39/1004439/5"}}},"commit":{"parents":[{"commit":"0e7f614e39497ba007c70d270f01a945fa3a805d","subject":"vpnaas: Use unique CIDRs per scenario test class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0e7f614e39497ba007c70d270f01a945fa3a805d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-09-07 15:03:46.000000000","tz":120},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-09-08 08:36:04.000000000","tz":0},"subject":"DNM \u003d\u003d Test ``neutron-tempest-plugin-vpnaas`` scenario classes","message":"DNM \u003d\u003d Test ``neutron-tempest-plugin-vpnaas`` scenario classes\n\nStress-test the unique-CIDR fix by running the VPNaaS CI job five\ntimes in parallel and duplicating each ``Vpnaas*in*`` scenario class\nfive times per job.\n\nRelated-Bug: #1882220\nAssisted-By: Composer 2.5\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I695292f10aa5b15499122c3fc92ec945afee7e9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/77eb622a3e816fe2c54635e8f2d5ee77ca6ee9ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/77eb622a3e816fe2c54635e8f2d5ee77ca6ee9ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
