)]}'
{"id":"openstack%2Fneutron-tempest-plugin~1006184","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ia9e211200df7c662223afe81e06b5280126b2ef1","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{"35308":{"account":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"last_update":"2026-10-05 10:10:34.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-10-03 06:55:31.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ia9e211200df7c662223afe81e06b5280126b2ef1","subject":"[TaaS] Add MTU and Cross-tenant test to Tap-as-a-Service.","status":"NEW","created":"2026-09-18 10:08:32.000000000","updated":"2026-10-05 10:10:34.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cce9e8499dfba7d19a6865588395cf9b026c09f4","_number":1006184,"virtual_id_number":1006184,"owner":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-10-05 08:12:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":-1,"date":"2026-10-03 06:55:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 12:19:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-18 23:08:20.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"}],"messages":[{"id":"7ebd72e3998a6feca68e2507ba2561f6a7f60300","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-09-18 10:08:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"467cf20edd733ea3244ae8cc1bf3e4d10fd4d338","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 12:19:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/49e5de40e7224d5b80340cfe8c849402\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04f35fca46b84335bfc10353db63bdf5 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2c943a408ae4c8c888240097190de08 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bd5299ed34c4be795f57b19a1b543b8 : SUCCESS in 4m 29s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/fc441629fdcb4a0ca989a7e86b360281 : FAILURE in 1h 04m 42s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/17c0adc1bbb84dd6a6e79c4a5b80f774 : FAILURE in 1h 25m 45s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/5318502f4fff41299a75ccbd5aa279ef : TIMED_OUT in 2h 04m 00s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/f24fdefd6baa477ba4fe7f2c3cf6aae5 : FAILURE in 1h 29m 16s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/00d5069cd09647bf804bf0a65aea9975 : TIMED_OUT in 2h 04m 33s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/95218b255a14450c956202178894e7d5 : FAILURE in 1h 52m 29s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/f7f96d7e18324e349feff0dc7ad09894 : FAILURE in 1h 37m 51s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/6f662fc092314ffea3f2dbdf0eb4735f : FAILURE in 1h 49m 53s","accounts_in_message":[],"_revision_number":1},{"id":"6cfa1fb6e9ce7499874774fe51f284c64e60b2cb","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-18 23:08:20.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/5a39d95c415a456a9388fa29b69c3f00\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/0293636a829d4c5ba64bb7dd458a0be1 : SUCCESS in 4h 07m 53s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/8ad15584933a4ca4adb43b8e8f508ac5 : SUCCESS in 3h 44m 25s","accounts_in_message":[],"_revision_number":1},{"id":"21522937cea007b15ed5dc9d4f9f383c6c03581e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-09-19 03:07:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"a2a4027e8253090b0591a517170a6b4888ee6808","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 05:23:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/959b349bd8b3438c80d6781e25ea1d06\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47fe8c32f5d040388b50f2537a140254 : RETRY_LIMIT\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/736bb3008bb245758499c21a67930b4d : RETRY_LIMIT\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f57bdbd5867a4d40978771582fe7a13d : RETRY_LIMIT\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/d0ae04c202224032b18834da7eda9b6d : RETRY_LIMIT\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/0dbaa3e8d2304197a5001cad3438ec84 : RETRY_LIMIT\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/5f905850484f4c70a2c03f5bac2ad6fc : SUCCESS in 1h 45m 08s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/a2004345375147588a22772528613a10 : RETRY_LIMIT\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/43b5016f3c3542fcab4f383658605e90 : SUCCESS in 1h 46m 04s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/15c0d508cc524ef78d97efda4bacd61b : SUCCESS in 1h 51m 07s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/1375dd314fe7420dbbfd9fa2cb1ac4af : SUCCESS in 1h 44m 45s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/191debfe0a1b42029d8d883faa44dd82 : SUCCESS in 1h 45m 21s","accounts_in_message":[],"_revision_number":2},{"id":"03600f9e59d433c2bcf72766724dd9bbf760ab0a","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-19 06:55:11.000000000","message":"Patch Set 2: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/906f4bef6aeb46cfb5bfa3c9d9aac6f0\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/5b55d7b460474c04b78512e4ad05c3c2 : SUCCESS in 3h 46m 48s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/6bf66a0cb47745bc9377590d85ddfa97 : FAILURE in 3h 32m 09s","accounts_in_message":[],"_revision_number":2},{"id":"54f6931705831c101d6673828582c06848840940","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-09-21 06:26:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"696e7bd5b82702dda7afe89a387f205829d218fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 08:37:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1aad568e66814d769138db6f978c441c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e600339ca4e47ca89b32ee78f94da76 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fb3757ba4eb40ccb584705ec5c8fca2 : SUCCESS in 5m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c973ad05bdeb47dea404bd06ba660cb0 : SUCCESS in 4m 49s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/a992cf90807f4588ac0d227c9b124612 : SUCCESS in 1h 27m 55s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/bf51238afa0847bf8c89b17e1a5285d6 : SUCCESS in 1h 54m 05s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/f7c524ad2cb64c2b9616e77afcb59d7d : SUCCESS in 1h 09m 09s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/6d9150c01e664c1a90007a7016ba14ef : SUCCESS in 1h 45m 34s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/3447c622235b48f79a9af91d03e2c7d5 : SUCCESS in 1h 42m 40s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/6895dff09cfd4653824dd20753e9e817 : SUCCESS in 1h 56m 50s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/063028c9986f48a8bb96ea6673c02b11 : SUCCESS in 48m 56s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/c6d51410ad564a89902aa6a2195b30de : TIMED_OUT in 2h 05m 27s","accounts_in_message":[],"_revision_number":2},{"id":"524de041f0014a21662f2e2557f9869f5c445d6e","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-21 10:52:58.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/2f6aa80e63ab4301b729312becc8668e\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/fc69350dad154878bf077119af235baa : SUCCESS in 4h 24m 56s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/61bc8b93151d483a9349bd4af9f5170e : SUCCESS in 3h 49m 47s","accounts_in_message":[],"_revision_number":2},{"id":"ae172a09aae0f5672f9f9f4db3457e14dfaa23d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-09-30 08:36:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"63416ea8ff4e20caaf2f2327d81da78d64a7d8c1","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-30 10:30:11.000000000","message":"Patch Set 3: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/05a89c1135cd4536979b95d4a12a4245\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/a326211be9cb42eba03fcd865b923fba : FAILURE in 5m 13s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/8285df4bc6014992a20c43f6924815d8 : SKIPPED Skipped due to failed job openstack-meta-content-provider","accounts_in_message":[],"_revision_number":3},{"id":"d90e2dae844394eb9fd906def3ca71c667dea217","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 10:49:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6240a851c3b47aa84050866cafa2c5f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb751ba06d184ee9bd028d3af297f315 : FAILURE in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af0dd3a962ce48e49d89005fbcada79c : SUCCESS in 3m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/502dd08cee2f46e48fbc3e645fef84be : SUCCESS in 3m 48s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/6e058137c53f4b60b2dedf2aefd9f17a : FAILURE in 1h 30m 51s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/855fda2de45d4ee5a482b6a7c7b0e214 : SUCCESS in 52m 03s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/c061cea49121462cbc0faa2c21c48801 : TIMED_OUT in 2h 04m 39s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/bf25d35c2bc84eb1bff8127ae53cbf1e : SUCCESS in 45m 53s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/453b75bb056249468676d3cf1aa8d699 : SUCCESS in 1h 51m 33s\n- neutron-tempest-plugin-tap-as-a-service-2026-2 https://zuul.opendev.org/t/openstack/build/083af0c3bb464d2eab2a2229e1cbc597 : TIMED_OUT in 2h 04m 27s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/e3f580ca80de49108a2bb687be21ed54 : SUCCESS in 54m 02s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/2bb6f40fbf6146df882cfdbe2fcb0e75 : SUCCESS in 2h 05m 14s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/e4db08e5425c43b39e4a8b1db9c0dba4 : SUCCESS in 1h 44m 52s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-2 https://zuul.opendev.org/t/openstack/build/2a9fd876fb994aed8973b58a200cdbd1 : SUCCESS in 50m 03s","accounts_in_message":[],"_revision_number":3},{"id":"26fe8a9e3f6cbc30deed4cbddf20bcac070aa243","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-10-01 09:32:55.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"590d0f330b16aadd7d607ca7fb21e5bdbcec6ad9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 11:43:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/030cbd4b9fb943a49a8a0b3f2c30ba46\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/840f8c7114b24812a86cd7dc9529069a : FAILURE in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c580eec8cf8f47098bbf750dc5cdd266 : SUCCESS in 3m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d0ae97865ea4568b0901274c62a3871 : SUCCESS in 3m 28s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/6805b7376f7548d2b6df7b3804a5255b : TIMED_OUT in 2h 03m 59s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/5c62c53791fe4ab1a641e40e0406f4eb : SUCCESS in 1h 42m 33s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/91ce2fff7da64e39b48c6d2c6873006b : SUCCESS in 1h 49m 02s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/1be4eb90c5564a2c8062855af12a7b98 : SUCCESS in 1h 42m 58s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/1976ff4fed584531ba512e019e908f5f : SUCCESS in 1h 08m 31s\n- neutron-tempest-plugin-tap-as-a-service-2026-2 https://zuul.opendev.org/t/openstack/build/0877c194e0e9439283b9cbf4b0aa453c : SUCCESS in 1h 40m 49s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/54c501ef33284f25984a132d39ba45a1 : SUCCESS in 1h 48m 27s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/c062cd26721f4bcf86b35c333c4601a9 : TIMED_OUT in 2h 04m 40s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/c0a19b0d9be04b23b938656e5eab1755 : SUCCESS in 54m 48s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-2 https://zuul.opendev.org/t/openstack/build/32ab93e884d94795ba52989c905691d4 : SUCCESS in 1h 53m 38s","accounts_in_message":[],"_revision_number":4},{"id":"59f8054d29efc8e06ef38b4bc7f8a7690c6699a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-10-02 09:46:59.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"ec0368211446cd4d5d38e32104394027ac6100ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 11:42:03.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3bd8fb5ea3fe45a2b1e83a5932dd771a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15e8fc28b1bc452e8bb65d30d9e2579f : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2a368e613fb44eaad186495da4a7e1f : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd67d7060e254cebaf49bdbc690b2486 : SUCCESS in 4m 03s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/8ab6a507baaf42788d8c344953a8159e : SUCCESS in 1h 49m 28s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/bbcfa0e569aa4fe58dae0cf956ca387b : SUCCESS in 54m 50s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/4a0997d7e39f4500bda8129fb6c677a4 : SUCCESS in 1h 44m 22s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/cbc74530fe9a4e2ea8c023325abb9fb9 : SUCCESS in 1h 40m 56s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/d1d79cc0aa1c45b1a140e8b6e31d32a3 : SUCCESS in 46m 22s\n- neutron-tempest-plugin-tap-as-a-service-2026-2 https://zuul.opendev.org/t/openstack/build/dff00eb879194514a09cba1e6f499af2 : SUCCESS in 1h 41m 37s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/01448a1089d04d02b63f004063d3e2f1 : SUCCESS in 1h 24m 28s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/d3cd632fe0774dfda817c8fadf59f202 : SUCCESS in 1h 32m 53s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/466d95b76f1d4f5ca809a461c27ee35a : SUCCESS in 52m 05s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-2 https://zuul.opendev.org/t/openstack/build/c0e09ca6882c4404909465c1154a1e83 : FAILURE in 1h 18m 49s","accounts_in_message":[],"_revision_number":5},{"id":"46b71b463290bef19f8c1ef5e620b98fda87d318","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-10-03 06:55:31.000000000","message":"Patch Set 5: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/e6d7812a881f4d5581fe6afcfe900b9e\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/4112d708e5434b8797a9154d2c750c2b : SUCCESS in 5h 36m 27s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/28ebc8235ed84f8aaf8408a97b319f26 : FAILURE in 3h 31m 41s","accounts_in_message":[],"_revision_number":5},{"id":"6ff4b649b09c4a1d343d319b819081427e768661","author":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"date":"2026-10-05 06:37:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c9e529b91a0ce8853ffde74308f95ad93f01be6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 08:12:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20feaaf4f4264238afcca60ed0edaf51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/137c71f43e414d24a360f49fdff2d0d6 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ff205a9ad444d25b2a27f66af4d87f0 : SUCCESS in 2m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/504e30af8ac84898a76aaa652bbe9b50 : SUCCESS in 3m 45s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/ac929360881a4c2786bfc6bb4a99b69b : SUCCESS in 45m 46s\n- neutron-tempest-plugin-tap-as-a-service-ovn https://zuul.opendev.org/t/openstack/build/ece23c745c654bf4b7ca045b66f49b99 : SUCCESS in 44m 19s\n- neutron-tempest-plugin-tap-as-a-service-2025-1 https://zuul.opendev.org/t/openstack/build/e5f154786865443199806b32c06e87ef : SUCCESS in 47m 18s\n- neutron-tempest-plugin-tap-as-a-service-2025-2 https://zuul.opendev.org/t/openstack/build/0e3e364bc2db47dca1e714683a0210a7 : SUCCESS in 46m 00s\n- neutron-tempest-plugin-tap-as-a-service-2026-1 https://zuul.opendev.org/t/openstack/build/29e46435b3a647c7abfc08c6fbb6d014 : SUCCESS in 48m 56s\n- neutron-tempest-plugin-tap-as-a-service-2026-2 https://zuul.opendev.org/t/openstack/build/411f015b46804d4dbf8109cd233e2d1a : SUCCESS in 40m 17s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/78ec4e492d7a490e9b720cb9a30d39ca : SUCCESS in 51m 25s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/82716f4dfbfc427f8d79ee9895d2393d : SUCCESS in 47m 22s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-1 https://zuul.opendev.org/t/openstack/build/468db82f3f5746b798911549316b7bc3 : SUCCESS in 1h 27m 02s\n- neutron-tempest-plugin-tap-as-a-service-ovn-2026-2 https://zuul.opendev.org/t/openstack/build/65e459d389c34be19a08e9e0f95e4700 : SUCCESS in 1h 33m 44s","accounts_in_message":[],"_revision_number":5},{"id":"cce9e8499dfba7d19a6865588395cf9b026c09f4","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-10-05 10:10:34.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/700dd844dbe74ceab42a2f4df26f1202\n\n- openstack-meta-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/9a954479fa9848d582ab18cf8782f1ef : SUCCESS in 3h 31m 29s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/8316a02e00f14e4cb82a6ead04b21594 : FAILURE in 3h 15m 27s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f66c0f6490a3c46090c67474c9920d975afe0c67","revisions":{"63f913e11f7060eb3b3341594f57c7f8ce8dbc38":{"kind":"REWORK","_number":1,"created":"2026-09-18 10:08:32.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/84/1006184/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/84/1006184/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/1006184/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/1006184/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/1006184/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/1006184/1"}}},"commit":{"parents":[{"commit":"2520421c72395c739a5ae9cc3cfab94ba1f58699","subject":"Merge \"bgp: Enable Neutron BGP multinode job in experimental queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2520421c72395c739a5ae9cc3cfab94ba1f58699"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-09-18 10:01:58.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-09-18 10:07:29.000000000","tz":330},"subject":"[TaaS] Add MTU test to Tap-as-a-Service.","message":"[TaaS] Add MTU test to Tap-as-a-Service.\n\nThis patch intends to add a test that tests\nthe tap mirroring with the highest possible MTU\nwithin the deployment.\n\nAssisted-By: Claude Opus 5\nChange-Id: Ia9e211200df7c662223afe81e06b5280126b2ef1\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/63f913e11f7060eb3b3341594f57c7f8ce8dbc38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/63f913e11f7060eb3b3341594f57c7f8ce8dbc38"}]},"branch":"refs/heads/master"},"138fa7876ac1dc186183712fe65064d4915d47f1":{"kind":"REWORK","_number":2,"created":"2026-09-19 03:07:01.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/84/1006184/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/84/1006184/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/1006184/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/1006184/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/1006184/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/1006184/2"}}},"commit":{"parents":[{"commit":"2520421c72395c739a5ae9cc3cfab94ba1f58699","subject":"Merge \"bgp: Enable Neutron BGP multinode job in experimental queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2520421c72395c739a5ae9cc3cfab94ba1f58699"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-09-18 10:01:58.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-09-19 03:06:48.000000000","tz":330},"subject":"[TaaS] Add MTU test to Tap-as-a-Service.","message":"[TaaS] Add MTU test to Tap-as-a-Service.\n\nThis patch intends to add a test that tests\nthe tap mirroring with the highest possible MTU\nwithin the deployment.\n\nAssisted-By: Claude Opus 5\nChange-Id: Ia9e211200df7c662223afe81e06b5280126b2ef1\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/138fa7876ac1dc186183712fe65064d4915d47f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/138fa7876ac1dc186183712fe65064d4915d47f1"}]},"branch":"refs/heads/master"},"9d5ca1024bc32d3f18c759f39b3cbf4df759ff2f":{"kind":"REWORK","_number":3,"created":"2026-09-30 08:36:29.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/84/1006184/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/84/1006184/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/1006184/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/1006184/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/1006184/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/1006184/3"}}},"commit":{"parents":[{"commit":"2520421c72395c739a5ae9cc3cfab94ba1f58699","subject":"Merge \"bgp: Enable Neutron BGP multinode job in experimental queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2520421c72395c739a5ae9cc3cfab94ba1f58699"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-09-18 10:01:58.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-09-30 08:36:17.000000000","tz":330},"subject":"[TaaS] Add MTU and Cross-tenant test to Tap-as-a-Service.","message":"[TaaS] Add MTU and Cross-tenant test to Tap-as-a-Service.\n\nThis patch intends to adds tests that :\n1. Test the tap mirroring with the highest possible MTU\nwithin the deployment.\n2. Test the tap mirror works for multiple tenant.\n3. Test the other tenants cannot access tap traffic of\nsource VM.\n\nAssisted-By: Claude Opus 5\nChange-Id: Ia9e211200df7c662223afe81e06b5280126b2ef1\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9d5ca1024bc32d3f18c759f39b3cbf4df759ff2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9d5ca1024bc32d3f18c759f39b3cbf4df759ff2f"}]},"branch":"refs/heads/master"},"d5e08cfd70516163b813fac51a41e4c91f0fd85c":{"kind":"REWORK","_number":4,"created":"2026-10-01 09:32:55.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/84/1006184/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/84/1006184/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/1006184/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/1006184/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/1006184/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/1006184/4"}}},"commit":{"parents":[{"commit":"2520421c72395c739a5ae9cc3cfab94ba1f58699","subject":"Merge \"bgp: Enable Neutron BGP multinode job in experimental queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2520421c72395c739a5ae9cc3cfab94ba1f58699"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-09-18 10:01:58.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-10-01 09:32:42.000000000","tz":330},"subject":"[TaaS] Add MTU and Cross-tenant test to Tap-as-a-Service.","message":"[TaaS] Add MTU and Cross-tenant test to Tap-as-a-Service.\n\nThis patch intends to adds tests that :\n1. Test the tap mirroring with the highest possible MTU\nwithin the deployment.\n2. Test the tap mirror works for multiple tenant.\n3. Test the other tenants cannot access tap traffic of\nsource VM.\n\nAssisted-By: Claude Opus 5\nChange-Id: Ia9e211200df7c662223afe81e06b5280126b2ef1\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d5e08cfd70516163b813fac51a41e4c91f0fd85c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d5e08cfd70516163b813fac51a41e4c91f0fd85c"}]},"branch":"refs/heads/master"},"f66c0f6490a3c46090c67474c9920d975afe0c67":{"kind":"REWORK","_number":5,"created":"2026-10-02 09:46:59.000000000","uploader":{"_account_id":35308,"name":"Bharath M V","email":"bmv@redhat.com","username":"bmv"},"ref":"refs/changes/84/1006184/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/84/1006184/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/1006184/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/1006184/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/1006184/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/84/1006184/5"}}},"commit":{"parents":[{"commit":"2520421c72395c739a5ae9cc3cfab94ba1f58699","subject":"Merge \"bgp: Enable Neutron BGP multinode job in experimental queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2520421c72395c739a5ae9cc3cfab94ba1f58699"}]}],"author":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-09-18 10:01:58.000000000","tz":330},"committer":{"name":"Bharath M V","email":"bmv@redhat.com","date":"2026-10-02 09:46:43.000000000","tz":330},"subject":"[TaaS] Add MTU and Cross-tenant test to Tap-as-a-Service.","message":"[TaaS] Add MTU and Cross-tenant test to Tap-as-a-Service.\n\nThis patch intends to adds tests that :\n1. Test the tap mirroring with the highest possible MTU\nwithin the deployment.\n2. Test the tap mirror works for multiple tenant.\n3. Test the other tenants cannot access tap traffic of\nsource VM.\n\nAssisted-By: Claude Opus 5\nChange-Id: Ia9e211200df7c662223afe81e06b5280126b2ef1\nSigned-off-by: Bharath M V \u003cbmv@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f66c0f6490a3c46090c67474c9920d975afe0c67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f66c0f6490a3c46090c67474c9920d975afe0c67"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
