)]}'
{"id":"openstack%2Fneutron-tempest-plugin~584229","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ida6e75d0215283c15b7a1c0ce0f473169afe8a66","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/1780094","hashtags":[],"change_id":"Ida6e75d0215283c15b7a1c0ce0f473169afe8a66","subject":"Add api test for DVR coverting back to centralized","status":"MERGED","created":"2018-07-20 05:22:26.000000000","updated":"2019-04-26 04:36:18.000000000","submitted":"2019-04-26 04:36:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"584229-1556253378752-36e30e48","meta_rev_id":"a994ef0946bb10224533cc56b636d253f3f18c01","_number":584229,"virtual_id_number":584229,"owner":{"_account_id":28373,"name":"Kailun Qin","email":"kailun.qin@hotmail.com","username":"kailun.qin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"date":"2019-04-25 04:15:48.000000000","_account_id":28373,"name":"Kailun Qin","email":"kailun.qin@hotmail.com","username":"kailun.qin"},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-04-26 04:36:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"all":[{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":28373,"name":"Kailun Qin","email":"kailun.qin@hotmail.com","username":"kailun.qin"},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":2,"date":"2019-04-25 21:34:34.000000000","_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2019-04-25 21:46:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":28373,"name":"Kailun Qin","email":"kailun.qin@hotmail.com","username":"kailun.qin"},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":1,"date":"2019-04-25 21:46:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":28373,"name":"Kailun Qin","email":"kailun.qin@hotmail.com","username":"kailun.qin"},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"_account_id":28373,"name":"Kailun Qin","email":"kailun.qin@hotmail.com","username":"kailun.qin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-07-20 08:27:15.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2018-07-20 08:27:25.000000000","updated_by":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"reviewer":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"state":"REVIEWER"},{"updated":"2018-07-20 08:27:35.000000000","updated_by":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"reviewer":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"state":"REVIEWER"},{"updated":"2019-04-18 14:35:27.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2019-04-18 14:38:45.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-04-20 07:33:34.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-04-22 11:18:20.000000000","updated_by":{"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"},"reviewer":{"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"},"state":"REVIEWER"},{"updated":"2019-04-22 17:04:48.000000000","updated_by":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"reviewer":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"state":"REVIEWER"},{"updated":"2019-04-25 21:34:34.000000000","updated_by":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"reviewer":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"state":"REVIEWER"},{"updated":"2019-04-25 21:46:55.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-04-26 04:36:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b50d17016f76e46e2451ca651872ff15e85d990","author":{"_account_id":28373,"name":"Kailun Qin","email":"kailun.qin@hotmail.com","username":"kailun.qin"},"date":"2018-07-20 05:22:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d83ad39b5ede0dafeb1eeb4eb8484fb122089f2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-20 08:06:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/29/584229/1/check/openstack-tox-pep8/2464be0/ : SUCCESS in 4m 53s\n- neutron-tempest-plugin-api http://logs.openstack.org/29/584229/1/check/neutron-tempest-plugin-api/3dbecfc/ : SUCCESS in 49m 40s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/29/584229/1/check/neutron-tempest-plugin-designate-scenario/be7257b/ : SUCCESS in 29m 21s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/29/584229/1/check/neutron-tempest-plugin-dvr-multinode-scenario/e8186d5/ : SUCCESS in 2h 42m 20s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/29/584229/1/check/neutron-tempest-plugin-scenario-linuxbridge/480e8cd/ : SUCCESS in 1h 38m 57s\n- build-openstack-sphinx-docs http://logs.openstack.org/29/584229/1/check/build-openstack-sphinx-docs/6ef8640/html/ : SUCCESS in 5m 22s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/29/584229/1/check/neutron-tempest-plugin-api-queens/e1af942/ : FAILURE in 49m 02s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/29/584229/1/check/neutron-tempest-plugin-designate-scenario-queens/2a20a88/ : SUCCESS in 31m 33s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/29/584229/1/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/92fe768/ : FAILURE in 2h 14m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/29/584229/1/check/neutron-tempest-plugin-scenario-linuxbridge-queens/386c366/ : SUCCESS in 1h 26m 21s","accounts_in_message":[],"_revision_number":1},{"id":"3e57f25e9abd63a7c1c45d829886d8786cc95571","author":{"_account_id":28373,"name":"Kailun Qin","email":"kailun.qin@hotmail.com","username":"kailun.qin"},"date":"2018-07-20 08:28:32.000000000","message":"Patch Set 1:\n\nNeed to cherry-pick commit 886e24155358afe7ad167ba1602d3939d68c3d5f to stable/queens to pass neutron-tempest-plugin-api-queens Zuul check.","accounts_in_message":[],"_revision_number":1},{"id":"fa95015ea413b33c2d7c779ea71f32bbac6dd3a3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-07-20 10:27:50.000000000","message":"Patch Set 1: Code-Review-1\n\nLooks good for me. Let\u0027s wait for cherry-pick of fix to stable/queens and we can go with this one IMO :)\n-1 for now only because this test is failing for queens branch now","accounts_in_message":[],"_revision_number":1},{"id":"512b0906fdc3d9b326940116fa88e5600bafa1e8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-08-30 21:45:02.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Jenkins the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1},{"id":"d8889f29196d6abcb271b798f58842f20dc09e1d","author":{"_account_id":28373,"name":"Kailun Qin","email":"kailun.qin@hotmail.com","username":"kailun.qin"},"date":"2018-08-31 01:47:36.000000000","message":"Restored\n\nI tried to cherry-pick of fix to stable/queens however quite a few dependencies like [1] are hard to be all backported to stable/queens. So I guess waiting for zuul check moving forward to use api-rocky might be a better idea.\n\n[1] https://review.openstack.org/#/c/591673/","accounts_in_message":[],"_revision_number":1},{"id":"4caa67a18391f703d40af34fb1d9039a230dc5a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-31 03:40:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/29/584229/1/check/openstack-tox-pep8/8379dc3/ : SUCCESS in 4m 12s\n- neutron-tempest-plugin-api http://logs.openstack.org/29/584229/1/check/neutron-tempest-plugin-api/46b6b52/ : SUCCESS in 40m 35s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/29/584229/1/check/neutron-tempest-plugin-designate-scenario/be551e4/ : FAILURE in 48m 17s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/29/584229/1/check/neutron-tempest-plugin-dvr-multinode-scenario/c0f9b4b/ : FAILURE in 5m 43s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/29/584229/1/check/neutron-tempest-plugin-scenario-linuxbridge/21a0aa0/ : SUCCESS in 1h 45m 16s\n- build-openstack-sphinx-docs http://logs.openstack.org/29/584229/1/check/build-openstack-sphinx-docs/43becc3/html/ : SUCCESS in 4m 29s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/29/584229/1/check/neutron-tempest-plugin-api-queens/9adc6f3/ : FAILURE in 48m 44s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/29/584229/1/check/neutron-tempest-plugin-designate-scenario-queens/aa29c17/ : SUCCESS in 30m 02s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/29/584229/1/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/698f322/ : FAILURE in 6m 06s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/29/584229/1/check/neutron-tempest-plugin-scenario-linuxbridge-queens/354e22f/ : FAILURE in 1h 51m 15s","accounts_in_message":[],"_revision_number":1},{"id":"6c12906aaf8a88a07d29df2755d8e7477894e892","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-16 08:39:40.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Jenkins the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1},{"id":"6dcf23d70788feb8b46a3f130bdb1106832f6631","author":{"_account_id":28373,"name":"Kailun Qin","email":"kailun.qin@hotmail.com","username":"kailun.qin"},"date":"2019-04-18 03:25:23.000000000","message":"Restored\n\nreactive the patch, will back port 886e24155358afe7ad167ba1602d3939d68c3d5f to stable/queens to pass neutron-tempest-plugin-api-queens Zuul check.","accounts_in_message":[],"_revision_number":1},{"id":"d37f505d5c38eae947b3d8c2e0fafd52aaff7a7e","author":{"_account_id":28373,"name":"Kailun Qin","email":"kailun.qin@hotmail.com","username":"kailun.qin"},"date":"2019-04-18 04:13:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4fbd2ec9841bb574cd21b182396209716d1097c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 14:17:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/29/584229/2/check/neutron-tempest-plugin-api/6025a41/ : SUCCESS in 54m 49s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/29/584229/2/check/neutron-tempest-plugin-designate-scenario/8e31262/ : SUCCESS in 37m 22s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/29/584229/2/check/neutron-tempest-plugin-dvr-multinode-scenario/9c87618/ : FAILURE in 2h 00m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/29/584229/2/check/neutron-tempest-plugin-scenario-linuxbridge/d7a149b/ : SUCCESS in 1h 09m 45s\n- build-openstack-sphinx-docs http://logs.openstack.org/29/584229/2/check/build-openstack-sphinx-docs/cdd9c86/html/ : SUCCESS in 5m 33s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/29/584229/2/check/neutron-tempest-plugin-api-queens/35d053a/ : SUCCESS in 49m 16s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/29/584229/2/check/neutron-tempest-plugin-designate-scenario-queens/72ff436/ : SUCCESS in 34m 05s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/29/584229/2/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/29ab21e/ : FAILURE in 1h 23m 01s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/29/584229/2/check/neutron-tempest-plugin-scenario-linuxbridge-queens/799f258/ : SUCCESS in 55m 50s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/29/584229/2/check/neutron-tempest-plugin-api-rocky/d49cb36/ : SUCCESS in 48m 59s\n- neutron-tempest-plugin-designate-scenario-rocky http://logs.openstack.org/29/584229/2/check/neutron-tempest-plugin-designate-scenario-rocky/40b6770/ : SUCCESS in 33m 24s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky http://logs.openstack.org/29/584229/2/check/neutron-tempest-plugin-dvr-multinode-scenario-rocky/a6aceed/ : FAILURE in 1h 41m 55s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky http://logs.openstack.org/29/584229/2/check/neutron-tempest-plugin-scenario-linuxbridge-rocky/f4e5549/ : SUCCESS in 54m 02s\n- neutron-tempest-plugin-api-stein http://logs.openstack.org/29/584229/2/check/neutron-tempest-plugin-api-stein/3e3b435/ : SUCCESS in 54m 27s\n- neutron-tempest-plugin-designate-scenario-stein http://logs.openstack.org/29/584229/2/check/neutron-tempest-plugin-designate-scenario-stein/34fa3a6/ : SUCCESS in 32m 54s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein http://logs.openstack.org/29/584229/2/check/neutron-tempest-plugin-dvr-multinode-scenario-stein/55b5a7d/ : FAILURE in 1h 52m 46s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein http://logs.openstack.org/29/584229/2/check/neutron-tempest-plugin-scenario-linuxbridge-stein/86dbbde/ : SUCCESS in 1h 12m 09s\n- openstack-tox-pep8 http://logs.openstack.org/29/584229/2/check/openstack-tox-pep8/fd08160/ : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":2},{"id":"30a02aa86dd5c0ba48875bf2ad00a851eaef27f0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-20 07:33:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9634cc2f5aa547c21d3e89856ef635f8986f7f83","author":{"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"},"date":"2019-04-22 11:18:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b2cd72c7767d04c4af6d2fb291b5c60821db63d9","author":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"date":"2019-04-22 17:04:48.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d6ecbd16a099c9757ce6ad3c6f84ca20ccb7bdc7","author":{"_account_id":28373,"name":"Kailun Qin","email":"kailun.qin@hotmail.com","username":"kailun.qin"},"date":"2019-04-24 10:49:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5911dd52863820b22125b2ca7fd75d8010a4108b","author":{"_account_id":28373,"name":"Kailun Qin","email":"kailun.qin@hotmail.com","username":"kailun.qin"},"date":"2019-04-24 10:49:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bdff2e470e8a41c26ad95072232b82713093deb6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 16:42:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/29/584229/3/check/neutron-tempest-plugin-api/8a5a053/ : SUCCESS in 50m 26s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/29/584229/3/check/neutron-tempest-plugin-designate-scenario/32f1f2f/ : SUCCESS in 31m 06s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/29/584229/3/check/neutron-tempest-plugin-dvr-multinode-scenario/ba79f82/ : FAILURE in 1h 59m 34s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/29/584229/3/check/neutron-tempest-plugin-scenario-linuxbridge/de79ec4/ : SUCCESS in 1h 06m 23s\n- build-openstack-sphinx-docs http://logs.openstack.org/29/584229/3/check/build-openstack-sphinx-docs/a7b99f9/html/ : SUCCESS in 5m 02s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/29/584229/3/check/neutron-tempest-plugin-api-queens/5bba9c1/ : FAILURE in 46m 15s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/29/584229/3/check/neutron-tempest-plugin-designate-scenario-queens/b7efe48/ : SUCCESS in 27m 18s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/29/584229/3/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/153d4eb/ : FAILURE in 1h 10m 06s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/29/584229/3/check/neutron-tempest-plugin-scenario-linuxbridge-queens/5721e38/ : SUCCESS in 57m 28s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/29/584229/3/check/neutron-tempest-plugin-api-rocky/a997dea/ : SUCCESS in 46m 40s\n- neutron-tempest-plugin-designate-scenario-rocky http://logs.openstack.org/29/584229/3/check/neutron-tempest-plugin-designate-scenario-rocky/4177a7c/ : SUCCESS in 33m 01s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky http://logs.openstack.org/29/584229/3/check/neutron-tempest-plugin-dvr-multinode-scenario-rocky/b711cfd/ : SUCCESS in 1h 18m 20s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky http://logs.openstack.org/29/584229/3/check/neutron-tempest-plugin-scenario-linuxbridge-rocky/720e2fa/ : SUCCESS in 1h 06m 14s\n- neutron-tempest-plugin-api-stein http://logs.openstack.org/29/584229/3/check/neutron-tempest-plugin-api-stein/100acef/ : SUCCESS in 55m 11s\n- neutron-tempest-plugin-designate-scenario-stein http://logs.openstack.org/29/584229/3/check/neutron-tempest-plugin-designate-scenario-stein/25ab425/ : SUCCESS in 33m 39s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein http://logs.openstack.org/29/584229/3/check/neutron-tempest-plugin-dvr-multinode-scenario-stein/e425bca/ : FAILURE in 1h 32m 32s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein http://logs.openstack.org/29/584229/3/check/neutron-tempest-plugin-scenario-linuxbridge-stein/7524e97/ : SUCCESS in 58m 03s\n- openstack-tox-pep8 http://logs.openstack.org/29/584229/3/check/openstack-tox-pep8/4923fb6/ : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":3},{"id":"741c61b2d08fe56a6abecadb2bd77bb1ebffff5b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-04-25 01:09:42.000000000","message":"Patch Set 3:\n\nThe stable/queens change has not merged yet, close...","accounts_in_message":[],"_revision_number":3},{"id":"47c2cfa09dd84175a8b2dad1cb2bc97c69847bb7","author":{"_account_id":28373,"name":"Kailun Qin","email":"kailun.qin@hotmail.com","username":"kailun.qin"},"date":"2019-04-25 04:14:07.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"3a799dce57011fd56364ebbef6d36c53afebf28f","author":{"_account_id":28373,"name":"Kailun Qin","email":"kailun.qin@hotmail.com","username":"kailun.qin"},"date":"2019-04-25 04:15:48.000000000","message":"Patch Set 4:\n\nThe stable/queens change has been merged: https://review.opendev.org/#/c/591867/.","accounts_in_message":[],"_revision_number":4},{"id":"89f4bb835cc3110656e148e24018d3dbf2f36faf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 11:17:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/29/584229/4/check/neutron-tempest-plugin-api/33a4401/ : SUCCESS in 52m 55s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/29/584229/4/check/neutron-tempest-plugin-designate-scenario/0069ac6/ : SUCCESS in 34m 27s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/29/584229/4/check/neutron-tempest-plugin-dvr-multinode-scenario/601def6/ : FAILURE in 1h 53m 35s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/29/584229/4/check/neutron-tempest-plugin-scenario-linuxbridge/925c26e/ : SUCCESS in 1h 21m 04s\n- build-openstack-sphinx-docs http://logs.openstack.org/29/584229/4/check/build-openstack-sphinx-docs/ec039d8/html/ : SUCCESS in 4m 06s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/29/584229/4/check/neutron-tempest-plugin-api-queens/c8245c2/ : SUCCESS in 49m 50s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/29/584229/4/check/neutron-tempest-plugin-designate-scenario-queens/2066af4/ : SUCCESS in 27m 24s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/29/584229/4/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/7f061a1/ : FAILURE in 1h 22m 57s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/29/584229/4/check/neutron-tempest-plugin-scenario-linuxbridge-queens/1c41f46/ : SUCCESS in 47m 24s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/29/584229/4/check/neutron-tempest-plugin-api-rocky/ebbb252/ : SUCCESS in 48m 30s\n- neutron-tempest-plugin-designate-scenario-rocky http://logs.openstack.org/29/584229/4/check/neutron-tempest-plugin-designate-scenario-rocky/e456789/ : SUCCESS in 30m 06s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky http://logs.openstack.org/29/584229/4/check/neutron-tempest-plugin-dvr-multinode-scenario-rocky/71c7ae1/ : FAILURE in 1h 19m 57s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky http://logs.openstack.org/29/584229/4/check/neutron-tempest-plugin-scenario-linuxbridge-rocky/a99ccd9/ : SUCCESS in 55m 40s\n- neutron-tempest-plugin-api-stein http://logs.openstack.org/29/584229/4/check/neutron-tempest-plugin-api-stein/89f47ca/ : SUCCESS in 50m 15s\n- neutron-tempest-plugin-designate-scenario-stein http://logs.openstack.org/29/584229/4/check/neutron-tempest-plugin-designate-scenario-stein/da19be8/ : SUCCESS in 34m 34s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein http://logs.openstack.org/29/584229/4/check/neutron-tempest-plugin-dvr-multinode-scenario-stein/d915330/ : FAILURE in 1h 39m 34s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein http://logs.openstack.org/29/584229/4/check/neutron-tempest-plugin-scenario-linuxbridge-stein/6cff6c4/ : SUCCESS in 1h 00m 28s\n- openstack-tox-pep8 http://logs.openstack.org/29/584229/4/check/openstack-tox-pep8/ce7478d/ : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":4},{"id":"534c08cd9c1c4bb1c3872ce9e32441a172607144","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2019-04-25 21:34:34.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"d8dce63db3311edf780e1973781d916b95cb08e1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-04-25 21:46:55.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"a911d9972d8dfc76a7523886ee1430a625d78c8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 21:47:27.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c32349a45723d56ed5d83d8adf505abd0c262122","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 04:36:18.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/29/584229/4/gate/neutron-tempest-plugin-api/e5f142d/ : SUCCESS in 51m 21s\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/29/584229/4/gate/neutron-tempest-plugin-scenario-linuxbridge/0f39c90/ : SUCCESS in 1h 08m 12s\n- build-openstack-sphinx-docs http://logs.openstack.org/29/584229/4/gate/build-openstack-sphinx-docs/72df0f4/html/ : SUCCESS in 4m 30s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/29/584229/4/gate/neutron-tempest-plugin-api-queens/1985ec8/ : SUCCESS in 47m 30s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/29/584229/4/gate/neutron-tempest-plugin-api-rocky/04a99eb/ : SUCCESS in 51m 54s\n- neutron-tempest-plugin-api-stein http://logs.openstack.org/29/584229/4/gate/neutron-tempest-plugin-api-stein/efd6033/ : SUCCESS in 48m 58s\n- openstack-tox-pep8 http://logs.openstack.org/29/584229/4/gate/openstack-tox-pep8/a851837/ : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":4},{"id":"c57a555c2fdfdacf9a983b270c8ffd15dd63818f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 04:36:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ce246d0b487d9ddc66aa0abe749ab54fca8476a9","revisions":{"ebc449f04bea2670e7d9083dd99f8d4dbbc79313":{"kind":"REWORK","_number":1,"created":"2018-07-20 05:22:26.000000000","uploader":{"_account_id":28373,"name":"Kailun Qin","email":"kailun.qin@hotmail.com","username":"kailun.qin"},"ref":"refs/changes/29/584229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/29/584229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/584229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/584229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/584229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/584229/1"}}},"commit":{"parents":[{"commit":"3e1912286ea7a9274ea229fff31ee0484dc23638","subject":"Merge \"Use latest Ubuntu 16.04 LTS guest image.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3e1912286ea7a9274ea229fff31ee0484dc23638"}]}],"author":{"name":"Kailun Qin","email":"kailun.qin@intel.com","date":"2018-07-20 13:38:07.000000000","tz":480},"committer":{"name":"Kailun Qin","email":"kailun.qin@intel.com","date":"2018-07-20 14:19:46.000000000","tz":480},"subject":"Add api test for DVR coverting back to centralized","message":"Add api test for DVR coverting back to centralized\n\nThis patch adds a test case for DVR routers which converts a centralized\nrouter to distributed firstly, followed by making a transition back to\ncentralized. It checks whether the attributes are updated or retained as\nexpected.\n\nRelated-Bug: #1780094\nChange-Id: Ida6e75d0215283c15b7a1c0ce0f473169afe8a66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ebc449f04bea2670e7d9083dd99f8d4dbbc79313"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ebc449f04bea2670e7d9083dd99f8d4dbbc79313"}]},"branch":"refs/heads/master"},"e0ec76878ffbb33908fe921676238e38b614e002":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2019-04-18 04:13:37.000000000","uploader":{"_account_id":28373,"name":"Kailun Qin","email":"kailun.qin@hotmail.com","username":"kailun.qin"},"ref":"refs/changes/29/584229/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/29/584229/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/584229/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/584229/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/584229/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/584229/2"}}},"commit":{"parents":[{"commit":"a57f80d1809a7d2a58e36ccd7d82c082f0be7080","subject":"Add definition of jobs running on stable/stein branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a57f80d1809a7d2a58e36ccd7d82c082f0be7080"}]}],"author":{"name":"Kailun Qin","email":"kailun.qin@intel.com","date":"2018-07-20 13:38:07.000000000","tz":480},"committer":{"name":"Kailun Qin","email":"kailun.qin@intel.com","date":"2019-04-18 13:06:09.000000000","tz":480},"subject":"Add api test for DVR coverting back to centralized","message":"Add api test for DVR coverting back to centralized\n\nThis patch adds a test case for DVR routers which converts a centralized\nrouter to distributed firstly, followed by making a transition back to\ncentralized. It checks whether the attributes are updated or retained as\nexpected.\n\nRelated-Bug: #1780094\nChange-Id: Ida6e75d0215283c15b7a1c0ce0f473169afe8a66\nDepends-On: https://review.openstack.org/591867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e0ec76878ffbb33908fe921676238e38b614e002"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e0ec76878ffbb33908fe921676238e38b614e002"}]},"branch":"refs/heads/master"},"47b08b62bb689e75dda792bccb6649799b1222eb":{"kind":"REWORK","_number":3,"created":"2019-04-24 10:49:29.000000000","uploader":{"_account_id":28373,"name":"Kailun Qin","email":"kailun.qin@hotmail.com","username":"kailun.qin"},"ref":"refs/changes/29/584229/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/29/584229/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/584229/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/584229/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/584229/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/584229/3"}}},"commit":{"parents":[{"commit":"46be4325b57ba58cd3a20c33789eff336d2973cb","subject":"Merge \"Add additional config option \"default_image_is_advanced\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/46be4325b57ba58cd3a20c33789eff336d2973cb"}]}],"author":{"name":"Kailun Qin","email":"kailun.qin@intel.com","date":"2018-07-20 13:38:07.000000000","tz":480},"committer":{"name":"Kailun Qin","email":"kailun.qin@intel.com","date":"2019-04-24 19:44:54.000000000","tz":480},"subject":"Add api test for DVR coverting back to centralized","message":"Add api test for DVR coverting back to centralized\n\nThis patch adds a test case for DVR routers which converts a centralized\nrouter to distributed firstly, followed by making a transition back to\ncentralized. It checks whether the attributes are updated or retained as\nexpected.\n\nRelated-Bug: #1780094\nChange-Id: Ida6e75d0215283c15b7a1c0ce0f473169afe8a66\nDepends-On: https://review.openstack.org/591867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/47b08b62bb689e75dda792bccb6649799b1222eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/47b08b62bb689e75dda792bccb6649799b1222eb"}]},"branch":"refs/heads/master"},"ce246d0b487d9ddc66aa0abe749ab54fca8476a9":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-04-25 04:14:07.000000000","uploader":{"_account_id":28373,"name":"Kailun Qin","email":"kailun.qin@hotmail.com","username":"kailun.qin"},"ref":"refs/changes/29/584229/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/29/584229/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/584229/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/584229/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/584229/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/29/584229/4"}}},"commit":{"parents":[{"commit":"46be4325b57ba58cd3a20c33789eff336d2973cb","subject":"Merge \"Add additional config option \"default_image_is_advanced\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/46be4325b57ba58cd3a20c33789eff336d2973cb"}]}],"author":{"name":"Kailun Qin","email":"kailun.qin@intel.com","date":"2018-07-20 13:38:07.000000000","tz":480},"committer":{"name":"Kailun Qin","email":"kailun.qin@intel.com","date":"2019-04-25 13:09:01.000000000","tz":480},"subject":"Add api test for DVR coverting back to centralized","message":"Add api test for DVR coverting back to centralized\n\nThis patch adds a test case for DVR routers which converts a centralized\nrouter to distributed firstly, followed by making a transition back to\ncentralized. It checks whether the attributes are updated or retained as\nexpected.\n\nRelated-Bug: #1780094\nChange-Id: Ida6e75d0215283c15b7a1c0ce0f473169afe8a66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ce246d0b487d9ddc66aa0abe749ab54fca8476a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ce246d0b487d9ddc66aa0abe749ab54fca8476a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
