)]}'
{"id":"openstack%2Fneutron-tempest-plugin~618557","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I47cc5656ea2e1e6e085d4d6602dbc76f87463cc3","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/1717302","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-08-28 23:43:20.000000000","reason":"Change was submitted"},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-08-26 23:44:39.000000000","reason":"\u003cGERRIT_ACCOUNT_9656\u003e replied on the change","reason_account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2024-08-28 23:43:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I47cc5656ea2e1e6e085d4d6602dbc76f87463cc3","subject":"Remove unstable_test from FIP test_east_west","status":"MERGED","created":"2018-11-16 16:29:31.000000000","updated":"2024-08-28 23:43:20.000000000","submitted":"2024-08-28 23:43:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"618557-bug/1717302","meta_rev_id":"f4d3d5a56fc70d8e37232aeabcad9841ab5bea28","_number":618557,"virtual_id_number":618557,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-28 23:43:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2024-08-28 00:11:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2024-08-27 15:16:52.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":2,"date":"2024-08-28 19:07:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":1,"date":"2024-08-28 19:07:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-11-19 17:05:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-11-19 21:40:57.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-08-26 21:15:34.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2024-08-28 00:11:34.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"}],"messages":[{"id":"9b7969c922ef27b7076d7a4b3a823590814c582a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-11-16 16:29:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80f068217b91e7e418bb0686fb84ffd5510debf4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-16 19:00:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-api/e6bd131/ : SUCCESS in 48m 28s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-designate-scenario/9017d2f/ : SUCCESS in 33m 53s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-dvr-multinode-scenario/ce48f7d/ : FAILURE in 2h 28m 58s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-scenario-linuxbridge/a081266/ : SUCCESS in 1h 01m 29s\n- build-openstack-sphinx-docs http://logs.openstack.org/57/618557/1/check/build-openstack-sphinx-docs/b3bf73b/html/ : SUCCESS in 4m 37s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-api-queens/31c9253/ : SUCCESS in 46m 48s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-designate-scenario-queens/1957a06/ : SUCCESS in 25m 23s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/b2d34fc/ : FAILURE in 2h 12m 47s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-scenario-linuxbridge-queens/b1b5d72/ : SUCCESS in 56m 50s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-api-rocky/7fe54d7/ : SUCCESS in 47m 05s\n- neutron-tempest-plugin-designate-scenario-rocky http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-designate-scenario-rocky/0d3cfdd/ : SUCCESS in 36m 38s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-dvr-multinode-scenario-rocky/ea68c5c/ : SUCCESS in 1h 39m 55s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-scenario-linuxbridge-rocky/94d6702/ : SUCCESS in 51m 29s\n- openstack-tox-pep8 http://logs.openstack.org/57/618557/1/check/openstack-tox-pep8/c21e586/ : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":1},{"id":"4d7a065bb18888615242c551c0f87d9fbd7013cc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-11-16 19:57:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"faf7cf87fdbf7de62b37651494d9d6e492489696","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-16 22:59:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-api/af38825/ : SUCCESS in 48m 04s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-designate-scenario/29dbe17/ : SUCCESS in 32m 35s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-dvr-multinode-scenario/3b38eee/ : FAILURE in 3h 01m 06s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-scenario-linuxbridge/db3cd6f/ : SUCCESS in 44m 14s\n- build-openstack-sphinx-docs http://logs.openstack.org/57/618557/1/check/build-openstack-sphinx-docs/d3f0130/html/ : SUCCESS in 4m 14s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-api-queens/683e1ff/ : SUCCESS in 46m 14s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-designate-scenario-queens/c4c55f0/ : SUCCESS in 28m 33s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/f0ba871/ : FAILURE in 1h 22m 04s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-scenario-linuxbridge-queens/cf8fa89/ : SUCCESS in 1h 08m 49s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-api-rocky/274222c/ : SUCCESS in 38m 34s\n- neutron-tempest-plugin-designate-scenario-rocky http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-designate-scenario-rocky/8a98c75/ : SUCCESS in 31m 31s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-dvr-multinode-scenario-rocky/1f98da4/ : FAILURE in 1h 44m 13s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky http://logs.openstack.org/57/618557/1/check/neutron-tempest-plugin-scenario-linuxbridge-rocky/fb1ff9e/ : SUCCESS in 45m 33s\n- openstack-tox-pep8 http://logs.openstack.org/57/618557/1/check/openstack-tox-pep8/d470753/ : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"75f07dbd1ecdb50e8fa18bed6401078c6c7d4aa7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-11-19 14:11:42.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6c2c86c69e0b1fd0b19b550e24275538632eade4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-19 17:05:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/57/618557/2/check/neutron-tempest-plugin-api/c311e79/ : SUCCESS in 45m 44s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/57/618557/2/check/neutron-tempest-plugin-designate-scenario/0fde7aa/ : SUCCESS in 30m 00s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/57/618557/2/check/neutron-tempest-plugin-dvr-multinode-scenario/5d0f012/ : FAILURE in 2h 03m 16s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/57/618557/2/check/neutron-tempest-plugin-scenario-linuxbridge/821c97f/ : SUCCESS in 1h 05m 13s\n- build-openstack-sphinx-docs http://logs.openstack.org/57/618557/2/check/build-openstack-sphinx-docs/6b7e3c4/html/ : SUCCESS in 4m 27s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/57/618557/2/check/neutron-tempest-plugin-api-queens/affc9b7/ : SUCCESS in 48m 05s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/57/618557/2/check/neutron-tempest-plugin-designate-scenario-queens/6889560/ : SUCCESS in 30m 22s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/57/618557/2/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/58d58f9/ : FAILURE in 1h 27m 59s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/57/618557/2/check/neutron-tempest-plugin-scenario-linuxbridge-queens/0237361/ : SUCCESS in 53m 31s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/57/618557/2/check/neutron-tempest-plugin-api-rocky/9596ec2/ : SUCCESS in 49m 23s\n- neutron-tempest-plugin-designate-scenario-rocky http://logs.openstack.org/57/618557/2/check/neutron-tempest-plugin-designate-scenario-rocky/21ac0eb/ : SUCCESS in 35m 01s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky http://logs.openstack.org/57/618557/2/check/neutron-tempest-plugin-dvr-multinode-scenario-rocky/573d9b5/ : FAILURE in 2h 49m 33s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky http://logs.openstack.org/57/618557/2/check/neutron-tempest-plugin-scenario-linuxbridge-rocky/a53a38a/ : SUCCESS in 1h 05m 29s\n- openstack-tox-pep8 http://logs.openstack.org/57/618557/2/check/openstack-tox-pep8/390b407/ : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"94c2b4eab75f53f1db77f7640f443ce76b6b1eab","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-11-22 07:42:08.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"eb287d6541e41621acb34df7cee62f3378fb8867","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-12-03 12:36:59.000000000","message":"Abandoned\n\nIt\u0027s not completely fixed yet for sure :(","accounts_in_message":[],"_revision_number":2},{"id":"ddf1f027dcb5eb9fd8861b5714112838f8d79ce8","tag":"autogenerated:gerrit:restore","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2024-08-26 21:05:18.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"66c3212844d49cceb2589e59edb97f39b82433d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 21:05:52.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron-tempest-plugin for 618557,2","accounts_in_message":[],"_revision_number":2},{"id":"15fc8665afe4a36fd3a2e12384ea4a509557be66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2024-08-26 21:08:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4ac89a85018ea0205d9974c444c09c169a7bd165","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2024-08-26 21:15:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"200269c59d5328689063b1905687e722d2925751","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 23:17:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da421bfaad5a4a0f9e61c0ab979fff1d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae310788c7eb4891988e9c9f0aa3b7af : SUCCESS in 2m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/911ef7d2aa7a4d1b9a9927b054e40962 : SUCCESS in 1h 21m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ba045e8d28f84fe2a6db18c931b3997c : SUCCESS in 1h 38m 06s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/e8b9faae58f047bbada79c7f05d4857a : SUCCESS in 1h 36m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b39e02726f7c4df38f16fec017a60e05 : SUCCESS in 1h 54m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7ebfcbc6c85d46b593fa58d4170c0754 : SUCCESS in 36m 01s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/7deda582e8144e4db0e87dfb23892e93 : SUCCESS in 1h 31m 04s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/d7e559c6680e42539168dc3e18c389c9 : SUCCESS in 1h 20m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/06cd031cbe764ab886bf0be4b5e12367 : SUCCESS in 1h 26m 09s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/c58a6bbbda1547e48fbd2db4598ed265 : SUCCESS in 1h 34m 17s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/b0fb6410d69f4aee9237ed8ab54ce44a : SUCCESS in 35m 29s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/4635bfaee1194ce0b77756d722849a4a : SUCCESS in 1h 17m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/a316fb358a3f4e76a694fd2cf32a4640 : SUCCESS in 1h 20m 46s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/2d2539bf579d4456b68605f2c4387c47 : SUCCESS in 2h 01m 07s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/3f6dda0b6bca4033ae6e70c20f008077 : SUCCESS in 32m 22s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/b2a7e947bddb4ee99005334755fed5fd : SUCCESS in 1h 12m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/dcb3a8803cc1492fb4c1b10e1b92c708 : SUCCESS in 1h 13m 11s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/3fdcb722fc054e0f9124ed6ad8eff4ec : SUCCESS in 1h 36m 36s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/8f8ba9873253459893a3b164ad51a276 : SUCCESS in 30m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aa5bc4087d6453caad4736bd032fe57 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":4},{"id":"fe2e84bfd7097918ad6d5971162da23845d9129c","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-26 23:44:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6b1864467a910cbc0bf4e8cb91f7e36c728aa058","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 01:50:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23925e1b90134527b822d6b80f8fa219\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebabd59072df497897f24d87f12c21ee : SUCCESS in 3m 26s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f14a4bbdceb5427186bb06acf8a007a7 : SUCCESS in 1h 23m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7b7ec60b0249499f9f1efdf130d6a41f : SUCCESS in 2h 04m 31s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/c915e9417a6e4bb9aff8b51fbe30ce5d : SUCCESS in 1h 35m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/687a7df8c42542779a6b690731431308 : SUCCESS in 1h 46m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/813aa2ff2c9043fb949a1bfaef3a61d8 : SUCCESS in 32m 30s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/4c9fba742db44f2f9532e2db76108ee3 : SUCCESS in 1h 52m 43s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/db7eff61d0d145d5901b212f180c748e : SUCCESS in 1h 09m 25s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/62af1a59f2bf48f4996fab5b0c214c74 : SUCCESS in 1h 23m 37s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/bce58f91c234439eb63c78333f5c2c4b : SUCCESS in 1h 44m 29s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/419b808669a74efabf11909acf386299 : SUCCESS in 32m 14s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/723e0b9fa19c48cba2eec1c8eeaeeac4 : SUCCESS in 1h 09m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/91b02da7fe644307afecf927c7633dc0 : SUCCESS in 1h 25m 54s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/ed95bcca66fe407cab454b49d1bf7e93 : SUCCESS in 1h 40m 58s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/26fed15c6a0e4712be1425765a3e2a99 : SUCCESS in 35m 58s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/3c858f3800b3449aa9169e3b5958be2d : SUCCESS in 1h 07m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/c8b732ee45c4482a82e635ef8d81ad59 : SUCCESS in 1h 44m 47s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/32791631135f4314a2f2758347d04333 : SUCCESS in 1h 37m 52s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/dc77237bc0a24ac38364c14c5cf5f1a6 : SUCCESS in 33m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a0dc42c970d403ebba89da0c47926aa : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":4},{"id":"7d4eb59f9d0b41b8f8220ae337181c6527997290","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-27 13:51:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7ff7d2d0659eea9bab453c4902eff754acb18458","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-27 15:16:52.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"058b7f35a6b8eea98fb3cead65fd05ad5aa6a349","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 16:21:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4453e8e72dc74071b72bab235722679d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb60cac6e5714757abeea8f0cdf685ae : SUCCESS in 3m 37s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/12223ad41517454b9b4c2a17cee92073 : SUCCESS in 1h 23m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3cd07c5ad892442d85d72c89ac80c435 : SUCCESS in 2h 26m 13s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/042b9af554a74d158bf8abde20322945 : SUCCESS in 1h 37m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/134ec049137744c8995299b5e66394af : SUCCESS in 2h 04m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/32d8f7a413004385a93106703741545e : SUCCESS in 37m 38s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/4d942ed49a1d46548b580e280180027e : SUCCESS in 1h 52m 29s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/b1dfd1d1519940e2850116f4ea97dc2c : SUCCESS in 1h 15m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/2aca19f439b4467bbb1320188ce70f48 : SUCCESS in 1h 52m 07s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/af3d47e52f104e18bbb0bf98699b9e91 : SUCCESS in 2h 12m 41s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/a63ddcc3be7a4671995475afbf8dae84 : SUCCESS in 32m 27s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/59c277f3fa2147619b300bd1c139c448 : SUCCESS in 1h 38m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/d11a37828e6149048910690e6bad61ad : SUCCESS in 1h 35m 47s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/4ad94bc71ddb48e0b54ef707796d1bb6 : SUCCESS in 2h 08m 43s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/597e87b34b56491facad766e2416683d : SUCCESS in 31m 35s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/e689d65ebf5b4b51b87af97cf85eae59 : SUCCESS in 1h 52m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/83f75cd46a9f4a55b14aaadeccebdf93 : SUCCESS in 1h 19m 34s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/9bf6f1f24608402eb5c765b70d1d6566 : SUCCESS in 2h 26m 46s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/58781160eba64498a6d75b7e82fd9846 : SUCCESS in 31m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3e6a86a2fb045088da9f591023c4ccb : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":4},{"id":"aef118801c525eec8806e519f4fdd31af03baefa","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-27 20:27:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"746834368934b538b36275662656902d9bd4bf5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 23:05:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2e6b7d293924420a79741ab10b521fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfc74bb1427b43a48a595ebf4a099caa : SUCCESS in 4m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/19b17da55e824c39a5ab6844d69d54bf : SUCCESS in 1h 46m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f8aafb22f6184cdbb7de75f7be292258 : SUCCESS in 2h 16m 07s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/fee7ac024b934130a6cf91c5b91bf6f1 : SUCCESS in 1h 53m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/eb861f905e104231946d347f83df882c : SUCCESS in 2h 35m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fae6b7fad58240809444b8cfbafe7466 : SUCCESS in 44m 18s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/22b424b606dc4e89866e05d962702070 : SUCCESS in 1h 45m 08s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/1c86434e19f049a2b0a7cd39018f1a2a : SUCCESS in 1h 38m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/8221ed5947a741ef98f855d701668df5 : SUCCESS in 1h 40m 12s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/36bc5c5345e4439d9a50a3ef3447c106 : SUCCESS in 1h 51m 34s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/b7d75c461db24605b64f83cd2e4ebdde : SUCCESS in 30m 05s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/0896e6447d84405ebd6a7aa68ae166e4 : SUCCESS in 1h 29m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/2627e65ed8e14e41b9bddbe42beabac1 : SUCCESS in 1h 18m 31s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/6f04f704b9cc45ed9d0804ab611f7477 : SUCCESS in 2h 03m 17s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/283a9c70fe2f494c86f572fcb1056f89 : SUCCESS in 37m 52s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/6f74da5856814d1794185571f1ff3c2b : SUCCESS in 1h 51m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/1ec1d29eb6bf414abd87337efdb02487 : SUCCESS in 2h 08m 12s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/3f1953b8fb61490ca77ebd3681ce123e : SUCCESS in 2h 21m 52s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/1adf4428fccc43f58217689b607a6888 : SUCCESS in 45m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5487bc456784a16a41b492836b35c65 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":4},{"id":"6bcaef1f56e2f3d1813c19541063a2ca5e7b110a","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-08-28 00:11:34.000000000","message":"Patch Set 4: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/1527c16209584a3ea1f8578fbac02de9\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/f1d5d6a75d494b1ab1668f4d6966c2c9 : SUCCESS in 3h 42m 32s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/1f61491fa8a34ae98005c8f92aae30eb : SUCCESS in 2h 53m 52s","accounts_in_message":[],"_revision_number":4},{"id":"c136d8e2dd3e27381d73bda6e11e8b2b747330cf","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-28 19:07:18.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"a6f7ccc6002d2ffd5cb78c45eb2ff1761be294d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 19:08:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f54cf0dfa53727f20b58081ee9dc2b2c419c3575","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 23:43:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc8a5e2bb4934065a7578a20f2f90ae5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfe27af1bbb84f93b611b3048ebfbca3 : SUCCESS in 5m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/500d3e0a500647c4bf36fccaf788c31e : SUCCESS in 1h 18m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5c664951e843421c900ba23b80cdc1cb : SUCCESS in 1h 37m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/13397fc760574548b834b571700508d8 : SUCCESS in 2h 32m 57s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/9b0e2c95fd4646e3a62ba7b7705f0fbb : SUCCESS in 1h 24m 23s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/26c5aab5d2a94b69af2a60ecb860fb54 : SUCCESS in 2h 17m 17s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/0a9cf7351ac34ba296cd285bc925fad9 : SUCCESS in 2h 15m 02s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/cbfecabbbd684c4dae92424fb294730b : SUCCESS in 2h 11m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94ed12f2b0ab4387b199f80accbde02e : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":4},{"id":"f4d3d5a56fc70d8e37232aeabcad9841ab5bea28","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 23:43:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8a74ef325955f1869e7873ce8867a72a61a24c41","revisions":{"c7729aa8306d77d09217b2302ee5fb36034f0995":{"kind":"REWORK","_number":1,"created":"2018-11-16 16:29:31.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/57/618557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/57/618557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/57/618557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/57/618557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/57/618557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/57/618557/1"}}},"commit":{"parents":[{"commit":"cf38b77328dbf94f1323f96f68aa77124a6f4a7b","subject":"Merge \"test for minimum bw ingress direction\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/cf38b77328dbf94f1323f96f68aa77124a6f4a7b"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2018-11-16 16:28:33.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2018-11-16 16:28:33.000000000","tz":60},"subject":"DNM Remove unstable_test from FIP test_east_west","message":"DNM Remove unstable_test from FIP test_east_west\n\nI just want to check how exactly it\u0027s failing in gate.\n\nChange-Id: I47cc5656ea2e1e6e085d4d6602dbc76f87463cc3\nRelated-Bug: #1717302\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c7729aa8306d77d09217b2302ee5fb36034f0995"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c7729aa8306d77d09217b2302ee5fb36034f0995"}]},"branch":"refs/heads/master"},"ad244405afc705c99044ae1fd9f86cc430835d02":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-11-19 14:11:42.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/57/618557/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/57/618557/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/57/618557/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/57/618557/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/57/618557/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/57/618557/2"}}},"commit":{"parents":[{"commit":"cf38b77328dbf94f1323f96f68aa77124a6f4a7b","subject":"Merge \"test for minimum bw ingress direction\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/cf38b77328dbf94f1323f96f68aa77124a6f4a7b"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2018-11-16 16:28:33.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2018-11-19 14:10:56.000000000","tz":60},"subject":"Remove unstable_test from FIP test_east_west","message":"Remove unstable_test from FIP test_east_west\n\nDepends-On: https://review.openstack.org/#/c/618750/\n\nChange-Id: I47cc5656ea2e1e6e085d4d6602dbc76f87463cc3\nRelated-Bug: #1717302\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ad244405afc705c99044ae1fd9f86cc430835d02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ad244405afc705c99044ae1fd9f86cc430835d02"}]},"branch":"refs/heads/master"},"e9d112cd8745b8d1cca12e1aba4b1089dd75bb44":{"kind":"REWORK","_number":3,"created":"2024-08-26 21:08:29.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/57/618557/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/57/618557/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/57/618557/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/57/618557/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/57/618557/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/57/618557/3"}}},"commit":{"parents":[{"commit":"20dec4e823d8217c45f2c7f7891e5b78ec04596e","subject":"Merge \"Switch networking-sfc job to be run with eventlet\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/20dec4e823d8217c45f2c7f7891e5b78ec04596e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2018-11-16 16:28:33.000000000","tz":60},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2024-08-26 21:08:28.000000000","tz":-300},"subject":"Remove unstable_test from FIP test_east_west","message":"Remove unstable_test from FIP test_east_west\n\nDepends-On: https://review.openstack.org/#/c/618750/\n\nChange-Id: I47cc5656ea2e1e6e085d4d6602dbc76f87463cc3\nRelated-Bug: #1717302\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e9d112cd8745b8d1cca12e1aba4b1089dd75bb44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e9d112cd8745b8d1cca12e1aba4b1089dd75bb44"}]},"branch":"refs/heads/master"},"8a74ef325955f1869e7873ce8867a72a61a24c41":{"kind":"REWORK","_number":4,"created":"2024-08-26 21:15:32.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/57/618557/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/57/618557/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/57/618557/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/57/618557/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/57/618557/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/57/618557/4"}}},"commit":{"parents":[{"commit":"20dec4e823d8217c45f2c7f7891e5b78ec04596e","subject":"Merge \"Switch networking-sfc job to be run with eventlet\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/20dec4e823d8217c45f2c7f7891e5b78ec04596e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2018-11-16 16:28:33.000000000","tz":60},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2024-08-26 21:15:25.000000000","tz":-300},"subject":"Remove unstable_test from FIP test_east_west","message":"Remove unstable_test from FIP test_east_west\n\nDepends-On: https://review.openstack.org/#/c/618750/\n\nChange-Id: I47cc5656ea2e1e6e085d4d6602dbc76f87463cc3\nRelated-Bug: #1717302\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8a74ef325955f1869e7873ce8867a72a61a24c41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8a74ef325955f1869e7873ce8867a72a61a24c41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
