)]}'
{"id":"openstack%2Fneutron-tempest-plugin~619207","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ic78ee64a9350863ff5bd4b4baa4b20770e1b9d47","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"trunk-subports-fix","hashtags":[],"change_id":"Ic78ee64a9350863ff5bd4b4baa4b20770e1b9d47","subject":"Add additional config option \"default_image_is_advanced\"","status":"MERGED","created":"2018-11-21 10:48:07.000000000","updated":"2019-04-23 20:51:21.000000000","submitted":"2019-04-23 20:51:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":48,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"619207-1556052681320-9fef03c1","meta_rev_id":"094fcf4958804b8a9cd7c63228449cec295062aa","_number":619207,"virtual_id_number":619207,"owner":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"date":"2019-04-23 11:00:14.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16615,"name":"abregman","email":"abregman@redhat.com","username":"abregman"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2019-04-23 20:51:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-18 08:51:07.000000000","_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},{"value":0,"_account_id":10034,"name":"Sanjay Upadhyay","email":"supadhya@redhat.com","username":"saneax"},{"value":0,"_account_id":26899,"name":"Roee Agiman","email":"ragiman@redhat.com","username":"ragiman"},{"value":0,"_account_id":28245,"name":"Vadim Khitrin","email":"vkhitrin@redhat.com","username":"vkhitrin@redhat.com"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":28094,"name":"Maxim Babushkin","email":"babushkin.max@gmail.com","username":"mbabushk"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"disliked":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"all":[{"value":-1,"date":"2019-04-23 10:38:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2019-04-17 07:20:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16615,"name":"abregman","email":"abregman@redhat.com","username":"abregman"},{"value":2,"date":"2019-04-23 12:42:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2019-04-15 11:37:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},{"value":2,"date":"2019-04-23 11:11:34.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},{"value":0,"_account_id":10034,"name":"Sanjay Upadhyay","email":"supadhya@redhat.com","username":"saneax"},{"value":1,"date":"2019-04-15 11:45:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26899,"name":"Roee Agiman","email":"ragiman@redhat.com","username":"ragiman"},{"value":0,"_account_id":28245,"name":"Vadim Khitrin","email":"vkhitrin@redhat.com","username":"vkhitrin@redhat.com"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":28094,"name":"Maxim Babushkin","email":"babushkin.max@gmail.com","username":"mbabushk"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16615,"name":"abregman","email":"abregman@redhat.com","username":"abregman"},{"value":1,"date":"2019-04-23 12:42:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},{"value":0,"date":"2019-04-15 11:40:04.000000000","_account_id":10034,"name":"Sanjay Upadhyay","email":"supadhya@redhat.com","username":"saneax"},{"value":0,"_account_id":26899,"name":"Roee Agiman","email":"ragiman@redhat.com","username":"ragiman"},{"value":0,"date":"2019-04-15 11:39:30.000000000","_account_id":28245,"name":"Vadim Khitrin","email":"vkhitrin@redhat.com","username":"vkhitrin@redhat.com"},{"value":0,"date":"2019-04-15 12:31:18.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"date":"2019-04-15 11:39:20.000000000","_account_id":28094,"name":"Maxim Babushkin","email":"babushkin.max@gmail.com","username":"mbabushk"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16615,"name":"abregman","email":"abregman@redhat.com","username":"abregman"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},{"value":0,"_account_id":10034,"name":"Sanjay Upadhyay","email":"supadhya@redhat.com","username":"saneax"},{"value":0,"_account_id":26899,"name":"Roee Agiman","email":"ragiman@redhat.com","username":"ragiman"},{"value":0,"_account_id":28245,"name":"Vadim Khitrin","email":"vkhitrin@redhat.com","username":"vkhitrin@redhat.com"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":28094,"name":"Maxim Babushkin","email":"babushkin.max@gmail.com","username":"mbabushk"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":10034,"name":"Sanjay Upadhyay","email":"supadhya@redhat.com","username":"saneax"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16615,"name":"abregman","email":"abregman@redhat.com","username":"abregman"},{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":26899,"name":"Roee Agiman","email":"ragiman@redhat.com","username":"ragiman"},{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"_account_id":28094,"name":"Maxim Babushkin","email":"babushkin.max@gmail.com","username":"mbabushk"},{"_account_id":28245,"name":"Vadim Khitrin","email":"vkhitrin@redhat.com","username":"vkhitrin@redhat.com"},{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-11-21 11:12:16.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2018-11-29 23:41:12.000000000","updated_by":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"reviewer":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"state":"REVIEWER"},{"updated":"2019-04-12 07:25:43.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-04-14 06:10:18.000000000","updated_by":{"_account_id":16615,"name":"abregman","email":"abregman@redhat.com","username":"abregman"},"reviewer":{"_account_id":16615,"name":"abregman","email":"abregman@redhat.com","username":"abregman"},"state":"REVIEWER"},{"updated":"2019-04-15 11:39:20.000000000","updated_by":{"_account_id":28094,"name":"Maxim Babushkin","email":"babushkin.max@gmail.com","username":"mbabushk"},"reviewer":{"_account_id":28094,"name":"Maxim Babushkin","email":"babushkin.max@gmail.com","username":"mbabushk"},"state":"REVIEWER"},{"updated":"2019-04-15 11:39:30.000000000","updated_by":{"_account_id":28245,"name":"Vadim Khitrin","email":"vkhitrin@redhat.com","username":"vkhitrin@redhat.com"},"reviewer":{"_account_id":28245,"name":"Vadim Khitrin","email":"vkhitrin@redhat.com","username":"vkhitrin@redhat.com"},"state":"REVIEWER"},{"updated":"2019-04-15 11:40:04.000000000","updated_by":{"_account_id":10034,"name":"Sanjay Upadhyay","email":"supadhya@redhat.com","username":"saneax"},"reviewer":{"_account_id":10034,"name":"Sanjay Upadhyay","email":"supadhya@redhat.com","username":"saneax"},"state":"REVIEWER"},{"updated":"2019-04-15 11:45:35.000000000","updated_by":{"_account_id":26899,"name":"Roee Agiman","email":"ragiman@redhat.com","username":"ragiman"},"reviewer":{"_account_id":26899,"name":"Roee Agiman","email":"ragiman@redhat.com","username":"ragiman"},"state":"REVIEWER"},{"updated":"2019-04-15 12:31:18.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-04-17 07:20:33.000000000","updated_by":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"reviewer":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"state":"REVIEWER"},{"updated":"2019-04-18 08:51:07.000000000","updated_by":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"reviewer":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"state":"REVIEWER"},{"updated":"2019-04-23 10:38:46.000000000","updated_by":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"reviewer":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"state":"REVIEWER"},{"updated":"2019-04-23 11:00:14.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-04-23 11:11:34.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-04-23 12:42:25.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-04-23 20:51:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f04a9f4534868c7337c43d1d22f611df96111b29","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-11-21 10:48:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84b6d91f2da354b9254d32da6957ac969f436be7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-11-21 11:05:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"60066ebb539145265b79e4b74888fe86e297ff58","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2018-11-21 11:12:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"32d2462ecfccc383c9203d930932adada2b8888c","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2018-11-21 12:07:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c690ee51b40de5b7e7159643e36f8ac1fdc3c61c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-21 13:30:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/07/619207/1/check/neutron-tempest-plugin-api/b6eba34/ : SUCCESS in 50m 59s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/07/619207/1/check/neutron-tempest-plugin-designate-scenario/64f684d/ : SUCCESS in 32m 34s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/07/619207/1/check/neutron-tempest-plugin-dvr-multinode-scenario/da80807/ : FAILURE in 2h 23m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/07/619207/1/check/neutron-tempest-plugin-scenario-linuxbridge/a333900/ : FAILURE in 58m 21s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/619207/1/check/build-openstack-sphinx-docs/b8e7f3c/html/ : SUCCESS in 5m 16s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/07/619207/1/check/neutron-tempest-plugin-api-queens/852214d/ : SUCCESS in 43m 29s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/07/619207/1/check/neutron-tempest-plugin-designate-scenario-queens/2016366/ : SUCCESS in 41m 16s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/07/619207/1/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/8b76cc6/ : FAILURE in 2h 08m 28s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/07/619207/1/check/neutron-tempest-plugin-scenario-linuxbridge-queens/7912e7e/ : FAILURE in 1h 10m 00s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/07/619207/1/check/neutron-tempest-plugin-api-rocky/d6a1e69/ : SUCCESS in 44m 41s\n- neutron-tempest-plugin-designate-scenario-rocky http://logs.openstack.org/07/619207/1/check/neutron-tempest-plugin-designate-scenario-rocky/e681799/ : SUCCESS in 32m 20s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky http://logs.openstack.org/07/619207/1/check/neutron-tempest-plugin-dvr-multinode-scenario-rocky/dd8db61/ : FAILURE in 2h 34m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky http://logs.openstack.org/07/619207/1/check/neutron-tempest-plugin-scenario-linuxbridge-rocky/fb86cb5/ : FAILURE in 1h 00m 11s\n- openstack-tox-pep8 http://logs.openstack.org/07/619207/1/check/openstack-tox-pep8/b1baca3/ : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":1},{"id":"1cb3135114bb640eac9afac68b76765ebc9a81f1","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-11-21 16:33:52.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ba7b1c5463b88d306714a06b7c8344faea3a07a2","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-11-21 16:34:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ea287122577ce0416cf8f10658e5396a6016c3b8","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-11-21 18:44:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"eb598133d8e867c6b1cfed13462ef2143b819dae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-21 19:47:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/07/619207/2/check/neutron-tempest-plugin-api/d41dbf6/ : SUCCESS in 44m 42s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/07/619207/2/check/neutron-tempest-plugin-designate-scenario/9f14d39/ : SUCCESS in 33m 44s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/07/619207/2/check/neutron-tempest-plugin-dvr-multinode-scenario/6eb1f2f/ : FAILURE in 2h 53m 00s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/07/619207/2/check/neutron-tempest-plugin-scenario-linuxbridge/4390460/ : SUCCESS in 1h 01m 11s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/619207/2/check/build-openstack-sphinx-docs/7590cda/html/ : SUCCESS in 6m 56s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/07/619207/2/check/neutron-tempest-plugin-api-queens/558489e/ : SUCCESS in 47m 18s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/07/619207/2/check/neutron-tempest-plugin-designate-scenario-queens/327b71c/ : SUCCESS in 35m 00s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/07/619207/2/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/ad980e3/ : FAILURE in 2h 11m 29s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/07/619207/2/check/neutron-tempest-plugin-scenario-linuxbridge-queens/c13f03c/ : SUCCESS in 1h 01m 27s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/07/619207/2/check/neutron-tempest-plugin-api-rocky/e643c6c/ : SUCCESS in 54m 24s\n- neutron-tempest-plugin-designate-scenario-rocky http://logs.openstack.org/07/619207/2/check/neutron-tempest-plugin-designate-scenario-rocky/d7adc99/ : SUCCESS in 31m 17s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky http://logs.openstack.org/07/619207/2/check/neutron-tempest-plugin-dvr-multinode-scenario-rocky/6532368/ : FAILURE in 2h 08m 46s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky http://logs.openstack.org/07/619207/2/check/neutron-tempest-plugin-scenario-linuxbridge-rocky/9c75902/ : SUCCESS in 58m 31s\n- openstack-tox-pep8 http://logs.openstack.org/07/619207/2/check/openstack-tox-pep8/a531b41/ : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"9f93aa80b27be6d4315db04c558d46303126b1bf","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-11-21 21:23:21.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6c66f58d03d1504af7f19047d757de8be3620eb4","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-11-22 09:20:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"19ddbe7f30b0d2c15c1f1714c6c27ff40ba8b7dd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-11-22 13:56:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"253c17037c6977ac42dfe760ed475d9303b3bb0c","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-11-22 14:50:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"086c156202e7a9d4cbde35180787e3034561a150","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-11-22 15:57:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e9855846c7c7f3ff34c729160c47c7fb02f6ffab","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-11-22 15:57:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e1929bd9353565b69b55e9b7228324c457f17cab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-22 17:06:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/07/619207/3/check/neutron-tempest-plugin-api/72220fe/ : FAILURE in 30m 17s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/07/619207/3/check/neutron-tempest-plugin-designate-scenario/3b97d20/ : FAILURE in 29m 07s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/07/619207/3/check/neutron-tempest-plugin-dvr-multinode-scenario/30db687/ : FAILURE in 49m 11s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/07/619207/3/check/neutron-tempest-plugin-scenario-linuxbridge/9de6907/ : FAILURE in 29m 31s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/619207/3/check/build-openstack-sphinx-docs/4c4f071/html/ : SUCCESS in 7m 38s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/07/619207/3/check/neutron-tempest-plugin-api-queens/970c240/ : FAILURE in 31m 28s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/07/619207/3/check/neutron-tempest-plugin-designate-scenario-queens/d1d43f4/ : FAILURE in 31m 06s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/07/619207/3/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/a2ba842/ : FAILURE in 40m 29s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/07/619207/3/check/neutron-tempest-plugin-scenario-linuxbridge-queens/3dfe42a/ : FAILURE in 27m 00s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/07/619207/3/check/neutron-tempest-plugin-api-rocky/10f86df/ : FAILURE in 26m 56s\n- neutron-tempest-plugin-designate-scenario-rocky http://logs.openstack.org/07/619207/3/check/neutron-tempest-plugin-designate-scenario-rocky/df4c212/ : TIMED_OUT in 1h 04m 57s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky http://logs.openstack.org/07/619207/3/check/neutron-tempest-plugin-dvr-multinode-scenario-rocky/292eef8/ : FAILURE in 43m 33s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky http://logs.openstack.org/07/619207/3/check/neutron-tempest-plugin-scenario-linuxbridge-rocky/27ac9dd/ : FAILURE in 36m 00s\n- openstack-tox-pep8 http://logs.openstack.org/07/619207/3/check/openstack-tox-pep8/c662dca/ : FAILURE in 4m 36s","accounts_in_message":[],"_revision_number":3},{"id":"b2eca4d3c161712ab705dd854f84a57f3a4aad07","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2018-11-23 07:15:54.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1683c0dc7cf1878d91febbe71bbe88efd2c7e0ab","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2018-11-23 07:18:14.000000000","message":"Patch Set 3:\n\nHello can you please figure out how to fix tempest.conf code instead of changing netutron-tempest-plugin code?","accounts_in_message":[],"_revision_number":3},{"id":"6bd0bf92f57bb0edbb7d85c73d41f03ca33872b7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-11-23 08:21:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ddca8ad3cc986802b941a0883048c0cf6568cd48","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-11-25 13:24:56.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ad800f9d2c57e47856f4d31fd3dd347c6d4cba34","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-11-26 16:22:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f9b244f0cb1f16d793046104f5bec35ecf2d9e72","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-11-26 16:24:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3291852da61bcc1d340569d1e90a39e502bc7400","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-11-26 16:27:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"97fe11ed8018777004e4ca8472b6b03707b70677","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-26 21:59:10.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/07/619207/6/check/neutron-tempest-plugin-api/1c1e57b/ : SUCCESS in 58m 36s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/07/619207/6/check/neutron-tempest-plugin-designate-scenario/12046e8/ : SUCCESS in 28m 41s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/07/619207/6/check/neutron-tempest-plugin-dvr-multinode-scenario/3ea252d/ : FAILURE in 2h 30m 33s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/07/619207/6/check/neutron-tempest-plugin-scenario-linuxbridge/574ae39/ : SUCCESS in 52m 53s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/619207/6/check/build-openstack-sphinx-docs/14b10c8/html/ : SUCCESS in 6m 27s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/07/619207/6/check/neutron-tempest-plugin-api-queens/55f8bdc/ : SUCCESS in 43m 01s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/07/619207/6/check/neutron-tempest-plugin-designate-scenario-queens/43dec52/ : SUCCESS in 31m 14s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/07/619207/6/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/bcafd46/ : FAILURE in 2h 36m 49s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/07/619207/6/check/neutron-tempest-plugin-scenario-linuxbridge-queens/c2bd00d/ : SUCCESS in 1h 04m 32s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/07/619207/6/check/neutron-tempest-plugin-api-rocky/17eaddb/ : SUCCESS in 43m 43s\n- neutron-tempest-plugin-designate-scenario-rocky http://logs.openstack.org/07/619207/6/check/neutron-tempest-plugin-designate-scenario-rocky/814d7d5/ : SUCCESS in 32m 25s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky http://logs.openstack.org/07/619207/6/check/neutron-tempest-plugin-dvr-multinode-scenario-rocky/4792a79/ : FAILURE in 2h 52m 44s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky http://logs.openstack.org/07/619207/6/check/neutron-tempest-plugin-scenario-linuxbridge-rocky/00b77ae/ : SUCCESS in 1h 06m 31s\n- openstack-tox-pep8 http://logs.openstack.org/07/619207/6/check/openstack-tox-pep8/dd0eea8/ : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":6},{"id":"2c7751fe9596c19129da29f6ae55d3068f0c4f77","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-11-27 08:39:48.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8dfd908ab7c7421b42c1f0da607b67573925d90c","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-11-27 12:24:07.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"88b4a70a5d7f492d63f675aaf3b0d07f1f82a346","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2018-11-27 23:10:11.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9492a615ae6b40e19f662485ab8099a617664c9c","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-11-28 06:14:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1b3dd478f2436f8cc73b207b4a868c644898f490","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-11-28 06:25:30.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f9c9298ff6f53b49593ef0dacfa08504e061868d","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-11-28 17:24:37.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"588bc72e19cb681711c0d7438f01129cee8d9e26","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2018-11-28 18:56:47.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6d0aa2a4f595c1fff3ba53178f5a39f54cd37faf","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-11-29 09:05:48.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"03ac1307619fa4e236ed611f1a5cffa392315ce1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-11-29 21:55:30.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f0d7c061dc0b5a4e45fb7f864539a060900e6ea1","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2018-11-29 23:41:12.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e4ef97a982ba93a1a644bef5d5dc6b6a8eb706c8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-11-30 08:18:52.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e723e18aaa7d587065546d72e4656fdec8d97f7f","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-12-02 15:20:28.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ac51cfd144f2dbfac19e741dd166d65a46f14268","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-12-09 09:48:25.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6691163aaeadcc17c53f9405506c8932ebe17652","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-12-09 09:50:24.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c672da0577dff1b4511e703f7a646383ae97b550","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-09 11:49:27.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/07/619207/8/check/neutron-tempest-plugin-api/ed1a7c7/ : SUCCESS in 42m 01s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/07/619207/8/check/neutron-tempest-plugin-designate-scenario/1367613/ : SUCCESS in 31m 13s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/07/619207/8/check/neutron-tempest-plugin-dvr-multinode-scenario/b4e9c19/ : FAILURE in 1h 57m 08s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/07/619207/8/check/neutron-tempest-plugin-scenario-linuxbridge/71966ca/ : SUCCESS in 51m 53s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/619207/8/check/build-openstack-sphinx-docs/e046c7c/html/ : SUCCESS in 4m 51s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/07/619207/8/check/neutron-tempest-plugin-api-queens/0d51d58/ : SUCCESS in 37m 01s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/07/619207/8/check/neutron-tempest-plugin-designate-scenario-queens/fd3587d/ : SUCCESS in 29m 10s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/07/619207/8/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/1759ec3/ : FAILURE in 1h 18m 47s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/07/619207/8/check/neutron-tempest-plugin-scenario-linuxbridge-queens/b49a3a7/ : SUCCESS in 42m 42s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/07/619207/8/check/neutron-tempest-plugin-api-rocky/7ea6d9a/ : SUCCESS in 43m 04s\n- neutron-tempest-plugin-designate-scenario-rocky http://logs.openstack.org/07/619207/8/check/neutron-tempest-plugin-designate-scenario-rocky/7912f42/ : SUCCESS in 24m 27s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky http://logs.openstack.org/07/619207/8/check/neutron-tempest-plugin-dvr-multinode-scenario-rocky/9b72491/ : FAILURE in 1h 40m 00s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky http://logs.openstack.org/07/619207/8/check/neutron-tempest-plugin-scenario-linuxbridge-rocky/1e64b5b/ : SUCCESS in 43m 02s\n- openstack-tox-pep8 http://logs.openstack.org/07/619207/8/check/openstack-tox-pep8/336caf1/ : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":8},{"id":"1df4f7b574b652ce54d4e529a92ad3ad7ab09e64","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-12-12 09:41:28.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)\n\nPlease note that there are also other tests which currently are using advanced_image (mtu) so You should make modifications for all of them that all would be consistent.","accounts_in_message":[],"_revision_number":8},{"id":"1554aee2402d73e813de50d199b3490e5a740832","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-12-13 08:56:37.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"9c1f6c5b3763eac53b4fd1624b420aa2d696c60f","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-12-13 08:57:22.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9998fe6392bb8ae7061412c3656176f60ea7725d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-13 11:28:11.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/07/619207/9/check/neutron-tempest-plugin-api/7c4eafe/ : SUCCESS in 47m 57s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/07/619207/9/check/neutron-tempest-plugin-designate-scenario/572bbe5/ : SUCCESS in 34m 07s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/07/619207/9/check/neutron-tempest-plugin-dvr-multinode-scenario/76ac65b/ : FAILURE in 2h 27m 04s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/07/619207/9/check/neutron-tempest-plugin-scenario-linuxbridge/4e1b055/ : FAILURE in 50m 54s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/619207/9/check/build-openstack-sphinx-docs/9dabc24/html/ : SUCCESS in 4m 30s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/07/619207/9/check/neutron-tempest-plugin-api-queens/59f327e/ : SUCCESS in 49m 48s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/07/619207/9/check/neutron-tempest-plugin-designate-scenario-queens/4aeedeb/ : SUCCESS in 31m 07s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/07/619207/9/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/d548230/ : FAILURE in 9m 51s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/07/619207/9/check/neutron-tempest-plugin-scenario-linuxbridge-queens/26c3f69/ : FAILURE in 53m 13s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/07/619207/9/check/neutron-tempest-plugin-api-rocky/6487cb4/ : SUCCESS in 49m 53s\n- neutron-tempest-plugin-designate-scenario-rocky http://logs.openstack.org/07/619207/9/check/neutron-tempest-plugin-designate-scenario-rocky/8659a21/ : SUCCESS in 30m 21s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky http://logs.openstack.org/07/619207/9/check/neutron-tempest-plugin-dvr-multinode-scenario-rocky/fb1a5a1/ : FAILURE in 1h 20m 48s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky http://logs.openstack.org/07/619207/9/check/neutron-tempest-plugin-scenario-linuxbridge-rocky/607deff/ : FAILURE in 52m 11s\n- openstack-tox-pep8 http://logs.openstack.org/07/619207/9/check/openstack-tox-pep8/3e73ace/ : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":9},{"id":"dd51df135df5c5f7f3a0b8e5d4d421b52f8843e4","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-12-13 12:06:44.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"169d0389042f34481ce21b5f0088dee899be2945","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-12-13 12:10:46.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"bcaf6022c90e01e7aec5c4a62c754e8010a9f789","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-13 14:15:43.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/07/619207/11/check/neutron-tempest-plugin-api/2e3c62b/ : SUCCESS in 41m 24s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/07/619207/11/check/neutron-tempest-plugin-designate-scenario/60e569a/ : SUCCESS in 29m 57s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/07/619207/11/check/neutron-tempest-plugin-dvr-multinode-scenario/3a1fb0c/ : FAILURE in 2h 01m 43s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/07/619207/11/check/neutron-tempest-plugin-scenario-linuxbridge/3cc4e6a/ : SUCCESS in 53m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/619207/11/check/build-openstack-sphinx-docs/d219a41/html/ : SUCCESS in 5m 26s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/07/619207/11/check/neutron-tempest-plugin-api-queens/544996e/ : SUCCESS in 44m 49s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/07/619207/11/check/neutron-tempest-plugin-designate-scenario-queens/3a21e23/ : SUCCESS in 29m 50s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/07/619207/11/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/da3d025/ : FAILURE in 15m 04s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/07/619207/11/check/neutron-tempest-plugin-scenario-linuxbridge-queens/ecddcd1/ : SUCCESS in 55m 14s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/07/619207/11/check/neutron-tempest-plugin-api-rocky/c0f89a3/ : SUCCESS in 59m 01s\n- neutron-tempest-plugin-designate-scenario-rocky http://logs.openstack.org/07/619207/11/check/neutron-tempest-plugin-designate-scenario-rocky/0af8e65/ : SUCCESS in 31m 34s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky http://logs.openstack.org/07/619207/11/check/neutron-tempest-plugin-dvr-multinode-scenario-rocky/e0ab23b/ : SUCCESS in 1h 19m 31s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky http://logs.openstack.org/07/619207/11/check/neutron-tempest-plugin-scenario-linuxbridge-rocky/68d7628/ : SUCCESS in 1h 02m 30s\n- openstack-tox-pep8 http://logs.openstack.org/07/619207/11/check/openstack-tox-pep8/56737d4/ : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":11},{"id":"748fdc889985ce58a77c652bdc4287dc586ad336","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-12-20 09:23:58.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)\n\nThe same \"advanced_image_ref\" option is used also in test_mtu https://github.com/openstack/neutron-tempest-plugin/blob/master/neutron_tempest_plugin/scenario/test_mtu.py - You should adjust this code to use this new config option too.","accounts_in_message":[],"_revision_number":11},{"id":"1f2d1191e26c33dace82822d9f0c20369723aa9b","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-12-20 09:32:00.000000000","message":"Patch Set 11:\n\n(1 comment)\n\nThanks for commenting regards the test_mtu, i will update next patch,","accounts_in_message":[],"_revision_number":11},{"id":"016e496dbc5d70e785507e088e66da6e93e3a4dc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-12-20 09:55:20.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"cc111f749c06cc2384b31b76a1d96d9bd6ab8535","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2018-12-27 09:04:45.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"835415648a9ae62baa7187c561fc469408bab701","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-12-27 09:36:15.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"228d90ed5601af0bc91175fea231e5062cdd0d58","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-11 15:17:50.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"6b50013f0ca0fc78c7eccada844f3549dfd386c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 17:23:32.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/07/619207/12/check/neutron-tempest-plugin-api/18e74e2/ : SUCCESS in 49m 03s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/07/619207/12/check/neutron-tempest-plugin-designate-scenario/1a9ed69/ : SUCCESS in 32m 05s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/07/619207/12/check/neutron-tempest-plugin-dvr-multinode-scenario/1fb220b/ : SUCCESS in 1h 35m 07s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/07/619207/12/check/neutron-tempest-plugin-scenario-linuxbridge/5b83c54/ : SUCCESS in 1h 20m 05s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/619207/12/check/build-openstack-sphinx-docs/c7b3b73/html/ : SUCCESS in 5m 18s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/07/619207/12/check/neutron-tempest-plugin-api-queens/7c60da9/ : SUCCESS in 42m 41s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/07/619207/12/check/neutron-tempest-plugin-designate-scenario-queens/a5c0c8c/ : SUCCESS in 40m 39s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/07/619207/12/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/4c42ce9/ : FAILURE in 1h 22m 52s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/07/619207/12/check/neutron-tempest-plugin-scenario-linuxbridge-queens/06b2da8/ : SUCCESS in 59m 40s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/07/619207/12/check/neutron-tempest-plugin-api-rocky/4a891ff/ : SUCCESS in 51m 59s\n- neutron-tempest-plugin-designate-scenario-rocky http://logs.openstack.org/07/619207/12/check/neutron-tempest-plugin-designate-scenario-rocky/8a0ff26/ : SUCCESS in 34m 27s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky http://logs.openstack.org/07/619207/12/check/neutron-tempest-plugin-dvr-multinode-scenario-rocky/0812ff3/ : SUCCESS in 1h 30m 25s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky http://logs.openstack.org/07/619207/12/check/neutron-tempest-plugin-scenario-linuxbridge-rocky/c6152a3/ : SUCCESS in 1h 08m 11s\n- neutron-tempest-plugin-api-stein http://logs.openstack.org/07/619207/12/check/neutron-tempest-plugin-api-stein/6f41acb/ : SUCCESS in 44m 29s\n- neutron-tempest-plugin-designate-scenario-stein http://logs.openstack.org/07/619207/12/check/neutron-tempest-plugin-designate-scenario-stein/dec2bea/ : SUCCESS in 37m 01s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein http://logs.openstack.org/07/619207/12/check/neutron-tempest-plugin-dvr-multinode-scenario-stein/8035775/ : FAILURE in 1h 59m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein http://logs.openstack.org/07/619207/12/check/neutron-tempest-plugin-scenario-linuxbridge-stein/8b8a094/ : SUCCESS in 1h 10m 41s\n- openstack-tox-pep8 http://logs.openstack.org/07/619207/12/check/openstack-tox-pep8/5f911ad/ : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":12},{"id":"3021d8dfbf75189dc45d4302fb44e8853502eb27","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-04-11 19:20:18.000000000","message":"Patch Set 12: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"5cf4c37892f4ba8ba74ba3b2fa385560192c9744","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-11 20:15:34.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"e337d8d517b68550155df220187cc2cc769be5d8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-11 20:17:07.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"8436a6a3e5427095da176ddedc206e7fa7ef995f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-11 20:17:32.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"6704f4d25d55891099e07752a8836774e54f9188","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 00:49:52.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-api/3643a02/ : SUCCESS in 1h 02m 46s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-designate-scenario/d9c0897/ : SUCCESS in 41m 28s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-dvr-multinode-scenario/3b2d902/ : SUCCESS in 1h 45m 56s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-scenario-linuxbridge/4909e5c/ : SUCCESS in 1h 11m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/619207/14/check/build-openstack-sphinx-docs/14976dd/html/ : SUCCESS in 7m 25s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-api-queens/427eb75/ : SUCCESS in 45m 49s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-designate-scenario-queens/e7a2b06/ : SUCCESS in 35m 00s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/3f06ef3/ : FAILURE in 1h 22m 45s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-scenario-linuxbridge-queens/1cd4bf2/ : FAILURE in 1h 40m 45s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-api-rocky/866b487/ : SUCCESS in 48m 49s\n- neutron-tempest-plugin-designate-scenario-rocky http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-designate-scenario-rocky/15c50bf/ : SUCCESS in 33m 42s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-dvr-multinode-scenario-rocky/0527494/ : SUCCESS in 1h 18m 06s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-scenario-linuxbridge-rocky/2c7019e/ : SUCCESS in 46m 46s\n- neutron-tempest-plugin-api-stein http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-api-stein/0338eea/ : SUCCESS in 46m 03s\n- neutron-tempest-plugin-designate-scenario-stein http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-designate-scenario-stein/d6dd0bc/ : SUCCESS in 37m 15s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-dvr-multinode-scenario-stein/9b508e1/ : SUCCESS in 1h 36m 58s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-scenario-linuxbridge-stein/2f07c65/ : SUCCESS in 1h 06m 19s\n- openstack-tox-pep8 http://logs.openstack.org/07/619207/14/check/openstack-tox-pep8/7c1772c/ : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":14},{"id":"dc22168404f73b6215f5e988bf425ac1c475fba8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-12 07:33:42.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"37ff27e796e250f247cdc6e072eafcc028e604da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 10:03:11.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-api/21aa847/ : SUCCESS in 46m 20s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-designate-scenario/dc2f8d5/ : SUCCESS in 32m 16s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-dvr-multinode-scenario/eb605f6/ : SUCCESS in 1h 32m 06s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-scenario-linuxbridge/4453c22/ : SUCCESS in 1h 11m 52s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/619207/14/check/build-openstack-sphinx-docs/d04951a/html/ : SUCCESS in 6m 13s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-api-queens/258eadf/ : SUCCESS in 47m 56s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-designate-scenario-queens/320243e/ : SUCCESS in 33m 44s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/efa434e/ : FAILURE in 1h 25m 56s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-scenario-linuxbridge-queens/1e682ff/ : SUCCESS in 45m 10s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-api-rocky/542ca7e/ : SUCCESS in 57m 08s\n- neutron-tempest-plugin-designate-scenario-rocky http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-designate-scenario-rocky/00a3188/ : SUCCESS in 40m 41s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-dvr-multinode-scenario-rocky/d2d0bad/ : FAILURE in 1h 26m 09s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-scenario-linuxbridge-rocky/18a31af/ : SUCCESS in 1h 06m 39s\n- neutron-tempest-plugin-api-stein http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-api-stein/0368292/ : SUCCESS in 45m 19s\n- neutron-tempest-plugin-designate-scenario-stein http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-designate-scenario-stein/741e38e/ : SUCCESS in 36m 58s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-dvr-multinode-scenario-stein/fd99162/ : FAILURE in 1h 59m 11s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein http://logs.openstack.org/07/619207/14/check/neutron-tempest-plugin-scenario-linuxbridge-stein/bf710bf/ : SUCCESS in 1h 05m 55s\n- openstack-tox-pep8 http://logs.openstack.org/07/619207/14/check/openstack-tox-pep8/08b8c0f/ : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":14},{"id":"a2f8d4449a41e38ae98f47197623fe3f75083e03","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-04-12 14:00:27.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"98220abcfc2ce24df0206feea95e1fb6df7d8f09","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2019-04-14 08:57:36.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"84c218024dcbbe78e57e4236deb630402424b252","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-15 07:48:29.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"5dfc8b3bef2c665877858e3d048b8431ddcb1f38","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2019-04-15 08:37:49.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"77b226acf85ba6321d4593eb29984de2d2b95ad4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 09:50:14.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/07/619207/15/check/neutron-tempest-plugin-api/006d76d/ : SUCCESS in 48m 29s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/07/619207/15/check/neutron-tempest-plugin-designate-scenario/911400c/ : SUCCESS in 32m 15s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/07/619207/15/check/neutron-tempest-plugin-dvr-multinode-scenario/d0ebdfa/ : SUCCESS in 1h 27m 42s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/07/619207/15/check/neutron-tempest-plugin-scenario-linuxbridge/07b5123/ : SUCCESS in 1h 02m 16s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/619207/15/check/build-openstack-sphinx-docs/bcbfe14/html/ : SUCCESS in 5m 42s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/07/619207/15/check/neutron-tempest-plugin-api-queens/6951c8a/ : SUCCESS in 41m 48s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/07/619207/15/check/neutron-tempest-plugin-designate-scenario-queens/21d5c77/ : SUCCESS in 28m 33s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/07/619207/15/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/7f78f5c/ : FAILURE in 1h 50m 03s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/07/619207/15/check/neutron-tempest-plugin-scenario-linuxbridge-queens/9eb7465/ : SUCCESS in 44m 45s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/07/619207/15/check/neutron-tempest-plugin-api-rocky/b0ba253/ : SUCCESS in 43m 07s\n- neutron-tempest-plugin-designate-scenario-rocky http://logs.openstack.org/07/619207/15/check/neutron-tempest-plugin-designate-scenario-rocky/0f383c4/ : SUCCESS in 27m 49s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky http://logs.openstack.org/07/619207/15/check/neutron-tempest-plugin-dvr-multinode-scenario-rocky/508ebad/ : FAILURE in 1h 20m 51s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky http://logs.openstack.org/07/619207/15/check/neutron-tempest-plugin-scenario-linuxbridge-rocky/2928d93/ : SUCCESS in 48m 50s\n- neutron-tempest-plugin-api-stein http://logs.openstack.org/07/619207/15/check/neutron-tempest-plugin-api-stein/e9ffd5d/ : SUCCESS in 46m 52s\n- neutron-tempest-plugin-designate-scenario-stein http://logs.openstack.org/07/619207/15/check/neutron-tempest-plugin-designate-scenario-stein/9295d77/ : SUCCESS in 36m 31s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein http://logs.openstack.org/07/619207/15/check/neutron-tempest-plugin-dvr-multinode-scenario-stein/bc06c92/ : FAILURE in 1h 37m 39s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein http://logs.openstack.org/07/619207/15/check/neutron-tempest-plugin-scenario-linuxbridge-stein/fc4fc30/ : SUCCESS in 1h 00m 35s\n- openstack-tox-pep8 http://logs.openstack.org/07/619207/15/check/openstack-tox-pep8/44fbc8a/ : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":15},{"id":"f3ba38319dfef0614a0056de723b4f86c5ef2449","author":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"date":"2019-04-15 11:37:23.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"769e0cd6b5739f6a26ef588e2fb3f9ea0403fa98","author":{"_account_id":26899,"name":"Roee Agiman","email":"ragiman@redhat.com","username":"ragiman"},"date":"2019-04-15 11:45:35.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"2995b421f32de572a09faac402ac29278caacd0d","author":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"date":"2019-04-17 07:20:33.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"9a7510ad3d7b8a40acd9877ddd08a66b222d3e6a","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2019-04-18 08:51:07.000000000","message":"Patch Set 15:\n\nAccording to this addition, will we have to introduce/add this condition for each test? I mean, in tempest.scenario.test_network_basic_ops.TestNetworkBasicOps tests if we want to run them with RHEL, then we will have to change the test?","accounts_in_message":[],"_revision_number":15},{"id":"fc55dee7aaee7915b9a78f37ec059fe4464bc3da","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-18 08:56:17.000000000","message":"Patch Set 15:\n\n(1 comment)\n\n\u003e According to this addition, will we have to introduce/add this\n \u003e condition for each test? I mean, in tempest.scenario.test_network_basic_ops.TestNetworkBasicOps\n \u003e tests if we want to run them with RHEL, then we will have to change\n \u003e the test?\n\nNo. You can configure RHEL image as default image and it will be used in every test in same way like cirros is. For most of the tests it doesn\u0027t matter what image is used.\nBut there is couple of tests which really require \"advanced\" image because cirros don\u0027t have all necessary features to run such test. And for those tests we are doing this change to allow user to say in tempest\u0027s config file \"my default image is advanced and You can use it for this test too and it should work\".","accounts_in_message":[],"_revision_number":15},{"id":"9f2a1278a9e022d4f862c987052824394bca7ea0","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2019-04-23 10:38:13.000000000","message":"Patch Set 15:\n\nThis is making test cases that requires images even more complex to write. Why not to add additional advanced image to the cloud instead and let test cases to decide witch one to use? Those few test cases that require an advance images could be easily skept if only base image is available. Please note that in most of cases a CirrOS image is better than an advanced one because it is cheaper to run (VMs booting time is probably the major time killer in our tempest based CI jobs). Therefore I see no good reason not having two separate images and let test cases to decide which one to use, instead of forcing test cases having only one. Test cases are already too complex and tempest.conf managing too. Test cases are already too slow and time consumed doing unnecessary operations like booting an advanced image instead a cirros one has a visible cost for CI infrastructure should be better spent running more tests.","accounts_in_message":[],"_revision_number":15},{"id":"18bb583d60c93730f070b5896e5b7096ba18cbf7","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2019-04-23 10:38:46.000000000","message":"Patch Set 15: Code-Review-1","accounts_in_message":[],"_revision_number":15},{"id":"669342d208f57679d5777a8fd666b3023b25c83d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-23 11:00:14.000000000","message":"Patch Set 15:\n\n\u003e This is making test cases that requires images even more complex to\n \u003e write. Why not to add additional advanced image to the cloud\n \u003e instead and let test cases to decide witch one to use? Those few\n \u003e test cases that require an advance images could be easily skept if\n \u003e only base image is available. Please note that in most of cases a\n \u003e CirrOS image is better than an advanced one because it is cheaper\n \u003e to run (VMs booting time is probably the major time killer in our\n \u003e tempest based CI jobs). Therefore I see no good reason not having\n \u003e two separate images and let test cases to decide which one to use,\n \u003e instead of forcing test cases having only one. Test cases are\n \u003e already too complex and tempest.conf managing too. Test cases are\n \u003e already too slow and time consumed doing unnecessary operations\n \u003e like booting an advanced image instead a cirros one has a visible\n \u003e cost for CI infrastructure should be better spent running more\n \u003e tests.\n\nIn upstream CI, where we are using devstack it is totally true that it\u0027s better to use cirros and it is easy to configure to images. And that is still done here.\n\nBut please note that it is easy to configure two images easy because we are using neutron-tempest-plugin\u0027s devstack plugin which can download proper image, upload it to glance and set its ID in tempest\u0027s config file.\n\nUnfortunatelly it isn\u0027t the case always e.g. in D/S CI (like we have in RH for example). If there is no devstack used to deploy cloud, there is no easy way to download another image and configure it in tempest\u0027s config file.\nSo, if in such CI job already e.g. RHEL image is used as guest image, it may be very trivial change to add only this one config option with boolean value there and test will run and use \"default\" image as an advanced one.\n\nPlease also note that using advanced image like RHEL or Ubuntu shouldn\u0027t be very expensive if nested virtualisation can be used. Unfortunatelly we can\u0027t use it in U/S CI as it causes some problems on some cloud providers. That\u0027s why it is much easier for us to use 2 different images to makes test run faster.","accounts_in_message":[],"_revision_number":15},{"id":"d4a77353475c192a4a00eea990cfe3dd54970f3e","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-04-23 11:11:34.000000000","message":"Patch Set 15: Code-Review+2\n\n(1 comment)\n\nI don\u0027t have any problem with this, and I think it is a good solution given the problem at hand.","accounts_in_message":[],"_revision_number":15},{"id":"8c05dd4b78440867c2717608e2fc99726cca62ab","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-04-23 12:42:25.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"055e7464e0f1811839936a8a7f3eb6de86f3ed65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 12:42:59.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"474bb2bfe52a25ed0281191888654bf9fa7bbefd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 19:33:16.000000000","message":"Patch Set 15:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"745c0255bbc847c88416310b786c60d6c1af7688","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 20:51:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15},{"id":"30a648a78dfce4455f3692e547e238e1f610ff5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 20:51:21.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/07/619207/15/gate/neutron-tempest-plugin-api/5ce50ca/ : SUCCESS in 51m 17s\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/07/619207/15/gate/neutron-tempest-plugin-scenario-linuxbridge/edda914/ : SUCCESS in 1h 17m 25s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/619207/15/gate/build-openstack-sphinx-docs/8c8578a/html/ : SUCCESS in 6m 22s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/07/619207/15/gate/neutron-tempest-plugin-api-queens/69553cf/ : SUCCESS in 48m 23s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/07/619207/15/gate/neutron-tempest-plugin-api-rocky/b4d00a4/ : SUCCESS in 54m 55s\n- neutron-tempest-plugin-api-stein http://logs.openstack.org/07/619207/15/gate/neutron-tempest-plugin-api-stein/82e7d69/ : SUCCESS in 53m 20s\n- openstack-tox-pep8 http://logs.openstack.org/07/619207/15/gate/openstack-tox-pep8/ff887ba/ : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"fa1081ad9f607b7a45a406eca7fbf48de206409b","revisions":{"d2528f994103f92080e6aaf153e6276593ba91f2":{"kind":"REWORK","_number":1,"created":"2018-11-21 10:48:07.000000000","uploader":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"ref":"refs/changes/07/619207/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/07/619207/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/1"}}},"commit":{"parents":[{"commit":"6f9bb77d9872a1cac4e1815ce13d97b8a143b3a8","subject":"Fix race condition in test_security_groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6f9bb77d9872a1cac4e1815ce13d97b8a143b3a8"}]}],"author":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-11-21 10:46:57.000000000","tz":120},"committer":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-11-21 10:46:57.000000000","tz":120},"subject":"Fix test-trunk tests based on NFV ci failures","message":"Fix test-trunk tests based on NFV ci failures\n\nChange-Id: Ic78ee64a9350863ff5bd4b4baa4b20770e1b9d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d2528f994103f92080e6aaf153e6276593ba91f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d2528f994103f92080e6aaf153e6276593ba91f2"}]},"branch":"refs/heads/master"},"208263a5bad6cfd79bc931f443f2b6339070c273":{"kind":"REWORK","_number":2,"created":"2018-11-21 16:34:03.000000000","uploader":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"ref":"refs/changes/07/619207/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/07/619207/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/2"}}},"commit":{"parents":[{"commit":"6f9bb77d9872a1cac4e1815ce13d97b8a143b3a8","subject":"Fix race condition in test_security_groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6f9bb77d9872a1cac4e1815ce13d97b8a143b3a8"}]}],"author":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-11-21 10:46:57.000000000","tz":120},"committer":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-11-21 14:51:18.000000000","tz":120},"subject":"Fix test-trunk tests based on NFV ci failures","message":"Fix test-trunk tests based on NFV ci failures\n\nNFV CI is running this tests continuously\nThis patch\nhttps://review.openstack.org/#/c/609762/16/neutron_tempest_plugin/config.py\nbroke CI, since it removed plugin_option image_is_advanced, type boolean.\nSo our suggestions here, is to add another var check to if query.\nIn this case, we use the advanced_image_ref as non None var as indicator for\nimage_is_advanced instead, and apply another Non check advanced_image_flavor_ref.\nOnce can still use compute.flavor_ref compute.image_ref as non cirros.\n\nChange-Id: Ic78ee64a9350863ff5bd4b4baa4b20770e1b9d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/208263a5bad6cfd79bc931f443f2b6339070c273"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/208263a5bad6cfd79bc931f443f2b6339070c273"}]},"branch":"refs/heads/master"},"8fcc5fcab0631ce2e2bd23f3df461593dc7efb43":{"kind":"REWORK","_number":3,"created":"2018-11-22 15:57:34.000000000","uploader":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"ref":"refs/changes/07/619207/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/07/619207/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/3"}}},"commit":{"parents":[{"commit":"6f9bb77d9872a1cac4e1815ce13d97b8a143b3a8","subject":"Fix race condition in test_security_groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6f9bb77d9872a1cac4e1815ce13d97b8a143b3a8"}]}],"author":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-11-21 10:46:57.000000000","tz":120},"committer":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-11-22 15:56:31.000000000","tz":120},"subject":"Fix test-trunk tests based on NFV ci failures","message":"Fix test-trunk tests based on NFV ci failures\n\nNFV CI is running this tests continuously\nThis patch\nhttps://review.openstack.org/#/c/609762/16/neutron_tempest_plugin/config.py\nbroke NFV CI, since it removed plugin_option image_is_advanced, type boolean.\nSo my suggestions here:\n Is to reverting back existence of image_is_advanced.\nAnd set configuration advanced parameters to equivalent parameters In compute\n\nChange-Id: Ic78ee64a9350863ff5bd4b4baa4b20770e1b9d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8fcc5fcab0631ce2e2bd23f3df461593dc7efb43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8fcc5fcab0631ce2e2bd23f3df461593dc7efb43"}]},"branch":"refs/heads/master"},"085f0140f32b39335380d76d2db507b580ff79c4":{"kind":"REWORK","_number":4,"created":"2018-11-26 16:22:58.000000000","uploader":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"ref":"refs/changes/07/619207/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/07/619207/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/4"}}},"commit":{"parents":[{"commit":"6f9bb77d9872a1cac4e1815ce13d97b8a143b3a8","subject":"Fix race condition in test_security_groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6f9bb77d9872a1cac4e1815ce13d97b8a143b3a8"}]}],"author":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-11-21 10:46:57.000000000","tz":120},"committer":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-11-26 16:20:31.000000000","tz":120},"subject":"Fix test-trunk tests based on NFV ci failures","message":"Fix test-trunk tests based on NFV ci failures\n\nNFV CI is running this tests continuously\nThis patch\nhttps://review.openstack.org/#/c/609762/16/neutron_tempest_plugin/config.py\nbroke NFV CI, since it removed plugin_option image_is_advanced, type boolean.\nSo new suggestions here:\n Is adding new config_param default_image_is_cirros default value True.\n Adding in lifecycle param value check before using advanced_image params\n\nChange-Id: Ic78ee64a9350863ff5bd4b4baa4b20770e1b9d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/085f0140f32b39335380d76d2db507b580ff79c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/085f0140f32b39335380d76d2db507b580ff79c4"}]},"branch":"refs/heads/master"},"a642d7d0c79a1f8dafdbaf710ade733a8bc74823":{"kind":"REWORK","_number":5,"created":"2018-11-26 16:24:29.000000000","uploader":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"ref":"refs/changes/07/619207/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/07/619207/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/5"}}},"commit":{"parents":[{"commit":"6f9bb77d9872a1cac4e1815ce13d97b8a143b3a8","subject":"Fix race condition in test_security_groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6f9bb77d9872a1cac4e1815ce13d97b8a143b3a8"}]}],"author":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-11-21 10:46:57.000000000","tz":120},"committer":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-11-26 16:24:18.000000000","tz":120},"subject":"Fix test-trunk tests based on NFV ci failures","message":"Fix test-trunk tests based on NFV ci failures\n\nNFV CI is running this tests continuously\nThis patch\nhttps://review.openstack.org/#/c/609762/16/neutron_tempest_plugin/config.py\nbroke NFV CI, since it removed plugin_option image_is_advanced, type boolean.\nSo new suggestions here:\n Is adding new config_param default_image_is_cirros default value True.\n Adding in lifecycle param value check before using advanced_image params\n\nChange-Id: Ic78ee64a9350863ff5bd4b4baa4b20770e1b9d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a642d7d0c79a1f8dafdbaf710ade733a8bc74823"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a642d7d0c79a1f8dafdbaf710ade733a8bc74823"}]},"branch":"refs/heads/master"},"c1f71df74a9a10acc29b18dccd99d4aaaf189991":{"kind":"REWORK","_number":6,"created":"2018-11-26 16:27:24.000000000","uploader":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"ref":"refs/changes/07/619207/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/07/619207/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/6"}}},"commit":{"parents":[{"commit":"6f9bb77d9872a1cac4e1815ce13d97b8a143b3a8","subject":"Fix race condition in test_security_groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6f9bb77d9872a1cac4e1815ce13d97b8a143b3a8"}]}],"author":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-11-21 10:46:57.000000000","tz":120},"committer":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-11-26 16:27:13.000000000","tz":120},"subject":"Fix test-trunk tests based on NFV ci failures","message":"Fix test-trunk tests based on NFV ci failures\n\nNFV CI is running this tests continuously\nThis patch\nhttps://review.openstack.org/#/c/609762/16/neutron_tempest_plugin/config.py\nbroke NFV CI, since it removed plugin_option image_is_advanced, type boolean.\nSo new suggestions here:\n Is adding new config_param default_image_is_cirros default value True.\n Adding in lifecycle param value check before using advanced_image params\n\nChange-Id: Ic78ee64a9350863ff5bd4b4baa4b20770e1b9d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c1f71df74a9a10acc29b18dccd99d4aaaf189991"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c1f71df74a9a10acc29b18dccd99d4aaaf189991"}]},"branch":"refs/heads/master"},"7873246099c3dc6de20d366d1ece1cffb3ca3960":{"kind":"REWORK","_number":7,"created":"2018-12-09 09:48:25.000000000","uploader":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"ref":"refs/changes/07/619207/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/07/619207/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/7"}}},"commit":{"parents":[{"commit":"6f9bb77d9872a1cac4e1815ce13d97b8a143b3a8","subject":"Fix race condition in test_security_groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6f9bb77d9872a1cac4e1815ce13d97b8a143b3a8"}]}],"author":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-11-21 10:46:57.000000000","tz":120},"committer":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-12-09 09:40:04.000000000","tz":120},"subject":"Fix test-trunk tests based on NFV ci failures","message":"Fix test-trunk tests based on NFV ci failures\n\nNFV CI is running this tests continuously\nThis patch\nhttps://review.openstack.org/#/c/609762/16/neutron_tempest_plugin/config.py\nbroke NFV CI, since it removed plugin_option image_is_advanced, type boolean.\nSo suggestions here as follow:\n 1) Reverting back existence of image_is_advanced.\n 2) Add another check for advanced vars image/flavor/ssh ref is not None\n\nBoth conditions advanced fulfilled test run as advanced neutron way,\nIn case only (1) using default image and flavors as per NFV CI\n\nChange-Id: Ic78ee64a9350863ff5bd4b4baa4b20770e1b9d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7873246099c3dc6de20d366d1ece1cffb3ca3960"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7873246099c3dc6de20d366d1ece1cffb3ca3960"}]},"branch":"refs/heads/master"},"ccf0ef7a646a8a2a16e590c77436123a4e01923b":{"kind":"REWORK","_number":8,"created":"2018-12-09 09:50:24.000000000","uploader":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"ref":"refs/changes/07/619207/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/07/619207/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/8"}}},"commit":{"parents":[{"commit":"6f9bb77d9872a1cac4e1815ce13d97b8a143b3a8","subject":"Fix race condition in test_security_groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6f9bb77d9872a1cac4e1815ce13d97b8a143b3a8"}]}],"author":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-11-21 10:46:57.000000000","tz":120},"committer":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-12-09 09:50:15.000000000","tz":120},"subject":"Fix test-trunk tests based on NFV ci failures","message":"Fix test-trunk tests based on NFV ci failures\n\nNFV CI is running this tests continuously\nThis patch\nhttps://review.openstack.org/#/c/609762/16/neutron_tempest_plugin/config.py\nbroke NFV CI, since it removed plugin_option image_is_advanced, type boolean.\nSo suggestions here as follow:\n 1) Reverting back existence of image_is_advanced.\n 2) Add another check for advanced vars image/flavor/ssh ref is not None\n\nBoth conditions advanced fulfilled test run as advanced neutron way,\nIn case only (1) using default image and flavors as per NFV CI\n\nChange-Id: Ic78ee64a9350863ff5bd4b4baa4b20770e1b9d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ccf0ef7a646a8a2a16e590c77436123a4e01923b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ccf0ef7a646a8a2a16e590c77436123a4e01923b"}]},"branch":"refs/heads/master"},"8eaef667f5cee3f02683ab8193e1cd339380b5d4":{"kind":"REWORK","_number":9,"created":"2018-12-13 08:56:37.000000000","uploader":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"ref":"refs/changes/07/619207/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/07/619207/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/9"}}},"commit":{"parents":[{"commit":"6f9bb77d9872a1cac4e1815ce13d97b8a143b3a8","subject":"Fix race condition in test_security_groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6f9bb77d9872a1cac4e1815ce13d97b8a143b3a8"}]}],"author":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-11-21 10:46:57.000000000","tz":120},"committer":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-12-13 08:50:22.000000000","tz":120},"subject":"Fix test-trunk tests based on NFV ci failures","message":"Fix test-trunk tests based on NFV ci failures\n\nNFV CI is running this tests continuously\nThis patch\nhttps://review.openstack.org/#/c/609762/16/neutron_tempest_plugin/config.py\nbroke NFV CI, since it removed plugin_option image_is_advanced, type boolean.\nSo suggestions here as follow:\n 1) Adding config_param default_image_is_cirros default value True.\n 2) Add another check for default_image_is_cirros exsitance, not to harm defaults\nof neutron tests\n\nBoth conditions advanced fulfilled test run as advanced neutron way,\nIn case only (1) using default image and flavors as per NFV CI\n\nChange-Id: Ic78ee64a9350863ff5bd4b4baa4b20770e1b9d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8eaef667f5cee3f02683ab8193e1cd339380b5d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8eaef667f5cee3f02683ab8193e1cd339380b5d4"}]},"branch":"refs/heads/master"},"8706e7adba1c777a92eed43e584bc76f3edb4283":{"kind":"REWORK","_number":10,"created":"2018-12-13 12:06:44.000000000","uploader":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"ref":"refs/changes/07/619207/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/07/619207/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/10"}}},"commit":{"parents":[{"commit":"6f9bb77d9872a1cac4e1815ce13d97b8a143b3a8","subject":"Fix race condition in test_security_groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6f9bb77d9872a1cac4e1815ce13d97b8a143b3a8"}]}],"author":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-11-21 10:46:57.000000000","tz":120},"committer":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-12-13 12:06:23.000000000","tz":120},"subject":"Fix test-trunk tests based on NFV ci failures","message":"Fix test-trunk tests based on NFV ci failures\n\nNFV CI is running this tests continuously\nThis patch\nhttps://review.openstack.org/#/c/609762/16/neutron_tempest_plugin/config.py\nbroke NFV CI, since it removed plugin_option image_is_advanced, type boolean.\nSo suggestions here as follow:\n 1) Adding config_param default_image_is_cirros default value True.\n 2) Add another check for default_image_is_cirros exsitance, not to harm defaults\nof neutron tests\n\nBoth conditions advanced fulfilled test run as advanced neutron way,\nIn case only (1) using default image and flavors as per NFV CI\n\nChange-Id: Ic78ee64a9350863ff5bd4b4baa4b20770e1b9d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8706e7adba1c777a92eed43e584bc76f3edb4283"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8706e7adba1c777a92eed43e584bc76f3edb4283"}]},"branch":"refs/heads/master"},"dc35f4959962668d16af5781d37fd97ea9292e44":{"kind":"REWORK","_number":11,"created":"2018-12-13 12:10:46.000000000","uploader":{"_account_id":20171,"name":"Yariv","email":"yrachman@redhat.com","username":"Yarboa"},"ref":"refs/changes/07/619207/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/07/619207/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/11"}}},"commit":{"parents":[{"commit":"6f9bb77d9872a1cac4e1815ce13d97b8a143b3a8","subject":"Fix race condition in test_security_groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6f9bb77d9872a1cac4e1815ce13d97b8a143b3a8"}]}],"author":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-11-21 10:46:57.000000000","tz":120},"committer":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-12-13 12:10:21.000000000","tz":120},"subject":"Fix test-trunk tests based on NFV ci failures","message":"Fix test-trunk tests based on NFV ci failures\n\nNFV CI is running this tests continuously\nThis patch\nhttps://review.openstack.org/#/c/609762/16/neutron_tempest_plugin/config.py\nbroke NFV CI, since it removed plugin_option image_is_advanced, type boolean.\nSo suggestions here as follow:\n 1) Adding config_param default_image_is_cirros default value True.\n 2) Add another check for default_image_is_cirros exsitance, not to harm defaults\nof neutron tests\n\nBoth conditions advanced fulfilled test run as advanced neutron way,\nIn case only (1) using default image and flavors as per NFV CI\n\nChange-Id: Ic78ee64a9350863ff5bd4b4baa4b20770e1b9d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/dc35f4959962668d16af5781d37fd97ea9292e44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/dc35f4959962668d16af5781d37fd97ea9292e44"}]},"branch":"refs/heads/master"},"130a6aae1fea282f518d8466887f63c3144a5347":{"kind":"REWORK","_number":12,"created":"2019-04-11 15:17:50.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/07/619207/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/07/619207/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/12"}}},"commit":{"parents":[{"commit":"a57f80d1809a7d2a58e36ccd7d82c082f0be7080","subject":"Add definition of jobs running on stable/stein branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a57f80d1809a7d2a58e36ccd7d82c082f0be7080"}]}],"author":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-11-21 10:46:57.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-04-11 15:17:48.000000000","tz":120},"subject":"Add additional config option \"default_image_is_advanced\"","message":"Add additional config option \"default_image_is_advanced\"\n\nIn patch [1] we added new config option to configure \"advanced\" image\nwhich we are using in some tests which require it and can\u0027t run using\ndefault Cirros image.\nUnfortunatelly this made life of some downstream users harder as they\nhave CI systems which don\u0027t use devstack and don\u0027t have easy way to\nload another \"advanced\" image and use it for those tests.\n\nSo now this patch adds one ore config option \"default_image_is_advanced\"\nwhich has boolean value and defaults to False.\nIf this is set to True, tests which require advanced image will not be\nskipped and default image will be used for those tests.\n\n[1] https://review.openstack.org/#/c/609762/16/neutron_tempest_plugin/config.py\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: Ic78ee64a9350863ff5bd4b4baa4b20770e1b9d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/130a6aae1fea282f518d8466887f63c3144a5347"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/130a6aae1fea282f518d8466887f63c3144a5347"}]},"branch":"refs/heads/master"},"3bb49a6a68efe2023367b73a511e8f1a50a027c9":{"kind":"REWORK","_number":13,"created":"2019-04-11 20:15:34.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/07/619207/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/07/619207/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/13"}}},"commit":{"parents":[{"commit":"a57f80d1809a7d2a58e36ccd7d82c082f0be7080","subject":"Add definition of jobs running on stable/stein branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a57f80d1809a7d2a58e36ccd7d82c082f0be7080"}]}],"author":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-11-21 10:46:57.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-04-11 20:15:23.000000000","tz":120},"subject":"Add additional config option \"default_image_is_advanced\"","message":"Add additional config option \"default_image_is_advanced\"\n\nIn patch [1] we added new config option to configure \"advanced\" image\nwhich we are using in some tests which require it and can\u0027t run using\ndefault Cirros image.\nUnfortunatelly this made life of some downstream users harder as they\nhave CI systems which don\u0027t use devstack and don\u0027t have easy way to\nload another \"advanced\" image and use it for those tests.\n\nSo now this patch adds one ore config option \"default_image_is_advanced\"\nwhich has boolean value and defaults to False.\nIf this is set to True, tests which require advanced image will not be\nskipped and default image will be used for those tests.\n\n[1] https://review.openstack.org/#/c/609762/16/neutron_tempest_plugin/config.py\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: Ic78ee64a9350863ff5bd4b4baa4b20770e1b9d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3bb49a6a68efe2023367b73a511e8f1a50a027c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3bb49a6a68efe2023367b73a511e8f1a50a027c9"}]},"branch":"refs/heads/master"},"f286b69d4dbeca0408e6afbc21730d5dccef4eef":{"kind":"REWORK","_number":14,"created":"2019-04-11 20:17:32.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/07/619207/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/07/619207/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/14"}}},"commit":{"parents":[{"commit":"a57f80d1809a7d2a58e36ccd7d82c082f0be7080","subject":"Add definition of jobs running on stable/stein branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a57f80d1809a7d2a58e36ccd7d82c082f0be7080"}]}],"author":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-11-21 10:46:57.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-04-11 20:17:22.000000000","tz":120},"subject":"Add additional config option \"default_image_is_advanced\"","message":"Add additional config option \"default_image_is_advanced\"\n\nIn patch [1] we added new config option to configure \"advanced\" image\nwhich we are using in some tests which require it and can\u0027t run using\ndefault Cirros image.\nUnfortunatelly this made life of some downstream users harder as they\nhave CI systems which don\u0027t use devstack and don\u0027t have easy way to\nload another \"advanced\" image and use it for those tests.\n\nSo now this patch adds one ore config option \"default_image_is_advanced\"\nwhich has boolean value and defaults to False.\nIf this is set to True, tests which require advanced image will not be\nskipped and default image will be used for those tests.\n\n[1] https://review.openstack.org/#/c/609762/16/neutron_tempest_plugin/config.py\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: Ic78ee64a9350863ff5bd4b4baa4b20770e1b9d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f286b69d4dbeca0408e6afbc21730d5dccef4eef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f286b69d4dbeca0408e6afbc21730d5dccef4eef"}]},"branch":"refs/heads/master"},"fa1081ad9f607b7a45a406eca7fbf48de206409b":{"kind":"REWORK","_number":15,"created":"2019-04-15 07:48:29.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/07/619207/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/07/619207/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/619207/15"}}},"commit":{"parents":[{"commit":"a57f80d1809a7d2a58e36ccd7d82c082f0be7080","subject":"Add definition of jobs running on stable/stein branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a57f80d1809a7d2a58e36ccd7d82c082f0be7080"}]}],"author":{"name":"Yariv Rachmani","email":"yrachman@redhat.com","date":"2018-11-21 10:46:57.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-04-15 07:47:12.000000000","tz":0},"subject":"Add additional config option \"default_image_is_advanced\"","message":"Add additional config option \"default_image_is_advanced\"\n\nIn patch [1] we added new config option to configure \"advanced\" image\nwhich we are using in some tests which require it and can\u0027t run using\ndefault Cirros image.\nUnfortunatelly this made life of some downstream users harder as they\nhave CI systems which don\u0027t use devstack and don\u0027t have easy way to\nload another \"advanced\" image and use it for those tests.\n\nSo now this patch adds one ore config option \"default_image_is_advanced\"\nwhich has boolean value and defaults to False.\nIf this is set to True, tests which require advanced image will not be\nskipped and default image will be used for those tests.\n\n[1] https://review.openstack.org/#/c/609762/16/neutron_tempest_plugin/config.py\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: Ic78ee64a9350863ff5bd4b4baa4b20770e1b9d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/fa1081ad9f607b7a45a406eca7fbf48de206409b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/fa1081ad9f607b7a45a406eca7fbf48de206409b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
