)]}'
{"id":"openstack%2Fneutron-tempest-plugin~660175","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I728cd8252d27e27e91bd95e4734d9db470dee35a","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/1824248","hashtags":[],"change_id":"I728cd8252d27e27e91bd95e4734d9db470dee35a","subject":"Add API test case to check if SG displays all rules","status":"MERGED","created":"2019-05-20 16:53:04.000000000","updated":"2019-05-28 17:43:05.000000000","submitted":"2019-05-28 17:43:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660175-1559065385665-0bd62daf","meta_rev_id":"2a0fc2f86a5f5a2d7f487033ea5938f5456055cc","_number":660175,"virtual_id_number":660175,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2019-05-28 17:43:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2019-05-24 22:45:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-05-24 21:39:34.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2019-05-24 22:45:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-23 14:01:58.000000000","updated_by":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"reviewer":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"state":"REVIEWER"},{"updated":"2019-05-24 15:21:16.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2019-05-24 21:39:34.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-05-24 22:45:01.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-05-28 17:43:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5359c77e983502ade1f8103a514a25fed223a07c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-20 16:53:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df9ad4518a039aa3953716541e09d987635a7de4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 22:39:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/75/660175/1/check/neutron-tempest-plugin-api/091bbb6/ : SUCCESS in 52m 09s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/75/660175/1/check/neutron-tempest-plugin-designate-scenario/f985f21/ : SUCCESS in 39m 31s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/75/660175/1/check/neutron-tempest-plugin-dvr-multinode-scenario/484eabe/ : FAILURE in 1h 42m 23s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/75/660175/1/check/neutron-tempest-plugin-scenario-linuxbridge/b94cdf0/ : SUCCESS in 1h 06m 56s\n- build-openstack-sphinx-docs http://logs.openstack.org/75/660175/1/check/build-openstack-sphinx-docs/ce40e91/html/ : SUCCESS in 4m 18s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/75/660175/1/check/neutron-tempest-plugin-api-queens/a435f70/ : FAILURE in 46m 35s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/75/660175/1/check/neutron-tempest-plugin-designate-scenario-queens/cc7b267/ : SUCCESS in 30m 41s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/75/660175/1/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/c2579ae/ : FAILURE in 1h 34m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/75/660175/1/check/neutron-tempest-plugin-scenario-linuxbridge-queens/665ecc0/ : SUCCESS in 59m 55s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/75/660175/1/check/neutron-tempest-plugin-api-rocky/dbb1ae2/ : FAILURE in 49m 01s\n- neutron-tempest-plugin-designate-scenario-rocky http://logs.openstack.org/75/660175/1/check/neutron-tempest-plugin-designate-scenario-rocky/35a8ffd/ : SUCCESS in 33m 39s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky http://logs.openstack.org/75/660175/1/check/neutron-tempest-plugin-dvr-multinode-scenario-rocky/5fe8ec5/ : FAILURE in 1h 33m 23s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky http://logs.openstack.org/75/660175/1/check/neutron-tempest-plugin-scenario-linuxbridge-rocky/be7222b/ : SUCCESS in 46m 55s\n- neutron-tempest-plugin-api-stein http://logs.openstack.org/75/660175/1/check/neutron-tempest-plugin-api-stein/7cc9b29/ : FAILURE in 44m 39s\n- neutron-tempest-plugin-designate-scenario-stein http://logs.openstack.org/75/660175/1/check/neutron-tempest-plugin-designate-scenario-stein/feb7151/ : SUCCESS in 33m 57s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein http://logs.openstack.org/75/660175/1/check/neutron-tempest-plugin-dvr-multinode-scenario-stein/46cc4ed/ : FAILURE in 1h 50m 30s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein http://logs.openstack.org/75/660175/1/check/neutron-tempest-plugin-scenario-linuxbridge-stein/c92cca4/ : SUCCESS in 1h 20m 22s\n- openstack-tox-pep8 http://logs.openstack.org/75/660175/1/check/openstack-tox-pep8/5347117/ : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"3f33e4062f83ec880119716325e0d454d99badcd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-21 12:28:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cfbb991cf823ffbecb68f1b5a03b849a5b23b1cf","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-05-21 16:40:57.000000000","message":"Patch Set 1:\n\nI wonder how this will pass if it\u0027s run against a stable branch?","accounts_in_message":[],"_revision_number":1},{"id":"3c24cbd523897256b911679bfaf269c1fcb5947e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-21 17:12:55.000000000","message":"Patch Set 1:\n\n\u003e I wonder how this will pass if it\u0027s run against a stable branch?\n\nThat is interesting question :)\nAt least it proves that fix works fine.\n\nI see 3 possible solutions here:\n - use API extension to make this change discoverable (I don\u0027t think it is good idea),\n - wait with this patch until patch in neutron will be merged and back ported up to Queens,\n - add this test to blacklist in stable branches job for now - and remove it when patch in neutron will be backported to stable branch.\n\nIMHO 3rd solution is the best but what do You think?","accounts_in_message":[],"_revision_number":1},{"id":"84dbba5734f7e26860ec3c9b56422f55ad4884f7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-05-21 18:06:38.000000000","message":"Patch Set 1:\n\n\u003e - add this test to blacklist in stable branches job for now - and remove it when patch in neutron will be backported to stable branch.\n\nHow do we do this, isn\u0027t it a chicken/egg problem?  We can\u0027t blacklist the test until it\u0027s added and we can\u0027t add it until it\u0027s blacklisted :)\nOr do you just mark it unstable here?\n\nAnd it raises a question - won\u0027t we still possibly break downstream consumers if they don\u0027t have the neutron change?  Just thinking out loud.","accounts_in_message":[],"_revision_number":1},{"id":"1e6c441582229a8b7034d349e29dd6ff4c88b30a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-21 20:25:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"169e3aaee3f8fd2a5beda710cf289fbf1a240a49","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-21 20:27:35.000000000","message":"Patch Set 2:\n\n\u003e \u003e - add this test to blacklist in stable branches job for now - and\n \u003e remove it when patch in neutron will be backported to stable\n \u003e branch.\n \u003e \n \u003e How do we do this, isn\u0027t it a chicken/egg problem?  We can\u0027t\n \u003e blacklist the test until it\u0027s added and we can\u0027t add it until it\u0027s\n \u003e blacklisted :)\n \u003e Or do you just mark it unstable here?\n\nI just added this test to be blacklisted in tempest regex in specific jobs. That should works IMO but lets see how it will be :)\n\n \u003e \n \u003e And it raises a question - won\u0027t we still possibly break downstream\n \u003e consumers if they don\u0027t have the neutron change?  Just thinking out\n \u003e loud.\n\nYes, basically that can break some d/s product which is using neutron-tempest-plugin from master branch and will run it against any stable release up to Queens.\nBut I really don\u0027t know what we can do we it except backport this fix ASAP to stable branches. Do You have any idea what we can do with this?","accounts_in_message":[],"_revision_number":2},{"id":"e2328e7722ae30c0385fbb1efd2f230e19ec8bfa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 00:37:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/75/660175/2/check/openstack-tox-docs/6e5ad0d/html/ : SUCCESS in 4m 25s\n- neutron-tempest-plugin-api http://logs.openstack.org/75/660175/2/check/neutron-tempest-plugin-api/e744788/ : SUCCESS in 43m 39s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/75/660175/2/check/neutron-tempest-plugin-designate-scenario/6cd0cce/ : SUCCESS in 32m 57s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/75/660175/2/check/neutron-tempest-plugin-dvr-multinode-scenario/0623d31/ : SUCCESS in 1h 43m 56s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/75/660175/2/check/neutron-tempest-plugin-scenario-linuxbridge/4782fa5/ : SUCCESS in 1h 05m 19s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/75/660175/2/check/neutron-tempest-plugin-api-queens/7818780/ : SUCCESS in 46m 32s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/75/660175/2/check/neutron-tempest-plugin-designate-scenario-queens/5d28a26/ : SUCCESS in 30m 57s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/75/660175/2/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/e930030/ : FAILURE in 1h 15m 47s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/75/660175/2/check/neutron-tempest-plugin-scenario-linuxbridge-queens/0be680d/ : SUCCESS in 58m 25s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/75/660175/2/check/neutron-tempest-plugin-api-rocky/bce6021/ : SUCCESS in 53m 43s\n- neutron-tempest-plugin-designate-scenario-rocky http://logs.openstack.org/75/660175/2/check/neutron-tempest-plugin-designate-scenario-rocky/3b48346/ : SUCCESS in 41m 42s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky http://logs.openstack.org/75/660175/2/check/neutron-tempest-plugin-dvr-multinode-scenario-rocky/0b6a4a2/ : FAILURE in 1h 15m 06s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky http://logs.openstack.org/75/660175/2/check/neutron-tempest-plugin-scenario-linuxbridge-rocky/46eebbd/ : SUCCESS in 54m 34s\n- neutron-tempest-plugin-api-stein http://logs.openstack.org/75/660175/2/check/neutron-tempest-plugin-api-stein/4b46152/ : SUCCESS in 58m 24s\n- neutron-tempest-plugin-designate-scenario-stein http://logs.openstack.org/75/660175/2/check/neutron-tempest-plugin-designate-scenario-stein/5b94db6/ : SUCCESS in 30m 29s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein http://logs.openstack.org/75/660175/2/check/neutron-tempest-plugin-dvr-multinode-scenario-stein/21cee5a/ : FAILURE in 1h 58m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein http://logs.openstack.org/75/660175/2/check/neutron-tempest-plugin-scenario-linuxbridge-stein/bddbeb8/ : SUCCESS in 1h 11m 26s\n- openstack-tox-pep8 http://logs.openstack.org/75/660175/2/check/openstack-tox-pep8/832d61c/ : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":2},{"id":"97508e7d9552b2f71c01293a444037073d5d6f51","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-05-22 03:11:22.000000000","message":"Patch Set 2:\n\nLet\u0027s see what Nate thinks, I guess I\u0027m Ok with this even though it\u0027s a little bit of a hack using the blacklist.","accounts_in_message":[],"_revision_number":2},{"id":"619d8e93ee9be3119352bef816a4de9b723ee8b9","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-05-22 19:29:45.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nDo we have any stadium projects that depend on neutron, that pull neutron from pip instead of pulling from master?  I\u0027m thinking of the issues we have had with neutron-lib and needing to wait for a published release for fixes, and I wonder since nearly all stadium projects consume neutron still if we might get ourselves in the same situation where a stadium project is broken until we get a new release of neutron.  So do we have any stadium projects that have jobs where tox_install_siblings is false?  \n\nIf that is the case, I don\u0027t see a way around putting a skip in for the test at least until we hit T-1.","accounts_in_message":[],"_revision_number":2},{"id":"4ec97d171845778ca93b244902e6c0e7a64f57ab","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-22 19:37:51.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n \u003e \n \u003e Do we have any stadium projects that depend on neutron, that pull\n \u003e neutron from pip instead of pulling from master?  I\u0027m thinking of\n \u003e the issues we have had with neutron-lib and needing to wait for a\n \u003e published release for fixes, and I wonder since nearly all stadium\n \u003e projects consume neutron still if we might get ourselves in the\n \u003e same situation where a stadium project is broken until we get a new\n \u003e release of neutron.  So do we have any stadium projects that have\n \u003e jobs where tox_install_siblings is false?\n \u003e \n \u003e If that is the case, I don\u0027t see a way around putting a skip in for\n \u003e the test at least until we hit T-1.\n\nLooking at http://codesearch.openstack.org/?q\u003d%5C%5Eneutron_tempest_plugin%5C.api\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d I found that only networking-ovn is running neutron_tempest_plugin.api tests but use neutron from master branch.\n\nOther possible solution for that issue is to propose backport of neutron patch now and make this one depend on all backport patches too. Then it should run fine for all stable releases here.\nWould it be better?","accounts_in_message":[],"_revision_number":2},{"id":"02603d77175d53423797540a9b2487c2cfe67a7e","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2019-05-23 14:01:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"28cda35914f6f22d023d6eccc5bdc2e9059a4e08","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-05-24 12:56:26.000000000","message":"Patch Set 2:\n\n\u003e \u003e (1 comment)\n \u003e \u003e\n \u003e \u003e Do we have any stadium projects that depend on neutron, that pull\n \u003e \u003e neutron from pip instead of pulling from master?  I\u0027m thinking of\n \u003e \u003e the issues we have had with neutron-lib and needing to wait for a\n \u003e \u003e published release for fixes, and I wonder since nearly all\n \u003e stadium\n \u003e \u003e projects consume neutron still if we might get ourselves in the\n \u003e \u003e same situation where a stadium project is broken until we get a\n \u003e new\n \u003e \u003e release of neutron.  So do we have any stadium projects that have\n \u003e \u003e jobs where tox_install_siblings is false?\n \u003e \u003e\n \u003e \u003e If that is the case, I don\u0027t see a way around putting a skip in\n \u003e for\n \u003e \u003e the test at least until we hit T-1.\n \u003e \n \u003e Looking at http://codesearch.openstack.org/?q\u003d%5C%5Eneutron_tempest_plugin%5C.api\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d\n \u003e I found that only networking-ovn is running neutron_tempest_plugin.api\n \u003e tests but use neutron from master branch.\n \u003e \n \u003e Other possible solution for that issue is to propose backport of\n \u003e neutron patch now and make this one depend on all backport patches\n \u003e too. Then it should run fine for all stable releases here.\n \u003e Would it be better?\n\nYes, I think that is the safest solution all around.","accounts_in_message":[],"_revision_number":2},{"id":"016c21ac31ff4b4fa9fe6dee4adaa435db278429","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-24 13:18:45.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"27c674ef0e3972c19037f32e18a7ccf342476a23","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-24 13:18:59.000000000","message":"Patch Set 3:\n\n\u003e \u003e \u003e (1 comment)\n \u003e \u003e \u003e\n \u003e \u003e \u003e Do we have any stadium projects that depend on neutron, that\n \u003e pull\n \u003e \u003e \u003e neutron from pip instead of pulling from master?  I\u0027m thinking\n \u003e of\n \u003e \u003e \u003e the issues we have had with neutron-lib and needing to wait for\n \u003e a\n \u003e \u003e \u003e published release for fixes, and I wonder since nearly all\n \u003e \u003e stadium\n \u003e \u003e \u003e projects consume neutron still if we might get ourselves in the\n \u003e \u003e \u003e same situation where a stadium project is broken until we get a\n \u003e \u003e new\n \u003e \u003e \u003e release of neutron.  So do we have any stadium projects that\n \u003e have\n \u003e \u003e \u003e jobs where tox_install_siblings is false?\n \u003e \u003e \u003e\n \u003e \u003e \u003e If that is the case, I don\u0027t see a way around putting a skip in\n \u003e \u003e for\n \u003e \u003e \u003e the test at least until we hit T-1.\n \u003e \u003e\n \u003e \u003e Looking at http://codesearch.openstack.org/?q\u003d%5C%5Eneutron_tempest_plugin%5C.api\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d\n \u003e \u003e I found that only networking-ovn is running neutron_tempest_plugin.api\n \u003e \u003e tests but use neutron from master branch.\n \u003e \u003e\n \u003e \u003e Other possible solution for that issue is to propose backport of\n \u003e \u003e neutron patch now and make this one depend on all backport\n \u003e patches\n \u003e \u003e too. Then it should run fine for all stable releases here.\n \u003e \u003e Would it be better?\n \u003e \n \u003e Yes, I think that is the safest solution all around.\n\nDone","accounts_in_message":[],"_revision_number":3},{"id":"6e5812d27d3612ea10355b40b6c77b91d5d36ec8","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-05-24 13:30:34.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ff0af8539460e63cbb584a199145f40305a8c4c3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-05-24 13:41:59.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1e7a4ab9de01f00af1b61400f628f5eae7338b27","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-24 13:51:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4b59e9f54312da31843b4607f1fd33993f75697b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-24 13:52:31.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"14aa4eb256334443d95fc1271100a6ccde8462ad","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-05-24 15:21:16.000000000","message":"Patch Set 4: Code-Review+1\n\nOk, accepted","accounts_in_message":[],"_revision_number":4},{"id":"e7d7e2f2f06e3aa70225b6410ff276bb968a0e5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 17:57:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/75/660175/4/check/openstack-tox-docs/f422bdc/html/ : SUCCESS in 6m 33s\n- neutron-tempest-plugin-api http://logs.openstack.org/75/660175/4/check/neutron-tempest-plugin-api/cf5c81b/ : SUCCESS in 49m 38s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/75/660175/4/check/neutron-tempest-plugin-designate-scenario/217854d/ : SUCCESS in 32m 04s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/75/660175/4/check/neutron-tempest-plugin-dvr-multinode-scenario/f939758/ : FAILURE in 1h 52m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/75/660175/4/check/neutron-tempest-plugin-scenario-linuxbridge/d5615a4/ : SUCCESS in 1h 24m 20s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/75/660175/4/check/neutron-tempest-plugin-api-queens/0fa8774/ : SUCCESS in 46m 17s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/75/660175/4/check/neutron-tempest-plugin-designate-scenario-queens/bbd8020/ : SUCCESS in 31m 14s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/75/660175/4/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/1e186cf/ : FAILURE in 1h 36m 37s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/75/660175/4/check/neutron-tempest-plugin-scenario-linuxbridge-queens/a89c9b5/ : SUCCESS in 51m 27s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/75/660175/4/check/neutron-tempest-plugin-api-rocky/3152451/ : SUCCESS in 46m 58s\n- neutron-tempest-plugin-designate-scenario-rocky http://logs.openstack.org/75/660175/4/check/neutron-tempest-plugin-designate-scenario-rocky/8cb9993/ : SUCCESS in 29m 38s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky http://logs.openstack.org/75/660175/4/check/neutron-tempest-plugin-dvr-multinode-scenario-rocky/c6ad9a1/ : FAILURE in 1h 34m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky http://logs.openstack.org/75/660175/4/check/neutron-tempest-plugin-scenario-linuxbridge-rocky/3ac5753/ : SUCCESS in 59m 07s\n- neutron-tempest-plugin-api-stein http://logs.openstack.org/75/660175/4/check/neutron-tempest-plugin-api-stein/9250faa/ : SUCCESS in 53m 49s\n- neutron-tempest-plugin-designate-scenario-stein http://logs.openstack.org/75/660175/4/check/neutron-tempest-plugin-designate-scenario-stein/014243a/ : SUCCESS in 35m 12s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein http://logs.openstack.org/75/660175/4/check/neutron-tempest-plugin-dvr-multinode-scenario-stein/5b2f8db/ : FAILURE in 1h 46m 16s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein http://logs.openstack.org/75/660175/4/check/neutron-tempest-plugin-scenario-linuxbridge-stein/881d510/ : SUCCESS in 1h 30m 16s\n- openstack-tox-pep8 http://logs.openstack.org/75/660175/4/check/openstack-tox-pep8/2727853/ : FAILURE in 4m 57s","accounts_in_message":[],"_revision_number":4},{"id":"35fee786f5e083f359a4788f81c2a02b5437f4df","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-24 19:02:57.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"dc8098f12d2e61aefe8faaa713711ae9b3097c83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 21:10:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/75/660175/5/check/openstack-tox-docs/7841f2b/html/ : SUCCESS in 4m 49s\n- neutron-tempest-plugin-api http://logs.openstack.org/75/660175/5/check/neutron-tempest-plugin-api/8e8c6cb/ : SUCCESS in 55m 55s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/75/660175/5/check/neutron-tempest-plugin-designate-scenario/209c1e1/ : SUCCESS in 32m 38s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/75/660175/5/check/neutron-tempest-plugin-dvr-multinode-scenario/6ee440a/ : FAILURE in 1h 46m 30s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/75/660175/5/check/neutron-tempest-plugin-scenario-linuxbridge/c4f654a/ : SUCCESS in 1h 11m 35s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/75/660175/5/check/neutron-tempest-plugin-api-queens/9a89812/ : SUCCESS in 48m 32s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/75/660175/5/check/neutron-tempest-plugin-designate-scenario-queens/d3f7b7d/ : SUCCESS in 33m 02s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/75/660175/5/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/e14d4ef/ : FAILURE in 1h 14m 12s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/75/660175/5/check/neutron-tempest-plugin-scenario-linuxbridge-queens/02376f1/ : SUCCESS in 54m 59s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/75/660175/5/check/neutron-tempest-plugin-api-rocky/115232d/ : SUCCESS in 45m 34s\n- neutron-tempest-plugin-designate-scenario-rocky http://logs.openstack.org/75/660175/5/check/neutron-tempest-plugin-designate-scenario-rocky/a6a0d5c/ : SUCCESS in 34m 54s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky http://logs.openstack.org/75/660175/5/check/neutron-tempest-plugin-dvr-multinode-scenario-rocky/5b4d880/ : FAILURE in 1h 16m 09s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky http://logs.openstack.org/75/660175/5/check/neutron-tempest-plugin-scenario-linuxbridge-rocky/ed668a6/ : SUCCESS in 57m 32s\n- neutron-tempest-plugin-api-stein http://logs.openstack.org/75/660175/5/check/neutron-tempest-plugin-api-stein/c226909/ : SUCCESS in 42m 00s\n- neutron-tempest-plugin-designate-scenario-stein http://logs.openstack.org/75/660175/5/check/neutron-tempest-plugin-designate-scenario-stein/6f8239e/ : SUCCESS in 30m 48s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein http://logs.openstack.org/75/660175/5/check/neutron-tempest-plugin-dvr-multinode-scenario-stein/2bcef4a/ : FAILURE in 1h 37m 35s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein http://logs.openstack.org/75/660175/5/check/neutron-tempest-plugin-scenario-linuxbridge-stein/4b9386c/ : SUCCESS in 1h 08m 35s\n- openstack-tox-pep8 http://logs.openstack.org/75/660175/5/check/openstack-tox-pep8/f46619e/ : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":5},{"id":"0b43c54a7c6171179ae9cacced330a3645208305","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-05-24 21:39:34.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"159ee611db32ccd76b1a746ee23884b8d29832a3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-05-24 22:45:01.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nThis is going to get stuck waiting for the neutron backports, but hopefully those merge quickly","accounts_in_message":[],"_revision_number":5},{"id":"85451353c9cf16422215218a3a1e61cbfa438833","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 14:14:11.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1d528a284a94cae831a7a402f95c7d2ca89382ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 17:43:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"76caa84802f8fceb75326348fccf3f41b630350a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 17:43:05.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/75/660175/5/gate/openstack-tox-docs/eb1ca03/html/ : SUCCESS in 4m 24s\n- neutron-tempest-plugin-api http://logs.openstack.org/75/660175/5/gate/neutron-tempest-plugin-api/fd1b21e/ : SUCCESS in 53m 26s\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/75/660175/5/gate/neutron-tempest-plugin-scenario-linuxbridge/9610d55/ : SUCCESS in 1h 10m 35s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/75/660175/5/gate/neutron-tempest-plugin-bgpvpn-bagpipe/5176bc0/ : SUCCESS in 1h 35m 39s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/75/660175/5/gate/neutron-tempest-plugin-api-queens/d0ba68b/ : SUCCESS in 50m 31s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/75/660175/5/gate/neutron-tempest-plugin-api-rocky/2579858/ : SUCCESS in 56m 45s\n- neutron-tempest-plugin-api-stein http://logs.openstack.org/75/660175/5/gate/neutron-tempest-plugin-api-stein/3821b8c/ : SUCCESS in 56m 46s\n- openstack-tox-pep8 http://logs.openstack.org/75/660175/5/gate/openstack-tox-pep8/d0f3135/ : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"87c3f941a3dd168d64226a4e238545ed69e05383","revisions":{"9ac910d9966b9507ebe952f09f03b7498053d796":{"kind":"REWORK","_number":1,"created":"2019-05-20 16:53:04.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/75/660175/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/75/660175/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/660175/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/660175/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/660175/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/660175/1"}}},"commit":{"parents":[{"commit":"1a987ecb4d310328cc4c2c5628dc5018dc7afc48","subject":"Migrate networking-sfc tests to neutron-tempest-plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1a987ecb4d310328cc4c2c5628dc5018dc7afc48"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-20 16:50:53.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-20 16:50:53.000000000","tz":120},"subject":"Add API test case to check if SG displays all rules","message":"Add API test case to check if SG displays all rules\n\nThis patch adds new API test which checks if owner of security group\ncan see rules which belongs to his security group even if rule was\ncreated and belongs to other user (admin).\n\nDepends-On: https://review.opendev.org/660174\n\nChange-Id: I728cd8252d27e27e91bd95e4734d9db470dee35a\nRelated-Bug: #1824248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9ac910d9966b9507ebe952f09f03b7498053d796"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9ac910d9966b9507ebe952f09f03b7498053d796"}]},"branch":"refs/heads/master"},"096a3abf72b2e3bd1c841f9e70baf09868c3a922":{"kind":"REWORK","_number":2,"created":"2019-05-21 20:25:09.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/75/660175/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/75/660175/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/660175/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/660175/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/660175/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/660175/2"}}},"commit":{"parents":[{"commit":"1a987ecb4d310328cc4c2c5628dc5018dc7afc48","subject":"Migrate networking-sfc tests to neutron-tempest-plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1a987ecb4d310328cc4c2c5628dc5018dc7afc48"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-20 16:50:53.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-21 20:24:57.000000000","tz":120},"subject":"Add API test case to check if SG displays all rules","message":"Add API test case to check if SG displays all rules\n\nThis patch adds new API test which checks if owner of security group\ncan see rules which belongs to his security group even if rule was\ncreated and belongs to other user (admin).\n\nDepends-On: https://review.opendev.org/660174\n\nChange-Id: I728cd8252d27e27e91bd95e4734d9db470dee35a\nRelated-Bug: #1824248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/096a3abf72b2e3bd1c841f9e70baf09868c3a922"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/096a3abf72b2e3bd1c841f9e70baf09868c3a922"}]},"branch":"refs/heads/master"},"9d60f32fcafa11ba4d51f6d146b30dfd8e8639ec":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-05-24 13:18:45.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/75/660175/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/75/660175/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/660175/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/660175/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/660175/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/660175/3"}}},"commit":{"parents":[{"commit":"1a987ecb4d310328cc4c2c5628dc5018dc7afc48","subject":"Migrate networking-sfc tests to neutron-tempest-plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1a987ecb4d310328cc4c2c5628dc5018dc7afc48"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-20 16:50:53.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-24 13:18:40.000000000","tz":0},"subject":"Add API test case to check if SG displays all rules","message":"Add API test case to check if SG displays all rules\n\nThis patch adds new API test which checks if owner of security group\ncan see rules which belongs to his security group even if rule was\ncreated and belongs to other user (admin).\n\nDepends-On: https://review.opendev.org/660174\nDepends-On: https://review.opendev.org/661281\nDepends-On: https://review.opendev.org/661283\nDepends-On: https://review.opendev.org/661284\n\nChange-Id: I728cd8252d27e27e91bd95e4734d9db470dee35a\nRelated-Bug: #1824248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9d60f32fcafa11ba4d51f6d146b30dfd8e8639ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9d60f32fcafa11ba4d51f6d146b30dfd8e8639ec"}]},"branch":"refs/heads/master"},"b731ec90f3e4615e9c95070b34188de71ce5cbd0":{"kind":"REWORK","_number":4,"created":"2019-05-24 13:51:12.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/75/660175/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/75/660175/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/660175/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/660175/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/660175/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/660175/4"}}},"commit":{"parents":[{"commit":"1a987ecb4d310328cc4c2c5628dc5018dc7afc48","subject":"Migrate networking-sfc tests to neutron-tempest-plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1a987ecb4d310328cc4c2c5628dc5018dc7afc48"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-20 16:50:53.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-24 13:50:06.000000000","tz":120},"subject":"Add API test case to check if SG displays all rules","message":"Add API test case to check if SG displays all rules\n\nThis patch adds new API test which checks if owner of security group\ncan see rules which belongs to his security group even if rule was\ncreated and belongs to other user (admin).\n\nPatch for master branch:\nDepends-On: https://review.opendev.org/660174\n\nBackport to stable/Stein:\nDepends-On: https://review.opendev.org/661281\n\nBackport to stable/Rocky:\nDepends-On: https://review.opendev.org/661283\n\nBackport to stable/Queens:\nDepends-On: https://review.opendev.org/661284\n\nChange-Id: I728cd8252d27e27e91bd95e4734d9db470dee35a\nRelated-Bug: #1824248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b731ec90f3e4615e9c95070b34188de71ce5cbd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b731ec90f3e4615e9c95070b34188de71ce5cbd0"}]},"branch":"refs/heads/master"},"87c3f941a3dd168d64226a4e238545ed69e05383":{"kind":"REWORK","_number":5,"created":"2019-05-24 19:02:57.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/75/660175/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/75/660175/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/660175/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/660175/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/660175/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/660175/5"}}},"commit":{"parents":[{"commit":"1a987ecb4d310328cc4c2c5628dc5018dc7afc48","subject":"Migrate networking-sfc tests to neutron-tempest-plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1a987ecb4d310328cc4c2c5628dc5018dc7afc48"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-20 16:50:53.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-24 19:02:54.000000000","tz":0},"subject":"Add API test case to check if SG displays all rules","message":"Add API test case to check if SG displays all rules\n\nThis patch adds new API test which checks if owner of security group\ncan see rules which belongs to his security group even if rule was\ncreated and belongs to other user (admin).\n\nPatch for master branch:\nDepends-On: https://review.opendev.org/660174\n\nBackport to stable/Stein:\nDepends-On: https://review.opendev.org/661281\n\nBackport to stable/Rocky:\nDepends-On: https://review.opendev.org/661283\n\nBackport to stable/Queens:\nDepends-On: https://review.opendev.org/661284\n\nChange-Id: I728cd8252d27e27e91bd95e4734d9db470dee35a\nRelated-Bug: #1824248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/87c3f941a3dd168d64226a4e238545ed69e05383"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/87c3f941a3dd168d64226a4e238545ed69e05383"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
