)]}'
{"id":"openstack%2Fneutron-tempest-plugin~660245","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Icf0980b8f1a8d884b05621f5298c13cad5f05387","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"dropping-py35-testing","hashtags":[],"change_id":"Icf0980b8f1a8d884b05621f5298c13cad5f05387","subject":"Dropping support for py35","status":"MERGED","created":"2019-05-21 03:13:06.000000000","updated":"2019-05-21 17:13:32.000000000","submitted":"2019-05-21 17:13:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660245-1558458812137-3b81190c","meta_rev_id":"e464e6d4df8a3d0aa11507e4ea0882b721328ed4","_number":660245,"virtual_id_number":660245,"owner":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2019-05-21 09:13:56.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-05-21 17:13:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2019-05-21 07:15:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":2,"date":"2019-05-21 13:01:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":1,"date":"2019-05-21 13:01:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 07:15:15.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-05-21 13:01:02.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-05-21 17:13:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8504828a87788c658dfac4110a27f34f4d2997a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-05-21 03:13:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2268153ee4f3937c63b8b444b8d8eaecc99c39d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 03:21:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/45/660245/1/check/test-release-openstack/53bdf5c/ : SUCCESS in 3m 30s\n- build-openstack-sphinx-docs http://logs.openstack.org/45/660245/1/check/build-openstack-sphinx-docs/2bbef18/html/ : SUCCESS in 5m 16s\n- openstack-tox-pep8 http://logs.openstack.org/45/660245/1/check/openstack-tox-pep8/bab0ca1/ : SUCCESS in 8m 30s","accounts_in_message":[],"_revision_number":1},{"id":"cd55de5613ed09fd19204bc858bd9b044bb973ba","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-21 07:14:49.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ac329e71920a36e7689f54612109ec554f05cfc3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-21 07:15:15.000000000","message":"Patch Set 2: Code-Review+2\n\nI only updated commit message according to real change in this repo :)","accounts_in_message":[],"_revision_number":2},{"id":"2857d6435496d835e9d1a66f410f9929300f5e4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 07:35:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/45/660245/2/check/test-release-openstack/c05f143/ : SUCCESS in 3m 24s\n- build-openstack-sphinx-docs http://logs.openstack.org/45/660245/2/check/build-openstack-sphinx-docs/0d533b3/html/ : SUCCESS in 4m 18s\n- openstack-tox-pep8 http://logs.openstack.org/45/660245/2/check/openstack-tox-pep8/31f30ca/ : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":2},{"id":"ea9f37a817954d2c4cb6a5dac3e6a293114fa314","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-05-21 09:13:56.000000000","message":"Patch Set 2:\n\n\u003e I only updated commit message according to real change in this repo\n \u003e :)\n\nThanks.","accounts_in_message":[],"_revision_number":2},{"id":"f5dc61c95c1395f2f50f4f75a2abdb4095800dcb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-05-21 13:01:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bd883d42b05aa4b2cd68e5f9f7b9b0a1a316f90b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 13:01:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"102dd2c4d42cdb6f7cf378f3b3b8326a2ccfc5ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 17:13:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/45/660245/2/gate/test-release-openstack/e0a47cf/ : SUCCESS in 3m 19s\n- build-openstack-sphinx-docs http://logs.openstack.org/45/660245/2/gate/build-openstack-sphinx-docs/a0bcdd5/html/ : SUCCESS in 5m 13s\n- openstack-tox-pep8 http://logs.openstack.org/45/660245/2/gate/openstack-tox-pep8/936f99e/ : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":2},{"id":"64c06e2a40eeef7481bb582f0a0e41863529689f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 17:13:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4f5afdce6bbb5868817119443ef6b125196dd36d","revisions":{"ce5d1edb88d9e02c33cadb7fd63aa32dbfb54746":{"kind":"REWORK","_number":1,"created":"2019-05-21 03:13:06.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/45/660245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/45/660245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/660245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/660245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/660245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/660245/1"}}},"commit":{"parents":[{"commit":"1a987ecb4d310328cc4c2c5628dc5018dc7afc48","subject":"Migrate networking-sfc tests to neutron-tempest-plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1a987ecb4d310328cc4c2c5628dc5018dc7afc48"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-05-21 03:08:31.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-05-21 03:11:33.000000000","tz":480},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: Icf0980b8f1a8d884b05621f5298c13cad5f05387\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ce5d1edb88d9e02c33cadb7fd63aa32dbfb54746"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ce5d1edb88d9e02c33cadb7fd63aa32dbfb54746"}]},"branch":"refs/heads/master"},"4f5afdce6bbb5868817119443ef6b125196dd36d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-21 07:14:49.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/45/660245/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/45/660245/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/660245/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/660245/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/660245/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/660245/2"}}},"commit":{"parents":[{"commit":"1a987ecb4d310328cc4c2c5628dc5018dc7afc48","subject":"Migrate networking-sfc tests to neutron-tempest-plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1a987ecb4d310328cc4c2c5628dc5018dc7afc48"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-05-21 03:08:31.000000000","tz":480},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-05-21 07:14:43.000000000","tz":0},"subject":"Dropping support for py35","message":"Dropping support for py35\n\nAs per below ML thread, we are good to drop the py35\nsupport now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: Icf0980b8f1a8d884b05621f5298c13cad5f05387\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4f5afdce6bbb5868817119443ef6b125196dd36d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4f5afdce6bbb5868817119443ef6b125196dd36d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
