)]}'
{"id":"openstack%2Fneutron-tempest-plugin~671297","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I61dc252920154e7e0998eb2c7c1f026814796cdf","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"make-running-stadium-tests-configurable","hashtags":[],"change_id":"I61dc252920154e7e0998eb2c7c1f026814796cdf","subject":"Add possibility to disable running stadium projects tests","status":"MERGED","created":"2019-07-17 13:40:16.000000000","updated":"2019-07-19 16:55:18.000000000","submitted":"2019-07-19 16:55:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671297-1563555318100-8d65fae4","meta_rev_id":"642cc40666b78265a790cb8ee1d2d41fde55c759","_number":671297,"virtual_id_number":671297,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2019-07-19 16:55:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-07-19 14:42:28.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-07-19 14:41:24.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-07-19 14:42:28.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-17 13:44:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-07-17 13:44:51.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-07-17 13:44:54.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-07-17 14:25:43.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-07-19 14:41:24.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-07-19 14:42:28.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-07-19 16:55:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dee23fe300513436daa229f68478f4a4574636a7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-07-17 13:40:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ab62016453500e353cfb8156720b1924fdb3c23","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-07-17 13:45:09.000000000","message":"Patch Set 1:\n\nContext of this patch is in http://eavesdrop.openstack.org/irclogs/%23openstack-neutron/%23openstack-neutron.2019-07-17.log.html#t2019-07-17T08:34:37","accounts_in_message":[],"_revision_number":1},{"id":"fc205b688f6a9a482208807311883089ec9de7b9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-07-17 14:25:43.000000000","message":"Patch Set 1:\n\n(3 comments)\n\nyou can also add the patch on stable/stein rocky, queens devstack or neutron\u0027s devstack plugins side to set these config option as false.\n\nor other way is to make these default config options as False and set it to True in devstack or devstack plugins. This is easy way to avoid multiple patches on stable branches.","accounts_in_message":[],"_revision_number":1},{"id":"a7fcadcca37bb0b1d1215473048aaa3f958f7d33","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-07-17 15:45:47.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1b1c7baee11a8ceaad0df14fd71e3484256c33f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 15:58:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/97/671297/1/check/openstack-tox-docs/723f60d/html/ : SUCCESS in 4m 12s\n- neutron-tempest-plugin-api http://logs.openstack.org/97/671297/1/check/neutron-tempest-plugin-api/aed6795/ : SUCCESS in 54m 20s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/97/671297/1/check/neutron-tempest-plugin-designate-scenario/0d9b2e1/ : SUCCESS in 36m 54s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/97/671297/1/check/neutron-tempest-plugin-dvr-multinode-scenario/b54ea40/ : FAILURE in 1h 38m 03s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/97/671297/1/check/neutron-tempest-plugin-scenario-linuxbridge/4ebe9ec/ : SUCCESS in 1h 02m 27s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/97/671297/1/check/neutron-tempest-plugin-api-queens/198e383/ : SUCCESS in 49m 13s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/97/671297/1/check/neutron-tempest-plugin-designate-scenario-queens/9ae784a/ : SUCCESS in 33m 48s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/97/671297/1/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/7c2454a/ : FAILURE in 1h 33m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/97/671297/1/check/neutron-tempest-plugin-scenario-linuxbridge-queens/935f31e/ : SUCCESS in 56m 34s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/97/671297/1/check/neutron-tempest-plugin-api-rocky/1e8f418/ : SUCCESS in 51m 30s\n- neutron-tempest-plugin-designate-scenario-rocky http://logs.openstack.org/97/671297/1/check/neutron-tempest-plugin-designate-scenario-rocky/2e4c4f0/ : SUCCESS in 31m 21s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky http://logs.openstack.org/97/671297/1/check/neutron-tempest-plugin-dvr-multinode-scenario-rocky/e40b78a/ : FAILURE in 1h 35m 47s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky http://logs.openstack.org/97/671297/1/check/neutron-tempest-plugin-scenario-linuxbridge-rocky/2222a1c/ : SUCCESS in 53m 16s\n- neutron-tempest-plugin-api-stein http://logs.openstack.org/97/671297/1/check/neutron-tempest-plugin-api-stein/a1ca862/ : SUCCESS in 44m 05s\n- neutron-tempest-plugin-designate-scenario-stein http://logs.openstack.org/97/671297/1/check/neutron-tempest-plugin-designate-scenario-stein/7b75e36/ : SUCCESS in 30m 50s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein http://logs.openstack.org/97/671297/1/check/neutron-tempest-plugin-dvr-multinode-scenario-stein/2c00b31/ : FAILURE in 1h 46m 38s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein http://logs.openstack.org/97/671297/1/check/neutron-tempest-plugin-scenario-linuxbridge-stein/2bfa34f/ : SUCCESS in 1h 06m 36s\n- openstack-tox-pep8 http://logs.openstack.org/97/671297/1/check/openstack-tox-pep8/4aa2093/ : SUCCESS in 3m 30s\n- neutron-tempest-plugin-sfc http://logs.openstack.org/97/671297/1/check/neutron-tempest-plugin-sfc/d39988b/ : SUCCESS in 1h 34m 56s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/97/671297/1/check/neutron-tempest-plugin-bgpvpn-bagpipe/7a522ca/ : SUCCESS in 1h 40m 42s\n- neutron-tempest-plugin-fwaas http://logs.openstack.org/97/671297/1/check/neutron-tempest-plugin-fwaas/c152378/ : SUCCESS in 37m 13s","accounts_in_message":[],"_revision_number":1},{"id":"317f4535ff2b2acacc3cf8dec81d97628bf12219","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-07-18 12:25:57.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cbc13ea67dbc6be16482c5f60ce4f6567fcc8363","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-07-18 12:26:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e5e5c044b9c17580e70416c5f29f024f5367d430","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-07-18 12:35:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1d0cb4ed937117d1acfe34473dab0a788d24108d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 14:36:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/97/671297/2/check/openstack-tox-docs/e591eeb/html/ : SUCCESS in 3m 35s\n- neutron-tempest-plugin-api http://logs.openstack.org/97/671297/2/check/neutron-tempest-plugin-api/6b8e6d1/ : SUCCESS in 50m 02s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/97/671297/2/check/neutron-tempest-plugin-designate-scenario/af61f4e/ : SUCCESS in 44m 16s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/97/671297/2/check/neutron-tempest-plugin-dvr-multinode-scenario/82d7174/ : SUCCESS in 1h 39m 04s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/97/671297/2/check/neutron-tempest-plugin-scenario-linuxbridge/76815bd/ : SUCCESS in 1h 22m 15s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/97/671297/2/check/neutron-tempest-plugin-api-queens/8fe7d82/ : SUCCESS in 48m 02s\n- neutron-tempest-plugin-designate-scenario-queens http://logs.openstack.org/97/671297/2/check/neutron-tempest-plugin-designate-scenario-queens/548a7cf/ : SUCCESS in 32m 31s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens http://logs.openstack.org/97/671297/2/check/neutron-tempest-plugin-dvr-multinode-scenario-queens/597b732/ : FAILURE in 1h 59m 33s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens http://logs.openstack.org/97/671297/2/check/neutron-tempest-plugin-scenario-linuxbridge-queens/e21828d/ : SUCCESS in 58m 41s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/97/671297/2/check/neutron-tempest-plugin-api-rocky/0837c40/ : SUCCESS in 1h 08m 31s\n- neutron-tempest-plugin-designate-scenario-rocky http://logs.openstack.org/97/671297/2/check/neutron-tempest-plugin-designate-scenario-rocky/b5cd99f/ : SUCCESS in 31m 43s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky http://logs.openstack.org/97/671297/2/check/neutron-tempest-plugin-dvr-multinode-scenario-rocky/eccb4e3/ : SUCCESS in 1h 27m 16s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky http://logs.openstack.org/97/671297/2/check/neutron-tempest-plugin-scenario-linuxbridge-rocky/82acbab/ : SUCCESS in 1h 29m 20s\n- neutron-tempest-plugin-api-stein http://logs.openstack.org/97/671297/2/check/neutron-tempest-plugin-api-stein/1a08849/ : SUCCESS in 48m 19s\n- neutron-tempest-plugin-designate-scenario-stein http://logs.openstack.org/97/671297/2/check/neutron-tempest-plugin-designate-scenario-stein/8f87dae/ : SUCCESS in 40m 59s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein http://logs.openstack.org/97/671297/2/check/neutron-tempest-plugin-dvr-multinode-scenario-stein/4193b44/ : SUCCESS in 1h 36m 30s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein http://logs.openstack.org/97/671297/2/check/neutron-tempest-plugin-scenario-linuxbridge-stein/3970eeb/ : SUCCESS in 1h 24m 43s\n- openstack-tox-pep8 http://logs.openstack.org/97/671297/2/check/openstack-tox-pep8/dab8238/ : SUCCESS in 6m 24s\n- neutron-tempest-plugin-sfc http://logs.openstack.org/97/671297/2/check/neutron-tempest-plugin-sfc/f135db2/ : SUCCESS in 1h 40m 30s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/97/671297/2/check/neutron-tempest-plugin-bgpvpn-bagpipe/d18961b/ : SUCCESS in 1h 34m 37s\n- neutron-tempest-plugin-fwaas http://logs.openstack.org/97/671297/2/check/neutron-tempest-plugin-fwaas/915f91a/ : SUCCESS in 42m 17s","accounts_in_message":[],"_revision_number":2},{"id":"d299ac50fdc52873291a9fa7ddac96ee937dd102","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-07-19 14:41:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6a01cd279d541df8b1d8be364fd4ee19f77afb01","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-07-19 14:42:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3e9dd1c4facb88d8a2ad7ff1cf46c9469b379fd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 14:42:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"326e05b4a0c6038ed86dad5c3d46c58f96669030","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 15:09:50.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6d0833b34c7f029ccc686126ecfece2fbe7e7315","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 16:55:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/97/671297/2/gate/openstack-tox-docs/0ad4e23/html/ : SUCCESS in 6m 11s\n- neutron-tempest-plugin-api http://logs.openstack.org/97/671297/2/gate/neutron-tempest-plugin-api/1369b2d/ : SUCCESS in 58m 07s\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/97/671297/2/gate/neutron-tempest-plugin-scenario-linuxbridge/8ca32dd/ : SUCCESS in 56m 21s\n- neutron-tempest-plugin-api-queens http://logs.openstack.org/97/671297/2/gate/neutron-tempest-plugin-api-queens/1777355/ : SUCCESS in 51m 13s\n- neutron-tempest-plugin-api-rocky http://logs.openstack.org/97/671297/2/gate/neutron-tempest-plugin-api-rocky/66adc11/ : SUCCESS in 1h 23m 35s\n- neutron-tempest-plugin-api-stein http://logs.openstack.org/97/671297/2/gate/neutron-tempest-plugin-api-stein/e806db3/ : SUCCESS in 56m 40s\n- openstack-tox-pep8 http://logs.openstack.org/97/671297/2/gate/openstack-tox-pep8/f255c59/ : SUCCESS in 5m 18s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/97/671297/2/gate/neutron-tempest-plugin-bgpvpn-bagpipe/e863d45/ : SUCCESS in 1h 39m 21s\n- neutron-tempest-plugin-fwaas http://logs.openstack.org/97/671297/2/gate/neutron-tempest-plugin-fwaas/ae32c68/ : SUCCESS in 48m 32s","accounts_in_message":[],"_revision_number":2},{"id":"a33532bbd074ae7e27bb99de0ab69c5e3efd6912","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 16:55:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a7bb1619d43b413eb8d5849eb6df8d0dee260660","revisions":{"7af348cad0ffd6b1f6d87c9523945d03799087e2":{"kind":"REWORK","_number":1,"created":"2019-07-17 13:40:16.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/97/671297/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/97/671297/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/97/671297/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/97/671297/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/97/671297/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/97/671297/1"}}},"commit":{"parents":[{"commit":"5cec35e997e006f123099a9c5e8fe588e63ffbde","subject":"Merge \"Add logging of servers console in scenario test_port_forwardings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5cec35e997e006f123099a9c5e8fe588e63ffbde"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-07-17 13:34:46.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-07-17 13:39:59.000000000","tz":120},"subject":"Add possibility to disable running stadium projects tests","message":"Add possibility to disable running stadium projects tests\n\nThis patch adds possibility to disable running of tests related to:\n* networking-bgpvpn,\n* neutron-fwaas,\n* networking-sfc\n\nwhich were moved to neutron-tempest-plugin repo recently.\nThis will be useful when tests will be run for one of\nstable branches up to Stein, in which those tests are still\nin project\u0027s repo.\n\nChange-Id: I61dc252920154e7e0998eb2c7c1f026814796cdf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7af348cad0ffd6b1f6d87c9523945d03799087e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7af348cad0ffd6b1f6d87c9523945d03799087e2"}]},"branch":"refs/heads/master"},"a7bb1619d43b413eb8d5849eb6df8d0dee260660":{"kind":"REWORK","_number":2,"created":"2019-07-18 12:26:21.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/97/671297/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/97/671297/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/97/671297/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/97/671297/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/97/671297/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/97/671297/2"}}},"commit":{"parents":[{"commit":"5cec35e997e006f123099a9c5e8fe588e63ffbde","subject":"Merge \"Add logging of servers console in scenario test_port_forwardings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5cec35e997e006f123099a9c5e8fe588e63ffbde"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-07-17 13:34:46.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-07-18 12:26:08.000000000","tz":120},"subject":"Add possibility to disable running stadium projects tests","message":"Add possibility to disable running stadium projects tests\n\nThis patch adds possibility to disable running of tests related to:\n* networking-bgpvpn,\n* neutron-fwaas,\n* networking-sfc\n\nwhich were moved to neutron-tempest-plugin repo recently.\nThis will be useful when tests will be run for one of\nstable branches up to Stein, in which those tests are still\nin project\u0027s repo.\n\nChange-Id: I61dc252920154e7e0998eb2c7c1f026814796cdf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a7bb1619d43b413eb8d5849eb6df8d0dee260660"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a7bb1619d43b413eb8d5849eb6df8d0dee260660"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
