)]}'
{"id":"openstack%2Fneutron-tempest-plugin~672858","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ic86546c18dbde251b7cc77955f591677c0de1fd4","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"sphinx","hashtags":[],"change_id":"Ic86546c18dbde251b7cc77955f591677c0de1fd4","subject":"Blacklist sphinx 2.1.0 (autodoc bug)","status":"MERGED","created":"2019-07-26 03:30:57.000000000","updated":"2019-07-27 02:43:08.000000000","submitted":"2019-07-27 02:43:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672858-1564195388348-520616fe","meta_rev_id":"16d98e7029eb1a6b70232f0bd6348e05a52e7de8","_number":672858,"virtual_id_number":672858,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-07-27 02:43:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2019-07-27 00:23:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2019-07-26 10:48:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":1,"date":"2019-07-27 00:23:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-26 10:48:38.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-07-27 00:23:29.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-07-27 02:43:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"09c1b804976a146e22c20e732fe79f4309b24ddf","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-26 03:30:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e84a3ab5923e69599a68d57edc45ec27dfc59723","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 03:35:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/58/672858/1/check/openstack-tox-docs/32cd92a/html/ : SUCCESS in 3m 15s\n- requirements-check http://logs.openstack.org/58/672858/1/check/requirements-check/278962d/ : SUCCESS in 2m 01s\n- openstack-tox-pep8 http://logs.openstack.org/58/672858/1/check/openstack-tox-pep8/86fad54/ : SUCCESS in 3m 24s\n- build-openstack-releasenotes http://logs.openstack.org/58/672858/1/check/build-openstack-releasenotes/3c88e8f/html/ : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"37d6c6254757323fbc05347e4c6a05e13a7027e3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-07-26 10:48:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1fa134ec8198a616005aa33e8e551f0c27b95375","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-07-27 00:23:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"caa8ec121ba12e918db3ca6b4793c380a4f0ff81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 00:24:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47a6c47d9a11b46bdd4c20139360548f359e6c45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 02:43:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"26f428e10eaf2170a8e7bf067ae436d304bae0d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 02:43:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/58/672858/1/gate/openstack-tox-docs/266d4dc/html/ : SUCCESS in 3m 03s\n- requirements-check http://logs.openstack.org/58/672858/1/gate/requirements-check/740787c/ : SUCCESS in 2m 04s\n- openstack-tox-pep8 http://logs.openstack.org/58/672858/1/gate/openstack-tox-pep8/87d1b94/ : SUCCESS in 3m 22s\n- build-openstack-releasenotes http://logs.openstack.org/58/672858/1/gate/build-openstack-releasenotes/30a6497/html/ : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d5bf8d6a016ad674cc3dbd79a8638be0b481082","revisions":{"7d5bf8d6a016ad674cc3dbd79a8638be0b481082":{"kind":"REWORK","_number":1,"created":"2019-07-26 03:30:57.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/58/672858/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/58/672858/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/58/672858/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/58/672858/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/58/672858/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/58/672858/1"}}},"commit":{"parents":[{"commit":"940ceee18274fd36404e746c01c6775810bb7a7b","subject":"Merge \"Test multi-cast messaging between VMs.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/940ceee18274fd36404e746c01c6775810bb7a7b"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-26 03:29:51.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-26 03:31:10.000000000","tz":480},"subject":"Blacklist sphinx 2.1.0 (autodoc bug)","message":"Blacklist sphinx 2.1.0 (autodoc bug)\n\nSee https://github.com/sphinx-doc/sphinx/issues/6440 for upstream details\nDepend-On: https://review.opendev.org/#/c/663060/\n\nChange-Id: Ic86546c18dbde251b7cc77955f591677c0de1fd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7d5bf8d6a016ad674cc3dbd79a8638be0b481082"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7d5bf8d6a016ad674cc3dbd79a8638be0b481082"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
