)]}'
{"id":"openstack%2Fneutron-tempest-plugin~677653","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Iec14343578ca587143a55bbd3d602790730d31b6","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"677653","hashtags":[],"change_id":"Iec14343578ca587143a55bbd3d602790730d31b6","subject":"Clean routers in DvrRoutersNegativeTestExtended","status":"MERGED","created":"2019-08-21 07:27:42.000000000","updated":"2019-08-30 03:15:10.000000000","submitted":"2019-08-30 03:15:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677653-1567134910238-6ccbb2a0","meta_rev_id":"79e5999ff102f32c674061468f8fff0968665b52","_number":677653,"virtual_id_number":677653,"owner":{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":14743,"name":"Viktor Tikkanen","email":"viktor.tikkanen@nokia.com","username":"vitikkan"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-08-30 03:15:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30156,"name":"Igor D.C.","email":"igor.duarte.cardoso@intel.com","username":"igordc"},{"value":0,"date":"2019-08-29 15:00:53.000000000","_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},{"value":0,"_account_id":24797,"name":"Matt Welch","email":"matt.welch@intel.com","username":"mattw4"},{"value":0,"_account_id":14749,"name":"Juha Kosonen","email":"juha.kosonen@nokia.com","username":"jukosone"},{"value":0,"_account_id":16590,"name":"Morgan Richomme","email":"morgan.richomme@orange.com","username":"Morgan22"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":14743,"name":"Viktor Tikkanen","email":"viktor.tikkanen@nokia.com","username":"vitikkan"},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2019-08-29 10:32:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14743,"name":"Viktor Tikkanen","email":"viktor.tikkanen@nokia.com","username":"vitikkan"},{"value":2,"date":"2019-08-29 10:31:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-08-29 21:30:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-29 17:01:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30156,"name":"Igor D.C.","email":"igor.duarte.cardoso@intel.com","username":"igordc"},{"value":0,"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},{"value":1,"date":"2019-08-29 16:34:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24797,"name":"Matt Welch","email":"matt.welch@intel.com","username":"mattw4"},{"value":1,"date":"2019-08-29 10:08:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14749,"name":"Juha Kosonen","email":"juha.kosonen@nokia.com","username":"jukosone"},{"value":0,"_account_id":16590,"name":"Morgan Richomme","email":"morgan.richomme@orange.com","username":"Morgan22"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":14743,"name":"Viktor Tikkanen","email":"viktor.tikkanen@nokia.com","username":"vitikkan"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-08-29 21:30:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30156,"name":"Igor D.C.","email":"igor.duarte.cardoso@intel.com","username":"igordc"},{"value":0,"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},{"value":0,"_account_id":24797,"name":"Matt Welch","email":"matt.welch@intel.com","username":"mattw4"},{"value":0,"_account_id":14749,"name":"Juha Kosonen","email":"juha.kosonen@nokia.com","username":"jukosone"},{"value":0,"_account_id":16590,"name":"Morgan Richomme","email":"morgan.richomme@orange.com","username":"Morgan22"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":14743,"name":"Viktor Tikkanen","email":"viktor.tikkanen@nokia.com","username":"vitikkan"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30156,"name":"Igor D.C.","email":"igor.duarte.cardoso@intel.com","username":"igordc"},{"value":0,"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},{"value":0,"_account_id":24797,"name":"Matt Welch","email":"matt.welch@intel.com","username":"mattw4"},{"value":0,"_account_id":14749,"name":"Juha Kosonen","email":"juha.kosonen@nokia.com","username":"jukosone"},{"value":0,"_account_id":16590,"name":"Morgan Richomme","email":"morgan.richomme@orange.com","username":"Morgan22"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14743,"name":"Viktor Tikkanen","email":"viktor.tikkanen@nokia.com","username":"vitikkan"},{"_account_id":14749,"name":"Juha Kosonen","email":"juha.kosonen@nokia.com","username":"jukosone"},{"_account_id":16590,"name":"Morgan Richomme","email":"morgan.richomme@orange.com","username":"Morgan22"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24797,"name":"Matt Welch","email":"matt.welch@intel.com","username":"mattw4"},{"_account_id":30156,"name":"Igor D.C.","email":"igor.duarte.cardoso@intel.com","username":"igordc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-21 07:28:14.000000000","updated_by":{"_account_id":16590,"name":"Morgan Richomme","email":"morgan.richomme@orange.com","username":"Morgan22"},"reviewer":{"_account_id":16590,"name":"Morgan Richomme","email":"morgan.richomme@orange.com","username":"Morgan22"},"state":"REVIEWER"},{"updated":"2019-08-21 07:28:37.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-08-29 10:08:14.000000000","updated_by":{"_account_id":14749,"name":"Juha Kosonen","email":"juha.kosonen@nokia.com","username":"jukosone"},"reviewer":{"_account_id":14749,"name":"Juha Kosonen","email":"juha.kosonen@nokia.com","username":"jukosone"},"state":"REVIEWER"},{"updated":"2019-08-29 10:31:23.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-08-29 10:32:13.000000000","updated_by":{"_account_id":14743,"name":"Viktor Tikkanen","email":"viktor.tikkanen@nokia.com","username":"vitikkan"},"reviewer":{"_account_id":14743,"name":"Viktor Tikkanen","email":"viktor.tikkanen@nokia.com","username":"vitikkan"},"state":"REVIEWER"},{"updated":"2019-08-29 16:34:13.000000000","updated_by":{"_account_id":24797,"name":"Matt Welch","email":"matt.welch@intel.com","username":"mattw4"},"reviewer":{"_account_id":24797,"name":"Matt Welch","email":"matt.welch@intel.com","username":"mattw4"},"state":"REVIEWER"},{"updated":"2019-08-29 17:01:48.000000000","updated_by":{"_account_id":30156,"name":"Igor D.C.","email":"igor.duarte.cardoso@intel.com","username":"igordc"},"reviewer":{"_account_id":30156,"name":"Igor D.C.","email":"igor.duarte.cardoso@intel.com","username":"igordc"},"state":"REVIEWER"},{"updated":"2019-08-29 21:30:20.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-08-30 03:15:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b1eb28f31f6762c2ec5be63a54a985c33f74d712","author":{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},"date":"2019-08-21 07:27:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ed8e304cc47d14227dee7429156c47fad85ebc7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 09:38:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bccf7d7a68bb4429a050734b4a36527a : SUCCESS in 4m 13s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/52a9b100850a477b8e8209a34a86615a : SUCCESS in 57m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fa31e4183f6e4fbb80a45c7a7a53779b : SUCCESS in 35m 26s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/86136c374f274754915a802e56884e6d : SUCCESS in 1h 43m 35s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/58b429286c064555b27b920828bae008 : SUCCESS in 1h 11m 17s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fda97aa4e7234228abc7564dc1de96a3 : SUCCESS in 1h 08m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fd2cc0eda6864017a926c62c3b05ffe6 : SUCCESS in 1h 04m 55s\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/b04fbd04adda4b64a1c6779f11582485 : SUCCESS in 49m 17s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/2f98f9675f3b4252aaa0424485051b56 : SUCCESS in 32m 42s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/bca93c10dcbd4ae7ab1e2a878c8a43b4 : FAILURE in 1h 32m 30s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/6010a50c5536471fb3e01d77d7ee7bba : SUCCESS in 1h 18m 30s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/0ac1bf7fd5304b98888c31cb9b8c3cd6 : SUCCESS in 1h 27m 16s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/fd60a83f086e4b8b85e786113e6b2ea2 : SUCCESS in 48m 26s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/211ad8b51ee648ba9cf30cfed5ffa252 : SUCCESS in 31m 09s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/993ba1a8ef49401d97f02656b229851e : SUCCESS in 1h 39m 40s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/e859d5ff1f47449da6a4e86f4407ca86 : SUCCESS in 1h 06m 23s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/87547ec56b4a4118b56bce38cc934165 : SUCCESS in 1h 12m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/c9c2b9bf8c1e4ec38f0e12341b12bcfc : SUCCESS in 1h 13m 11s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/61383ae4ca1940dfb98d6cf84fcee61d : FAILURE in 49m 31s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/6a1b1144080e44318ec14e3f92db7b51 : SUCCESS in 36m 45s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/07493824c4a2450cb74702ad3eac3c51 : FAILURE in 1h 39m 47s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/59e067e8c8ce48b8aa41751ff3085a91 : SUCCESS in 1h 21m 28s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/de785b3fe6e14d7d9443a9ab4c08077f : SUCCESS in 1h 14m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/7871261a7f704d59a9e0a39258a25326 : SUCCESS in 1h 08m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ef3fe570ce94a3db076b308feac038f : SUCCESS in 5m 07s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/e894847b57d74ea49faeede962385a58 : SUCCESS in 1h 40m 21s","accounts_in_message":[],"_revision_number":1},{"id":"48ae545cc92f670a18abef2e38781c29f7371484","author":{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},"date":"2019-08-21 09:42:34.000000000","message":"Patch Set 1:\n\nrecheck / reverify","accounts_in_message":[],"_revision_number":1},{"id":"bc7a252f2fab5d26cb61ff157f3e74d64f31bce3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 11:55:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ac3e6d84225469b890277ee8b2957a2 : SUCCESS in 5m 38s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/451f1d70ee334ee4b7101f946022e74c : SUCCESS in 53m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3f41a8345c13495fbf206c7e56e34651 : SUCCESS in 34m 53s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/5f2727f6dbf24c11af1450ad5b365703 : FAILURE in 1h 39m 42s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4f39c7c6ea7f4653bf8b5eb1425bac11 : SUCCESS in 1h 17m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3a58875085b74b54adbcee53785670d7 : SUCCESS in 1h 12m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/57a4583ae5044f5196da4de2f65e2a63 : SUCCESS in 1h 18m 13s\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/bf9d441f97ac439fba9802707ac6e359 : SUCCESS in 51m 41s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/a1e9c0e2caa443c286d426b633914feb : SUCCESS in 31m 07s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/f24d8847f11d4b93b5cd4d96e31a462c : FAILURE in 1h 34m 45s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/21007dd0f2754f589bc73bd3123e1cab : SUCCESS in 1h 08m 47s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/ca6f79d0ae0743c7a30eb0d808a8b62b : SUCCESS in 1h 05m 15s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/da04ffe6fdcf4d1080f85a2d1736673c : SUCCESS in 48m 58s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/40b5ec1344e7485f9474749883549769 : SUCCESS in 34m 07s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/4de452c4dc624bca91f9ab9d9b8778d5 : FAILURE in 1h 47m 02s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/2834ae75c6bd4eb7a86beaee678eab99 : SUCCESS in 1h 09m 55s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/9b97d6652c174a5c83f44142a23bf946 : SUCCESS in 1h 05m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/301c6c05c5174c418e2ec0a5cb96378a : SUCCESS in 1h 10m 37s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/2d2cc34b50e7495caa7628ba358f1522 : SUCCESS in 52m 39s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/0dade47185c8486297c36d768fad1582 : SUCCESS in 46m 08s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/5d29f549d5594bfb9cfddce31f925bf1 : FAILURE in 1h 36m 38s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/5f3cc3fde21045558526c5cf76c4a53c : SUCCESS in 1h 16m 31s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/3ed5b74dbd72432c98bf5f8139a48c27 : SUCCESS in 1h 09m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/000b7928beb04adab3f492f1525f0c2b : SUCCESS in 1h 03m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7c2b48fc279413da6a4e087dedddf72 : SUCCESS in 3m 38s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/2719c1516b7b4cf291c18a654adcd056 : SUCCESS in 1h 53m 44s","accounts_in_message":[],"_revision_number":1},{"id":"f5c28e8d7bb854861566b5a80973e8778d131544","author":{"_account_id":14743,"name":"Viktor Tikkanen","email":"viktor.tikkanen@nokia.com","username":"vitikkan"},"date":"2019-08-21 11:56:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a9f8b4a7090d54429a86d0cfed7e777b739c8ae7","author":{"_account_id":14749,"name":"Juha Kosonen","email":"juha.kosonen@nokia.com","username":"jukosone"},"date":"2019-08-21 12:27:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8ee916c311f360fe5f55e63b6c71a7ab9165f889","author":{"_account_id":30156,"name":"Igor D.C.","email":"igor.duarte.cardoso@intel.com","username":"igordc"},"date":"2019-08-22 14:11:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d1c6a0d2af724d6c0fde388a5f4b07c0a6c725b3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-08-26 15:52:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f20fda70e7b6ce0ac22c01f3c30c053d5da21249","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-08-28 12:37:12.000000000","message":"Patch Set 1: Code-Review+2\n\nI\u0027m fine merging this, we can clean things up in a follow-on if we feel like it.","accounts_in_message":[],"_revision_number":1},{"id":"0e659d07f3055a4ccb543798dd70864429e587c7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-08-28 15:00:43.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ab66e9ac7a4833a792c5f47044922282e81c098a","author":{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},"date":"2019-08-29 08:43:54.000000000","message":"Patch Set 1:\n\nIt should have been a fix about the previous patch.\nOk to switch to the helper listed if it works.","accounts_in_message":[],"_revision_number":1},{"id":"f699152af4fb2ef1a264219bae4e40a71c39674f","author":{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},"date":"2019-08-29 09:50:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fa44b73f1ea38ac070845a01bf46fcd835fa862e","author":{"_account_id":14749,"name":"Juha Kosonen","email":"juha.kosonen@nokia.com","username":"jukosone"},"date":"2019-08-29 10:08:14.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3e29d586699af54af0c98de5c440cbcc744af737","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-08-29 10:31:23.000000000","message":"Patch Set 2: Code-Review+2\n\nI hope that CI will be happy with this change :)","accounts_in_message":[],"_revision_number":2},{"id":"30e770215281d9590f526b863f2e130d5c8bfa32","author":{"_account_id":14743,"name":"Viktor Tikkanen","email":"viktor.tikkanen@nokia.com","username":"vitikkan"},"date":"2019-08-29 10:32:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"082d909de575a63675d734ad8b915b2def0c8728","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 12:52:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/068896e274bb40128cb0c2731380c44c : SUCCESS in 4m 03s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a1e4c7b7f2e141c091735fad69122600 : SUCCESS in 53m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b0d1407968eb4784bad09144e971278f : SUCCESS in 39m 03s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/7ecb8b13cdcd4dd9a7cde48ef0c0e7f6 : FAILURE in 1h 41m 41s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/48b1929496bc48e890e718008f202ac1 : SUCCESS in 1h 04m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2144bf9894af420f95ff384c6ae5ad6f : SUCCESS in 1h 04m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b081506b041f456596484fcea7dbfd2b : SUCCESS in 1h 00m 52s\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/7c33eb786fd149c69245c60286091910 : SUCCESS in 53m 48s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/4a7fd0fdd6b84002864756bd1eca87d4 : SUCCESS in 44m 56s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/b640f9a2b0ad420983eff838baceb43d : FAILURE in 1h 39m 16s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/9ad873bb8fc04612befd7e370d668940 : FAILURE in 1h 06m 19s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/54d9c64811e4468484c53a9278bc91a1 : SUCCESS in 49m 21s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/6b603028fd944238a140939a6774d8e1 : SUCCESS in 56m 42s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/0ee014b73d6e495488bf8743b910c99e : SUCCESS in 30m 11s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/491fedf04caf4a1d8d8a4c6144625e39 : FAILURE in 1h 55m 48s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/2bd300f71cdc469f833d3187e3e2a8a4 : SUCCESS in 59m 06s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/0ed231954c234b51b1ea8254841284b7 : SUCCESS in 1h 18m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/5f990886aa6a46b5a9351f419d9993df : SUCCESS in 1h 19m 10s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/1e8c18f9b52b4c868f90fe2dddb89fb5 : SUCCESS in 52m 48s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/9585cd49ed8049399943b7c51e2fa71e : SUCCESS in 44m 20s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/1e62e383f62f41c4b830c1d2fdbf2da0 : FAILURE in 1h 32m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/f6131d0c24304b2795a5ed866c913ef6 : SUCCESS in 57m 44s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/c729bd21a3b0483e9ebf0a50c25f995d : SUCCESS in 1h 03m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/f27980674cfc4ca2834dd00e195742f5 : SUCCESS in 54m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/737fdfabfb3a44a9a781d973c2f74f30 : SUCCESS in 4m 44s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/6c5c82c45ecc4be8b2efa8e8c205c3b4 : SUCCESS in 1h 44m 50s","accounts_in_message":[],"_revision_number":2},{"id":"3bfeb34efafa023caa7d5f811da58708a2ee3442","author":{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},"date":"2019-08-29 15:00:53.000000000","message":"Patch Set 2:\n\nrecheck / reverify","accounts_in_message":[],"_revision_number":2},{"id":"c954f2d2816287198875477180f977d6385f7c02","author":{"_account_id":24797,"name":"Matt Welch","email":"matt.welch@intel.com","username":"mattw4"},"date":"2019-08-29 16:34:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3bf05aae70922093d46fc2267225f1b899c7b632","author":{"_account_id":30156,"name":"Igor D.C.","email":"igor.duarte.cardoso@intel.com","username":"igordc"},"date":"2019-08-29 17:01:48.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6f5e9833b41982cbc7c5a6536fea4a5dfff1c3ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 20:46:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea16d1d4593c4932bda8e816746f8213 : SUCCESS in 6m 22s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ae190afbc23b454a85a0667339257926 : SUCCESS in 1h 06m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c1ba845998e942a08052eadd715e552b : SUCCESS in 38m 24s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/97ba73a92dde4e7a8c27d55e9819372d : SUCCESS in 1h 23m 04s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/338d887c3ed644db9c77facf08f24591 : SUCCESS in 1h 12m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6589859295904104814a840b0c35d409 : SUCCESS in 1h 15m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e67173eb64574601ad7b847bf47a0144 : SUCCESS in 1h 01m 31s\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/635fcecfdd7e447489071c3d29f27f1b : SUCCESS in 1h 02m 20s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/d1e5af95d788457489edaebfadeba9a7 : SUCCESS in 35m 42s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/6de5bf8a9045466ea52a8e82a241270e : FAILURE in 1h 35m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/5b86451e861244cf8f905158a23bb42e : SUCCESS in 1h 05m 29s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/ed5f3e3b099b425eaf8475e8a56c6ebe : SUCCESS in 57m 29s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/090c7e108fd444d3b9c7a5ba2ffaabcb : SUCCESS in 56m 22s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/b9c09a772c784c4d9fbc42ce5f8a7040 : SUCCESS in 34m 31s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/db2e7331baf9487e908ab6172c7dbc75 : FAILURE in 1h 46m 51s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/7b51e8ff05714945bfce9307e1eea327 : SUCCESS in 1h 17m 15s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/d55c307977df4e10a3962be52f4af2ea : SUCCESS in 1h 05m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/d5bea73aedf14e6f96c5f59955a779ee : SUCCESS in 1h 05m 58s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/9c36463ccc8346f592f98e2092ff07f0 : SUCCESS in 47m 30s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/2ca68406713c451fbb4be8d0b6d584ac : SUCCESS in 46m 18s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/7001f0731c994bf09f197127c4f0ee77 : FAILURE in 1h 53m 28s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/5479ddef430c4a8b8de6b1c16476c797 : SUCCESS in 1h 15m 54s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/5272f48dfb8d4ce5809383d0f88d4ba0 : SUCCESS in 58m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/2a592e5069bb4c3ab5f6161ac94cc0dc : SUCCESS in 59m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/795f5559ec3e4195b26e6b45424aea17 : SUCCESS in 6m 09s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d834fd7ac9d140fc83ea339f70564eb4 : SUCCESS in 1h 36m 40s","accounts_in_message":[],"_revision_number":2},{"id":"c7f9ba48ea7b6b6e2079816bf0fa236d43371e66","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-08-29 21:30:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nI started https://review.opendev.org/#/c/679351/ to change the rest of the calls over.","accounts_in_message":[],"_revision_number":2},{"id":"8a27574b3481064f8c058ad2a57bd75ada3fc069","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 21:30:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"330c7272a554bfd1a328e9b11cc67c19b41a29f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 03:15:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9d1a8bc69c8ea0840e2dfc0f4b979ff2a257e32d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 03:15:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f39a730366c412c8f406bf5d981e6b6 : SUCCESS in 4m 58s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1d4d5a12904144fcb63608c3726c3a18 : SUCCESS in 52m 00s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/746f0bf51d8043d5ab70ca076dd3ef86 : SUCCESS in 1h 14m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/13f42039f8a4469bbf6451a59c05eecc : SUCCESS in 58m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b2a6980764ed46278be2df98e7331736 : SUCCESS in 1h 01m 53s\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/e23087522bd346fcace0391da35a65bc : SUCCESS in 1h 09m 26s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/01f9b4cf228d473da5c6378bedbdc22b : SUCCESS in 51m 33s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/b7a36e366d2c40c69d8c857f9bd942be : SUCCESS in 57m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f58669e5d24a40b3b04d101025b38777 : SUCCESS in 4m 49s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/3fbb7a98147a4134ae4f2ee89bb60bd4 : SUCCESS in 1h 47m 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0a3705f937bbe16aeffea4bc434a87655f282ea2","revisions":{"6569bdeea85c4a46270fa70ff01b3bff1756e3b6":{"kind":"REWORK","_number":1,"created":"2019-08-21 07:27:42.000000000","uploader":{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},"ref":"refs/changes/53/677653/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/53/677653/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/53/677653/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/53/677653/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/53/677653/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/53/677653/1"}}},"commit":{"parents":[{"commit":"d11f4ec31ab1cf7965671817f2733c362765ebb1","subject":"Start enforcing E125 flake8 directive","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d11f4ec31ab1cf7965671817f2733c362765ebb1"}]}],"author":{"name":"Cédric Ollivier","email":"ollivier.cedric@gmail.com","date":"2019-08-21 07:26:37.000000000","tz":120},"committer":{"name":"Cédric Ollivier","email":"ollivier.cedric@gmail.com","date":"2019-08-21 07:27:40.000000000","tz":120},"subject":"Clean routers in DvrRoutersNegativeTestExtended","message":"Clean routers in DvrRoutersNegativeTestExtended\n\nIt completes the initial patch \"Add tempest tests for DVR router state\nmanagement\" which doesn\u0027t clean 2 routers [1].\n\n[1] https://review.opendev.org/#/c/668760/\n\nChange-Id: Iec14343578ca587143a55bbd3d602790730d31b6\nSigned-off-by: Cédric Ollivier \u003collivier.cedric@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6569bdeea85c4a46270fa70ff01b3bff1756e3b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6569bdeea85c4a46270fa70ff01b3bff1756e3b6"}]},"branch":"refs/heads/master"},"0a3705f937bbe16aeffea4bc434a87655f282ea2":{"kind":"REWORK","_number":2,"created":"2019-08-29 09:50:02.000000000","uploader":{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},"ref":"refs/changes/53/677653/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/53/677653/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/53/677653/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/53/677653/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/53/677653/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/53/677653/2"}}},"commit":{"parents":[{"commit":"d11f4ec31ab1cf7965671817f2733c362765ebb1","subject":"Start enforcing E125 flake8 directive","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d11f4ec31ab1cf7965671817f2733c362765ebb1"}]}],"author":{"name":"Cédric Ollivier","email":"ollivier.cedric@gmail.com","date":"2019-08-21 07:26:37.000000000","tz":120},"committer":{"name":"Cédric Ollivier","email":"ollivier.cedric@gmail.com","date":"2019-08-29 09:49:52.000000000","tz":120},"subject":"Clean routers in DvrRoutersNegativeTestExtended","message":"Clean routers in DvrRoutersNegativeTestExtended\n\nIt completes the initial patch \"Add tempest tests for DVR router state\nmanagement\" which doesn\u0027t clean 2 routers [1].\n\n[1] https://review.opendev.org/#/c/668760/\n\nChange-Id: Iec14343578ca587143a55bbd3d602790730d31b6\nSigned-off-by: Cédric Ollivier \u003collivier.cedric@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0a3705f937bbe16aeffea4bc434a87655f282ea2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0a3705f937bbe16aeffea4bc434a87655f282ea2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
