)]}'
{"id":"openstack%2Fneutron-tempest-plugin~686043","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ia757540fee60657a801d34a731b91f3a78c17e25","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"ipv6-only-deployment-and-testing","hashtags":[],"change_id":"Ia757540fee60657a801d34a731b91f3a78c17e25","subject":"[train][goal] Define new neutron-tempest-ipv6-only job","status":"ABANDONED","created":"2019-10-01 20:50:22.000000000","updated":"2020-09-25 15:37:18.000000000","total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8aaaefcc58f53891f87a757993eb0a84de0e5367","_number":686043,"virtual_id_number":686043,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"date":"2020-09-25 15:37:18.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-30 23:34:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-01 23:54:56.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-10-02 05:02:51.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-10-08 14:40:46.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-11-08 11:00:58.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-03-30 23:34:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-25 15:37:18.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"baf0158e5301d36b906842da5709bc3811372921","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-01 20:50:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6127c9f4a97dc95892717d86ae192faae1511cd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 20:50:38.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron-tempest-plugin on branch master.  The error was:\n\n  while scanning a double-quoted scalar\n  in \"openstack/neutron-tempest-plugin/.zuul.yaml@master\", line 949, column 27\nfound unknown escape character \u0027[\u0027\n  in \"openstack/neutron-tempest-plugin/.zuul.yaml@master\", line 949, column 29","accounts_in_message":[],"_revision_number":1},{"id":"0c446a5de92d38b024290a68ff1d31bd60e47969","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-01 20:50:42.000000000","message":"Topic set to ipv6-only-deployment-and-testing","accounts_in_message":[],"_revision_number":1},{"id":"06b53cd0aeae29cb509066b6e61c1aa0ded2a85a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-01 20:52:50.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8f915b459d699031613c5ef16b50e625e8ae3e46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 20:53:04.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron-tempest-plugin on branch master.  The error\nwas:\n\n  The key \"devstack_localrc\" appears more than once; duplicate keys are\n  not permitted.\n\nThe error appears in the following stanza:\n\n  \u003c\u003c: *base_vars\n        # This job does not need to run all the tests. Smoke and IPv6 tests are enough\n        # to test the Ipv6 deployment.\n        tempest_test_regex: \u0027\\[.*\\bsmoke|ipv6|test_network_v6\\b.*\\]\u0027\n        devstack_services:\n          neutron-log: true\n        devstack_local_conf:\n          post-config:\n            # NOTE(slaweq): We can get rid of this hardcoded absolute path when\n            # devstack-tempest job will be switched to use lib/neutron instead of\n            # lib/neutron-legacy\n            /$NEUTRON_CORE_PLUGIN_CONF:\n              AGENT:\n                tunnel_types: gre,vxlan\n              network_log:\n                local_output_log_base: /tmp/test_log.log\n        devstack_localrc:\n          PHYSICAL_NETWORK: default\n          DOWNLOAD_DEFAULT_IMAGES: false\n          IMAGE_URLS: \"http://download.cirros-cloud.net/0.3.4/cirros-0.3.4-i386-disk.img,https://cloud-images.ubuntu.com/releases/xenial/release/ubuntu-16.04-server-cloudimg-amd64-disk1.img\"\n          DEFAULT_IMAGE_NAME: cirros-0.3.4-i386-disk\n          ADVANCED_IMAGE_NAME: ubuntu-16.04-server-cloudimg-amd64-disk1\n          ADVANCED_INSTANCE_TYPE: ds512M\n          ADVANCED_INSTANCE_USER: ubuntu\n          BUILD_TIMEOUT: 784\n        network_api_extensions_stadium:\n          - flow_classifier\n          - sfc\n          - bgpvpn\n          - bgpvpn-routes-control\n          - bgp\n          - bgp_dragent_scheduler\n          - bgp_4byte_asn\n          - fwaas_v2\n        devstack_localrc:\n          NETWORK_API_EXTENSIONS: \"{{ (network_api_extensions_common + network_api_extensions_stadium) | join(\u0027,\u0027) }}\"\n        irrelevant-files: *tempest-irrelevant-files\n\n  in \"openstack/neutron-tempest-plugin/.zuul.yaml@master\", line 946, column 7","accounts_in_message":[],"_revision_number":2},{"id":"1af0bf0c15db93eb095e464f754441dcc2b8c6d6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-01 20:54:19.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"a5309e8f5638e8293408263c8e302a16ea616b62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 00:21:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c1d7db6901f4d6fa1820f42fb29b5fb : SUCCESS in 4m 30s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ecdc351b8b874286b577402e0ac91aad : SUCCESS in 1h 07m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/018814a4e9cd42e3b9ecf90562d7858f : SUCCESS in 35m 54s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/7bfc4655291e4cb1a161578013f66bd5 : SUCCESS in 1h 26m 48s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1bab124d2837495cbbad0574aabec0cb : SUCCESS in 1h 04m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a4d4e963d3694333867c1e30fdeafd85 : SUCCESS in 1h 08m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f3da99b8648343349030b2994cda8cb6 : SUCCESS in 1h 07m 17s\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/aa4b8974b36e47a7920ce43129166818 : SUCCESS in 44m 29s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/b4bfcd9183354f6393a5b94501690843 : SUCCESS in 39m 35s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/f7f650fe7d9643b6bb7ffb49d56ec156 : FAILURE in 1h 37m 58s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/8ea47cd919d3481f9771fb7325c52cb7 : SUCCESS in 1h 03m 01s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/87f02a2729de4ca483c6947b02ed641f : SUCCESS in 51m 32s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/e7eab1a43df04a56a082e4ebf8399cc1 : SUCCESS in 54m 54s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/14dc66e599784106b8bd7ac778d13433 : SUCCESS in 34m 50s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/b04235b6338844f0876a21313a3d0cb1 : SUCCESS in 1h 42m 08s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/e3571977c098422980b8cbd44b229771 : SUCCESS in 1h 17m 41s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/f65490a590f245b68487c34c017c7a1b : SUCCESS in 1h 10m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/47c718e2433e4490a8bf77811589c14f : SUCCESS in 56m 33s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/65786b7c2ea94e08aeb8fb535c873a25 : SUCCESS in 57m 18s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/dc5e125e39114323b15dcaab64e68ea9 : SUCCESS in 37m 11s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/18f5493727d24e53b77a0a059dc29187 : SUCCESS in 1h 21m 02s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/ddfb34a1fd6e469a9714341f724346d7 : SUCCESS in 1h 01m 37s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/371d7e0f8a4c46f3be68d0af02c9dd2d : SUCCESS in 59m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/4110dde94f1d4bbfa732a4e38b762e0d : SUCCESS in 1h 07m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec7eae334cb440fbb19cac2115fd15a1 : SUCCESS in 3m 36s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/a1974389a4b144139d7a0b576d8d5d73 : SUCCESS in 1h 49m 47s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/1dea02a8c5834bfca41febed3def5bbe : SUCCESS in 1h 55m 02s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/1faf36675af247688d045fc59e34376e : SUCCESS in 41m 58s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/ebfd41b2054e4118ab6dde4edfe138ee : SUCCESS in 48m 01s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3b37147985048209f5a5b24d7c4c219 : FAILURE in 2h 48m 41s","accounts_in_message":[],"_revision_number":3},{"id":"b258a886d7b786e2a85b3b735f9a126aabdf3e23","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-02 02:36:46.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"635bbb6b4fda17ce00c6d979024e8d36e81ec743","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-02 05:02:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"19b5dd803a8633dcc7f8bc3c17f5387858aba859","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 05:14:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f84d5cf48c2d4c548c95f862717fbad8 : SUCCESS in 3m 24s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cda3c7f2d678450f8bc750f834372cfa : SUCCESS in 54m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c769f3cb80ac4b9785319146b0109c0c : SUCCESS in 34m 34s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/c1542833b3de4221902359e327cb2128 : FAILURE in 1h 26m 13s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/171e705a5a844571a4364169c79ca48f : SUCCESS in 1h 01m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c1e3d36c65b64121a818bdd87b9c5008 : SUCCESS in 52m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/12833d7c294a449d9dac016379cb4951 : SUCCESS in 58m 19s\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/8381b195c64f494d86d9b14ccb153860 : SUCCESS in 49m 14s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/949d3638b2db477b892f23e05fa927d9 : SUCCESS in 33m 24s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/684f416bffe246a9ad4c59e776602764 : FAILURE in 1h 26m 20s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/1bc25eecb3024963a45d2ac4ebcfcef6 : SUCCESS in 58m 45s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/2fa032c401334486957cc162e9a39d05 : SUCCESS in 1h 27m 07s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/127f2faed64648328a39ca9854ebeedd : SUCCESS in 46m 53s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/ae5a1840d1f249c4b94ef89b68e5daae : SUCCESS in 29m 36s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/e6dad15eb6154cbf845ce920371beafb : SUCCESS in 1h 31m 25s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/a971ee2a0cec4a3e9944d98ae5a945ea : SUCCESS in 1h 07m 11s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/a92fb8f7472e453c8fc49ab9bb14a8b3 : SUCCESS in 1h 00m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/72b1d9631e8e4a998e117299fd31408f : SUCCESS in 57m 27s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/ec583d5e26784c4895f7a33423553f4b : SUCCESS in 48m 00s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/8c359dde1e1c496ebefc45254ab40fe0 : SUCCESS in 38m 04s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/3371bf585a3c41959de711c70bdf9685 : SUCCESS in 1h 22m 33s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/f42c581df3b54101b94f4a1aef7cf252 : SUCCESS in 1h 09m 00s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/9c90ceb915824e45bebffb8664b07e47 : SUCCESS in 1h 09m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/66b54de2f73a4b1a9d956f483787d6d3 : SUCCESS in 1h 01m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50948c413fac4fc6bbb40212ab85c007 : SUCCESS in 3m 08s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/1e3f81bd06a14cf99bf87c6968d2d913 : SUCCESS in 1h 47m 27s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/97dd2589084b485ba2fe2ecb3caff236 : SUCCESS in 1h 31m 58s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/59a8b1d3ab7f4444ba3972826b4452b8 : SUCCESS in 34m 39s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/e6c80596c1854930aac26a9dd46fa8d4 : SUCCESS in 52m 49s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5740a93f174a49ebb14062a1e3c52428 : FAILURE in 2h 31m 15s","accounts_in_message":[],"_revision_number":4},{"id":"33cef00e20f1690a6a302a6e9dee34d937b930f5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-02 18:26:30.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b81d40b3476fe8a151e1fc471bd28d38a171b5d1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-02 19:00:39.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"fb0e2a9b81d4d13c9f5cf751d9908546b0b65780","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 21:49:42.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fb44d4e6e574e5294a1bcd31ea7ef68 : SUCCESS in 4m 35s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a8a5cc836e804631bf074af20ee852a1 : SUCCESS in 1h 03m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0eb89311371e45c688bb8438b06baa41 : SUCCESS in 35m 30s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/03adad1bdd5d418e9b65362a0157f3c8 : SUCCESS in 1h 35m 11s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ca242b484f89444da26b0b90b995ab58 : SUCCESS in 1h 08m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b5e0056814df4f9ba18130a7b400ae07 : SUCCESS in 1h 00m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/49466505c31142a6a20d48c5bd6868ad : SUCCESS in 53m 32s\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/3ec12acdee8443f1b0bd827c94fa1fcc : SUCCESS in 51m 23s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/d08f6d249bf9403fb6b5807fe295da27 : SUCCESS in 34m 58s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/91bad342b65049b198e1dc033e3fd0aa : FAILURE in 1h 49m 04s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/b2059f1de2b447e1a577a7c6622ebd5d : SUCCESS in 1h 07m 46s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/b213ecb4645d49948f636efad121d066 : SUCCESS in 59m 36s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/dc71600ad0464cf6a7b11d3ec784e2ec : SUCCESS in 53m 44s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/ea50b41700da44fb962157b3d35b7649 : SUCCESS in 32m 44s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/4f70b3f7478f46db8d257030de30b0d9 : FAILURE in 1h 19m 16s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/2706ceee678d4389914fae40302283b2 : SUCCESS in 58m 41s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/78cf2c7004884a74a1b229e576e482de : SUCCESS in 1h 09m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/2351b724439944c6bf7ab34503534156 : SUCCESS in 1h 12m 46s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/3c24d25f0fc148d3b305dca5063aa6ac : SUCCESS in 1h 25m 52s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/a7e73abe4bbb4559be691519b2c83f5f : SUCCESS in 40m 49s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/9bdbdb8857d04f8eac3ba43a2c98a228 : FAILURE in 1h 28m 57s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/9d7cbf53b5cc4c269a212d8320ee974d : SUCCESS in 1h 08m 12s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/cab74bc6070943c6acc0d62988ab74d2 : SUCCESS in 54m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/fd5faaf9af454dc2979cf3b3e23935f5 : SUCCESS in 1h 00m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/109e6556d52f4698b158e59f868a574b : SUCCESS in 3m 03s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/5b926d7727a845b4a5ccfc98fcafb763 : FAILURE in 1h 51m 08s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/0b6ec95d118d4d73bf674350fc1fcba9 : SUCCESS in 1h 41m 16s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/f637a3b841dd4033990fe202c81ea1f4 : SUCCESS in 42m 05s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/a89b72a2cd3c41f6b01936ac726d283e : SUCCESS in 49m 18s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b73c29257ddd43dfa6e2fb24720b8d2e : FAILURE in 2h 32m 46s","accounts_in_message":[],"_revision_number":5},{"id":"ed38d105d1ca61b24bc6ba66382e66f756002c18","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-02 22:08:34.000000000","message":"Patch Set 5:\n\nrecheck timeout","accounts_in_message":[],"_revision_number":5},{"id":"1d4427f4dbae484cf01b4c44ff7315242c9cb981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 01:14:26.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1e4905893f94656862902130871766d : SUCCESS in 4m 56s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e1716b7607a249d199cee319d0b3d466 : SUCCESS in 49m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d09f4d31f79749c18e5abf74d2f20bc2 : SUCCESS in 36m 10s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/462d8e2edc5f43f58324211253aba7db : SUCCESS in 1h 35m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/98af958c950b4d7ba3996c08775df534 : SUCCESS in 1h 01m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c3b7d88d9eb448b2ad1d20535063c47c : SUCCESS in 1h 04m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6d97213086c444a889df493a5aa9d33b : SUCCESS in 1h 15m 00s\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/5fe6857ab1a741bfb9866952e4a6cbcc : SUCCESS in 1h 03m 41s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/4655854c514d47e58742be1a79dcfe85 : SUCCESS in 40m 48s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/e61f6d6a0fc641e997c13d7d180646d4 : FAILURE in 1h 19m 21s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/5b6de604487344bc95cece4a5d72b89d : SUCCESS in 1h 12m 38s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/19febab3d1eb44f28e3aba1b7d28d371 : SUCCESS in 1h 00m 35s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/081e0804da154518b04040f5fc5bb240 : SUCCESS in 1h 31m 24s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/e476f7dc69d341768ef6362b90608725 : SUCCESS in 39m 42s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/104321f86fd845dd92e326b9fd4f428a : SUCCESS in 1h 36m 35s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/f6b9982ccda849b0bfc302bf169d660b : SUCCESS in 1h 11m 03s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/6ac44bfd1cbc480699670a25e74658e1 : SUCCESS in 1h 16m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/e49d47a2f53b48709d8e624e2b1dee2c : SUCCESS in 1h 17m 34s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/11cba59cd9474925acf166be69b90e2c : SUCCESS in 1h 06m 00s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/9d03a4ff6fdf43a8ad6e9f234436b6fd : SUCCESS in 37m 20s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/34741d425ddb43e8b26aa02b8b9c04ea : SUCCESS in 1h 19m 09s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/0470078256424eb7a904addf7c3cbefa : SUCCESS in 1h 15m 35s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/870295dc2ee849e09e79d7337a28ff8e : SUCCESS in 1h 21m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/73fd192e4a86499e8bcf58cba25ca204 : SUCCESS in 1h 04m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a83e6e5942f449f7b68e6a2155139e86 : SUCCESS in 3m 12s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/22c7c6334314446cb076287e848d45e1 : SUCCESS in 1h 54m 25s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/77c1a6d2b5344aeeb4ecf1c0cad445dc : SUCCESS in 1h 37m 42s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/4a8754523756435abae74831bf2bca3c : SUCCESS in 44m 03s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/511b606bf50d4605ad21ec8d4d90081d : SUCCESS in 51m 50s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e109ac89468c4571bce609df91615c4f : FAILURE in 2h 28m 40s","accounts_in_message":[],"_revision_number":5},{"id":"29a7a1bdbe36f7f75af69a3fe9e933cf8cda9522","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-16 12:57:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2e56c73c9c34ad98d655e0a5033898dbacade3e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 16:04:01.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87b3182d0e6d470b95060973b4c41bca : SUCCESS in 3m 48s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c7b3e5cc770a4e068052d5dad17252bb : SUCCESS in 54m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c887fa2f18f14f10835617dcc691bf51 : SUCCESS in 43m 46s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/5e97dd1f58ae4c829c366ba26d465706 : SUCCESS in 1h 32m 29s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/094d0f13a1ea49c89cc8bc91d0067790 : SUCCESS in 1h 22m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7667c7049e0b4349af92dbf9182a4088 : SUCCESS in 1h 33m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b6c618708ead4465b8115c3d68ed287a : SUCCESS in 1h 08m 19s\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/85072ce7c6ce45f2a471d1287eecfcdd : SUCCESS in 54m 24s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/195981ee0bc44cbc87571e26d64ac6d6 : SUCCESS in 34m 41s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/00aaaef603f546d3a157eadc9ca66b99 : FAILURE in 1h 29m 56s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/0e6dc460702c4f518a7b1d22c99947f0 : SUCCESS in 1h 01m 24s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/14a493b8de9d40ba81dd702fec4ed65a : SUCCESS in 1h 00m 43s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/0439f69434a8481eb71bb6efc4460901 : SUCCESS in 46m 42s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/bed6a5f241ac42a98d6e29d6190be537 : SUCCESS in 35m 38s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/f4060035195e43f98485d8ebfc64cd35 : SUCCESS in 1h 34m 29s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/1f3d6518b5394ecb80060847d69024a5 : SUCCESS in 53m 55s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/d3016d560a7e4f1b98e7c702f62b7cf4 : SUCCESS in 1h 17m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/71815cd53b894668b531bce6a357f4a1 : SUCCESS in 1h 17m 58s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/97e5c3d897594072bdf0327b3cc29702 : SUCCESS in 46m 48s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/80906a91afda403faa5535b9de05792f : SUCCESS in 33m 55s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/05f9145ff41644818fce190f6dfd3a0b : SUCCESS in 1h 24m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/20b4e9f518a54999a15739f14088ad52 : SUCCESS in 1h 11m 42s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/3f905aad2dcb4ca88bd74e25b3589b9b : SUCCESS in 1h 08m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/7416f68ca6a74cb6adc43ad5f7a790a3 : SUCCESS in 1h 04m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75816586ad464f69ab40c8076335caef : SUCCESS in 3m 45s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/b3978b528a66494bbf395dd6c7a4f419 : SUCCESS in 1h 53m 00s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9a5eb4fd7b4b4401b75a6e665761cc2c : SUCCESS in 1h 27m 51s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/8070fbf5255c43da8c1354b5664eae2e : SUCCESS in 40m 24s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/82c0e469c6c840cda66fab4e8224fb85 : SUCCESS in 53m 32s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f0e3e1b35604b01afbaa864116ae751 : FAILURE in 2h 35m 27s","accounts_in_message":[],"_revision_number":6},{"id":"77d35586ad77912210012ae682733ca8a8c39ee4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-16 19:55:31.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"28223c46daa869fc808c04fee63b2c7e52677f80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 21:58:40.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e622d2164cc14bada7b0b71467246aad : SUCCESS in 4m 35s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0c1f6cafa906426d965dbfbb91da0d97 : SUCCESS in 44m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aad291a604a84966aab8b08cb29397b8 : SUCCESS in 37m 44s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/737f60958a5f4dfa9b80d94187855284 : SUCCESS in 1h 34m 51s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ec8ae9ef910747d4ae7ca78612b2af88 : SUCCESS in 1h 12m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/aea655ba12cc4cc6b995ed0dcf865b11 : SUCCESS in 1h 04m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/55eeae102c1340a9916a709fa942afb1 : SUCCESS in 1h 04m 32s\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/37647b1db71e4a3e9aa3669fd6e7dc50 : SUCCESS in 46m 27s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/3462dfb1da1344ccb6c0723c1e1133e9 : SUCCESS in 29m 31s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/ef8d068fde474b6caaa931a31d5e40c9 : FAILURE in 1h 34m 11s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/6f738680a4d2423592c787d0a37a9d02 : SUCCESS in 1h 00m 04s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/3d3f9e190502499497ad0bab5a9fb8a4 : SUCCESS in 1h 01m 15s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/9efec5897b944da3873c3f6b527b925e : SUCCESS in 55m 18s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/064bc079044247678d7c869d69fc6839 : SUCCESS in 35m 07s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/c7c24d14b5684744a19a6a2127acdc14 : SUCCESS in 1h 39m 44s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/05d4e5dd9e2b44b2b53c8746c51480f7 : SUCCESS in 1h 09m 29s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/b806a5d9129943a68a7cd933bf0f246e : SUCCESS in 1h 04m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/c6adec51f2cc4359bb4bc62ba1c37845 : SUCCESS in 1h 12m 16s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/f9e6d21c90654d3584adfe579608df44 : SUCCESS in 52m 27s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/7fdd54b264004b839b14fb5ae9791c98 : SUCCESS in 33m 17s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/be0432846bc943e1ab0d28d4bd23d72f : SUCCESS in 1h 46m 52s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/bda488ddc7f9404da10c9606702f2af1 : SUCCESS in 1h 05m 36s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/48d044e5110b46fc8f4a5c6fc2c6c4ce : SUCCESS in 1h 20m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/31740de0fb8d476aadda3e7363df24d2 : SUCCESS in 1h 06m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b96845f1f334f988513a88f241441c3 : SUCCESS in 4m 44s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/be519897016246419256937bc2b243d6 : SUCCESS in 1h 58m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d31e30e9151c41d9a3d0f843699b8f13 : SUCCESS in 1h 37m 03s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/464586e129b440189199a1ad1a206c4f : SUCCESS in 35m 35s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/50cd8cc7154f4462aa0571190d47ab76 : SUCCESS in 51m 48s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bf88c1ad6e3444c8dc421da48aaf811 : FAILURE in 42m 22s","accounts_in_message":[],"_revision_number":7},{"id":"e46f0718232dded1231103ebe6a9a5041bc26609","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-07 04:26:56.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0be5a490392cdbc17859112b6c3d0bf6d6232855","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-07 04:28:25.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"cb7fd3d77cb220b6f48388f5079a4217236cae19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 04:28:36.000000000","message":"Patch Set 8: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"2339b5d81e25eb17838b43bf462cb448c92ee074","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-07 07:24:00.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"e85ed82709e852642dbf681f20dacae2e48106f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 09:27:52.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f412b5f9af304312abcabb1b1ef3b9e9 : SUCCESS in 3m 59s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4115449436454eb8a46b27675257e7fc : SUCCESS in 45m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d22303d5fc314c719aeffd979d733a6a : SUCCESS in 37m 41s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/11811a8037cf4f93b83f60d0e0599144 : FAILURE in 1h 34m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/34983685660a47e0946c2f00cbf82561 : SUCCESS in 1h 02m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/707e8b71d4114af1bca552eaa1068519 : SUCCESS in 52m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f8e0a12130894e0f9e00590dd5f88f1b : SUCCESS in 1h 01m 04s\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/008be5fb72824824a8c90f5e31f35e64 : SUCCESS in 53m 37s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/c05b4de40c764a1c909cef3b8783479c : SUCCESS in 30m 58s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/4b4436cc1c8643a6a3d6c376c9fe556d : SUCCESS in 1h 11m 09s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/8794362afb7b4612aeb0fbf73fa18259 : SUCCESS in 47m 32s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/31789a0fd276443093394cb7ff07dfae : SUCCESS in 52m 37s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/8bd411978260446e8bd8f04a691295cf : SUCCESS in 47m 05s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/5342fe39c6a94de78f8035bbdb6a2f87 : SUCCESS in 31m 57s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/56a4a9bdcde94d40be817b88431f56a3 : SUCCESS in 1h 23m 54s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/f9cc42d7e9a24f3a8a5365fbd60b6a38 : SUCCESS in 59m 18s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/a632aab025cb4393957c2cf30d023a6a : SUCCESS in 54m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/544b780276ae42f494fa43222e746f25 : SUCCESS in 53m 27s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/3b194ca91aec4010bcf1d5e3cf81021e : SUCCESS in 47m 20s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/5f7463e5918943c7a96b0c64b059cc4b : SUCCESS in 31m 08s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/2509ba644573489fb1681a23415e8f2d : FAILURE in 1h 34m 49s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/2827b99b11554130a8519020a28402ac : SUCCESS in 1h 09m 17s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/549ffb54de9f41b59bd5ff5d975da947 : SUCCESS in 1h 29m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/babed94f1e0a4201b30b1093d154483e : SUCCESS in 1h 08m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22c62657d45d42798b74135593155ff5 : SUCCESS in 3m 37s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/707e1815ef2b45b48fc6e0ab9625438a : SUCCESS in 1h 32m 31s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/785bb761fbbb428bb35dc5c9f4a395fc : SUCCESS in 1h 25m 46s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/0be8680a08ff45efa99dd6a261479b32 : SUCCESS in 48m 07s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/9fa7894a1f474e0fa33e64edeeeda318 : FAILURE in 1h 52m 12s (non-voting)\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3b6c2f7d67d4759adbdb55d5eb3b4bc : FAILURE in 35m 23s","accounts_in_message":[],"_revision_number":9},{"id":"fd11965b5e1214b64a643f94afa279b431e364ff","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-11-08 11:00:58.000000000","message":"Patch Set 9:\n\n(4 comments)\n\nLooks like bagpipe isn\u0027t ready for IPv6.","accounts_in_message":[],"_revision_number":9},{"id":"d36d5d667f27168873d4c83a63e5c9fa52d26150","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-09 12:02:09.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"df6b3d840cf8cc2ecf32353a00e3047128d8bcc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 12:02:24.000000000","message":"Patch Set 10: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron-tempest-plugin on branch master.  The error\nwas:\n\n  Job neutron-tempest-ipv6-only not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - build-openstack-docs-pti\n        - neutron-tempest-plugin-jobs\n        - neutron-tempest-plugin-jobs-rocky\n        - neutron-tempest-plugin-jobs-stein\n        - neutron-tempest-plugin-jobs-train\n        - check-requirements\n        - tempest-plugin-jobs\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - neutron-tempest-plugin-sfc\n          - neutron-tempest-plugin-sfc-train\n          - neutron-tempest-plugin-bgpvpn-bagpipe\n          - neutron-tempest-plugin-bgpvpn-bagpipe-train\n          - neutron-tempest-plugin-fwaas\n          - neutron-tempest-plugin-fwaas-train\n          - neutron-tempest-plugin-dynamic-routing:\n              # TODO(slaweq): switch it to be voting when bug\n              # https://bugs.launchpad.net/neutron/+bug/1850626 will be fixed\n              voting: false\n          - neutron-tempest-ipv6-only\n      gate:\n        jobs:\n          - neutron-tempest-plugin-sfc\n          - neutron-tempest-plugin-bgpvpn-bagpipe\n          - neutron-tempest-plugin-fwaas\n          # TODO(slaweq): bring it back to gate queue\n          # https://bugs.launchpad.net/neutron/+bug/1850626 will be fixed\n          # - neutron-tempest-plugin-dynamic-routing\n          - neutron-tempest-ipv6-only\n\n  in \"openstack/neutron-tempest-plugin/.zuul.yaml@master\", line 1286, column 3","accounts_in_message":[],"_revision_number":10},{"id":"c28b3226813744bbb7a5d4708fbc4986ae582526","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-09 12:03:39.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"e45763502e0bc0c20fe67485539111dcd6b8b8ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 13:53:58.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac00e55a2e0243ababc4970604c00ce4 : SUCCESS in 5m 10s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/34a5fcfbd03840929a56f7a44613d15b : SUCCESS in 49m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/08aea6d7c2f6466fb90278d00ab7a17e : SUCCESS in 32m 14s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/ee44cc1daeb641be9165cc57428f82ee : FAILURE in 1h 38m 35s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a91795668bc04febb2fb4703d1c8cf78 : SUCCESS in 1h 09m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d48ef4af2002403684e2020720850e42 : SUCCESS in 52m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f134e13696194e9ab7ecb356ae3589d9 : SUCCESS in 49m 52s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/90f24ea90b9c4d24aec250b4ddf4e6e9 : SUCCESS in 46m 58s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/dc02c40df550402aa0030982047b7c83 : SUCCESS in 33m 46s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/0b94f28c53b24b2aa3e192c9408c9cc7 : SUCCESS in 1h 26m 11s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/f7561eb66ef7461ab63b69c65f8fd790 : SUCCESS in 59m 50s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/2f589a15128343ddb6f1571aa8f3dc4d : SUCCESS in 1h 01m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/942901663360472d97e4a2e56de1555c : SUCCESS in 1h 05m 38s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/a97f3e79572f450587e451337a6dbbcd : SUCCESS in 57m 27s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/d88e265654a14f9c855d5875885f9bd3 : SUCCESS in 34m 58s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/049fd716d66e4d11b4508b3478504bdf : SUCCESS in 1h 31m 42s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/55e57ac3a3134311b5417cafb4588172 : SUCCESS in 1h 24m 46s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/e3c6f0ca6da54ce892e992d4b4e30adb : SUCCESS in 49m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/e5058c993a31486795c878e0b21d83a7 : SUCCESS in 52m 06s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/97404402665f4cae9881639aed8d7143 : SUCCESS in 50m 52s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/c37c4dc2e45d486cb2b5740e809d1144 : SUCCESS in 39m 52s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/0e6a80e876e24652ba082a6489eb0f32 : SUCCESS in 1h 34m 10s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/6b13029dd08f40b6a8feecd66ac858b3 : SUCCESS in 58m 52s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/44999692cf07412594bbdb1399799fae : SUCCESS in 56m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/8be05de07e744c74866ae4d03b88fe66 : SUCCESS in 59m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1530e5ee444407696ad700960627c9b : SUCCESS in 3m 47s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/202701b96d3c4f4184cbece264b56c91 : SUCCESS in 1h 42m 07s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/3c571997827643e6a84a104ae6d77262 : SUCCESS in 1h 43m 41s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/8f0d8d2a741946629e656d3fc8542c44 : SUCCESS in 1h 32m 39s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/577ba0fab33c435ca8709885b3994fdf : SUCCESS in 1h 29m 15s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/23c74220d63d49a0bc824fe34c438172 : SUCCESS in 38m 51s\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/97a7372dae564e3cac581d9cc9061203 : SUCCESS in 44m 18s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/dd03fda84d59488bbf0b3bf5e887f88b : SUCCESS in 46m 37s (non-voting)\n- neutron-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/b2ceb710d6d44e2c84389dff9de7457c : FAILURE in 37m 42s","accounts_in_message":[],"_revision_number":11},{"id":"3db581aeaa686d812f1e37623d483034d735c039","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-12-12 00:41:10.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"7ad0efc8f5d25b4f1a4018ea84950656145002c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 00:41:25.000000000","message":"Patch Set 12: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron-tempest-plugin on branch master.  The error was:\n\n  while parsing a block collection\n  in \"openstack/neutron-tempest-plugin/.zuul.yaml@master\", line 1, column 1\nexpected \u003cblock end\u003e, but found \u0027?\u0027\n  in \"openstack/neutron-tempest-plugin/.zuul.yaml@master\", line 1170, column 1","accounts_in_message":[],"_revision_number":12},{"id":"7ee317956e959af103ffb99ffe6887c12deccd1d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-12-12 00:42:23.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"5a8168d6055cf069cf6bde611a36c441071f7b78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 00:42:39.000000000","message":"Patch Set 13: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron-tempest-plugin on branch master.  The error was:\n\n  while parsing a block collection\n  in \"openstack/neutron-tempest-plugin/.zuul.yaml@master\", line 1, column 1\nexpected \u003cblock end\u003e, but found \u0027?\u0027\n  in \"openstack/neutron-tempest-plugin/.zuul.yaml@master\", line 1170, column 1","accounts_in_message":[],"_revision_number":13},{"id":"14f123437824f8affeac94dd124ee1b5bc28d797","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-12-12 00:44:06.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"97b3b176844f8ec8397d58e085820d8877464793","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 02:41:12.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf4b14fcea9b40368dfdccda729e5425 : SUCCESS in 3m 28s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/29b51535e2b2442f94f342999bb7ebe8 : SUCCESS in 59m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/491a5f56b5ad46b0a37cef1014c9297d : SUCCESS in 35m 52s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/d44b93de6d5c43cc8aab0a128bc823b6 : FAILURE in 1h 42m 54s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/33b7ef93e4ea4508a6d5e89a76304e51 : SUCCESS in 1h 10m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0b389ab609d24aab964c3d112766e017 : SUCCESS in 1h 21m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/25e60e185020497ca1a844f9322b5dbf : SUCCESS in 1h 08m 09s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/579af1104d584a30b8e9d2a54d151e2d : SUCCESS in 49m 32s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/4c3d9572d63d47f2a471c29c62d0ad91 : SUCCESS in 34m 14s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/20a67780fce4472a9bcfe40ad94e6b73 : FAILURE in 1h 42m 12s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/7a1b40a83f604c52ac24c9d42d311141 : SUCCESS in 1h 07m 54s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/8b64ecf3881147d39e1f785fb570fc71 : SUCCESS in 1h 05m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/9e0f6ed0d4a24302bffeb7673b7a00d3 : SUCCESS in 1h 11m 10s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/c52f882e1c3f4adc874ddd7dbfeac1aa : SUCCESS in 50m 34s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/db270145fe8943b78a43694e6c988737 : SUCCESS in 44m 33s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/1811d414ca1747739ef0ddbf01c34b03 : SUCCESS in 1h 35m 23s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/66c6e737d6a74f439163bdbf4ca390ec : SUCCESS in 1h 04m 08s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/840aac21701a4a7384d40425212c37ab : SUCCESS in 1h 00m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/e36e610372574911ba3c3d72ee77d11e : SUCCESS in 1h 14m 32s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/d5e58eab8d0e47ce96980383cccfe6cf : SUCCESS in 54m 18s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/13b72eaa7cc74e85a0d68c9ee0eabcd0 : SUCCESS in 39m 48s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/16ad7561ae974729abd92f7801c5526e : SUCCESS in 1h 18m 26s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/6d9e314ec236435bb01d9b2d622e6c2d : SUCCESS in 57m 46s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/3dc41ff299cb44e3bc7ef4228d3f451a : SUCCESS in 1h 02m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/56ef7ddaf02648749faf158303bf0f17 : SUCCESS in 1h 02m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9fd7bc7a7c546789fe8145ce62117e8 : SUCCESS in 3m 29s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/a7db1a828f0143f0aa1a12af711ad7de : SUCCESS in 1h 39m 07s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/38504a8dc3454dbbbdbbe33a050b48da : SUCCESS in 1h 46m 06s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/292ab299458441a78140bb19c50ea75f : SUCCESS in 1h 43m 57s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/2a35320c6b3641fa945bf24d25b6479d : SUCCESS in 1h 42m 49s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/0500249fee6b4579b102b67befed1208 : SUCCESS in 38m 06s\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/9777c80883ff4915b3bb3d5b981985d9 : SUCCESS in 49m 27s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/9496a67a8e2d486c8ab772c7fe3705d9 : SUCCESS in 1h 00m 57s (non-voting)\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/d2f8497d531642dda5445cea5b0df13d : SUCCESS in 43m 10s\n- neutron-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/a5d722388e0e43babe14b04ddf15b9e7 : FAILURE in 37m 08s","accounts_in_message":[],"_revision_number":14},{"id":"8ed53d7de56597e0a580cbf6b848e37025972a18","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-12 09:36:10.000000000","message":"Patch Set 14:\n\n(4 comments)\n\nIt seems that bagpipe isn\u0027t ready for IPv6 peers. I reported bug: https://bugs.launchpad.net/networking-bagpipe/+bug/1856154","accounts_in_message":[],"_revision_number":14},{"id":"a331ab923434c8fc39d43825807659ff38ca3c47","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-12 14:14:11.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"a7a102cf404e8259a64e771de2e466bc81b93858","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 16:52:45.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6adae3d1b3547c1baabedc27842621b : SUCCESS in 4m 22s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/11871ae7945341fcac2b93ca91e99e4b : SUCCESS in 49m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9b3d7a10621c4eec8c8f8e41b72e0cc1 : SUCCESS in 35m 58s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/7528283ab2c048a7aa07b43d561f0483 : FAILURE in 1h 33m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cc6004c1dff34c8a9f1118215053ca5b : FAILURE in 1h 36m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/02253c6db5eb4db3b7cfc2d790745cae : SUCCESS in 1h 30m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ad5202ea1a9e4c818b5f55ba38ee1f46 : SUCCESS in 1h 24m 53s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/43699a1076984a6c9acac8f29d1dcc77 : SUCCESS in 50m 51s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/f0fc60e45c3c43aeb398af2226c68564 : SUCCESS in 38m 57s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/b89653e17c3a45fbb85375f49587af73 : FAILURE in 1h 31m 34s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/95c35b402256478598c02b427742a9bd : SUCCESS in 1h 07m 27s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/3e7f54e6389945009322f6f9fd9cf0ff : SUCCESS in 1h 14m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/a02a2905690f450283882dc788a26c66 : SUCCESS in 1h 01m 53s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/990abea30c4543709d9a329cdf7eb294 : SUCCESS in 50m 35s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/e0964ac12f4f44339c7a52f4abe584fc : SUCCESS in 36m 25s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/f05d69650aa249809edc53549d598c16 : SUCCESS in 1h 15m 24s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/b7fec7b9748943c087a69b5d9dca5ef3 : SUCCESS in 1h 00m 52s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/b1edf6bded7a43519de2c6d8ee405201 : SUCCESS in 1h 01m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/c5b25f407dbf4d15956b38b200d2050c : SUCCESS in 1h 17m 20s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/7c8df72c835c4738aefefc0f353b5677 : SUCCESS in 1h 06m 35s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/f92194b08eb840a5b810261a87742bb3 : SUCCESS in 42m 44s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/0f9cedb388a548e48982de8369bbd11a : FAILURE in 1h 57m 37s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/9651860374c54a019fa814919c3c641d : SUCCESS in 1h 12m 51s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/27b8a990764e4e6fb0b36dd82c93ea87 : SUCCESS in 1h 06m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/70d63d7dd6144270b19c2288304d3ccb : SUCCESS in 1h 09m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6113ecdeb3e74aef825fe0368d77fd4b : SUCCESS in 5m 25s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/98c636890a9a4812bb1b910de6486cad : SUCCESS in 1h 42m 06s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/4764fa7376a946469f773c6fb310c095 : SUCCESS in 1h 46m 08s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/bbdc73593b8f43a98ba14fb1e09e74c2 : SUCCESS in 1h 35m 26s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/1df159d0dcb146fe95999fecb0764753 : SUCCESS in 1h 31m 42s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/921d675283064fd3aa2ce5d22bbb144c : SUCCESS in 45m 14s\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/cb45cce4b77747e78d75d28a563cb75e : SUCCESS in 40m 47s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/1463ed34a9cb4f758be58354da53d4e9 : SUCCESS in 51m 25s (non-voting)\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/f05df47859b74b68a6ff31456f17984d : SUCCESS in 42m 42s\n- neutron-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/0e63dc1a5dfb4470a095e34ac85afa86 : FAILURE in 2h 32m 59s","accounts_in_message":[],"_revision_number":15},{"id":"be6baa582b17bc0e65ba2b94aed287d19e9b7c85","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-10 10:28:03.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"382e6ee8a35793c5e26f66474eabb09b9648ba8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 13:11:28.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef559b1bc03a49ebb781fd2b1bf8bacc : SUCCESS in 4m 30s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ec8d2af4686c41d6a3e3f2e165af1987 : SUCCESS in 55m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ae3dec0dd7304d30af5470e01746b7c1 : SUCCESS in 36m 53s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/a2ae2c23bb4540ef80cb72c58130ad85 : FAILURE in 2h 01m 57s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/93b6f0525e4e4c43a04fe7c47d1b9569 : SUCCESS in 1h 37m 19s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6d485e584d1541f8a3269ae8e6214778 : SUCCESS in 1h 20m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0c240fa1cdfe47ae97d7a493802437ca : SUCCESS in 1h 09m 40s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/a54140a4ae1c4edfbd71ad832d54b0d1 : SUCCESS in 55m 11s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/44ecf6ab750144d4bf0d2d192622d654 : SUCCESS in 35m 31s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/8244517e235543fb8b82eae9e936c240 : FAILURE in 1h 46m 48s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/78d19317324a4c7981fbfc2eef8c3f04 : SUCCESS in 1h 18m 46s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/c595b3173f9b486eb48cbfe66fa1101d : SUCCESS in 1h 00m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/8f5ceb6fc70f43de92667330c8f547fa : SUCCESS in 1h 06m 25s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/14f2b7778e264755bb9fefdca512a7d4 : SUCCESS in 49m 06s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/75ad5cdbddad4a029b6056deee962fd5 : SUCCESS in 32m 28s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/e8476503f34e493097c6ae7402bbceaa : FAILURE in 1h 24m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/cdcfd1789c764b118df2c69be56df3b6 : SUCCESS in 1h 21m 13s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/840c05ddbbc34339998f0f33888bc306 : SUCCESS in 1h 14m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/3ea44cf4cb234d11bb7e929d236de013 : SUCCESS in 1h 15m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9554d5fbba3446608391979fb7b04b58 : SUCCESS in 3m 53s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/b655289f73fe40fea803d27707edd01f : SUCCESS in 1h 40m 57s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/e28a50daf94441e48bfd4d927d0833db : SUCCESS in 1h 40m 09s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/b92f383ee5f84e17b5ab3c13db7e4a8e : SUCCESS in 1h 31m 16s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/69bf99a8bc354079b07c597d0edbbca3 : SUCCESS in 1h 29m 44s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/14cf2a2886084e8fb7310c0a850bbe7c : SUCCESS in 34m 50s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/877725c2327442eda87d67a51a7e6b67 : SUCCESS in 40m 14s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/8f5533f9c4094fa98ecaafbc11ce73b3 : SUCCESS in 49m 11s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/dac6739d9bdd4b8a9b126abc2e3f3e14 : SUCCESS in 34m 53s\n- neutron-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/8b7f8da1789c49b8946a45466ea79368 : FAILURE in 2h 37m 29s","accounts_in_message":[],"_revision_number":16},{"id":"869ab27710c372893b351a84d3accdba35f9a041","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-16 14:07:55.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"f5ca8cdfb74f1337fed21f3b17af7a2deb1aeb18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 14:09:24.000000000","message":"Patch Set 17: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron-tempest-plugin on branch master.  The error was:\n\n  while scanning a double-quoted scalar\n  in \"openstack/neutron-tempest-plugin/.zuul.yaml@master\", line 1375, column 27\nfound unknown escape character \u0027[\u0027\n  in \"openstack/neutron-tempest-plugin/.zuul.yaml@master\", line 1376, column 44","accounts_in_message":[],"_revision_number":17},{"id":"714922e23ee46cfe4021392cd9afc7f0e704216b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-17 10:56:06.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"68aeb206aa410d80857313cc2dfb7846f581e0fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 10:56:25.000000000","message":"Patch Set 18: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron-tempest-plugin on branch master.  The error was:\n\n  while scanning a double-quoted scalar\n  in \"openstack/neutron-tempest-plugin/.zuul.yaml@master\", line 1375, column 27\nfound unknown escape character \u0027[\u0027\n  in \"openstack/neutron-tempest-plugin/.zuul.yaml@master\", line 1376, column 43","accounts_in_message":[],"_revision_number":18},{"id":"27b77e6b8cf8bf8140aad5b8439e735f77971ef4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-17 11:11:43.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"6ae31285c65050f190c50c96bab8cde12c4ffbb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 11:12:01.000000000","message":"Patch Set 19: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron-tempest-plugin on branch master.  The error was:\n\n  while scanning a double-quoted scalar\n  in \"openstack/neutron-tempest-plugin/.zuul.yaml@master\", line 1375, column 27\nfound unknown escape character \u0027[\u0027\n  in \"openstack/neutron-tempest-plugin/.zuul.yaml@master\", line 1377, column 15","accounts_in_message":[],"_revision_number":19},{"id":"587add4d80fdcfac43db59236a506b95a18e6c69","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-17 11:16:35.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"2ebaf6bc10a80971e3451bdcb8a2021adf96a6c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 11:16:59.000000000","message":"Patch Set 20: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron-tempest-plugin on branch master.  The error was:\n\n  while scanning a double-quoted scalar\n  in \"openstack/neutron-tempest-plugin/.zuul.yaml@master\", line 1375, column 27\nfound unknown escape character \u0027[\u0027\n  in \"openstack/neutron-tempest-plugin/.zuul.yaml@master\", line 1377, column 15","accounts_in_message":[],"_revision_number":20},{"id":"6784d41e052084a42bde480b547d397ebf1d2acf","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-30 20:03:14.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"bd5e369f57ee46f9f83f6b5e00483ade95ce07bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 23:34:25.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0fdb45d01494447bb1e7f542a6f85dc : SUCCESS in 4m 37s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ae43be32ebc645f7a46908e2e75dbf81 : SUCCESS in 1h 11m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4ce7a9e142464c94ac2a532a4f1d54e5 : FAILURE in 32m 43s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/6ea83f57e2f645028b2f9d2d0aeb2bcf : FAILURE in 1h 55m 00s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/365f6bc3e0184cb4b86549e0c35f3f86 : SUCCESS in 1h 29m 52s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3c4f26dca786425597eefc562bdb5bb2 : SUCCESS in 1h 06m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4d9bc3db075240acbca399ea97f38ae2 : SUCCESS in 1h 31m 38s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/00484c0c92c441bc91c69bd9c2c97fec : SUCCESS in 57m 49s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/62b6315cbae74ddea70d145ef5203550 : SUCCESS in 36m 45s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/23ebc54103ac4a1a8da0e962046c07c1 : FAILURE in 1h 48m 29s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/89ee09ccba65487aba26d2613d1d5e90 : SUCCESS in 1h 25m 07s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/3f47331d809c4174a8301bb3163f997d : SUCCESS in 1h 01m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/d3f0e3fc09c34d89a46653caf11b94be : SUCCESS in 1h 19m 59s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/efd67a6900dc460a9dbdb169951a3719 : SUCCESS in 53m 52s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/75088e09e5eb46f9a12c01518d590ac4 : SUCCESS in 37m 37s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/d6ef61557bd5481eb676b36dc7144e2c : FAILURE in 2h 27m 35s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/d9a523c9955b4524b23c40d034a7bfd0 : SUCCESS in 1h 06m 12s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/055341f71d414dcd8cafbd7d5f99224c : SUCCESS in 1h 26m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/04a0dc97076e4649905bfdd7815c602c : SUCCESS in 1h 34m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac10baf370fd4cb1ba974973c46cf0c7 : SUCCESS in 4m 54s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/b749a65352084728a23002c01f4fe813 : SUCCESS in 1h 40m 37s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/73ffa215758249ef8dc8538d4af759d7 : SUCCESS in 1h 37m 10s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/dbfc0d47d2504cacada3569067984a35 : SUCCESS in 1h 26m 57s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/ef010d6c85e54ff9a4cda27e327839fc : SUCCESS in 1h 26m 03s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/1aaf514f193b4d97828fac6e54210748 : FAILURE in 46m 09s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/f3fd6890aae2450fa4f0704cbf287cd7 : SUCCESS in 42m 48s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/2e0620db82134a5c9184480de61c8fb2 : SUCCESS in 48m 11s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/a670cfbba1b24f99a310e1368264ae39 : SUCCESS in 41m 54s\n- neutron-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/9ae9879fcfd049478b270c0e249d94b9 : TIMED_OUT in 3h 06m 29s","accounts_in_message":[],"_revision_number":21},{"id":"8aaaefcc58f53891f87a757993eb0a84de0e5367","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-25 15:37:18.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"b8f401455e7875fa101983c51953df2c5e7453a3","revisions":{"bb5949880ed07445095cb35eca4c806d89308585":{"kind":"REWORK","_number":1,"created":"2019-10-01 20:50:22.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/43/686043/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/43/686043/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/1"}}},"commit":{"parents":[{"commit":"09ee9e809cff778d4e67d9ff76602ffea8913628","subject":"Merge \"Move neutron-dynamic-routing BGP tests from stadium\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/09ee9e809cff778d4e67d9ff76602ffea8913628"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:44:34.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:50:20.000000000","tz":0},"subject":"[train][goal] Define new neutron-tempest-ipv6-only job","message":"[train][goal] Define new neutron-tempest-ipv6-only job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nwe need to define the integration job which deploy services\non IPv6 and perform testing to make sure service listen and communicate\nover IPv6 properly.\n\nCurrent \u0027tempest-ipv6-only\u0027 IPv6 job (running on neutron gate) only cover neutron\nand run only smoke+ipv6 tests present in Tempest. But neutron-tempest-plugin has\nmore coverage for IPv6 and its stadium projects tests.\n\nThis commit adds the new job \u0027neutron-tempest-ipv6-only\u0027 which can deploy neutron\nand other required services on IPv6 and run all required tests for Neutron and\nits stadium projects IPv6 related tests.\n\nAfter this, we can remove the tempest-ipv6-only from neutron gate and use this new\njob instead.\n\nStory: #2005477\nTask: #35908\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia757540fee60657a801d34a731b91f3a78c17e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bb5949880ed07445095cb35eca4c806d89308585"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bb5949880ed07445095cb35eca4c806d89308585"}]},"branch":"refs/heads/master"},"c046118e8f26661824541ec4b6c992e72e781e20":{"kind":"REWORK","_number":2,"created":"2019-10-01 20:52:50.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/43/686043/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/43/686043/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/2"}}},"commit":{"parents":[{"commit":"09ee9e809cff778d4e67d9ff76602ffea8913628","subject":"Merge \"Move neutron-dynamic-routing BGP tests from stadium\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/09ee9e809cff778d4e67d9ff76602ffea8913628"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:44:34.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:52:48.000000000","tz":0},"subject":"[train][goal] Define new neutron-tempest-ipv6-only job","message":"[train][goal] Define new neutron-tempest-ipv6-only job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nwe need to define the integration job which deploy services\non IPv6 and perform testing to make sure service listen and communicate\nover IPv6 properly.\n\nCurrent \u0027tempest-ipv6-only\u0027 IPv6 job (running on neutron gate) only cover neutron\nand run only smoke+ipv6 tests present in Tempest. But neutron-tempest-plugin has\nmore coverage for IPv6 and its stadium projects tests.\n\nThis commit adds the new job \u0027neutron-tempest-ipv6-only\u0027 which can deploy neutron\nand other required services on IPv6 and run all required tests for Neutron and\nits stadium projects IPv6 related tests.\n\nAfter this, we can remove the tempest-ipv6-only from neutron gate and use this new\njob instead.\n\nStory: #2005477\nTask: #35908\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia757540fee60657a801d34a731b91f3a78c17e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c046118e8f26661824541ec4b6c992e72e781e20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c046118e8f26661824541ec4b6c992e72e781e20"}]},"branch":"refs/heads/master"},"d204dd67874e404c234a6097751ebdd98212076c":{"kind":"REWORK","_number":3,"created":"2019-10-01 20:54:19.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/43/686043/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/43/686043/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/3"}}},"commit":{"parents":[{"commit":"09ee9e809cff778d4e67d9ff76602ffea8913628","subject":"Merge \"Move neutron-dynamic-routing BGP tests from stadium\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/09ee9e809cff778d4e67d9ff76602ffea8913628"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:44:34.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:54:16.000000000","tz":0},"subject":"[train][goal] Define new neutron-tempest-ipv6-only job","message":"[train][goal] Define new neutron-tempest-ipv6-only job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nwe need to define the integration job which deploy services\non IPv6 and perform testing to make sure service listen and communicate\nover IPv6 properly.\n\nCurrent \u0027tempest-ipv6-only\u0027 IPv6 job (running on neutron gate) only cover neutron\nand run only smoke+ipv6 tests present in Tempest. But neutron-tempest-plugin has\nmore coverage for IPv6 and its stadium projects tests.\n\nThis commit adds the new job \u0027neutron-tempest-ipv6-only\u0027 which can deploy neutron\nand other required services on IPv6 and run all required tests for Neutron and\nits stadium projects IPv6 related tests.\n\nAfter this, we can remove the tempest-ipv6-only from neutron gate and use this new\njob instead.\n\nStory: #2005477\nTask: #35908\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia757540fee60657a801d34a731b91f3a78c17e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d204dd67874e404c234a6097751ebdd98212076c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d204dd67874e404c234a6097751ebdd98212076c"}]},"branch":"refs/heads/master"},"51b6506b614940c1b0475fe22a4378c802cf5c93":{"kind":"REWORK","_number":4,"created":"2019-10-02 02:36:46.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/43/686043/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/43/686043/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/4"}}},"commit":{"parents":[{"commit":"09ee9e809cff778d4e67d9ff76602ffea8913628","subject":"Merge \"Move neutron-dynamic-routing BGP tests from stadium\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/09ee9e809cff778d4e67d9ff76602ffea8913628"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:44:34.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-02 02:36:44.000000000","tz":0},"subject":"[train][goal] Define new neutron-tempest-ipv6-only job","message":"[train][goal] Define new neutron-tempest-ipv6-only job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nwe need to define the integration job which deploy services\non IPv6 and perform testing to make sure service listen and communicate\nover IPv6 properly.\n\nCurrent \u0027tempest-ipv6-only\u0027 IPv6 job (running on neutron gate) only cover neutron\nand run only smoke+ipv6 tests present in Tempest. But neutron-tempest-plugin has\nmore coverage for IPv6 and its stadium projects tests.\n\nThis commit adds the new job \u0027neutron-tempest-ipv6-only\u0027 which can deploy neutron\nand other required services on IPv6 and run all required tests for Neutron and\nits stadium projects IPv6 related tests.\n\nAfter this, we can remove the tempest-ipv6-only from neutron gate and use this new\njob instead.\n\nStory: #2005477\nTask: #35908\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia757540fee60657a801d34a731b91f3a78c17e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/51b6506b614940c1b0475fe22a4378c802cf5c93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/51b6506b614940c1b0475fe22a4378c802cf5c93"}]},"branch":"refs/heads/master"},"c240d4b267ed5f2aa44374a145a0cc9cbd9cd24f":{"kind":"REWORK","_number":5,"created":"2019-10-02 19:00:39.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/43/686043/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/43/686043/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/5"}}},"commit":{"parents":[{"commit":"09ee9e809cff778d4e67d9ff76602ffea8913628","subject":"Merge \"Move neutron-dynamic-routing BGP tests from stadium\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/09ee9e809cff778d4e67d9ff76602ffea8913628"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:44:34.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-02 19:00:36.000000000","tz":0},"subject":"[train][goal] Define new neutron-tempest-ipv6-only job","message":"[train][goal] Define new neutron-tempest-ipv6-only job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nwe need to define the integration job which deploy services\non IPv6 and perform testing to make sure service listen and communicate\nover IPv6 properly.\n\nCurrent \u0027tempest-ipv6-only\u0027 IPv6 job (running on neutron gate) only cover neutron\nand run only smoke+ipv6 tests present in Tempest. But neutron-tempest-plugin has\nmore coverage for IPv6 and its stadium projects tests.\n\nThis commit adds the new job \u0027neutron-tempest-ipv6-only\u0027 which can deploy neutron\nand other required services on IPv6 and run all required tests for Neutron and\nits stadium projects IPv6 related tests.\n\nAfter this, we can remove the tempest-ipv6-only from neutron gate and use this new\njob instead.\n\nStory: #2005477\nTask: #35908\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia757540fee60657a801d34a731b91f3a78c17e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c240d4b267ed5f2aa44374a145a0cc9cbd9cd24f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c240d4b267ed5f2aa44374a145a0cc9cbd9cd24f"}]},"branch":"refs/heads/master"},"4b2f551393995c1f323ea1d1adc55c0349d152cd":{"kind":"REWORK","_number":6,"created":"2019-10-16 12:57:18.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/43/686043/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/43/686043/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/6"}}},"commit":{"parents":[{"commit":"09ee9e809cff778d4e67d9ff76602ffea8913628","subject":"Merge \"Move neutron-dynamic-routing BGP tests from stadium\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/09ee9e809cff778d4e67d9ff76602ffea8913628"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:44:34.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-10-16 02:45:52.000000000","tz":120},"subject":"[train][goal] Define new neutron-tempest-ipv6-only job","message":"[train][goal] Define new neutron-tempest-ipv6-only job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nwe need to define the integration job which deploy services\non IPv6 and perform testing to make sure service listen and communicate\nover IPv6 properly.\n\nCurrent \u0027tempest-ipv6-only\u0027 IPv6 job (running on neutron gate) only cover neutron\nand run only smoke+ipv6 tests present in Tempest. But neutron-tempest-plugin has\nmore coverage for IPv6 and its stadium projects tests.\n\nThis commit adds the new job \u0027neutron-tempest-ipv6-only\u0027 which can deploy neutron\nand other required services on IPv6 and run all required tests for Neutron and\nits stadium projects IPv6 related tests.\n\nAfter this, we can remove the tempest-ipv6-only from neutron gate and use this new\njob instead.\n\nStory: #2005477\nTask: #35908\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia757540fee60657a801d34a731b91f3a78c17e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4b2f551393995c1f323ea1d1adc55c0349d152cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4b2f551393995c1f323ea1d1adc55c0349d152cd"}]},"branch":"refs/heads/master"},"7e9bfa977f3ddc954c1c4315714b118885df8661":{"kind":"REWORK","_number":7,"created":"2019-10-16 19:55:31.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/43/686043/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/43/686043/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/7"}}},"commit":{"parents":[{"commit":"09ee9e809cff778d4e67d9ff76602ffea8913628","subject":"Merge \"Move neutron-dynamic-routing BGP tests from stadium\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/09ee9e809cff778d4e67d9ff76602ffea8913628"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:44:34.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-10-16 19:55:16.000000000","tz":120},"subject":"[train][goal] Define new neutron-tempest-ipv6-only job","message":"[train][goal] Define new neutron-tempest-ipv6-only job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nwe need to define the integration job which deploy services\non IPv6 and perform testing to make sure service listen and communicate\nover IPv6 properly.\n\nCurrent \u0027tempest-ipv6-only\u0027 IPv6 job (running on neutron gate) only cover neutron\nand run only smoke+ipv6 tests present in Tempest. But neutron-tempest-plugin has\nmore coverage for IPv6 and its stadium projects tests.\n\nThis commit adds the new job \u0027neutron-tempest-ipv6-only\u0027 which can deploy neutron\nand other required services on IPv6 and run all required tests for Neutron and\nits stadium projects IPv6 related tests.\n\nAfter this, we can remove the tempest-ipv6-only from neutron gate and use this new\njob instead.\n\nStory: #2005477\nTask: #35908\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia757540fee60657a801d34a731b91f3a78c17e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7e9bfa977f3ddc954c1c4315714b118885df8661"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7e9bfa977f3ddc954c1c4315714b118885df8661"}]},"branch":"refs/heads/master"},"50460919a0ce6e49e37c4ce455f268a5627a1cf3":{"kind":"REWORK","_number":8,"created":"2019-11-07 04:28:25.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/43/686043/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/43/686043/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/8"}}},"commit":{"parents":[{"commit":"09ee9e809cff778d4e67d9ff76602ffea8913628","subject":"Merge \"Move neutron-dynamic-routing BGP tests from stadium\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/09ee9e809cff778d4e67d9ff76602ffea8913628"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:44:34.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-07 04:28:21.000000000","tz":0},"subject":"[train][goal] Define new neutron-tempest-ipv6-only job","message":"[train][goal] Define new neutron-tempest-ipv6-only job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nwe need to define the integration job which deploy services\non IPv6 and perform testing to make sure service listen and communicate\nover IPv6 properly.\n\nCurrent \u0027tempest-ipv6-only\u0027 IPv6 job (running on neutron gate) only cover neutron\nand run only smoke+ipv6 tests present in Tempest. But neutron-tempest-plugin has\nmore coverage for IPv6 and its stadium projects tests.\n\nThis commit adds the new job \u0027neutron-tempest-ipv6-only\u0027 which can deploy neutron\nand other required services on IPv6 and run all required tests for Neutron and\nits stadium projects IPv6 related tests.\n\nAfter this, we can remove the tempest-ipv6-only from neutron gate and use this new\njob instead.\n\nStory: #2005477\nTask: #35908\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia757540fee60657a801d34a731b91f3a78c17e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/50460919a0ce6e49e37c4ce455f268a5627a1cf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/50460919a0ce6e49e37c4ce455f268a5627a1cf3"}]},"branch":"refs/heads/master"},"1900ef0fa8b83eeaf77679b647766cf4e1eae171":{"kind":"REWORK","_number":9,"created":"2019-11-07 07:24:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/43/686043/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/43/686043/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/9"}}},"commit":{"parents":[{"commit":"1fadb3589b97075ebb2be0dbf564a60dfff8f4e4","subject":"Merge \"Fix list of API extensions for dvr and designate jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1fadb3589b97075ebb2be0dbf564a60dfff8f4e4"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:44:34.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-07 07:23:43.000000000","tz":0},"subject":"[train][goal] Define new neutron-tempest-ipv6-only job","message":"[train][goal] Define new neutron-tempest-ipv6-only job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nwe need to define the integration job which deploy services\non IPv6 and perform testing to make sure service listen and communicate\nover IPv6 properly.\n\nCurrent \u0027tempest-ipv6-only\u0027 IPv6 job (running on neutron gate) only cover neutron\nand run only smoke+ipv6 tests present in Tempest. But neutron-tempest-plugin has\nmore coverage for IPv6 and its stadium projects tests.\n\nThis commit adds the new job \u0027neutron-tempest-ipv6-only\u0027 which can deploy neutron\nand other required services on IPv6 and run all required tests for Neutron and\nits stadium projects IPv6 related tests.\n\nAfter this, we can remove the tempest-ipv6-only from neutron gate and use this new\njob instead.\n\nStory: #2005477\nTask: #35908\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia757540fee60657a801d34a731b91f3a78c17e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1900ef0fa8b83eeaf77679b647766cf4e1eae171"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1900ef0fa8b83eeaf77679b647766cf4e1eae171"}]},"branch":"refs/heads/master"},"dc28147dcc9413ab75ae39a9550ac5265fdb904b":{"kind":"REWORK","_number":10,"created":"2019-12-09 12:02:09.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/43/686043/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/43/686043/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/10"}}},"commit":{"parents":[{"commit":"da39450b378dc3db54c936c81c44324792eec027","subject":"Merge \"Queens jobs will now be run with same version of plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/da39450b378dc3db54c936c81c44324792eec027"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:44:34.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-12-09 12:01:33.000000000","tz":60},"subject":"[train][goal] Define new neutron-tempest-ipv6-only job","message":"[train][goal] Define new neutron-tempest-ipv6-only job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nwe need to define the integration job which deploy services\non IPv6 and perform testing to make sure service listen and communicate\nover IPv6 properly.\n\nCurrent \u0027tempest-ipv6-only\u0027 IPv6 job (running on neutron gate) only cover neutron\nand run only smoke+ipv6 tests present in Tempest. But neutron-tempest-plugin has\nmore coverage for IPv6 and its stadium projects tests.\n\nThis commit adds the new job \u0027neutron-tempest-ipv6-only\u0027 which can deploy neutron\nand other required services on IPv6 and run all required tests for Neutron and\nits stadium projects IPv6 related tests.\n\nAfter this, we can remove the tempest-ipv6-only from neutron gate and use this new\njob instead.\n\nStory: #2005477\nTask: #35908\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia757540fee60657a801d34a731b91f3a78c17e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/dc28147dcc9413ab75ae39a9550ac5265fdb904b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/dc28147dcc9413ab75ae39a9550ac5265fdb904b"}]},"branch":"refs/heads/master"},"5cba49da668562e975c8626c9c88994914f60965":{"kind":"REWORK","_number":11,"created":"2019-12-09 12:03:39.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/43/686043/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/43/686043/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/11"}}},"commit":{"parents":[{"commit":"da39450b378dc3db54c936c81c44324792eec027","subject":"Merge \"Queens jobs will now be run with same version of plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/da39450b378dc3db54c936c81c44324792eec027"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:44:34.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-12-09 12:03:34.000000000","tz":0},"subject":"[train][goal] Define new neutron-tempest-ipv6-only job","message":"[train][goal] Define new neutron-tempest-ipv6-only job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nwe need to define the integration job which deploy services\non IPv6 and perform testing to make sure service listen and communicate\nover IPv6 properly.\n\nCurrent \u0027tempest-ipv6-only\u0027 IPv6 job (running on neutron gate) only cover neutron\nand run only smoke+ipv6 tests present in Tempest. But neutron-tempest-plugin has\nmore coverage for IPv6 and its stadium projects tests.\n\nThis commit adds the new job \u0027neutron-tempest-ipv6-only\u0027 which can deploy neutron\nand other required services on IPv6 and run all required tests for Neutron and\nits stadium projects IPv6 related tests.\n\nAfter this, we can remove the tempest-ipv6-only from neutron gate and use this new\njob instead.\n\nStory: #2005477\nTask: #35908\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia757540fee60657a801d34a731b91f3a78c17e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5cba49da668562e975c8626c9c88994914f60965"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5cba49da668562e975c8626c9c88994914f60965"}]},"branch":"refs/heads/master"},"24b5da754d29ed99db17b8d1e480bc6ff3ab60cf":{"kind":"REWORK","_number":12,"created":"2019-12-12 00:41:10.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/43/686043/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/43/686043/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/12"}}},"commit":{"parents":[{"commit":"1cf724f7d70cfdc854a09fc429fee7024f55c3df","subject":"Merge \"Migrate neutron-vpnaas tests to neutron-tempest-plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1cf724f7d70cfdc854a09fc429fee7024f55c3df"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:44:34.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-12-12 00:40:59.000000000","tz":0},"subject":"[train][goal] Define new neutron-tempest-ipv6-only job","message":"[train][goal] Define new neutron-tempest-ipv6-only job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nwe need to define the integration job which deploy services\non IPv6 and perform testing to make sure service listen and communicate\nover IPv6 properly.\n\nCurrent \u0027tempest-ipv6-only\u0027 IPv6 job (running on neutron gate) only cover neutron\nand run only smoke+ipv6 tests present in Tempest. But neutron-tempest-plugin has\nmore coverage for IPv6 and its stadium projects tests.\n\nThis commit adds the new job \u0027neutron-tempest-ipv6-only\u0027 which can deploy neutron\nand other required services on IPv6 and run all required tests for Neutron and\nits stadium projects IPv6 related tests.\n\nAfter this, we can remove the tempest-ipv6-only from neutron gate and use this new\njob instead.\n\nStory: #2005477\nTask: #35908\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia757540fee60657a801d34a731b91f3a78c17e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/24b5da754d29ed99db17b8d1e480bc6ff3ab60cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/24b5da754d29ed99db17b8d1e480bc6ff3ab60cf"}]},"branch":"refs/heads/master"},"88f3225ae8bdbc9dd465741d3eb3baaf2c1c00ff":{"kind":"REWORK","_number":13,"created":"2019-12-12 00:42:23.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/43/686043/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/43/686043/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/13"}}},"commit":{"parents":[{"commit":"1cf724f7d70cfdc854a09fc429fee7024f55c3df","subject":"Merge \"Migrate neutron-vpnaas tests to neutron-tempest-plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1cf724f7d70cfdc854a09fc429fee7024f55c3df"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:44:34.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-12-12 00:42:21.000000000","tz":0},"subject":"[train][goal] Define new neutron-tempest-ipv6-only job","message":"[train][goal] Define new neutron-tempest-ipv6-only job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nwe need to define the integration job which deploy services\non IPv6 and perform testing to make sure service listen and communicate\nover IPv6 properly.\n\nCurrent \u0027tempest-ipv6-only\u0027 IPv6 job (running on neutron gate) only cover neutron\nand run only smoke+ipv6 tests present in Tempest. But neutron-tempest-plugin has\nmore coverage for IPv6 and its stadium projects tests.\n\nThis commit adds the new job \u0027neutron-tempest-ipv6-only\u0027 which can deploy neutron\nand other required services on IPv6 and run all required tests for Neutron and\nits stadium projects IPv6 related tests.\n\nAfter this, we can remove the tempest-ipv6-only from neutron gate and use this new\njob instead.\n\nStory: #2005477\nTask: #35908\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia757540fee60657a801d34a731b91f3a78c17e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/88f3225ae8bdbc9dd465741d3eb3baaf2c1c00ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/88f3225ae8bdbc9dd465741d3eb3baaf2c1c00ff"}]},"branch":"refs/heads/master"},"c8d73dbfe870dcfb1685ae3c4f2337c7c5ec0b04":{"kind":"REWORK","_number":14,"created":"2019-12-12 00:44:06.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/43/686043/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/43/686043/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/14"}}},"commit":{"parents":[{"commit":"1cf724f7d70cfdc854a09fc429fee7024f55c3df","subject":"Merge \"Migrate neutron-vpnaas tests to neutron-tempest-plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1cf724f7d70cfdc854a09fc429fee7024f55c3df"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:44:34.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-12-12 00:44:04.000000000","tz":0},"subject":"[train][goal] Define new neutron-tempest-ipv6-only job","message":"[train][goal] Define new neutron-tempest-ipv6-only job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nwe need to define the integration job which deploy services\non IPv6 and perform testing to make sure service listen and communicate\nover IPv6 properly.\n\nCurrent \u0027tempest-ipv6-only\u0027 IPv6 job (running on neutron gate) only cover neutron\nand run only smoke+ipv6 tests present in Tempest. But neutron-tempest-plugin has\nmore coverage for IPv6 and its stadium projects tests.\n\nThis commit adds the new job \u0027neutron-tempest-ipv6-only\u0027 which can deploy neutron\nand other required services on IPv6 and run all required tests for Neutron and\nits stadium projects IPv6 related tests.\n\nAfter this, we can remove the tempest-ipv6-only from neutron gate and use this new\njob instead.\n\nStory: #2005477\nTask: #35908\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia757540fee60657a801d34a731b91f3a78c17e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c8d73dbfe870dcfb1685ae3c4f2337c7c5ec0b04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c8d73dbfe870dcfb1685ae3c4f2337c7c5ec0b04"}]},"branch":"refs/heads/master"},"b8687bb99241b81d902c3455da5c1ee5651af8b2":{"kind":"REWORK","_number":15,"created":"2019-12-12 14:14:11.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/43/686043/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/43/686043/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/15"}}},"commit":{"parents":[{"commit":"1cf724f7d70cfdc854a09fc429fee7024f55c3df","subject":"Merge \"Migrate neutron-vpnaas tests to neutron-tempest-plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1cf724f7d70cfdc854a09fc429fee7024f55c3df"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:44:34.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-12-12 14:13:53.000000000","tz":60},"subject":"[train][goal] Define new neutron-tempest-ipv6-only job","message":"[train][goal] Define new neutron-tempest-ipv6-only job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nwe need to define the integration job which deploy services\non IPv6 and perform testing to make sure service listen and communicate\nover IPv6 properly.\n\nCurrent \u0027tempest-ipv6-only\u0027 IPv6 job (running on neutron gate) only cover neutron\nand run only smoke+ipv6 tests present in Tempest. But neutron-tempest-plugin has\nmore coverage for IPv6 and its stadium projects tests.\n\nThis commit adds the new job \u0027neutron-tempest-ipv6-only\u0027 which can deploy neutron\nand other required services on IPv6 and run all required tests for Neutron and\nits stadium projects IPv6 related tests.\n\nAfter this, we can remove the tempest-ipv6-only from neutron gate and use this new\njob instead.\n\nStory: #2005477\nTask: #35908\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia757540fee60657a801d34a731b91f3a78c17e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b8687bb99241b81d902c3455da5c1ee5651af8b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b8687bb99241b81d902c3455da5c1ee5651af8b2"}]},"branch":"refs/heads/master"},"1e9e510f847d5a4423c377322deac61e8f4cec26":{"kind":"REWORK","_number":16,"created":"2020-03-10 10:28:03.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/43/686043/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/43/686043/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/16"}}},"commit":{"parents":[{"commit":"87fc61a6aba524a261312c3bf068c58ad121efbf","subject":"Merge \"Add tagging of ports during bulk creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/87fc61a6aba524a261312c3bf068c58ad121efbf"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:44:34.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-10 10:27:06.000000000","tz":60},"subject":"[train][goal] Define new neutron-tempest-ipv6-only job","message":"[train][goal] Define new neutron-tempest-ipv6-only job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nwe need to define the integration job which deploy services\non IPv6 and perform testing to make sure service listen and communicate\nover IPv6 properly.\n\nCurrent \u0027tempest-ipv6-only\u0027 IPv6 job (running on neutron gate) only cover neutron\nand run only smoke+ipv6 tests present in Tempest. But neutron-tempest-plugin has\nmore coverage for IPv6 and its stadium projects tests.\n\nThis commit adds the new job \u0027neutron-tempest-ipv6-only\u0027 which can deploy neutron\nand other required services on IPv6 and run all required tests for Neutron and\nits stadium projects IPv6 related tests.\n\nAfter this, we can remove the tempest-ipv6-only from neutron gate and use this new\njob instead.\n\nStory: #2005477\nTask: #35908\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia757540fee60657a801d34a731b91f3a78c17e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1e9e510f847d5a4423c377322deac61e8f4cec26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1e9e510f847d5a4423c377322deac61e8f4cec26"}]},"branch":"refs/heads/master"},"ffe5269ccdc786a60c4bf918ba2821a3aa1acac4":{"kind":"REWORK","_number":17,"created":"2020-03-16 14:07:55.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/43/686043/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/43/686043/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/17"}}},"commit":{"parents":[{"commit":"87fc61a6aba524a261312c3bf068c58ad121efbf","subject":"Merge \"Add tagging of ports during bulk creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/87fc61a6aba524a261312c3bf068c58ad121efbf"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:44:34.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-14 15:27:39.000000000","tz":60},"subject":"[train][goal] Define new neutron-tempest-ipv6-only job","message":"[train][goal] Define new neutron-tempest-ipv6-only job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nwe need to define the integration job which deploy services\non IPv6 and perform testing to make sure service listen and communicate\nover IPv6 properly.\n\nCurrent \u0027tempest-ipv6-only\u0027 IPv6 job (running on neutron gate) only cover neutron\nand run only smoke+ipv6 tests present in Tempest. But neutron-tempest-plugin has\nmore coverage for IPv6 and its stadium projects tests.\n\nThis commit adds the new job \u0027neutron-tempest-ipv6-only\u0027 which can deploy neutron\nand other required services on IPv6 and run all required tests for Neutron and\nits stadium projects IPv6 related tests.\n\nAfter this, we can remove the tempest-ipv6-only from neutron gate and use this new\njob instead.\n\nStory: #2005477\nTask: #35908\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia757540fee60657a801d34a731b91f3a78c17e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ffe5269ccdc786a60c4bf918ba2821a3aa1acac4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ffe5269ccdc786a60c4bf918ba2821a3aa1acac4"}]},"branch":"refs/heads/master"},"71403dfd9351b94cd9eea353f6f7e95a2d59d23e":{"kind":"REWORK","_number":18,"created":"2020-03-17 10:56:06.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/43/686043/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/43/686043/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/18"}}},"commit":{"parents":[{"commit":"87fc61a6aba524a261312c3bf068c58ad121efbf","subject":"Merge \"Add tagging of ports during bulk creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/87fc61a6aba524a261312c3bf068c58ad121efbf"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:44:34.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-17 10:56:02.000000000","tz":0},"subject":"[train][goal] Define new neutron-tempest-ipv6-only job","message":"[train][goal] Define new neutron-tempest-ipv6-only job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nwe need to define the integration job which deploy services\non IPv6 and perform testing to make sure service listen and communicate\nover IPv6 properly.\n\nCurrent \u0027tempest-ipv6-only\u0027 IPv6 job (running on neutron gate) only cover neutron\nand run only smoke+ipv6 tests present in Tempest. But neutron-tempest-plugin has\nmore coverage for IPv6 and its stadium projects tests.\n\nThis commit adds the new job \u0027neutron-tempest-ipv6-only\u0027 which can deploy neutron\nand other required services on IPv6 and run all required tests for Neutron and\nits stadium projects IPv6 related tests.\n\nAfter this, we can remove the tempest-ipv6-only from neutron gate and use this new\njob instead.\n\nStory: #2005477\nTask: #35908\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia757540fee60657a801d34a731b91f3a78c17e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/71403dfd9351b94cd9eea353f6f7e95a2d59d23e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/71403dfd9351b94cd9eea353f6f7e95a2d59d23e"}]},"branch":"refs/heads/master"},"2c4b220bb1e802938a73f1ebb90913845c90a455":{"kind":"REWORK","_number":19,"created":"2020-03-17 11:11:43.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/43/686043/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/43/686043/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/19"}}},"commit":{"parents":[{"commit":"87fc61a6aba524a261312c3bf068c58ad121efbf","subject":"Merge \"Add tagging of ports during bulk creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/87fc61a6aba524a261312c3bf068c58ad121efbf"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:44:34.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-17 11:11:40.000000000","tz":0},"subject":"[train][goal] Define new neutron-tempest-ipv6-only job","message":"[train][goal] Define new neutron-tempest-ipv6-only job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nwe need to define the integration job which deploy services\non IPv6 and perform testing to make sure service listen and communicate\nover IPv6 properly.\n\nCurrent \u0027tempest-ipv6-only\u0027 IPv6 job (running on neutron gate) only cover neutron\nand run only smoke+ipv6 tests present in Tempest. But neutron-tempest-plugin has\nmore coverage for IPv6 and its stadium projects tests.\n\nThis commit adds the new job \u0027neutron-tempest-ipv6-only\u0027 which can deploy neutron\nand other required services on IPv6 and run all required tests for Neutron and\nits stadium projects IPv6 related tests.\n\nAfter this, we can remove the tempest-ipv6-only from neutron gate and use this new\njob instead.\n\nStory: #2005477\nTask: #35908\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia757540fee60657a801d34a731b91f3a78c17e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2c4b220bb1e802938a73f1ebb90913845c90a455"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2c4b220bb1e802938a73f1ebb90913845c90a455"}]},"branch":"refs/heads/master"},"bf2f78b1c764fd10219431f82ed95d3eeef31152":{"kind":"REWORK","_number":20,"created":"2020-03-17 11:16:35.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/43/686043/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/43/686043/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/20"}}},"commit":{"parents":[{"commit":"87fc61a6aba524a261312c3bf068c58ad121efbf","subject":"Merge \"Add tagging of ports during bulk creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/87fc61a6aba524a261312c3bf068c58ad121efbf"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:44:34.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-17 11:16:30.000000000","tz":0},"subject":"[train][goal] Define new neutron-tempest-ipv6-only job","message":"[train][goal] Define new neutron-tempest-ipv6-only job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nwe need to define the integration job which deploy services\non IPv6 and perform testing to make sure service listen and communicate\nover IPv6 properly.\n\nCurrent \u0027tempest-ipv6-only\u0027 IPv6 job (running on neutron gate) only cover neutron\nand run only smoke+ipv6 tests present in Tempest. But neutron-tempest-plugin has\nmore coverage for IPv6 and its stadium projects tests.\n\nThis commit adds the new job \u0027neutron-tempest-ipv6-only\u0027 which can deploy neutron\nand other required services on IPv6 and run all required tests for Neutron and\nits stadium projects IPv6 related tests.\n\nAfter this, we can remove the tempest-ipv6-only from neutron gate and use this new\njob instead.\n\nStory: #2005477\nTask: #35908\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia757540fee60657a801d34a731b91f3a78c17e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bf2f78b1c764fd10219431f82ed95d3eeef31152"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bf2f78b1c764fd10219431f82ed95d3eeef31152"}]},"branch":"refs/heads/master"},"b8f401455e7875fa101983c51953df2c5e7453a3":{"kind":"REWORK","_number":21,"created":"2020-03-30 20:03:14.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/43/686043/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/43/686043/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/43/686043/21"}}},"commit":{"parents":[{"commit":"87fc61a6aba524a261312c3bf068c58ad121efbf","subject":"Merge \"Add tagging of ports during bulk creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/87fc61a6aba524a261312c3bf068c58ad121efbf"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-01 20:44:34.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-30 20:02:54.000000000","tz":120},"subject":"[train][goal] Define new neutron-tempest-ipv6-only job","message":"[train][goal] Define new neutron-tempest-ipv6-only job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nwe need to define the integration job which deploy services\non IPv6 and perform testing to make sure service listen and communicate\nover IPv6 properly.\n\nCurrent \u0027tempest-ipv6-only\u0027 IPv6 job (running on neutron gate) only cover neutron\nand run only smoke+ipv6 tests present in Tempest. But neutron-tempest-plugin has\nmore coverage for IPv6 and its stadium projects tests.\n\nThis commit adds the new job \u0027neutron-tempest-ipv6-only\u0027 which can deploy neutron\nand other required services on IPv6 and run all required tests for Neutron and\nits stadium projects IPv6 related tests.\n\nAfter this, we can remove the tempest-ipv6-only from neutron gate and use this new\njob instead.\n\nStory: #2005477\nTask: #35908\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia757540fee60657a801d34a731b91f3a78c17e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b8f401455e7875fa101983c51953df2c5e7453a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b8f401455e7875fa101983c51953df2c5e7453a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
