)]}'
{"id":"openstack%2Fneutron-tempest-plugin~699369","triplet_id":"openstack%2Fneutron-tempest-plugin~master~If5a363cb9582b44e780165d537cebd217962db4c","project":"openstack/neutron-tempest-plugin","branch":"master","hashtags":[],"change_id":"If5a363cb9582b44e780165d537cebd217962db4c","subject":"[DNM] Change cirros image used for tests","status":"ABANDONED","created":"2019-12-17 08:30:13.000000000","updated":"2020-03-05 10:42:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5b335472a2c0542b3590be2511ae23d5ac13a374","_number":699369,"virtual_id_number":699369,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-17 11:47:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"all":[{"value":-1,"date":"2019-12-17 10:05:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"all":[{"value":-1,"date":"2019-12-17 10:05:52.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-17 11:47:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"34bdc16c3abde66e6c763a3a53d4dbbf445795b5","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-12-17 08:30:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"552148accb11987d7b286883a94d9e784cfaac1b","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-12-17 10:05:52.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b90a0a9d3509a35cf3e1b50e123b226ae45cce5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 11:47:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfc7f63e5cf14d48934a2a6692fc005c : SUCCESS in 4m 06s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/116c74866bb54e1695dfcc5e35097e14 : SUCCESS in 53m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/afee97442752432cbb02a67baed0151f : SUCCESS in 33m 05s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/784aaaed294d413a8680b2645d446b26 : TIMED_OUT in 3h 04m 57s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/74bc9890a5b447e8a755384f2e361c22 : TIMED_OUT in 2h 51m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3459cb6c21e6459b895d53846c5cc87b : TIMED_OUT in 2h 52m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dff0552b0f7c4687a89e4f91606d5a82 : TIMED_OUT in 2h 52m 26s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/1ea6bf0efc334b959bf6dba1ac405fbc : SUCCESS in 49m 46s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/d0c203534bf742be99917ffadcbf2377 : SUCCESS in 44m 26s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/08b9db2f02484597bf6b22982a4ea4bd : TIMED_OUT in 3h 08m 39s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/56deab01baa6496f950b33591ca95fb6 : TIMED_OUT in 2h 52m 59s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/eccee020544a43348216bd6e4e013bd8 : TIMED_OUT in 2h 55m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/e79de37623d04afba072faadfdeacf86 : TIMED_OUT in 2h 55m 59s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/7027b009baed48bd934d0a82e64f0866 : SUCCESS in 1h 00m 54s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/97e122552e774dc0873a47d19a2883e7 : SUCCESS in 46m 53s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/8ad70aeded4d4b24b40d8c3f9144bba7 : TIMED_OUT in 3h 05m 21s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/301b9260af484f1d88c8c6fc4e942fe2 : TIMED_OUT in 2h 53m 16s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/341a1bef6a354f7891e8142c9f61c09d : TIMED_OUT in 2h 52m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/d56d01db3a264e22b40f1c811d56e217 : TIMED_OUT in 2h 53m 24s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/ec96a72c2ac94e319ef912998dc0110d : SUCCESS in 1h 12m 06s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/090989e232924f4d9b1326b3b6e92b77 : SUCCESS in 57m 25s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/d4294b72f21e487b84d9cba955045c7d : TIMED_OUT in 3h 07m 05s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/436b0a91e8164bd2adced100ca2d9405 : TIMED_OUT in 2h 53m 01s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/e26c0bb9e8864d589138e213d51bd754 : TIMED_OUT in 2h 51m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/35ed5c11676945db9829a40d283209e9 : TIMED_OUT in 2h 52m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f7600eb67784dec87501515de365de0 : SUCCESS in 4m 11s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/92c80edb3c8649d19bb47bd7e3e1a09e : SUCCESS in 1h 30m 34s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/1520acb7bc544b118500e1e465ed34d9 : SUCCESS in 2h 00m 52s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/925549d6a6cf4498bde686de3053a9f8 : SUCCESS in 2h 05m 16s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/6ec865f1726248fa90100d041d0b4d25 : SUCCESS in 2h 05m 07s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/cb456a56700845dfbf8b383f31139519 : FAILURE in 38m 45s\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/82cd219b1355435080fbf4cd33b25fdc : SUCCESS in 39m 10s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/39dfcee47a544d28a8a3a931e838f7d3 : SUCCESS in 50m 00s (non-voting)\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/1e7de4ebdf2d4ab7b6c1adaa85215c3a : SUCCESS in 1h 10m 03s","accounts_in_message":[],"_revision_number":1},{"id":"5b335472a2c0542b3590be2511ae23d5ac13a374","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-03-05 10:42:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"70c2739b334fc6cad68c46a18ae75881dc6e3a1d","revisions":{"70c2739b334fc6cad68c46a18ae75881dc6e3a1d":{"kind":"REWORK","_number":1,"created":"2019-12-17 08:30:13.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/69/699369/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/69/699369/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/699369/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/699369/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/699369/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/699369/1"}}},"commit":{"parents":[{"commit":"616a27a1b0125e6d2b820bb1cb0d7b623a0f3394","subject":"Merge \"Add stable/train jobs definitions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/616a27a1b0125e6d2b820bb1cb0d7b623a0f3394"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-12-17 08:31:06.000000000","tz":60},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-12-17 08:31:06.000000000","tz":60},"subject":"[DNM] Change cirros image used for tests","message":"[DNM] Change cirros image used for tests\n\nWhile checking bug https://bugs.launchpad.net/neutron/+bug/1856523\nwe found that potentially problem is in cirros.\nThis patch uses modified cirros image that has removed \u0027--silent\u0027\nfrom CURL.\n\nChange-Id: If5a363cb9582b44e780165d537cebd217962db4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/70c2739b334fc6cad68c46a18ae75881dc6e3a1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/70c2739b334fc6cad68c46a18ae75881dc6e3a1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
