)]}'
{"id":"openstack%2Fneutron-tempest-plugin~702868","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I16823d2578d35b53e2a50c22ff2ab4537df520b1","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/1859988","hashtags":[],"change_id":"I16823d2578d35b53e2a50c22ff2ab4537df520b1","subject":"Use queens-em tempest for Queens jobs","status":"ABANDONED","created":"2020-01-16 14:00:58.000000000","updated":"2020-01-23 09:34:13.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fa83b37ef4ff582a053955f7bce480555540e15d","_number":702868,"virtual_id_number":702868,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-01-16 16:30:14.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"date":"2020-01-17 08:03:47.000000000","_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-01-16 16:22:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"date":"2020-01-20 09:56:56.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-01-17 12:39:27.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"date":"2020-01-17 13:07:02.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"date":"2020-01-16 15:20:30.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-16 15:20:30.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-01-16 16:22:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-16 16:30:14.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-01-17 08:03:47.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-01-17 12:39:27.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-01-17 13:07:02.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"0a61bb45253b04430af3b7628076aa6d8fbc74c0","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-16 14:00:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7afa0ebff61cfd946293766c56cd8a3b94d2eee2","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-16 14:02:36.000000000","message":"Patch Set 1:\n\nTesting with added Depends-On in https://review.opendev.org/#/c/701900/ backport","accounts_in_message":[],"_revision_number":1},{"id":"40452320cb4d3abe40f42880fde07dd7f71b8834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 16:22:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae3fb52c932041df8d6b1cd028f7850c : SUCCESS in 5m 20s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/029d084efbd544809fbe89561f0b0f3a : SUCCESS in 51m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4aa5ad24ecca4e8fb9981045d68a0a35 : SUCCESS in 45m 51s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/0971a65954814bfeb98198ab90a8f7eb : FAILURE in 40m 00s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/bd5d3512bec64c58bf9a25315da5775f : FAILURE in 28m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f184c922958d466faebece6e2cccecbd : SUCCESS in 1h 05m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2d1e0c31e12a46a58459787c8590e18a : SUCCESS in 1h 27m 51s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/7dfb086332564100a05ff5dcc4a00913 : FAILURE in 39m 36s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/56f5b560c629428fa6767792b38abc8b : FAILURE in 36m 15s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/108a03722368496ea986c639dd8daf5a : FAILURE in 46m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/30e1197e2af34868b8dafaef50077e00 : FAILURE in 34m 24s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/529aac8c1cff4cc3932dcbf6fcc74407 : FAILURE in 33m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/8f4eca04de8e4654a540a104a8c00f94 : FAILURE in 34m 29s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/c1f4bf07fc664945a4eb3e1743e074eb : SUCCESS in 48m 47s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/91e61d508de04300862738ba2c3378aa : FAILURE in 31m 55s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/3e4ba2ed55ff46318e1f48cd4625af1d : FAILURE in 41m 44s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/3c7f101a029143218d9f788d12c409a1 : SUCCESS in 1h 14m 25s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/697f143c573446ebbcee0340dc7fee93 : SUCCESS in 1h 07m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/2e4958cd778e40678aad4d00d4b25419 : SUCCESS in 1h 21m 21s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/741ef07b60054ce2b9d0bd9459c22a27 : SUCCESS in 1h 10m 42s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/efea1d589a33485183692a0bc18e8e4b : SUCCESS in 44m 01s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/7c1004cd10184abda5803795bcb5423f : SUCCESS in 2h 14m 54s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/67cae7620ed143fca54217e4f38c858c : FAILURE in 38m 36s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/3e1ed0f8164846b4b9dd141627bf45ae : SUCCESS in 1h 21m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/10673ab3aa144c07afdf8aff7a94e93d : SUCCESS in 1h 24m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d193792a468143ff9d25ba92cf08a45c : SUCCESS in 5m 41s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/717f86aca7604339b1ffaefee06c9a17 : SUCCESS in 1h 50m 48s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/aa935ad614214cbda046f6d91956b746 : SUCCESS in 2h 13m 22s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/8d52deff823440b699308cc7dd8d86ee : SUCCESS in 1h 30m 57s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/9945da6799bd408c9806f05f668fa4cd : SUCCESS in 1h 38m 48s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/8e8c5dd8e50a4756afd8a3a23c376548 : SUCCESS in 45m 04s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/db6543827a884b9abd260c6b711bd876 : SUCCESS in 39m 35s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/a6d80044f5f04aff9cb7a2b62bd5d55c : SUCCESS in 1h 01m 40s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/7264c936864a4ff19aefc9a62a742774 : FAILURE in 56m 32s","accounts_in_message":[],"_revision_number":1},{"id":"5b1709526924b9882b39f63887e64ec02129fedd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-16 16:30:14.000000000","message":"Patch Set 1:\n\nI\u0027m having the same problem in my patches:\n\n2020-01-16 14:44:48.731166 | controller | Connecting to download.cirros-cloud.net (download.cirros-cloud.net)|2607:f298:6:a036::bd6:a72a|:80... failed: Network is unreachable.","accounts_in_message":[],"_revision_number":1},{"id":"7b8ead555000d77c09eb03bd76ad7e4bf8968e63","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-17 08:03:47.000000000","message":"Patch Set 1:\n\n\u003e I\u0027m having the same problem in my patches:\n \u003e \n \u003e 2020-01-16 14:44:48.731166 | controller | Connecting to\n \u003e download.cirros-cloud.net (download.cirros-cloud.net)|2607:f298:6:a036::bd6:a72a|:80...\n \u003e failed: Network is unreachable.\n\nHi, Based on this mail thread: http://lists.openstack.org/pipermail/openstack-discuss/2020-January/012024.html I suppose the cirros \"cache\" with cirros 0.4.0 works only for master devstack.","accounts_in_message":[],"_revision_number":1},{"id":"02c34c634052e24038a8e8110a39ba7a640c919d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-17 09:09:07.000000000","message":"Patch Set 1:\n\nHe true, I had the same issue manually when one of my test scripts failed to download cirros image, looks like it is back atm","accounts_in_message":[],"_revision_number":1},{"id":"4f45945610a1e24eb0339eff05d45ffcb5df3f4d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-17 09:09:13.000000000","message":"Patch Set 1:\n\nrecheck cirros","accounts_in_message":[],"_revision_number":1},{"id":"6611d808a0334b23a01011ff615e4d8528af761c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-17 10:40:30.000000000","message":"Patch Set 1:\n\nNow to wait for rocky fix and python2 EOL drama bug","accounts_in_message":[],"_revision_number":1},{"id":"17dce03292e60f9c20850c7958aff5a71abbb52f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-17 12:39:27.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"3d2f31902e6c20fee52103c980161de8c82781a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 12:40:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc0f36dbd6c24a179e4b9df48748a832 : SUCCESS in 7m 27s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/607f9a356ccd4a1185227b026b62ec68 : SUCCESS in 50m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/167ca0206b1c44a4abb3b3209eb03882 : SUCCESS in 34m 28s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/8efad728fce540e693a4d412ff94b382 : FAILURE in 1h 36m 43s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/55341b0a3c764128a54ae31ada780e5d : SUCCESS in 1h 14m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ea3ec61a453c4959a44066c261960208 : SUCCESS in 1h 00m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/979fb24f71964ac4afd46296eccf3211 : SUCCESS in 1h 02m 07s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/8639c9e08e5a4dc09219679cece86a4a : FAILURE in 32m 50s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/44c3d5dd40224c388e0df569a64c84ac : FAILURE in 26m 33s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/df06cdc8017b43fda891a0f9e702ea75 : FAILURE in 46m 55s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/4ebde3aa7416466e8363703c08fa8fcb : FAILURE in 35m 15s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/06cdb0ce4e4d4f0f80bc98abfd72b117 : FAILURE in 34m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/ce4af029590443f2ab07e0e298bbab46 : FAILURE in 35m 00s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/b157ed8643004b608b4b7e9206860745 : SUCCESS in 49m 09s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/922462c05e6c43908f15876b51b9f20f : SUCCESS in 38m 08s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/48c20174a4be4a30a836b174cf7aa163 : FAILURE in 1h 47m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/4e761d2fb21c46c0a91107a8ad629a9f : SUCCESS in 1h 00m 06s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/b28345e10af34f739b4906ccee9a097e : SUCCESS in 1h 04m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/f1b46a44057b48e28da4cc8283c7f9a1 : SUCCESS in 51m 55s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/bfad3532e314406c92723f20fe1b0531 : SUCCESS in 52m 41s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/3bcf28e20d0b4e6f9622f9766e48e7c7 : SUCCESS in 46m 30s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/c8ec10fe409c4c6480882a26abce7694 : SUCCESS in 1h 39m 18s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/2541f0ddad2743b8b8da4743d0ee9689 : SUCCESS in 59m 10s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/f5c94e47a0ed4868be9c24c5c6d16ae2 : SUCCESS in 1h 02m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/d0fbe3cdf6c7431982b4194f85edb5ff : SUCCESS in 56m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7beec963d46e4351b4b15045943bc9d5 : SUCCESS in 5m 04s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/a8db9a64b5e14ae5a73a6ef0dccfe2b3 : SUCCESS in 1h 46m 12s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/98de6b9a340542a0b92100a8b71c6ad1 : SUCCESS in 1h 39m 27s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/6d8c1f909f02465bb9cf9e1a5ba0f43a : SUCCESS in 1h 44m 31s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/2b759af7fdc0464db547ab2f2fcfd3a0 : SUCCESS in 1h 47m 48s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/f356733629c3478b8423262b38f1db9e : SUCCESS in 40m 54s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/bf99c9fe0d9a4f4590cf50ca796cdfde : SUCCESS in 42m 31s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/faab137b5ee047f8ba28923a42bfe739 : SUCCESS in 51m 56s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/a9fa1abc7c0340b0964755053b0b4ff7 : SUCCESS in 35m 49s","accounts_in_message":[],"_revision_number":1},{"id":"f9731a18adc580e7444c4c13f9d150b0ad2edb84","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-17 12:53:09.000000000","message":"Patch Set 1: Workflow-1\n\nOK per https://bugs.launchpad.net/neutorn/+bug/1860033/comments/5 we also need to set u-c to work around the python2 eol drama issue","accounts_in_message":[],"_revision_number":1},{"id":"e80f4176d295b33c550123e525e90e543e640b22","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-20 09:56:56.000000000","message":"Patch Set 1: -Workflow\n\nRest of the fix may be in devstack itself, or set upper-constraints in job definition here.\nDiscussion at http://lists.openstack.org/pipermail/openstack-discuss/2020-January/012052.html","accounts_in_message":[],"_revision_number":1},{"id":"9b6df61cc0dbbb83717c874e8c5b8fc2e87e8f26","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-23 09:34:13.000000000","message":"Abandoned\n\nFull fix in progress at https://review.opendev.org/#/c/703903/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2a1bd21deadf43af21f57a2982ffc646f2600f9","revisions":{"a2a1bd21deadf43af21f57a2982ffc646f2600f9":{"kind":"REWORK","_number":1,"created":"2020-01-16 14:00:58.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/68/702868/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/68/702868/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/702868/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/702868/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/702868/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/702868/1"}}},"commit":{"parents":[{"commit":"560737404a9d00879c1f2da4f523bb1a03599164","subject":"Merge \"Revert \"Switch neutron-dynamic-routing job to be non-voting temporaray\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/560737404a9d00879c1f2da4f523bb1a03599164"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-01-16 13:56:36.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-01-16 13:56:36.000000000","tz":60},"subject":"Use queens-em tempest for Queens jobs","message":"Use queens-em tempest for Queens jobs\n\nJobs already use a pinned version of the plugin for\nneutron-tempest-plugin itself, but recently Queens backports started to\nfail on tempest lib code\n\nAlso set fixed version (queens-em tag) of tempest for these jobs\n\nChange-Id: I16823d2578d35b53e2a50c22ff2ab4537df520b1\nCloses-Bug: #1859988\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a2a1bd21deadf43af21f57a2982ffc646f2600f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a2a1bd21deadf43af21f57a2982ffc646f2600f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
