)]}'
{"id":"openstack%2Fneutron-tempest-plugin~702987","triplet_id":"openstack%2Fneutron-tempest-plugin~master~If937a1901316b1906968013a159c0961c34e2563","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/1860033","hashtags":[],"change_id":"If937a1901316b1906968013a159c0961c34e2563","subject":"DNM: TEST for bug 1860033","status":"ABANDONED","created":"2020-01-16 21:12:24.000000000","updated":"2020-01-17 18:51:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b2f592e925ec5bdefea5745c5ae274be9713cbb8","_number":702987,"virtual_id_number":702987,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-01-16 23:33:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-16 23:33:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9558fa9e953dbe343e5b95fca8cf45f8f8270810","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-16 21:12:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46245afefdd1f972e92bad36689c7618080c8e93","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-16 21:13:45.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"31d8a7f56e2309ac0643a46bb14ac78b31dae8b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 21:23:47.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"cae27b763a9a6236c1f56578cbe134874e67738d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-16 21:24:56.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"20e6e3329a8e702b7c477263b40c9220348b63a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 23:33:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91212d3bbfe64c2c93391abb2c81adf9 : SUCCESS in 3m 46s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4dedf852263c48cc8c20118c23d26148 : SUCCESS in 44m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cd8e1cc081104b99891e523d66c3a638 : SUCCESS in 35m 30s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/0cd22e8c6ce44fa3b5e3be7c2f43f9c7 : FAILURE in 2h 02m 08s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ba6d99e7007646f0b9695aa9b4243773 : SUCCESS in 1h 04m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/095c8476024f4870bc610190f92849f9 : SUCCESS in 1h 15m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0f5ceb6c785c4291a01192e6f1b463a3 : SUCCESS in 1h 03m 54s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/2eabc183ca5a49978a0554b7f7c0b166 : FAILURE in 34m 20s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/6a92044097ac4674ba88c52263b10fe7 : FAILURE in 29m 14s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/cd195abe226a4c199a83566f610e5edb : FAILURE in 43m 16s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/4cfd715153c848f5b6c758479b1537cc : FAILURE in 28m 20s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/70241b9111914e3280e9c34ebee2e817 : FAILURE in 30m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/d25f68236bcd4a5da40c12e288bcb7bd : FAILURE in 29m 27s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/2309126ffb5f494ca1c6f3718a4f42be : SUCCESS in 49m 02s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/aea513188f5540cba8a9ab265d640966 : SUCCESS in 34m 06s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/aa9155f9ceaa4fe3a819c86c2473b5cb : SUCCESS in 1h 36m 58s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/c105508f85fe46ca9a53f6f8b057f39a : SUCCESS in 1h 05m 57s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/21b7917bf13943edb317a85eb192708b : SUCCESS in 1h 07m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/6d690c74503f4dd382142e3db084020e : SUCCESS in 1h 03m 36s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/d6a1edddcef2452a809b90234f901df1 : SUCCESS in 53m 42s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/803dc84d08ee467c835ae515e0ea717d : SUCCESS in 35m 31s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/177010c96179428199f4b9a5fdda3c5e : SUCCESS in 1h 32m 34s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/9a67a6ebf8cf4600970d05544153f81c : SUCCESS in 1h 10m 35s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/8ab1ce8f1fbc42aca79ecb29976f6205 : SUCCESS in 1h 09m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/d7f22d6a9100485d87811964c4633079 : SUCCESS in 1h 07m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af83bac4d45144eeb2efd66900136ce3 : SUCCESS in 4m 27s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/a75c21d5f60844dd970fd212fb429fe6 : SUCCESS in 1h 41m 07s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/5b58a011e2d84eea9fbd2bd554f3cabd : SUCCESS in 1h 43m 48s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/524de62dc48c48d8aadc00d21cd53807 : SUCCESS in 1h 35m 00s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/edb93a46fe334dd89c408106bd01d8ed : SUCCESS in 1h 29m 32s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/cf210fcf8dea4d48be34c5b77c35c8cf : SUCCESS in 40m 44s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/1113090b61bb4ccbab59137fab079973 : SUCCESS in 38m 10s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/f93d012758af46c9b64fce5a4f4932e0 : SUCCESS in 1h 11m 00s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/426e2e5ab89b4263887062f18c24c131 : SUCCESS in 40m 25s","accounts_in_message":[],"_revision_number":2},{"id":"127f0316b0ed2ba622ecae8da77db3a54b32e2c2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-17 18:51:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ff4c6a3c7be9689926e0f5546c03d95d13066739","revisions":{"5af2eaf9948fd894879a9159d0090c0a08cf7f41":{"kind":"REWORK","_number":1,"created":"2020-01-16 21:12:24.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/87/702987/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/87/702987/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/702987/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/702987/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/702987/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/702987/1"}}},"commit":{"parents":[{"commit":"560737404a9d00879c1f2da4f523bb1a03599164","subject":"Merge \"Revert \"Switch neutron-dynamic-routing job to be non-voting temporaray\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/560737404a9d00879c1f2da4f523bb1a03599164"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-16 21:11:36.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-16 21:11:36.000000000","tz":-300},"subject":"DNM: TEST for bug 1860033","message":"DNM: TEST for bug 1860033\n\nDepends-on: https://review.opendev.org/702986\nChange-Id: If937a1901316b1906968013a159c0961c34e2563\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5af2eaf9948fd894879a9159d0090c0a08cf7f41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5af2eaf9948fd894879a9159d0090c0a08cf7f41"}]},"branch":"refs/heads/master"},"ff4c6a3c7be9689926e0f5546c03d95d13066739":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-01-16 21:24:56.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/87/702987/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/87/702987/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/702987/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/702987/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/702987/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/87/702987/2"}}},"commit":{"parents":[{"commit":"560737404a9d00879c1f2da4f523bb1a03599164","subject":"Merge \"Revert \"Switch neutron-dynamic-routing job to be non-voting temporaray\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/560737404a9d00879c1f2da4f523bb1a03599164"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-16 21:11:36.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-16 21:24:37.000000000","tz":-300},"subject":"DNM: TEST for bug 1860033","message":"DNM: TEST for bug 1860033\n\nTest #2\n\nDepends-on: https://review.opendev.org/702986\nChange-Id: If937a1901316b1906968013a159c0961c34e2563\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ff4c6a3c7be9689926e0f5546c03d95d13066739"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ff4c6a3c7be9689926e0f5546c03d95d13066739"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
