)]}'
{"id":"openstack%2Fneutron-tempest-plugin~703903","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I07beb6e7bdcbaf7ffe72d47f9edb90df40641e68","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/1859988","hashtags":[],"change_id":"I07beb6e7bdcbaf7ffe72d47f9edb90df40641e68","subject":"Remove TEMPEST_BRANCH var from neutron queens job.","status":"MERGED","created":"2020-01-23 00:35:14.000000000","updated":"2020-01-25 13:42:04.000000000","submitted":"2020-01-25 13:42:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703903-1579959724079-cc57de1a","meta_rev_id":"990262044bd6880942c52fdbf6da43a5945cc187","_number":703903,"virtual_id_number":703903,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-25 13:42:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-01-24 14:12:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2020-01-24 14:12:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-01-24 16:25:47.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2020-01-24 16:25:47.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-01-24 14:12:57.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2020-01-24 14:12:57.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-23 23:02:04.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-01-24 09:19:37.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-01-24 14:08:24.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-01-24 14:12:13.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-01-24 16:25:47.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-01-25 13:42:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2b6774e65e3e879c9cab70c804416e96f4d0c7e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-23 00:35:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8cc55cd0cd18e551142119a3aea4ffaf5c2885ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 03:20:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b51341ed30004f088ba80072df10f0b3 : SUCCESS in 4m 42s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/440bebe2f22c4ff0a8b8d94b93ee2344 : SUCCESS in 48m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/efc4795742c245a5bbea73ecc55d4f9b : SUCCESS in 32m 39s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/bf99906092da4eaf91c6557b2339de32 : FAILURE in 1h 39m 34s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ff5a2706a1c442e5b726d9010df059f8 : SUCCESS in 1h 04m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4fe8e97f8473428f9609994bb3891119 : SUCCESS in 1h 28m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fb7020772e744964b517944d5d8bb6e5 : SUCCESS in 59m 38s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/22a2b2a942bd4639901f0f5417f15185 : SUCCESS in 48m 17s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/25e0b86ca352485b96570fbcf22a4732 : SUCCESS in 35m 23s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/d91139e62eff4f6ca780d5e14661ee63 : FAILURE in 1h 32m 16s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/0409bf58f0ea42518b33b9c9d7952e4b : SUCCESS in 56m 05s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/4436b55257ae427598bf7aeea9fb0ce9 : SUCCESS in 1h 00m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/697942de834a4cdb9fc5aead878bfe94 : SUCCESS in 53m 40s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/c20c6c5d87294388a4781e34782beccd : SUCCESS in 48m 49s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/43e1a8cfeece47638e880b9f14634e06 : SUCCESS in 33m 53s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/70d9d96673f24ac1a369a339ea77c847 : FAILURE in 1h 32m 05s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/220feb0f3fa94c32b03bb9606f16d99c : SUCCESS in 1h 13m 19s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/10457821ede7492cbd65c0bc6aad53e7 : SUCCESS in 1h 01m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/79a8ec39cb4d455381e87095f700273f : SUCCESS in 1h 16m 06s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/b2a4390bb4e543f698cec30fd876ff4e : SUCCESS in 48m 45s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/de86b70419b042138ad9fd8abcffdb23 : SUCCESS in 38m 24s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/b5430705c63641c08a8d99bdc37b8878 : FAILURE in 1h 27m 20s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/64b1c958e23447d58fb61984b5bb9fe7 : SUCCESS in 1h 14m 47s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/f31b338fc462406a94309a1be33696da : SUCCESS in 1h 30m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/598881dc8e9d4a279098aa5ae1ea1772 : SUCCESS in 1h 19m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ceab8f3964f4fe3978096f34a4d640f : SUCCESS in 4m 30s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/1ccb9ef9ea26432aa1daca991031d858 : SUCCESS in 1h 53m 20s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/7993c586fabc45a59b4bf9aa279ef44b : SUCCESS in 1h 40m 55s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/5ff12d6289844e5e84c0fc7a5cb4a533 : SUCCESS in 1h 23m 14s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/f272d248612f43d28211692509e64839 : SUCCESS in 1h 20m 44s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/ee85984b6e864adc9bc254e94d784ac3 : SUCCESS in 35m 33s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/1f8accb07c194c83bc893b89c23d59f6 : SUCCESS in 40m 01s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/b126ba60188b466b842d37a8848bc8ba : SUCCESS in 45m 44s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/66f9f3fd01c84d7eb82a738379bc4845 : SUCCESS in 39m 30s","accounts_in_message":[],"_revision_number":1},{"id":"583dcd00b4f649adc4695f64ff0bbcb30eed2a49","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-23 09:38:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6f01ac83588bccdf4533cdaf50436499ec0861cf","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-23 11:06:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cbecc0c4b73d6bc8af648c0aceccbdb82c9d73b9","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-23 11:07:38.000000000","message":"Patch Set 3:\n\nReverted my attempt, I still see --black-regex parameter added in https://review.opendev.org/#/c/703680 test run","accounts_in_message":[],"_revision_number":3},{"id":"81e939943a36d01cd3705961e1425bfb5e855bc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 13:13:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96b3df07b4e3431887653af11494382f : SUCCESS in 6m 55s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/360fb30a7b6648f6bb6924d0f0d20891 : SUCCESS in 49m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/66b5d1c8bf3f4ac98659b8f694395c7e : SUCCESS in 59m 02s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/24c111008ea6459095b1ee683b4e0888 : FAILURE in 1h 50m 32s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f484d1562f954be0a52a68a4ab4dcbb2 : SUCCESS in 1h 08m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ce7a182176c14dfe94610c550f637a69 : SUCCESS in 1h 02m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5384b89758fe4ef6a2cf669e393bb908 : SUCCESS in 1h 03m 58s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/bd3d2b23aed24af29fd720ec0dc9d297 : SUCCESS in 51m 21s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/63591753a161465b836e640cfb94dbdf : SUCCESS in 36m 01s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/8c0aba0dca3a4a36b6cdffae737e0c8d : SUCCESS in 1h 36m 48s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/4716a4f7d3604a8089a5cae4767cd502 : SUCCESS in 1h 00m 26s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/d225edcb46f645eba860dfd4201b4ae2 : SUCCESS in 1h 07m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/09666f7e5d964c7bb558d8afda217f0e : SUCCESS in 1h 18m 34s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/731c9dc8d0fa49809d93975dcc4700c0 : SUCCESS in 48m 58s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/8462021f9b9843c5b5eee30db7ac2f27 : SUCCESS in 45m 07s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/039ef9c667fb4db79b72c806d70b28ad : FAILURE in 1h 30m 11s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/1ad389fbc2fd418bac651b26f277e49e : SUCCESS in 1h 14m 23s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/ea112d7abd3349a89c4208be538e68f1 : SUCCESS in 1h 05m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/655ecfb1b89d4bef81c3886aaa8d12f1 : SUCCESS in 1h 36m 52s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/c48bc01aee684501adc6ee315070c30e : SUCCESS in 56m 00s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/dc2b35d32e3c43afb2ed0f8dbadcc9e0 : SUCCESS in 46m 36s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/92a8d8661ea64078827aa290eb6c9660 : FAILURE in 1h 31m 08s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/8f93185b639d4d0e880adb7d0efe37ca : SUCCESS in 1h 35m 05s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/944dadbf9e0a4ca29de0118b432a7540 : SUCCESS in 1h 28m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/100005705e5c4450817508acb9c91fdc : SUCCESS in 1h 21m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d0e279e30254b3d8369cd80f8de43b3 : SUCCESS in 5m 13s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/ab2f7c6e06ce4db0a5290d1cb703cbb6 : SUCCESS in 1h 55m 47s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/6bf17717004b4074b22902caf57a92db : SUCCESS in 1h 53m 05s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/dd8612a646b742a7bcc8003b6e19f7c0 : SUCCESS in 1h 35m 58s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/e9e2ecdf892941528cbb9e46252c568c : SUCCESS in 1h 34m 18s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/b6cb70d36d284e7fb50b54b76be38491 : SUCCESS in 41m 31s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/3d61668f6add4881b2f51463b0e9f398 : FAILURE in 42m 18s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/43aec22991d3441e9572910b9231e5d4 : SUCCESS in 54m 28s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/adf98f8505e54df7b5a8275699b66ef5 : SUCCESS in 43m 26s","accounts_in_message":[],"_revision_number":3},{"id":"ec27dbae2e6bce5aa6e5fb928026684e87bf3823","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-23 14:53:02.000000000","message":"Patch Set 3:\n\n\u003e Reverted my attempt, I still see --black-regex parameter added in\n \u003e https://review.opendev.org/#/c/703680 test run\n\nI think we are hitting other issue here and that should be fixed somewhere. let me find.\n\n tempest run: error: unrecognized arguments: --black-regex\u003d","accounts_in_message":[],"_revision_number":3},{"id":"bf54291dade130ed30633246de176c1f2927aaa4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-23 15:13:37.000000000","message":"Patch Set 3:\n\n\u003e \u003e Reverted my attempt, I still see --black-regex parameter added in\n \u003e \u003e https://review.opendev.org/#/c/703680 test run\n \u003e \n \u003e I think we are hitting other issue here and that should be fixed\n \u003e somewhere. let me find.\n \u003e \n \u003e tempest run: error: unrecognized arguments: --black-regex\u003d\n\nTrying with tag 19 which has the --black-regex available. - https://review.opendev.org/#/c/703679/","accounts_in_message":[],"_revision_number":3},{"id":"ce48bcb8daf15631fa5886f793caf106f326ee3b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-23 18:08:00.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c19b5ae4e21f25b6ddb726ab5ab418e3f6d6642a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-23 18:08:44.000000000","message":"Patch Set 4:\n\n\u003e \u003e \u003e Reverted my attempt, I still see --black-regex parameter added\n \u003e in\n \u003e \u003e \u003e https://review.opendev.org/#/c/703680 test run\n \u003e \u003e\n \u003e \u003e I think we are hitting other issue here and that should be fixed\n \u003e \u003e somewhere. let me find.\n \u003e \u003e\n \u003e \u003e tempest run: error: unrecognized arguments: --black-regex\u003d\n \u003e \n \u003e Trying with tag 19 which has the --black-regex available. -\n \u003e https://review.opendev.org/#/c/703679/\n\n--black-regex issue is fixed but neutron-tempest-plugin tag 0.7.0 is not compatible for stable/queens - https://review.opendev.org/#/c/704049/1","accounts_in_message":[],"_revision_number":4},{"id":"5c8538ead62ffd83e17287d125319951b6d51216","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 21:52:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/005e00b302834063a566fdae5caaf5c8 : SUCCESS in 3m 59s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/54bfcc19f812414daf9021bdc7ebaeb7 : SUCCESS in 1h 05m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2a4f29ce470d40088f3ed9b92859853e : SUCCESS in 36m 23s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/1f52cf59df2e41f584b166f3b2d5e361 : FAILURE in 1h 49m 04s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8692347d7c7341b48d8b7c1b2d5624d4 : SUCCESS in 1h 00m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/71cee22f9bca4591bb4b01b6b56f2e22 : SUCCESS in 57m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c6ec46e17b7345a8868bd34c0dc8856a : SUCCESS in 56m 45s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/9ecce0ee24f4475bae6490a169d0f676 : SUCCESS in 1h 03m 28s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/4afdf74952634ae5b0624d407d80685e : SUCCESS in 41m 12s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/0ef9deb82b05475bbd354e15b48cfac0 : SUCCESS in 1h 45m 25s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/d732f002b9e94ec6a0a4fb51f649ffa5 : SUCCESS in 1h 35m 48s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/b0266ae2b2fb41b2be71d5aeb9613a9c : SUCCESS in 1h 22m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/c54c1aa214134bd0b10de13fbbc111c0 : SUCCESS in 1h 04m 12s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/867a6a2377174f50bc617ec34269f33f : SUCCESS in 52m 59s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/64a294338179445a91ce704a320e623c : SUCCESS in 39m 15s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/e05072783e024d55a920c54ee0800a0d : SUCCESS in 1h 25m 46s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/d2d46fb3699c4163850e61cd3c02643b : SUCCESS in 1h 33m 15s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/22110a83e8534126926605facaf91d43 : SUCCESS in 1h 07m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/a450af3ca6f94564b4a58164cd1616b1 : SUCCESS in 1h 00m 47s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/4767528cd9e245e59ca269c636760dff : SUCCESS in 53m 20s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/696786d7e87946b1a59eac8c1e7363e0 : SUCCESS in 37m 41s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/41304b329944457bbe8c47025b1e9bb8 : SUCCESS in 1h 24m 00s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/8ba51dba0ff24b09bf7a4aab75a9d049 : SUCCESS in 1h 19m 12s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/c00d865764b048c98f2044f42f0a0210 : SUCCESS in 1h 07m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/c12319832da14923b0ea191fd7ed39a7 : SUCCESS in 1h 21m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66ed848a9fa240a88bc031baa9d153f5 : SUCCESS in 5m 51s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/5befd89947cf4383b3ac46860b928cef : SUCCESS in 1h 54m 55s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/2f78b0653e3f4071a79e94eb93956953 : SUCCESS in 2h 14m 22s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/b2819c9dcf4d4b9eb7e3956dd4bc0746 : SUCCESS in 1h 37m 57s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/55ca0b208e1e4b429fcec07ec07f1fcb : SUCCESS in 1h 31m 33s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/c050e0cc6c1f4a00b20959b32dc9b268 : SUCCESS in 49m 01s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/06bbc63f0aa54f2eada721e72bdf8419 : SUCCESS in 37m 14s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/2b66aea5de0c4fab90f885de8fdc331c : SUCCESS in 56m 57s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/96d9e1e98c4744a9be024c4ede745e9a : SUCCESS in 38m 00s","accounts_in_message":[],"_revision_number":4},{"id":"f8aff08755db79e2abb598fb8d21338cb671ff65","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-01-23 23:02:04.000000000","message":"Patch Set 4: Code-Review+2\n\n(3 comments)\n\nSome nits in the commit message in case you roll out another patch","accounts_in_message":[],"_revision_number":4},{"id":"f4fc4bc30d7789f00796519d8fcc7687130c7689","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-24 09:19:37.000000000","message":"Patch Set 4: Code-Review+1\n\nGetting close to passing tests with other patches, this one is good to go indeed","accounts_in_message":[],"_revision_number":4},{"id":"2fb31ef97b91118daa58a4b0bae2ca9c0f9bb403","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-24 14:12:57.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"fc915a078de277973e63305b7896407de4a56923","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-01-24 16:25:47.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(6 comments)\n\nNits in the commit message are not severe enough to hold this uo.","accounts_in_message":[],"_revision_number":5},{"id":"89e9bb49e3c57c402cbca03e23a6124a987ab1d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 16:43:10.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/282ef18da1ec4c0eb8be5ac15aaa579d : SUCCESS in 7m 41s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/61a38b9bb7d64cd1ac4e4864bf190961 : SUCCESS in 58m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/94b36635a886482c9a4fdeba6782fbac : SUCCESS in 36m 03s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/5ddb913826034921baef4fd9dde54903 : SUCCESS in 1h 43m 06s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ec2994b7ce0445e4ad5b7c30c84de1d5 : SUCCESS in 1h 18m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4ecbb1759cd7473fa57956964cdde601 : SUCCESS in 1h 09m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d3379349c6934dfc8cb3ed58bd991b64 : SUCCESS in 1h 06m 49s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/cd51167575194b4a9e775fb575261c7c : SUCCESS in 1h 00m 44s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/84d10766df8e4401928e9724a2872357 : SUCCESS in 34m 09s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/3570d4f25f6846fab75c418cb18a9577 : FAILURE in 1h 42m 26s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/0d47eebccc5f452cb2d6c4dde7e3911a : SUCCESS in 1h 01m 52s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/42905b97a2704905b1a10130a4b44c09 : SUCCESS in 1h 17m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/149b4ee0bfd84e3a90a3b43219fe7cf3 : FAILURE in 1h 27m 19s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/bd4b22cde4884a2bb40a573eae91ffda : SUCCESS in 59m 48s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/92d71e2e8857426e81514e11d779890c : SUCCESS in 37m 43s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/5b583fc806d84a39bcb548f5eabe943b : FAILURE in 1h 31m 39s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/d4bcd88e720341a7988be260c5fd745a : SUCCESS in 1h 09m 52s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/9a364c2a7fec46eca1aa55f52e606fd1 : SUCCESS in 1h 15m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/ec50619255ef4381984f06ccfff4c8de : FAILURE in 26m 30s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/21e736580daa476b908972dbb89cb074 : SUCCESS in 51m 42s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/a9d2151f3f8f497282165eb44c82a42b : SUCCESS in 44m 30s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/2fdf9f32bb33454eab17ae8ac0ee5e4f : SUCCESS in 1h 31m 12s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/ffde42a0a61e4d54a5cedc00e689e75a : SUCCESS in 1h 03m 34s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/022129796a534f8a9e40aea996ebfb92 : SUCCESS in 1h 26m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/f21d8084114c4f738dc623892e0798d6 : SUCCESS in 1h 07m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad66c65aee40469192a01bc3718e1ee6 : SUCCESS in 4m 30s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/f625c126b5804f659239fe6bba179a12 : SUCCESS in 1h 53m 51s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/748ec3f4d8fc44a9b2905a726a1a42d7 : SUCCESS in 1h 51m 44s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/fa72374fdcc84672a5fcd1fe87ee266e : SUCCESS in 1h 35m 49s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/0589a8f471fc4344b13138ad528fc151 : SUCCESS in 1h 36m 24s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/2018802f6eaa41f183ee8ecb79fedb9b : SUCCESS in 41m 31s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/46c893756f444c1ca8880b46b10dcae3 : FAILURE in 45m 16s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/cd4f9cdebdb848d99ac599b70850f6ae : SUCCESS in 1h 11m 59s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/e9cc9a1a86d849bbb30df85881c9583a : FAILURE in 59m 26s","accounts_in_message":[],"_revision_number":5},{"id":"9ae58df95f2d33c962a949483d53045cc0d61256","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-25 07:32:33.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"7da2f83c2b977928e1e39d189129e0136267ae14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 10:12:07.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26285b19f5ca440ea2cb67d4f81d2f11 : SUCCESS in 4m 50s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2fa5e743ce394f5695504955940e48de : FAILURE in 1h 01m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/13a5a5c46abc4e6abedcec78e0c8856b : SUCCESS in 37m 25s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/a38456ca733d4792a7862221b251d7e6 : FAILURE in 1h 37m 37s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/48bdf7def8274641bc2f2d0cd0a83109 : SUCCESS in 1h 10m 16s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/805d40d29e3a4694aac09618672d9831 : SUCCESS in 1h 10m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/865855a1e143499fbe5451651cda34e6 : SUCCESS in 1h 02m 53s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/6c8854e0f9a741a2b478069840a87e6a : SUCCESS in 1h 03m 10s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/b84d2e40e7574cb8895a6c3c5133ba64 : SUCCESS in 36m 53s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/a084f02b4e9f4b84b67068ccd585ad1c : SUCCESS in 1h 40m 34s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/7fa05121463f4fdbb40460fca43fd986 : SUCCESS in 1h 00m 06s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/0bda0e4bc9fb48e0a5b64cc6cf65c170 : SUCCESS in 58m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/b210459491914b91bddbd9f275d5fe6a : FAILURE in 1h 01m 14s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/a24759dc7e954932b00af6b963412872 : SUCCESS in 51m 37s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/23d5aa8c7d2346e2a6a1af3442c704b1 : TIMED_OUT in 1h 06m 45s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/812405efc1a549eb8f140af344559299 : FAILURE in 1h 38m 10s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/259aa77821334e95b6efba342633cb88 : SUCCESS in 1h 07m 15s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/d35e2af48c6649f99fec85c39e41ca91 : SUCCESS in 1h 03m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/65c9a469afb24a248e6eba719f6ae092 : SUCCESS in 57m 18s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/f58374160f2e4c1eb16d6423a5ad74f6 : FAILURE in 1h 03m 30s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/5c56f0bb605d4eb9b4a24d782c71b9b1 : SUCCESS in 39m 00s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/35b39bdb04244393b96649edbb50f123 : FAILURE in 1h 26m 21s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/a8c2eadb05ab4339a5703dee84c29704 : SUCCESS in 1h 11m 58s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/7c7d8055b1f24387978af75b36255678 : SUCCESS in 1h 10m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/69e0b316cf3246a0988aaf49c96f0c4d : FAILURE in 1h 06m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2f738b54bd24387b886e80b4af8cf16 : SUCCESS in 4m 42s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/922977dbc5084bce97d3f724c508230c : SUCCESS in 1h 58m 30s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/379122f6ab274a9898ff9512fe29ace2 : FAILURE in 2h 33m 49s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/5a30cba85ef64547afbc9b836ac23120 : SUCCESS in 1h 32m 28s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/d42bd828d13b474faad89227b547eeb1 : TIMED_OUT in 2h 05m 40s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/53b1a941f1514c4f8d311947f124ebb8 : SUCCESS in 39m 11s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/7d6ca44ad7f44f239dbd676e0174b924 : SUCCESS in 46m 38s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/f89c1209b41a4741a6aa7fa6d9845ce9 : SUCCESS in 1h 37m 50s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/2c355e18cab141da9332ba35213c4701 : FAILURE in 47m 04s","accounts_in_message":[],"_revision_number":5},{"id":"24bb480dbdcc708655f8f2c974c91f7cf1e3e51d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-25 10:25:09.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"60b235144be4eeb20f23c1b445a6bbeec77fe8a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 12:03:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c38cbea382304db2b1937f70a7a92c53 : SUCCESS in 3m 54s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f61a433587274b3dab5a2f34a130e927 : SUCCESS in 48m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/00a8d98f8b1d42ea86e00eed78a00c98 : SUCCESS in 33m 12s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/4f56ece4a1a4473299332f0000d65c43 : FAILURE in 1h 21m 59s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d1ff56eaef734b399959c0fd8b63bfe4 : SUCCESS in 1h 08m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6a0f7de0edcc4d91ba96cac96ebb34da : SUCCESS in 1h 07m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6b6d17a018264f7dacaac294e9d6fca1 : SUCCESS in 55m 14s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/3986f45d2fd249b7b89adece180bc658 : SUCCESS in 49m 05s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/6127d83189bd4eb684806568302006d7 : SUCCESS in 34m 55s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/9cb5706f8abe4d42985335c6c0d57b0e : SUCCESS in 1h 31m 59s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/384eb49fafba4bba913b63ba17b4f8d5 : SUCCESS in 57m 11s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/5a8d0b959a504efcbe8aae547d7df18f : SUCCESS in 1h 02m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/c78c6f8235cb4fffa80c2699a473f5a5 : SUCCESS in 59m 46s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/c741c388a8a14a9081c96b6ff0dfe0dd : SUCCESS in 47m 44s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/5439ee5dddba45c39b6bbb2296c6bd4d : SUCCESS in 36m 55s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/9d7d06c0b00e41878be31d117549b5b5 : SUCCESS in 1h 18m 17s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/468420a3ee79493386c153fc7ed5bb11 : SUCCESS in 1h 00m 03s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/8fb8201fb7194bf795aa0c87c93fefdf : SUCCESS in 1h 09m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/4334f0dc1af54fb6b77c7667367611ba : SUCCESS in 56m 26s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/4a3e5e54ea764cb4a82f3ca385a51114 : SUCCESS in 50m 22s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/40e84e97ee4f4a63b8ac640d81012a00 : SUCCESS in 37m 15s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/8c2385b52a87424aa1d04d4591b5321d : SUCCESS in 1h 29m 34s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/89f7f543416340b3921cfeebc38613e5 : SUCCESS in 1h 09m 05s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/afd0158edff24145a2283e049c0f15cf : SUCCESS in 1h 01m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/2946c86097c0446e92a4ff4439e9bebe : SUCCESS in 1h 01m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77d8c0810b8f4ae7b6e323dc44faab94 : SUCCESS in 3m 54s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/28a4a52ae4b8432dbd5a6274dc1762c8 : SUCCESS in 1h 36m 51s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/b0ee2063bb7446568d252e8f80b1be14 : SUCCESS in 1h 36m 54s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/998a0eb61c5c4b17b23625363a439ff4 : SUCCESS in 1h 24m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/32b773c30c6e46d2a70e26dab98dac33 : SUCCESS in 1h 25m 17s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/40b2fdc106c54d5daae0d026a8479e32 : SUCCESS in 33m 15s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/d02975a0d70b44b5887fef2915379ce5 : SUCCESS in 35m 22s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/a26290d267ff4ffaae8ae36dcedcbf9d : SUCCESS in 44m 12s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/157184c8da4e4507b198e2e95f4fb741 : SUCCESS in 36m 00s","accounts_in_message":[],"_revision_number":5},{"id":"5eda44c0710a61481c4e0af7021ae07fdb37c921","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 12:03:38.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0ac455ec76630054d241de85e05662264d3b214f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 13:42:03.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9584be73bb694a0fb9293ae1ca96e05f : SUCCESS in 3m 45s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9e50463650474070b5bcb392ed57d640 : SUCCESS in 53m 14s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/61094db1619941bcb7704818c74043d1 : SUCCESS in 52m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b2cb6d6998374374abf3d8e6612f041c : SUCCESS in 53m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e9beb53f66604c668c2c341b916bd9da : SUCCESS in 1h 01m 02s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/53bc4e6a4e094857bef1dc675e8e0b36 : SUCCESS in 47m 13s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/b02e3042a46d4e4e9b2b242b0f486953 : SUCCESS in 50m 13s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/f7ae6ad2703748499b40c2b23bfbd245 : SUCCESS in 48m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4807b4340b444413b63a6cf79497a088 : SUCCESS in 3m 57s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/deed9c211cca4fb8b5dacfab0c41f3b7 : SUCCESS in 1h 37m 24s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7e49f14801df4c48acbbc2ef4bfc2533 : SUCCESS in 1h 27m 43s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/95b012f1bc9e4e19b6d7aa566ed6ac06 : SUCCESS in 45m 58s","accounts_in_message":[],"_revision_number":5},{"id":"4809053c983cc93287c11523ae17ac4fef35bbb6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 13:42:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0327bc227b7aa15f15328b32309fecd506a27830","revisions":{"db08df13b7ee2ac6a67f7f978f6d84dc777432f0":{"kind":"REWORK","_number":1,"created":"2020-01-23 00:35:14.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/03/703903/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/03/703903/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/703903/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/703903/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/703903/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/703903/1"}}},"commit":{"parents":[{"commit":"560737404a9d00879c1f2da4f523bb1a03599164","subject":"Merge \"Revert \"Switch neutron-dynamic-routing job to be non-voting temporaray\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/560737404a9d00879c1f2da4f523bb1a03599164"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-23 00:23:15.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-23 00:35:12.000000000","tz":-360},"subject":"Remove TEMPEST_BRANCH var from neutron queens job.","message":"Remove TEMPEST_BRANCH var from neutron queens job.\n\nIn devstack install_Tempest, ref is taken by TEMPEST_BRANCH var[1]. TEMPEST_BRANCH var on zuul job\ndefinitoin is priority than the default value in devstack[2]. When we set the queens compatible Tempest\ntag[1] (with review style not by tag name) on devstack stable/queens, that is not taken by devstack\nbecause neutron queens job define the smae with different value[4].\n\nAlso Tag name (for example queens-em) never work because of devstack logic. Tempest tag name only work\nif RECLONE is true otherwise it should be gerrit ref style. So always Tempest master is installed\non queens jobs.\n\nLet\u0027s remove the TEMPEST_BRANCH from neutron jobs and rely on devstack deafult value.\n\nCLoses-Bug: 1859988\n[1] https://github.com/openstack/devstack/blob/stable/queens/lib/tempest#L658\n[2] https://github.com/openstack/devstack/blob/786c485cf6d664163b4ee3d390c62693a9e0d036/stackrc#L313\n[3] https://review.opendev.org/#/c/703679/\n[4] https://zuul.opendev.org/t/openstack/build/0cc32b278afe4e51b787d87d480f7092/log/job-output.txt#12370\n\nChange-Id: I07beb6e7bdcbaf7ffe72d47f9edb90df40641e68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/db08df13b7ee2ac6a67f7f978f6d84dc777432f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/db08df13b7ee2ac6a67f7f978f6d84dc777432f0"}]},"branch":"refs/heads/master"},"a76d9fef67f1f7ada8281c8281e1232623567d4d":{"kind":"REWORK","_number":2,"created":"2020-01-23 09:38:36.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/03/703903/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/03/703903/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/703903/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/703903/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/703903/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/703903/2"}}},"commit":{"parents":[{"commit":"560737404a9d00879c1f2da4f523bb1a03599164","subject":"Merge \"Revert \"Switch neutron-dynamic-routing job to be non-voting temporaray\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/560737404a9d00879c1f2da4f523bb1a03599164"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-23 00:23:15.000000000","tz":-360},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-01-23 09:37:01.000000000","tz":60},"subject":"Remove TEMPEST_BRANCH var from neutron queens job.","message":"Remove TEMPEST_BRANCH var from neutron queens job.\n\nIn devstack install_Tempest, ref is taken by TEMPEST_BRANCH var[1].\nTEMPEST_BRANCH var on zuul job definition is priority than the default\nvalue in devstack[2]. When we set the queens compatible Tempest tag[1]\n(with review style not by tag name) on devstack stable/queens, that is\nnot taken by devstack because neutron queens job define the smae with\ndifferent value[4].\n\nAlso Tag name (for example queens-em) never work because of devstack\nlogic. Tempest tag name only work if RECLONE is true otherwise it should\nbe gerrit ref style. So always Tempest master is installed on queens\njobs.\n\nLet\u0027s remove the TEMPEST_BRANCH from neutron jobs and rely on devstack\ndeafult value.\n\nTemporarily comment tempest_black_regex definition in queens jobs to see\nif this is the cause of errors with pinned tempest\n\nCLoses-Bug: 1859988\n[1] https://github.com/openstack/devstack/blob/stable/queens/lib/tempest#L658\n[2] https://github.com/openstack/devstack/blob/786c485cf6d664163b4ee3d390c62693a9e0d036/stackrc#L313\n[3] https://review.opendev.org/#/c/703679/\n[4] https://zuul.opendev.org/t/openstack/build/0cc32b278afe4e51b787d87d480f7092/log/job-output.txt#12370\n\nChange-Id: I07beb6e7bdcbaf7ffe72d47f9edb90df40641e68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a76d9fef67f1f7ada8281c8281e1232623567d4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a76d9fef67f1f7ada8281c8281e1232623567d4d"}]},"branch":"refs/heads/master"},"cf35a68f49b0a0b4554e3b494a83a998d9792051":{"kind":"REWORK","_number":3,"created":"2020-01-23 11:06:59.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/03/703903/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/03/703903/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/703903/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/703903/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/703903/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/703903/3"}}},"commit":{"parents":[{"commit":"560737404a9d00879c1f2da4f523bb1a03599164","subject":"Merge \"Revert \"Switch neutron-dynamic-routing job to be non-voting temporaray\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/560737404a9d00879c1f2da4f523bb1a03599164"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-23 00:23:15.000000000","tz":-360},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-01-23 11:06:46.000000000","tz":60},"subject":"Remove TEMPEST_BRANCH var from neutron queens job.","message":"Remove TEMPEST_BRANCH var from neutron queens job.\n\nIn devstack install_Tempest, ref is taken by TEMPEST_BRANCH var[1]. TEMPEST_BRANCH var on zuul job\ndefinitoin is priority than the default value in devstack[2]. When we set the queens compatible Tempest\ntag[1] (with review style not by tag name) on devstack stable/queens, that is not taken by devstack\nbecause neutron queens job define the smae with different value[4].\n\nAlso Tag name (for example queens-em) never work because of devstack logic. Tempest tag name only work\nif RECLONE is true otherwise it should be gerrit ref style. So always Tempest master is installed\non queens jobs.\n\nLet\u0027s remove the TEMPEST_BRANCH from neutron jobs and rely on devstack deafult value.\n\nCLoses-Bug: 1859988\n[1] https://github.com/openstack/devstack/blob/stable/queens/lib/tempest#L658\n[2] https://github.com/openstack/devstack/blob/786c485cf6d664163b4ee3d390c62693a9e0d036/stackrc#L313\n[3] https://review.opendev.org/#/c/703679/\n[4] https://zuul.opendev.org/t/openstack/build/0cc32b278afe4e51b787d87d480f7092/log/job-output.txt#12370\n\nChange-Id: I07beb6e7bdcbaf7ffe72d47f9edb90df40641e68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/cf35a68f49b0a0b4554e3b494a83a998d9792051"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/cf35a68f49b0a0b4554e3b494a83a998d9792051"}]},"branch":"refs/heads/master"},"1c2661e3aae33d3a4cce08de2772523899e274d7":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-01-23 18:08:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/03/703903/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/03/703903/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/703903/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/703903/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/703903/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/703903/4"}}},"commit":{"parents":[{"commit":"9ed4488161675dc39887bc47e4646526744a40ba","subject":"Fix neutron-tempest-plugin tag for queens jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9ed4488161675dc39887bc47e4646526744a40ba"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-23 00:23:15.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-01-23 18:08:00.000000000","tz":0},"subject":"Remove TEMPEST_BRANCH var from neutron queens job.","message":"Remove TEMPEST_BRANCH var from neutron queens job.\n\nIn devstack install_Tempest, ref is taken by TEMPEST_BRANCH var[1]. TEMPEST_BRANCH var on zuul job\ndefinitoin is priority than the default value in devstack[2]. When we set the queens compatible Tempest\ntag[1] (with review style not by tag name) on devstack stable/queens, that is not taken by devstack\nbecause neutron queens job define the smae with different value[4].\n\nAlso Tag name (for example queens-em) never work because of devstack logic. Tempest tag name only work\nif RECLONE is true otherwise it should be gerrit ref style. So always Tempest master is installed\non queens jobs.\n\nLet\u0027s remove the TEMPEST_BRANCH from neutron jobs and rely on devstack deafult value.\n\nCLoses-Bug: 1859988\n[1] https://github.com/openstack/devstack/blob/stable/queens/lib/tempest#L658\n[2] https://github.com/openstack/devstack/blob/786c485cf6d664163b4ee3d390c62693a9e0d036/stackrc#L313\n[3] https://review.opendev.org/#/c/703679/\n[4] https://zuul.opendev.org/t/openstack/build/0cc32b278afe4e51b787d87d480f7092/log/job-output.txt#12370\n\nChange-Id: I07beb6e7bdcbaf7ffe72d47f9edb90df40641e68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1c2661e3aae33d3a4cce08de2772523899e274d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1c2661e3aae33d3a4cce08de2772523899e274d7"}]},"branch":"refs/heads/master"},"0327bc227b7aa15f15328b32309fecd506a27830":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-01-24 14:12:57.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/03/703903/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/03/703903/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/703903/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/703903/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/703903/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/703903/5"}}},"commit":{"parents":[{"commit":"1b63b3d7aa5b3b446c62cf9137106ba4d9230e7a","subject":"Fix neutron-tempest-plugin tag for queens jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1b63b3d7aa5b3b446c62cf9137106ba4d9230e7a"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-23 00:23:15.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-01-24 14:12:57.000000000","tz":0},"subject":"Remove TEMPEST_BRANCH var from neutron queens job.","message":"Remove TEMPEST_BRANCH var from neutron queens job.\n\nIn devstack install_Tempest, ref is taken by TEMPEST_BRANCH var[1]. TEMPEST_BRANCH var on zuul job\ndefinitoin is priority than the default value in devstack[2]. When we set the queens compatible Tempest\ntag[1] (with review style not by tag name) on devstack stable/queens, that is not taken by devstack\nbecause neutron queens job define the smae with different value[4].\n\nAlso Tag name (for example queens-em) never work because of devstack logic. Tempest tag name only work\nif RECLONE is true otherwise it should be gerrit ref style. So always Tempest master is installed\non queens jobs.\n\nLet\u0027s remove the TEMPEST_BRANCH from neutron jobs and rely on devstack deafult value.\n\nCLoses-Bug: 1859988\n[1] https://github.com/openstack/devstack/blob/stable/queens/lib/tempest#L658\n[2] https://github.com/openstack/devstack/blob/786c485cf6d664163b4ee3d390c62693a9e0d036/stackrc#L313\n[3] https://review.opendev.org/#/c/703679/\n[4] https://zuul.opendev.org/t/openstack/build/0cc32b278afe4e51b787d87d480f7092/log/job-output.txt#12370\n\nChange-Id: I07beb6e7bdcbaf7ffe72d47f9edb90df40641e68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0327bc227b7aa15f15328b32309fecd506a27830"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0327bc227b7aa15f15328b32309fecd506a27830"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
