)]}'
{"id":"openstack%2Fneutron-tempest-plugin~703909","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I8912b1a123ac52c99b65b93884702e792fa7ea86","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"neutron-api-ext","hashtags":[],"change_id":"I8912b1a123ac52c99b65b93884702e792fa7ea86","subject":"Optimize the way to define the api_extensions","status":"ABANDONED","created":"2020-01-23 02:17:45.000000000","updated":"2020-09-25 15:36:54.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e004a76d3e0511394433628c815efb14d84fd547","_number":703909,"virtual_id_number":703909,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"date":"2020-09-25 15:36:54.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-08 22:54:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-24 15:28:14.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-05-08 22:54:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-25 15:36:54.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"42b6feab2698ed31462eb552ebfcd1070e1b50f1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-23 02:17:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b90948db908d508c6de62f5c02c93dffebe1726c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-23 02:18:43.000000000","message":"Patch Set 1:\n\nif this work properly then I will do for other stable branch jobs too.","accounts_in_message":[],"_revision_number":1},{"id":"16571b5e53e7be414838707b6d37b383890c5fe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 04:15:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9506c03d2d28443496563043bcd23ebd : SUCCESS in 4m 02s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6bdef2739a874beeb4e10f35bd77c580 : SUCCESS in 46m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c95623f3cbe54d7b94d73acddebd1bed : SUCCESS in 36m 04s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/4aed20a7438f4cd08bd034ce93fccb59 : FAILURE in 1h 32m 56s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/69f37fd5ea1f48298d518655bc2ac724 : SUCCESS in 1h 12m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0a7c30913b1a458f98a42c3e7bfca0d5 : SUCCESS in 1h 07m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/288913a75a4a4ab285adf0bf56209722 : SUCCESS in 54m 18s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/0ecefa970a184e47a434d08ed2f44450 : SUCCESS in 56m 10s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/d9618bbdc3ed4a20b01212f8bbd62e74 : SUCCESS in 34m 41s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/7882d7b245b3472383e11f7b6f7a2db3 : SUCCESS in 1h 35m 35s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/5d4671ff4d3c4aecbd8c9d24e564f514 : SUCCESS in 1h 05m 43s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/d60ae8b20d924835ba2d392034d6f43d : SUCCESS in 1h 09m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/e99f8a0c61a34cb898425015dc4194d9 : SUCCESS in 1h 00m 57s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/a9cd9b241260407e86cbd930bec48043 : SUCCESS in 56m 46s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/c1dae133850e461bb561a548de894dcc : SUCCESS in 34m 15s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/5be5885979d04a49a673820a3ce08b90 : FAILURE in 1h 35m 10s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/4dcb85b521fc42f0a127001204e1e0e2 : SUCCESS in 1h 12m 28s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/ffacd5b3bfa44214bddbc3302bb175a0 : SUCCESS in 1h 03m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/760ef8383a604cee9300ef502cbcab1b : SUCCESS in 55m 34s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/8914b4903f894bd5b45ee0a6f577e935 : FAILURE in 49m 51s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/60fbca93839d4b4abd15b480b0239f94 : FAILURE in 33m 55s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/6b3c7730d86d4f1cb1d43e357469382e : SUCCESS in 1h 36m 18s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/c5b9cbe3da8040e5a85ae1f7b9e1cdc1 : SUCCESS in 1h 01m 21s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/f4f5a0e1c6a1470cbc6aa3e6509a820b : SUCCESS in 1h 04m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/f472ad6a4def4ac995a577cac4388ef4 : SUCCESS in 54m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47e160aa6ed248f5947069bc4df4f0a4 : SUCCESS in 3m 35s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/15155769232d482b92fbd6a0fea798e1 : SUCCESS in 1h 36m 06s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/d239e29059b947098b68811c61e348c2 : SUCCESS in 1h 52m 01s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a6d809f64e8b4a7db81f34b90c0fe916 : SUCCESS in 1h 18m 38s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/6946843b243f4fd793aaab8145fa39ba : SUCCESS in 1h 32m 08s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/0aa2b716ebc74900b1dc42a51364ad32 : SUCCESS in 36m 19s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/655da9b3d8614c0584662ee1e3539af9 : SUCCESS in 43m 37s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/d2af27363af44e938ed0c2079160f7b8 : SUCCESS in 46m 58s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/ce8b17a16764433db65b8ea97e1219c1 : SUCCESS in 42m 58s","accounts_in_message":[],"_revision_number":1},{"id":"9e782b4a902a5ab9f3acf2fe03bf39ba704327fb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-23 17:00:37.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"da34bc693a3ccf33b71af41e875d4511f8d217d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 20:54:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d910b652aada48aaa8bb2096e86b97e6 : SUCCESS in 6m 13s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b7d6b693b2b847aca1e2817a6b0463d7 : SUCCESS in 1h 09m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/876f84b0cd3e4f6bbc76b413b14142c7 : SUCCESS in 34m 39s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/612b63929fcc4d799d21b0988bf62584 : FAILURE in 1h 46m 39s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b83b0bc16d224fc8b2b89a826aab508d : SUCCESS in 1h 12m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3678657a1081401994d282401f8b04b9 : SUCCESS in 56m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9dfbf53b84c446528913cd2efc3068f6 : SUCCESS in 56m 53s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/dbcfb180c2194e338630a9ead9f5df0a : SUCCESS in 56m 45s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/d6ef4e57ab664072bff38bc0cfe76875 : SUCCESS in 31m 20s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/c130b72219824642bf860c5c13a97129 : FAILURE in 2h 00m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/174a21e78134499fad525991ab5fa8d0 : SUCCESS in 1h 14m 49s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/36931b3831d3470c86a0883de647c043 : SUCCESS in 1h 03m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/92006158e9f54e9f866e21a24600b1d2 : SUCCESS in 58m 13s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/a63c8aaa31e045e49c983e14290faf74 : SUCCESS in 52m 45s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/30faea13d9a445ca900b546d2e765a49 : SUCCESS in 40m 22s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/be07a589983c44e6b6c35c318a11da85 : SUCCESS in 1h 52m 01s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/9e5cc0f72b06409b92fb1e79e7b1144d : SUCCESS in 1h 03m 09s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/e24fe6c90ef04687a771819805109277 : SUCCESS in 1h 17m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/6cd3aa3fc738454aaca590979fc9378d : SUCCESS in 1h 15m 10s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/88c5f277320642b0a8efea48c44a296c : FAILURE in 51m 08s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/269d8366cda54122a895d3cd626ba956 : FAILURE in 31m 48s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/0e30d475670e42e1a32d8c4cec4bde2d : FAILURE in 1h 34m 44s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/521a71e6a41844bf96ada7615d5a988c : SUCCESS in 1h 31m 02s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/b71a60848eb1411c91b8f94002a0df45 : SUCCESS in 58m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/35ecf87fb9c14ae3bb1e308bd3adf3d5 : SUCCESS in 1h 08m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47fa327995e0445aaeccc83ead4ab5f7 : SUCCESS in 4m 02s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/ffef21b044b14ec2afebc4d0c6d85fea : SUCCESS in 1h 43m 28s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/32345b00557c4c5e9b2cd6b3610b08a4 : SUCCESS in 1h 48m 01s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9b9acf69662e4441a1e8002e0150c70e : SUCCESS in 1h 39m 12s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/0836ec4b1e65428b878003e522b37d12 : SUCCESS in 1h 23m 13s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/db1dc8cc451f4a38b187af20b75b668d : FAILURE in 37m 39s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/4a969d98c05047b29d6b44115cc1c34a : SUCCESS in 39m 15s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/838b0cad1afb4d3e82d77248da28f693 : SUCCESS in 49m 05s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/62227fecfb164c40b3b7e303d75aa845 : SUCCESS in 40m 17s","accounts_in_message":[],"_revision_number":2},{"id":"af1fc4b39f53f0b83079648c2ce60add38619217","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-24 15:15:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ee424cbe5bb7a66b84f3058ecf6240e047344f56","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-24 15:16:06.000000000","message":"Patch Set 2:\n\nthis is merged now - https://review.opendev.org/#/c/687176/1","accounts_in_message":[],"_revision_number":2},{"id":"cb6043f44e1fbc476038624df44110a55463130b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-24 15:16:24.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"95c0868181d6d2a007104fd6c4eecb7c34eb2116","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-24 15:28:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bb6d83f8217cc3fc062f84e38cb5692da306ed28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 19:06:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f17e5b6f5ec4df1a6681557f5de801a : SUCCESS in 4m 45s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/41864bb187af46e195775109d773a30e : SUCCESS in 51m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7f06240cebef472986a4876f22927811 : SUCCESS in 38m 06s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/7441151ff7cb4a1594a34d92ac2bc020 : FAILURE in 1h 28m 43s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8eb202feaf1342c3955f007e29443130 : SUCCESS in 1h 03m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0084ee108ba248eeaf877a938526dc9a : SUCCESS in 1h 10m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7acf8fcffbc14f25a1d78766d7a8b4d0 : SUCCESS in 1h 01m 53s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/8207a2ce8f3b4b55a6ec9745f9d6848b : SUCCESS in 1h 02m 50s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/a8aa70e497a4457cb6eb3ecf1f7a1e5e : SUCCESS in 40m 46s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/cc7c0e62b96842848603d3a40d889965 : FAILURE in 1h 41m 26s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/0271fb5d2eb2476abc95479d6fc51bcc : SUCCESS in 1h 10m 51s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/a3e3c19949b14f1eb6b050956b276b8a : SUCCESS in 1h 13m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/cfe58c551eae4ccbbce5849cb350c8f6 : SUCCESS in 1h 10m 38s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/799fd09e267d49d7a429a040fd0f9307 : SUCCESS in 1h 03m 37s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/f7746dbca90e418882ea30aea294071e : SUCCESS in 34m 49s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/bab1b9dc42dd4e6f8dae9cab1460d8db : SUCCESS in 1h 32m 25s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/ea4f370c336a46f4abb2aa3db8ca50ee : SUCCESS in 1h 12m 59s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/4e2ed67ac52b4acdae5c437f4ab2d24f : SUCCESS in 56m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/dcea92649e974aa08ce3c4ddb10bb4fe : SUCCESS in 1h 02m 59s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/0c3ad612bf324be9a862711a6084802d : FAILURE in 53m 02s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/fe0baee925854ad8aad816b77b38138d : FAILURE in 43m 43s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/e95982df3fb64bfcb1eccc369ee3e88d : SUCCESS in 1h 50m 01s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/a82689cede984696b2cd9ed2131c0266 : SUCCESS in 1h 21m 08s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/346abc88d9c74664ae5962315cc38a47 : SUCCESS in 1h 04m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/49c44f7a57b44c37b39f4085bf538047 : SUCCESS in 1h 00m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3d62395ff944677b715d7001618c97f : SUCCESS in 5m 16s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/40b2ddf0254c447ea220187f4e476256 : SUCCESS in 1h 40m 04s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/7e675f19deeb42c49db8561c363cb124 : SUCCESS in 1h 51m 35s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/3c57e97b8b124bf189fdd4da39600650 : SUCCESS in 1h 41m 07s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/dd44cb2ed2444bb0b81b9025ef79d4c0 : SUCCESS in 1h 59m 51s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/4832f0d79706404088caab590da9cf1b : SUCCESS in 35m 25s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/ba5b6b7f3b534102b9ba369711aa2e32 : SUCCESS in 37m 12s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/6558b3b5c2f549db9052d0d59c1033a5 : SUCCESS in 55m 57s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/a5463a3b246348dcb9c8e663b0c31a9c : FAILURE in 47m 05s","accounts_in_message":[],"_revision_number":3},{"id":"a74b7abc5c0961c837a1f3a4fe4a96da6d92db8c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-27 14:15:48.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)\n\ndevstack bakport to all stable branches are merged not, let me do it for other branches also.","accounts_in_message":[],"_revision_number":3},{"id":"6eb6379e9deed80aca824fdcab8918e65e38a373","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-06 00:53:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d3051d116802cb833f8e6fa787c124fc0c4a40cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 03:19:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2d34ed2eb334192824d74fd22498f8f : SUCCESS in 4m 00s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/957bd364fd5e4df196b81567fcae4783 : SUCCESS in 45m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1d5f4bcad03347c4ad393962212ae95d : SUCCESS in 34m 33s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/53c0a7624adb4c698c2ab32a0b8d2f3e : FAILURE in 1h 52m 46s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/38d23a27c0844e099db4a894b1db33e4 : SUCCESS in 1h 25m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fe9262273c04416fb2e5afa38853929d : SUCCESS in 1h 11m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/852982f2d55149c0ad24288cf7e32998 : SUCCESS in 1h 10m 05s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/74845d7aa7e6490780b87be2fbc9db98 : SUCCESS in 53m 48s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/b15266f267dd4eb8be970ab30b298a10 : SUCCESS in 32m 07s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/c4f8eab1050d459b8a045e6bf385f0a1 : FAILURE in 2h 04m 10s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/f81c2a69364b49948455d857facbb65c : SUCCESS in 2h 05m 50s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/9b784d8f3d794d41acd2d85364ad83a5 : SUCCESS in 1h 28m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/5ca975595f2c4c7f9ba1abae3479229e : SUCCESS in 1h 11m 24s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/7e5db8bbaf03438fa5d9d9a46b3a9088 : FAILURE in 1h 03m 53s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/276583ff904649d9ad1db623d3044684 : FAILURE in 36m 57s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/65da9bdada79470b881c6987c78a1d4f : FAILURE in 1h 53m 44s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/f9e58fca910b465e85b52185f6f7e7ac : SUCCESS in 1h 34m 59s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/af00fa9d067a4bb79c205dc097d796fe : SUCCESS in 1h 17m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/d06d39ed25c6486db37be5439fbe7b4d : SUCCESS in 1h 55m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/360dae413be14511816a902741b9a7f6 : SUCCESS in 4m 30s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/337c65103d3b495c9e5e13fd107d8b0a : SUCCESS in 1h 38m 14s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/24f4c7ceb5c248f6a0aab414a356d360 : SUCCESS in 1h 40m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/59040249f06c4de38423553970bc579b : SUCCESS in 1h 22m 33s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/e743ede16e154370a1573a1c00fad333 : SUCCESS in 1h 27m 47s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/9a0d8ff64ae844278310e998b2ee2034 : SUCCESS in 36m 58s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/349c75c2eafc402896e8a20374100414 : SUCCESS in 36m 00s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/3d9e35932b244b08af4cf75f9c6a463d : SUCCESS in 50m 07s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/a38093d513f840af9010a19eb5fafc26 : SUCCESS in 37m 12s","accounts_in_message":[],"_revision_number":4},{"id":"cd25e333202f69f7781227af0fd8ab09d41b82ac","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-06 10:20:54.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"999cc06114bdc4ea0fdbf14af55226f44845c7c1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-08 17:56:55.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2cd566bd0b2c8af12f839d1b4fb3acb0903a08d2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-08 20:00:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6a0e6e9240add2840e47eae1e4c75b6a1eb17c53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 20:00:32.000000000","message":"Patch Set 5: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron-tempest-plugin on branch master.  The error was:\n\n  found undefined alias \u0027api_extensions_master\u0027\n  in \"openstack/neutron-tempest-plugin/.zuul.yaml@master\", line 277, column 31","accounts_in_message":[],"_revision_number":5},{"id":"e61cde4371c6ffabebeaa034cb91ce7d0f5a0934","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-08 20:40:40.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5785dd5dcae4ecd4e2b6712312df20eea72ba734","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-08 20:43:00.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"7d0e707936b0838d129517f54cc0b216a9dd608c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 22:54:49.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d783636af75d4219b474bba6e02381c6 : SUCCESS in 5m 09s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f26c12c545064d0f8e14916e6c48b595 : FAILURE in 58m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6525f4ade0474762816704ef64f6684d : SUCCESS in 34m 33s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/656de55b163b4a96a202ab322d04c854 : FAILURE in 1h 54m 31s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/04e74271e46f400ea31395964c3b4ada : FAILURE in 1h 15m 38s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7e2a7e3cd21249a296fb7c31ebf9ca7f : FAILURE in 1h 11m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ab234be52ba847d5983d28ff57ee7d83 : FAILURE in 1h 15m 06s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/96f6ff43da524dcab650654b9bf09bc3 : SUCCESS in 54m 34s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/5a8ecbc394fe4efd971b24b5fcfc9d1f : SUCCESS in 37m 17s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/57e252ca805e4489a0bcf3939e732c70 : FAILURE in 1h 53m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/06054776a349489997782f8ed0b4877a : FAILURE in 1h 09m 24s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/291854a8308b4f568de9f849106aecb1 : FAILURE in 1h 16m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/950a246f572b459891c9ae5655965b3d : FAILURE in 1h 01m 38s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/dbee1dd690564d9ebec5c3901d363ddf : SUCCESS in 56m 31s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/5f143a641f4e48fda94787ab7e9c3633 : SUCCESS in 41m 41s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/c9503368ba2c4130adafbb2706318d73 : FAILURE in 2h 06m 17s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/947b08b7828644c2bb91fb6d945b6ca1 : FAILURE in 1h 12m 01s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/e91198ee52884a9aac9a3af924e4931f : FAILURE in 1h 36m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/79521cea08bd4aca9302bef5218159a2 : FAILURE in 1h 00m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cda2af4fd924b31984e63140a873963 : FAILURE in 4m 11s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/c9fefb7fba98427ba609002dbcdd23a3 : RETRY_LIMIT in 9m 00s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/66a04bb51d5d41f4a09409b455e16edb : RETRY_LIMIT in 6m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/e1d1a95904d44ca5bf8862055aa0cfaa : SUCCESS in 1h 37m 55s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/88462a4762f541ebb7484df3687d8af8 : SUCCESS in 42m 16s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/4d7e4a226dfc4bf2a369e8d0cd58cb4c : SUCCESS in 52m 00s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/aa116e088562493dafde590b98f48866 : SUCCESS in 36m 04s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/668af6e1f2304b4185843334920d69af : FAILURE in 1h 02m 54s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/f4b09c1955e94df8b17199ad69c4ca5f : SUCCESS in 45m 35s","accounts_in_message":[],"_revision_number":7},{"id":"7603b1284961f2fee411e7214135843cc7d20e18","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-25 15:36:54.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"a68563357d0233774cea478d6cf9ce5488975d89","revisions":{"7400b9ca59d4fabd148e2c6f0b8178bb1cb570dc":{"kind":"REWORK","_number":1,"created":"2020-01-23 02:17:45.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/09/703909/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/09/703909/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/1"}}},"commit":{"parents":[{"commit":"560737404a9d00879c1f2da4f523bb1a03599164","subject":"Merge \"Revert \"Switch neutron-dynamic-routing job to be non-voting temporaray\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/560737404a9d00879c1f2da4f523bb1a03599164"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-23 02:08:30.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-23 02:08:30.000000000","tz":-360},"subject":"Optimize the way to define the api_extensions for train jobs","message":"Optimize the way to define the api_extensions for train jobs\n\nDevstack now has a way to append the api_extensions to the\ndefault extensiones defined in devstack side\n- https://review.opendev.org/#/c/666508/\n\nstable/train default extensions are defined here\n- https://review.opendev.org/#/c/687176/\n\nThis commit appends the additional api_extensions enabled\nin train jobs to the devstack stable/train.\n\nI did diff of existing list form job to what we have in default\ndevstack list and appended the additional one.\n\nChange-Id: I8912b1a123ac52c99b65b93884702e792fa7ea86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7400b9ca59d4fabd148e2c6f0b8178bb1cb570dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7400b9ca59d4fabd148e2c6f0b8178bb1cb570dc"}]},"branch":"refs/heads/master"},"0269302832b55e4a31aa154ca532fd7893fa1619":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-01-23 17:00:37.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/09/703909/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/09/703909/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/2"}}},"commit":{"parents":[{"commit":"560737404a9d00879c1f2da4f523bb1a03599164","subject":"Merge \"Revert \"Switch neutron-dynamic-routing job to be non-voting temporaray\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/560737404a9d00879c1f2da4f523bb1a03599164"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-23 02:08:30.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-01-23 17:00:34.000000000","tz":0},"subject":"Optimize the way to define the api_extensions for train jobs","message":"Optimize the way to define the api_extensions for train jobs\n\nDevstack now has a way to append the api_extensions to the\ndefault extensiones defined in devstack side\n- https://review.opendev.org/#/c/666508/\n\nstable/train default extensions are defined here\n- https://review.opendev.org/#/c/687176/\n\nThis commit appends the additional api_extensions enabled\nin train jobs to the devstack stable/train.\n\nI did diff of existing list form job to what we have in default\ndevstack list and appended the additional one.\n\nDepends-On: https://review.opendev.org/#/c/687176/\n\nChange-Id: I8912b1a123ac52c99b65b93884702e792fa7ea86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0269302832b55e4a31aa154ca532fd7893fa1619"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0269302832b55e4a31aa154ca532fd7893fa1619"}]},"branch":"refs/heads/master"},"20b02e5e01319e21dbba688eee64560bf00d38b3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-01-24 15:16:24.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/09/703909/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/09/703909/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/3"}}},"commit":{"parents":[{"commit":"560737404a9d00879c1f2da4f523bb1a03599164","subject":"Merge \"Revert \"Switch neutron-dynamic-routing job to be non-voting temporaray\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/560737404a9d00879c1f2da4f523bb1a03599164"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-23 02:08:30.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-01-24 15:16:22.000000000","tz":0},"subject":"Optimize the way to define the api_extensions for train jobs","message":"Optimize the way to define the api_extensions for train jobs\n\nDevstack now has a way to append the api_extensions to the\ndefault extensiones defined in devstack side\n- https://review.opendev.org/#/c/666508/\n\nstable/train default extensions are defined here\n- https://review.opendev.org/#/c/687176/\n\nThis commit appends the additional api_extensions enabled\nin train jobs to the devstack stable/train.\n\nI did diff of existing list form job to what we have in default\ndevstack list and appended the additional one.\n\nChange-Id: I8912b1a123ac52c99b65b93884702e792fa7ea86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/20b02e5e01319e21dbba688eee64560bf00d38b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/20b02e5e01319e21dbba688eee64560bf00d38b3"}]},"branch":"refs/heads/master"},"32470d22f84df92b60dc1c66a940e5804d39e82a":{"kind":"REWORK","_number":4,"created":"2020-05-06 00:53:42.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/09/703909/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/09/703909/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/4"}}},"commit":{"parents":[{"commit":"4bc1b00213cf660648cad1916fe6497ac29b2e78","subject":"Merge \"Ensure that external network don\u0027t have any ports before deletion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4bc1b00213cf660648cad1916fe6497ac29b2e78"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-23 02:08:30.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-06 00:52:18.000000000","tz":-300},"subject":"Optimize the way to define the api_extensions for train jobs","message":"Optimize the way to define the api_extensions for train jobs\n\nDevstack now has a way to append the api_extensions to the\ndefault extensiones defined in devstack side\n- https://review.opendev.org/#/c/666508/\n\nstable/train default extensions are defined here\n- https://review.opendev.org/#/c/687176/\n\nThis commit appends the additional api_extensions enabled\nin train jobs to the devstack stable/train.\n\nI did diff of existing list form job to what we have in default\ndevstack list and appended the additional one.\n\nChange-Id: I8912b1a123ac52c99b65b93884702e792fa7ea86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/32470d22f84df92b60dc1c66a940e5804d39e82a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/32470d22f84df92b60dc1c66a940e5804d39e82a"}]},"branch":"refs/heads/master"},"918480b712e62d73f95ac7003a1934e02cc95f8d":{"kind":"REWORK","_number":5,"created":"2020-05-08 20:00:15.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/09/703909/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/09/703909/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/5"}}},"commit":{"parents":[{"commit":"30e13ba32426bf0ce177bcfd0d3df3ebcd4f4a7c","subject":"Merge \"Test security group creation with insufficient rules quota\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/30e13ba32426bf0ce177bcfd0d3df3ebcd4f4a7c"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-23 02:08:30.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-08 19:55:13.000000000","tz":-300},"subject":"Optimize the way to define the api_extensions","message":"Optimize the way to define the api_extensions\n\nDevstack now has a way to append the api_extensions to the\ndefault extensiones defined in devstack side\n- https://review.opendev.org/#/c/666508/\n\nmaster devstack set the \u0027all\u0027 api extensions so master\njob does not need to set the api extensions. \u0027all\u0027 means\nhere is every extensions will be considered as enabled.\n\nstable branch devstack set default set of extensions for each\nstable branch. Neutron stable jobs only needs to append\nthe additional api_extensions enabled in that stable\nbranch job.\n\nChange-Id: I8912b1a123ac52c99b65b93884702e792fa7ea86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/918480b712e62d73f95ac7003a1934e02cc95f8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/918480b712e62d73f95ac7003a1934e02cc95f8d"}]},"branch":"refs/heads/master"},"3b91da54d808f1b406aff9805443525f577d8338":{"kind":"REWORK","_number":6,"created":"2020-05-08 20:40:40.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/09/703909/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/09/703909/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/6"}}},"commit":{"parents":[{"commit":"30e13ba32426bf0ce177bcfd0d3df3ebcd4f4a7c","subject":"Merge \"Test security group creation with insufficient rules quota\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/30e13ba32426bf0ce177bcfd0d3df3ebcd4f4a7c"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-23 02:08:30.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-08 20:39:05.000000000","tz":-300},"subject":"Optimize the way to define the api_extensions","message":"Optimize the way to define the api_extensions\n\nDevstack now has a way to append the api_extensions to the\ndefault extensiones defined in devstack side\n- https://review.opendev.org/#/c/666508/\n\nmaster devstack set the \u0027all\u0027 api extensions so master\njob does not need to set the api extensions. \u0027all\u0027 means\nhere is every extensions will be considered as enabled.\n\nstable branch devstack set default set of extensions for each\nstable branch. Neutron stable jobs only needs to append\nthe additional api_extensions enabled in that stable\nbranch job.\n\nChange-Id: I8912b1a123ac52c99b65b93884702e792fa7ea86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3b91da54d808f1b406aff9805443525f577d8338"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3b91da54d808f1b406aff9805443525f577d8338"}]},"branch":"refs/heads/master"},"a68563357d0233774cea478d6cf9ce5488975d89":{"kind":"REWORK","_number":7,"created":"2020-05-08 20:43:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/09/703909/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/09/703909/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/703909/7"}}},"commit":{"parents":[{"commit":"30e13ba32426bf0ce177bcfd0d3df3ebcd4f4a7c","subject":"Merge \"Test security group creation with insufficient rules quota\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/30e13ba32426bf0ce177bcfd0d3df3ebcd4f4a7c"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-23 02:08:30.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-08 20:42:59.000000000","tz":0},"subject":"Optimize the way to define the api_extensions","message":"Optimize the way to define the api_extensions\n\nDevstack now has a way to append the api_extensions to the\ndefault extensiones defined in devstack side\n- https://review.opendev.org/#/c/666508/\n\nmaster devstack set the \u0027all\u0027 api extensions so master\njob does not need to set the api extensions. \u0027all\u0027 means\nhere is every extensions will be considered as enabled.\n\nstable branch devstack set default set of extensions for each\nstable branch. Neutron stable jobs only needs to append\nthe additional api_extensions enabled in that stable\nbranch job.\n\nChange-Id: I8912b1a123ac52c99b65b93884702e792fa7ea86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a68563357d0233774cea478d6cf9ce5488975d89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a68563357d0233774cea478d6cf9ce5488975d89"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
