)]}'
{"id":"openstack%2Fneutron-tempest-plugin~704809","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I64443361efde18dba3876af504947772df1f7f61","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bp/neutron-classifier-neutron-qos","hashtags":[],"change_id":"I64443361efde18dba3876af504947772df1f7f61","subject":"WIP: classification API tests","status":"ABANDONED","created":"2020-01-29 14:06:09.000000000","updated":"2020-04-09 10:49:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"315cef9d4c7c3bcccf5bb3ac6c64a6644686b6fa","_number":704809,"virtual_id_number":704809,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-01-31 14:49:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-29 14:11:09.000000000","updated_by":{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},"reviewer":{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},"state":"REVIEWER"},{"updated":"2020-01-31 14:49:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"acfa87a8038fdf9a9ada6b93114cec12e1bb108e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-29 14:06:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e71ac53da2c03572887ca560aa1fe360acf2ce20","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-29 14:21:17.000000000","message":"Patch Set 1:\n\nTests for classification are on the way.\nFor me the test_delete_classification_group failed as the GET classification group failed with:\nFile \"/opt/stack/neutron/neutron/services/classification/plugin.py\", line 55, in _make_db_dict\nfor key in obj.fields.keys():\nAttributeError: \u0027NoneType\u0027 object has no attribute \u0027fields\u0027\n\nIt is a good question (at least for me) to have these tests here or to have them in tempest","accounts_in_message":[],"_revision_number":1},{"id":"4ee4d600e683855cbd22f1401f5689054b5de883","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 16:35:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e085dbd4645e4414b5e79652ba9f70f0 : SUCCESS in 5m 25s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/003b58b1530645ecb3b6e6cc5f528bed : SUCCESS in 1h 03m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1b63fd0b39a844a8a21c1beb580c9304 : SUCCESS in 50m 01s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/bf350fa16a7d479cbbb3a0b42e1a1b95 : FAILURE in 2h 26m 46s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/818258f9e4b14028bc5e96709597bf47 : SUCCESS in 1h 07m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6e2147d9eda844e2ac1fcbff3c2c478b : SUCCESS in 1h 10m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/14a0335819024d82bc589d97cbbb94ce : SUCCESS in 55m 26s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/71c0273187594e63a7eb5a5325d9bac0 : SUCCESS in 55m 30s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/e5cbf0e87d17499e8189b1f4459afafd : SUCCESS in 42m 13s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/9c977d7b5dda496eb3f7efb585f4d6d7 : SUCCESS in 1h 43m 53s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/1029ac271e484fe59cae5fbfa4387aa4 : SUCCESS in 1h 24m 48s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/89a53578828849f5adf0949d59fa5e1e : SUCCESS in 1h 26m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/0e388d01a49647a0ba5057ed9140db69 : SUCCESS in 1h 06m 41s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/8a2b9a5e880148fcb9da8a2b98d7d906 : SUCCESS in 46m 29s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/da6920303e754acdbbc88816868e3cd5 : SUCCESS in 36m 37s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/f36e2219e2e04758a48e7b02df063dfc : FAILURE in 1h 21m 12s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/be73f31983704e16a185e8adef4c62f8 : SUCCESS in 1h 07m 07s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/6ec32c9508a74a9cbad1bec26327b785 : SUCCESS in 1h 03m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/c61eed9e75a448869d56754369af0321 : SUCCESS in 55m 11s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/04d386a9400d445aa16c355e5b2afef1 : SUCCESS in 1h 01m 00s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/7575a41823c34bf2ae076de8b84913d6 : SUCCESS in 34m 14s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/b7ce0f2b295a4bee85faff6b24a5db5f : FAILURE in 1h 26m 39s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/e279991a508748d3ad6ef70a2479e4ee : SUCCESS in 1h 29m 46s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/3d19c98b745045e2b11f3dfd25ad73e3 : FAILURE in 1h 01m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/665b7846525043cc97987c5254f35995 : SUCCESS in 59m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42368f28862541658faabd68eb3afe67 : FAILURE in 3m 32s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/4502b50521ef4cbb998e4d6b68359d1a : SUCCESS in 1h 38m 14s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/2f4d653400a747a4a794f053f7423367 : SUCCESS in 1h 52m 39s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/490c44e803ed4dca89795f9012e0739f : SUCCESS in 1h 38m 58s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/10f4a3e9bf27401a8e66929a9ed68a68 : SUCCESS in 1h 31m 08s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/e294aa8403b8463e94cb769ead3d00b2 : SUCCESS in 37m 30s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/7b45ca7acde440bea69d0601ec09760a : SUCCESS in 41m 19s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/4210988159c44fc98c52d66656c7045c : SUCCESS in 52m 15s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/7f7b518d28e144919df13d522a536dd1 : SUCCESS in 46m 13s","accounts_in_message":[],"_revision_number":1},{"id":"2a8d04da36f7474be29dba347697eb9a39b7ee15","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-31 12:47:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6e57188a505c0ac2de0f777652c8a280286426b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 14:49:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0395b4f319eb48b280e46c0bff2490cd : SUCCESS in 4m 12s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/897b848ada674640a019e6bb2b688a87 : SUCCESS in 55m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cd15f65cb6b945c58ee2e96806c2a7f6 : SUCCESS in 37m 45s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/72b711774ac74d57afe227221bc015ad : FAILURE in 1h 32m 04s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/689ad056e5104c3a9378d84eef27694b : SUCCESS in 1h 15m 20s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f5696f77dead4f69a36e3a01347789ca : SUCCESS in 1h 14m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3918a210449a46ed8f63360ce5bfa0f4 : SUCCESS in 1h 16m 00s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/8d98fa8d62754bf696fd059822fa23bb : FAILURE in 31m 25s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/acccddd1648f4ed8a6919ae9f8a9c7fb : FAILURE in 33m 44s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/b73e954067894819b48f2778b0a3ba44 : FAILURE in 51m 17s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/e2884e9d32b740e58fc5a8dca4e1a415 : FAILURE in 32m 23s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/76ad694e1b474e40b78a9159b0d76fab : FAILURE in 36m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/cb65a068e1194a3c9305d7c6bb5071e2 : FAILURE in 32m 21s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/e2ee25d700db4330b5b184aacce8b3ed : SUCCESS in 56m 11s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/b6e0c22050d44a29b2e81301a82897b2 : SUCCESS in 37m 50s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/c64fc7022ccb47e2b28ffd74b08500d1 : FAILURE in 1h 32m 11s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/0ef573de6905477aa76a2e600a4f0e9c : SUCCESS in 1h 21m 47s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/81725c10af5d4c3c8911dc40f7887215 : SUCCESS in 1h 08m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/f52c1e21bec24994b1d4e1f9cc3f26cf : SUCCESS in 1h 04m 09s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/ae5dfd43a9744a0ba7e8ec1189108078 : SUCCESS in 1h 00m 26s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/8e6e99bfff1846b49790651ec165d15e : SUCCESS in 41m 15s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/6a5493405fd8478faae9a3f04c9e5bb3 : FAILURE in 1h 30m 24s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/1b5479d0062145db8b47c96ba3aad57a : SUCCESS in 1h 07m 56s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/e139585782fe4cd696eabba66d364d69 : SUCCESS in 1h 06m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/2799c8e5b8fc4d768e45a8e8b3f26004 : SUCCESS in 1h 00m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a6ae0cfa223478787af6925290d0cdb : SUCCESS in 5m 08s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/85915e627e0d45e1843f6cc181f7a93b : SUCCESS in 1h 44m 27s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/47c6e996501e409c803da1424faece13 : SUCCESS in 1h 55m 44s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/e038fb53ac1c4c1b8f15ec7c3875c0f0 : SUCCESS in 1h 23m 11s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/8fe599c1f43f4d6e8fba98b4e5055c42 : SUCCESS in 1h 32m 19s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/9f2b9a1f40ce4305804aa9da91810af4 : SUCCESS in 39m 50s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/5988334a944e4158afe5d2b64964ac70 : FAILURE in 37m 25s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/e49e4121df7c4236bb0c971696ec6aa3 : SUCCESS in 47m 48s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/e87f03a9df234834a042cb0bb7a2d18a : SUCCESS in 42m 03s","accounts_in_message":[],"_revision_number":2},{"id":"bc3e5e6dbef319a2f4ffdd519d36706c6acee04e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-09 10:49:45.000000000","message":"Abandoned\n\nThe feature is blocked due to lack of designers, if it will be continued this patch can be restored.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c757426473b7aca892c39c0179c44201959d2aa2","revisions":{"280927917f1718d0af5abf1f5ee7e44832265f5d":{"kind":"REWORK","_number":1,"created":"2020-01-29 14:06:09.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/09/704809/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/09/704809/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/704809/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/704809/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/704809/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/704809/1"}}},"commit":{"parents":[{"commit":"eb7d434bc537e0e9983edd732adadc3fd6aabc06","subject":"Remove unused method \"create_network_keystone_v3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/eb7d434bc537e0e9983edd732adadc3fd6aabc06"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-24 15:57:35.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-29 13:59:20.000000000","tz":60},"subject":"WIP: classification API tests","message":"WIP: classification API tests\n\nRelated Blueprint: blueprint neutron-classifier-neutron-qos\nChange-Id: I64443361efde18dba3876af504947772df1f7f61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/280927917f1718d0af5abf1f5ee7e44832265f5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/280927917f1718d0af5abf1f5ee7e44832265f5d"}]},"branch":"refs/heads/master"},"c757426473b7aca892c39c0179c44201959d2aa2":{"kind":"REWORK","_number":2,"created":"2020-01-31 12:47:18.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/09/704809/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/09/704809/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/704809/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/704809/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/704809/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/09/704809/2"}}},"commit":{"parents":[{"commit":"eb7d434bc537e0e9983edd732adadc3fd6aabc06","subject":"Remove unused method \"create_network_keystone_v3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/eb7d434bc537e0e9983edd732adadc3fd6aabc06"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-24 15:57:35.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-31 12:46:51.000000000","tz":60},"subject":"WIP: classification API tests","message":"WIP: classification API tests\n\nDepends-On: https://review.opendev.org/670050\nRelated Blueprint: blueprint neutron-classifier-neutron-qos\nChange-Id: I64443361efde18dba3876af504947772df1f7f61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c757426473b7aca892c39c0179c44201959d2aa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c757426473b7aca892c39c0179c44201959d2aa2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
