)]}'
{"id":"openstack%2Fneutron-tempest-plugin~704940","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ib94518597bdf3d9f3049643d3242db632769de6b","project":"openstack/neutron-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ib94518597bdf3d9f3049643d3242db632769de6b","subject":"Adding pattern to check_remote_connectivity function","status":"MERGED","created":"2020-01-30 08:43:48.000000000","updated":"2020-01-31 00:26:18.000000000","submitted":"2020-01-31 00:26:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704940-1580430378581-f323dc7b","meta_rev_id":"af8186592e85d0067438ce9848e0349a3026d11d","_number":704940,"virtual_id_number":704940,"owner":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-31 00:26:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},{"value":0,"_account_id":18894,"name":"Toni Freger","email":"tfreger@redhat.com"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2020-01-30 11:17:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":1,"date":"2020-01-30 11:21:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},{"value":2,"date":"2020-01-30 21:08:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},{"value":0,"_account_id":18894,"name":"Toni Freger","email":"tfreger@redhat.com"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},{"value":1,"date":"2020-01-30 21:08:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},{"value":0,"_account_id":18894,"name":"Toni Freger","email":"tfreger@redhat.com"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},{"value":0,"_account_id":18894,"name":"Toni Freger","email":"tfreger@redhat.com"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":18894,"name":"Toni Freger","email":"tfreger@redhat.com"},{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-30 08:43:59.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2020-01-30 08:45:56.000000000","updated_by":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"reviewer":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"state":"REVIEWER"},{"updated":"2020-01-30 08:46:00.000000000","updated_by":{"_account_id":18894,"name":"Toni Freger","email":"tfreger@redhat.com"},"reviewer":{"_account_id":18894,"name":"Toni Freger","email":"tfreger@redhat.com"},"state":"REVIEWER"},{"updated":"2020-01-30 11:17:42.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-01-30 11:21:37.000000000","updated_by":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"reviewer":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"state":"REVIEWER"},{"updated":"2020-01-30 21:08:52.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-01-31 00:26:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ed9db1fb3dd90e222959e50286e0f5066ce25863","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2020-01-30 08:43:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab9befd94bab0a4ebe3a2c02deacd4bf4170d79c","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2020-01-30 08:43:59.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"4fecde570c7eacf30c0270686189509be0eb223d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-30 09:12:57.000000000","message":"Patch Set 1:\n\nI\u0027m OK with this patch, but I have a couple of questions:\n- Where this functionality is needed?\n- Do you have a launchpad bug to describe this need?","accounts_in_message":[],"_revision_number":1},{"id":"1cfbf7058370ac71aeef66d23eea182e28bb8086","author":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"date":"2020-01-30 09:38:34.000000000","message":"Patch Set 1:\n\n\u003e I\u0027m OK with this patch, but I have a couple of questions:\n \u003e - Where this functionality is needed?\n \u003e - Do you have a launchpad bug to describe this need?\n\nHi Rodolfo, this patch was my idea. We need this functionality in ovn downstream dvr/l3ha tests which use tcpdump in order to validate that traffic is passing through specific openstack node.\nAnd also this functionality of ping can be useful by itself in order to test if traffic routing is not affected by some specific data in a packet. \n\nEduardo, it really would be nice if the commit message will have some explanation how this functionality can be useful.","accounts_in_message":[],"_revision_number":1},{"id":"4c61be9116ade5ba4b890130ad764dc0d501766f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 10:05:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47e57bbc44564c3bb0e6337c1270742e : SUCCESS in 3m 48s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/aafdda15e7514987967b994fd0a299c8 : FAILURE in 34m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/58cf83a1692344e393704ddf54ca63f7 : FAILURE in 28m 34s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/16c748bc371a42ab87632c20492ec021 : FAILURE in 44m 23s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7a62be623c3d45de9e09fbe3220a06a2 : FAILURE in 32m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/75fcec9a28024636ae110aad4fad90f1 : FAILURE in 40m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9b2c58912ad844d5a93e4df4feeca026 : FAILURE in 33m 10s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/252e2a89dfdf4eae979fecfc9b48ecc7 : FAILURE in 34m 52s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/f556f60bc7a343db98025d7f08c9e7fc : FAILURE in 28m 56s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/68b9f96be8604426acbd294e12350373 : FAILURE in 45m 48s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/b5aaa57e78f24f5894c3d3645a19d536 : FAILURE in 32m 13s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/9631dfc30ac54820a034c81531f242fc : FAILURE in 37m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/6a78fe4e95474923bc5fb8aa4cec4b28 : FAILURE in 31m 56s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/d6ddaa2aab6642dbbc2294c44dca445e : FAILURE in 34m 23s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/f9570d8aad274b63a4b94e14cc3bb00c : FAILURE in 32m 39s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/6272acd170e34c78ad209235b65f5c91 : FAILURE in 45m 28s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/d6b6421a68164fad8f598ca54fb03b09 : FAILURE in 30m 50s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/f4d4c552668d480aa9e84f7f553b3d26 : FAILURE in 42m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/1a47f49daa3f4f108a47aa6b6cd07a51 : FAILURE in 32m 56s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/9e9522c79f79407e842a12ad9e196c6a : FAILURE in 39m 33s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/b9353fe1cd034a1db2e1ac11060754d5 : FAILURE in 30m 31s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/cd80586844dd4c22a25d40680b8f342e : FAILURE in 48m 59s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/cfcdd782e858423f86c0e5f694b10a07 : FAILURE in 34m 22s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/d4d389b2d67342a69cc83225eab6d858 : FAILURE in 33m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/38ef25e11a4f42839956a3cf0d8a2e7e : FAILURE in 34m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7fe015250c64157893d0418eb122fd4 : FAILURE in 3m 29s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/972d4fd491414516aad2ed43eaeb1747 : FAILURE in 30m 50s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/816cc155ea5e4806b343c4be1265e24c : FAILURE in 36m 01s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/076f5b3db50748378038eeb65c46b423 : FAILURE in 34m 42s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/df94f3d25d2148b5b97c4ecd84eee8ba : FAILURE in 33m 44s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/72ec0624222545b385775cb5ea2ab343 : FAILURE in 31m 43s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/8c478d71b41143e1bfbcda14057004eb : FAILURE in 38m 58s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/c36338b79371474c90e7e320e9ff0894 : FAILURE in 33m 24s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/2f2d3a854bf64fd3acab655c257c3757 : FAILURE in 36m 07s","accounts_in_message":[],"_revision_number":1},{"id":"5e8c0c644ed734baf34ece4f1cb61c0d68dce7b3","author":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"date":"2020-01-30 10:14:43.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ad2b605a7e5f854f385ea6271e4967d636be81e","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2020-01-30 10:53:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5797f24a17f8b306df005c79905c4be21ef814bf","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2020-01-30 10:54:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0cc470d18c02120ea9808eb0e3d21a791fa26a3e","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2020-01-30 11:15:30.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"4dc2622552c259730e3598614235a97349da70cc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-30 11:17:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"408a34b07dfbd8b7cb984165907acb189b5236db","author":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"date":"2020-01-30 11:21:37.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9440b04f8c837fadc8e51ce45c35ddd9aae7253f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 14:25:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16c6aa8e97074c048b61ca8291d37cb3 : SUCCESS in 5m 33s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/71d7c97cd5274db7bd165ffa91b7ce6c : SUCCESS in 1h 07m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9e05e07d627344a1a4429145a6fcc594 : SUCCESS in 41m 12s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/a8f1fce0d0bb43f8918f2d35424a5985 : FAILURE in 1h 43m 24s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8118dfb74d73429fb745ed031d20ccf9 : SUCCESS in 1h 12m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/626461c4057f48a4bb50642e0b386399 : SUCCESS in 1h 12m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/86fd68a96a7949e9b9b55b66672038fe : SUCCESS in 1h 13m 50s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/1a7e7e5952314c6cb6a8d550493147c6 : SUCCESS in 53m 44s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/e89892c030c94f1c8bdad52783dccead : SUCCESS in 40m 56s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/8f43700fbac44b88951c3d84f437dd19 : SUCCESS in 1h 52m 29s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/3918e1e90dd94029bb02563e31a94918 : SUCCESS in 1h 35m 59s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/db06248935f548958d547e481156b558 : SUCCESS in 1h 23m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/c7ccd2717a1a471fab94b4cad49dec22 : SUCCESS in 1h 15m 42s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/9270eaea6c244cf39d0d8c2571eb2c4d : SUCCESS in 1h 01m 49s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/bd37b99e944049d29cf480df786407bd : SUCCESS in 35m 49s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/29061e3f2c2d4fa182fd1e6e0fa47ac0 : SUCCESS in 1h 33m 13s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/7c4f45fbdc8a4b0f950e6738fe971155 : SUCCESS in 1h 07m 37s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/a3c6b571f98241d3afdb9d34b2393d12 : SUCCESS in 1h 09m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/b446c00a93d746f39c4fe9b170f96e03 : SUCCESS in 1h 07m 53s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/51448c8738164d9c9c3ec4ba27bc9f13 : SUCCESS in 53m 00s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/da8da45eb6a94a78bf3e410295db8be2 : SUCCESS in 44m 16s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/494307f5645d4540b123216dff1a5424 : SUCCESS in 1h 42m 32s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/8ff1a4a178fe4d56a1628efc8659d24e : SUCCESS in 1h 05m 03s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/5702572e3243433dbb50e46e22ecd823 : SUCCESS in 1h 06m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/8b9ffb5c98f7421f84d21431fec4088e : SUCCESS in 1h 04m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/151ad688050c470594acaf413099b062 : SUCCESS in 4m 46s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/1b396512c68640848e11eee1bcfe5b8b : SUCCESS in 1h 41m 43s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/965e7680ddd6492bbcc11b4aafeb1b78 : SUCCESS in 1h 43m 59s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/2bd2559ceb2d47bc8511a9b610322b72 : SUCCESS in 1h 27m 56s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/c582b0d45b2e4016a1828b62ec35d596 : SUCCESS in 1h 38m 18s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/b52201dd18b5485bbcbfcfcd861aaa65 : FAILURE in 40m 00s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/aafb9dec3f4f47f1b3a59274c8c1aa26 : SUCCESS in 47m 46s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/2007d59127d940a7a78f3a8ff9105252 : SUCCESS in 49m 00s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/f87684cde5bb4126b919f84fa0fcd858 : SUCCESS in 45m 04s","accounts_in_message":[],"_revision_number":3},{"id":"8aa99c43737fbc5bc1153422f5c31294ebd69122","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-30 21:08:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ddc0c089c2fc887347da5400585594b51f113966","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 21:09:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"31fe2c5442e179858558ee562d797634127f16cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 00:26:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/889a50a09d2f494fb028f987a2edd722 : SUCCESS in 6m 00s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/41a3e917a0414b97ad14bc422eb9cef5 : SUCCESS in 57m 34s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7341edd1aa014174a887af3929773c1e : SUCCESS in 1h 15m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/685608a2d8354e749d8bc9ee8cc73040 : SUCCESS in 1h 07m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4fd3bae5ba5846eea17a5f4f623517cc : SUCCESS in 1h 21m 26s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/4824cc38d70a405d9b3a9f0655020b9a : SUCCESS in 59m 45s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/7ce2d9779cdc4ca0b97e6c5e3dbfea80 : SUCCESS in 1h 01m 32s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/19346e7318ad4724adff56fa2fccd364 : SUCCESS in 54m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed90b11504bf420b99e419f5657adbd0 : SUCCESS in 4m 26s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/e34e312c85b7475baa8fdfd7e4c05a23 : SUCCESS in 1h 56m 54s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9b65adebcc204f08bd477ca8c5bc7255 : SUCCESS in 1h 26m 25s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/e5cce9b274b14c4aad561ff34c60b359 : SUCCESS in 50m 49s","accounts_in_message":[],"_revision_number":3},{"id":"af8186592e85d0067438ce9848e0349a3026d11d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 00:26:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f2b60545a585169f4d17653c024320f499cc8026","revisions":{"f571dbbf34e0e98b25a5973d603932922f13dfb8":{"kind":"REWORK","_number":1,"created":"2020-01-30 08:43:48.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/40/704940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/40/704940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/704940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/704940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/704940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/704940/1"}}},"commit":{"parents":[{"commit":"92959e541270d9c62599914c93f53eb71e6b941d","subject":"Merge \"Add debugging information in case of SSH connection error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/92959e541270d9c62599914c93f53eb71e6b941d"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2020-01-30 08:37:22.000000000","tz":60},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2020-01-30 08:41:49.000000000","tz":60},"subject":"Adding pattern to check_remote_connectivity function","message":"Adding pattern to check_remote_connectivity function\n\nPing messages sent by this function will include a pattern of bytes\nthat are repeated until message size is completed.\nInput format required is a string with hex digits\n\nChange-Id: Ib94518597bdf3d9f3049643d3242db632769de6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f571dbbf34e0e98b25a5973d603932922f13dfb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f571dbbf34e0e98b25a5973d603932922f13dfb8"}]},"branch":"refs/heads/master"},"fb0aadc9a8a528ccb4187a3984bc65ef017fa6f4":{"kind":"REWORK","_number":2,"created":"2020-01-30 10:53:02.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/40/704940/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/40/704940/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/704940/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/704940/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/704940/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/704940/2"}}},"commit":{"parents":[{"commit":"92959e541270d9c62599914c93f53eb71e6b941d","subject":"Merge \"Add debugging information in case of SSH connection error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/92959e541270d9c62599914c93f53eb71e6b941d"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2020-01-30 08:37:22.000000000","tz":60},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2020-01-30 10:47:05.000000000","tz":60},"subject":"Adding pattern to check_remote_connectivity function","message":"Adding pattern to check_remote_connectivity function\n\nPing messages sent by this function will include a pattern of bytes\nthat are repeated until message size is completed.\nInput format required is a string with hex digits.\n\nThis is useful when these messages are captured, in order to identify\nand differentiate clearly messages from different tests.\nIt can be used to validate that traffic routing is not affected by\npacket data as well.\n\nChange-Id: Ib94518597bdf3d9f3049643d3242db632769de6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/fb0aadc9a8a528ccb4187a3984bc65ef017fa6f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/fb0aadc9a8a528ccb4187a3984bc65ef017fa6f4"}]},"branch":"refs/heads/master"},"f2b60545a585169f4d17653c024320f499cc8026":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-01-30 11:15:30.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/40/704940/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/40/704940/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/704940/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/704940/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/704940/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/704940/3"}}},"commit":{"parents":[{"commit":"92959e541270d9c62599914c93f53eb71e6b941d","subject":"Merge \"Add debugging information in case of SSH connection error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/92959e541270d9c62599914c93f53eb71e6b941d"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2020-01-30 08:37:22.000000000","tz":60},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2020-01-30 11:14:49.000000000","tz":60},"subject":"Adding pattern to check_remote_connectivity function","message":"Adding pattern to check_remote_connectivity function\n\nPing messages sent by this function will include a pattern of bytes\nthat are repeated until message size is completed.\nInput format required is a string with hex digits.\n\nThis is useful when these messages are captured, in order to identify\nand differentiate clearly messages from different tests.\nIt can be used to validate that traffic routing is not affected by\npacket data as well.\n\nCloses-Bug: #1861397\n\nChange-Id: Ib94518597bdf3d9f3049643d3242db632769de6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f2b60545a585169f4d17653c024320f499cc8026"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f2b60545a585169f4d17653c024320f499cc8026"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
