)]}'
{"id":"openstack%2Fneutron-tempest-plugin~706119","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ic04c9d1d2562398bd3640bfd064971bc82235502","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"Ic04c9d1d2562398bd3640bfd064971bc82235502","hashtags":[],"change_id":"Ic04c9d1d2562398bd3640bfd064971bc82235502","subject":"Modify rocky jobs to run with correct py version","status":"MERGED","created":"2020-02-05 21:22:24.000000000","updated":"2020-02-12 12:16:49.000000000","submitted":"2020-02-12 12:16:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706119-1581509809572-86a07e3f","meta_rev_id":"13011f23737a20a4eeb2cc37f0987f88d5b43e85","_number":706119,"virtual_id_number":706119,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-12 12:16:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-07 13:37:32.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2020-02-12 09:38:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-02-10 09:39:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-02-10 19:07:35.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2020-02-12 09:38:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-10 09:39:44.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-02-10 19:07:35.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-02-12 09:38:34.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-02-12 12:16:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8ddec5861fda12e3798d986b7794051f9c607142","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-05 21:22:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4bf89760bf1b4ae632aad74d99ff3f19fd17952","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-02-05 22:50:20.000000000","message":"Patch Set 1:\n\nLooks like the rocky jobs all got hit with the \"Could not find a version that satisfies the requirement oslo.concurrency\u003d\u003d\u003d4.0.0\" issue.","accounts_in_message":[],"_revision_number":1},{"id":"86ac6eb8f3525a60e21cd4e83f2ccb0946698083","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 23:46:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57330c1ed8d94cf692c6c702dff847e4 : SUCCESS in 5m 10s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/83d45277bb704408891162719b01b4b8 : SUCCESS in 50m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7952b5fd523e41c0b7216a4582c63cbf : SUCCESS in 33m 50s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/b6aba7b1dc334b579f5fecf3092b8eac : FAILURE in 1h 43m 43s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e452a00b689444c9bf898eafa6a5ff36 : SUCCESS in 1h 34m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1c6f15cb9ae3468aa66d59d2ab7c94b3 : SUCCESS in 1h 12m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c9637f8e2ccd4777a540560d923b6f16 : SUCCESS in 1h 02m 59s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/5ec1fe9f969f47a0a4751351a0d8c996 : FAILURE in 24m 16s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/ee538ccd46b8434dbe24d1ca1f34a9e3 : FAILURE in 23m 20s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/6d71f25c11e74d47a15e5fc6b249dbdc : FAILURE in 38m 53s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/c6a768b29df64ec99532f8532a261596 : FAILURE in 19m 41s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/9aa057f2f8344ef7bd02e2fe84ff418b : FAILURE in 24m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/4db5f22100a44a858632b8984da9afb6 : FAILURE in 26m 46s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/3cb032cbfd5b45b6a3054de5beff7098 : SUCCESS in 1h 27m 31s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/b38abdee27d94d25ab3901eaab30fea8 : SUCCESS in 42m 16s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/19a412ec63404dc39e090b221b1be2fd : FAILURE in 1h 24m 01s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/ba04fa1809264d98aa58db686856933c : SUCCESS in 1h 11m 29s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/eed6d5493aa84d09865fe3ea0253fae5 : SUCCESS in 1h 03m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/1cf08e447881407cabe7cc4979fd9678 : SUCCESS in 1h 47m 39s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/73499467e27841a78c5d8b6e65dd8757 : SUCCESS in 59m 23s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/312cf63d8c9444d99759dc6859f99fbe : SUCCESS in 53m 47s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/9b5cf4711233438fbb18af52178d7f4f : FAILURE in 1h 42m 38s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/cc926a680f6d4bd5832e70fa8e1fe4aa : SUCCESS in 1h 47m 21s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/ad2138bdc6124970afea497cb1d28992 : SUCCESS in 1h 08m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/2e36d6d8be6d4260b14679d8f20614c1 : SUCCESS in 1h 38m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1993a207d6542b3bd4f70d56f341b0f : SUCCESS in 6m 35s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/5b51def626d9425f9afceecb317c781a : SUCCESS in 1h 50m 02s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/46e38850cea74d4bba1cc2d985f73984 : SUCCESS in 2h 19m 33s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/1c61920de9474d8391b80d885d43fa11 : SUCCESS in 1h 36m 39s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/efd50733dd6f4e66b199d81c068c4348 : SUCCESS in 1h 21m 37s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/ed18f791172b487f8892166ca19a4e1a : SUCCESS in 58m 25s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/de4d426adf5f4833a379d640a2ce3301 : SUCCESS in 41m 22s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/468a6dd1138f4368975bdd2084fd6f52 : SUCCESS in 54m 46s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/b52582848bb54dbd88d7dcfce2073c67 : SUCCESS in 46m 23s","accounts_in_message":[],"_revision_number":1},{"id":"e8c00414f420afa4101673be3ac4f06cee2990ca","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-02-06 09:12:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1fb382345c56172f0f69020084119f5a72778b5a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-06 14:54:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e8d0a0917b3e91a0cd4ae9ab64748e2fc1ef83e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-02-07 09:42:09.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"7781f946646f3eac389587dce2a790d2a3228576","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-02-07 09:43:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5f284b77f4b6e160dcb73c819a0e10c2d93060ad","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-02-07 09:48:02.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8c12a81e200bb860fd3733d7fb3e555e9a81c2bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 12:24:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a083b99251947b8829aec805e7a0524 : SUCCESS in 5m 53s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1a84a39af5944f0788e0b0577ee269c2 : SUCCESS in 53m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d996686ddf11458aaf2c97a05e4af1a1 : SUCCESS in 36m 12s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/237e89cbc8fe49b3bc6efb5447a8e7b3 : FAILURE in 1h 52m 18s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/232c6f84a3bc4b999996b2ac0598af32 : SUCCESS in 1h 00m 31s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b7d9e3b061ac4287937a565fd5d4f3d0 : SUCCESS in 1h 06m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9f40a8eedc5d46b2bc8d3fd462075054 : SUCCESS in 56m 12s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/0d275e781d044f09abcb21a393c0f52e : SUCCESS in 59m 20s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/080799159b694e689d40b5c886992ea6 : SUCCESS in 35m 58s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/f712e3fbe002480f92420003c5b9cba2 : FAILURE in 1h 47m 32s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/8f5c16374a20435590e3229510afb807 : SUCCESS in 1h 05m 05s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/2e01bcc37e4f44c1985daf9bb4f14108 : SUCCESS in 1h 00m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/25cec14d8f334da4b833aac0cc0c6880 : SUCCESS in 1h 15m 26s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/2e056fe4f4a6446ead69421cb8ab8f9e : SUCCESS in 1h 03m 38s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/905c6f01e0224163b6b20b094875cefc : SUCCESS in 37m 37s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/d9c9708572994ee5bf5bd7abbec2195b : FAILURE in 1h 56m 03s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/dfbc78e2c1e24130b3508f9c0140a05b : SUCCESS in 1h 10m 41s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/2ce3ca139b674c51852c41786abbbb36 : SUCCESS in 1h 22m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/11d9eef3583b4d369111dffca62d9dc5 : SUCCESS in 59m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29bcad1b59a947a5ba3e14ba766a05fd : SUCCESS in 3m 50s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/678eab8917124af4bb98bccccff10682 : SUCCESS in 1h 43m 29s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/f039125b39fc44269998dc04979d71da : SUCCESS in 1h 44m 15s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/494219fe5935442a8f2912cbfad47e4d : SUCCESS in 1h 34m 10s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/c78ab4dc94d54e1aa7e398dee6998420 : SUCCESS in 1h 26m 26s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/9ec74628e20f41ed8dea1c9dc5b6e15c : SUCCESS in 36m 34s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/69f6c4d8a08842ac82a6059b403f218b : SUCCESS in 36m 25s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/ffe7ebfffdee4c4b8c0dd661e8c91271 : SUCCESS in 49m 53s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/2081b92f345041f5a156fd7909785de4 : SUCCESS in 38m 59s","accounts_in_message":[],"_revision_number":2},{"id":"2ef19bf2e5eda1ccda8009391a0b02c11eb7d58e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-07 13:37:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9eabda4df4dcd1a0797af13ec258d121af1a7cff","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-02-10 09:39:44.000000000","message":"Patch Set 2: Code-Review+2\n\nI think we are now good to go with this. It\u0027s passing on next patch: https://review.opendev.org/#/c/706451/2","accounts_in_message":[],"_revision_number":2},{"id":"e00e155df6734909c16a298bd4a24f085c2432d9","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-02-10 19:07:35.000000000","message":"Patch Set 2: Code-Review+2\n\nWaiting for Rodolfo to remove his -1.","accounts_in_message":[],"_revision_number":2},{"id":"076326815365ef6bdba2d98f1f2961daa5eaa084","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-02-12 09:38:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0e170ed471d561230013770db98d702682f92433","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 09:39:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"53749c28541c67394ea026b515fd806169cf5d70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 12:16:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04da940745e145aab6726b899f245fc6 : SUCCESS in 7m 58s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2536ead0117441c5889ea8eecb5b82cf : SUCCESS in 51m 49s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c3f2bfc57f7240fe97b1eecd8a8917b3 : SUCCESS in 1h 16m 48s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/48862334d13e489d8c9efbb32ab3ff9f : SUCCESS in 1h 06m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d1072bb00f2b4b59af79caf1b60406a7 : SUCCESS in 1h 02m 23s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/0be14a0338d94df483b925c8f2c4fe5d : SUCCESS in 56m 41s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/f85db68c9f524767a62d1b6694dd002c : SUCCESS in 1h 03m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7a129f2e9154ddc97a4b24a0f01d7b4 : SUCCESS in 6m 03s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/df809a47e0304bc289a35fc103460a64 : SUCCESS in 1h 49m 03s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/e3294afdc48c4508814b3fd3174a7095 : SUCCESS in 1h 40m 39s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/eeb58465d57a4f4fa254ca335a0c384e : SUCCESS in 51m 13s","accounts_in_message":[],"_revision_number":2},{"id":"e70a1659712afb3ca80a0a8560a09527338d8dc5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 12:16:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d44ac533581eb6e9aecfc8110fdfa5d383201ba7","revisions":{"22ecedce47ef074499231e618296bc3ed5a2c34d":{"kind":"REWORK","_number":1,"created":"2020-02-05 21:22:24.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/19/706119/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/19/706119/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/19/706119/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/19/706119/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/19/706119/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/19/706119/1"}}},"commit":{"parents":[{"commit":"4dea8060f05fab427cd042934674f08b03927e05","subject":"Remove \"unstable_test\" decorator local implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4dea8060f05fab427cd042934674f08b03927e05"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-02-05 21:17:50.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-02-05 21:17:50.000000000","tz":-360},"subject":"Modify rocky jobs to run with correct py version","message":"Modify rocky jobs to run with correct py version\n\nOpenStak ussuri release is python3-only which means\nstable branhces jobs running on neutron-tempest-plugin\nmaster gate should run with py3. neutron-tempest-plugin\nwill drop the py2 support soon so it will not be possible\nto test the stable branch with py2.\n\nRocky jobs are modified to run on py2 on stable/rocky gate\nand on py3 for master gate.\n\nChange-Id: Ic04c9d1d2562398bd3640bfd064971bc82235502\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/22ecedce47ef074499231e618296bc3ed5a2c34d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/22ecedce47ef074499231e618296bc3ed5a2c34d"}]},"branch":"refs/heads/master"},"d44ac533581eb6e9aecfc8110fdfa5d383201ba7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-02-07 09:42:09.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/19/706119/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/19/706119/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/19/706119/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/19/706119/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/19/706119/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/19/706119/2"}}},"commit":{"parents":[{"commit":"de952714e262f88c927ab6bb2cc5dfbb5daaf9c7","subject":"Drop all Rocky jobs from both check and gate queues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/de952714e262f88c927ab6bb2cc5dfbb5daaf9c7"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-02-05 21:17:50.000000000","tz":-360},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-02-07 09:39:08.000000000","tz":60},"subject":"Modify rocky jobs to run with correct py version","message":"Modify rocky jobs to run with correct py version\n\nOpenStak ussuri release is python3-only which means\nstable branhces jobs running on neutron-tempest-plugin\nmaster gate should run with py3. neutron-tempest-plugin\nwill drop the py2 support soon so it will not be possible\nto test the stable branch with py2.\n\nRocky jobs are modified to run on py2 on stable/rocky gate\nand on py3 for master gate.\n\nChange-Id: Ic04c9d1d2562398bd3640bfd064971bc82235502\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d44ac533581eb6e9aecfc8110fdfa5d383201ba7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d44ac533581eb6e9aecfc8110fdfa5d383201ba7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
