)]}'
{"id":"openstack%2Fneutron-tempest-plugin~708842","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ie961293215c48883c962debe888b0bca418e3eb9","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"test_sg_rules_quota_increased","hashtags":[],"change_id":"Ie961293215c48883c962debe888b0bca418e3eb9","subject":"Adding new test of security group rules quota increased.","status":"MERGED","created":"2020-02-20 10:24:15.000000000","updated":"2020-03-11 11:56:57.000000000","submitted":"2020-03-11 11:56:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708842-1583927817294-eb552905","meta_rev_id":"e717d20c06600bfe6c27131126aa2287a1e697aa","_number":708842,"virtual_id_number":708842,"owner":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-11 11:56:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2020-03-11 10:20:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-03-10 15:42:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":2,"date":"2020-03-10 13:36:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2020-03-11 10:20:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-20 11:19:07.000000000","updated_by":{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},"reviewer":{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},"state":"REVIEWER"},{"updated":"2020-02-23 10:30:37.000000000","updated_by":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"reviewer":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"state":"REVIEWER"},{"updated":"2020-02-23 12:30:17.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"state":"REVIEWER"},{"updated":"2020-03-10 13:36:47.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-03-10 15:42:43.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-03-11 10:20:01.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-03-11 11:56:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0325142f49730ae72ea7daa8ad5a1a86212759a2","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-02-20 10:24:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4697acad9032c38a709ad8dee51f4516482554f","author":{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},"date":"2020-02-20 11:19:07.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nYour test suppose to fail","accounts_in_message":[],"_revision_number":1},{"id":"38e75fc63bee8d98dde37ccac352c9a8ddf8527a","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-02-20 11:27:07.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ad9da239f5f161b5831a52a4d3936d9e24016f6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 13:29:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e06cf5008d494ea98f78e2e58c04eb57 : SUCCESS in 5m 11s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d8e168968b214b658d1c455373814d35 : FAILURE in 1h 02m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/005d446065844de881ee82e6b827ec37 : SUCCESS in 35m 05s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/c7dd073404354c2cbf174effa25b52be : FAILURE in 2h 00m 42s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/214e3d6e57ee48eab346fd3df9d4db1d : SUCCESS in 1h 11m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/da7e87ea85fa43a397f163b925210bf0 : SUCCESS in 1h 16m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/08da519c5aea4df295f6bc98bf05363a : SUCCESS in 1h 12m 27s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/43da1c270ead4b84aa020793aeffc9a0 : FAILURE in 1h 15m 19s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/5e801988a4a54e7989288b2c826df827 : FAILURE in 22m 45s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/2046d1823f664c668cf07c9fb000cd4d : SUCCESS in 1h 50m 17s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/55aa3f3af19d458da1aebac935a66714 : SUCCESS in 1h 15m 09s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/9fa7d696303545d4859c50efc1ea12fd : SUCCESS in 1h 14m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/3fb3b87510284c40bf90bf7cfcd4ef06 : SUCCESS in 1h 25m 41s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/66ab6f563e834b3eb8359b9f105ae827 : FAILURE in 1h 16m 49s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/3c7998072d784b988575c9130d57f9ee : SUCCESS in 36m 53s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/03479d1e28f149f69d76279833cf033a : SUCCESS in 1h 26m 13s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/cb28411fd9a048b08f87c219da159129 : SUCCESS in 1h 26m 37s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/df7c05c447504da3b46119e8160dc871 : SUCCESS in 1h 04m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/6d52b88b3911409a85adc04bbe78a94d : SUCCESS in 1h 05m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fdc91b809534945bbf2be8f2dfb29c7 : FAILURE in 4m 42s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/5762f2874377426ea6388122654da143 : SUCCESS in 1h 36m 14s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/e6304ac173d24d10b1ca6212889b49ee : SUCCESS in 1h 36m 58s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c5b41b309cee4e17b06fe210667480a0 : FAILURE in 32m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/1838a5374cbb4818a63f7a93746b567d : SUCCESS in 1h 24m 18s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/bdf6adc29dca4b9abccd385456bb5325 : SUCCESS in 37m 28s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/e9a9aee2c2e5413196ccff2bdb9d854f : SUCCESS in 40m 55s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/fdcb60a428a6447e9ec2a250e6be4b9d : SUCCESS in 47m 58s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/80cbc09881f448a99c354f8921445672 : SUCCESS in 38m 53s","accounts_in_message":[],"_revision_number":2},{"id":"9b7edd14f136a649b57b953480fa6299f00b871d","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-02-23 10:27:50.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c93145a6175c2041af5f247519df92ec2768e546","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-02-23 10:29:44.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"5c618cdd5318df837728b069829310827738c223","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 12:24:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a0694c0174348328417dee584718324 : SUCCESS in 4m 43s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/91080bdd6c5745e08e5189ac4ab656bd : FAILURE in 1h 00m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7aa7213e2e324b179604c20a1806f676 : SUCCESS in 34m 54s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/64c8de7cebf44d7984e4f68d1efbfead : FAILURE in 1h 42m 04s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/aee76f7b8fc541f985be39609226861c : SUCCESS in 1h 12m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b7c37bc074e545bb90c375845ba56be1 : SUCCESS in 1h 24m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/406db718523f482cb3dd45481e5b2a94 : SUCCESS in 58m 31s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/c5a8f64b540b475982e1938cf062c629 : FAILURE in 1h 03m 41s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/87ff3787a6b9428eaa90f5e6c785a772 : SUCCESS in 38m 41s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/80044e6aba044097bbf4eeb4035f044b : SUCCESS in 1h 34m 00s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/713b590bc8a342888ca8dd10b99372f3 : SUCCESS in 1h 07m 07s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/02658464f7694e7ba3e1e5c91c8c20e2 : SUCCESS in 1h 16m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/2f689bfaedff40649021b10386ca56c8 : SUCCESS in 1h 10m 57s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/a58e35502a2b4974a4a56ad804861628 : FAILURE in 1h 13m 29s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/128ad25000204fb49440cbd711f8a389 : SUCCESS in 34m 50s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/7fc7f6be89c44383a9263ac851ad775a : SUCCESS in 1h 37m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/8b4540f6f3a24dd8a83ca48bd9001d4d : SUCCESS in 1h 22m 15s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/afb0a6cdad5c411a93b5a1d99483b664 : SUCCESS in 1h 11m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/7013d6e70bf542b6befde04d45bcc111 : SUCCESS in 1h 02m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38d1dcba6ff74c8eaad7822b6d306f04 : FAILURE in 5m 25s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/1217d591e5fe47729c7c30d83db1b77a : SUCCESS in 1h 44m 30s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/1fa69012eee64828a7c3a59a3ab55fdf : SUCCESS in 1h 42m 43s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/21ef02fcdd23431da128c33fff187a3f : FAILURE in 31m 29s (non-voting)\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/d081eeb3df994ee0b56fec86ee826732 : SUCCESS in 1h 27m 33s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/218a8285097841c1be0c8e2961b51537 : SUCCESS in 36m 23s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/9a49a111d3534da58b7360d055d4f9aa : SUCCESS in 35m 40s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/fde7dcd870af4a00ab2a8da21dd8ea18 : SUCCESS in 47m 42s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/4f8e012a771a4e04ad1de813e38132c6 : SUCCESS in 40m 56s","accounts_in_message":[],"_revision_number":4},{"id":"0697385a86349aeae4529717e074e8887e3b584a","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-02-23 12:30:17.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a6e624394da3f2a0b4b2f44320f70d26c44cf944","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-02-23 12:37:23.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"8b4049a4665cb24d61e3260644d4cf7a64c286fd","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-02-23 12:58:09.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"b9fd2b978634c3acd4b643918f6c2ce4bca0547f","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-02-23 12:59:26.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"8f3134907ebe0d4e1d75a6d10ecca15fb297e3a0","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-02-23 12:59:55.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c15f9a0a977ffb4b19f0af16d3243099043a3e99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 15:07:57.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c97a5b86c62242949e90f9164b0fc058 : SUCCESS in 5m 17s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e897617900014aa6963e0c8250dff8fc : FAILURE in 1h 05m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1716117a80494aa1882892ccb34eed74 : SUCCESS in 32m 02s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/9e028c6950e64bd6a1fb41ffdb58c885 : FAILURE in 2h 06m 12s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7a7ec927dda843e8b752984f93f432b8 : SUCCESS in 1h 29m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/bd6d1ef0dd90439f89b749f4bbe0deed : SUCCESS in 1h 06m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2d5546e9bee14a60a31b20368164d43b : SUCCESS in 1h 14m 52s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/d327cf3fe10041f0909d2b314ed55295 : FAILURE in 1h 07m 36s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/89bf79f448b54d018a184932f9860217 : SUCCESS in 39m 44s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/1bafd2705e694af7b858743661ac6ffc : SUCCESS in 1h 34m 02s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/22f87cd35d0746809df13ff1e633b01d : SUCCESS in 1h 24m 44s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/c22cdf74d130486bbc18fb1e79ec02aa : SUCCESS in 1h 04m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/ea7e3020237e4f7491cc7a1d294c7aea : SUCCESS in 1h 10m 48s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/1b9b89e2ef6041d586787debf2534edf : FAILURE in 1h 13m 08s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/aa8a0f5ebb1d4d3891ccee7293ecaeec : SUCCESS in 35m 16s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/a0f4ea61e96a4fa9b0a19b58cc333753 : SUCCESS in 1h 38m 01s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/428a27f87e6e4de8b2b0d0cd92a9272b : SUCCESS in 1h 26m 31s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/2be5f49d40ea4b5d9f433916993ff45d : SUCCESS in 1h 02m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/809f9b6b5a5a44cc85757a7e33ad96bd : SUCCESS in 1h 12m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fea9793eeaf4c68b8d3b1a22fe69a4a : FAILURE in 4m 25s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/4e92607b8ef548e9906b60b2ff3219fb : SUCCESS in 1h 47m 41s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/989ed81e43b54c2ba67a62f8e727c795 : SUCCESS in 1h 40m 52s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/618ba467f7b342f6845e58755eff372a : FAILURE in 31m 35s (non-voting)\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/8bd33646b8c448df9bf40009055a27f5 : SUCCESS in 1h 15m 47s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/6a4652e478dd4526bec11a6b3fa28eb1 : SUCCESS in 40m 28s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/d73c421a549343128fd47552839049d2 : SUCCESS in 37m 48s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/0c0460a2753d431da69dd9dc84ec22b1 : SUCCESS in 48m 42s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/25583faf139c46d5a28d4c55d9c88210 : SUCCESS in 39m 36s","accounts_in_message":[],"_revision_number":7},{"id":"ceccc65449035e98d541b66ef624a6866f9b2bec","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-02-23 20:06:52.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"e837736b1abf863e21349d3744c82b31b21e21c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 22:42:04.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a191cd1d51044357bd98ed953f4314ee : SUCCESS in 4m 55s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/09a71ef3d6cc4fd9bae585978b5e5149 : FAILURE in 1h 08m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4a0dc207e171463f9e92193c549c660b : SUCCESS in 32m 39s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/00476e98ae654d2aac308e20e4fe73b9 : FAILURE in 2h 32m 59s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ac9778a8180242fcb50c0ab0520a2fdf : SUCCESS in 1h 18m 49s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1538af12b6704d74a2f094aa1bda8b66 : SUCCESS in 1h 07m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a07cfa3f5dd64cbaae0f20dafce367bf : SUCCESS in 1h 17m 32s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/b031403fe7304fcc858837f5cadd4d2d : FAILURE in 1h 08m 46s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/f5703b7692c74cd1b5a2f328cbd08dfc : SUCCESS in 35m 14s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/cb1fbe2d59c44853ab2bfd3794e87851 : SUCCESS in 1h 42m 59s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/85409eb80aac4a4b98c8c351220b5feb : SUCCESS in 1h 10m 17s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/efe4d835e95d4c12921d025bb3decf95 : SUCCESS in 1h 02m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/506e3c16b72c46a9884095f8d4860432 : SUCCESS in 1h 02m 03s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/013c539a08d34483b5d1a24fa64135ee : FAILURE in 1h 04m 32s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/e699f6eeedf64fc789c171a77c477735 : SUCCESS in 38m 05s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/640b9688eccf4ae8a0919bbca0e9d6c4 : FAILURE in 1h 32m 34s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/9d246d07afd749adbc407a4808ad17f6 : SUCCESS in 1h 11m 20s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/bc1f32895f8843508ce363f3480dfbbc : SUCCESS in 1h 12m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/84f781051dda4a2d9d2b57964e949e15 : SUCCESS in 1h 05m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37f9c15bf15d4123a7458530cd37a004 : SUCCESS in 4m 54s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/743fef8d153b4193abee05889634ae6b : SUCCESS in 2h 03m 00s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/f8b27934d9284771806d6337e95cd922 : SUCCESS in 1h 42m 32s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d9a8c5d8eed44421918bf27927172b1b : FAILURE in 34m 14s (non-voting)\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/1982ebd0d97c44699277a6d28c139243 : SUCCESS in 1h 28m 31s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/3694782d6cd849b2b34c875a2fbb470e : SUCCESS in 38m 20s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/560b190f1bd843be9efd302bbec6e31b : SUCCESS in 39m 20s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/7bf73548a1344e97993aa34e24a06780 : SUCCESS in 50m 36s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/dbcd871d528d4be49818c0db3c84ca11 : SUCCESS in 42m 21s","accounts_in_message":[],"_revision_number":8},{"id":"d047754033ca414cefd483458c5f28318cf70774","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-02-25 17:00:10.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"dd94582a8f4899a1623cac0289d9e54e8cb7d46b","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-02-27 07:44:23.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"a5dcff83daa2e3e3dc127384add46d871cc4bea4","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-02-27 07:46:53.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"892ba76bd8be77d8b37b1218c16b97da82aa1c5f","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-02-27 07:47:48.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"162b73de5f6b0ac0420ce66806e4a189dc8909c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 15:26:02.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/735c7a78989e4536bb1dde1757670f8a : SUCCESS in 6m 03s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e3fc2899e8784255bd45c3c4018b0ebd : RETRY_LIMIT in 32m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2abb85d51fb64509ad5806316589acd9 : SUCCESS in 36m 49s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/8d2faab77a9b45d4bf3ec684aee5b85a : RETRY_LIMIT in 10m 33s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1e7e037e7a5147069cff6de674e6b551 : RETRY_LIMIT in 32m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7e1e18d91ab946b6bc2af266a48ce1dd : RETRY_LIMIT in 33m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1ffa3263c4c14b3c9397084c12e67768 : RETRY_LIMIT in 32m 10s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/f4c6e4d0bd35471ba7d720a4ec2c9d65 : RETRY_LIMIT in 33m 08s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/9eb7ba1a7ae9479a800335c9287dd43a : SUCCESS in 33m 55s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/ff803d9b7e8745f1963160724df6e366 : RETRY_LIMIT in 29m 26s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/3f21ecbe7c564160816bccb0cdcc1545 : RETRY_LIMIT in 32m 08s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/1d4ec40f20464f9ea6101cffe8e3d88f : RETRY_LIMIT in 31m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/84a6a1ea588343b9bc28a440e4dd984a : RETRY_LIMIT in 32m 25s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/b1863327281149e5a1fdc0d967465b94 : RETRY_LIMIT in 46m 02s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/ad89518b4c2f4508bf71b9074ae7e433 : SUCCESS in 53m 58s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/c664a218c1b743048150ac98a692b817 : RETRY_LIMIT in 33m 46s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/5bf7343f5f6c48e0be543a5455aa33cf : RETRY_LIMIT in 31m 04s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/47e75c92fc874a48a15393e8ddaa8fc5 : RETRY_LIMIT in 1h 21m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/173f466fca6140c2b74b395c1cd02476 : RETRY_LIMIT in 32m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a99d31abb7842399615aab22c609f82 : FAILURE in 5m 26s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/409139bd3b4e4af1bf1c89c047939f52 : RETRY_LIMIT in 32m 18s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/7d8b51529c684d0da0e198d453032a5e : RETRY_LIMIT in 32m 56s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/19cd95a2bb4d41408a6de4db0e930524 : RETRY_LIMIT in 42m 06s (non-voting)\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/80bc74ef07b34d738a8d8d06e6ac3bc6 : RETRY_LIMIT in 31m 53s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/5add6a1cf2a348889e7698d3b6afc722 : SUCCESS in 40m 05s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/6a698fe2ed364d248c1874fed2c3e33b : SUCCESS in 40m 10s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/97d7c7b963a44705b16e26c735e4bd66 : SUCCESS in 52m 38s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/3b6f9c7ea11e489f859e9271024bd65b : SUCCESS in 44m 34s","accounts_in_message":[],"_revision_number":11},{"id":"e272534379c104f0fc6298a57dac1065acd3b629","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-03-01 08:26:03.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"cda32ad0196ec8e09b58d22b65f9dd04d9f11dfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-01 11:17:23.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86870eb3d38146188b4aa091a5eb9d6f : SUCCESS in 5m 19s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a889df27c7114f36b23778cd16e5635b : FAILURE in 1h 30m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/373e51f095874e689a001bb059436411 : SUCCESS in 42m 27s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/35445a3ff16649a5bca20fbc03c91b80 : FAILURE in 2h 44m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b82b9d1e4bd049729a906e1a5fd99dc3 : SUCCESS in 1h 31m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8e23d8a24be044dd939f20486b0a4d17 : SUCCESS in 1h 32m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5e682cde2c1a48e09a46cbe8b06c3567 : SUCCESS in 1h 01m 24s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/209386ad1ceb486b8a1450c25fe8b276 : FAILURE in 59m 59s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/dc2e5a5677634122ad6770e5becf2109 : SUCCESS in 39m 52s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/1705bf7281c94e71b07361bc696c68a2 : SUCCESS in 1h 52m 53s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/ea00b52190be4914b2bcb73829a20034 : SUCCESS in 1h 09m 45s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/840d9a1040e14bf8a3d3e4a20c70306b : SUCCESS in 1h 12m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/d302ae0279aa476baf06bd63c5a5481e : SUCCESS in 1h 11m 39s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/6900c0722c0441009d14afa449bec26a : FAILURE in 56m 46s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/6d25c1623d86491a9a7c07f7ff51e265 : SUCCESS in 38m 06s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/6e371f22288144d2ba7a9f66628adffa : FAILURE in 1h 34m 46s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/7d9aed7ef3414c1586c153a8954b72aa : SUCCESS in 1h 07m 22s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/0a76987447a24c2ba961881d82c106ed : SUCCESS in 1h 11m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/c5db346642d84c028ed7aa6627f9e2cb : SUCCESS in 1h 38m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/625a2630d5504bcba09f39d5580e75a7 : SUCCESS in 6m 58s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/70150086628643bd8f5ba3cd8265543b : SUCCESS in 1h 40m 48s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/2e759aca414f49bbb1a3f87946245094 : SUCCESS in 1h 39m 51s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/55946e30ed1842ae8537a83cc7ea5c1c : SUCCESS in 1h 23m 09s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/43b5c1e143ee4eba8ddf57fda0c97060 : SUCCESS in 1h 43m 46s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/59f7ac5e519c4ac1a220f6288aa3597c : SUCCESS in 41m 47s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/bf77868d523f4d558a2ccae32f83ffbb : SUCCESS in 39m 38s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/45aaf7ada427415b9269bf3b5df0f548 : SUCCESS in 50m 14s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/e01a192282364a18bfcc880774bda0ba : SUCCESS in 40m 09s","accounts_in_message":[],"_revision_number":12},{"id":"91252f6c5c5f7b1d8a577714a8d21ad80c80eb8b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-02 11:35:08.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"4e99d852fe8998939c7bb78be94cb621ae9006c7","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-03-03 05:51:18.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"98ce9c01b583e998892760e56df99ef8ae471690","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-03-03 05:53:31.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"34f1960a5ed77c759af6c7223e5e847498c27e72","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-03-03 06:29:46.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"703b0325b6ea2f8f75de8ab1eaaf1ce76498a427","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-03-03 06:46:42.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"483ce8dec9fef341b4a8c4a8062948c829aa70ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 09:06:38.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97ba3bca6b2541b1b728b48f84660722 : SUCCESS in 9m 00s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/300dd8ccb73546f1bc9336523107a418 : FAILURE in 33m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6887b1b79baa41f78bf5fa175a81547a : FAILURE in 36m 25s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/7e7668e1e00249c381508f9498c23ed9 : FAILURE in 57m 07s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c48cc2db0dda430eb5c92a097f7a461f : FAILURE in 36m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/986da6c547b949b99ad5a031330ae898 : FAILURE in 37m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dc0c4912b41946079337faad6e27ff8b : FAILURE in 42m 03s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/08e150d8e7d148ed9e69620739809ec6 : FAILURE in 41m 20s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/8a0356ac9c404b17818971c67be2c5f0 : FAILURE in 43m 17s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/22a53809b3ae48afadd8ce187b565d8f : FAILURE in 49m 22s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/9a955f7acd06452da11c34c463b68617 : FAILURE in 35m 58s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/03962e2b66da4c50b99398927fbdf6c5 : FAILURE in 37m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/bb1fa0f1e4c54ad9aa2df344dbd65cb9 : FAILURE in 37m 37s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/f82204a08a6b4a6387bec17b3833f218 : FAILURE in 36m 18s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/7d449277cb694faea6b52add2c045306 : FAILURE in 44m 00s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/de807cc799304d2daa6bbc488204863a : FAILURE in 1h 01m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/92bd04d8f5e3421999d05c236dcb2954 : FAILURE in 37m 57s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/6b8e060a076246b2afc4ea20ae468a3d : FAILURE in 33m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/795b65dc46964f6680abaa8bfeb758cf : FAILURE in 34m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d72aff3f16c4451eba8bc7936bdf2e9c : FAILURE in 6m 17s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/30e1cc9cf90849be84e2887d832fa814 : FAILURE in 43m 39s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/57878ba5875b4a7399399eaa58d661e4 : FAILURE in 42m 49s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/22852b33df3f4b799b3369b737112e07 : FAILURE in 41m 07s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/454aac0e5e7441e9ad59113ed9fac8cb : FAILURE in 42m 01s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/6956771cf40a4bb9bc3c770c319c60f6 : FAILURE in 43m 14s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/5eed50102aae4468a3235fd05f461169 : FAILURE in 44m 04s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/de4f9b030df9406c8a8ece260d45db4d : FAILURE in 42m 35s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/50063b7746fc4f44b1806776dcbe7bac : FAILURE in 40m 09s","accounts_in_message":[],"_revision_number":15},{"id":"c86450c1c09cc6a42f40722ad26d28a7d1e6986a","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-03-03 10:36:41.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"f924fb11f29b0b314b5533a55c9daa40e8c45cab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 13:32:42.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e144326cad854ee29d15a1359efa397c : SUCCESS in 6m 21s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/39b1dce640634d27aeb3483a21c4dc22 : FAILURE in 1h 05m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/967b85fa8ce1420e8f24168cfabc4ed9 : SUCCESS in 42m 03s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/0c3f29b8fc5d42e4b99e39c5df706f07 : FAILURE in 1h 51m 01s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4c902bcb45694c148da3cbaf883475f9 : SUCCESS in 1h 06m 38s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1289196621fe4959b4b50b36b457e2e2 : SUCCESS in 1h 04m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b7561adfde2c48cdbedec7d3981d341a : SUCCESS in 1h 15m 23s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/43a73f8d935f433ebb83ee13b25037e0 : FAILURE in 49m 14s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/59cc7a4b77c649c8ba5d774b41d76684 : SUCCESS in 33m 41s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/6a2b5b124f4145b88a8640c8ca81754b : SUCCESS in 1h 32m 03s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/713364eb60ef4fc8b00f83f03dc74235 : SUCCESS in 1h 34m 22s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/478af50d97894e1a8d6c1df9631c3654 : SUCCESS in 1h 07m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/8b510827363341b79c5fc0202ed309e9 : SUCCESS in 1h 16m 07s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/8d501bc4a3034f17855cb092a34def7c : FAILURE in 1h 02m 26s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/ab1246d723d34eb7861774e820041177 : SUCCESS in 36m 32s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/65deef45243145a3b5d545dd02fd79a8 : SUCCESS in 1h 32m 55s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/f6a029c6572a4c93a76cd337ed787bcb : SUCCESS in 1h 08m 08s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/f38269463c944846909f1a35f86257cf : SUCCESS in 1h 12m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/eea0317f71f0424cb35d6fd1706e3078 : SUCCESS in 1h 17m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e64e598868ba4e4a9c8d35dd526a2eb2 : FAILURE in 4m 42s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/3e1880ccad214bc6baebb05a18d9cf04 : SUCCESS in 1h 35m 07s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/5c55e64942954b6e8477fc3be392226a : SUCCESS in 1h 38m 33s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/39fdc57df11c43deb6ee9c7dbd386644 : SUCCESS in 1h 29m 11s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/7b189378ec944f0b8f43ebfa9dc107cf : SUCCESS in 1h 34m 04s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/fb3069e7132d419e9050696485380fb4 : FAILURE in 40m 44s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/e59bb7fe0e804adaaad0678dc423b9a3 : FAILURE in 45m 28s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/ba8f180112cb42fcac7e03cd49e318b6 : SUCCESS in 45m 38s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/9d938b99de61406289cde1f8fbaae347 : SUCCESS in 37m 13s","accounts_in_message":[],"_revision_number":16},{"id":"85afa5db6fbfc4e33ba01126223f25759f9cab37","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-03-04 15:47:21.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"ffbe6a44a81061c601a7d753374f607889003a69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 20:48:55.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41a10dc2efa84c58955555ad6604d5ad : SUCCESS in 4m 09s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/023409d1c147451eb0a6903846a4c81f : FAILURE in 56m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/262c95ac9e98491e8c38323841c328ba : SUCCESS in 39m 51s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/ed072052ca934dc4b58b45135213f521 : SUCCESS in 1h 31m 16s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4ffc0d68e772428f9696e60b0068e891 : SUCCESS in 1h 15m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9fbd2410e4fa4b8b98ecffe169029f80 : SUCCESS in 1h 03m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1db4570b32fe45dfae1a65941ebbc67f : SUCCESS in 1h 15m 14s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/ef667916938f40d982f036db4d6ead47 : FAILURE in 52m 05s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/babb53641abf4227b131124e272684c2 : SUCCESS in 38m 59s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/e0c8d00ec5234aff83a2a20067cce37a : FAILURE in 1h 51m 57s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/92088109fff242b2b0ad4db0bcb37289 : SUCCESS in 1h 21m 17s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/a97ea95bb49a4a26915d5b0a0b025171 : SUCCESS in 1h 09m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/5e71c81689324c7e8dec4e1ab9563ccc : SUCCESS in 1h 01m 06s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/9527c1acb861496aa99b7fb9d81a939c : FAILURE in 49m 17s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/434d2bfcfd044907a3c812e2f9e4bc02 : SUCCESS in 38m 12s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/e851f134e15546478f6a6cfd4d846afe : FAILURE in 1h 50m 35s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/c2fa8fbca0e848ee9530644056150d60 : SUCCESS in 1h 10m 41s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/b1ce03ebbb194a1ba831db93629b7c16 : SUCCESS in 1h 23m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/4afb95449a25417faf1c0b17a4ebbfe4 : SUCCESS in 1h 15m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5977e2eafb541dbb956d24017471e69 : SUCCESS in 6m 02s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/5e598cf58e3d4040b11aeb13b08603c9 : SUCCESS in 1h 32m 18s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/a2fa0f03afb342e38aa49f5e94b05826 : SUCCESS in 1h 38m 13s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/8cb8458e6b104f178c0cb0df4c9e0ac4 : SUCCESS in 1h 25m 51s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/c230795341264254a262ceb8a0c839de : SUCCESS in 1h 35m 37s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/8ef8b44dc94b459f9f20e4371019f191 : SUCCESS in 38m 43s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/07a9fcaf68b14943b30b441f23d41f70 : SUCCESS in 43m 42s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/aa36b1f089f8487dae7b2349461d9159 : SUCCESS in 52m 09s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/04eb84f773cc4476ae057ed671c5ca34 : SUCCESS in 45m 28s","accounts_in_message":[],"_revision_number":17},{"id":"32012024444cfa23e95541989b77b16d96c135d1","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-03-05 07:37:28.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"48ed3f19c79c2bafb37becf33c36c109866b8a72","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-03-05 07:46:14.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"2b6f83579d25b54e74854b23a1ce14b9810efc66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 09:53:41.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fca6f4408df44f0992e5065f1d51b2ef : SUCCESS in 4m 49s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9da7fae56a0746a99aae2340344450b9 : FAILURE in 1h 20m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c206542ece0644c79beef7343139a67c : SUCCESS in 47m 14s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/62780cd8f8f5469989140c48b2fcc246 : FAILURE in 1h 50m 11s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0b0e10e2182d4367a65149f179e8903d : SUCCESS in 1h 12m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b8456eafaa90491eac65590afb097139 : SUCCESS in 1h 24m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2eb8ec67e400431d8c5424c232585fba : SUCCESS in 1h 10m 44s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/9b7443c659364364a717a174b6f6cebe : FAILURE in 1h 02m 26s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/c0b1d6b361ad46f088846f1a784df10e : SUCCESS in 37m 39s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/b094be2cf820442cbb3ab1f7eea692e7 : FAILURE in 1h 52m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/f2403cc038704ff28aa3317725ac4b34 : SUCCESS in 1h 09m 05s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/a5174cd59d08404f88d98bc3348d66b5 : SUCCESS in 1h 13m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/ab0d86c1652c42f58ba4b7a711805ebd : FAILURE in 1h 04m 53s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/ca64f72969034ef68a87a64d9019bc72 : FAILURE in 1h 12m 00s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/d088e11cb8a6449084625ccf28dbc7d5 : SUCCESS in 37m 51s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/403dbedc19ac45ed9716b5a748b7ced8 : SUCCESS in 1h 34m 55s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/ed005dc8ebff4cc282b2a3d432600ae8 : FAILURE in 2h 01m 28s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/dacbaddbd63c446696b7860e3d602042 : SUCCESS in 1h 33m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/9ffe600b197d48c6a6aadcb5618b96f0 : SUCCESS in 1h 09m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a285f414c1748b1aba6ddfedd9f3d62 : SUCCESS in 4m 12s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/f3205fd6e9bc4a269d9595c062f3b436 : SUCCESS in 2h 12m 57s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/36de275bfc90451db49d119fac7dca37 : SUCCESS in 1h 41m 51s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c94b4305931d4b1da1399a0c9c07548d : SUCCESS in 1h 29m 34s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/4c46e4c06a224d3599bcad05a8281bba : SUCCESS in 1h 37m 50s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/a1ecc63670c44c09bcbecbbbd3900b10 : SUCCESS in 39m 46s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/021400f3b8aa4d90870790e25ce4c700 : SUCCESS in 37m 35s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/76b34c5cad3d44cf85137d1f9fede6f8 : SUCCESS in 52m 12s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/767ee216e5704a4c9e0196941b328441 : SUCCESS in 39m 52s","accounts_in_message":[],"_revision_number":18},{"id":"6c047c7c52816a582153b4cf0631a2ed53877095","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-03-06 16:25:12.000000000","message":"Patch Set 18:\n\nrecheck Length too long: 7024536","accounts_in_message":[],"_revision_number":18},{"id":"ea95184778e13f2485c1bcb32d358c04b076fda9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 18:51:55.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba837c4c82a744bfb3220900fb38c6c6 : SUCCESS in 5m 35s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2022efffee484482ab5547e0b1b8d61a : FAILURE in 1h 09m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/849f8891c572490ab33dba872a6bccc1 : SUCCESS in 36m 23s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/dc59847f2e12461cae02a35941b8f480 : FAILURE in 2h 17m 16s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2892eeac2973435da600e748dfd65e50 : SUCCESS in 1h 24m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9a2928cc15d44f26938aba0932833f91 : SUCCESS in 1h 10m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/331073314e304dcfab95876eefe92269 : SUCCESS in 1h 14m 02s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/76531840a43248beae71296fe350eb5b : FAILURE in 1h 07m 57s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/73cc284f981e451387e92b3e44c7e9f5 : SUCCESS in 33m 31s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/6feafa3b65d144dd933a10b39b48c42b : FAILURE in 1h 24m 15s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/1ab41995e0dd4633a849fc0a5b5140b5 : SUCCESS in 1h 12m 18s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/62ce283ebbbe48fcae14e43bf623afb4 : SUCCESS in 1h 07m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/6c30ac62ae0647b1bf588cf8f2753ac5 : SUCCESS in 1h 08m 05s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/136e9bd172dd49a5a972db29962e9913 : FAILURE in 1h 14m 32s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/e5827e67117e4dde94d3c0c8a4b4d302 : SUCCESS in 40m 47s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/d1bd5bc6774345fbb4addc550359d964 : FAILURE in 1h 55m 39s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/e2e85fc0e2754403ba7f0ef75c87b6d7 : SUCCESS in 1h 23m 00s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/fa583d38a0d84cbdbac494c345e908cb : SUCCESS in 1h 08m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/c4f90d12cb44422eb702ef58c4b3ffd6 : SUCCESS in 1h 05m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe55271600734192b3fb4408cdd84be0 : SUCCESS in 6m 22s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/74627174629442f8817a7792834ec80c : SUCCESS in 1h 43m 25s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/40b9cef2cded441384ebf266e24e6513 : SUCCESS in 1h 45m 03s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7862cfc2d29e432ea53e0ad3544fa4de : SUCCESS in 1h 33m 23s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/70615b79c8d6470fb144cd68a74a4e83 : SUCCESS in 1h 28m 36s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/726a26e28cee4257b591f4affdeeae3e : SUCCESS in 42m 14s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/d6f20446c7494cfa8a5f39b517042870 : SUCCESS in 40m 17s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/649e599fea684877a75ce1d7aa3a005f : SUCCESS in 52m 39s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/f321a03589df44459b76d86a291f56c6 : SUCCESS in 39m 29s","accounts_in_message":[],"_revision_number":18},{"id":"ca2efb85e06a5c228ab73772a8b96b6922a76be5","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-03-09 11:18:05.000000000","message":"Patch Set 18:\n\nRunning the test locally it passes fine. My theory is we hit a logging limitation like we had some time ago (where \"fix\" was to limit logging).\n\nThis is visible when comparing job-output.txt size, 5.6M for a passing recent one, compared to 14M for the failing one here.\n\nMost of it I think is created by the API calls to create SG rules:\n2020-03-06 17:33:05.985365 | controller |             Body: {\"security_group_rule\": {\"project_id\": \"1fb4765510ad4b3fb10b9ad421fbf757\", \"port_range_max\": 3159, \"port_range_min\": 3159, \"protocol\": \"tcp\", \"ethertype\": \"IPv4\", \"remote_ip_prefix\": \"0.0.0.0/0\", \"direction\": \"ingress\", \"security_group_id\": \"a882371d-f86e-44ec-a772-d963903052c5\"}}\n\nI see 2 ways of fixing it: disabling logging for these helper API calls in the test, or lower quotas for it.\n\nThat later option sounds good to me, no need to exercise neutron-server in API tests and we do not need 1000s of security groups to test the quota increase","accounts_in_message":[],"_revision_number":18},{"id":"f00197c49a5b625aca72a4f699366cadaa996791","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-03-09 12:41:02.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"2dc6bf61a81e90d002fe8a4a38c482306312b4d0","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-03-09 13:04:00.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"4379c56682df3f480f61829aad025ed5774c3b26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 15:14:01.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f8ccc5e6e7445b5a7d517879f7bb3db : SUCCESS in 5m 14s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/42de89e2ec0a4379bd560f70d3ac1919 : SUCCESS in 1h 01m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2c107248cbad46248506a78731cdde21 : SUCCESS in 34m 56s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/a162f48adc82465aa4392c5bfbca9b63 : FAILURE in 1h 52m 40s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f670eff641184d72a7456d488593c564 : FAILURE in 2h 01m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ba35e4e260a2497089f6aa58742a0a7f : SUCCESS in 1h 50m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/318f120d86ba4d99974c1c0f48b97bc7 : SUCCESS in 1h 01m 03s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/64bafffcb9d8449ba553820d11e3f6f4 : SUCCESS in 1h 06m 24s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/677d05da39da4f2b81d527be39900ba0 : SUCCESS in 39m 34s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/0ba2bb8badac4ed394fea57c4b3cc222 : FAILURE in 1h 47m 37s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/f08a051b2ca64a809c4b0c847a3fe6e4 : SUCCESS in 1h 37m 35s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/a037a842386847c19f4246fd4bc1a040 : SUCCESS in 1h 19m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/f237e5e2a8d9444fbf7dbbaf71122543 : SUCCESS in 1h 31m 27s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/bb3caac9bf414104ad08ab8704d62267 : SUCCESS in 49m 14s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/0c99151e0abb476a96fc0914aa3386d2 : SUCCESS in 34m 01s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/08b7170496bd4d37bc01cc26ee98a1c0 : FAILURE in 1h 37m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/345731808bb44b598d18ad8ff3a35978 : SUCCESS in 1h 56m 47s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/49330722eaf949cfa462e1779bf2451d : SUCCESS in 1h 06m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/6186d717ce1749f3bfc4600435f7da44 : SUCCESS in 54m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92da8587e2ae4a6aace96fd9275eacde : SUCCESS in 4m 25s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/0e5bbfc6080a4e1881bf5131dd7e7d7c : SUCCESS in 1h 59m 28s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/3d3a673e162043c6bf1ffb4340dd0216 : SUCCESS in 1h 46m 31s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/6a453f71f3c14e5cab461f9ebec2d252 : SUCCESS in 1h 30m 38s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/e6d4ca1c75b241edaef601ce1a772456 : SUCCESS in 1h 35m 39s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/4d94fac2e2ee4e8c816f22760fb22ee3 : SUCCESS in 49m 50s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/2af3d28a48e249828480007f9bcef2a8 : SUCCESS in 36m 51s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/aacfaf24819747d08eba60a2342232f6 : SUCCESS in 51m 00s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/ae77c8f23c864f14841a46cc567506ac : SUCCESS in 36m 26s","accounts_in_message":[],"_revision_number":20},{"id":"9a55f824959dc0f0294f0ba02135ab8d1c83cc43","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-03-09 15:28:13.000000000","message":"Patch Set 20:\n\nrecheck linuxbridge","accounts_in_message":[],"_revision_number":20},{"id":"84b15debdbed91408ddd621da77a0e46546be71d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 17:25:56.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/613b23fcf6c042bdb5f1ad677a61c075 : SUCCESS in 5m 37s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/17233763843f45febedcd818d8ba813d : SUCCESS in 49m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bf9e92c38399443697841590d9f39223 : SUCCESS in 37m 16s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/d6161a741af14d1891d4cb58a2265a63 : FAILURE in 1h 42m 24s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/98f3374d8c1d4171b363dae0b5b39ba7 : SUCCESS in 1h 16m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2ccfbd62ae8b457c8cd0b9f47c02bf83 : SUCCESS in 1h 10m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b9b5a77eb074445197a4a1f18fab3244 : SUCCESS in 1h 03m 58s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/2640294abff34099afdc7f73c1abdbc5 : SUCCESS in 55m 35s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/1e3e1c2a6d7d42c3b5282c825317d2c3 : SUCCESS in 35m 57s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/5e46105c71914f6991315d49faad4c3f : FAILURE in 1h 40m 57s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/2a396b0aef92482caaa40dbf1fa56394 : SUCCESS in 1h 10m 32s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/98e28bbf2a5b4c8fbbc9393b8e924716 : SUCCESS in 1h 23m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/99dda84d252b4892a89cd388e9c1e234 : SUCCESS in 1h 04m 25s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/c5fd6d022bdc43bab489506cc3f406a2 : SUCCESS in 55m 08s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/9c17bd789c5e4d95a85b476d5da0763c : SUCCESS in 33m 55s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/c6831997849640a897504048856f0044 : SUCCESS in 1h 19m 41s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/7c79d6c9cc31481cb634db15ddcd63c6 : SUCCESS in 1h 23m 43s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/f1a87dd7e79a49499787b124bdcf71f0 : SUCCESS in 1h 00m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/c1e11336e18649eaa43840396e9cd316 : SUCCESS in 1h 38m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2e5c46c91b5474dafcd0f77bf27d2e2 : SUCCESS in 5m 48s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/35efe3ca1c7b4cc8948cb8ac2a07fea0 : SUCCESS in 1h 53m 53s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/0a1ceff254a64a2191e34170a88523d0 : SUCCESS in 1h 54m 14s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/88d3d25517c64e20a5d8562173631354 : SUCCESS in 1h 37m 18s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/e23094332fba4a5b85fd33bc1519a14d : SUCCESS in 1h 21m 24s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/6031970ae8de4918b127eec209dd4003 : SUCCESS in 43m 25s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/bdfe83ba17a94b08aa9f05e02ed35c18 : SUCCESS in 40m 51s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/ffa98fe21d314b29b7f5e8ffbe79ebf0 : SUCCESS in 51m 55s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/05699d35c6554d7f804683906d0dff13 : SUCCESS in 39m 02s","accounts_in_message":[],"_revision_number":20},{"id":"d15fdfc178ca93b11938671e6d4f64fdb3e88359","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-03-09 19:45:00.000000000","message":"Patch Set 20: Code-Review+1\n\nLooks good to me and passing fine in gates with lower initial quota (also faster to test and fine for an API test)","accounts_in_message":[],"_revision_number":20},{"id":"d1647caea03232af6e2b9488a7fd0f2e45727d01","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-10 09:40:28.000000000","message":"Patch Set 20: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"021651adff17d288b80365fd31f77db1cdd3aab7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-03-10 12:01:02.000000000","message":"Patch Set 20:\n\n(1 comment)\n\nCode looks OK apart from this comment from Slawek. But the code is almost ready.","accounts_in_message":[],"_revision_number":20},{"id":"17c4733aaa5da7d901af922d66e10ec2dd4829e4","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-03-10 12:42:08.000000000","message":"Patch Set 21: Published edit on patch set 20.","accounts_in_message":[],"_revision_number":21},{"id":"b64f10f6f855a46b065a024cc7ce5ee0fe9d297c","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-03-10 12:42:54.000000000","message":"Patch Set 20:\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"4d16f82e41f091efe81a093e3eaede45e44a3057","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-10 13:09:30.000000000","message":"Patch Set 21: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"51c8f6fc2dd8c93929f2632a918f0b7e971e7acd","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-03-10 13:29:18.000000000","message":"Patch Set 22: Published edit on patch set 21.","accounts_in_message":[],"_revision_number":22},{"id":"856e31dc7b1fb1c2472d4ba7abec132703abb0f0","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-03-10 13:35:26.000000000","message":"Patch Set 23: Published edit on patch set 22.","accounts_in_message":[],"_revision_number":23},{"id":"c971122c31346ece67a9d9c2e3e6988eb12f919b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-10 13:36:47.000000000","message":"Patch Set 23: Code-Review+2\n\nseems fine for me now. thx :)","accounts_in_message":[],"_revision_number":23},{"id":"682c0634b84dc759d1b4a0c12335258a3697dbab","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-03-10 15:42:43.000000000","message":"Patch Set 23: Code-Review+1\n\nand test passing in API jobs","accounts_in_message":[],"_revision_number":23},{"id":"936998ad4e8643846f3694b82e51201c15904a04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 15:56:14.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af01519dd2f44354943f2a205f8f09f2 : SUCCESS in 5m 53s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8546f764094141b2ac30016bbe3ec38d : SUCCESS in 1h 01m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/462ba8cebbe84257bc7485379052b7b4 : SUCCESS in 41m 07s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/910f6da29f46439e87b2dbefae0470f9 : FAILURE in 1h 50m 02s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e81e202aec0244d2b0fb7009c0c69c55 : SUCCESS in 1h 40m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/37a35e35be514bb789d7fb52f3286550 : SUCCESS in 1h 29m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/54d5e8f410a647cd91382c3e70ce8c86 : SUCCESS in 1h 08m 56s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/ce4937a430134943ac79e3ca7661740a : SUCCESS in 1h 01m 02s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/080702b929ff469c80fc7dd0f030ab22 : SUCCESS in 38m 31s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/9a23c74cd5b74fc393def68f28a02a53 : FAILURE in 1h 46m 12s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/e469cdaa44284000a7c38f5e16cac603 : SUCCESS in 1h 44m 14s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/cb3058fe551c403da83eb5388e2a58f4 : SUCCESS in 54m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/38b448e8287846fc93144c8ef824316d : SUCCESS in 1h 42m 25s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/c0564cdf0ece4dd89f2780629a133102 : SUCCESS in 1h 00m 10s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/67f9ef65e4bf4f98ad0b49c409cd722f : SUCCESS in 41m 20s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/00cb2cb7f71b4bba997c9567a4a9e633 : SUCCESS in 1h 41m 13s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/7fec3d56127c400bb948e7d87cd2505c : SUCCESS in 1h 30m 38s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/13cc877542884702b9364cd9eda00527 : SUCCESS in 1h 05m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/9910c216ef2d4ec28f91c6dfda99b8b6 : SUCCESS in 1h 04m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d5ca0594b914702a016f8b591973bf8 : SUCCESS in 4m 59s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/3588b96bac4d45bda007760f4d12af95 : SUCCESS in 1h 45m 29s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/0ae6bc98671b4f8aa949dbd2ae41fcb8 : SUCCESS in 1h 49m 55s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/208b0f1b655e4ea38b02c3abeef32910 : SUCCESS in 1h 23m 54s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/345d2ae2f6d24b90aaeb6c454c3a3f47 : SUCCESS in 1h 26m 08s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/a98c92ecb5ed41ef87485fb66a09d9d7 : SUCCESS in 36m 36s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/39b52a20121f4588814ffecb9b0cd1d9 : SUCCESS in 37m 22s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/3a9f03f2dd5c45d3b9e19fdfc66f81ec : SUCCESS in 55m 13s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/24f49d7cb0bd4371b7ac83592eede39d : SUCCESS in 36m 55s","accounts_in_message":[],"_revision_number":23},{"id":"e45dfb53f74951dae2d06a56ef19be2457f6c7fc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-03-11 10:20:01.000000000","message":"Patch Set 23: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":23},{"id":"755e8c26319e999d687cb205cee0cd17263cb652","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 10:20:13.000000000","message":"Patch Set 23: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":23},{"id":"c601f1a9a632afec9e01fda3f82ba47dbbcf546a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 11:56:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":23},{"id":"ef31a08856a21ef05c605def71a1ca557181ed51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 11:56:57.000000000","message":"Patch Set 23: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bece285615bc467f99910bfa8f9ca1c7 : SUCCESS in 5m 17s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c490ac6903ba4d5c8d02932cffdfed09 : SUCCESS in 58m 41s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e7e140c105634155bba6fe0e56c7cc25 : SUCCESS in 1h 12m 21s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5ac0c519e16940bab391b0fb4b5bf85d : SUCCESS in 1h 13m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2bf9ffc2db2042e093ad180514645827 : SUCCESS in 1h 02m 02s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/60b4e2bdca8341f6b9d1b67d70516045 : SUCCESS in 55m 17s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/8c3d60ed0fa74a549f00d68731cf62fc : SUCCESS in 51m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73626aafbabd4ccbaed8a8b37fa0cddf : SUCCESS in 4m 00s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/d4837a5d2a9647a3a6346b871dacdf7c : SUCCESS in 1h 35m 04s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/31deab8693ea46caa9d26ac84fefe8cb : SUCCESS in 1h 23m 19s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/1dbfe1fc833e40adbc3aa96d3b36fcd8 : SUCCESS in 48m 13s","accounts_in_message":[],"_revision_number":23}],"current_revision_number":23,"current_revision":"4e329cf99c8af6092a1e47488dce91d2dd1f4e94","revisions":{"748db7ba56e5f990328a6ee76839104efcbc933c":{"kind":"REWORK","_number":1,"created":"2020-02-20 10:24:15.000000000","uploader":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"ref":"refs/changes/42/708842/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/42/708842/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/1"}}},"commit":{"parents":[{"commit":"af2bb87f08a4f6d7f9bad88d9b69733adbfc3112","subject":"Helper class for BaseSecGroupRulesQuota tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/af2bb87f08a4f6d7f9bad88d9b69733adbfc3112"}]}],"author":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"committer":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"subject":"Adding new test of security group rules quota increased.","message":"Adding new test of security group rules quota increased.\n\nThe test is checking that it is possible to create\nnew security group rules after increasing the quota parameter.\n\nChange-Id: Ie961293215c48883c962debe888b0bca418e3eb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/748db7ba56e5f990328a6ee76839104efcbc933c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/748db7ba56e5f990328a6ee76839104efcbc933c"}]},"branch":"refs/heads/master"},"305816e9e711e2df2d3fa9f109989ce48ec1637a":{"kind":"REWORK","_number":2,"created":"2020-02-20 11:27:07.000000000","uploader":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"ref":"refs/changes/42/708842/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/42/708842/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/2"}}},"commit":{"parents":[{"commit":"af2bb87f08a4f6d7f9bad88d9b69733adbfc3112","subject":"Helper class for BaseSecGroupRulesQuota tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/af2bb87f08a4f6d7f9bad88d9b69733adbfc3112"}]}],"author":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"committer":{"name":"Eran Kuris","email":"ekuris@redhat.com","date":"2020-02-20 11:27:02.000000000","tz":0},"subject":"Adding new test of security group rules quota increased.","message":"Adding new test of security group rules quota increased.\n\nThe test is checking that it is possible to create\nnew security group rules after increasing the quota parameter.\n\nChange-Id: Ie961293215c48883c962debe888b0bca418e3eb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/305816e9e711e2df2d3fa9f109989ce48ec1637a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/305816e9e711e2df2d3fa9f109989ce48ec1637a"}]},"branch":"refs/heads/master"},"a687da9a2fae3756476f259fb6136887cef91b6c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-02-23 10:27:50.000000000","uploader":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"ref":"refs/changes/42/708842/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/42/708842/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/3"}}},"commit":{"parents":[{"commit":"a022dc3be21c073a70558bb51e0e0d3efb748516","subject":"Helper class for BaseSecGroupRulesQuota tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a022dc3be21c073a70558bb51e0e0d3efb748516"}]}],"author":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"committer":{"name":"Eran Kuris","email":"ekuris@redhat.com","date":"2020-02-23 10:27:50.000000000","tz":0},"subject":"Adding new test of security group rules quota increased.","message":"Adding new test of security group rules quota increased.\n\nThe test is checking that it is possible to create\nnew security group rules after increasing the quota parameter.\n\nChange-Id: Ie961293215c48883c962debe888b0bca418e3eb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a687da9a2fae3756476f259fb6136887cef91b6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a687da9a2fae3756476f259fb6136887cef91b6c"}]},"branch":"refs/heads/master"},"9c1d18b96b035b0f00edf242426e414463af3551":{"kind":"REWORK","_number":4,"created":"2020-02-23 10:29:44.000000000","uploader":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"ref":"refs/changes/42/708842/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/42/708842/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/4"}}},"commit":{"parents":[{"commit":"a022dc3be21c073a70558bb51e0e0d3efb748516","subject":"Helper class for BaseSecGroupRulesQuota tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a022dc3be21c073a70558bb51e0e0d3efb748516"}]}],"author":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"committer":{"name":"Eran Kuris","email":"ekuris@redhat.com","date":"2020-02-23 10:29:34.000000000","tz":0},"subject":"Adding new test of security group rules quota increased.","message":"Adding new test of security group rules quota increased.\n\nThe test is checking that it is possible to create\nnew security group rules after increasing the quota parameter.\n\nChange-Id: Ie961293215c48883c962debe888b0bca418e3eb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9c1d18b96b035b0f00edf242426e414463af3551"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9c1d18b96b035b0f00edf242426e414463af3551"}]},"branch":"refs/heads/master"},"1facdcb248a466419badc3fe45ffe505d4471163":{"kind":"REWORK","_number":5,"created":"2020-02-23 12:37:23.000000000","uploader":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"ref":"refs/changes/42/708842/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/42/708842/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/5"}}},"commit":{"parents":[{"commit":"a022dc3be21c073a70558bb51e0e0d3efb748516","subject":"Helper class for BaseSecGroupRulesQuota tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a022dc3be21c073a70558bb51e0e0d3efb748516"}]}],"author":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"committer":{"name":"Eran Kuris","email":"ekuris@redhat.com","date":"2020-02-23 12:35:20.000000000","tz":0},"subject":"Adding new test of security group rules quota increased.","message":"Adding new test of security group rules quota increased.\n\nThe test is checking that it is possible to create\nnew security group rules after increasing the quota parameter.\n\nChange-Id: Ie961293215c48883c962debe888b0bca418e3eb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1facdcb248a466419badc3fe45ffe505d4471163"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1facdcb248a466419badc3fe45ffe505d4471163"}]},"branch":"refs/heads/master"},"f139a07ed499b47ac1a9cd03426b89d76c96a017":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-02-23 12:58:09.000000000","uploader":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"ref":"refs/changes/42/708842/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/42/708842/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/6"}}},"commit":{"parents":[{"commit":"69432e004917ab42b8af9391d39120dfc58c257d","subject":"Helper class for BaseSecGroupRulesQuota tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/69432e004917ab42b8af9391d39120dfc58c257d"}]}],"author":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"committer":{"name":"Eran Kuris","email":"ekuris@redhat.com","date":"2020-02-23 12:58:09.000000000","tz":0},"subject":"Adding new test of security group rules quota increased.","message":"Adding new test of security group rules quota increased.\n\nThe test is checking that it is possible to create\nnew security group rules after increasing the quota parameter.\n\nChange-Id: Ie961293215c48883c962debe888b0bca418e3eb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f139a07ed499b47ac1a9cd03426b89d76c96a017"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f139a07ed499b47ac1a9cd03426b89d76c96a017"}]},"branch":"refs/heads/master"},"9a283c3d58c17e803d4367e5f2dbf06ae679fcf4":{"kind":"REWORK","_number":7,"created":"2020-02-23 12:59:26.000000000","uploader":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"ref":"refs/changes/42/708842/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/42/708842/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/7"}}},"commit":{"parents":[{"commit":"69432e004917ab42b8af9391d39120dfc58c257d","subject":"Helper class for BaseSecGroupRulesQuota tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/69432e004917ab42b8af9391d39120dfc58c257d"}]}],"author":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"committer":{"name":"Eran Kuris","email":"ekuris@redhat.com","date":"2020-02-23 12:59:19.000000000","tz":0},"subject":"Adding new test of security group rules quota increased.","message":"Adding new test of security group rules quota increased.\n\nThe test is checking that it is possible to create\nnew security group rules after increasing the quota parameter.\n\nChange-Id: Ie961293215c48883c962debe888b0bca418e3eb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9a283c3d58c17e803d4367e5f2dbf06ae679fcf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9a283c3d58c17e803d4367e5f2dbf06ae679fcf4"}]},"branch":"refs/heads/master"},"74d8d5660b82724eaf91ebfe695887f092a9a6b1":{"kind":"REWORK","_number":8,"created":"2020-02-23 20:06:52.000000000","uploader":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"ref":"refs/changes/42/708842/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/42/708842/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/8"}}},"commit":{"parents":[{"commit":"69432e004917ab42b8af9391d39120dfc58c257d","subject":"Helper class for BaseSecGroupRulesQuota tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/69432e004917ab42b8af9391d39120dfc58c257d"}]}],"author":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"committer":{"name":"Eran Kuris","email":"ekuris@redhat.com","date":"2020-02-23 20:06:24.000000000","tz":0},"subject":"Adding new test of security group rules quota increased.","message":"Adding new test of security group rules quota increased.\n\nThe test is checking that it is possible to create\nnew security group rules after increasing the quota parameter.\n\nChange-Id: Ie961293215c48883c962debe888b0bca418e3eb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/74d8d5660b82724eaf91ebfe695887f092a9a6b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/74d8d5660b82724eaf91ebfe695887f092a9a6b1"}]},"branch":"refs/heads/master"},"63a6fe412a714677c1d9e10cb6144b70b9693fb2":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-02-27 07:44:23.000000000","uploader":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"ref":"refs/changes/42/708842/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/42/708842/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/9"}}},"commit":{"parents":[{"commit":"123befaeeccddeb30a28e5c581ced2c0771ff7a4","subject":"Helper class for BaseSecGroupRulesQuota tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/123befaeeccddeb30a28e5c581ced2c0771ff7a4"}]}],"author":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"committer":{"name":"Eran Kuris","email":"ekuris@redhat.com","date":"2020-02-27 07:44:23.000000000","tz":0},"subject":"Adding new test of security group rules quota increased.","message":"Adding new test of security group rules quota increased.\n\nThe test is checking that it is possible to create\nnew security group rules after increasing the quota parameter.\n\nChange-Id: Ie961293215c48883c962debe888b0bca418e3eb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/63a6fe412a714677c1d9e10cb6144b70b9693fb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/63a6fe412a714677c1d9e10cb6144b70b9693fb2"}]},"branch":"refs/heads/master"},"06634163c0ab599d2b5e25948a3c94bcf63946a0":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-02-27 07:46:53.000000000","uploader":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"ref":"refs/changes/42/708842/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/42/708842/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/10"}}},"commit":{"parents":[{"commit":"dee69e0f5c1eb5693355768aa8accc4cc2676820","subject":"Helper class for BaseSecGroupRulesQuota tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/dee69e0f5c1eb5693355768aa8accc4cc2676820"}]}],"author":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"committer":{"name":"Eran Kuris","email":"ekuris@redhat.com","date":"2020-02-27 07:46:53.000000000","tz":0},"subject":"Adding new test of security group rules quota increased.","message":"Adding new test of security group rules quota increased.\n\nThe test is checking that it is possible to create\nnew security group rules after increasing the quota parameter.\n\nChange-Id: Ie961293215c48883c962debe888b0bca418e3eb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/06634163c0ab599d2b5e25948a3c94bcf63946a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/06634163c0ab599d2b5e25948a3c94bcf63946a0"}]},"branch":"refs/heads/master"},"c9f986490e981bd564fd1311ce8630ed32cb5c2e":{"kind":"REWORK","_number":11,"created":"2020-02-27 07:47:48.000000000","uploader":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"ref":"refs/changes/42/708842/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/42/708842/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/11"}}},"commit":{"parents":[{"commit":"dee69e0f5c1eb5693355768aa8accc4cc2676820","subject":"Helper class for BaseSecGroupRulesQuota tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/dee69e0f5c1eb5693355768aa8accc4cc2676820"}]}],"author":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"committer":{"name":"Eran Kuris","email":"ekuris@redhat.com","date":"2020-02-27 07:47:42.000000000","tz":0},"subject":"Adding new test of security group rules quota increased.","message":"Adding new test of security group rules quota increased.\n\nThe test is checking that it is possible to create\nnew security group rules after increasing the quota parameter.\n\nChange-Id: Ie961293215c48883c962debe888b0bca418e3eb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c9f986490e981bd564fd1311ce8630ed32cb5c2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c9f986490e981bd564fd1311ce8630ed32cb5c2e"}]},"branch":"refs/heads/master"},"e6d3c532da06402238b0b960ef32790d949ce3d8":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2020-03-01 08:26:03.000000000","uploader":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"ref":"refs/changes/42/708842/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/42/708842/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/12"}}},"commit":{"parents":[{"commit":"913ea88c6c0e6a8495b93d4137026692b33c880d","subject":"Helper class for BaseSecGroupRulesQuota tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/913ea88c6c0e6a8495b93d4137026692b33c880d"}]}],"author":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"committer":{"name":"Eran Kuris","email":"ekuris@redhat.com","date":"2020-03-01 08:26:03.000000000","tz":0},"subject":"Adding new test of security group rules quota increased.","message":"Adding new test of security group rules quota increased.\n\nThe test is checking that it is possible to create\nnew security group rules after increasing the quota parameter.\n\nChange-Id: Ie961293215c48883c962debe888b0bca418e3eb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e6d3c532da06402238b0b960ef32790d949ce3d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e6d3c532da06402238b0b960ef32790d949ce3d8"}]},"branch":"refs/heads/master"},"342d5da3f8b4bb9855c77907eac2445adde08e58":{"kind":"REWORK","_number":13,"created":"2020-03-03 05:51:18.000000000","uploader":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"ref":"refs/changes/42/708842/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/42/708842/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/13"}}},"commit":{"parents":[{"commit":"913ea88c6c0e6a8495b93d4137026692b33c880d","subject":"Helper class for BaseSecGroupRulesQuota tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/913ea88c6c0e6a8495b93d4137026692b33c880d"}]}],"author":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"committer":{"name":"Eran Kuris","email":"ekuris@redhat.com","date":"2020-03-03 05:49:11.000000000","tz":0},"subject":"Adding new test of security group rules quota increased.","message":"Adding new test of security group rules quota increased.\n\nThe test is checking that it is possible to create\nnew security group rules after increasing the quota parameter.\n\nChange-Id: Ie961293215c48883c962debe888b0bca418e3eb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/342d5da3f8b4bb9855c77907eac2445adde08e58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/342d5da3f8b4bb9855c77907eac2445adde08e58"}]},"branch":"refs/heads/master"},"4e47ef88ba87973607362af1b7d688e1e909682c":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2020-03-03 06:29:46.000000000","uploader":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"ref":"refs/changes/42/708842/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/42/708842/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/14"}}},"commit":{"parents":[{"commit":"b3a88e74d773cc809fdc6c208054de295114dcac","subject":"Helper class for BaseSecGroupRulesQuota tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b3a88e74d773cc809fdc6c208054de295114dcac"}]}],"author":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"committer":{"name":"Eran Kuris","email":"ekuris@redhat.com","date":"2020-03-03 06:29:46.000000000","tz":0},"subject":"Adding new test of security group rules quota increased.","message":"Adding new test of security group rules quota increased.\n\nThe test is checking that it is possible to create\nnew security group rules after increasing the quota parameter.\n\nChange-Id: Ie961293215c48883c962debe888b0bca418e3eb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4e47ef88ba87973607362af1b7d688e1e909682c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4e47ef88ba87973607362af1b7d688e1e909682c"}]},"branch":"refs/heads/master"},"aeaecc8d57486bca5b4b3137f6ecf6d909a6928d":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2020-03-03 06:46:42.000000000","uploader":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"ref":"refs/changes/42/708842/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/42/708842/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/15"}}},"commit":{"parents":[{"commit":"56141acf48de2b19fd7b8aaf85526adc480a55cc","subject":"Helper class for BaseSecGroupRulesQuota tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/56141acf48de2b19fd7b8aaf85526adc480a55cc"}]}],"author":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"committer":{"name":"Eran Kuris","email":"ekuris@redhat.com","date":"2020-03-03 06:46:42.000000000","tz":0},"subject":"Adding new test of security group rules quota increased.","message":"Adding new test of security group rules quota increased.\n\nThe test is checking that it is possible to create\nnew security group rules after increasing the quota parameter.\n\nChange-Id: Ie961293215c48883c962debe888b0bca418e3eb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/aeaecc8d57486bca5b4b3137f6ecf6d909a6928d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/aeaecc8d57486bca5b4b3137f6ecf6d909a6928d"}]},"branch":"refs/heads/master"},"7abc0857695ce9d8c043cbd5415515a4afd74285":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2020-03-03 10:36:41.000000000","uploader":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"ref":"refs/changes/42/708842/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/42/708842/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/16"}}},"commit":{"parents":[{"commit":"ea8947100635e955196567cfc9fefa0a320e152b","subject":"Helper class for BaseSecGroupRulesQuota tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ea8947100635e955196567cfc9fefa0a320e152b"}]}],"author":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"committer":{"name":"Eran Kuris","email":"ekuris@redhat.com","date":"2020-03-03 10:36:41.000000000","tz":0},"subject":"Adding new test of security group rules quota increased.","message":"Adding new test of security group rules quota increased.\n\nThe test is checking that it is possible to create\nnew security group rules after increasing the quota parameter.\n\nChange-Id: Ie961293215c48883c962debe888b0bca418e3eb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7abc0857695ce9d8c043cbd5415515a4afd74285"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7abc0857695ce9d8c043cbd5415515a4afd74285"}]},"branch":"refs/heads/master"},"48c348350060ac21ca1bdc8faf84a98f282f1270":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2020-03-04 15:47:21.000000000","uploader":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"ref":"refs/changes/42/708842/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/42/708842/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/17"}}},"commit":{"parents":[{"commit":"1b896ca1968affede6a611f3e1e717037c43ca84","subject":"Helper class for BaseSecGroupRulesQuota tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1b896ca1968affede6a611f3e1e717037c43ca84"}]}],"author":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"committer":{"name":"Eran Kuris","email":"ekuris@redhat.com","date":"2020-03-04 15:47:21.000000000","tz":0},"subject":"Adding new test of security group rules quota increased.","message":"Adding new test of security group rules quota increased.\n\nThe test is checking that it is possible to create\nnew security group rules after increasing the quota parameter.\n\nChange-Id: Ie961293215c48883c962debe888b0bca418e3eb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/48c348350060ac21ca1bdc8faf84a98f282f1270"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/48c348350060ac21ca1bdc8faf84a98f282f1270"}]},"branch":"refs/heads/master"},"1f4bd6b27f6698a917935aead323c9ad61002334":{"kind":"REWORK","_number":18,"created":"2020-03-05 07:37:28.000000000","uploader":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"ref":"refs/changes/42/708842/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/42/708842/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/18"}}},"commit":{"parents":[{"commit":"1b896ca1968affede6a611f3e1e717037c43ca84","subject":"Helper class for BaseSecGroupRulesQuota tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1b896ca1968affede6a611f3e1e717037c43ca84"}]}],"author":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"committer":{"name":"Eran Kuris","email":"ekuris@redhat.com","date":"2020-03-05 07:37:21.000000000","tz":0},"subject":"Adding new test of security group rules quota increased.","message":"Adding new test of security group rules quota increased.\n\nThe test is checking that it is possible to create\nnew security group rules after increasing the quota parameter.\n\nChange-Id: Ie961293215c48883c962debe888b0bca418e3eb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1f4bd6b27f6698a917935aead323c9ad61002334"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1f4bd6b27f6698a917935aead323c9ad61002334"}]},"branch":"refs/heads/master"},"6aa8f02fb0bee54599b3fee9836de3d9f9061601":{"kind":"REWORK","_number":19,"created":"2020-03-09 12:41:02.000000000","uploader":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"ref":"refs/changes/42/708842/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/42/708842/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/19"}}},"commit":{"parents":[{"commit":"1b896ca1968affede6a611f3e1e717037c43ca84","subject":"Helper class for BaseSecGroupRulesQuota tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1b896ca1968affede6a611f3e1e717037c43ca84"}]}],"author":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"committer":{"name":"Eran Kuris","email":"ekuris@redhat.com","date":"2020-03-09 12:40:34.000000000","tz":0},"subject":"Adding new test of security group rules quota increased.","message":"Adding new test of security group rules quota increased.\n\nThe test is checking that it is possible to create\nnew security group rules after increasing the quota parameter.\n\nChange-Id: Ie961293215c48883c962debe888b0bca418e3eb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6aa8f02fb0bee54599b3fee9836de3d9f9061601"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6aa8f02fb0bee54599b3fee9836de3d9f9061601"}]},"branch":"refs/heads/master"},"8fc22f4ae8c3e5177d4584cbf36f83004d6b2e94":{"kind":"REWORK","_number":20,"created":"2020-03-09 13:04:00.000000000","uploader":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"ref":"refs/changes/42/708842/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/42/708842/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/20"}}},"commit":{"parents":[{"commit":"1b896ca1968affede6a611f3e1e717037c43ca84","subject":"Helper class for BaseSecGroupRulesQuota tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1b896ca1968affede6a611f3e1e717037c43ca84"}]}],"author":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"committer":{"name":"Eran Kuris","email":"ekuris@redhat.com","date":"2020-03-09 13:03:54.000000000","tz":0},"subject":"Adding new test of security group rules quota increased.","message":"Adding new test of security group rules quota increased.\n\nThe test is checking that it is possible to create\nnew security group rules after increasing the quota parameter.\n\nChange-Id: Ie961293215c48883c962debe888b0bca418e3eb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8fc22f4ae8c3e5177d4584cbf36f83004d6b2e94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8fc22f4ae8c3e5177d4584cbf36f83004d6b2e94"}]},"branch":"refs/heads/master"},"d08e7b0e806ba85accf5ce856c286f7c662d9f4d":{"kind":"REWORK","_number":21,"created":"2020-03-10 12:42:08.000000000","uploader":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"ref":"refs/changes/42/708842/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/42/708842/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/21"}}},"commit":{"parents":[{"commit":"1b896ca1968affede6a611f3e1e717037c43ca84","subject":"Helper class for BaseSecGroupRulesQuota tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1b896ca1968affede6a611f3e1e717037c43ca84"}]}],"author":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"committer":{"name":"Eran Kuris","email":"ekuris@redhat.com","date":"2020-03-10 12:41:55.000000000","tz":0},"subject":"Adding new test of security group rules quota increased.","message":"Adding new test of security group rules quota increased.\n\nThe test is checking that it is possible to create\nnew security group rules after increasing the quota parameter.\n\nChange-Id: Ie961293215c48883c962debe888b0bca418e3eb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d08e7b0e806ba85accf5ce856c286f7c662d9f4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d08e7b0e806ba85accf5ce856c286f7c662d9f4d"}]},"branch":"refs/heads/master"},"7ef1fc115b78cdaa7c85a86f529cf038f01a1e1c":{"kind":"REWORK","_number":22,"created":"2020-03-10 13:29:18.000000000","uploader":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"ref":"refs/changes/42/708842/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/42/708842/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/22"}}},"commit":{"parents":[{"commit":"1b896ca1968affede6a611f3e1e717037c43ca84","subject":"Helper class for BaseSecGroupRulesQuota tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1b896ca1968affede6a611f3e1e717037c43ca84"}]}],"author":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"committer":{"name":"Eran Kuris","email":"ekuris@redhat.com","date":"2020-03-10 13:29:10.000000000","tz":0},"subject":"Adding new test of security group rules quota increased.","message":"Adding new test of security group rules quota increased.\n\nThe test is checking that it is possible to create\nnew security group rules after increasing the quota parameter.\n\nChange-Id: Ie961293215c48883c962debe888b0bca418e3eb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7ef1fc115b78cdaa7c85a86f529cf038f01a1e1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7ef1fc115b78cdaa7c85a86f529cf038f01a1e1c"}]},"branch":"refs/heads/master"},"4e329cf99c8af6092a1e47488dce91d2dd1f4e94":{"kind":"REWORK","_number":23,"created":"2020-03-10 13:35:26.000000000","uploader":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"ref":"refs/changes/42/708842/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/42/708842/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/42/708842/23"}}},"commit":{"parents":[{"commit":"1b896ca1968affede6a611f3e1e717037c43ca84","subject":"Helper class for BaseSecGroupRulesQuota tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1b896ca1968affede6a611f3e1e717037c43ca84"}]}],"author":{"name":"ekuris","email":"ekuris@redhat.com","date":"2020-02-20 10:14:20.000000000","tz":120},"committer":{"name":"Eran Kuris","email":"ekuris@redhat.com","date":"2020-03-10 13:35:21.000000000","tz":0},"subject":"Adding new test of security group rules quota increased.","message":"Adding new test of security group rules quota increased.\n\nThe test is checking that it is possible to create\nnew security group rules after increasing the quota parameter.\n\nChange-Id: Ie961293215c48883c962debe888b0bca418e3eb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4e329cf99c8af6092a1e47488dce91d2dd1f4e94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4e329cf99c8af6092a1e47488dce91d2dd1f4e94"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
