)]}'
{"id":"openstack%2Fneutron-tempest-plugin~711656","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ie6e26acd5e9c3acd2ee86a9c354c4a2934aa4bce","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/1862032","hashtags":[],"change_id":"Ie6e26acd5e9c3acd2ee86a9c354c4a2934aa4bce","subject":"Add tempest api tests for subnet pools RBAC","status":"MERGED","created":"2020-03-06 16:48:26.000000000","updated":"2020-04-14 13:07:46.000000000","submitted":"2020-04-14 13:07:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711656-1586869666380-69063b05","meta_rev_id":"0ebc738631f3e4eafa4b72df56bf34f6a29ab1a8","_number":711656,"virtual_id_number":711656,"owner":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-14 13:07:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2020-04-10 12:21:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},{"value":2,"date":"2020-04-10 12:21:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2020-04-14 09:11:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-04-10 12:21:10.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"date":"2020-04-10 12:21:10.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"date":"2020-04-10 12:21:10.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"date":"2020-04-10 12:21:10.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-08 08:02:37.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-04-08 08:02:42.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-04-08 08:02:46.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-04-08 08:02:52.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-04-08 08:38:22.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-04-14 09:11:08.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-04-14 13:07:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"726e5137fc2e1ad3860355042eb391895a22744a","author":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"date":"2020-03-06 16:48:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c884fa462596ffff69140d40d996f614fa9aeefe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 16:48:45.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"60d85881def81e1695f61665b2502060e832ee5b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-10 10:42:59.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"4d1cfadb8c2d70da207623c43fe42a783eb8093a","author":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"date":"2020-03-11 19:47:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a45109b9b580ab1f40dc20a952068b2a5726a9d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 00:07:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99959925852a472aba625c88944dbd9e : SUCCESS in 5m 36s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5e583a4e676f4eaeb3aaf0ff3fc1d995 : SUCCESS in 58m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/86395eda93b04cbab50e6b881ecc9c18 : SUCCESS in 38m 05s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/581d2e5326b341b3a956a8d315fe2a7d : FAILURE in 1h 54m 25s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2c837011b28d4f99ac9c7791cd20fba6 : SUCCESS in 1h 09m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7ff85db99a594f49862c4a9fdcab7d04 : SUCCESS in 1h 24m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/01cc1935397f45b983ae040ed33e9f5c : SUCCESS in 1h 00m 44s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/9f12d26a874b4530a256256c78a38f60 : SUCCESS in 1h 04m 11s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/0b81c05d952243ca805a76fa76db45bd : SUCCESS in 38m 54s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/ec950b6587ee45c0813388861a3dc1c2 : FAILURE in 1h 33m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/8f82d45de48c47f780711ee389af25c9 : SUCCESS in 1h 28m 11s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/bb7f06fdd3a24b969254b976a7b0e4b9 : SUCCESS in 1h 21m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/4ffc4bcd0c54448792a32d5b1af79736 : SUCCESS in 1h 06m 07s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/28dd0ee20690485388d11bfa735346ac : SUCCESS in 58m 17s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/46ecdb4f427f4733a4f064770387cf3c : SUCCESS in 32m 09s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/252212940aa44937bec31edc7879b9f2 : FAILURE in 1h 21m 11s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/8af88104ee5c43f680e3df7200a6c5dc : SUCCESS in 1h 29m 30s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/178042e565cc4f9b917181fab54de033 : SUCCESS in 1h 13m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/91a332194ab640f194061daff1acf473 : SUCCESS in 1h 07m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ddf6b912b474ecf9991f1a4215c3cd0 : SUCCESS in 4m 23s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/6e04ba285c0c4907a80cf9e3f03bfc68 : SUCCESS in 1h 51m 48s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/3d8b2ebf88e04053b1d9958e08e81915 : SUCCESS in 1h 35m 18s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a9dc23fe499a4634acf5705a88d0655d : SUCCESS in 1h 33m 45s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/7d7e8a8551814d79bf152d6e234aa0c4 : SUCCESS in 1h 39m 52s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/51b61d251c4f43f988a7afbcdbb6e3b1 : SUCCESS in 36m 37s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/a9ba9a08ba6a454485e242ebc87d528d : SUCCESS in 40m 16s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/e75f2234dfb64cac9cd6080046dc180f : SUCCESS in 52m 44s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/914e902e5d3641f2b075eb0cd53c1124 : SUCCESS in 43m 57s","accounts_in_message":[],"_revision_number":1},{"id":"67e0ffde4d611bdb22ab92f8b3063693d0c6b075","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-19 09:45:08.000000000","message":"Patch Set 1: Code-Review-1\n\nSame comment here like for address pools tests patch, You need to add api extension to the list, otherwise test will be skipped","accounts_in_message":[],"_revision_number":1},{"id":"b127da319cd12c1b8b2849347a1e06972da8d69d","author":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"date":"2020-03-19 14:32:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d7db36bb9bc0b1e7f563b7cda756c3d0dc08f5ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 14:40:41.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"ca504a334d9ecf08cf8e9ee7be87876030a4877d","author":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"date":"2020-03-19 14:42:50.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"becdc902909f4a27a8a48d6eb07a9dca73584fe3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 14:52:16.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"06946c7b3dbab6806da80674d96df64c7e4709e3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-19 15:30:06.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"447e6926de6f696f89a944bcfbe1911aeb6edc63","author":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"date":"2020-03-26 12:37:34.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"43d2a7ac97eccf5e0e6afd8a94da66a38817bd54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 15:55:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b411e68d1e0476781e9c9664f587ee7 : SUCCESS in 6m 16s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7b27864106d94ca39e91b8e447d65893 : SUCCESS in 54m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/35b9894a4fb14b9ea4514a27f474c480 : SUCCESS in 39m 45s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/7087eb16ebce4c0cbb65f5e6684fffaa : FAILURE in 2h 39m 44s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ec377dc2d59a47f5a4f1e2c4480732e3 : SUCCESS in 1h 23m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b6f758ce742e42278ab5f0de324fff27 : SUCCESS in 59m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/304b6b23d6db41aa915b1d05e9048c76 : SUCCESS in 1h 19m 47s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/f94d878d761243449225361039304548 : SUCCESS in 59m 09s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/5cd1eeadf6544671a9260fb5aad51a40 : SUCCESS in 41m 37s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/231b487772ba40f9915696f1b89ca568 : FAILURE in 2h 00m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/7587b1fd1f144681836e0c9d15e11184 : SUCCESS in 1h 23m 44s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/affb3ccd351242acbd846c3b4e83fb9d : FAILURE in 14m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/7705ab0bbf3a4678b99e1b75d14d71d0 : FAILURE in 16m 06s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/152c78d505874190b9324d43da6c4e32 : SUCCESS in 1h 01m 02s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/a2795c2b530148a5a2e98ef06da85bd6 : SUCCESS in 51m 29s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/a540ee67bf294484b819596524e47d8d : FAILURE in 1h 41m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/8b9300c75148406ca072a82728e3e17d : SUCCESS in 1h 32m 41s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/5cdb0bdce12c42419c3099bac4f439d5 : SUCCESS in 1h 09m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/6c5c7f7f5a7d49a2a269819788563ad0 : SUCCESS in 1h 02m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ae0cca259a94a4b973118d09fcf46ea : SUCCESS in 5m 16s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/9f119a876b504c5bb978812b1a86f56a : SUCCESS in 1h 47m 23s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/ec354df0e6a44340a0f8783ddcf8b594 : SUCCESS in 1h 49m 37s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/65f439997f5e4e88a6a6058b6c8766a4 : SUCCESS in 1h 37m 03s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/87aed66a69f94a858cfeaf07dad426c8 : SUCCESS in 1h 42m 12s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/6de31bde71464ac7ba498680bd47955d : FAILURE in 42m 50s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/99d263d470a14aa3ab393cbe43a8acb6 : SUCCESS in 58m 36s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/a0539c9b4be84fa9a52606ffad144826 : SUCCESS in 1h 08m 44s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/a5b0a8a6accb49a196765fc1d4505f9c : SUCCESS in 40m 44s","accounts_in_message":[],"_revision_number":4},{"id":"ae4b7f23abdc55ad6fa5376227792422bdd632a7","author":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"date":"2020-03-29 13:07:35.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"5ff981b8ff9a6596efa7d755f4dc9691249b7da0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 15:17:33.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/782e193b65fb4719a73f81ed452b65a5 : SUCCESS in 5m 28s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/71e0287c993243609fe9944f15767e4a : SUCCESS in 1h 12m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8020511df5234db1bbad8912a7e5490d : FAILURE in 23m 36s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/efe08d47d2ea4c9c973faff2d646ef3b : FAILURE in 1h 46m 51s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/79d0e24869f24e8f9f1a6e46d2ba07dc : SUCCESS in 1h 29m 46s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/328781fc0c7f4c0e81162d26d953e2e6 : SUCCESS in 1h 10m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1092d78dea2f42c888a0bb366dd36528 : SUCCESS in 1h 04m 16s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/1c060387911f4db2ad5a5d1d1b8cf342 : SUCCESS in 55m 56s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/d7627c038a1046b58551184336a38493 : SUCCESS in 35m 02s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/198f68740e6c4161b7315bc317b3a3b2 : FAILURE in 1h 26m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/b405b71e09604239b94b3a395f27b8c4 : SUCCESS in 1h 10m 44s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/d29a7ed5d74f4646a72b4842a1ca7797 : SUCCESS in 1h 27m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/b9635141d2dd47a89c6116996e4210d9 : SUCCESS in 1h 01m 49s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/893820d46d1e4ea290331f44d05a3ff5 : SUCCESS in 50m 06s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/129fab4480124256b9b3d429648b892b : SUCCESS in 36m 00s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/498fde2203ed49538337bbf87d7270e1 : SUCCESS in 2h 04m 17s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/37c5e6539f7d4841a31e7a88900ed974 : SUCCESS in 1h 14m 11s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/b9821ae65ba042438670beb85cc6ba25 : SUCCESS in 1h 17m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/f30fb093ac354b68958a6a6537ad2a9e : SUCCESS in 1h 39m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dc94840e5a64c5db5f4dcd7e72bb0d1 : SUCCESS in 3m 48s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/69faef3d8f2343d29d83c2a9f8976a19 : SUCCESS in 1h 48m 15s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/c7abd8772c8c441695e487073e1724f6 : SUCCESS in 1h 31m 30s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a4e22c570a8546ffb13b854633062097 : SUCCESS in 1h 36m 30s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/3da400b729d045b2a95ed03cb6560c6d : SUCCESS in 1h 37m 59s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/2a7f18aa85534023815c15077b6ab99d : SUCCESS in 33m 31s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/866b6cceba704a5dae240baef5cfbc27 : SUCCESS in 42m 03s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/ccd03ac0d6c849bc873a756bd2f998c9 : SUCCESS in 46m 06s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/e85615a4e9404bf5a59ca493969da207 : SUCCESS in 41m 45s","accounts_in_message":[],"_revision_number":4},{"id":"e67d7ddb0ea568eedeee78c2115e337f97c501c6","author":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"date":"2020-03-30 17:43:53.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a983c6d9115262b44c5db1a719750bef51d22e22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 20:41:14.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90b5f88ee64c4fb7925eb1c3e16c7e32 : SUCCESS in 7m 02s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/edfb20d1251745178361c14113148e13 : SUCCESS in 50m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3bc01e5a269a4b7dad853079e7653eea : FAILURE in 31m 37s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/7e38a1f065ef44cca82854ae6e4f0a9b : SUCCESS in 1h 37m 23s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a493ad7d3ef04c80b03fcdfb2d5d8c8c : SUCCESS in 1h 26m 16s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/54cff95355234637948174f0cecfcb0c : SUCCESS in 1h 13m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/389113a3ab244e6abad841bc22228049 : SUCCESS in 59m 43s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/bac29154e32d46c7b4ec7b262842bb9c : SUCCESS in 56m 18s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/dbcc6ad160754e74b83bfe8f5f93de04 : SUCCESS in 36m 22s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/91c3344d0ef74a09b808056f2e3c2efb : SUCCESS in 1h 25m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/7ecb2ed278b040cab9a5aa552027679c : SUCCESS in 1h 08m 39s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/5bce298ea38943b29926e7425cd02797 : SUCCESS in 1h 04m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/25c0fcc7c85940f0bd0ca467a86ba5fb : SUCCESS in 1h 02m 07s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/985b6df84c8442cdbd92ffa82cad0a7e : SUCCESS in 1h 10m 05s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/3e77a8b4e84d4ba1a2c0c33b323b75a1 : SUCCESS in 34m 45s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/cf00d9471b934f0889491dc9a130cc5c : FAILURE in 2h 04m 46s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/6a4cff24667f4f0cb355864d0aafb53b : SUCCESS in 1h 15m 57s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/045ec3f91c6646969faef5773b36492d : SUCCESS in 1h 13m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/4f24a0a6160d4cc69642a7de2c197da6 : SUCCESS in 1h 03m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ea1dcf8d6514f8fbda4e8701161b211 : SUCCESS in 5m 42s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/f75445405d0846e8952df9823e224f16 : SUCCESS in 1h 59m 56s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/ab2e8c21f253437288c95286358ea616 : SUCCESS in 1h 57m 28s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/2e773877dd2a4913987c6874bc143471 : FAILURE in 1h 41m 09s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/fc834d52bb7749079bafc9b763dcdd55 : SUCCESS in 1h 22m 28s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/10c7c28f8c5f4b928ddcb86aaafd47a6 : SUCCESS in 33m 17s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/d91988a14e9e4c13b8a554e5fa31d3ef : FAILURE in 37m 23s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/ecd2d752d1634a6bb886d102c903d91b : SUCCESS in 45m 03s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/addccdc018a9400ea5b2584b9a575d95 : SUCCESS in 43m 09s","accounts_in_message":[],"_revision_number":4},{"id":"a5fe4007abe1dd9c85ffbdde8204d0d9dc8edc91","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-30 20:56:53.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"23ef924dd77242e719f033cc217685dc62f591e9","author":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"date":"2020-04-03 13:37:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c8659ae49844347709a5d9988489b605fb195e2b","author":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"date":"2020-04-03 13:37:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"64dd3575f1844016becc792b493bb1b305990da2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 13:44:43.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"37a297d6c7f45d13086e37b8e247116904a6d45a","author":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"date":"2020-04-03 13:54:48.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"60d3838b02ac966617cb22d9ee1f02949e25c1c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 16:15:44.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/125aa48395904175878fb29276283a7f : SUCCESS in 5m 05s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0853e4bac7754342aa200a484b3da484 : FAILURE in 49m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ac02c65c19974b97ab4602fa3e86502b : SUCCESS in 35m 36s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/a332326a11054766846fd80771354448 : FAILURE in 2h 17m 39s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cf6d3ad1d6304fbca3d83f9ffb169863 : SUCCESS in 1h 05m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/203025fcc72d49088c319f03debf4a98 : SUCCESS in 1h 17m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cdc23c00edad456c802996ff41d20069 : SUCCESS in 1h 31m 02s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/e970b8912e2d41a18dff2eabfb55420c : SUCCESS in 1h 01m 29s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/7bedec7708e0493196dce9e259c3b3f5 : SUCCESS in 39m 16s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/9c5069de3c394d1ca211b5aabc7d6b2b : SUCCESS in 1h 41m 31s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/e241176a79f243a19ba5564aab9abca3 : SUCCESS in 1h 17m 32s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/ec29b06c823d4791b5268f74c0d211a2 : SUCCESS in 1h 08m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/2d7ace2247c440688959a6707a038814 : SUCCESS in 1h 08m 09s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/1e150ec9162b4a80885c4a12084d2f69 : SUCCESS in 55m 52s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/83c3d36d1ca34ab9bdbef207ee34de84 : SUCCESS in 38m 12s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/b80a450486d84610815670314a723c40 : FAILURE in 1h 40m 26s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/d72968aee1164dbfa156d85c0ed5574a : SUCCESS in 1h 13m 07s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/38b72d809a134a2f8fa4da35fcb46906 : SUCCESS in 1h 13m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/b3db33b7476747639c55f9c1e243c19a : SUCCESS in 1h 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63d278ba1bab42059418d7b74ef8b65d : SUCCESS in 4m 48s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/904f6dd9fce34cc28b6b6b149ae80efe : SUCCESS in 1h 41m 59s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/1a6735a3f62b4c3c95947b59d97744c7 : SUCCESS in 1h 47m 41s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/b86da4e7f19448d49723097680f9e25d : SUCCESS in 1h 44m 51s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/d043a42602fd4eff88434b44d65aabac : SUCCESS in 1h 30m 02s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/263d169dc9aa49a99af6af2c3fda505b : SUCCESS in 41m 23s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/bdfcf494446a45fcb4267e326ffafd29 : SUCCESS in 47m 38s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/4abbb047fe5a4006b45e3a71dfe0f818 : SUCCESS in 48m 38s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/c578f052766b484087efcf122ebb3446 : SUCCESS in 39m 41s","accounts_in_message":[],"_revision_number":6},{"id":"ff5c22f644c0da2648b849b4c36fb894ef8b13cc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-06 09:40:19.000000000","message":"Patch Set 6: Code-Review-1\n\nfailure in neutron-tempest-plugin-api job seems related to this change https://3b0b3de87306dfdfc0b8-85249f43a4823dd63eb58b0df16bb6f5.ssl.cf5.rackcdn.com/711656/6/check/neutron-tempest-plugin-api/0853e4b/testr_results.html\nPlease check it.","accounts_in_message":[],"_revision_number":6},{"id":"1250215fb4a970fa51ea1a24e84379ecd9e668c9","author":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"date":"2020-04-06 12:14:50.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"88d8c68e177ffd9877c60e029cb80aa0959b4533","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 15:21:29.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe162b4b6f884165b0df4afc7a05b002 : SUCCESS in 4m 20s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4b66b0c8afad4bcb8546ea289dcf4cd8 : SUCCESS in 58m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/195c9fb52394472b88ad4e331f081661 : SUCCESS in 38m 35s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/ef02f7bcc6814a3eb36493c8edf4037b : FAILURE in 1h 50m 49s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fda3ed93c4494a1ea2fe649420d2ec98 : SUCCESS in 1h 43m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/26bbb6242c4d4c1e9005389be9fcb6dc : SUCCESS in 1h 07m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1506f58fe4034f5db5c93d6a1c9fac57 : SUCCESS in 1h 02m 09s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/584dd979a647440fb447cebda9751754 : SUCCESS in 55m 29s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/4c4ffeb413c344c787a031726fce6081 : SUCCESS in 39m 20s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/0ea50eb4963b464a8795c0a087455863 : FAILURE in 1h 57m 09s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/1e0f61b336304cf9ab7c3915fd5defb1 : FAILURE in 1h 24m 01s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/b4ae361e38b744f59a613d880128fafc : SUCCESS in 1h 07m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/00c3f8fc71b34e53966367dd2ab78f00 : SUCCESS in 1h 14m 31s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/63ec8d42d929422db31618f88ddf3020 : SUCCESS in 1h 06m 25s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/35601f205403481a8830171481029624 : SUCCESS in 53m 42s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/3b972a83a0fa4809ba44a6e161226505 : SUCCESS in 1h 55m 13s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/d0eac0f34a0f4838b197cd36661cc798 : SUCCESS in 1h 27m 10s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/46c4b917e7f24f8499d08a8cca8538b1 : SUCCESS in 1h 27m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/3e27184f548e42ddbf74716c4abb1674 : SUCCESS in 1h 12m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74c3feec504f4dfeb471e003d23a5e73 : SUCCESS in 4m 43s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/2a75d35354e94136ba4f0dc8ee0d2ea7 : SUCCESS in 1h 57m 53s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/77d71c1fe15743e9ba4f01a4e8258020 : SUCCESS in 2h 26m 59s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9dc9ce8febad4489a9539518300a5dc7 : SUCCESS in 1h 34m 50s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/b9bc54027b754662b165b8d6c11def74 : SUCCESS in 1h 31m 25s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/8e2a8964b3d54e218b174a0655f25d2a : FAILURE in 35m 46s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/d6425085176d499d91f2d160a6d2c51e : SUCCESS in 39m 59s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/c9aa73fd9e8b4eacbb8efc3199288283 : SUCCESS in 52m 00s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/0624bc61c68148efac469cddd7fc2aa8 : SUCCESS in 38m 39s","accounts_in_message":[],"_revision_number":7},{"id":"d6049f63125abd08df43db2d305499e1b1acf67e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-07 09:39:10.000000000","message":"Patch Set 7: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"9fa1b38ebb6798c40030de9fcad93e951fa2484b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-07 15:44:06.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"9127b824ba981bd5588cfa454367ca1fa4e6b0ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 16:03:56.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf5a19289f3c45bc92f2f57d76aabde6 : SUCCESS in 6m 06s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/87e38a153981449aba0051878a7b9cb8 : SUCCESS in 50m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/afcdccb0b37646c786cbb87d3db2cea1 : SUCCESS in 33m 05s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/5779871d57d941ac8d0a69124743cc70 : FAILURE in 1h 45m 48s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/19ef781249354c348bbdf77ad121477e : SUCCESS in 1h 04m 28s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/906e90f30b0b4386a3b3863d98aead86 : SUCCESS in 1h 04m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7862f512f16c4ce3a9565d0d3433d66a : SUCCESS in 1h 17m 52s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/50b56e86b09945eea6de00701b3bd213 : SUCCESS in 51m 00s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/c6eaedfdf4794e62b603ebf4a285e9b2 : SUCCESS in 41m 52s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/40a1c75dc0114ed39a20372b5b4e3101 : FAILURE in 2h 00m 23s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/871a622f9e9948baa90071169014ac40 : SUCCESS in 1h 08m 07s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/bba017e81dcd4eb3afdfba9d09894550 : SUCCESS in 1h 07m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/660a468beb90402dad0c5df3bf7fee7a : SUCCESS in 1h 18m 35s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/31d395df21304d349fa0ff775b164ace : SUCCESS in 1h 09m 55s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/d122cc635bec4ce0951f764b57673e80 : SUCCESS in 42m 51s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/68d1eef6474846a8a691719be7686226 : SUCCESS in 1h 52m 56s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/bdfbaeb2c47e44078075022650dabe69 : SUCCESS in 1h 46m 21s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/40ba5637e716490f8adad799484e0c4b : SUCCESS in 1h 20m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/e89ce0ab380f4f47913722761b68453b : SUCCESS in 1h 06m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/268381f065834fc3ba2ce893f8f36133 : SUCCESS in 5m 54s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/244d4178b11b4171ac9022962306c295 : SUCCESS in 1h 40m 56s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/529a914a199143d7ac316009648c73da : SUCCESS in 2h 03m 40s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9bdd576ae2ba439b8bce03b1df37ac30 : SUCCESS in 1h 36m 58s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/cd250f3c6c894b60bff11431649cbbbf : SUCCESS in 1h 41m 02s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/a5f8daa6afaf4a9eb43e45085390862b : SUCCESS in 39m 59s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/cadc819deffc472fa5f8cbde3fb17e91 : SUCCESS in 45m 34s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/c24ad3d05ee541188a6352eabeb31788 : SUCCESS in 53m 31s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/9c1f7d346c2844bf91d4c09423fa5dc1 : SUCCESS in 46m 52s","accounts_in_message":[],"_revision_number":7},{"id":"d2da14429afe2927e08b1841e132591c82296971","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-08 08:38:22.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"cb0fe128beeace5abc715a924f5605b31c9b5a53","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-08 08:41:01.000000000","message":"Patch Set 7:\n\nThe one skipped test will pass I suppose when addressscope tests will be merged and extension will be enabled","accounts_in_message":[],"_revision_number":7},{"id":"bdaca3c72eeeea458ad5c9a93fb342e494e07a53","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-09 21:42:06.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"1d9586e0fa6f8269c7c8a5ba20e515cf404f46ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 06:25:32.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"1c9846895eaf144d127d3f2f264f8f1e682d0dc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 09:22:15.000000000","message":"Patch Set 7: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"ef615ccb85752243cfeb07c17c65c1700f642b5a","author":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"date":"2020-04-10 12:21:10.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"e5d9eed4af8e907144f3342d5eefdf959c8f6f01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 14:52:26.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e14306486144af6b38e0d2471bc28eb : SUCCESS in 4m 22s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0846ab7cf20044fa9aed704b0283ec72 : SUCCESS in 57m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6b1877495ffa486aab4a4d075782e2d2 : SUCCESS in 32m 49s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/9480ee12c32a4381bfdf9d2dbd10dc4c : FAILURE in 1h 38m 43s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3fbba42229044f91a5034a90b30ca556 : SUCCESS in 1h 29m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/36571b761bba497a9b04267e290aecdc : SUCCESS in 1h 15m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b9c33af3df2f4b6fa84b93de16b499f0 : SUCCESS in 1h 07m 46s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/48ec9de842804e009e89d65af371cbf7 : SUCCESS in 53m 43s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/a9d6b6c254da409f949b35cda1701001 : SUCCESS in 37m 19s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/32a80e0a8a6c4472bbe656372e043242 : FAILURE in 2h 24m 25s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/6f60af8dadc940f8a3bed37d98d8fd49 : SUCCESS in 1h 10m 42s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/14eb6cee73aa46d38e3da552bc5786b3 : SUCCESS in 1h 31m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/b21fe6dd9f0d43caa851d5a4c888d73c : SUCCESS in 1h 05m 17s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/251d805dfffb47e0831ed94ccd6d068e : SUCCESS in 52m 27s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/ce9e60f272c74953a6d4493a84b21475 : SUCCESS in 38m 23s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/33cbd38eb0cc40ada54e3f7435ebf9b2 : FAILURE in 1h 45m 09s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/9360c91b1e994be88af7ebdc0310b562 : SUCCESS in 1h 12m 04s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/81cbc45b6d5649769354904162fc5a56 : SUCCESS in 1h 36m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/f3afce2d2d4847f5b2ed38becc57ef38 : SUCCESS in 1h 00m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9ef9eaf404048fa84a72a9761f3b7ab : SUCCESS in 4m 38s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/17f9dc107d374859aa58f8b6e4ad0007 : SUCCESS in 1h 41m 42s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/6d09204f5d734c40825a20a72207ee3f : SUCCESS in 1h 38m 33s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/fee236cf99884800972bbbbca2b4a556 : SUCCESS in 1h 21m 35s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/d31842358c6e4fd99484ba16793b26f3 : SUCCESS in 1h 12m 15s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/beb3b582c8a445e085012caa7bf3128e : FAILURE in 36m 06s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/b3d86888f4074adaa9d2c37a266f7892 : SUCCESS in 39m 52s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/53398eb5833b49c1afc96f432c805cc6 : SUCCESS in 47m 07s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/a62a9200b03e4e78b95b4907b8f7d65f : SUCCESS in 39m 42s","accounts_in_message":[],"_revision_number":8},{"id":"6b2e7bb3ad92aafd3957e03b1be88a2f8a6066d1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-14 09:11:08.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"a394677706e24a8d3bdf232d6694948b2cefd35e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 09:12:09.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"6d7a8f8183fb9be61dd076de1ace2aa1f959d962","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 13:07:43.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a8d2031db744640aa000d9f676f0933 : SUCCESS in 5m 18s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a1b989357b824ac281ba11c43faf0741 : SUCCESS in 54m 22s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e908e3d6d5b8447d9d0c36f0cb9b3469 : SUCCESS in 1h 12m 31s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a4d7c65bfff64f33883c5427f37742d0 : SUCCESS in 1h 17m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5ac7810ddd3c4c5e92f81f907a3d510c : SUCCESS in 1h 20m 43s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/f1553a12ff1e4d6994f2b57cd62d06fd : SUCCESS in 1h 03m 52s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/f327b57795414bffa48223291d74b1e8 : SUCCESS in 1h 03m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bd206a9755a40869a4eee9a4ffbd93a : SUCCESS in 6m 17s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/82f88160df454b6aa4ad6a9897b92c20 : SUCCESS in 1h 48m 07s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/1e4f6e12b55c4752bf2ef4df89f5d2d4 : SUCCESS in 1h 34m 15s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/c9654755cd4348ffb27279cf771c2fa7 : SUCCESS in 50m 08s","accounts_in_message":[],"_revision_number":8},{"id":"7e2a786bf6fadb6a0a23fb46854637e75e4f0be2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 13:07:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"649c018b3e7998e796ca9d8ac8624664b8f03f98","revisions":{"e377494faf65c354b633047ac8e5bb925828a89c":{"kind":"REWORK","_number":1,"created":"2020-03-06 16:48:26.000000000","uploader":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"ref":"refs/changes/56/711656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/56/711656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/1"}}},"commit":{"parents":[{"commit":"a80f80057f6f67f3d9a0b85df7427bc5354eafab","subject":"Add tempest api tests for address scopes RBAC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a80f80057f6f67f3d9a0b85df7427bc5354eafab"}]}],"author":{"name":"Igor Malinovskiy","email":"u.glide@gmail.com","date":"2020-03-06 15:30:47.000000000","tz":120},"committer":{"name":"Igor Malinovskiy","email":"u.glide@gmail.com","date":"2020-03-06 16:47:00.000000000","tz":120},"subject":"Add tempest api tests for subnet pools RBAC","message":"Add tempest api tests for subnet pools RBAC\n\nChange-Id: Ie6e26acd5e9c3acd2ee86a9c354c4a2934aa4bce\nPartial-Bug: #1862032\nDepends-On: https://review.opendev.org/#/c/710755/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e377494faf65c354b633047ac8e5bb925828a89c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e377494faf65c354b633047ac8e5bb925828a89c"}]},"branch":"refs/heads/master"},"c806ec50638158baa9776e1383998f00553d551a":{"kind":"REWORK","_number":2,"created":"2020-03-19 14:32:18.000000000","uploader":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"ref":"refs/changes/56/711656/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/56/711656/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/2"}}},"commit":{"parents":[{"commit":"08d3954268e0282b6d52a8faa0ee12b0048e4e5e","subject":"Add tempest api tests for address scopes RBAC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/08d3954268e0282b6d52a8faa0ee12b0048e4e5e"}]}],"author":{"name":"Igor Malinovskiy","email":"u.glide@gmail.com","date":"2020-03-06 15:30:47.000000000","tz":120},"committer":{"name":"Igor Malinovskiy","email":"u.glide@gmail.com","date":"2020-03-19 14:32:01.000000000","tz":120},"subject":"Add tempest api tests for subnet pools RBAC","message":"Add tempest api tests for subnet pools RBAC\n\nChange-Id: Ie6e26acd5e9c3acd2ee86a9c354c4a2934aa4bce\nPartial-Bug: #1862032\nDepends-On: https://review.opendev.org/#/c/710755/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c806ec50638158baa9776e1383998f00553d551a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c806ec50638158baa9776e1383998f00553d551a"}]},"branch":"refs/heads/master"},"b1a5ae1f9819e6c29f16546dfe18e52749169353":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-03-19 14:42:50.000000000","uploader":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"ref":"refs/changes/56/711656/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/56/711656/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/3"}}},"commit":{"parents":[{"commit":"b82476000ce29b010ed2246d364e4ed027c7dc1a","subject":"Merge \"Adding new test of security group rules quota increased.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b82476000ce29b010ed2246d364e4ed027c7dc1a"}]}],"author":{"name":"Igor Malinovskiy","email":"u.glide@gmail.com","date":"2020-03-06 15:30:47.000000000","tz":120},"committer":{"name":"Igor Malinovskiy","email":"u.glide@gmail.com","date":"2020-03-19 14:42:28.000000000","tz":120},"subject":"Add tempest api tests for subnet pools RBAC","message":"Add tempest api tests for subnet pools RBAC\n\nChange-Id: Ie6e26acd5e9c3acd2ee86a9c354c4a2934aa4bce\nPartial-Bug: #1862032\nDepends-On: https://review.opendev.org/#/c/710755/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b1a5ae1f9819e6c29f16546dfe18e52749169353"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b1a5ae1f9819e6c29f16546dfe18e52749169353"}]},"branch":"refs/heads/master"},"8974c3d8d2914c588391fab0ed25d1936c75e4e5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-03-26 12:37:34.000000000","uploader":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"ref":"refs/changes/56/711656/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/56/711656/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/4"}}},"commit":{"parents":[{"commit":"22d7cc950ba228ec00383b1ee6447c64ea713cb0","subject":"Merge \"Fix how nc is called in qos test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/22d7cc950ba228ec00383b1ee6447c64ea713cb0"}]}],"author":{"name":"Igor Malinovskiy","email":"u.glide@gmail.com","date":"2020-03-06 15:30:47.000000000","tz":120},"committer":{"name":"Igor Malinovskiy","email":"u.glide@gmail.com","date":"2020-03-26 12:37:31.000000000","tz":0},"subject":"Add tempest api tests for subnet pools RBAC","message":"Add tempest api tests for subnet pools RBAC\n\nChange-Id: Ie6e26acd5e9c3acd2ee86a9c354c4a2934aa4bce\nPartial-Bug: #1862032\nDepends-On: https://review.opendev.org/#/c/710755/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8974c3d8d2914c588391fab0ed25d1936c75e4e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8974c3d8d2914c588391fab0ed25d1936c75e4e5"}]},"branch":"refs/heads/master"},"6e74aae71b4eeb97967267804f3e14bb422a16c1":{"kind":"REWORK","_number":5,"created":"2020-04-03 13:37:17.000000000","uploader":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"ref":"refs/changes/56/711656/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/56/711656/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/5"}}},"commit":{"parents":[{"commit":"22d7cc950ba228ec00383b1ee6447c64ea713cb0","subject":"Merge \"Fix how nc is called in qos test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/22d7cc950ba228ec00383b1ee6447c64ea713cb0"}]}],"author":{"name":"Igor Malinovskiy","email":"u.glide@gmail.com","date":"2020-03-06 15:30:47.000000000","tz":120},"committer":{"name":"Igor Malinovskiy","email":"u.glide@gmail.com","date":"2020-04-03 13:35:20.000000000","tz":180},"subject":"Add tempest api tests for subnet pools RBAC","message":"Add tempest api tests for subnet pools RBAC\n\nChange-Id: Ie6e26acd5e9c3acd2ee86a9c354c4a2934aa4bce\nPartial-Bug: #1862032\nDepends-On: https://review.opendev.org/#/c/710755/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6e74aae71b4eeb97967267804f3e14bb422a16c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6e74aae71b4eeb97967267804f3e14bb422a16c1"}]},"branch":"refs/heads/master"},"75bf61a451038f5c5732c64064ca7bbc8bb0c49f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-04-03 13:54:48.000000000","uploader":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"ref":"refs/changes/56/711656/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/56/711656/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/6"}}},"commit":{"parents":[{"commit":"a33bd6ee1feb7e23a9333ebba3a88e886bed1b78","subject":"Disable negative subnetpool test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a33bd6ee1feb7e23a9333ebba3a88e886bed1b78"}]}],"author":{"name":"Igor Malinovskiy","email":"u.glide@gmail.com","date":"2020-03-06 15:30:47.000000000","tz":120},"committer":{"name":"Igor Malinovskiy","email":"u.glide@gmail.com","date":"2020-04-03 13:54:29.000000000","tz":180},"subject":"Add tempest api tests for subnet pools RBAC","message":"Add tempest api tests for subnet pools RBAC\n\nChange-Id: Ie6e26acd5e9c3acd2ee86a9c354c4a2934aa4bce\nPartial-Bug: #1862032\nDepends-On: https://review.opendev.org/#/c/710755/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/75bf61a451038f5c5732c64064ca7bbc8bb0c49f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/75bf61a451038f5c5732c64064ca7bbc8bb0c49f"}]},"branch":"refs/heads/master"},"051af731d9b80d449056cceff494a7705e8f37f0":{"kind":"REWORK","_number":7,"created":"2020-04-06 12:14:50.000000000","uploader":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"ref":"refs/changes/56/711656/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/56/711656/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/7"}}},"commit":{"parents":[{"commit":"a33bd6ee1feb7e23a9333ebba3a88e886bed1b78","subject":"Disable negative subnetpool test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a33bd6ee1feb7e23a9333ebba3a88e886bed1b78"}]}],"author":{"name":"Igor Malinovskiy","email":"u.glide@gmail.com","date":"2020-03-06 15:30:47.000000000","tz":120},"committer":{"name":"Igor Malinovskiy","email":"u.glide@gmail.com","date":"2020-04-06 12:14:24.000000000","tz":180},"subject":"Add tempest api tests for subnet pools RBAC","message":"Add tempest api tests for subnet pools RBAC\n\nChange-Id: Ie6e26acd5e9c3acd2ee86a9c354c4a2934aa4bce\nPartial-Bug: #1862032\nDepends-On: https://review.opendev.org/#/c/710755/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/051af731d9b80d449056cceff494a7705e8f37f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/051af731d9b80d449056cceff494a7705e8f37f0"}]},"branch":"refs/heads/master"},"649c018b3e7998e796ca9d8ac8624664b8f03f98":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-04-10 12:21:10.000000000","uploader":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"ref":"refs/changes/56/711656/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/56/711656/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/56/711656/8"}}},"commit":{"parents":[{"commit":"89c38f4aba54c343c38454791ed84a0df0af46e5","subject":"Merge \"Added new API test_sg_rules_quota_values.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/89c38f4aba54c343c38454791ed84a0df0af46e5"}]}],"author":{"name":"Igor Malinovskiy","email":"u.glide@gmail.com","date":"2020-03-06 15:30:47.000000000","tz":120},"committer":{"name":"Igor Malinovskiy","email":"u.glide@gmail.com","date":"2020-04-10 12:20:49.000000000","tz":180},"subject":"Add tempest api tests for subnet pools RBAC","message":"Add tempest api tests for subnet pools RBAC\n\nChange-Id: Ie6e26acd5e9c3acd2ee86a9c354c4a2934aa4bce\nPartial-Bug: #1862032\nDepends-On: https://review.opendev.org/#/c/710755/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/649c018b3e7998e796ca9d8ac8624664b8f03f98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/649c018b3e7998e796ca9d8ac8624664b8f03f98"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
