)]}'
{"id":"openstack%2Fneutron-tempest-plugin~712626","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I0215731f05c7e7e4f955766ab334d8bd3de6d43f","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"attached_to_network","hashtags":[],"change_id":"I0215731f05c7e7e4f955766ab334d8bd3de6d43f","subject":"Add test \"test_create_instance_in_network_with_attached_policy\"","status":"MERGED","created":"2020-03-12 09:21:11.000000000","updated":"2020-03-17 20:30:29.000000000","submitted":"2020-03-17 20:30:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712626-1584477029684-c6fdaf72","meta_rev_id":"24340ecf8e0934ad9dc3483759d35a91941c56ce","_number":712626,"virtual_id_number":712626,"owner":{"_account_id":30920,"name":"Noy fridman","email":"nfridman@redhat.com","username":"nfridman"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-03-17 09:05:45.000000000","_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2020-03-15 13:17:55.000000000","_account_id":30920,"name":"Noy fridman","email":"nfridman@redhat.com","username":"nfridman"},{"value":0,"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-17 20:30:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"value":2,"date":"2020-03-17 15:33:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":30920,"name":"Noy fridman","email":"nfridman@redhat.com","username":"nfridman"},{"value":1,"date":"2020-03-17 09:00:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},{"value":2,"date":"2020-03-17 14:21:42.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-16 07:38:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"value":1,"date":"2020-03-17 15:33:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":30920,"name":"Noy fridman","email":"nfridman@redhat.com","username":"nfridman"},{"value":0,"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":30920,"name":"Noy fridman","email":"nfridman@redhat.com","username":"nfridman"},{"value":0,"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"_account_id":30920,"name":"Noy fridman","email":"nfridman@redhat.com","username":"nfridman"},{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-16 07:38:58.000000000","updated_by":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"reviewer":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"state":"REVIEWER"},{"updated":"2020-03-17 09:00:40.000000000","updated_by":{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},"reviewer":{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},"state":"REVIEWER"},{"updated":"2020-03-17 09:05:45.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"state":"REVIEWER"},{"updated":"2020-03-17 14:21:42.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-03-17 15:33:42.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-03-17 20:30:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c5033b9047fdd02ca97bf6db500ccbc033178ce","author":{"_account_id":30920,"name":"Noy fridman","email":"nfridman@redhat.com","username":"nfridman"},"date":"2020-03-12 09:21:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b49f0fe207f4a4e41b2c506caf279c708aef5699","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 11:28:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67f432b77a164d529d96df704a4514b6 : SUCCESS in 3m 56s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1403ebdde96c4b5a9023ca35d2b4233b : SUCCESS in 54m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/517c1dcd34e440129b76fbc36c7c033a : SUCCESS in 35m 21s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/0754fecaeac6431caf8edaff16a813ed : FAILURE in 1h 56m 06s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6b3645c7117049ac9cf04c6598a7d3fe : SUCCESS in 1h 21m 17s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7116d2c01b0c4ddeb702b61aa1ffd785 : SUCCESS in 1h 20m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2c47adb26afe4ec0a37d0e7f8c4960e2 : SUCCESS in 1h 28m 40s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/f738dd2f74b448dc927a3744e32a4b0d : SUCCESS in 50m 34s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/2f797f01c1b141e398819ca6d911886b : SUCCESS in 36m 36s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/33e28d8b86e24323a9ec2cd3c0de8c3a : FAILURE in 1h 40m 20s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/69fa53b9bcfd452d81fb17435a995df1 : SUCCESS in 1h 14m 50s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/0cf864ab5e7a4f7994f49b4ec6aefdc0 : SUCCESS in 1h 16m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/e2cf20d740354afb8a35bd6b7e508051 : SUCCESS in 1h 07m 31s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/acc750b9968f4f548438be4ecd4738f6 : SUCCESS in 1h 00m 42s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/ba50b7aa87604e3dab7045812d0d9112 : SUCCESS in 38m 22s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/083d3d11e2414fc9b43730d7a2feb90d : SUCCESS in 1h 34m 00s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/d43b079bc0f449648e6152b86a9e049e : SUCCESS in 1h 14m 48s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/baed16e1826a46d6a3f341f3e6441aa1 : SUCCESS in 1h 15m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/a79e4c5dca5f489db6620120c4e7b69b : SUCCESS in 1h 15m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c9786d738cb4754a44503a08323bfa4 : FAILURE in 3m 42s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/14d8fc6e774c49ff9b068226ed322024 : SUCCESS in 1h 53m 50s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/e7057dfab99948afbb36e8ff38cd6a30 : SUCCESS in 1h 46m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/67aff9a6dde54ebeb45afb5c26f835e1 : SUCCESS in 1h 24m 58s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/659c5a2dcee745c9a393b0df593bffba : SUCCESS in 1h 30m 08s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/755ef0fd997f4b35b7dff16c87f56403 : SUCCESS in 42m 33s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/6dd999faa30948eaa6f3d1e54529449e : SUCCESS in 41m 48s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/cd43f374c5754f8aaf0b8329a5772d20 : SUCCESS in 51m 56s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/601b935e0ec24251ba55a14fe14eab05 : SUCCESS in 40m 45s","accounts_in_message":[],"_revision_number":1},{"id":"c0e959eabf0783426fe565f34c0100b47a1e9c2f","author":{"_account_id":30920,"name":"Noy fridman","email":"nfridman@redhat.com","username":"nfridman"},"date":"2020-03-12 11:50:59.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e4e9a1a8e0333a9e430bcf69f290b1eadd6eb5f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 13:41:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8c5a51ff3eb4aada29f00ad3131eaae : SUCCESS in 5m 04s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b0edccae903b46b4a25e5d4750bd3710 : SUCCESS in 1h 07m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/70f5efa81b0640df9fd571ef0f4c9cc4 : SUCCESS in 35m 58s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/3dd056c0cd7e43f9be9d69d52f3083f8 : FAILURE in 1h 45m 33s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9f0d6413349b48498e9eee3225c6a356 : SUCCESS in 1h 21m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ba9ac55e46b64aea86fddbbda71591d7 : SUCCESS in 1h 04m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a0c5bead5bfc4fa3b0241fa6810e3a83 : SUCCESS in 1h 06m 46s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/40779615b4074acd91e1b20c727a800a : SUCCESS in 51m 58s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/f68b150f00c749c8b10378ba911b4f99 : SUCCESS in 39m 58s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/9a9059c975004da6be994f01c86dae3d : FAILURE in 1h 31m 41s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/936aefb3f4234c3cb0a0d2b956041de3 : SUCCESS in 1h 00m 02s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/728c136f594b402c842271827687357b : SUCCESS in 1h 02m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/3cd2666c2bcf460faf24d5a7cd35f1a2 : SUCCESS in 54m 19s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/1536fb85937d4609a9340f35ff08158a : SUCCESS in 47m 39s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/41b6b09f64464eb99fc4b6ce3f600338 : SUCCESS in 33m 18s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/fd0c6f9fd3304226b65dd4445ddf50ed : SUCCESS in 1h 21m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/bd5f7c82f9a94a22966e3edcea2c378a : SUCCESS in 1h 07m 39s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/ad92ff0db76a4084a34e1df656b97684 : SUCCESS in 1h 10m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/f378aa61104c4bdd8edeca337db5fb16 : SUCCESS in 1h 03m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/129dda46513f43a9b385fd288672d011 : SUCCESS in 4m 08s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/d490761a86084fc09da5f0ce0b32cf5b : SUCCESS in 1h 38m 53s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/9e74921d28e043e6996bb8a64bc34d03 : SUCCESS in 1h 36m 31s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/6c0fffa1159d48b28b20a6f602250f5c : SUCCESS in 1h 20m 33s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/383fea25b43a40878b7f89c24aab2611 : SUCCESS in 1h 19m 58s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/266d3e9a406847629e4b03475340ab3c : SUCCESS in 35m 00s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/2b7d92a8768a4a0da0d3a1694ae2f72a : SUCCESS in 37m 39s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/4dacc3a7f30f46b19f11273dae14fa50 : SUCCESS in 46m 33s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/8de4f14256c6458faca242ba0abb5413 : SUCCESS in 36m 58s","accounts_in_message":[],"_revision_number":2},{"id":"e7290f2837adfd5f687b116762cb5e165a177593","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2020-03-12 14:32:43.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n1) I think that it would be great to complete the \"End To End\" scenario in this particular test case, in fact you have all needed stuff to accomplish that.\nSo in addition to \"IDs comparisons\" done in assertions, I\u0027d add file download, to ensure that configured and actual bandwidth limit are the same.\n2) Comments are missing, it would be better to add \"test case scenario\" (after def test_...) + in other places where you think it could be relevant.","accounts_in_message":[],"_revision_number":2},{"id":"7c6fa54aceb8cfc162bcc1ecc2ef9c67da285f23","author":{"_account_id":30920,"name":"Noy fridman","email":"nfridman@redhat.com","username":"nfridman"},"date":"2020-03-15 09:28:48.000000000","message":"Patch Set 2:\n\nHi Arkady, \nThe end-to-end scenario is already done in \"test_qos_basic_and_update\", so once the policy is attached, there is no difference between the cases.\nHowever,the propose of this test is to ensure that the launching process (including creation and connectivity) is done successfully.","accounts_in_message":[],"_revision_number":2},{"id":"917cf36f26bbe86a0a13763a2d1678bc3992c06e","author":{"_account_id":30920,"name":"Noy fridman","email":"nfridman@redhat.com","username":"nfridman"},"date":"2020-03-15 09:29:55.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b87c691c062692f98912038d362971e3b24dd74b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-15 12:37:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9b944c6ddff48b4929e66871fd39605 : SUCCESS in 3m 38s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7e58813b2d7e4d62b50aacfaa697fa45 : SUCCESS in 53m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/513ef9c280874be9b8ddeeb57a4de8af : SUCCESS in 35m 44s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/991c4ffa024d4978aa551086b31af3e2 : FAILURE in 2h 27m 57s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fa69f89e45ae46be9c44c343a0fb6155 : SUCCESS in 1h 54m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/cf0b9df3a1034f90bb9d5b1377b228c7 : SUCCESS in 1h 21m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/768137789f8c4166802421d133a356ef : SUCCESS in 1h 49m 08s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/0b4bf3c74b80449c92752d7c63549d52 : SUCCESS in 1h 46m 18s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/79e49b4c43004993927c3c5368fd0b53 : SUCCESS in 50m 34s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/d02451b9df374a3ebb68c506244d5863 : FAILURE in 1h 15m 55s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/b6fa39dc9fd143299b870d2c3db70141 : FAILURE in 1h 53m 45s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/22459491b7ea4334851c6ac08058b484 : SUCCESS in 2h 08m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/9e41f74e328449afad1566bd10a7c55a : SUCCESS in 2h 03m 41s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/c77f280ac6cb4cd4bb56359a9aec2bd6 : SUCCESS in 1h 54m 58s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/eae6918714904ee0adcca692e280057e : SUCCESS in 59m 45s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/18ffe4fac29b4ded8eb60bf9a3d3b9de : TIMED_OUT in 3h 04m 09s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/711e1f11d98946bf95675ff6927a3c4f : FAILURE in 2h 10m 30s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/00b29d40fffc481ca4db0e886d27fdfd : FAILURE in 2h 11m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/75f77a58216948d88f01e29294b203f9 : FAILURE in 1h 56m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e87fd5a5d2f4c71835531fafbd57661 : SUCCESS in 4m 40s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/8f8a335afb65419794c029d2377337e8 : SUCCESS in 2h 32m 38s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/55c08a3cd6da4e8482a8c8fc20a9b2ee : SUCCESS in 2h 40m 25s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/de765e03a27d4816a363adee3fafac0f : SUCCESS in 1h 35m 48s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/03bbb0c35ae64de8843d7825fb9f5bdb : TIMED_OUT in 2h 05m 18s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/cae6b571815c4d009867ecf102da2264 : SUCCESS in 39m 09s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/4b0bbae74a1d4c4a9c259d3f378a7fbc : SUCCESS in 53m 11s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/775fb7d2280540659635f2cb90885773 : SUCCESS in 48m 56s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/0c7a8a00dd0841e4877d20449a411bff : SUCCESS in 33m 46s","accounts_in_message":[],"_revision_number":3},{"id":"c52abf12be3e9a227c806e88c03ee613084f247d","author":{"_account_id":30920,"name":"Noy fridman","email":"nfridman@redhat.com","username":"nfridman"},"date":"2020-03-15 13:17:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f7e60215213d8bbe6a6a8b02cc2186a9ba523bba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-15 16:07:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83d2b08371544960aff83e52419a2797 : SUCCESS in 5m 26s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bd4cd8b34eea4ba3a5a151e09271c5f8 : SUCCESS in 50m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/af294caaca7b4d56aee1d009aa1b429f : SUCCESS in 38m 05s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/f7c951c326d3496f83526e015f59b010 : FAILURE in 1h 57m 09s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/15bc9d1f392e46e49a1be6fd8a86998e : SUCCESS in 2h 47m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/bde6bdcb473a499d8acc6a24be5883fd : SUCCESS in 1h 06m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c2a8e5ea994f485187573dfb56c99d53 : SUCCESS in 1h 03m 50s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/c049de26fc1540c393135a198b2f2afa : SUCCESS in 59m 13s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/6e8ad02a2dcf429b9f4f0eb4bdb1fb76 : SUCCESS in 33m 02s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/a20f84e5493649ddaf9a6458562eb003 : FAILURE in 1h 42m 18s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/65964c82128f46f18d5c07a9c39dd024 : SUCCESS in 1h 15m 02s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/2698555eecc94c939336a0967f66da34 : SUCCESS in 1h 19m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/62bcd7ec5f0848e28816776281497154 : SUCCESS in 1h 05m 05s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/c2fb27a718e74c61ba5891016ba59848 : SUCCESS in 56m 30s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/42e9a3b2ad21471bafafea7f7cb51634 : SUCCESS in 33m 53s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/c3d8407206cb4cb990a67645508bed05 : SUCCESS in 1h 39m 33s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/e05d544505984c588fd48ce6e36b1376 : SUCCESS in 1h 16m 28s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/c2a73bfd5a954ca9a06e32994ed474dd : SUCCESS in 1h 10m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/559efb47082e4ec5be6af5588a8dc2d9 : SUCCESS in 1h 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50d5bdf0cec94b37a070cea1fe126735 : SUCCESS in 3m 37s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/cfb5e2cdb546492d87db8f5f4bd75044 : SUCCESS in 1h 43m 06s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/b98ba4d141584665854f3f630cc13e5b : SUCCESS in 1h 38m 58s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/3f8617fa474c402fb1259aa3199e1f0e : SUCCESS in 1h 25m 39s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/2744aee340d34d158caace82f5bb0ed9 : SUCCESS in 1h 35m 00s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/aafe428f92db4d519753fa169f04350e : SUCCESS in 38m 16s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/e55fca3a029344bb9c710e0bb30494ab : SUCCESS in 37m 26s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/1e0bfe2db9c040cea9d1476955458e9a : SUCCESS in 53m 41s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/89fb182568ce4bb18d7a461fd0633be9 : SUCCESS in 39m 34s","accounts_in_message":[],"_revision_number":3},{"id":"21514bd99dc79bb5ed6b8613c8900d73f0d1e25c","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2020-03-16 07:38:58.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8e1988a2a6e3316ef4804063199d023f7e5e2475","author":{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},"date":"2020-03-17 09:00:40.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2d667b4ce10b969e1e58fedc494cc7f894550293","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2020-03-17 09:05:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4fc9a56b875167c40b97f5e42780aa9a864e71e3","author":{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},"date":"2020-03-17 09:07:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c1d0ecb8594b3fc1d984466ac4c1badefbe3d7d2","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-03-17 14:21:42.000000000","message":"Patch Set 3: Code-Review+2\n\nLooks good to me.  Thanks!","accounts_in_message":[],"_revision_number":3},{"id":"77f6a3c36d4197749fad5606dc3e1df40ba1b9f3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-17 15:33:42.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"255224dd3804076ac2d974507851b64c307e0aa9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 15:36:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ce924431d8766e2bcce5433ed9e4928117bc5530","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 20:30:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1494e842cdfb4ec793b73bbdf233ebca : SUCCESS in 4m 36s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ca3baea6b59546c399e680196bc0fa22 : SUCCESS in 53m 09s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0df7d0759b77412a8d9c9bf3bac69773 : SUCCESS in 1h 17m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3ee84bde67bb48058a035503665c12b7 : SUCCESS in 1h 34m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b8c35a983e154a5a8b1ae6a4d71e9008 : SUCCESS in 1h 25m 13s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/dff47653cb5c49029754fe1f1d69e431 : SUCCESS in 1h 04m 09s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/a2ae200bd62f44f084a2851be7d1b27b : SUCCESS in 59m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/001e26c3b0c6484984e835d759b06418 : SUCCESS in 4m 03s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/68901d515e1640a7996d130240c36430 : SUCCESS in 1h 34m 50s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a6cb2ab502254c7b98e8e72f1e090613 : SUCCESS in 1h 34m 36s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/aa1283dd94f44fff9de2ee066198d4d1 : SUCCESS in 54m 51s","accounts_in_message":[],"_revision_number":3},{"id":"b4aeb26f23d86f72e8957e0b6fbe696445ed89ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 20:30:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5451fabce12a072447d8105334b58206daaeaa96","revisions":{"d1a0f61a022ecb8962910708477a67a607b7e272":{"kind":"REWORK","_number":1,"created":"2020-03-12 09:21:11.000000000","uploader":{"_account_id":30920,"name":"Noy fridman","email":"nfridman@redhat.com","username":"nfridman"},"ref":"refs/changes/26/712626/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/712626/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/712626/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/712626/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/712626/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/712626/1"}}},"commit":{"parents":[{"commit":"b82476000ce29b010ed2246d364e4ed027c7dc1a","subject":"Merge \"Adding new test of security group rules quota increased.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b82476000ce29b010ed2246d364e4ed027c7dc1a"}]}],"author":{"name":"nfridman","email":"nfridman@redhat.com","date":"2020-03-12 09:17:43.000000000","tz":120},"committer":{"name":"nfridman","email":"nfridman@redhat.com","date":"2020-03-12 09:17:43.000000000","tz":120},"subject":"Add test \"test_create_instance_in_network_with_attached_policy\"","message":"Add test \"test_create_instance_in_network_with_attached_policy\"\n\nThe test validates the ability to create new instances\nin a network that has attached QoS policy.\n\nChange-Id: I0215731f05c7e7e4f955766ab334d8bd3de6d43f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d1a0f61a022ecb8962910708477a67a607b7e272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d1a0f61a022ecb8962910708477a67a607b7e272"}]},"branch":"refs/heads/master"},"fb74e919d2c3a6d029849e8f98410651d734dacb":{"kind":"REWORK","_number":2,"created":"2020-03-12 11:50:59.000000000","uploader":{"_account_id":30920,"name":"Noy fridman","email":"nfridman@redhat.com","username":"nfridman"},"ref":"refs/changes/26/712626/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/712626/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/712626/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/712626/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/712626/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/712626/2"}}},"commit":{"parents":[{"commit":"b82476000ce29b010ed2246d364e4ed027c7dc1a","subject":"Merge \"Adding new test of security group rules quota increased.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b82476000ce29b010ed2246d364e4ed027c7dc1a"}]}],"author":{"name":"nfridman","email":"nfridman@redhat.com","date":"2020-03-12 09:17:43.000000000","tz":120},"committer":{"name":"Noy fridman","email":"nfridman@redhat.com","date":"2020-03-12 11:50:30.000000000","tz":0},"subject":"Add test \"test_create_instance_in_network_with_attached_policy\"","message":"Add test \"test_create_instance_in_network_with_attached_policy\"\n\nThe test validates the ability to create new instances\nin a network that has attached QoS policy.\n\nChange-Id: I0215731f05c7e7e4f955766ab334d8bd3de6d43f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/fb74e919d2c3a6d029849e8f98410651d734dacb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/fb74e919d2c3a6d029849e8f98410651d734dacb"}]},"branch":"refs/heads/master"},"5451fabce12a072447d8105334b58206daaeaa96":{"kind":"REWORK","_number":3,"created":"2020-03-15 09:29:55.000000000","uploader":{"_account_id":30920,"name":"Noy fridman","email":"nfridman@redhat.com","username":"nfridman"},"ref":"refs/changes/26/712626/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/712626/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/712626/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/712626/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/712626/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/712626/3"}}},"commit":{"parents":[{"commit":"b82476000ce29b010ed2246d364e4ed027c7dc1a","subject":"Merge \"Adding new test of security group rules quota increased.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b82476000ce29b010ed2246d364e4ed027c7dc1a"}]}],"author":{"name":"nfridman","email":"nfridman@redhat.com","date":"2020-03-12 09:17:43.000000000","tz":120},"committer":{"name":"Noy fridman","email":"nfridman@redhat.com","date":"2020-03-15 09:29:50.000000000","tz":0},"subject":"Add test \"test_create_instance_in_network_with_attached_policy\"","message":"Add test \"test_create_instance_in_network_with_attached_policy\"\n\nThe test validates the ability to create new instances\nin a network that has attached QoS policy.\n\nChange-Id: I0215731f05c7e7e4f955766ab334d8bd3de6d43f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5451fabce12a072447d8105334b58206daaeaa96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5451fabce12a072447d8105334b58206daaeaa96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
