)]}'
{"id":"openstack%2Fneutron-tempest-plugin~721262","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I4046aaf0e7f4171e6ce5751b79b84c943ccf56cd","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"multicast-enhancement","hashtags":[],"change_id":"I4046aaf0e7f4171e6ce5751b79b84c943ccf56cd","subject":"Fix test_multicast_between_vms_on_same_network","status":"MERGED","created":"2020-04-20 12:50:05.000000000","updated":"2020-04-27 11:42:17.000000000","submitted":"2020-04-27 11:42:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721262-1587987737083-8af36960","meta_rev_id":"abf853cb461dea1ccb4d91a7e4f0c1e7b3056425","_number":721262,"virtual_id_number":721262,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-23 12:27:22.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 11:42:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-04-27 09:58:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-04-23 15:16:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-22 14:06:51.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-04-27 09:58:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 18:59:23.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-04-21 15:47:31.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-04-22 14:06:51.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2020-04-23 15:16:10.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-04-27 09:58:41.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-04-27 11:42:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f937fc7f02a79433f9e325fcd3386f71c1d86b3f","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-04-20 12:50:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3937872d03b7fe0ab838639bf244a3e783361f61","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-04-20 12:52:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"51888ebf10e7fba01c7d56a5a833c555ee0597b7","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-04-20 12:54:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fdaa748554eb812c1bbe98026bde5abab9a47a63","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-04-20 12:55:11.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"51a2044a88c0ba488185e9009cdfdd7f7107cb6d","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-04-20 12:56:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d2d7a59e26420353cd1077ccf064cc209486dbba","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-04-20 13:07:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f654efed7b5d588237451c5ebf89121d5e20621","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-04-20 13:18:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8c2ae22441d3f16f69c3a5289a6629b6d7c0c878","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-04-20 15:33:58.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM thanks!","accounts_in_message":[],"_revision_number":2},{"id":"52671075f7922c4a0c6f9b5b17f10e3e650e50df","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-20 18:59:23.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nI\u0027ll push an update for the typo since it looks like zuul went to lunch","accounts_in_message":[],"_revision_number":2},{"id":"12cbe07dff46c9ff5db12882132999f930511203","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-20 19:01:01.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"ed7c6d201d819be200fa5322de2fde975fbd0fe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 02:10:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e6a64f188b148c896782fe01b140e01 : SUCCESS in 5m 06s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c6dcdfeb492949ac9632a43b8d47d83c : SUCCESS in 52m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5a5d095e91004c5785bb54998ec93acd : SUCCESS in 36m 05s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/0d3a04236222411cae6a4c13a8d8c88e : FAILURE in 1h 59m 14s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8ec18c14f76d46228a0b624e04e8b7d3 : SUCCESS in 1h 32m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5d443701bb7a4f238b36a08dd5f6dd00 : SUCCESS in 1h 01m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/601c56242a574c35810941ccc3e3c6d6 : SUCCESS in 1h 19m 33s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/753791b080aa430991e429bbd5cc54d2 : FAILURE in 21m 31s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/3228f130df0543e0aaa5880ccf9cbeae : FAILURE in 20m 14s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/94dab16feb6f44f9830b244bb641cd61 : FAILURE in 25m 54s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/5106be60d4de4356900b5289982a3742 : FAILURE in 21m 32s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/ddb9813259f24294808e2c52d2544e95 : FAILURE in 14m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/705f1908184e4b589f97106b45598230 : FAILURE in 18m 48s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/ab280ddcccc04a75940ab133cf146da4 : SUCCESS in 59m 44s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/f327b6b40aeb4c028013d83de31cf400 : SUCCESS in 38m 00s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/ec24110ca2054e0988fe4db636381d4a : FAILURE in 2h 20m 33s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/1b33d79aa9ef42188c2ff7f024ee2036 : SUCCESS in 1h 00m 55s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/46924ce99c434f7e8e141d751cbd36ef : SUCCESS in 1h 08m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/d7031481ce654b9da5a0be57ce81a571 : SUCCESS in 1h 09m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee4cf32876bf4b7cbad1f36a37ab084c : SUCCESS in 5m 22s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/dc3e895d82ca40ae8bc270eba6c3cb6f : SUCCESS in 1h 38m 03s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/66d9ef66827c4a06b057aaa36a5e0162 : SUCCESS in 1h 40m 49s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d2b7239bd79344edb8b6c324eefbfde4 : SUCCESS in 1h 18m 23s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/28cb3450a6ff409d8e9bd2fb2dd3a5f0 : SUCCESS in 1h 40m 00s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/173a53a1549e4cd5989ea1b0d278b791 : SUCCESS in 38m 10s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/d7a9219ac5aa46d28d4b7d665aa8f18c : SUCCESS in 44m 58s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/1dbc8ec02568428f8b1891f2fada3837 : SUCCESS in 45m 13s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/f3cee7f05fe04e0abfe8a865d967d237 : SUCCESS in 34m 57s","accounts_in_message":[],"_revision_number":3},{"id":"ab5484dad063f6c70ff0d089540dcbbce067bae3","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-04-21 15:16:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ceacce48564b8be0818d5a34df454eaa37e2c07f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 18:56:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c33b9300b9db4c67b61469d991969bf7 : POST_FAILURE in 3m 39s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/de273e37009c4fb7bbcde407a9378e04 : SUCCESS in 51m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e83512d6284b4285b55a50f902272593 : SUCCESS in 36m 27s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/dcc7fbbeef2a4ed2adbe2adbf7d31c00 : FAILURE in 2h 46m 57s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ce72579a43c94b839409e106f269e7d5 : SUCCESS in 1h 23m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/912cd9998be844718fddcc6486587c34 : SUCCESS in 1h 22m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3444ea04c4734feda8828457508dd4f4 : SUCCESS in 1h 08m 16s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/d46ef2e88ce347cdb75a393b84ecc8a7 : FAILURE in 21m 22s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/bbd7b7274c2d48138508033de434a351 : FAILURE in 24m 33s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/558bbf98608f4acd97f9c5ba708a07d9 : FAILURE in 34m 31s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/0aab8ddac12448ac852db770a6048f9e : FAILURE in 20m 04s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/1acabfedcbb2432f9d6b9354bc876370 : FAILURE in 16m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/ee998fda1196449b9c761bca7399277a : FAILURE in 22m 20s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/972d104c922143b3b9d05dc7bf322199 : SUCCESS in 52m 47s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/8ab16968a7e348319b0eaab31e36e6d2 : SUCCESS in 36m 34s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/1f891c60ef064eda87f640937c6d6194 : SUCCESS in 1h 53m 43s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/019425aad2314209ab90e5fab851380d : POST_FAILURE in 1h 16m 39s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/adb68b7cb0104cb89859fb4661582f20 : SUCCESS in 1h 05m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/cce7353d0cef41c882fb8a35b6ee3204 : SUCCESS in 1h 17m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d849141d9124ede869ba7aa71f7b97d : FAILURE in 4m 57s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/be33e62385d549649a2e2254b6c8c41a : SUCCESS in 1h 52m 03s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/ebc15f4622774ae5bd08695648c666bf : SUCCESS in 1h 53m 32s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/76d124a405dd42c38d5960f930f8e167 : SUCCESS in 1h 35m 15s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/087f3b9bb47644568e30a17a204ebe8e : SUCCESS in 1h 33m 31s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/eb2e311b3fbc44f8a422300769e7bbc3 : FAILURE in 39m 15s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/b9319e82e4574901bb2b80511c7a7dd7 : FAILURE in 41m 43s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/d729a8bb6f2b46ddafbdfbcd372dd5b4 : SUCCESS in 49m 59s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/54f38d69344e4fa58d037d5792b11a9d : SUCCESS in 42m 54s","accounts_in_message":[],"_revision_number":4},{"id":"476c02a53373f5fc6665de3aea4c7c7e7f3dd563","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-04-22 10:37:05.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"00dc56a5bec62dbc2d0d2ea1165f97d0455bffca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 12:34:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21aa487493d34db7a4d37b75d91355f2 : SUCCESS in 5m 49s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6de11310b38d4aa99758189473f40073 : SUCCESS in 56m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fef53ac6a32e4c5bb06300e3c0988867 : SUCCESS in 38m 27s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/81e69629689f453ea4957ee0d531d399 : FAILURE in 1h 45m 20s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/de0cc5641806460d83d9e114dfc39e2a : SUCCESS in 1h 18m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c937952f68db476391296dccb9f13660 : SUCCESS in 1h 08m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d7b03ac623ea41158fda2318e854ece3 : SUCCESS in 1h 03m 49s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/949ff574610e404586f35f7ce4e273a3 : SUCCESS in 51m 09s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/dc52b809462b44a6a22a065e39845722 : SUCCESS in 36m 49s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/ceb0df3d3cd14e6ca550611567a05598 : FAILURE in 1h 41m 49s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/8046ff39a077489280b8a65c14d1cba9 : SUCCESS in 1h 25m 16s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/0db8a4b7e8e845cebd53a79e8a2dc5cf : SUCCESS in 1h 28m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/849850fa781848988ae2f67747879565 : SUCCESS in 1h 00m 47s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/d727f08d91244c5a9ffe112e0db3ace2 : SUCCESS in 49m 56s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/89986d6b77f44ee09b34262e8c0e5731 : SUCCESS in 34m 28s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/d4f90399bd8446a195711375460092ca : FAILURE in 1h 54m 46s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/a04aa1bb7d0642f3adca052dccb2970c : SUCCESS in 1h 30m 48s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/971dce07fc71466ba4602a2a1799637b : SUCCESS in 1h 21m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/5312ef81df2947e0aa0cf69758e588d1 : SUCCESS in 1h 23m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34af5b36e60543ed9ca3653e50abd53a : SUCCESS in 5m 25s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/13b9001cf39f454f8467a84db85fd18d : SUCCESS in 1h 38m 52s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/b54c7dde03ba467a815ff6d12482275b : SUCCESS in 1h 38m 53s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/da8c86c07ad5468dbce9459ab4655102 : SUCCESS in 1h 18m 13s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/11f8169d256e45528ea5650541088c9a : SUCCESS in 1h 27m 45s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/67d5be6d5c184a5b82da70b1fa31fa47 : FAILURE in 36m 10s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/f0b127a3b0204c7bbc5bc8ce4365b001 : SUCCESS in 37m 01s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/df3e58abd58b4c9e83c70272868e502b : SUCCESS in 44m 41s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/11f30ed55ed14f9094989f8e4ebb54dd : SUCCESS in 40m 25s","accounts_in_message":[],"_revision_number":5},{"id":"3bc8c2a8bdf36b13750a09011a9bc39229a41a63","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-04-22 12:58:06.000000000","message":"Patch Set 5:\n\nThis is being tested at [0] and it works [1], the configs are being set properly on that patch too [2] [3] \n\n[0] https://review.opendev.org/#/c/721263/\n\n[1] https://da23047d68bf5f8bd56f-715185bc608cc1ee8bc59b9528031d6b.ssl.cf1.rackcdn.com/721263/2/check/neutron-ovn-tempest-ovs-release/0debd09/testr_results.html \n\n[2] https://da23047d68bf5f8bd56f-715185bc608cc1ee8bc59b9528031d6b.ssl.cf1.rackcdn.com/721263/2/check/neutron-ovn-tempest-ovs-release/0debd09/controller/logs/etc/neutron/neutron_conf.txt\n\n[3] https://da23047d68bf5f8bd56f-715185bc608cc1ee8bc59b9528031d6b.ssl.cf1.rackcdn.com/721263/2/check/neutron-ovn-tempest-ovs-release/0debd09/controller/logs/tempest_conf.txt","accounts_in_message":[],"_revision_number":5},{"id":"ab633aff1f40494427951bd679f873a0251c75cb","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-04-22 14:06:51.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"dfd314c20041ba05d7c51b5a31476e95df552d45","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-23 11:38:03.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6a810ffd46e04dddcfd78d5fddb41b9547c92a18","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-04-23 12:27:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"043a5fcc6347222bf0f8def520915974fa239406","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-23 15:16:10.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b457a141d19494110d74612c92a4f32dfb62ddb2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-27 09:58:41.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"401ffd1b19991e5bd51caa5a3cc9d26bcd25709d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 09:58:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f476c8c7f87005352ec3d9fc1a28af8f43948059","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 11:42:16.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/931ee59ac1454b6cadb34dd109de43cf : SUCCESS in 5m 52s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1bbff7254caa400685820a9bd1aa2361 : SUCCESS in 48m 50s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e76d7fb12aa846abb5e1a515593c3339 : SUCCESS in 1h 16m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2602150c76d64cef9caf31c961b65fbe : SUCCESS in 1h 30m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/82ef2d8fbf204d728c902725309c6fa7 : SUCCESS in 1h 04m 02s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/f35b32b09e854f5496ab120e83e1e2dc : SUCCESS in 58m 04s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/03a1d6d126b841c48c8f257f37d2bc07 : SUCCESS in 54m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12a335fd5db2471a894933163732e523 : SUCCESS in 6m 24s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/6c0fbc496c464427a14ca01b4929e324 : SUCCESS in 1h 38m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/6bf19ccbf76a411eaf6016581f8afc8a : SUCCESS in 1h 28m 59s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/f12b6962ffba4e01b470d794456831dd : SUCCESS in 48m 52s","accounts_in_message":[],"_revision_number":5},{"id":"fcea31266ed92e4a7de13273ada0ad4dd48fbbbe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 11:42:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"023396f59facd2caa28e992b85102ff2befcf467","revisions":{"35e96fd5d709562e936983c0aa48ea2edbf82ec0":{"kind":"REWORK","_number":1,"created":"2020-04-20 12:50:05.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/62/721262/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/62/721262/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/721262/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/721262/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/721262/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/721262/1"}}},"commit":{"parents":[{"commit":"cad2c81b5427af6a26293742d9b2605b73eebfa0","subject":"Merge \"Replace old requires-python with python-requires\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/cad2c81b5427af6a26293742d9b2605b73eebfa0"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-04-20 12:44:48.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-04-20 12:44:48.000000000","tz":60},"subject":"Fix test_multicast_between_vms_on_same_network","message":"Fix test_multicast_between_vms_on_same_network\n\nThe test_multicast_between_vms_on_same_network had a few errors, one\nrelated to the \"mcast_address\" being an IPAddress object and not a\nstring. This has been fixed by converting the object to its string\nrepresentation before trying to do a startswith() operation on it.\n\nThe second problem was related to the execution of the\nunregistered_traffic_receiver.sh script. The script was created by\nthe default SSH user in its home directory but, when SSH\u0027ing the VM to\nexecute it we had the become_root parameter set to True, which caused\nthe command to look for the script in the /root directory. This has been\nfixed by creating and executing the script from the /tmp directory cos\nit\u0027s userless.\n\nThis test also takes a while to run since it requires at least 3 VMs,\ntherefore this patch also marks it as \"slow\".\n\nChange-Id: I4046aaf0e7f4171e6ce5751b79b84c943ccf56cd\nCloses-Bug: #1873798\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/35e96fd5d709562e936983c0aa48ea2edbf82ec0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/35e96fd5d709562e936983c0aa48ea2edbf82ec0"}]},"branch":"refs/heads/master"},"c68244faa5f77cccd441374dc46c2d016685867e":{"kind":"REWORK","_number":2,"created":"2020-04-20 12:56:02.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/62/721262/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/62/721262/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/721262/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/721262/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/721262/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/721262/2"}}},"commit":{"parents":[{"commit":"cad2c81b5427af6a26293742d9b2605b73eebfa0","subject":"Merge \"Replace old requires-python with python-requires\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/cad2c81b5427af6a26293742d9b2605b73eebfa0"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-04-20 12:44:48.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-04-20 12:55:53.000000000","tz":60},"subject":"Fix test_multicast_between_vms_on_same_network","message":"Fix test_multicast_between_vms_on_same_network\n\nThe test_multicast_between_vms_on_same_network had a few errors, one\nrelated to the \"mcast_address\" being an IPAddress object and not a\nstring. This has been fixed by converting the object to its string\nrepresentation before trying to do a startswith() operation on it.\n\nThe second problem was related to the execution of the\nunregistered_traffic_receiver.sh script. The script was created by\nthe default SSH user in its home directory but, when SSH\u0027ing the VM to\nexecute it we had the become_root parameter set to True, which caused\nthe command to look for the script in the /root directory. This has been\nfixed by creating and executing the script from the /tmp directory cos\nit\u0027s userless.\n\nThis test also takes a while to run since it requires at least 3 VMs,\ntherefore this patch also marks it as \"slow\".\n\nChange-Id: I4046aaf0e7f4171e6ce5751b79b84c943ccf56cd\nCloses-Bug: #1873798\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c68244faa5f77cccd441374dc46c2d016685867e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c68244faa5f77cccd441374dc46c2d016685867e"}]},"branch":"refs/heads/master"},"c62aaeb4e4f8ef4b9e5eb6ce818958303c7bbc89":{"kind":"REWORK","_number":3,"created":"2020-04-20 19:01:01.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/721262/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/62/721262/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/721262/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/721262/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/721262/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/721262/3"}}},"commit":{"parents":[{"commit":"cad2c81b5427af6a26293742d9b2605b73eebfa0","subject":"Merge \"Replace old requires-python with python-requires\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/cad2c81b5427af6a26293742d9b2605b73eebfa0"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-04-20 12:44:48.000000000","tz":60},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2020-04-20 19:00:58.000000000","tz":0},"subject":"Fix test_multicast_between_vms_on_same_network","message":"Fix test_multicast_between_vms_on_same_network\n\nThe test_multicast_between_vms_on_same_network had a few errors, one\nrelated to the \"mcast_address\" being an IPAddress object and not a\nstring. This has been fixed by converting the object to its string\nrepresentation before trying to do a startswith() operation on it.\n\nThe second problem was related to the execution of the\nunregistered_traffic_receiver.sh script. The script was created by\nthe default SSH user in its home directory but, when SSH\u0027ing the VM to\nexecute it we had the become_root parameter set to True, which caused\nthe command to look for the script in the /root directory. This has been\nfixed by creating and executing the script from the /tmp directory cos\nit\u0027s userless.\n\nThis test also takes a while to run since it requires at least 3 VMs,\ntherefore this patch also marks it as \"slow\".\n\nChange-Id: I4046aaf0e7f4171e6ce5751b79b84c943ccf56cd\nCloses-Bug: #1873798\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c62aaeb4e4f8ef4b9e5eb6ce818958303c7bbc89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c62aaeb4e4f8ef4b9e5eb6ce818958303c7bbc89"}]},"branch":"refs/heads/master"},"d00841c1c35ac8ddd3978532742fe65ce4911ffb":{"kind":"REWORK","_number":4,"created":"2020-04-21 15:16:33.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/62/721262/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/62/721262/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/721262/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/721262/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/721262/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/721262/4"}}},"commit":{"parents":[{"commit":"cad2c81b5427af6a26293742d9b2605b73eebfa0","subject":"Merge \"Replace old requires-python with python-requires\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/cad2c81b5427af6a26293742d9b2605b73eebfa0"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-04-20 12:44:48.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-04-21 15:16:27.000000000","tz":60},"subject":"Fix test_multicast_between_vms_on_same_network","message":"Fix test_multicast_between_vms_on_same_network\n\nThe test_multicast_between_vms_on_same_network had a few errors, one\nrelated to the \"mcast_address\" being an IPAddress object and not a\nstring. This has been fixed by converting the object to its string\nrepresentation before trying to do a startswith() operation on it.\n\nThe second problem was related to the execution of the\nunregistered_traffic_receiver.sh script. The script was created by\nthe default SSH user in its home directory but, when SSH\u0027ing the VM to\nexecute it we had the become_root parameter set to True, which caused\nthe command to look for the script in the /root directory. This has been\nfixed by creating and executing the script from the /tmp directory cos\nit\u0027s userless.\n\nChange-Id: I4046aaf0e7f4171e6ce5751b79b84c943ccf56cd\nCloses-Bug: #1873798\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d00841c1c35ac8ddd3978532742fe65ce4911ffb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d00841c1c35ac8ddd3978532742fe65ce4911ffb"}]},"branch":"refs/heads/master"},"023396f59facd2caa28e992b85102ff2befcf467":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-04-22 10:37:05.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/62/721262/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/62/721262/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/721262/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/721262/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/721262/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/62/721262/5"}}},"commit":{"parents":[{"commit":"0bdc63d7c1c3ed192b7206ec9ac25e4380fabbd2","subject":"Only install plugin system-wide if INSTALL_TEMPEST is set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0bdc63d7c1c3ed192b7206ec9ac25e4380fabbd2"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-04-20 12:44:48.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-04-22 10:37:05.000000000","tz":0},"subject":"Fix test_multicast_between_vms_on_same_network","message":"Fix test_multicast_between_vms_on_same_network\n\nThe test_multicast_between_vms_on_same_network had a few errors, one\nrelated to the \"mcast_address\" being an IPAddress object and not a\nstring. This has been fixed by converting the object to its string\nrepresentation before trying to do a startswith() operation on it.\n\nThe second problem was related to the execution of the\nunregistered_traffic_receiver.sh script. The script was created by\nthe default SSH user in its home directory but, when SSH\u0027ing the VM to\nexecute it we had the become_root parameter set to True, which caused\nthe command to look for the script in the /root directory. This has been\nfixed by creating and executing the script from the /tmp directory cos\nit\u0027s userless.\n\nChange-Id: I4046aaf0e7f4171e6ce5751b79b84c943ccf56cd\nCloses-Bug: #1873798\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/023396f59facd2caa28e992b85102ff2befcf467"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/023396f59facd2caa28e992b85102ff2befcf467"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
