)]}'
{"id":"openstack%2Fneutron-tempest-plugin~733822","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I04056528ada7fda538cf8b285d9379626202d51d","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"run-scenario-serially","hashtags":[],"change_id":"I04056528ada7fda538cf8b285d9379626202d51d","subject":"DNM It\u0027s just test of running tests in scenario jobs in serial","status":"ABANDONED","created":"2020-06-05 12:39:21.000000000","updated":"2020-08-23 08:22:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"35da77f8cd6a0b71c4ae48a2e818343ddba17ca4","_number":733822,"virtual_id_number":733822,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-05 15:16:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-05 15:16:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d7289a84f4b9c56704003de398914404d801a2d3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-05 12:39:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d075de8f0453032fcd9b72f2c136d7c01cd34ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 15:16:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/009efcc63b094e56a5935ff9aaf6cb19 : SUCCESS in 4m 33s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/dfd712a74cea49daaaa8d41695dc4e95 : SUCCESS in 58m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/546148eb7c634b5fbc8c7045d9cecb43 : SUCCESS in 41m 45s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/e0b4a905c59c486c82797be8b16485b3 : FAILURE in 1h 51m 06s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/04f9ef386fc741aea3f554945a545c4d : FAILURE in 1h 58m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3bf9154cb726439d8ab235ce20ce4cc1 : SUCCESS in 1h 41m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1a76df5b9a7f4a829f89c724e24395d4 : SUCCESS in 1h 36m 16s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/92c971876403449da9ece1666facdb64 : SUCCESS in 48m 22s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/37659359ba4945a4986b0093bfc53cb9 : SUCCESS in 36m 54s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/0948f080684240cbb3ac5790b069247e : FAILURE in 1h 54m 37s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/f4c901f5724946a4bca13e1ee45633b1 : SUCCESS in 2h 08m 21s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/0d186e054ebb49eebfa9cd6553a4312c : SUCCESS in 1h 55m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/b8cd7c0ad5ae4d72a7f40d0d801c49ab : SUCCESS in 1h 51m 09s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/4a585c476e554c9aaf27c09f7c27e860 : SUCCESS in 54m 48s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/cfe4a95d78ed427fac91b6099f4adb7d : SUCCESS in 36m 30s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/9c491772ae4b4b40aa82bf8832ca4693 : FAILURE in 1h 35m 54s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/ee75b9c5430c4fdc99d5dd020788d2f7 : SUCCESS in 1h 57m 44s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/d7f6201ddc1d4ac7a1c3c25b08e5a692 : SUCCESS in 1h 46m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/bb416e6706ce47e59864e59357d1255a : SUCCESS in 1h 44m 16s\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/f49d050276694c65bd907b0b036874e7 : SUCCESS in 1h 00m 36s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/e3e02971670e420cabeaa6925c71b1ed : SUCCESS in 24m 21s\n- neutron-tempest-plugin-dvr-multinode-scenario-ussuri https://zuul.opendev.org/t/openstack/build/6d7700a467cc4bb38060b415bf2c9a96 : FAILURE in 2h 16m 20s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/6a37a66184cb4d5ebf2891a39a74c2fa : SUCCESS in 2h 18m 46s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/9834474f1f534286851bf2ebff102cec : SUCCESS in 2h 02m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/3da08bb0b8aa477c90e7a82278077eb4 : SUCCESS in 2h 00m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/715e5b1741f84b848bc404b37041a736 : SUCCESS in 4m 09s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/ae078f07cfcb4559b6743e545ce4e150 : SUCCESS in 1h 30m 35s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/96e78838c68c4a57ae9eff333ea08551 : SUCCESS in 1h 37m 34s\n- neutron-tempest-plugin-sfc-ussuri https://zuul.opendev.org/t/openstack/build/f867aa8ff6664945b312b773e3edb918 : SUCCESS in 1h 36m 43s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/b97321c7c17f41ed8cf12e0507ed9404 : FAILURE in 1h 26m 06s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/c36c9da9215947828e62fa654c2dc8ca : SUCCESS in 1h 32m 10s\n- neutron-tempest-plugin-bgpvpn-bagpipe-ussuri https://zuul.opendev.org/t/openstack/build/cf03e1cff4214a8abf2d1fcbae7d1af2 : SUCCESS in 1h 25m 24s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/a50ec536c6064e99a0f4cde50e70901c : FAILURE in 34m 18s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/c0371478490d4e8c87d13bd7befd5142 : SUCCESS in 37m 26s (non-voting)\n- neutron-tempest-plugin-fwaas-ussuri https://zuul.opendev.org/t/openstack/build/2fc26eb58a4740f18b4d74574751e354 : SUCCESS in 35m 26s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/76fee8a0575847eb9449243ce5e90ade : SUCCESS in 54m 07s\n- neutron-tempest-plugin-dynamic-routing-ussuri https://zuul.opendev.org/t/openstack/build/40beaccaf0b442a19570335b6a71b91c : SUCCESS in 53m 24s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/1ffec33e42834dd9a3aa2d10deb6bcf8 : SUCCESS in 38m 43s\n- neutron-tempest-plugin-vpnaas-ussuri https://zuul.opendev.org/t/openstack/build/2396a5840921405fbb190648c3ab1561 : FAILURE in 42m 47s","accounts_in_message":[],"_revision_number":1},{"id":"6ea3811cb23535d1a47b7bc96fcc41ba93d0354e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-23 08:22:50.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"101312f3706ef994c420d366edead53acb0942f1","revisions":{"101312f3706ef994c420d366edead53acb0942f1":{"kind":"REWORK","_number":1,"created":"2020-06-05 12:39:21.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/22/733822/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/22/733822/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/22/733822/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/22/733822/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/22/733822/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/22/733822/1"}}},"commit":{"parents":[{"commit":"e3478457c7756daf307eb67a82cbc6b39cf557d4","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e3478457c7756daf307eb67a82cbc6b39cf557d4"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-06-05 12:38:44.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-06-05 12:38:44.000000000","tz":120},"subject":"DNM It\u0027s just test of running tests in scenario jobs in serial","message":"DNM It\u0027s just test of running tests in scenario jobs in serial\n\nChange-Id: I04056528ada7fda538cf8b285d9379626202d51d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/101312f3706ef994c420d366edead53acb0942f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/101312f3706ef994c420d366edead53acb0942f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
