)]}'
{"id":"openstack%2Fneutron-tempest-plugin~741695","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I761f5f199b7a2dd045cdc503223333900bfeadf0","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/1887992","hashtags":[],"change_id":"I761f5f199b7a2dd045cdc503223333900bfeadf0","subject":"[WIP] Test GLANCE_STANDALONE\u003dFalse to fix glance in CI jobs","status":"ABANDONED","created":"2020-07-17 18:08:12.000000000","updated":"2020-07-18 23:07:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"97f21511f6c8bc1163e2aad3c3cd38a8f3e1ef84","_number":741695,"virtual_id_number":741695,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"date":"2020-07-18 11:47:07.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-18 14:25:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-18 08:42:08.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-07-18 11:47:07.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-07-18 14:25:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3bd20b47a824e8a57d4860f15826bd3ea7648d91","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-17 18:08:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb153f0502ac004124c7a52f12ed855bbf0e554b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 02:24:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37d317a2da8b4d54b8ab73a846d0abae : SUCCESS in 6m 21s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b0ee4a394b764f0b8f35ea60e7280b6c : SUCCESS in 51m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1db7f53bdebd4664b940611e9a997ee2 : SUCCESS in 31m 30s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c7c5ff039ba34d9986ce343506b141a6 : SUCCESS in 1h 08m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/502d80e87fb84ac88be21b1bb3fe336f : SUCCESS in 1h 21m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0ae5a425c376409a8ffc021f9a88d0ad : SUCCESS in 1h 05m 35s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/490c7e734987402d89146170777c2472 : SUCCESS in 47m 13s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/2b84b83e6de14aefb4175e4025bd5037 : SUCCESS in 56m 34s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/600db6de9b4648b8933d452e6c39eed5 : SUCCESS in 49m 02s\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/70dc96c015a14e3a831bf3c41d1515f0 : SUCCESS in 1h 26m 44s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/e6ab73f08ac74a8c925a326c96568f3d : SUCCESS in 1h 24m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/69270e3b1a9049cca3dfda29e5102af9 : SUCCESS in 1h 03m 50s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/60fd67b2a7974c809f87d138d1f6ba35 : SUCCESS in 57m 47s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/a171f5109dea4f60a441de283e4da1c0 : SUCCESS in 37m 20s\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/af9171a98d45409cba8a3b984ad44f72 : SUCCESS in 1h 32m 03s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/ecbcc24c7efd4a09b4fb40a6737414a0 : SUCCESS in 1h 25m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/5652dc6021d643979a59bed350c7e4f9 : SUCCESS in 1h 13m 36s\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/de0558be1681493d81eb0c0f04365c7d : SUCCESS in 1h 00m 34s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/d60af13652484da8bc235f8c84467c76 : SUCCESS in 34m 48s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/894712e271554fc1a6ead23c38c20571 : SUCCESS in 1h 06m 39s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/b5f357111a424b189839c995be5ea002 : SUCCESS in 58m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/9cca0a15a5074d8ab0582d12eec0f46c : SUCCESS in 1h 03m 42s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/ebabe013c0834b1bbe1aac458bcb7c06 : SUCCESS in 1h 29m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19ea956f8fdd4719a6fc1dd247d3e782 : SUCCESS in 4m 06s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/b540e4f7d70946b19c0f9934d9728d9f : SUCCESS in 1h 34m 23s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/34366f4206084131a4a06e2dfe975d84 : SUCCESS in 1h 25m 44s\n- neutron-tempest-plugin-sfc-ussuri https://zuul.opendev.org/t/openstack/build/1972072056714962b9b9de3718dc47c2 : SUCCESS in 1h 39m 18s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/ce2f04f14dde4e1a8d70fc00784f02a9 : SUCCESS in 1h 09m 22s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/9b319ba01b314df895e440d5457b23d9 : SUCCESS in 1h 19m 11s\n- neutron-tempest-plugin-bgpvpn-bagpipe-ussuri https://zuul.opendev.org/t/openstack/build/26cdec1d36d74a5da635aaf2b3f34499 : SUCCESS in 1h 12m 35s\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/98791295b8a04aa3b90af415445d5ea1 : FAILURE in 9m 01s (non-voting)\n- neutron-tempest-plugin-fwaas-ussuri https://zuul.opendev.org/t/openstack/build/5f2aad4a8dfa4db9a08dd82965ed4740 : FAILURE in 36m 25s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/fe1f7a89c5b34080ad21e8a0dd3e29dc : SUCCESS in 43m 03s\n- neutron-tempest-plugin-dynamic-routing-ussuri https://zuul.opendev.org/t/openstack/build/f801854605114fe49fca425433aef54e : SUCCESS in 48m 20s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/1c10fc1e9ba04d8c83a30fa29ff882f4 : SUCCESS in 39m 11s\n- neutron-tempest-plugin-vpnaas-ussuri https://zuul.opendev.org/t/openstack/build/1b77e1c7fba04bbab3ae73af383575c6 : SUCCESS in 37m 23s","accounts_in_message":[],"_revision_number":1},{"id":"79008efcc869a012be12613eadfc12b1a10ad1c2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-18 10:02:58.000000000","message":"Patch Set 1:\n\ndevstack patch 741687 is fixing the problem of neutron tempest plugin jobs when tls-proxy service is disabled. Probably this patch won\u0027t be needed. The neutron patch testing the referred one is working fine.","accounts_in_message":[],"_revision_number":1},{"id":"1abac9374259606cd39bbfd77738ecc9aedc87c8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-18 11:46:28.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9e3ae814b4b11107f360c0e60d8b1602da5d0568","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-18 11:47:07.000000000","message":"Patch Set 2:\n\ni added depends-on to test this other combination (not standalone and not tls) after the patch","accounts_in_message":[],"_revision_number":2},{"id":"7acacd96d9f899755e74b002adcfaefb80a3487f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 14:25:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22fcc5691a184d46b3973c8570c644a0 : SUCCESS in 4m 12s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c43159ecf9224754ae00601a5a95bc3a : SUCCESS in 47m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c5dee28c55f3492984a96056fe1f36b5 : SUCCESS in 33m 43s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/77e2c34da9df468eb8d77f38d1f949d4 : SUCCESS in 1h 20m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e7575cadde914c468730342da2b090dd : SUCCESS in 59m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/31aa8448ddf642c58f7cc3bff4313e5d : SUCCESS in 1h 07m 18s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9eabb025041f4aedb7b98b134449fc5e : SUCCESS in 51m 16s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/a0a140f5bf8d4d718b4cbc14aa74dd2c : SUCCESS in 53m 36s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/5a459a23584d40879952c77e34d6e17d : SUCCESS in 32m 03s\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/709fddbe61ff4f21a0d5367847cf46dd : SUCCESS in 1h 16m 20s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/8a93baea838d451e907afb6396212327 : SUCCESS in 1h 08m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/e22b8ae4c3fe4b65a3fccd4a379f5a22 : SUCCESS in 1h 16m 18s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/bb724b4bbe2f4b63929ebfb7c8549b6e : SUCCESS in 52m 39s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/780c70f9ffdb4b849df753f1a3d7eed7 : SUCCESS in 37m 00s\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/3b61cc0b6ae442c0ac8516040b6aeb59 : SUCCESS in 1h 20m 59s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/744c54101da4486fa019616542fb1860 : SUCCESS in 1h 09m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/1ca79cd59dd640db846d7a049a74ee8b : SUCCESS in 1h 00m 58s\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/c6d400e5ef6a425aa79bcfb9840535cb : SUCCESS in 54m 25s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/8aa602205a164d539c1ec2ecdafadb65 : SUCCESS in 28m 29s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/638b99aae0744125be387c440e9d07f1 : SUCCESS in 1h 13m 21s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/176a752f02794cb39ebc76ff95d53c05 : SUCCESS in 54m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/b4c3e81b4b9f40c7a2576e774b48c473 : SUCCESS in 1h 13m 24s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/83724e63afa64ec19fac57d833a11a95 : SUCCESS in 1h 03m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9d5784b2ad84281ba1563e861d8fb29 : SUCCESS in 7m 33s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/1d094d397d8343fc8f0d44b5502e27a7 : SUCCESS in 2h 35m 01s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/d78751da54f14691beea6b0149d24b26 : SUCCESS in 2h 36m 43s\n- neutron-tempest-plugin-sfc-ussuri https://zuul.opendev.org/t/openstack/build/aa730610e45843368b0f13e49ded66bd : SUCCESS in 2h 33m 37s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d68f07d662be4e65b05735766d67555f : TIMED_OUT in 2h 03m 41s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/520de5e2cbfa4280a5919e44ad3e38c2 : TIMED_OUT in 2h 03m 19s\n- neutron-tempest-plugin-bgpvpn-bagpipe-ussuri https://zuul.opendev.org/t/openstack/build/0562724a65a74a00bb3cb70c6e0a1d4d : TIMED_OUT in 2h 03m 25s\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/0383de10fbea45a5a24780bf7420f40e : FAILURE in 33m 46s (non-voting)\n- neutron-tempest-plugin-fwaas-ussuri https://zuul.opendev.org/t/openstack/build/2c81590d63ac462cab444fd99240e91b : SUCCESS in 1h 25m 56s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/8651af7b96324173a05f3942cf8cd693 : SUCCESS in 1h 43m 20s\n- neutron-tempest-plugin-dynamic-routing-ussuri https://zuul.opendev.org/t/openstack/build/d3bf82d1eeb646f39dec1ac4a8dc390a : SUCCESS in 1h 43m 41s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/962562120c804c6591fad6f929736251 : TIMED_OUT in 1h 09m 57s\n- neutron-tempest-plugin-vpnaas-ussuri https://zuul.opendev.org/t/openstack/build/96d469e17a1846f0809cbab029c2b1d2 : TIMED_OUT in 1h 10m 43s","accounts_in_message":[],"_revision_number":2},{"id":"597d8291960d8894eeea69d683b6372ca554afa7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-18 23:07:34.000000000","message":"Abandoned\n\ndevstack patch is working and merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ed097fc5526392dc4435eabce55838ccf625fa64","revisions":{"7b544216cf6c1ca22739e83c5150fdf41361bad9":{"kind":"REWORK","_number":1,"created":"2020-07-17 18:08:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/95/741695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/741695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/741695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/741695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/741695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/741695/1"}}},"commit":{"parents":[{"commit":"803e839077bd762cc0e5f6c08b55a62dad664485","subject":"Merge \"Fix neutron-dynamic-routing tests when DVR is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/803e839077bd762cc0e5f6c08b55a62dad664485"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-17 18:06:23.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-17 18:08:10.000000000","tz":0},"subject":"[WIP] Test GLANCE_STANDALONE\u003dFalse to fix glance in CI jobs","message":"[WIP] Test GLANCE_STANDALONE\u003dFalse to fix glance in CI jobs\n\nChange-Id: I761f5f199b7a2dd045cdc503223333900bfeadf0\nRelated-Bug: #1887992\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7b544216cf6c1ca22739e83c5150fdf41361bad9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7b544216cf6c1ca22739e83c5150fdf41361bad9"}]},"branch":"refs/heads/master"},"ed097fc5526392dc4435eabce55838ccf625fa64":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-07-18 11:46:28.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/95/741695/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/741695/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/741695/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/741695/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/741695/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/741695/2"}}},"commit":{"parents":[{"commit":"803e839077bd762cc0e5f6c08b55a62dad664485","subject":"Merge \"Fix neutron-dynamic-routing tests when DVR is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/803e839077bd762cc0e5f6c08b55a62dad664485"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-17 18:06:23.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-07-18 11:46:25.000000000","tz":0},"subject":"[WIP] Test GLANCE_STANDALONE\u003dFalse to fix glance in CI jobs","message":"[WIP] Test GLANCE_STANDALONE\u003dFalse to fix glance in CI jobs\n\nChange-Id: I761f5f199b7a2dd045cdc503223333900bfeadf0\nRelated-Bug: #1887992\nDepends-On: https://review.opendev.org/741687\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ed097fc5526392dc4435eabce55838ccf625fa64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ed097fc5526392dc4435eabce55838ccf625fa64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
