)]}'
{"id":"openstack%2Fneutron-tempest-plugin~755800","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I0ed17ad7632c1aacce6188d135f2295714e9a771","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"test_devstack_fix_setuptools","hashtags":[],"change_id":"I0ed17ad7632c1aacce6188d135f2295714e9a771","subject":"Test devstack change","status":"ABANDONED","created":"2020-10-02 14:16:07.000000000","updated":"2020-10-05 07:01:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c06570f82d2b0bde10fe1905a7e7c1a8c5fb34e1","_number":755800,"virtual_id_number":755800,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-02 20:03:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-02 20:03:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aabb3270d344ecc45e35463d86386a21786844e5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-10-02 14:16:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d208234c5b8ebf530c0adbf21b426202397fc909","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 20:03:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2eb91abaf17645dbbd5494e6a353fd82 : SUCCESS in 3m 29s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bc4c3bc93cc64c939ed3bda8513e35e3 : SUCCESS in 51m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/955aa034c15f4c37a7694fbb3415eb8b : SUCCESS in 31m 54s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cebf69a0aafd4b7da05ff0b047450e3a : SUCCESS in 1h 21m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9311512e80cb46b98dda685062008a84 : SUCCESS in 1h 24m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d99bc8cd2be544058080768aefb227c9 : SUCCESS in 1h 13m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/039b03a621754399b2a55ca579780e47 : SUCCESS in 1h 10m 14s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/e57e4f90e9b6410bac48947974528747 : SUCCESS in 53m 44s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/b25dc40493fd47e4937db464b0c7733b : SUCCESS in 31m 56s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/6f757f4909994c1e99ab77287ef4c8e9 : SUCCESS in 1h 10m 52s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/6abd9f19e8f547d8b192621397678316 : FAILURE in 1h 30m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/e49df65b392c45a290d51dd60aeb65b7 : SUCCESS in 1h 19m 01s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/88881138e57541bc85d8fdc1b277aef7 : SUCCESS in 53m 12s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/7228aa8aa6d741da817838457104cb67 : SUCCESS in 40m 30s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/d0a89cab359e426db0beb1aeea3ac22b : SUCCESS in 1h 25m 10s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/0f131c10c7784966964b35badf056257 : SUCCESS in 1h 14m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/dba95077e6b54a3ca96289cb8d67e456 : SUCCESS in 1h 14m 31s\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/cd0511d921924dc38e32635a88c7aa49 : SUCCESS in 47m 32s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/bb84b53f690a4c90be2ac1f173f7089b : SUCCESS in 52m 29s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/d74d2491b93142f88487ba9a75aa0c7e : SUCCESS in 1h 25m 58s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/a4a52959f73e4c2c85ed33076b21ba0a : SUCCESS in 1h 09m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/53ec59e32a5747e7b1e2bcc4cea1141b : SUCCESS in 1h 13m 53s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/ac3fcf2d22ff49e5a687026db6280e81 : SUCCESS in 1h 13m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fc1152c7ecf41e0958d43ec77a5f666 : SUCCESS in 4m 44s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/11ac417c18754647a0388cfc8b50a6c1 : SUCCESS in 1h 36m 50s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/3781d87b3cd0469eae71ebcf98a52d22 : SUCCESS in 1h 56m 46s\n- neutron-tempest-plugin-sfc-ussuri https://zuul.opendev.org/t/openstack/build/380e9f7c2c6c49ee85b8f9475b6fade9 : SUCCESS in 1h 33m 34s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/620d3819543a489c919b8f5d0fe40dd1 : SUCCESS in 1h 24m 29s (non-voting)\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/1b4417b2a9b2435c9157f0d01fe60cfd : SUCCESS in 1h 21m 30s\n- neutron-tempest-plugin-bgpvpn-bagpipe-ussuri https://zuul.opendev.org/t/openstack/build/f033b6f2011147f9b561a569fedbf78a : FAILURE in 1h 27m 52s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/16b103292e7f4724bbe3e91001c6fd31 : FAILURE in 13m 14s (non-voting)\n- neutron-tempest-plugin-fwaas-ussuri https://zuul.opendev.org/t/openstack/build/6702b61f646b4cecb063902b3c5346fc : FAILURE in 33m 37s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/10cb63d524a24706a0d6c3759960dee9 : SUCCESS in 46m 55s\n- neutron-tempest-plugin-dynamic-routing-ussuri https://zuul.opendev.org/t/openstack/build/a045b5378ca64f48aeaf229d9d689c90 : FAILURE in 53m 10s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/b034fdb3a3cb406499ecdac6e4f80995 : SUCCESS in 42m 26s\n- neutron-tempest-plugin-vpnaas-ussuri https://zuul.opendev.org/t/openstack/build/7e8a041d1cbe4112a39571e42c19f704 : SUCCESS in 41m 24s","accounts_in_message":[],"_revision_number":1},{"id":"fc33ac25c8058e73f3c90b674bf0d400bb170183","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-10-05 07:01:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"91a16bc8f7836e4382cc39c16650adfd5d1327ca","revisions":{"91a16bc8f7836e4382cc39c16650adfd5d1327ca":{"kind":"REWORK","_number":1,"created":"2020-10-02 14:16:07.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/00/755800/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/00/755800/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/00/755800/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/00/755800/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/00/755800/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/00/755800/1"}}},"commit":{"parents":[{"commit":"ff09b50d54acf36550238b2e73f8f3c360be1ee6","subject":"[SFC] Mark test test_update_port_pair_group_remove_port_pairs as unstable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ff09b50d54acf36550238b2e73f8f3c360be1ee6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-10-02 14:14:19.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-10-02 14:15:58.000000000","tz":0},"subject":"Test devstack change","message":"Test devstack change\n\nNOTE: I don\u0027t know if other versions than Train will work with this\ndepends on, but doesn\u0027t matter\n\nDepends-On: https://review.opendev.org/#/c/755794/\nChange-Id: I0ed17ad7632c1aacce6188d135f2295714e9a771\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/91a16bc8f7836e4382cc39c16650adfd5d1327ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/91a16bc8f7836e4382cc39c16650adfd5d1327ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
