)]}'
{"id":"openstack%2Fneutron-tempest-plugin~756369","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I0d46ff77c392b0468578954257e6a2232a9b408e","project":"openstack/neutron-tempest-plugin","branch":"master","hashtags":[],"change_id":"I0d46ff77c392b0468578954257e6a2232a9b408e","subject":"QOS DSCP marking traffic test","status":"ABANDONED","created":"2020-10-06 17:19:06.000000000","updated":"2021-04-14 13:54:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e4ae21515d8e9268c77b1cbd53521fbed83281ee","_number":756369,"virtual_id_number":756369,"owner":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-07 00:36:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-07 00:36:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-14 13:54:13.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"0609964e9250db719ef87b461e4d970f2d75dd4d","author":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"date":"2020-10-06 17:19:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c46875ac058d5e046dee3a2befca5b669f342887","author":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"date":"2020-10-06 17:20:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5e829058cf668225ca5cc363d413393191690966","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 00:36:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c336f39425b24e78a40df7e37132a3ae : SUCCESS in 3m 38s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5b5cea25c9d2433bac54a06837765a60 : SUCCESS in 52m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5be642d682464cfa8af3aa6ae568ac40 : SUCCESS in 39m 45s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3aff4e9ca9e34ea1aca406aa42c7c61d : FAILURE in 1h 24m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/de6ea6b67c884250a3745d58a0c78aba : FAILURE in 1h 39m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6da087d98f124aeaa8eedfd29d3b9cc9 : FAILURE in 1h 10m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/79d79a329ea94902a473c2d640de84a9 : FAILURE in 1h 13m 17s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/fc0bf8765b1248528a6af71574bad38f : SUCCESS in 48m 03s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/d0b13372f806410c82eda8254cc381f6 : SUCCESS in 29m 08s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/23caf2d103f648c38012352291d3db3c : FAILURE in 1h 43m 53s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/116147b22c2242e3a18d202010465b75 : FAILURE in 1h 07m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/c110639aa0b0401bbb28cb85fa99642e : FAILURE in 1h 07m 25s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/684b997070fc4b46a72791234df84f6c : SUCCESS in 48m 53s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/30ed452b0b114a8b983a24785ac85b47 : SUCCESS in 30m 58s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/d88fee1a2e0c4886ab90d65884473463 : FAILURE in 1h 12m 39s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/73c61907e0924ed39fc8fdab581c3f08 : FAILURE in 1h 12m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/3e1ecc2b48af48598dff123bd178ed6a : FAILURE in 1h 25m 23s\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/1cbe5e5be821476b860ba407b9d305d9 : SUCCESS in 58m 11s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/29992c5568564ce1a4d964d00ad9d1f2 : SUCCESS in 28m 47s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/b43351883d45444d8671489988dcf6fa : FAILURE in 1h 28m 14s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/31a642c524f54b349f440fe69f1d8b17 : FAILURE in 1h 22m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/1c39db918cc14de1be9618d1f14c5ee1 : FAILURE in 1h 03m 44s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/6027b670afe6418a9f225950c8fda26c : FAILURE in 1h 16m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b135f99e58e428ca91b97edf722426a : SUCCESS in 3m 31s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/6eb680cc586c4deab4bac43b88c5c771 : SUCCESS in 1h 07m 03s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/3e0e187ff62540b7b055c282149e0a3b : SUCCESS in 1h 45m 52s\n- neutron-tempest-plugin-sfc-ussuri https://zuul.opendev.org/t/openstack/build/9124168260b84306bf3b0fce44c5bcc3 : SUCCESS in 1h 34m 29s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/6c0c5a5b10d741c6885ce35e67dc5297 : SUCCESS in 1h 42m 28s (non-voting)\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/d58ceac4fd8c40009f23e7caadc44c37 : SUCCESS in 1h 28m 50s\n- neutron-tempest-plugin-bgpvpn-bagpipe-ussuri https://zuul.opendev.org/t/openstack/build/1256d5d965e34877b148f9cbc003dbe8 : FAILURE in 1h 24m 35s (non-voting)\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/3fcd806d3d1d4b0397f31888da556cad : FAILURE in 14m 53s (non-voting)\n- neutron-tempest-plugin-fwaas-ussuri https://zuul.opendev.org/t/openstack/build/8f7b45f01cb942d7a2c475a5c89b959a : SUCCESS in 39m 11s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/99b96806680246cc9840e5668b706e9b : SUCCESS in 54m 02s\n- neutron-tempest-plugin-dynamic-routing-ussuri https://zuul.opendev.org/t/openstack/build/015a5fa2208244c38782d3bff7db4f7f : SUCCESS in 45m 31s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/72bba3ddff464212ad44a022c1a6a149 : SUCCESS in 44m 02s\n- neutron-tempest-plugin-vpnaas-ussuri https://zuul.opendev.org/t/openstack/build/77cf96fa52fe43a6bdeaafe421874150 : SUCCESS in 40m 58s","accounts_in_message":[],"_revision_number":2},{"id":"e4ae21515d8e9268c77b1cbd53521fbed83281ee","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-14 13:54:14.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"162856800d602e824f618d8fa3cf240c37c95d08","revisions":{"4584bb00787e791746b8cccaa5ba9958dd25f3df":{"kind":"REWORK","_number":1,"created":"2020-10-06 17:19:06.000000000","uploader":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"ref":"refs/changes/69/756369/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/69/756369/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/756369/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/756369/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/756369/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/756369/1"}}},"commit":{"parents":[{"commit":"c3bec86663dc9d30529bc0641fb4f1994d5ecb6e","subject":"Update hacking for Focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c3bec86663dc9d30529bc0641fb4f1994d5ecb6e"}]}],"author":{"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","date":"2020-10-02 15:44:11.000000000","tz":120},"committer":{"name":"Tom Stappaerts","email":"tom.stappaerts@nuagenetworks.net","date":"2020-10-06 17:19:04.000000000","tz":120},"subject":"QOS DSCP marking traffic test","message":"QOS DSCP marking traffic test\n\nAdd a DSCP marking traffic test by applying\nDSCP marking on host A and pinging host B from it.\nUsing tcpdump the DSCP mark of arriving pings on\nhost B is verified.\n\nChange-Id: I0d46ff77c392b0468578954257e6a2232a9b408e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4584bb00787e791746b8cccaa5ba9958dd25f3df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4584bb00787e791746b8cccaa5ba9958dd25f3df"}]},"branch":"refs/heads/master"},"162856800d602e824f618d8fa3cf240c37c95d08":{"kind":"REWORK","_number":2,"created":"2020-10-06 17:20:19.000000000","uploader":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"ref":"refs/changes/69/756369/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/69/756369/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/756369/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/756369/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/756369/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/756369/2"}}},"commit":{"parents":[{"commit":"c3bec86663dc9d30529bc0641fb4f1994d5ecb6e","subject":"Update hacking for Focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c3bec86663dc9d30529bc0641fb4f1994d5ecb6e"}]}],"author":{"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","date":"2020-10-02 15:44:11.000000000","tz":120},"committer":{"name":"Tom Stappaerts","email":"tom.stappaerts@nuagenetworks.net","date":"2020-10-06 17:19:43.000000000","tz":120},"subject":"QOS DSCP marking traffic test","message":"QOS DSCP marking traffic test\n\nAdd a DSCP marking traffic test by applying\nDSCP marking on host A and pinging host B from it.\nUsing tcpdump the DSCP mark of arriving pings on\nhost B is verified.\n\nChange-Id: I0d46ff77c392b0468578954257e6a2232a9b408e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/162856800d602e824f618d8fa3cf240c37c95d08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/162856800d602e824f618d8fa3cf240c37c95d08"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
