)]}'
{"id":"openstack%2Fneutron-tempest-plugin~767650","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ib15d3dd79d518a0194762a3e9f4e46ff4ba4db43","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"neutron_14","hashtags":[],"change_id":"Ib15d3dd79d518a0194762a3e9f4e46ff4ba4db43","subject":"test insert firewall rule before/after rule in policy","status":"ABANDONED","created":"2020-12-18 01:57:34.000000000","updated":"2021-04-14 13:48:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a26cdb5eef6abde4b4b29e233251ad8ba877f773","_number":767650,"virtual_id_number":767650,"owner":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-18 03:45:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":-2,"date":"2020-12-18 08:58:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-18 03:45:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-18 08:57:44.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"5b72ff19aebc34d2bd4b949cbb3df6de2b0f1aba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2020-12-18 01:57:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed81691a1a82cdfc961eef040ab5e57ad482bc33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 03:45:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4beec937452475d80891abf46cae3dd : SUCCESS in 6m 05s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/33a6b1b161bb48aeaf9c87c25ce09eaf : SUCCESS in 1h 45m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4cf09a58b9cc4dfeb38b6820457fd890 : TIMED_OUT in 1h 02m 46s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/12fee286d1e241ffbcf4e1cfa5397a0d : SUCCESS in 1h 37m 24s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/63bfb0d0041a40e4834912230e32f559 : SUCCESS in 1h 43m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cdc2b445cd7c4698b9a6ab629c742bd3 : SUCCESS in 1h 29m 36s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/194c5cf6c66e4adbb4b3fba4a54e1508 : SUCCESS in 1h 25m 00s\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/b515ddc78ce94143852078dfbf2ff540 : SUCCESS in 40m 13s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/2d2f2a745252417096847931359192b1 : SUCCESS in 26m 38s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/d45133deaffa4d038d534de565e3c7b3 : SUCCESS in 1h 21m 46s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/bcab16cfae7a4ff8a9f0e6c121e90f5c : SUCCESS in 1h 26m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/d7e08dc633864d64bc12939452f5060e : SUCCESS in 1h 04m 01s\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/0b11c4c6acbe40919fbcb5afb6c6e177 : SUCCESS in 49m 47s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/0f406ad057ad466482d066cad47e4e13 : SUCCESS in 29m 47s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/590f40b5b47e4179b8a9597608ea877b : SUCCESS in 1h 45m 52s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/b0d605745d0a479cb93acb06222b769d : SUCCESS in 1h 08m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/7bb49834436e4f3a830257e1bea4a900 : SUCCESS in 1h 27m 28s\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/5f823fc3b58f4ef78878491a0286786a : SUCCESS in 54m 07s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/88375b8f5931486a9f7096e786a9cea7 : SUCCESS in 27m 55s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/4a7697a8be834e4ea74fb3b060e24ab3 : SUCCESS in 1h 31m 13s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/3eb7604307ae473a9d4d20a6505dbc08 : SUCCESS in 1h 30m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/7306e5d8171849eaa66edada39714139 : SUCCESS in 1h 28m 24s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/7c12d2b17f534c34a29e0876a74143e2 : SUCCESS in 1h 24m 43s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/a457e6ba344b4267a17b7af7e5fd5ba7 : SUCCESS in 50m 45s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/e0c263cbdc5f4f9eab5a234ddd8ea568 : SUCCESS in 35m 09s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-victoria https://zuul.opendev.org/t/openstack/build/dc007ec0757b4a308451a2afbd0787fa : SUCCESS in 1h 27m 59s\n- neutron-tempest-plugin-scenario-openvswitch-victoria https://zuul.opendev.org/t/openstack/build/f4ac3906b64c4c2d9a1e5f7cf7b4a20a : SUCCESS in 1h 19m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-victoria https://zuul.opendev.org/t/openstack/build/0dc7c027397c410e97bf1a25df962061 : SUCCESS in 1h 24m 06s\n- neutron-tempest-plugin-scenario-ovn-victoria https://zuul.opendev.org/t/openstack/build/d359d860cd4645cf9426d542d8122843 : SUCCESS in 1h 16m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/968d378ee5994a88bced2c6a7b1a9f82 : SUCCESS in 3m 07s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/a2bd46db764f4bc8bb2a0ae868e7555a : SUCCESS in 1h 28m 39s\n- neutron-tempest-plugin-sfc-train https://zuul.opendev.org/t/openstack/build/0780beff09964ebd942a756c5ff14dce : SUCCESS in 1h 27m 08s\n- neutron-tempest-plugin-sfc-ussuri https://zuul.opendev.org/t/openstack/build/c1353b0b52154ab3b86bb56e57dcd365 : SUCCESS in 1h 29m 32s\n- neutron-tempest-plugin-sfc-victoria https://zuul.opendev.org/t/openstack/build/3f541e555e34427d87e1101adb0293a5 : SUCCESS in 1h 32m 22s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/41ea729eb5864b939d37d9e94eddc0c2 : SUCCESS in 1h 13m 44s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/6001dd0cf7224c3ea25f80bd5e3aa492 : SUCCESS in 1h 06m 29s\n- neutron-tempest-plugin-bgpvpn-bagpipe-ussuri https://zuul.opendev.org/t/openstack/build/956478cdeea0411e97643f4e4710645a : SUCCESS in 1h 20m 58s\n- neutron-tempest-plugin-bgpvpn-bagpipe-victoria https://zuul.opendev.org/t/openstack/build/04bc400f9af14ff3ac9d7416ec00d0fe : SUCCESS in 1h 08m 31s\n- neutron-tempest-plugin-fwaas-train https://zuul.opendev.org/t/openstack/build/21480bd7c4174f19a6f51114cd61e143 : FAILURE in 10m 02s (non-voting)\n- neutron-tempest-plugin-fwaas-ussuri https://zuul.opendev.org/t/openstack/build/8ff7686775ec4fd7be9cb6170e30603f : FAILURE in 29m 00s (non-voting)\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/61824621f0434187b4c449ae42e774be : SUCCESS in 35m 04s\n- neutron-tempest-plugin-dynamic-routing-ussuri https://zuul.opendev.org/t/openstack/build/c1a4b966ea834e9b8087396af9c6ec2f : SUCCESS in 43m 16s\n- neutron-tempest-plugin-dynamic-routing-victoria https://zuul.opendev.org/t/openstack/build/3fdd000d2d824612a108aff144fb1472 : SUCCESS in 40m 41s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/5bdde1e12b134286a1938479a0af0c6b : SUCCESS in 34m 18s\n- neutron-tempest-plugin-vpnaas-ussuri https://zuul.opendev.org/t/openstack/build/ad78b690caaa403194d9c1c956c04d60 : SUCCESS in 34m 32s\n- neutron-tempest-plugin-vpnaas-victoria https://zuul.opendev.org/t/openstack/build/f3cbdc2920974fbb820b03621ad7026f : SUCCESS in 35m 56s","accounts_in_message":[],"_revision_number":1},{"id":"2f746df72c9fcafe10c9c1ccb00471de28b6d19b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-18 08:57:44.000000000","message":"Patch Set 1: Code-Review-1\n\nFWaaS isn\u0027t supported since couple of cycles already. We have those tests only for stable branches like Train. Jobs for fwaas aren\u0027t really stable. So based on those circumstances I think that it isn\u0027t good idea to add new fwaas tests now.","accounts_in_message":[],"_revision_number":1},{"id":"5a0f57e5b3ff58b48549fc81d90f37b5ad821fc3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-18 08:58:15.000000000","message":"Patch Set 1: Code-Review-2\n\nSorry, based on my last comment it should be -2, not -1","accounts_in_message":[],"_revision_number":1},{"id":"a26cdb5eef6abde4b4b29e233251ad8ba877f773","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-14 13:48:01.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment and currently blocked by a core reviewer with a -2. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and contacting the reviewer with the -2 on this review to ensure you address their concerns.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b00630c54aca8cf3ec6829d431d4345d39fe913","revisions":{"9b00630c54aca8cf3ec6829d431d4345d39fe913":{"kind":"REWORK","_number":1,"created":"2020-12-18 01:57:34.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/50/767650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/50/767650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/50/767650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/50/767650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/50/767650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/50/767650/1"}}},"commit":{"parents":[{"commit":"f165a64263ab4d47f19182b2b819b3d01fbc807f","subject":"Merge \"cases associated with the port\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f165a64263ab4d47f19182b2b819b3d01fbc807f"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2020-12-18 01:57:05.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2020-12-18 01:57:05.000000000","tz":480},"subject":"test insert firewall rule before/after rule in policy","message":"test insert firewall rule before/after rule in policy\n\nChange-Id: Ib15d3dd79d518a0194762a3e9f4e46ff4ba4db43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9b00630c54aca8cf3ec6829d431d4345d39fe913"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9b00630c54aca8cf3ec6829d431d4345d39fe913"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
