)]}'
{"id":"openstack%2Fneutron-tempest-plugin~797594","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I2b9b472a07455f5e67fc4adcc3b264b4ad8a9dbf","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/1933115","hashtags":[],"change_id":"I2b9b472a07455f5e67fc4adcc3b264b4ad8a9dbf","subject":"Fix required extensions for the subnet\u0027s service type API tests","status":"MERGED","created":"2021-06-23 07:38:59.000000000","updated":"2021-06-25 09:06:35.000000000","submitted":"2021-06-25 09:06:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"797594-bug/1933115","meta_rev_id":"273d151f38684aed51da66230b6ba5d802562ef6","_number":797594,"virtual_id_number":797594,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-25 09:06:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-06-23 13:55:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-06-25 06:59:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-06-25 06:59:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-23 07:52:36.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-06-23 11:46:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-25 06:59:16.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"3d4704ffed97e40f83df879ec9b06605c6743bf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-23 07:38:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"669da7cf4bc58de15772d9bc05be72e3d571864e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-23 07:52:36.000000000","message":"Patch Set 1: Code-Review+2\n\ngood catch (now I realize I didn\u0027t check the logs in the previous patch)","accounts_in_message":[],"_revision_number":1},{"id":"808ea9958e7575a0eb5be3ed608c94a478b27fd3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-23 08:01:28.000000000","message":"Patch Set 1: Code-Review-1\n\nWhere is this extension loaded? This extension is not loaded in any job.","accounts_in_message":[],"_revision_number":1},{"id":"4adafb5955bf31305a3d7643afcb5d878dabd25b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-23 09:14:27.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e Where is this extension loaded? This extension is not loaded in any job.\n\nIt is loaded e.g. in neutron-tempest-plugin-api job but it\u0027s true that we should include it in the extensions list in tempest\u0027s config.","accounts_in_message":[],"_revision_number":1},{"id":"5c4d09179738fe065d53128807e156935559d28d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-23 09:20:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"363082abe4778ba64fbc20f3ee586d343f2dcdd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 11:46:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/882bf879e20047bf94ba813910605f69 : SUCCESS in 5m 01s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e3b80b09671e442baea3cd09a1d45c07 : SUCCESS in 36m 30s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4a5b1835720d48a3962dfc73e6230d28 : SUCCESS in 1h 43m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d6d4bd7b013847c5a22bb254e26337d2 : SUCCESS in 1h 41m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cb237bdbbcdb45b08b4da010dc56beb1 : SUCCESS in 1h 13m 59s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/456ad91383794592abd1bab561142fbe : SUCCESS in 1h 39m 52s\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/07319887bdc04356b405faa578ba3f8f : SUCCESS in 44m 27s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/bc4b9dc50096466b9904b347966c4b52 : SUCCESS in 1h 54m 43s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/ae89958ed90342aeab6716c085d7dd1c : SUCCESS in 1h 28m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/cbacf25f235b40f89dfb0bf7a25c8e79 : SUCCESS in 1h 16m 48s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/3d2b69b9f1954bb3879ef08535584a88 : SUCCESS in 1h 47m 21s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/008635639bd04efa82100eb25e1c87cf : SUCCESS in 42m 34s\n- neutron-tempest-plugin-scenario-linuxbridge-victoria https://zuul.opendev.org/t/openstack/build/22310c50526b416eb56784eca70c6788 : SUCCESS in 1h 41m 15s\n- neutron-tempest-plugin-scenario-openvswitch-victoria https://zuul.opendev.org/t/openstack/build/5a8a89f754754cf5987fac7e2d580513 : SUCCESS in 1h 49m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-victoria https://zuul.opendev.org/t/openstack/build/07f349de759c470e806292f850cba8e3 : SUCCESS in 1h 13m 44s\n- neutron-tempest-plugin-scenario-ovn-victoria https://zuul.opendev.org/t/openstack/build/eaadd696e1ee4bd5a2ed6d4cee52a10e : SUCCESS in 1h 45m 50s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/b44c73776f4e42a8bd988bcd16d07de0 : SUCCESS in 41m 31s\n- neutron-tempest-plugin-scenario-linuxbridge-wallaby https://zuul.opendev.org/t/openstack/build/9aacab84d18d44128e6a32db4c2e3788 : SUCCESS in 2h 00m 04s\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/be67cd6043fc46b2921040b7fa71eb3c : SUCCESS in 1h 49m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-wallaby https://zuul.opendev.org/t/openstack/build/e7800219e80a429087e584d65a0e0400 : SUCCESS in 1h 20m 47s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/9630400b52764ba2835750440c1aeac6 : SUCCESS in 2h 23m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d704c47e6b28446dac84f6460bf8b274 : SUCCESS in 4m 04s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/cbe68e43c79844d29ccd56789baf9737 : SUCCESS in 1h 20m 11s\n- neutron-tempest-plugin-sfc-ussuri https://zuul.opendev.org/t/openstack/build/96a5105c61984b25b3b89a88a8fa8a06 : SUCCESS in 1h 13m 15s\n- neutron-tempest-plugin-sfc-victoria https://zuul.opendev.org/t/openstack/build/89cc793da995409a967a84c75f34857b : SUCCESS in 1h 14m 54s\n- neutron-tempest-plugin-sfc-wallaby https://zuul.opendev.org/t/openstack/build/8745308fda514aeb9ecf69cd266ae9b2 : SUCCESS in 1h 11m 56s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/fa8567a8cb04455bb9d31470d37642a6 : SUCCESS in 1h 07m 07s\n- neutron-tempest-plugin-bgpvpn-bagpipe-ussuri https://zuul.opendev.org/t/openstack/build/bf663728e442439c875aeeb007522c50 : SUCCESS in 1h 12m 04s\n- neutron-tempest-plugin-bgpvpn-bagpipe-victoria https://zuul.opendev.org/t/openstack/build/9bcf5d9a4164411a87244ea1daefd0fc : SUCCESS in 1h 13m 07s\n- neutron-tempest-plugin-bgpvpn-bagpipe-wallaby https://zuul.opendev.org/t/openstack/build/ad9fd87ab64f433d92590cba6943dbc9 : SUCCESS in 1h 08m 19s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/647caed6ddd44132a8b43a0f493344cc : SUCCESS in 39m 17s\n- neutron-tempest-plugin-dynamic-routing-ussuri https://zuul.opendev.org/t/openstack/build/d6988da0762e4b45b95b9b276ee6eaee : SUCCESS in 38m 58s\n- neutron-tempest-plugin-dynamic-routing-victoria https://zuul.opendev.org/t/openstack/build/38ef4fe5c6f740a2b7960d6dccb89d68 : SUCCESS in 48m 10s\n- neutron-tempest-plugin-dynamic-routing-wallaby https://zuul.opendev.org/t/openstack/build/213cf1299481412eb7b4eee7fca22a35 : SUCCESS in 47m 51s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/98be0ff3ec9b4c51a29994638f3122af : SUCCESS in 23m 43s\n- neutron-tempest-plugin-vpnaas-ussuri https://zuul.opendev.org/t/openstack/build/1866b974348c4b908df13d47d3889bf5 : SUCCESS in 34m 04s\n- neutron-tempest-plugin-vpnaas-victoria https://zuul.opendev.org/t/openstack/build/e946c799caa5415d93cf655baae02596 : SUCCESS in 31m 06s\n- neutron-tempest-plugin-vpnaas-wallaby https://zuul.opendev.org/t/openstack/build/d78585ac265c44e9b013f1024637f5c1 : SUCCESS in 34m 08s","accounts_in_message":[],"_revision_number":2},{"id":"d2c44ee00158c3ca008461694742c5b6aea5dcef","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-23 13:55:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3e30a606161ee07eeef1a721b7105c20088f54a3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-25 06:59:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nGood catch","accounts_in_message":[],"_revision_number":2},{"id":"bc9be4be2763b596a00a6e120a00f1a286ddb448","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 06:59:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3cf86c5684bd9677f186e55b9ebcf24b0359bb04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 09:06:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eff7e7bf94ed4c50bc87586d4fbdacd5 : SUCCESS in 6m 58s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9050ba12f89940e08920279397ed4455 : SUCCESS in 39m 23s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2ecb043f2e144844bb5b269c2e8c002a : SUCCESS in 2h 05m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/dd401166da0c49a3aadc0d723ba6afb1 : SUCCESS in 1h 42m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2ecbec0384f84031a33c4a2fb00aefb5 : SUCCESS in 1h 48m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a8bf1c1706c74c49b72cd4dfe2f4b9b5 : SUCCESS in 1h 17m 30s\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/cde0c28852504baca16a1d73b120b053 : SUCCESS in 49m 29s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/feb4e69609ea40c6b787b606e551c945 : SUCCESS in 50m 24s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/9313372fc8b84a3fb1cb4bf339442e35 : SUCCESS in 48m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21f4264e609442c8882659ff3c5941d1 : SUCCESS in 5m 22s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/08e7188abe744fd3b32217246708c87f : SUCCESS in 1h 18m 04s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/eaed8231f09a49758e82d30fb2fd1df8 : SUCCESS in 1h 09m 00s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/f5ea5e7f5f7e4d5fb1f5761ef80603f6 : SUCCESS in 39m 05s","accounts_in_message":[],"_revision_number":2},{"id":"273d151f38684aed51da66230b6ba5d802562ef6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 09:06:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a5cdede38739b52340d204ebe57c63638a19dcca","revisions":{"a43c5f92986f237ec96ddcd0c6ff32eb621df5d1":{"kind":"REWORK","_number":1,"created":"2021-06-23 07:38:59.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/94/797594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/94/797594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/94/797594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/94/797594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/94/797594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/94/797594/1"}}},"commit":{"parents":[{"commit":"47bf2c5054ac69117a6c1bc1e576f3d9e6614cb0","subject":"Merge \"Remove allowed_address_pair API tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/47bf2c5054ac69117a6c1bc1e576f3d9e6614cb0"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-23 07:37:04.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-23 07:37:04.000000000","tz":120},"subject":"Fix required extensions for the subnet\u0027s service type API tests","message":"Fix required extensions for the subnet\u0027s service type API tests\n\nIn [1] new tests for subnet service types were added but by mistake\n\u0027service-types\u0027 was set as API extension required for those tests.\nIn fact this API feature in the subnet is provided by the\n\u0027subnet-service-types\u0027 extension and this should be required for that\ntests.\n\n[1] https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/787718\n\nRelated-Bug: #1933115\nChange-Id: I2b9b472a07455f5e67fc4adcc3b264b4ad8a9dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a43c5f92986f237ec96ddcd0c6ff32eb621df5d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a43c5f92986f237ec96ddcd0c6ff32eb621df5d1"}]},"branch":"refs/heads/master"},"a5cdede38739b52340d204ebe57c63638a19dcca":{"kind":"REWORK","_number":2,"created":"2021-06-23 09:20:04.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/94/797594/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/94/797594/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/94/797594/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/94/797594/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/94/797594/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/94/797594/2"}}},"commit":{"parents":[{"commit":"47bf2c5054ac69117a6c1bc1e576f3d9e6614cb0","subject":"Merge \"Remove allowed_address_pair API tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/47bf2c5054ac69117a6c1bc1e576f3d9e6614cb0"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-23 07:37:04.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-23 09:18:54.000000000","tz":120},"subject":"Fix required extensions for the subnet\u0027s service type API tests","message":"Fix required extensions for the subnet\u0027s service type API tests\n\nIn [1] new tests for subnet service types were added but by mistake\n\u0027service-types\u0027 was set as API extension required for those tests.\nIn fact this API feature in the subnet is provided by the\n\u0027subnet-service-types\u0027 extension and this should be required for that\ntests.\n\nThis patch also adds \"subnet-service-types\" extension to the Tempest\nconf in jobs for all branches up to stable/queens becuase this API\nextension was available in Neutron since queens for sure so it\u0027s safe to\nadd it there.\n\n[1] https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/787718\n\nRelated-Bug: #1933115\nChange-Id: I2b9b472a07455f5e67fc4adcc3b264b4ad8a9dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a5cdede38739b52340d204ebe57c63638a19dcca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a5cdede38739b52340d204ebe57c63638a19dcca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
