)]}'
{"id":"openstack%2Fneutron-tempest-plugin~803315","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I43c0b7781ca2ed98156e125566115196aab530cf","project":"openstack/neutron-tempest-plugin","branch":"master","hashtags":[],"change_id":"I43c0b7781ca2ed98156e125566115196aab530cf","subject":"Fix StatefulConnection to not to fail on exit","status":"MERGED","created":"2021-08-03 07:25:52.000000000","updated":"2021-08-09 11:49:11.000000000","submitted":"2021-08-09 11:49:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"803315","meta_rev_id":"5530ba1e2ed44de23daccafa98bb4716174bae33","_number":803315,"virtual_id_number":803315,"owner":{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29088,"name":"Candido Campos Rivas","email":"ccamposr@redhat.com","username":"ccamposr"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-09 11:49:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"recommended":{"_account_id":29088,"name":"Candido Campos Rivas","email":"ccamposr@redhat.com","username":"ccamposr"},"all":[{"value":1,"date":"2021-08-05 16:01:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29088,"name":"Candido Campos Rivas","email":"ccamposr@redhat.com","username":"ccamposr"},{"value":2,"date":"2021-08-09 09:19:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-08-05 13:33:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-08-06 11:49:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-08 06:48:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":0,"_account_id":29088,"name":"Candido Campos Rivas","email":"ccamposr@redhat.com","username":"ccamposr"},{"value":1,"date":"2021-08-09 09:19:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29088,"name":"Candido Campos Rivas","email":"ccamposr@redhat.com","username":"ccamposr"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29088,"name":"Candido Campos Rivas","email":"ccamposr@redhat.com","username":"ccamposr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-03 09:32:54.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-08-03 10:10:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-03 10:14:51.000000000","updated_by":{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},"reviewer":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"state":"REVIEWER"},{"updated":"2021-08-03 10:14:51.000000000","updated_by":{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-08-03 10:14:51.000000000","updated_by":{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-08-05 08:08:52.000000000","updated_by":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"reviewer":{"_account_id":29088,"name":"Candido Campos Rivas","email":"ccamposr@redhat.com","username":"ccamposr"},"state":"REVIEWER"},{"updated":"2021-08-06 11:49:34.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"9d876747cbc37585c4271e77391fc293d9c8b8dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},"date":"2021-08-03 07:25:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"858093063d0eb829266b8a5d70b5899fee48b5f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 10:10:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0022175f8ac4586b0561a43ce8da24d : SUCCESS in 4m 15s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2c8ffd60dd614c6a89fd697f03f5c1bd : SUCCESS in 41m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/986dc5a532fd44d6992fd03d0d1dc470 : SUCCESS in 1h 37m 30s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/72ecdb13c85d4d5eb904a6c49b14648b : SUCCESS in 2h 00m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/53e9eb4643d843188610e19f04f023a3 : SUCCESS in 1h 28m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/aba4b506e567417db86a7e03eb8f122e : SUCCESS in 2h 38m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/09d2407a0b3a4996b536e3872cd73cd0 : SUCCESS in 33m 40s\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/088e06decbdb491ba599ad392c12ad75 : SUCCESS in 52m 05s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/97ec409b3f4b4b46b14a492091bea789 : SUCCESS in 2h 29m 05s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/eeba4ac62a7c4d90ae8e399ab8a36399 : SUCCESS in 2h 12m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/d5073cd30e5f40f4b70462304ba7b08b : SUCCESS in 1h 57m 58s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/0e27f7ad87e4489cad63de6c28eb9261 : SUCCESS in 2h 04m 13s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/af520f3d611f41408da41f1e4bed0692 : SUCCESS in 33m 17s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/4fa19077537044a69ef917935cb543d3 : SUCCESS in 53m 23s\n- neutron-tempest-plugin-scenario-linuxbridge-victoria https://zuul.opendev.org/t/openstack/build/b2f35990e21e4713870155d73130808c : SUCCESS in 1h 59m 17s\n- neutron-tempest-plugin-scenario-openvswitch-victoria https://zuul.opendev.org/t/openstack/build/0e35f845d8c248da8893f7fc843ea623 : SUCCESS in 1h 55m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-victoria https://zuul.opendev.org/t/openstack/build/bbfb51f0b1964afaa3c6fff643961708 : SUCCESS in 1h 26m 47s\n- neutron-tempest-plugin-scenario-ovn-victoria https://zuul.opendev.org/t/openstack/build/5bd269311db74777a7780bc16cd93828 : SUCCESS in 1h 51m 45s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/abe4de6a523646caa26e38eba1656620 : SUCCESS in 32m 57s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/8a2e5392007a4264876cec323c7f82e4 : SUCCESS in 57m 20s\n- neutron-tempest-plugin-scenario-linuxbridge-wallaby https://zuul.opendev.org/t/openstack/build/dc43264ce7ad48ca95069ac68b419385 : SUCCESS in 1h 59m 58s\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/8694104be663422c9fb0b0afb3527505 : SUCCESS in 2h 17m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-wallaby https://zuul.opendev.org/t/openstack/build/3f96dadfa3154a1a801dabb30af09e24 : SUCCESS in 1h 29m 02s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/3fc7300efe5243e790c3acb045a914ad : SUCCESS in 1h 59m 27s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/07297ee76f1e4dbfa58f340b4646d068 : SUCCESS in 42m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6ddbb929cff4a2681fb40e250c9bfbd : SUCCESS in 4m 11s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/9fae4cdf8a6a411fb9e4b2b37a2afd62 : SUCCESS in 1h 15m 59s\n- neutron-tempest-plugin-sfc-ussuri https://zuul.opendev.org/t/openstack/build/038445c432924e7d887160b50b4afeec : SUCCESS in 1h 24m 23s\n- neutron-tempest-plugin-sfc-victoria https://zuul.opendev.org/t/openstack/build/65c89a35985e461dace3a259f9c837ec : SUCCESS in 1h 22m 23s\n- neutron-tempest-plugin-sfc-wallaby https://zuul.opendev.org/t/openstack/build/71d83a783f3b405987a400a9298faf1e : SUCCESS in 1h 29m 26s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/52da24bbb7684b68b946cb1231a521a8 : SUCCESS in 1h 09m 50s\n- neutron-tempest-plugin-bgpvpn-bagpipe-ussuri https://zuul.opendev.org/t/openstack/build/9297c5006fe94ee39d7f19ac57301ea2 : SUCCESS in 1h 12m 43s\n- neutron-tempest-plugin-bgpvpn-bagpipe-victoria https://zuul.opendev.org/t/openstack/build/4218dea5416d43e8ba8b00895d3d95f3 : SUCCESS in 1h 20m 36s\n- neutron-tempest-plugin-bgpvpn-bagpipe-wallaby https://zuul.opendev.org/t/openstack/build/fbabf4f827f148acab2159bd2433d149 : SUCCESS in 1h 11m 21s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/a310cd7afa4849d68f281633c8249ef7 : SUCCESS in 43m 51s\n- neutron-tempest-plugin-dynamic-routing-ussuri https://zuul.opendev.org/t/openstack/build/82126bc62d7e42b889128197e3becdc7 : SUCCESS in 41m 46s\n- neutron-tempest-plugin-dynamic-routing-victoria https://zuul.opendev.org/t/openstack/build/111238e316044e9abea074f3295a08b5 : SUCCESS in 41m 01s\n- neutron-tempest-plugin-dynamic-routing-wallaby https://zuul.opendev.org/t/openstack/build/9b6a70a4b3794d5b902c12c636b6e1c8 : SUCCESS in 45m 30s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/97cc0f272dcb4e6aa6fd727519cd692b : SUCCESS in 32m 35s\n- neutron-tempest-plugin-vpnaas-ussuri https://zuul.opendev.org/t/openstack/build/e082c27dfc2a4a0aa5b92fa017069c95 : SUCCESS in 43m 40s\n- neutron-tempest-plugin-vpnaas-victoria https://zuul.opendev.org/t/openstack/build/4bbdf297a7924d71b67bb6081b09f195 : SUCCESS in 34m 56s\n- neutron-tempest-plugin-vpnaas-wallaby https://zuul.opendev.org/t/openstack/build/67d4b980e83548f8a7b97743a2ef600a : SUCCESS in 36m 09s","accounts_in_message":[],"_revision_number":1},{"id":"e240efaf5be972fff63a64d48a8f44622981dd10","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-04 09:10:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"62a1e00ba136e613da76b183caaee212a94e56d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},"date":"2021-08-05 08:12:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b887fef6d5777417370b0585d2bc6c6b56c27c5c","author":{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},"date":"2021-08-05 08:14:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c96f06bafb94ba63f5000abf770a575748432268","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 11:19:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/695fcc03bce540feb5edbf71cbd81962 : SUCCESS in 5m 26s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/be68628a62be43a0bd2d184aa3730d96 : SUCCESS in 35m 36s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/15ef3ce06d3744b493c6f11fcf771274 : SUCCESS in 2h 18m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5cabe06dbb4c4d11be9616a7cfdbb238 : SUCCESS in 1h 51m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9c9965c7fbdd4788ad265bf6810fad6f : SUCCESS in 1h 41m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/151e5f6709014ef69a7a0e8b85fcb070 : SUCCESS in 1h 48m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a8f30d8387584e289eded2bb3007740a : SUCCESS in 34m 05s\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/998c383b196b4aacac1edf71a637f13c : SUCCESS in 50m 33s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/cb1e4b0612b74ecfa9dbac38661e74c7 : SUCCESS in 2h 06m 55s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/d54ba4e2386c4b73ab528e8ec07d0b6f : SUCCESS in 1h 59m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/46b4289759f64301be4ef4dcc957c6e1 : SUCCESS in 1h 48m 06s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/221fc0fe6fed46c091ab40e38a335bed : SUCCESS in 2h 16m 42s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/ec358db5f39f42bba519865dfee7afdd : SUCCESS in 35m 16s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/6f1fc0cae6ea4ec797dd5b286a1ae423 : SUCCESS in 49m 15s\n- neutron-tempest-plugin-scenario-linuxbridge-victoria https://zuul.opendev.org/t/openstack/build/0fe93fb409704c4fb748046185f663f4 : SUCCESS in 1h 55m 54s\n- neutron-tempest-plugin-scenario-openvswitch-victoria https://zuul.opendev.org/t/openstack/build/11d5a55042f94c0295b5f7ba43d34395 : SUCCESS in 1h 53m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-victoria https://zuul.opendev.org/t/openstack/build/cec5743569c44a41b9478f849552a677 : SUCCESS in 1h 25m 50s\n- neutron-tempest-plugin-scenario-ovn-victoria https://zuul.opendev.org/t/openstack/build/1be839556e4f453f9207109250b906c7 : SUCCESS in 1h 35m 09s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/4f336c1c70804ddca902ba5c186f259a : SUCCESS in 28m 57s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/46ad3af471a948c6829d1246a7b166e8 : SUCCESS in 43m 33s\n- neutron-tempest-plugin-scenario-linuxbridge-wallaby https://zuul.opendev.org/t/openstack/build/6fc876416b5447218c12fdc8f8dfb47e : SUCCESS in 1h 53m 48s\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/a4b6a0a63d9941f7867f7bba4fe51fd7 : SUCCESS in 2h 28m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-wallaby https://zuul.opendev.org/t/openstack/build/03659526544844278d765c0ae2c2a1c3 : SUCCESS in 1h 27m 11s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/5bd607a56d074625b837fc94b202947e : SUCCESS in 2h 04m 17s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/fc64ff35bdcc4cd092cea0b229dc3e36 : SUCCESS in 34m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28d8fa1f90a74c67874afcab144353d3 : SUCCESS in 4m 29s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/026e5f3e71384ccfaa96833ab5c4842a : SUCCESS in 1h 16m 54s\n- neutron-tempest-plugin-sfc-ussuri https://zuul.opendev.org/t/openstack/build/6037f8d1f6014bafa1ce9715c63554db : SUCCESS in 1h 12m 54s\n- neutron-tempest-plugin-sfc-victoria https://zuul.opendev.org/t/openstack/build/a34f3e1bd2e549719bee65ecabd98e39 : SUCCESS in 1h 26m 32s\n- neutron-tempest-plugin-sfc-wallaby https://zuul.opendev.org/t/openstack/build/84e5c16484f246bc88b93103e5d075ed : SUCCESS in 1h 24m 14s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/f2b194b089a6417eb9fa446e86bfce46 : SUCCESS in 1h 10m 06s\n- neutron-tempest-plugin-bgpvpn-bagpipe-ussuri https://zuul.opendev.org/t/openstack/build/e7b83dc911fc457a83c5007bbe0d2814 : SUCCESS in 1h 10m 18s\n- neutron-tempest-plugin-bgpvpn-bagpipe-victoria https://zuul.opendev.org/t/openstack/build/e7be7590a6b94138ad6dc7040acb85ba : SUCCESS in 1h 21m 31s\n- neutron-tempest-plugin-bgpvpn-bagpipe-wallaby https://zuul.opendev.org/t/openstack/build/c6d53ec620d2483981f61b893f68b168 : SUCCESS in 1h 08m 32s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/f6dcbb1b749f478ca3d24fd872fcfb27 : SUCCESS in 44m 23s\n- neutron-tempest-plugin-dynamic-routing-ussuri https://zuul.opendev.org/t/openstack/build/188fc030117548cebbcaa1167dc9ea9b : SUCCESS in 44m 11s\n- neutron-tempest-plugin-dynamic-routing-victoria https://zuul.opendev.org/t/openstack/build/2bd979e4f0a34d7ba794a5d6172b1862 : SUCCESS in 43m 48s\n- neutron-tempest-plugin-dynamic-routing-wallaby https://zuul.opendev.org/t/openstack/build/272d4d263fc64be4a95e0d2ad4397383 : SUCCESS in 48m 19s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/e32bcea383db4673a99d27cd7279fe3f : SUCCESS in 29m 38s\n- neutron-tempest-plugin-vpnaas-ussuri https://zuul.opendev.org/t/openstack/build/ca20a8b60c40463fb9b16f773aef89fb : SUCCESS in 43m 15s\n- neutron-tempest-plugin-vpnaas-victoria https://zuul.opendev.org/t/openstack/build/895d37e89ae846f69bcee31f5773d6a1 : SUCCESS in 34m 24s\n- neutron-tempest-plugin-vpnaas-wallaby https://zuul.opendev.org/t/openstack/build/8137181bda994fdd8feb5d61572adb33 : SUCCESS in 35m 07s","accounts_in_message":[],"_revision_number":2},{"id":"07108df4f68d983f975b6d97f4ea62b0f196fccf","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-05 13:33:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b71786cfae3e2e0bd705f283ba7536de1ee12dc3","author":{"_account_id":29088,"name":"Candido Campos Rivas","email":"ccamposr@redhat.com","username":"ccamposr"},"date":"2021-08-05 16:01:07.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ce27304e99ded4cace734406ad296369eefb2716","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-06 11:49:34.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"533c3e38160ab1d30df515b9e7a27581ca92290e","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2021-08-08 06:48:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b992b2e4aab9302dd270b85d4c1c42b899e768fc","author":{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},"date":"2021-08-08 07:46:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6499e481f9ada54fc1c8b1a350e8bb8febb09ba4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-09 09:19:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2da2a1a219f888318368d0e4cd5cfa8d3e0f3e8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 09:20:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b093833c4eb8afb85c3fa8c1d5a71b6f8ad55b71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 11:49:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/093d8ce121a641f58c5b1c363e4089d4 : SUCCESS in 5m 41s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e0649385f0484ecc866e1b3c11aa1e3d : SUCCESS in 48m 38s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b4b38e2787344a30aa6e17e04ef56bd5 : SUCCESS in 2h 03m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a7a5103acea246cb8b3d1b9ee8a388da : SUCCESS in 2h 03m 43s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/df0a57a821b8406186de003e8be65fa4 : SUCCESS in 2h 20m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/aee1c4f580bb4e92ac31dd8ba87d04ff : SUCCESS in 1h 29m 24s\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/7fd4063f62464a77870e1ef6218867a1 : SUCCESS in 48m 21s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/bbcf50787afd41d4940f7de706526633 : SUCCESS in 46m 03s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/61efd1c03c1946bba0d993999b6fee4f : SUCCESS in 47m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ccab23c660842568ce4db067dad356e : SUCCESS in 4m 12s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/b55384ba34284bca8a1d6715708c11e8 : SUCCESS in 1h 22m 27s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/aa262037b84f4974a634fb613d3fde31 : SUCCESS in 1h 04m 21s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/1025a0666a9249b9a9abea999f08bccd : SUCCESS in 40m 30s","accounts_in_message":[],"_revision_number":2},{"id":"5530ba1e2ed44de23daccafa98bb4716174bae33","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 11:49:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5d1043b27afb7b9f9828e98ba426c1e4d075d9e9","revisions":{"9ab16804b5d6ed07dd49bd3db1ac06549adfe6d5":{"kind":"REWORK","_number":1,"created":"2021-08-03 07:25:52.000000000","uploader":{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},"ref":"refs/changes/15/803315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/15/803315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/15/803315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/15/803315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/15/803315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/15/803315/1"}}},"commit":{"parents":[{"commit":"0e98f828abdd9397b16a025519bb8cc88e06fb82","subject":"Merge \"Enable tls-proxy in jobs where it was disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0e98f828abdd9397b16a025519bb8cc88e06fb82"}]}],"author":{"name":"Alex Katz","email":"akatz@redhat.com","date":"2021-08-03 07:21:43.000000000","tz":180},"committer":{"name":"Alex Katz","email":"akatz@redhat.com","date":"2021-08-03 07:21:43.000000000","tz":180},"subject":"Fix StatefulConnection to not to fail on exit","message":"Fix StatefulConnection to not to fail on exit\n\nSome of the images require `tail` process to be explicitly stopped\nafter the `tail -f ... | nc ...` command has been executed and `nc`\nprocess is finished. But most of the time `tail` process is ended\nautomatically that leads to fail of `killall tail` command.\n\nChange-Id: I43c0b7781ca2ed98156e125566115196aab530cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9ab16804b5d6ed07dd49bd3db1ac06549adfe6d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9ab16804b5d6ed07dd49bd3db1ac06549adfe6d5"}]},"branch":"refs/heads/master"},"5d1043b27afb7b9f9828e98ba426c1e4d075d9e9":{"kind":"REWORK","_number":2,"created":"2021-08-05 08:12:40.000000000","uploader":{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},"ref":"refs/changes/15/803315/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/15/803315/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/15/803315/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/15/803315/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/15/803315/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/15/803315/2"}}},"commit":{"parents":[{"commit":"0e98f828abdd9397b16a025519bb8cc88e06fb82","subject":"Merge \"Enable tls-proxy in jobs where it was disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0e98f828abdd9397b16a025519bb8cc88e06fb82"}]}],"author":{"name":"Alex Katz","email":"akatz@redhat.com","date":"2021-08-03 07:21:43.000000000","tz":180},"committer":{"name":"Alex Katz","email":"akatz@redhat.com","date":"2021-08-05 08:12:30.000000000","tz":180},"subject":"Fix StatefulConnection to not to fail on exit","message":"Fix StatefulConnection to not to fail on exit\n\nSome of the images require `tail` process to be explicitly stopped\nafter the `tail -f ... | nc ...` command has been executed and `nc`\nprocess is finished. But most of the time `tail` process is ended\nautomatically that leads to fail of `killall tail` command.\n\nChange-Id: I43c0b7781ca2ed98156e125566115196aab530cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5d1043b27afb7b9f9828e98ba426c1e4d075d9e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5d1043b27afb7b9f9828e98ba426c1e4d075d9e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
