)]}'
{"id":"openstack%2Fneutron-tempest-plugin~821965","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ic27c9f9996ce2a2541e93d4ba5a17b4269100842","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"dnm","hashtags":[],"change_id":"Ic27c9f9996ce2a2541e93d4ba5a17b4269100842","subject":"[DNM] Test switch","status":"ABANDONED","created":"2021-12-16 11:31:14.000000000","updated":"2021-12-21 05:38:04.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f0b47ef36b8b05ad87147ee2e28645e558b475f7","_number":821965,"virtual_id_number":821965,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-12-21 05:36:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-21 05:36:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b7c261207638cdfa494f0e289df4222d69933c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-12-16 11:31:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"559e11893a7bf661442b56f5670a8879d71361ee","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-12-16 11:35:16.000000000","message":"Abandoned\n\nNot worked need to wait for parent patch to merge https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/821939","accounts_in_message":[],"_revision_number":1},{"id":"ca064ed001a0661a14af45961a1dfb3231881bbe","tag":"autogenerated:gerrit:restore","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-12-21 04:54:07.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"d8a6745f0a4f4b3cf943f33a12268e60fae52f39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-12-21 04:54:15.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"667460a5c6ebff71f7e01787897f6ec2ffed7b0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-12-21 05:01:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8c0a1e4b22fab8e53e3cd8b63f9c22cc1b18eaaf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-12-21 05:03:31.000000000","message":"Abandoned\n\nsomehow not working need to dig further","accounts_in_message":[],"_revision_number":3},{"id":"56ac3ed362e2a0f414b8f74b384aea04e9386d7d","tag":"autogenerated:gerrit:restore","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-12-21 05:13:59.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"668f4444212e67a187f6ce6c574d3613e106e90a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-12-21 05:15:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"358d258608be7adba3d9e79b08e98168268c3607","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-12-21 05:20:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e730450d37b7507d80974dc63fc3fb40d5619bae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-12-21 05:24:09.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3b39fa44ab42150bd5dda21a144e2edf7d225a80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-12-21 05:31:28.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"faf5362f9d97515f4743ed3ccea6dd8760a0a46b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 05:36:13.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c901f0b12dd0417fbd7f21c347efc6e0 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ab26ac1280047c5ad1e03f179316a7f : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":7},{"id":"bfa9889fd89ce244bf90e5c0030d3a79f75a7506","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-12-21 05:38:00.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f0b47ef36b8b05ad87147ee2e28645e558b475f7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-12-21 05:38:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"d8bfb48c6bf87328ce8f75e357daf1bbbb96a23b","revisions":{"7c3686572fb07918556273cce4bb6d42502a0372":{"kind":"REWORK","_number":1,"created":"2021-12-16 11:31:14.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/65/821965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/65/821965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/1"}}},"commit":{"parents":[{"commit":"f3c2db18e909da06d4ecd1f4ac541dbd505262ac","subject":"Restructure move to nested virt nodes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f3c2db18e909da06d4ecd1f4ac541dbd505262ac"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-12-16 11:31:00.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-12-16 11:31:00.000000000","tz":330},"subject":"[DNM] Test switch","message":"[DNM] Test switch\n\nChange-Id: Ic27c9f9996ce2a2541e93d4ba5a17b4269100842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7c3686572fb07918556273cce4bb6d42502a0372"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7c3686572fb07918556273cce4bb6d42502a0372"}]},"branch":"refs/heads/master"},"7624f66b675cc2f33f1f10dd9c3e61c77f8b2480":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-12-21 04:54:15.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/65/821965/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/65/821965/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/2"}}},"commit":{"parents":[{"commit":"864d1b9a1655c36e9ab2b52390330fc5e129def7","subject":"Merge \"Set SG quota for specific project.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/864d1b9a1655c36e9ab2b52390330fc5e129def7"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-12-16 11:31:00.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2021-12-21 04:54:15.000000000","tz":0},"subject":"[DNM] Test switch","message":"[DNM] Test switch\n\nChange-Id: Ic27c9f9996ce2a2541e93d4ba5a17b4269100842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7624f66b675cc2f33f1f10dd9c3e61c77f8b2480"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7624f66b675cc2f33f1f10dd9c3e61c77f8b2480"}]},"branch":"refs/heads/master","description":"Rebase"},"de97f3be618ce2711c26b29ca963b43cf1651156":{"kind":"REWORK","_number":3,"created":"2021-12-21 05:01:42.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/65/821965/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/65/821965/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/3"}}},"commit":{"parents":[{"commit":"864d1b9a1655c36e9ab2b52390330fc5e129def7","subject":"Merge \"Set SG quota for specific project.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/864d1b9a1655c36e9ab2b52390330fc5e129def7"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-12-16 11:31:00.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-12-21 05:01:29.000000000","tz":330},"subject":"[DNM] Test switch","message":"[DNM] Test switch\n\nChange-Id: Ic27c9f9996ce2a2541e93d4ba5a17b4269100842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/de97f3be618ce2711c26b29ca963b43cf1651156"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/de97f3be618ce2711c26b29ca963b43cf1651156"}]},"branch":"refs/heads/master"},"a395fc1552e815e1e23fb71bc523c4332515cdc0":{"kind":"REWORK","_number":4,"created":"2021-12-21 05:15:10.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/65/821965/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/65/821965/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/4"}}},"commit":{"parents":[{"commit":"864d1b9a1655c36e9ab2b52390330fc5e129def7","subject":"Merge \"Set SG quota for specific project.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/864d1b9a1655c36e9ab2b52390330fc5e129def7"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-12-16 11:31:00.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-12-21 05:13:33.000000000","tz":330},"subject":"[DNM] Test switch","message":"[DNM] Test switch\n\nChange-Id: Ic27c9f9996ce2a2541e93d4ba5a17b4269100842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a395fc1552e815e1e23fb71bc523c4332515cdc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a395fc1552e815e1e23fb71bc523c4332515cdc0"}]},"branch":"refs/heads/master"},"eba0e2a85267f5e9c4cc6ac90090b93a62c122c3":{"kind":"REWORK","_number":5,"created":"2021-12-21 05:20:45.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/65/821965/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/65/821965/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/5"}}},"commit":{"parents":[{"commit":"864d1b9a1655c36e9ab2b52390330fc5e129def7","subject":"Merge \"Set SG quota for specific project.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/864d1b9a1655c36e9ab2b52390330fc5e129def7"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-12-16 11:31:00.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-12-21 05:20:19.000000000","tz":330},"subject":"[DNM] Test switch","message":"[DNM] Test switch\n\nChange-Id: Ic27c9f9996ce2a2541e93d4ba5a17b4269100842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/eba0e2a85267f5e9c4cc6ac90090b93a62c122c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/eba0e2a85267f5e9c4cc6ac90090b93a62c122c3"}]},"branch":"refs/heads/master"},"03f245a4062d54b472e6e285ac41edbbd3603299":{"kind":"REWORK","_number":6,"created":"2021-12-21 05:24:09.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/65/821965/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/65/821965/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/6"}}},"commit":{"parents":[{"commit":"864d1b9a1655c36e9ab2b52390330fc5e129def7","subject":"Merge \"Set SG quota for specific project.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/864d1b9a1655c36e9ab2b52390330fc5e129def7"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-12-16 11:31:00.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-12-21 05:23:58.000000000","tz":330},"subject":"[DNM] Test switch","message":"[DNM] Test switch\n\nChange-Id: Ic27c9f9996ce2a2541e93d4ba5a17b4269100842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/03f245a4062d54b472e6e285ac41edbbd3603299"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/03f245a4062d54b472e6e285ac41edbbd3603299"}]},"branch":"refs/heads/master"},"d8bfb48c6bf87328ce8f75e357daf1bbbb96a23b":{"kind":"REWORK","_number":7,"created":"2021-12-21 05:31:28.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/65/821965/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/65/821965/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/65/821965/7"}}},"commit":{"parents":[{"commit":"864d1b9a1655c36e9ab2b52390330fc5e129def7","subject":"Merge \"Set SG quota for specific project.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/864d1b9a1655c36e9ab2b52390330fc5e129def7"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-12-16 11:31:00.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-12-21 05:31:08.000000000","tz":330},"subject":"[DNM] Test switch","message":"[DNM] Test switch\n\nChange-Id: Ic27c9f9996ce2a2541e93d4ba5a17b4269100842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d8bfb48c6bf87328ce8f75e357daf1bbbb96a23b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d8bfb48c6bf87328ce8f75e357daf1bbbb96a23b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
