)]}'
{"id":"openstack%2Fneutron-tempest-plugin~827595","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I1619935c5a09aeab9608a7b06304efef7fef4ed2","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"neutron-ci-improvements","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-02-03 07:25:24.000000000","reason":"yatin replied on the change"}},"hashtags":[],"change_id":"I1619935c5a09aeab9608a7b06304efef7fef4ed2","subject":"[DNM] Test job triggers with parent patch","status":"ABANDONED","created":"2022-02-03 07:08:00.000000000","updated":"2022-02-03 07:35:48.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d53ec071defe819cc9b85c82e82c0a4f1a14a7cc","_number":827595,"virtual_id_number":827595,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-03 07:24:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7775d4b310f4572a851ef95116798eede60494b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-03 07:08:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"370eb4105be48ccc8667065b54ea3e0a39600e61","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-03 07:10:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7d864527469b2d00e6134d43c8bde2ee5d7c616b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-03 07:12:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c4cb9634a1d1c2048de7dd0416785b3715eb67db","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-03 07:16:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cb1a9b33c83767ab1df6464a2c21c721caad71c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-03 07:23:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0b5c19f57fb370a1cdac513ffa607331748ef9c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 07:24:25.000000000","message":"Patch Set 3: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron-tempest-plugin on branch master.  The error\nwas:\n\n  expected str @ data[\u0027irrelevant-files\u0027][32]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: neutron-tempest-plugin-scenario-openvswitch\n      parent: neutron-tempest-plugin-scenario-nested-switch\n      timeout: 10000\n      vars:\n        devstack_services:\n          # Disable OVN services\n          br-ex-tcpdump: false\n          br-int-flows: false\n          ovn-controller: false\n          ovn-northd: false\n          ovs-vswitchd: false\n          ovsdb-server: false\n          q-ovn-metadata-agent: false\n          # Neutron services\n          neutron-local-ip-static: true\n          q-agt: true\n          q-dhcp: true\n          q-l3: true\n          q-meta: true\n          q-metering: true\n        network_api_extensions: *api_extensions\n        network_api_extensions_openvswitch:\n          - local_ip\n        network_available_features: *available_features\n        devstack_localrc:\n          Q_AGENT: openvswitch\n          Q_ML2_TENANT_NETWORK_TYPE: vxlan\n          Q_ML2_PLUGIN_MECHANISM_DRIVERS: openvswitch\n          NETWORK_API_EXTENSIONS: \"{{ (network_api_extensions + network_api_extensions_openvswitch) | join(\u0027,\u0027) }}\"\n        devstack_local_conf:\n          post-config:\n            $NEUTRON_CONF:\n              DEFAULT:\n                enable_dvr: false\n                l3_ha: true\n            # NOTE(slaweq): We can get rid of this hardcoded absolute path when\n            # devstack-tempest job will be switched to use lib/neutron instead of\n            # lib/neutron-legacy\n            /$NEUTRON_CORE_PLUGIN_CONF:\n              agent:\n                tunnel_types: vxlan,gre\n              ovs:\n                tunnel_bridge: br-tun\n                bridge_mappings: public:br-ex\n          test-config:\n            $TEMPEST_CONFIG:\n              network-feature-enabled:\n                available_features: \"{{ network_available_features | join(\u0027,\u0027) }}\"\n              neutron_plugin_options:\n                available_type_drivers: flat,vlan,local,vxlan\n                firewall_driver: openvswitch\n      irrelevant-files: \u0026openvswitch-scenario-irrelevant-files\n        - ^(test-|)requirements.txt$\n        - lower-constraints.txt\n        - ^releasenotes/.*$\n        - ^doc/.*$\n        - ^setup.cfg$\n        - ^.*\\.rst$\n        - ^.*\\.conf\\.sample$\n        - ^neutron/locale/.*$\n        - ^neutron/tests/unit/.*$\n        - ^neutron/tests/fullstack/.*\n        - ^neutron/tests/functional/.*\n        - ^tools/.*$\n        - ^tox.ini$\n        - ^neutron/agent/ovn/.*$\n        - ^neutron/agent/windows/.*$\n        - ^neutron/plugins/ml2/drivers/linuxbridge/.*$\n        - ^neutron/plugins/ml2/drivers/macvtap/.*$\n        - ^neutron/plugins/ml2/drivers/mech_sriov/.*$\n        - ^neutron/plugins/ml2/drivers/ovn/.*$\n        - ^neutron/services/ovn_l3/.*$\n        - ^neutron/services/logapi/drivers/ovn/.*$\n        - ^neutron/services/portforwarding/drivers/ovn/.*$\n        - ^neutron/services/qos/drivers/linuxbridge/.*$\n        - ^neutron/services/qos/drivers/ovn/.*$\n        - ^neutron/services/trunk/drivers/linuxbridge/.*$\n        - ^neutron/services/trunk/drivers/ovn/.*$\n        - ^neutron/cmd/ovn/.*$\n        - ^neutron/common/ovn/.*$\n        - ^neutron_tempest_plugin/api/test_.*$\n        - ^neutron_tempest_plugin/(bgpvpn|fwaas|neutron_dynamic_routing|sfc|tap_as_a_service|vpnaas).*$\n        - ^neutron_tempest_plugin/services/bgp/.*$\n        - ^rally-jobs/.*$\n        - - ^vagrant/.*$\n        - ^zuul.d/(?!(project)).*\\.yaml\n\n  in \"openstack/neutron-tempest-plugin/zuul.d/master_jobs.yaml@master\", line 137, column 3","accounts_in_message":[],"_revision_number":3},{"id":"038112ed945b87f969f2f5d7788eb2050ed04bdc","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-03 07:25:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aab7506ded3f21d6afcb3d291605c84e0d48e676","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-03 07:27:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"41ae2a6f498047fd417a86a1202c90b99ad32365","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-03 07:30:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"afb97c9becbec1b40a7376c607c5fed8195e5e8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-03 07:31:41.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ef7b0f54ab1de000108566a55ac27adf3b3d9960","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-03 07:34:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d53ec071defe819cc9b85c82e82c0a4f1a14a7cc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-03 07:35:48.000000000","message":"Abandoned\n\nCan be resumed for more tests are needed","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4a2e6cfa4d95823d02ef7eeb9f1f8d5dbf76f45a","revisions":{"1f05bdbc57f94b0360bc6fd6b83e035c539622eb":{"kind":"REWORK","_number":1,"created":"2022-02-03 07:08:00.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/95/827595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/827595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/827595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/827595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/827595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/827595/1"}}},"commit":{"parents":[{"commit":"77bfe311d25417a90d9b7761e789d6a08c3f50c4","subject":"Update irrelevant-files to skip run of untouched jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/77bfe311d25417a90d9b7761e789d6a08c3f50c4"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-03 07:07:13.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-03 07:07:13.000000000","tz":330},"subject":"[DNM] Test job triggers with parent patch","message":"[DNM] Test job triggers with parent patch\n\nChange-Id: I1619935c5a09aeab9608a7b06304efef7fef4ed2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1f05bdbc57f94b0360bc6fd6b83e035c539622eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1f05bdbc57f94b0360bc6fd6b83e035c539622eb"}]},"branch":"refs/heads/master"},"f24692192d3c97c2ee6fedbd0886f3f0d0fe67fd":{"kind":"REWORK","_number":2,"created":"2022-02-03 07:12:04.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/95/827595/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/827595/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/827595/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/827595/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/827595/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/827595/2"}}},"commit":{"parents":[{"commit":"77bfe311d25417a90d9b7761e789d6a08c3f50c4","subject":"Update irrelevant-files to skip run of untouched jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/77bfe311d25417a90d9b7761e789d6a08c3f50c4"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-03 07:07:13.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-03 07:11:45.000000000","tz":330},"subject":"[DNM] Test job triggers with parent patch","message":"[DNM] Test job triggers with parent patch\n\nChange-Id: I1619935c5a09aeab9608a7b06304efef7fef4ed2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f24692192d3c97c2ee6fedbd0886f3f0d0fe67fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f24692192d3c97c2ee6fedbd0886f3f0d0fe67fd"}]},"branch":"refs/heads/master"},"b70f5602acf148ea7997732d11c29b0f7455d4bb":{"kind":"REWORK","_number":3,"created":"2022-02-03 07:23:44.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/95/827595/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/827595/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/827595/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/827595/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/827595/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/827595/3"}}},"commit":{"parents":[{"commit":"77bfe311d25417a90d9b7761e789d6a08c3f50c4","subject":"Update irrelevant-files to skip run of untouched jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/77bfe311d25417a90d9b7761e789d6a08c3f50c4"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-03 07:07:13.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-03 07:22:52.000000000","tz":330},"subject":"[DNM] Test job triggers with parent patch","message":"[DNM] Test job triggers with parent patch\n\nChange-Id: I1619935c5a09aeab9608a7b06304efef7fef4ed2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b70f5602acf148ea7997732d11c29b0f7455d4bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b70f5602acf148ea7997732d11c29b0f7455d4bb"}]},"branch":"refs/heads/master"},"1c8f2b7ba8e892e82ff70a2ddcf5cefda35d0bf1":{"kind":"REWORK","_number":4,"created":"2022-02-03 07:27:58.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/95/827595/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/827595/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/827595/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/827595/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/827595/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/827595/4"}}},"commit":{"parents":[{"commit":"77bfe311d25417a90d9b7761e789d6a08c3f50c4","subject":"Update irrelevant-files to skip run of untouched jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/77bfe311d25417a90d9b7761e789d6a08c3f50c4"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-03 07:07:13.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-03 07:27:38.000000000","tz":330},"subject":"[DNM] Test job triggers with parent patch","message":"[DNM] Test job triggers with parent patch\n\nChange-Id: I1619935c5a09aeab9608a7b06304efef7fef4ed2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1c8f2b7ba8e892e82ff70a2ddcf5cefda35d0bf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1c8f2b7ba8e892e82ff70a2ddcf5cefda35d0bf1"}]},"branch":"refs/heads/master"},"4a2e6cfa4d95823d02ef7eeb9f1f8d5dbf76f45a":{"kind":"REWORK","_number":5,"created":"2022-02-03 07:31:41.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/95/827595/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/95/827595/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/827595/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/827595/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/827595/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/95/827595/5"}}},"commit":{"parents":[{"commit":"77bfe311d25417a90d9b7761e789d6a08c3f50c4","subject":"Update irrelevant-files to skip run of untouched jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/77bfe311d25417a90d9b7761e789d6a08c3f50c4"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-03 07:07:13.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-03 07:31:10.000000000","tz":330},"subject":"[DNM] Test job triggers with parent patch","message":"[DNM] Test job triggers with parent patch\n\nChange-Id: I1619935c5a09aeab9608a7b06304efef7fef4ed2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4a2e6cfa4d95823d02ef7eeb9f1f8d5dbf76f45a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4a2e6cfa4d95823d02ef7eeb9f1f8d5dbf76f45a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
