)]}'
{"id":"openstack%2Fneutron-tempest-plugin~832911","triplet_id":"openstack%2Fneutron-tempest-plugin~master~If2badd6cbac13ddaf86fe81b48ebd69d2857f1b0","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"lp/1963907","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-03-10 12:17:40.000000000","reason":"Rodolfo Alonso replied on the change"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-03-10 13:50:29.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-03-10 12:23:10.000000000","reason":"Slawek Kaplonski replied on the change"},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2022-03-10 11:08:14.000000000","reason":"Alfredo Moralejo replied on the change"},"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2022-03-10 13:50:29.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-03-10 13:50:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If2badd6cbac13ddaf86fe81b48ebd69d2857f1b0","subject":"Move RULE_TYPE_MINIMUM_PACKET_RATE under setup method","status":"MERGED","created":"2022-03-10 04:24:07.000000000","updated":"2022-03-10 13:50:29.000000000","submitted":"2022-03-10 13:50:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832911-lp/1963907","meta_rev_id":"bf4529951dfa81b1b8db7d1451247ef1625490e0","_number":832911,"virtual_id_number":832911,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-10 13:50:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-03-10 12:17:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-03-10 08:32:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2022-03-10 11:34:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2022-03-10 08:17:46.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2022-03-10 08:58:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":2,"date":"2022-03-10 12:23:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-03-10 11:47:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":1,"date":"2022-03-10 12:23:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-10 04:24:33.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-03-10 06:50:19.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-03-10 06:50:19.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2022-03-10 06:50:54.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-03-10 06:50:54.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-03-10 06:57:49.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2022-03-10 07:32:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-10 08:58:58.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"state":"REVIEWER"},{"updated":"2022-03-10 11:34:36.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2022-03-10 11:47:09.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"}],"messages":[{"id":"9df516428f404c41d55b367e9d6d128f6f6cd17b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-03-10 04:24:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee02a9aba2f171ba94962a8f260c60c141cac421","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-03-10 04:47:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"164750235a2750abbbc809b29d3102e65d5ce754","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-03-10 05:41:11.000000000","message":"Abandoned\n\nhttps://review.opendev.org/c/openstack/neutron-tempest-plugin/+/832911","accounts_in_message":[],"_revision_number":2},{"id":"d328d39ab55d6f6ae5776a68cad44a39ff85bd3d","tag":"autogenerated:gerrit:restore","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-03-10 05:41:22.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"4832739fa5479307a02fb7f364c17a6705a3a5ae","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-03-10 06:22:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0bfffecfb626841a19cadfabec6da8e29e170bff","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-03-10 06:24:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1140fa980ea161cb11d1e08b8bdd20ed6ccf03d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-03-10 06:25:57.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"23fd26362113abc7b75490ca8fa57ed79828d452","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-03-10 06:47:53.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a69978dd37419241cbd425faa5c33f0ea028c0dc","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-03-10 06:57:49.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ead9f9aadc408cb39cb477dcb1c25f92719a28ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 07:32:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0a336006a8f419e83a023dbd2f1da3b : SUCCESS in 5m 05s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f8a62a9cb67347a7a542bf22f56cac81 : SUCCESS in 52m 18s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/6e7a754f00364f2399fec5b8b83679ec : SUCCESS in 52m 36s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/acbfabcbcaea4bdc8525cf5100926eed : SUCCESS in 48m 43s\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/75a49005b5fb43c7bb9fb39a9d45cf15 : SUCCESS in 1h 00m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d28c5da922e044dea8d7ec0d133194a8 : SUCCESS in 5m 29s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/f172d59328f340d18f58adad5c589e7a : SUCCESS in 33m 06s","accounts_in_message":[],"_revision_number":3},{"id":"21958b90e724b2dfb64e23ce7a4b849516e983b2","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-03-10 07:55:51.000000000","message":"Patch Set 3: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5cd694e2df52f00c2ed90e7c61b84abf4a29792f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-03-10 08:15:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"121d041537cc34404d0c8e4afe5ebf813defcead","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-03-10 08:16:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b788c655e3bf72d23bcdee0b488ff4e9260e07cc","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-03-10 08:17:46.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"2e431618499fc90ec316a8c0c380e34c45ac7763","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-03-10 08:32:41.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"69a21f5058aef52e20eed5fb877fe83470d219dc","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-03-10 08:58:58.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"59f02e5159097067b073cc061ae974583478b12e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 09:26:01.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2a4c1568d8243d2b9b0e6f5b089c1f1 : SUCCESS in 5m 33s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8dc4815dacbd4592ad25803ffdfd81b9 : SUCCESS in 37m 16s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/8606cb44165d4fee8756b1123a16c370 : SUCCESS in 45m 35s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/02f6a8622f2b4ea185370c2242caaf7b : SUCCESS in 51m 18s\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/4f5e83053d7d4b89ad7c2789b43ef374 : SUCCESS in 47m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa0c87e21b8f4906a1863c14fc345601 : SUCCESS in 5m 26s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/e1ef7521838e499fa88f917961a134fb : FAILURE in 40m 04s","accounts_in_message":[],"_revision_number":4},{"id":"ef878de818a6ade8a55c147b80935bdb6068c967","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-03-10 10:54:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"628064f396554d67396f4b88c3473ea40eb0d094","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-03-10 10:54:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d9c152cf77b0241e2ea37e79e972b3adaf3ddf13","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-03-10 11:08:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c7e510f7af136ec1ba5d1fa90ca102e270caa9a4","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-03-10 11:34:36.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"922d3397c05019ea24fb9efeab9b506af9ea111b","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-03-10 11:47:09.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"c9778608b5bb87d9be248ed02712feef6f4c3807","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 11:54:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/256c3169ad2a435a89c427e740bb47e2 : SUCCESS in 6m 05s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9c139f861d63499dbd889b1a07e309cc : SUCCESS in 49m 07s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/876f45a39e8048819e38590a3616fee2 : SUCCESS in 44m 45s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/3ecb4b5f1f464438802c436d4a19308f : SUCCESS in 43m 17s\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/392f14de1dfa48b58c0b97c1eac9536d : SUCCESS in 38m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9307597fbe364bad9e914dfcf51359e5 : SUCCESS in 5m 37s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/e4ef907c11e248a9bc1a233c0720cbec : SUCCESS in 26m 48s","accounts_in_message":[],"_revision_number":4},{"id":"cf6e5005a195bc26177cc23640143206b42d3867","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-10 12:17:40.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"5b434f465d62f4e198397b7b35089e571cbfcfdd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-10 12:23:10.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5c2cd4a7904d36749d9c0d6645700c4f042604cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 12:34:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"11f77cd6906d95d2e3ef8c51287c97e9fe8675f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 13:50:28.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ce58b7e8ee048c6aec6d1cb177c4f69 : SUCCESS in 6m 01s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3d9af45c28ed42168f24264474022a76 : SUCCESS in 47m 22s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/1f3d694eb4fa4332b2d225a2f44ef303 : SUCCESS in 44m 41s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/2dc8675cec5841a6b262e34b72ec2296 : SUCCESS in 46m 00s\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/0f7deddc8fd440f3ace5fc7227af5e3d : SUCCESS in 40m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82cfdd075a6d483fac1dad50ac888080 : SUCCESS in 7m 08s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/7eb57338cfcb40ecb2c8e21b7af061b7 : SUCCESS in 32m 13s","accounts_in_message":[],"_revision_number":4},{"id":"bf4529951dfa81b1b8db7d1451247ef1625490e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 13:50:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"255586cca98d0f49d44ed511279627b83995d925","revisions":{"466da5445356d58e9a2a1cd19b662eecef8d7775":{"kind":"REWORK","_number":1,"created":"2022-03-10 04:24:07.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/11/832911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/11/832911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/832911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/832911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/832911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/832911/1"}}},"commit":{"parents":[{"commit":"2ad4d583ec0ce4a5a48212a675cbf1762d121640","subject":"Honour ssh_key_type config for keypair and ssh clients","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2ad4d583ec0ce4a5a48212a675cbf1762d121640"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-03-10 04:13:37.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-03-10 04:24:02.000000000","tz":330},"subject":"Skip QosMinimumPpsRuleTestJSON when RULE_TYPE_MINIMUM_PACKET_RATE not supported","message":"Skip QosMinimumPpsRuleTestJSON when RULE_TYPE_MINIMUM_PACKET_RATE not supported\n\nOn Wallaby, We have pinned tempest[1] to particular commit[2]\nto fix lp#1963907.\n\nhttps://review.rdoproject.org/r/c/rdoinfo/+/40273 pinned\nneutron-tempest-plugin at master commit to fix neutron tempest\ntests in wallaby.\n\nBut the tempest tests are throwing\n```\nAttributeError: module \u0027neutron_lib.services.qos.constants\u0027\nhas no attribute \u0027RULE_TYPE_MINIMUM_PACKET_RATE\u0027\n```\n\nOn wallaby, neutron-lib\u003d2.10.2 is used in the RDO/Tripleo\nDeployment.\n\nhttps://review.opendev.org/c/openstack/neutron-lib/+/816447\nMoves RULE_TYPE_MINIMUM_PACKET_RATE to neutron-lib which is available\nin Yoga release and not available in wallaby neutron-lib\nleading to above failure.\n\nAdding the skip_checks to the test fixes the issue.\n\n[1]. https://review.opendev.org/c/openstack/tempest/+/822560\n[2]. https://review.rdoproject.org/r/c/rdoinfo/+/40228\n\nRelated-Bug: #1963907\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: If2badd6cbac13ddaf86fe81b48ebd69d2857f1b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/466da5445356d58e9a2a1cd19b662eecef8d7775"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/466da5445356d58e9a2a1cd19b662eecef8d7775"}]},"branch":"refs/heads/master"},"5475ea5eca5a3b2edfb98c6444b2ac2799c647ac":{"kind":"REWORK","_number":2,"created":"2022-03-10 04:47:45.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/11/832911/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/11/832911/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/832911/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/832911/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/832911/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/832911/2"}}},"commit":{"parents":[{"commit":"2ad4d583ec0ce4a5a48212a675cbf1762d121640","subject":"Honour ssh_key_type config for keypair and ssh clients","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2ad4d583ec0ce4a5a48212a675cbf1762d121640"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-03-10 04:13:37.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-03-10 04:47:34.000000000","tz":330},"subject":"Skip QosMinimumPpsRuleTestJSON when RULE_TYPE_MINIMUM_PACKET_RATE not supported","message":"Skip QosMinimumPpsRuleTestJSON when RULE_TYPE_MINIMUM_PACKET_RATE not supported\n\nOn Wallaby, We have pinned tempest[1] to particular commit[2]\nto fix lp#1963907.\n\nhttps://review.rdoproject.org/r/c/rdoinfo/+/40273 pinned\nneutron-tempest-plugin at master commit to fix neutron tempest\ntests in wallaby.\n\nBut the tempest tests are throwing\n```\nAttributeError: module \u0027neutron_lib.services.qos.constants\u0027\nhas no attribute \u0027RULE_TYPE_MINIMUM_PACKET_RATE\u0027\n```\n\nOn wallaby, neutron-lib\u003d2.10.2 is used in the RDO/Tripleo\nDeployment.\n\nhttps://review.opendev.org/c/openstack/neutron-lib/+/816447\nMoves RULE_TYPE_MINIMUM_PACKET_RATE to neutron-lib which is available\nin Yoga release and not available in wallaby neutron-lib\nleading to above failure.\n\nAdding the skip_checks to the test fixes the issue.\n\nIT also removes RULES_NAME unused var.\n\n[1]. https://review.opendev.org/c/openstack/tempest/+/822560\n[2]. https://review.rdoproject.org/r/c/rdoinfo/+/40228\n\nRelated-Bug: #1963907\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: If2badd6cbac13ddaf86fe81b48ebd69d2857f1b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5475ea5eca5a3b2edfb98c6444b2ac2799c647ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5475ea5eca5a3b2edfb98c6444b2ac2799c647ac"}]},"branch":"refs/heads/master"},"dd6f2c774fa2c0e70999926928c40979c6173b88":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-03-10 06:25:57.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/11/832911/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/11/832911/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/832911/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/832911/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/832911/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/832911/3"}}},"commit":{"parents":[{"commit":"2ad4d583ec0ce4a5a48212a675cbf1762d121640","subject":"Honour ssh_key_type config for keypair and ssh clients","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2ad4d583ec0ce4a5a48212a675cbf1762d121640"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-03-10 04:13:37.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-03-10 06:25:53.000000000","tz":330},"subject":"Skip QosMinimumPpsRuleTestJSON when RULE_TYPE_MINIMUM_PACKET_RATE not supported","message":"Skip QosMinimumPpsRuleTestJSON when RULE_TYPE_MINIMUM_PACKET_RATE not supported\n\nOn Wallaby, We have pinned tempest[1] to particular commit[2]\nto fix lp#1963907.\n\nhttps://review.rdoproject.org/r/c/rdoinfo/+/40273 pinned\nneutron-tempest-plugin at master commit to fix neutron tempest\ntests in wallaby.\n\nBut the tempest tests are throwing\n```\nAttributeError: module \u0027neutron_lib.services.qos.constants\u0027\nhas no attribute \u0027RULE_TYPE_MINIMUM_PACKET_RATE\u0027\n```\n\nOn wallaby, neutron-lib\u003d2.10.2 is used in the RDO/Tripleo\nDeployment.\n\nhttps://review.opendev.org/c/openstack/neutron-lib/+/816447\nMoves RULE_TYPE_MINIMUM_PACKET_RATE to neutron-lib which is available\nin Yoga release and not available in wallaby neutron-lib\nleading to above failure.\n\nAdding the skip_checks to the test fixes the issue.\n\n[1]. https://review.opendev.org/c/openstack/tempest/+/822560\n[2]. https://review.rdoproject.org/r/c/rdoinfo/+/40228\n\nRelated-Bug: #1963907\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: If2badd6cbac13ddaf86fe81b48ebd69d2857f1b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/dd6f2c774fa2c0e70999926928c40979c6173b88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/dd6f2c774fa2c0e70999926928c40979c6173b88"}]},"branch":"refs/heads/master"},"255586cca98d0f49d44ed511279627b83995d925":{"kind":"REWORK","_number":4,"created":"2022-03-10 08:15:47.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/11/832911/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/11/832911/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/832911/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/832911/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/832911/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/832911/4"}}},"commit":{"parents":[{"commit":"2ad4d583ec0ce4a5a48212a675cbf1762d121640","subject":"Honour ssh_key_type config for keypair and ssh clients","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2ad4d583ec0ce4a5a48212a675cbf1762d121640"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-03-10 04:13:37.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-03-10 08:12:41.000000000","tz":330},"subject":"Move RULE_TYPE_MINIMUM_PACKET_RATE under setup method","message":"Move RULE_TYPE_MINIMUM_PACKET_RATE under setup method\n\nOn Wallaby, We have pinned tempest[1] to particular commit[2]\nto fix lp#1963907.\n\nhttps://review.rdoproject.org/r/c/rdoinfo/+/40273 pinned\nneutron-tempest-plugin at master commit to fix neutron tempest\ntests in wallaby.\n\nBut the tempest tests are throwing\n```\nAttributeError: module \u0027neutron_lib.services.qos.constants\u0027\nhas no attribute \u0027RULE_TYPE_MINIMUM_PACKET_RATE\u0027\n```\n\nOn wallaby, neutron-lib\u003d2.10.2 is used in the RDO/Tripleo\nDeployment.\n\nhttps://review.opendev.org/c/openstack/neutron-lib/+/816447\nMoves RULE_TYPE_MINIMUM_PACKET_RATE to neutron-lib which is available\nin Yoga release and not available in wallaby neutron-lib\nleading to above failure.\n\nMoving the RULE_TYPE_MINIMUM_PACKET_RATE variable to the setup\nmethod fixes the issue.\n\n[1]. https://review.opendev.org/c/openstack/tempest/+/822560\n[2]. https://review.rdoproject.org/r/c/rdoinfo/+/40228\n\nRelated-Bug: #1963907\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: If2badd6cbac13ddaf86fe81b48ebd69d2857f1b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/255586cca98d0f49d44ed511279627b83995d925"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/255586cca98d0f49d44ed511279627b83995d925"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
